)]}'
{"id":"openstack%2Fswift~1006096","triplet_id":"openstack%2Fswift~master~I9b7b3c5c4cae425d4b7ad68e07226f956634a71e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-18 19:13:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b7b3c5c4cae425d4b7ad68e07226f956634a71e","subject":"s3api: Drop unnecessary backwards compat","status":"MERGED","created":"2026-09-17 18:37:22.000000000","updated":"2026-09-18 22:31:43.000000000","submitted":"2026-09-18 19:13:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006096","meta_rev_id":"0af5f8fac7dab160b89ac9f716c49c995e9615a5","_number":1006096,"virtual_id_number":1006096,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 19:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-09-18 17:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2026-09-18 18:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2026-09-18 18:05:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 19:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 19:50:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 17:00:58.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-09-18 18:05:15.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"bf0b419acd1945ebcfc8442278a27d0c295c81e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-17 18:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08a4ca74b5101218025936beb413f49a720d36de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be0a26df01e846daabb71ac48e9b67b6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4df6e1d8b7894f4eb6b80a7033490d8c : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e029264b81c217b4d4d407cd5a8045a2f7122de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:50:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef6dadc6a9c34c90819d904176e471de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ffd49082f4348be85f3b80c6cc3e487 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e23dfe6732704c7b98c90ec47854ed56 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/d9aec7acfef9477a80a2a1fe192271ad : SUCCESS in 34m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d06d3ee8b1c4da8a7dfae593a27ad2a : SUCCESS in 51m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61cdb65068e74815a6ceb8e57562ec5f : SUCCESS in 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a9d8e927ff647a1ac196b66069b7e79 : SUCCESS in 1h 04m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/940dbfefbe594f1ca2f5f9e01f9ee1f9 : SUCCESS in 12m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff26d473195a4cddb1b1e23d2b4b44c8 : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88d661dbe3a04adab532876878d61997 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f08bdd7afd9241699b2d114fe9da0c88 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9542fe8dea84d7d9a467e34f8c827ad : SUCCESS in 25m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e718cbbd87264731b857ccaf3ae7b629 : SUCCESS in 22m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f9bd788bcf1b48359a121127b573e057 : SUCCESS in 21m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0f0249ff51194867be303505fa55f735 : SUCCESS in 9m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/25078ca26410433db1e8c699df5a9ddc : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba5725dbbecf4dc3a7ff154b24c246a8 : SUCCESS in 9m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/01d72f93216a4d3e97e6e753bd1fafec : SUCCESS in 13m 49s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/81b3862efa2445b79c0ed066d3f9809e : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb8a47ecd5bc486d933cf4d30a58a7f3 : SUCCESS in 8m 12s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f8258b6861244f8fbfe9814e25847b8b : SUCCESS in 10m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cbbeb248a45485e8a837170979af425 : SUCCESS in 59m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f290eeb0b8164af7b2307796eabc2408 : SUCCESS in 52m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79e0fc66340545e29bfc550cab6b6150 : SUCCESS in 21m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b21b8f198334f23abc935fba32a3cc2 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e2a1ac12730473da5beb52b4ff75d6c : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8bf40acba6614c6eb86e1153590e82a0 : SUCCESS in 20m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77aac3806ae444979bd57190a7d32084 : SUCCESS in 56m 48s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1b73c1fcbf9a6a4ea94b3b05533c0ad9c5221f28","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-18 17:00:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"335c7d7073f365eb5ae348b9421ebc18eefa5863","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-18 18:05:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f717081dafa68f5a4ad0a00d7724182d8ea3999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:07:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9f9a4c5e9f08c42b4bc454d543f5b8e735a3c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:13:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1e7c9e7b154dacb8743cea07034d8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca16c6555c644feaa4799cf07153e55 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/255096ca1f4e4c95b7d9c48ff33a8caf : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/90fa1cde3dc24c2db02129bb46f67fbb : SUCCESS in 1h 04m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3083ea5c7a4e40f1a2da55c56fa7d381 : SUCCESS in 37m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/77b28c5daeb24e1680bce6b0058aa629 : SUCCESS in 41m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46354dc07f834ad3adaed4b6f96f0396 : SUCCESS in 1h 03m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/999087e1c198481e9de1caf866d423a4 : SUCCESS in 4m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/38e4b3b9f9934cc1b8a5745364e65ce5 : SUCCESS in 13m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/379b2fd3e03049e0bf5fe2434c38f4ab : SUCCESS in 22m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2a5a240227114b4893ef1bc1dec958c5 : SUCCESS in 27m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6ba395afb34c44239e8fd892b2e8a12f : SUCCESS in 23m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0f1ece994ed046ad8c7f925f4c036fe0 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ce6f0bc7ca1c440baad49c5514533dc2 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6a8ed16a435943839501b5bc7df73b4c : SUCCESS in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c24a91fff7c74bba83bcaf97fb6c21b0 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/662b16c23f4f4a5b855f3a7bc3f83493 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3409679e59d2441590eff1da8012a203 : SUCCESS in 40m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bf9702a1209471f9684a006aba822d6 : SUCCESS in 23m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d189867858e47f6bf351b71487d1d53 : SUCCESS in 55m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1609e27a3a740ecb200d02f9f45b590 : SUCCESS in 29m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5885ebd6df034a0fb74a8f5244f07bf7 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bbe665a37f24ac3a7d0c66b8c4190c2 : SUCCESS in 21m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/346bf54917724fe9b82af0e1df9c435a : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8d1b5575fa8cbcf746c32c1857511eb7e4657547","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:13:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b577854c0d50b9cbc3566d6461d562e3a020faa7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:14:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/516c003d2006437cb783ed1c36f1dea1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b49bdbb24aa49e286da7c7ce7d22c5a : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cb2288f8523c429cb220e48895fe9e8a : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0af5f8fac7dab160b89ac9f716c49c995e9615a5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 22:31:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31470a2b3be6e30b7b54626345c08b4c5b1ca8fc","revisions":{"31470a2b3be6e30b7b54626345c08b4c5b1ca8fc":{"kind":"REWORK","_number":1,"created":"2026-09-17 18:37:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/96/1006096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/1006096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/1006096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/1006096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/1006096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/1006096/1"}}},"commit":{"parents":[{"commit":"21709787c9faf63fca57e6e13b2cd52cb473c6da","subject":"Clarify s3api.utils sysmeta helper names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21709787c9faf63fca57e6e13b2cd52cb473c6da"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-17 18:36:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-17 18:36:41.000000000","tz":-420},"subject":"s3api: Drop unnecessary backwards compat","message":"s3api: Drop unnecessary backwards compat\n\nAlso, better-name the s3api-specific HeaderKeyDict subclass.\n\nRelated-Change: I29f988543605956614bcdb7801e2617392578af5\nChange-Id: I9b7b3c5c4cae425d4b7ad68e07226f956634a71e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31470a2b3be6e30b7b54626345c08b4c5b1ca8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31470a2b3be6e30b7b54626345c08b4c5b1ca8fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
