)]}'
{"id":"openstack%2Fswift~1006188","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I74354617c00839d20ecb4e3544c73559518d1433","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","hashtags":[],"change_id":"I74354617c00839d20ecb4e3544c73559518d1433","subject":"Merge remote-tracking branch \u0027upstream/master\u0027 into feature/threaded","status":"MERGED","created":"2026-09-18 10:31:54.000000000","updated":"2026-09-18 12:37:16.000000000","submitted":"2026-09-18 12:35:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006188-eventlet-removal","meta_rev_id":"2e2c40ffd70c387ac77d4fbea1415fcb95f5ca1b","_number":1006188,"virtual_id_number":1006188,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 12:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-09-18 10:39:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-09-18 10:39:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 11:32:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 11:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9505bfb3b602a4b376aec89f3f43f2ac94058bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-18 10:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26d06b1afcb7898f0c1863f1411db18ea3484154","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-18 10:39:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4594eb2a0f7825b35c97483b4cc3ebf683b8c87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:32:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b30647113c124f4d8a1eb8ecec7154ad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27c33b01aa9c4dc5b5d019ddfa2a159e : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e505cc7da40716f7a072a5a2d68d5dc12e24d206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:49:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce69e3fde4c4147a34b234357073904\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7104893a6110484d862df5617e905770 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a201dfc0c6824ea3ad2d64678bf11d14 : SUCCESS in 9m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fa348048b0544faa6605880c80ae856 : SUCCESS in 22m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8930d9c53d7a41abb90c6899331a4f55 : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc7eb0e535634678876df7742aa98f55 : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/763f77c0e8904d7f8a2b3dfaafeb4034 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cc71b22b3ad34094b39bcf6d8a53ebb5 : SUCCESS in 26m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1b8186d1e8ea46c6a0cf4e27b8fe6a8c : SUCCESS in 22m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ba65824611894f3291488c17e3d3141d : SUCCESS in 32m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1c598986acec44c9aa705ac526264175 : SUCCESS in 20m 24s\n- swift-tox-py3-no-eventlet https://zuul.opendev.org/t/openstack/build/954b6e9c2c6c4432b7c3ff2961d7b75a : SUCCESS in 21m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/db43905d8f1a49629ccd862ada1ce996 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/39863f582cb54332a114df064e2c21d1 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3525f9adbc334e77a4678d949cb5e64e : SUCCESS in 13m 35s\n- swift-tox-func-no-eventlet-py312 https://zuul.opendev.org/t/openstack/build/785cf7844d09433c9d4b47565a45cdff : SUCCESS in 14m 54s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba521a10c1b34d17be5f60905720eeec : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1da014ad26b422285105739b2963241 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/176f882a17514c01829ac0daafcaffcb : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/623ba0c7b5494093ab101d4ad0ec7174 : SUCCESS in 1h 11m 32s\n- swift-probetests-no-eventlet-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebda98807a644e0da65fb1a12aa7f7f1 : SUCCESS in 55m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8db414f1222a4ab38fbb1e42aefbec75 : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35aaa3e0539946958eb961e0965c67f6 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4694411e36a444faa8a5f2b81c48844e : SUCCESS in 29m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"893c0474b43ad326c766eb613dfae362abc60bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:50:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d4b8d0b5a30dda2d0fe5934f560ecf95f8f51d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:35:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f95fcd0d5a34443b97ad94fdc8d170c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a4e87ef6364b65bf0a3ba2ac7f616a : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcaa4306f5a54833af1f995508d8d482 : SUCCESS in 7m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/db7b9c1817cc4a3f94b620499204698e : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/622f493b18884e74ad84705c207f4dd0 : SUCCESS in 12m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4e6007e0d5414eb1a186cf5b58c0b3e9 : SUCCESS in 22m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d1c04306c94b4ba1a9a2afd6625be05f : SUCCESS in 23m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/98570820c17541a88026b41e58c7d5e6 : SUCCESS in 22m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a16daeca0a8e42f58582971efa2d9662 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b972634be1b14f628b8fb498bd717118 : SUCCESS in 9m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9f1cd4a65e6947e5b914996ce4c3c571 : SUCCESS in 17m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9176f9866eb54fcabcf985ea48924963 : SUCCESS in 9m 23s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/99d97ff3563b410aaab13d6a4140fecb : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c61e01ada104906bfe3d72924a4bc59 : SUCCESS in 38m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8073263552c14f1e8d667b39409d24c0 : SUCCESS in 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3991108efd0431ea51df5716420fb2d : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0de28a536d34a439ce3de3200960aee : SUCCESS in 23m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5cd0f0b96a4dad482f889d73b63cc04d6399b37c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:35:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e2c40ffd70c387ac77d4fbea1415fcb95f5ca1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:37:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e1d6c2c8680477ba2c20e9123406f15\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e37dace9fc4ad4934f48d31b53e541 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/44b6282aeec7493789da5e9138f69637 : SUCCESS in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37326b942b20703591ea08fd214c1713568cc856","revisions":{"37326b942b20703591ea08fd214c1713568cc856":{"kind":"REWORK","_number":1,"created":"2026-09-18 10:31:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/88/1006188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/1006188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/1006188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/1006188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/1006188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/1006188/1"}}},"commit":{"parents":[{"commit":"322d53a710425952111a07e52f3bb82a45534031","subject":"utils: count pending pile items under the lock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/322d53a710425952111a07e52f3bb82a45534031"}]},{"commit":"00bf38b9cf32746f507a3157634103de40a966f4","subject":"Merge \"add PrometheusClient for file-based metrics export\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00bf38b9cf32746f507a3157634103de40a966f4"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-09-18 10:24:25.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-09-18 10:25:53.000000000","tz":120},"subject":"Merge remote-tracking branch \u0027upstream/master\u0027 into feature/threaded","message":"Merge remote-tracking branch \u0027upstream/master\u0027 into feature/threaded\n\nChange-Id: I74354617c00839d20ecb4e3544c73559518d1433\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37326b942b20703591ea08fd214c1713568cc856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37326b942b20703591ea08fd214c1713568cc856"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
