)]}'
{"id":"openstack%2Fswift~1006324","triplet_id":"openstack%2Fswift~stable%2F2026.2~I6f4c412a7ec106e9d7dfc97a305776f2eb5c92c7","project":"openstack/swift","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"I6f4c412a7ec106e9d7dfc97a305776f2eb5c92c7","subject":"Update .gitreview for stable/2026.2","status":"MERGED","created":"2026-09-18 19:13:43.000000000","updated":"2026-09-20 05:44:00.000000000","submitted":"2026-09-20 05:42:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006324-create-2026.2","meta_rev_id":"2ee95c4976d56de764e2b30e50b89c8ec379e123","_number":1006324,"virtual_id_number":1006324,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-20 05:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-18 23:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-18 23:21:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 19:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 23:21:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"7041c160da516b74188d5c1638fe9dee483379f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-18 19:13:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8766f5b99c8b6ab17a9b90dd5ffbae56eb61266a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:47:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9f0f536e2041c98a756f2307ce30b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1196b79286f4b8589495f59a75d0b1c : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47caa17aaef3470ba3211c4bbf09446b : SUCCESS in 9m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f684b8cce33e423884e93cf305e893a7 : SUCCESS in 9m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5838f354691548f5a932641b04781da9 : SUCCESS in 13m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d4e91f0459a84d2a948131a9f826650b : SUCCESS in 9m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/73f9eb5fc56d4b26a7637e669ee8e0aa : SUCCESS in 25m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/897295b29ec84fe3b0f66f41f9088561 : SUCCESS in 20m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aed5b348cb6488a921762d00162cb57 : SUCCESS in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f7956dab0e45de86445cb28fbda929 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3878cba91c86695f9f8c5aa1fe3b04043a4d19d9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 23:21:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"344d4370ed3c8c0ed9a346578f338b8b32f641d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:22:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b553416056c2c59bd5afc7b718051b435af942c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:33:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fc31533fc184262814f26a07d073948\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3175bca68d2e42ea9085447893114c1e : RETRY_LIMIT\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/846a3eb3f9134fb9bce2ba05862fbe8c : RETRY_LIMIT\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f8d34f04a2ac450c80cab9cbf9818502 : RETRY_LIMIT (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/28539a6947684c57bbb90f61bc60c781 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a8749ebbe57b4868b971b42c22fdf3a2 : SUCCESS in 11m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2231160bcbf3499ca66400a975822560 : RETRY_LIMIT\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac8ea3bc11bf47f88a07cd2b25a00396 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":1},{"id":"7aecd39e444fd688e97adcb4c5d4dcac93ccd062","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-20 04:33:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b1f52dab5daba3ab83f8b9e0694c1f0a909f489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecbc056465dc4a6f854e0e90eb4fe454\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23c5a3e53d5491696fdee20a80a87e8 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92e36f8412534459b10a2612367620b4 : SUCCESS in 11m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49af83fa447d448ba62474dc43ae6056 : SUCCESS in 7m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8437fa31b87d401c8c378d0d88bad944 : SUCCESS in 23m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c143f09d356744a39353441ac3d9ef27 : SUCCESS in 9m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4df3fa8723f544d184a1111ab5147145 : SUCCESS in 23m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d931da0ebd2944e1b713a8c005584596 : SUCCESS in 16m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e19f7fb5c9f34ff8a3b85dbf4d7ced53 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8039c8076f8f45d2990d8fe131b6ede2 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d03d5b97e308b960193aadedfcd13e8da666bf9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:02:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17e64135025b72803daa00ca6adc4c77f83d726a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:42:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8131ca45641243e3bc024a9d008e65ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d513abf4eeec42c98fbcdd1736a83741 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86742574af5b4b6b9b31510968d890a8 : SUCCESS in 8m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/87993f8d1f8042ce9b8168ec861f81ca : SUCCESS in 6m 37s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e95bcd53cd794f5b88a3d1380a5c1523 : SUCCESS in 5m 06s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/53ed13dccb5145c58b9051ea4954582c : SUCCESS in 8m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/239927c7af344bbbae9fc1363c74614e : SUCCESS in 28m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b5f170eb9348b1ba3d92e80028460d : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"62c98432fc9ed577f33f25400a5671a9d72fef3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:42:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ee95c4976d56de764e2b30e50b89c8ec379e123","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:44:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6e6ff744caa43a3a71fce4d4909facc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a3d2c21ede4c8fad66dae2ad971f09 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a3c3b59084b64b96a151980f378a8214 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7bc1cbf56b250e04d4f7f654aee124ec526304e","revisions":{"a7bc1cbf56b250e04d4f7f654aee124ec526304e":{"kind":"REWORK","_number":1,"created":"2026-09-18 19:13:43.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/24/1006324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1006324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1006324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1006324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1006324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1006324/1"}}},"commit":{"parents":[{"commit":"3064444bf433dfb7ade3d91130b58f7bceae9add","subject":"CHANGELOG for 2.38.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3064444bf433dfb7ade3d91130b58f7bceae9add"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-18 19:13:41.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-18 19:13:41.000000000","tz":0},"subject":"Update .gitreview for stable/2026.2","message":"Update .gitreview for stable/2026.2\n\nChange-Id: I6f4c412a7ec106e9d7dfc97a305776f2eb5c92c7\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7bc1cbf56b250e04d4f7f654aee124ec526304e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7bc1cbf56b250e04d4f7f654aee124ec526304e"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
