)]}'
{"id":"openstack%2Fswift~1006325","triplet_id":"openstack%2Fswift~stable%2F2026.2~I1829e2790130365375271a1beb91d2f846faa302","project":"openstack/swift","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"I1829e2790130365375271a1beb91d2f846faa302","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","status":"MERGED","created":"2026-09-18 19:13:44.000000000","updated":"2026-09-20 06:25:47.000000000","submitted":"2026-09-20 06:24:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006325-create-2026.2","meta_rev_id":"63ae641c78b738ea92b886ffb53a161d940dee4b","_number":1006325,"virtual_id_number":1006325,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-20 06:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-18 23:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-18 23:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 20:05:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 20:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 23:22:42.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"963b81a27a09696d8860737c5b95476a15ff56cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-18 19:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08b0414c9a1ebf0c431e87e54fc1a5f232b27d05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 20:05:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed8d99cdedcb4101accf606f4320e888\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eab458e3f19f4643876884b585874942 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7150fe9e7cdc56386bad634a82a5ab7336ac0b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 20:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d6ccd4775546afac6a2458b6ae53d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c021faad3e48e6a7c408f8ccc9c355 : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef4b06118fdf4ba8b2c01c25f3179ac1 : SUCCESS in 8m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0dcdd1e9c5af46c5b81ed3e8f622be22 : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad9a76aaa2a34ea38dc518ed7358b2d9 : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8b9a429817c450aa5c76bf86a00e3d4 : SUCCESS in 11m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01bf2cf9b6c04f4488830c2a9e44ad48 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d0576e8d37e2468cbd212dda3ab572cf : SUCCESS in 27m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dcf6db3d685e4867b63e50c8d8aada10 : SUCCESS in 20m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ce3d9aff42764ae885fae2ab00cd6391 : SUCCESS in 23m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dc7b5e99ec9e4f8881a8be1b876e2256 : SUCCESS in 21m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e3ef48741d904838a5ac9b3a86e41dcb : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dff4aeae5529453fbc0929871b6b1f2b : SUCCESS in 14m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/359b41f30e22421da7c7d54a236c8f78 : SUCCESS in 17m 21s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ba2ec9eff9f404695bcd1ee189b1ecd : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39c8e314eeb3497bb0ffe0b82750947e : SUCCESS in 4m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/74d78fbd173c4b2c8e74c5d860c3c648 : FAILURE in 18m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/40fb99d0cf1b41c2ac2382e7815dacd9 : SUCCESS in 6m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdf7384d1acb4ae69cf8356ed4bb1bd8 : SUCCESS in 1h 04m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8c2fb21a518420cb05347bd681c0c86 : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97fc0a4c314244b89b634c169ffa9a33 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/07fb81312ed44dba92a9c9defb8b7d74 : SUCCESS in 19m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"7903aa1222a5aa91fdaa27da93de7f102a75b321","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 23:22:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9a44ce808bbbf8547a9c9b7a37bc9e398d1e2db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:33:53.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/swift/+/1006324 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"0a04356f6a8d94b2609d7ec9ad6428ca0ced778c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-20 04:33:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba76029fcd4379382b7de9b641522647e1b65747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:11:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dd3c87a6c4149b4b7947af80f10974e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57afb2f401684862bb20cdc7789ad94d : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f760ba7da17470c80a8d4624cd7bb63 : SUCCESS in 10m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee50f62a4fc44b13a2273898f5f573e7 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1debacefbce248e78079d085b181be88 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ea5a602ff684e0ba2047b2f3ee4a355 : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a1e4454d73949edb6188a0e1b49de2b : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63feb5ad88ae4373afd1eb27ffe285f7 : SUCCESS in 14m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c67624f1a22045a88e3e30341be7f274 : SUCCESS in 21m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9fdee033412b4c1899447004c88648dc : SUCCESS in 15m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d6665e07fa4d41e6ab8628fcad16155f : SUCCESS in 15m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1864f6e5dde44251a40a922a6ed651ef : SUCCESS in 9m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/96de4dc0b602407c8b86bb2f1da7d211 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c9c37be5899445b4ba39facf479fea75 : SUCCESS in 15m 42s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/233572621cc344f2845a400266657090 : SUCCESS in 9m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ddfa3f306114004937ce30e96772665 : SUCCESS in 6m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b7d98e70880a45cbbd47809100a5b566 : FAILURE in 15m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/02814311969747cabf641ef294126348 : SUCCESS in 8m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f53ea6e48304e34930e53a36d68ae7d : SUCCESS in 36m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e4ef00d277e478fa026f34055e66964 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd0cb6c8791840fc80c328af0a24a87e : SUCCESS in 3m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5d162b3aa094864ac473abe111c7c64 : SUCCESS in 11m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"07f3ae5af1bfe8105b69dcc7972df84bc3070ac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:11:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5675c40a9563205e26b286aec434f9a368bd9e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:46:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192ac7d0ec6b4fd799adcb0715370e7f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b93c904540d74fcdaff6cfa065600188 : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8930a7cb46b18487c0d9781f70f8fc01e6383dd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 06:24:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89a09cd523d84879ae3f506d9da9a149\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ee9662803d4773af258cf618d823c7 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0de40a30f5a64f0ca9dd8af59c825576 : SUCCESS in 10m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8b5b86d0d51741eb906028e67fddcbc1 : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e2c47e6f79b41a489f8babd1aae777d : SUCCESS in 17m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/30cc4704720e45658303032055b98f30 : SUCCESS in 21m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c1dd545db7654f05bcf51c105de4949c : SUCCESS in 20m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/82a05848f2f1424ba124a113319669bc : SUCCESS in 20m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e3921ce0893146299312dbf2966f8700 : SUCCESS in 8m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22a2bd0eafbc4fad8378523f3d9a1d1f : SUCCESS in 8m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5228b8198e60487994b22403fbf499bc : SUCCESS in 9m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/757f68e26d47478b9807cb1467eaa4d8 : SUCCESS in 4m 08s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/73f2b1fdeae54820982c755b00362d8e : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d5e989b6f0d47bb8403654d1308a049 : SUCCESS in 1h 05m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fb0673368f044c0b735196804bd0365 : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a8607c1df04903aad228a332ff7f6c : SUCCESS in 3m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce9bd6e682b44de48612ae2ccb89dcab : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fd97afca58cd0df84ccc0c1b1e2b88a9749d248f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 06:24:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"63ae641c78b738ea92b886ffb53a161d940dee4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 06:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43b1eb2e0256429982eb10ed099db7a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2a70db2d414787b7342616095ec78a : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8399aab157f4b88a7fcd5194aa2bd80 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/211a48221a804a4594d19c0185772706 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb4d8b5b21766886d1f8e4259126c7879277f83a","revisions":{"bb4d8b5b21766886d1f8e4259126c7879277f83a":{"kind":"REWORK","_number":1,"created":"2026-09-18 19:13:44.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/25/1006325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/1006325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/1006325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/1006325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/1006325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/1006325/1"}}},"commit":{"parents":[{"commit":"a7bc1cbf56b250e04d4f7f654aee124ec526304e","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7bc1cbf56b250e04d4f7f654aee124ec526304e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-18 19:13:43.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-18 19:13:43.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I1829e2790130365375271a1beb91d2f846faa302\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb4d8b5b21766886d1f8e4259126c7879277f83a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb4d8b5b21766886d1f8e4259126c7879277f83a"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
