)]}'
{"id":"openstack%2Fswift~1006659","triplet_id":"openstack%2Fswift~master~I2076c7bf18f1150d54c77e61fde62e0497aaeaa4","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I2076c7bf18f1150d54c77e61fde62e0497aaeaa4","subject":"sq? Check etag as soon as we can","status":"NEW","created":"2026-09-21 21:45:28.000000000","updated":"2026-09-21 23:11:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cdf58b6b674cc5a87bab972480494a7692a26fcf","_number":1006659,"virtual_id_number":1006659,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-21 23:11:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 22:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-21 23:11:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5911f1f58f56bb001444900aa6b82627242c7a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-21 21:45:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4255bfe7530c1e8b77bf69536e83b3c466ac636c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:54:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69941b392764395a7692d52f6122079\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3ba3a0b8dc374e188310a55f2adeeb51 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cdf58b6b674cc5a87bab972480494a7692a26fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 23:11:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a990ccbd1d244120b26cb4ad1a7c89c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb51f0cfba3941e29658659b1a31d6c5 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c61fbfe6bc44bee8f40b5f9e132b330 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/1665073aa9b74176a23d125137997348 : SUCCESS in 1h 06m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16e55d90027c4efc8b3063c235a33d25 : SUCCESS in 47m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a0e6109303f495b9602a5db1f36e708 : SUCCESS in 43m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9b13cd76c2f46289802f5ef1d24f480 : SUCCESS in 55m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27ce75ee46824d58bdd2e66bedf96606 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8f46b00e51648a48152007078e245a4 : SUCCESS in 22m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8cb979fbff8b4946945cd18d40a2aa84 : SUCCESS in 18m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6142fcd257c34880ab270fd8fe7285e8 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d1287a1e41354017a08e76730a1dea95 : SUCCESS in 26m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7502b0e882124155a998d2a8ed5ad4b8 : SUCCESS in 21m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5fc842b1116a4d099653dfc562290e6a : SUCCESS in 33m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4b382799155e40f7aaa8b96743a2cf1f : SUCCESS in 21m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/537178cd37674c7580dbbfc1ae89c93d : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3c50d4664b4f4bf2a79b22c607b590ec : SUCCESS in 17m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9fe6616e5a0b4360a2e3a04ffda7eaa9 : SUCCESS in 12m 56s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ef982848ce647d4b4f2056a6c33448a : SUCCESS in 18m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e92528c338e44d1afae949622a6606c : SUCCESS in 6m 07s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3bb7965849654381beed07dedec3ced2 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f15f9cdd7db349c98a8aeddc3a11dbcb : SUCCESS in 45m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c5e112eb72049c5b6daddb31d7bafc3 : SUCCESS in 54m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2adb46d8e9954ed69c1d36581727fe0e : SUCCESS in 54m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1d123f0568248d2a912492926c54c9a : SUCCESS in 30m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812c7f2d6d50451294c1ef5e95fc4301 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/730586a478be40cc9a521a728907eda1 : SUCCESS in 16m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20a07f0cd276425c8e0479f223548dfc : SUCCESS in 1h 00m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f3ad9da4145f84bb2b24fc84bd1082395d12edf","revisions":{"1f3ad9da4145f84bb2b24fc84bd1082395d12edf":{"kind":"REWORK","_number":1,"created":"2026-09-21 21:45:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/1006659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/1006659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/1006659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/1006659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/1006659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/1006659/1"}}},"commit":{"parents":[{"commit":"b5c27b87e3797db730e0cda57a253517af76bff1","subject":"Verify donor fragment ETags when rebuilding an EC fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5c27b87e3797db730e0cda57a253517af76bff1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-21 21:42:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-21 21:42:05.000000000","tz":-420},"subject":"sq? Check etag as soon as we can","message":"sq? Check etag as soon as we can\n\nChange-Id: I2076c7bf18f1150d54c77e61fde62e0497aaeaa4\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f3ad9da4145f84bb2b24fc84bd1082395d12edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f3ad9da4145f84bb2b24fc84bd1082395d12edf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
