)]}'
{"id":"openstack%2Fswift~1006998","triplet_id":"openstack%2Fswift~master~Ie01cac8dfc6d1faea1a76148a04eefa1a4e9ccf2","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ie01cac8dfc6d1faea1a76148a04eefa1a4e9ccf2","subject":"CHANGELOG: Include calendar version as well as code name","status":"MERGED","created":"2026-09-23 17:24:12.000000000","updated":"2026-09-23 23:20:54.000000000","submitted":"2026-09-23 23:18:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006998","meta_rev_id":"319cbce8de137a4f90739a90409260986835c313","_number":1006998,"virtual_id_number":1006998,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 23:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-23 22:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-23 22:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 18:03:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea6112fec76b7c2625d9851f594c8e866d359b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 17:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ffdc9808f29e889f9f86da9d52eb3fc5726cd0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 18:03:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d881ca214bf4ec09b3e00e112f5df04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4efc693023d4189840d8cebb6ace3d1 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8e40fa6af57410885d0515127b87cab : SUCCESS in 5m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/063e18881ee7441a9cafa9b78dc951ef : SUCCESS in 4m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84835241301c4771898da5bff9f386f7 : SUCCESS in 21m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c024096a8b274f20a561de38acfeea8a : SUCCESS in 9m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f374b3b36035456480418a4fcc1d5b6b : SUCCESS in 12m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3f3305b05f1143c2ac1b2ddde9b89455 : SUCCESS in 20m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e17ea93dfbac47df898d8bd18fdb2e70 : SUCCESS in 33m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed9838e0bf84dde8533a18c0177369f : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f81907c7ad892bc9e2f826bc7d9d6cf5df65c257","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 18:31:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb621e0cc57d40b8b95cffcfbf4f4d69\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1cea56eeabb4121b77365aa4f07e7a3 : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6a9eb04dcbd0c02c7fc3dedd3c3ea40c6c94a8dd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 22:00:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"159afb3df5702c0c4b458eef49225426e1daae2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 22:01:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cd734b67ce6e2d1419422e6c1e01e7191de7d01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:18:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5331a2d065b0466eb7942bd3bb20e7b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b41e45d26134dcebdcaac777ca17f31 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62c58d9b325c4eafb3fe776810a8ddc6 : SUCCESS in 8m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fcf8b72a242c4f08af39834b601d0de0 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/22f9c1ccc36045c49f5899bab00af44e : SUCCESS in 20m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dbf3d253e0124f01affb8f1a60074840 : SUCCESS in 17m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/09d69a2da67c452a9c725f99c87d519b : SUCCESS in 24m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9d1d7a68c5f74750a99d3aea691a48ef : SUCCESS in 16m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/036c3b818d16449a80ff4ba24035277b : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1930f380451a409ab2ddd3718ad920c8 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/83e48f127cb248209de5740f995dd8de : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1e4792a48614ebcb7a79d3f8c0e5244 : SUCCESS in 5m 16s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/686dc6371b094487af16f9b48866603b : SUCCESS in 8m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/439672fe46af4a08a09568940c00fe22 : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd4742a22d7b4e0dae82b9c098610485 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c1e34e3e99c65f770c173f825de6da4d08e3a57d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:18:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"319cbce8de137a4f90739a90409260986835c313","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:20:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b6889288ca4446b36a55e57d77382b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3818e92c0b3140a8adb636c56ab16e90 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/aaff10479df34bc484ae3438aaaa1a36 : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"142316f9606e9b4cac9e6ebe8845322f5e98ff83","revisions":{"142316f9606e9b4cac9e6ebe8845322f5e98ff83":{"kind":"REWORK","_number":1,"created":"2026-09-23 17:24:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/1006998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/1006998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/1006998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/1006998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/1006998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/1006998/1"}}},"commit":{"parents":[{"commit":"6b83d7f3fcf95539b17dc1e3f5f1074c29afc487","subject":"Merge \"tests: Use POLICIES.default in quarantine hashdir tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b83d7f3fcf95539b17dc1e3f5f1074c29afc487"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-23 17:23:35.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-23 17:23:39.000000000","tz":-420},"subject":"CHANGELOG: Include calendar version as well as code name","message":"CHANGELOG: Include calendar version as well as code name\n\nChange-Id: Ie01cac8dfc6d1faea1a76148a04eefa1a4e9ccf2\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/142316f9606e9b4cac9e6ebe8845322f5e98ff83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/142316f9606e9b4cac9e6ebe8845322f5e98ff83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
