)]}'
{"id":"openstack%2Fswift~1007046","triplet_id":"openstack%2Fswift~master~Ibe98ae41cba42997bbba00c74bb24bee22f32c05","project":"openstack/swift","branch":"master","hashtags":["CVE-2026-97149","OSSA-2026-041","ossa-2026-041"],"change_id":"Ibe98ae41cba42997bbba00c74bb24bee22f32c05","subject":"tempurl: Disallow X-Copy-From on tempurl requests","status":"MERGED","created":"2026-09-23 22:03:29.000000000","updated":"2026-09-24 06:43:21.000000000","submitted":"2026-09-24 00:32:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007046","meta_rev_id":"55749605aa82a94c8231d47b0dedadded24467af","_number":1007046,"virtual_id_number":1007046,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-24 00:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-23 22:12:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-23 22:12:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 22:03:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2026-09-23 23:07:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-23 23:19:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b6bebb62a255b3aac78883b3ac806abf85486ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 22:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"651d25b232bdcd63f0c4cad8a3543c876c5bd3f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 22:12:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e925b21ab2ca2fe70fb8e78568460fac0107f67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:07:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d38b74bd06a541788f9839a143cd6a5b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b77a37f5283f403d9c4c266e43b60473 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0c9a9652f826879875eb9b071b2671f5c26f642f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-23 23:10:34.000000000","message":"Hashtag added: OSSA-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"2f7ba971570c0a80237a2cadfcc36b9f149415c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:19:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bde5b20a6994629bf865de03440de99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a192b768ee845b9b80e2ab6c543dfda : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6ae8a3ac06548f1925acbfae20c0f4d : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/9c0701eda2c8468ab4ecda54270aa2ff : SUCCESS in 41m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3119ac186732415cae08e9c12bb83d3c : SUCCESS in 1h 10m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6c4b259261b44a1b81798f28ece0944 : SUCCESS in 51m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fa384caae1a4ca1b4a490afa1312c42 : SUCCESS in 55m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ada01a11f0c14288a42dd8899f743cfe : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cea2975679d484687f077c245c9eae1 : SUCCESS in 10m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc38a30347114a59b2b8bc857233611a : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a30bdf67a8a44b79d1d30d60a7596ca : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4d0de3a6c40d4ceabfaa95d09476970d : SUCCESS in 27m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/68014a4651664b45a7f85580e1630b1d : SUCCESS in 22m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/34d3e7d93d4142bbb7f6c52993725698 : SUCCESS in 31m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/851408e1d593424c9e97243c79e79106 : SUCCESS in 21m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/93d582e7c7c248a2a1fddce13faaa1ec : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/25e5972721254034bf065263bbadd279 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29455164dd7d454285676dce088bc594 : SUCCESS in 15m 36s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d071e93e115b4b6db577b0d66679728b : SUCCESS in 17m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e21cc4e8f31f49ac9ea916adfb61ff82 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/107866ec487c4af886d5758a198004cf : SUCCESS in 9m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/456e963081e14e9086ddfcf31cc74442 : SUCCESS in 37m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64fd5814974d4ebd90f69efedc5c67b6 : SUCCESS in 49m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b6899fa2a2b4bf2a8c51e3f3a9bc303 : SUCCESS in 52m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2eef932221640428f5cd2f646c2cd3e : SUCCESS in 27m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6162a400c694bbdb6f6d0a2682bedc2 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de35db00cbf14c8f96e5d2e2cc77f9c4 : SUCCESS in 19m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b8c6a5d8a934cacaede61caddf270b9 : SUCCESS in 57m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1f92131494ace1b3bc28b7754131eba46842bbff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:20:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86f7d0488a66ab8b196b7f1d2d3db751447273f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 00:32:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2e0c9f755b5419a942253aacb08983b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017d1dd11e724f66ad6cc5bf3ef59aa6 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c096806122d45e0a919d4d3b614c33d : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/a560bdccc8304ddfbf1be5a7e9f53ce6 : SUCCESS in 52m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1328d1654884bbc8e8e01a32de1c7bd : SUCCESS in 1h 07m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5b0a60688484dc0a8ad5dd986656e91 : SUCCESS in 30m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a4b41c71f394e0894e063b11bec458f : SUCCESS in 1h 03m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cb7352a41bc64786bcdf6e3ebef596f4 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c75e9cbbc7b54990a27601164baa5052 : SUCCESS in 12m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2e064f10997f4e94837107359ff0488c : SUCCESS in 19m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5ea4000d394f46909ef99e818fa94efd : SUCCESS in 21m 35s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f3a77fd73c9845baaa893afc701c17a5 : SUCCESS in 20m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9369edc649c44bf985e65ed7096251fc : SUCCESS in 8m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bdd17c2144764049bd2e69b6e55e0d58 : SUCCESS in 12m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/74c9ab5445ec4ad6ae800131f9c5bc95 : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/76aae2384b2c4ddd95d66aad67c8acb6 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e04d01ffcac42dea38c60051fbf7803 : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/54355d72478d446f99a08284368c99d8 : SUCCESS in 36m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9f2379f66f5439faa2a1bc0d94e5449 : SUCCESS in 34m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22070753f9eb4fdb8bce7a8945d5465e : SUCCESS in 52m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36e8b1a1a8b4dd6b99bb73c13fe0b40 : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882c372c81a842eeb05c4da294048c11 : SUCCESS in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c96c0095946464ab75e94fd7a3ad224 : SUCCESS in 18m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b29e88be30ea4ed6bec9b0eddebf2345 : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d8655217267fc191e98350e8b38a6f3842656fbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 00:32:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eccf3246c518015478e4f628d7277584b4065a4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 00:33:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f2d13320ef64caeba900b502ad45e9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4543bb6845747fb8ea6f1f6bb3e9362 : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c4775547af8c447e92355625c9040a0e : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55749605aa82a94c8231d47b0dedadded24467af","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 06:43:21.000000000","message":"Hashtags added: CVE-2026-97149, ossa-2026-041","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32943ef7b9691f0447c3cfa93bb2f492958283a3","revisions":{"32943ef7b9691f0447c3cfa93bb2f492958283a3":{"kind":"REWORK","_number":1,"created":"2026-09-23 22:03:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/1007046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/1007046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1007046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1007046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1007046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/1007046/1"}}},"commit":{"parents":[{"commit":"6b83d7f3fcf95539b17dc1e3f5f1074c29afc487","subject":"Merge \"tests: Use POLICIES.default in quarantine hashdir tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b83d7f3fcf95539b17dc1e3f5f1074c29afc487"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-09-10 13:41:44.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-23 22:03:13.000000000","tz":-420},"subject":"tempurl: Disallow X-Copy-From on tempurl requests","message":"tempurl: Disallow X-Copy-From on tempurl requests\n\nDisallow clients using the X-Copy-From header to make a copy of any\nobject within the account using a PUT tempurl by adding the x-copy-from\nheader to DISALLOWED_INCOMING_HEADERS. Requests that use the header now\nget a 400 response, like the requests that use X-Object-Manifest or\nX-Symlink-Target.\n\nX-Copy-From-Account is not in the list, similar to\nX-Symlink-Target-Account. These headers have no effect without\nX-Copy-From or X-Symlink-Target.\n\nCloses-Bug: #2166876\nChange-Id: Ibe98ae41cba42997bbba00c74bb24bee22f32c05\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32943ef7b9691f0447c3cfa93bb2f492958283a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32943ef7b9691f0447c3cfa93bb2f492958283a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
