)]}'
{"id":"openstack%2Fswift~1007047","triplet_id":"openstack%2Fswift~stable%2F2026.2~Ibe98ae41cba42997bbba00c74bb24bee22f32c05","project":"openstack/swift","branch":"stable/2026.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-25 14:41:47.000000000","reason":"Change was submitted"}},"hashtags":["CVE-2026-97149","OSSA-2026-041","ossa-2026-041"],"change_id":"Ibe98ae41cba42997bbba00c74bb24bee22f32c05","subject":"tempurl: Disallow X-Copy-From on tempurl requests","status":"MERGED","created":"2026-09-23 22:04:06.000000000","updated":"2026-09-25 14:43:57.000000000","submitted":"2026-09-25 14:41:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007047","meta_rev_id":"68d95c470a4d688a55fad508b17648928dfeee92","_number":1007047,"virtual_id_number":1007047,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 14:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-09-24 18:56:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-09-25 13:41:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-09-25 07:09:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-25 13:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-25 13:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 22:04:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2026-09-23 23:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-23 23:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-24 18:56:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-09-25 07:09:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-09-25 13:28:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-09-25 13:41:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"febadb7d6cf51e36be21900f1c04a555c9d65bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 22:04:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12f75f003078c7ccf0c6065ee6d4b5a063c52435","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:05:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c35a6db2c414a0fbd117bee73345043\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b8aaa425a8f477ea4b2dc40a7df338d : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":1},{"id":"020ff862cd2c1f7b07b40792dbd94bce22aca8f8","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-23 23:10:34.000000000","message":"Hashtag added: OSSA-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"194d94a463caa836ed75b08e9d400fa9d6198d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:20:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0877549f3744158a6461a6409ddc103\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb904fec37a434f897638a872c83d90 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19773803606d4e1ea353c739bf9c93a6 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/423347e5d3004f64b813ea4132ebb45d : SUCCESS in 1h 01m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b902ff6576d0435ca4945e28e84027cf : SUCCESS in 45m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee914d7a143e4f02a832812853add374 : SUCCESS in 56m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f5a334fb6974d7f8e98566dbdc3afe0 : SUCCESS in 33m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4377990b47c40f4bce7eb9fe2b53aa5 : SUCCESS in 10m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc7cd18b0112422cb97ab7029627dc15 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07efb5e21c0b4747a42d6d41078079f2 : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/695c8acd4c6b4461968506cb57fd3985 : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/562f8dd26ab447e585c43e743f3bbbd0 : SUCCESS in 25m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e681a5d3acae4c18a50d9c666ecb9835 : SUCCESS in 17m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a58d55cb86564274b7ea9b242be1cfa3 : SUCCESS in 23m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8ea44bd0e0ad4157861569abdd6588cf : SUCCESS in 16m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c5169d0e5c36461ba4c61fa8ba25543d : SUCCESS in 8m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/36667a13e030425c89f30f32940223ac : SUCCESS in 17m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6ca188a677b4ffb92e59f810c6264f7 : SUCCESS in 14m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffde16413acc4fc0b9bf5215f11d1248 : SUCCESS in 17m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0404f6050d674cff9e7224b5a23f90d3 : SUCCESS in 4m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ae5392a2e8943d4be34f67ce0e74a3f : FAILURE in 16m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/527b3049659945b18eea6c2522004567 : SUCCESS in 5m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/81a9028b1be847739fef55a040c066bc : SUCCESS in 51m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e5a6d03e4804323846e90e54f9b9bde : SUCCESS in 33m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/125f31f8fd1d48439d1022f067203ef5 : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a0cbec82ed34976a180d8ba82432ae1 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/393d29561ede4796ac901fb139e0e2e9 : SUCCESS in 2m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8524e7b99cf64093bd45d44c02683945 : SUCCESS in 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e9f33271686453792dc76656efa8424 : SUCCESS in 24m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"d8f39650207d6c309cb987d3de6ec88e1c850e6a","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 06:43:21.000000000","message":"Hashtags added: CVE-2026-97149, ossa-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"1521b1b7e7947facaadad11a9ff1659ca8415bd9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 18:56:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee246c47f5249491585fe91782b23e46216f9138","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-25 07:09:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b9c398633957683e3819e8e8f07c7b842b03297","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-09-25 13:28:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c964238d52445dcd6b63e8663c85ba8f1881f34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 13:29:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c23eb7529c1d3f7e1e8c7ac98357ca262748bc8a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-25 13:41:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6301266957a14580923e5b2986433be72643325b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 14:41:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc07ca448e64f619e1e0f562383e88e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f8335d506c4945bd7c0eb7bd193131 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3d698ce733442188c04f43afbb7acea : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/7c5f2fe9ecb94636a5ce3d79c5ec9fcd : SUCCESS in 52m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8169f62cdbab44c29862fc24b8dd0e78 : SUCCESS in 1h 06m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca07a0a0c6f54406877f84ce5ec68abd : SUCCESS in 28m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c400d2a3bd5f4e4e94535254e1da3434 : SUCCESS in 28m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/55c0b704c57a4e12abbf6201dba9d5f8 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/813a1906de7843debac05a166827ebbd : SUCCESS in 21m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/797406c9878a414b9ba0cb88acc35fce : SUCCESS in 9m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/00e7054074ee41c3bd5fe9687a1b3387 : SUCCESS in 31m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1c074ce554b04714aca0e6b2eebf63f3 : SUCCESS in 18m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c82db913a7974232ac345eacc220edda : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/426e58746835450a9b8352cb6017b9f1 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e2176ecc28f6443bbba3c3febe805525 : SUCCESS in 8m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e057a33685042a984dab2957d481f36 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d308a352a5f42d1872053403a7feca7 : SUCCESS in 8m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/24a71a1c15c44681b991d7f0041cf1b5 : SUCCESS in 53m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5692ebd3f424f24a9a2fb0f0239db2d : SUCCESS in 50m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6645ee659b7e46a28aafdc624c5592e1 : SUCCESS in 40m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ead97b941ae4ac6b2b75d8a2712bbe8 : SUCCESS in 24m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03d9b871d13e4f81842b63fb5f9fa08a : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be4b728a63f54cd1914529271574f71d : SUCCESS in 32m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4882cd24a9c645759ab09b4c0a56b524 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"70702f3d9b7e139a76962be6f00ca002c8dd2e93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 14:41:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"68d95c470a4d688a55fad508b17648928dfeee92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 14:43:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e161126deb943939a88e71cf656e89b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e68f7ae33d24c89b2179d428d1bf5a3 : SUCCESS in 1m 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8f5306afa80341bdbd4fd334fedf037d : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"844252daed8d24ed6df016d1512e6a385ebb60aa","revisions":{"844252daed8d24ed6df016d1512e6a385ebb60aa":{"kind":"REWORK","_number":1,"created":"2026-09-23 22:04:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/1007047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/1007047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/1007047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/1007047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/1007047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/1007047/1"}}},"commit":{"parents":[{"commit":"bb4d8b5b21766886d1f8e4259126c7879277f83a","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb4d8b5b21766886d1f8e4259126c7879277f83a"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-09-10 13:41:44.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-23 22:03:59.000000000","tz":-420},"subject":"tempurl: Disallow X-Copy-From on tempurl requests","message":"tempurl: Disallow X-Copy-From on tempurl requests\n\nDisallow clients using the X-Copy-From header to make a copy of any\nobject within the account using a PUT tempurl by adding the x-copy-from\nheader to DISALLOWED_INCOMING_HEADERS. Requests that use the header now\nget a 400 response, like the requests that use X-Object-Manifest or\nX-Symlink-Target.\n\nX-Copy-From-Account is not in the list, similar to\nX-Symlink-Target-Account. These headers have no effect without\nX-Copy-From or X-Symlink-Target.\n\nCloses-Bug: #2166876\nChange-Id: Ibe98ae41cba42997bbba00c74bb24bee22f32c05\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n(cherry picked from commit 32943ef7b9691f0447c3cfa93bb2f492958283a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/844252daed8d24ed6df016d1512e6a385ebb60aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/844252daed8d24ed6df016d1512e6a385ebb60aa"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
