)]}'
{"id":"openstack%2Fswift~1007048","triplet_id":"openstack%2Fswift~stable%2F2026.1~Ibe98ae41cba42997bbba00c74bb24bee22f32c05","project":"openstack/swift","branch":"stable/2026.1","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-25 07:09:22.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"removed_from_attention_set":{},"hashtags":["CVE-2026-97149","OSSA-2026-041","ossa-2026-041"],"change_id":"Ibe98ae41cba42997bbba00c74bb24bee22f32c05","subject":"tempurl: Disallow X-Copy-From on tempurl requests","status":"NEW","created":"2026-09-23 22:04:36.000000000","updated":"2026-09-25 07:09:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d1650ad204de259c0dc41cc27911e01b352326cc","_number":1007048,"virtual_id_number":1007048,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-23 23:18:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-09-25 07:09:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 22:04:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2026-09-23 22:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-23 23:18:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 07:09:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"4b143fab0a8fac64c02dd7d534e72599c9d1708f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 22:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88bce995ddfa86642e8149d71ff0633bb2594494","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 22:46:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f6fd33767534ff984e22f35f8fc011b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea73e8085b854b9dbb42430ea6817d42 : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":1},{"id":"702ef34f0a9d3ebe61b9704b6e0da5da73a39a9f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-23 23:10:34.000000000","message":"Hashtag added: OSSA-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"47f41ced86d27625eb242818fc24fde33a860183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:18:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daee701fc85847338fa93b3fdb05b762\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374f5844897c461d851cc373723ff453 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a541e3c3a9854baabc6e9f37524918e9 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ac4af6a7033d47b2afcf1c247070131f : SUCCESS in 1h 04m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d693b8b09fe847658be523fb855a35cf : SUCCESS in 46m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6994df8c55da46248e423eff21092fd2 : SUCCESS in 1h 00m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a61ecaecf97464daa058046dec04935 : SUCCESS in 1h 06m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c10d29f3a91480d95b29e8ca56c38db : SUCCESS in 11m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5599dacea2a142289ae2039a868f51aa : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6dca899aa734de4866942de6c852719 : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/10e8a38e0af743329f1725d5e90ae950 : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d2f0e4b36a1461f868bd2f978d29323 : SUCCESS in 24m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/11a709e5adba49b0afbe2d849b720542 : SUCCESS in 17m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8bdc2b1ced7540ebb39429814f2dcc0d : SUCCESS in 11m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/da6656d5f0d14fb1b95a32222e6edf99 : SUCCESS in 9m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3d283e8871dd40c88e982d1c28685841 : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cec6bf5d826b4de4a71e9a644d334a8b : SUCCESS in 8m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e1a51804a0f141768e546b034d503104 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f47e60770fe4ab1bc490cc72f13b16e : SUCCESS in 8m 19s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc4c450f84cd432989ea7c526f623304 : FAILURE in 28m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8cb14dffbb4345ca9f16fd77a68320fd : SUCCESS in 5m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/328aedffaeb5449189ced2f1423fe15c : SUCCESS in 1h 01m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cafd13c616a44908921485d7880dec94 : SUCCESS in 55m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9132fee6face487bae0c7ce82bc3edf9 : SUCCESS in 57m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34f85aacd3c64f3cb843def9a2b46b2b : SUCCESS in 26m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0a1ea10090457ab37dbe89a1376526 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9115ae1856684cc09a2852af79dda4af : SUCCESS in 18m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcd6a813daec4bb8bc1f04b4e8948c6a : SUCCESS in 43m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"0b7bbd594bc912a8210d051f1ff3d3764759e5ad","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 06:43:21.000000000","message":"Hashtags added: CVE-2026-97149, ossa-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"d1650ad204de259c0dc41cc27911e01b352326cc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-25 07:09:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3475161a007348beedc818b47baee175f4b55b7a","revisions":{"3475161a007348beedc818b47baee175f4b55b7a":{"kind":"REWORK","_number":1,"created":"2026-09-23 22:04:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/1007048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/1007048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/1007048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/1007048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/1007048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/1007048/1"}}},"commit":{"parents":[{"commit":"5104042f12efcaf90c28b656a378da8959004826","subject":"CHANGELOG for 2.37.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5104042f12efcaf90c28b656a378da8959004826"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-09-10 13:41:44.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-23 22:04:29.000000000","tz":-420},"subject":"tempurl: Disallow X-Copy-From on tempurl requests","message":"tempurl: Disallow X-Copy-From on tempurl requests\n\nDisallow clients using the X-Copy-From header to make a copy of any\nobject within the account using a PUT tempurl by adding the x-copy-from\nheader to DISALLOWED_INCOMING_HEADERS. Requests that use the header now\nget a 400 response, like the requests that use X-Object-Manifest or\nX-Symlink-Target.\n\nX-Copy-From-Account is not in the list, similar to\nX-Symlink-Target-Account. These headers have no effect without\nX-Copy-From or X-Symlink-Target.\n\nCloses-Bug: #2166876\nChange-Id: Ibe98ae41cba42997bbba00c74bb24bee22f32c05\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n(cherry picked from commit 32943ef7b9691f0447c3cfa93bb2f492958283a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3475161a007348beedc818b47baee175f4b55b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3475161a007348beedc818b47baee175f4b55b7a"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
