)]}'
{"id":"openstack%2Fswift~1007049","triplet_id":"openstack%2Fswift~stable%2F2025.2~Ibe98ae41cba42997bbba00c74bb24bee22f32c05","project":"openstack/swift","branch":"stable/2025.2","hashtags":["CVE-2026-97149","OSSA-2026-041","ossa-2026-041"],"change_id":"Ibe98ae41cba42997bbba00c74bb24bee22f32c05","subject":"tempurl: Disallow X-Copy-From on tempurl requests","status":"NEW","created":"2026-09-23 22:04:59.000000000","updated":"2026-09-24 06:43:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a940bc567fae9e56e8d61aa2e086543d0131cbf","_number":1007049,"virtual_id_number":1007049,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-23 23:14:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 22:04:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2026-09-23 22:46:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-23 23:14:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a4823aab9503905502bc0b7fb58091f148b1285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 22:04:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"197af6d11265e963bd97d5ab6a2f508bb185da4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 22:46:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b53b36241d244547a2c2a3351b98dded\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e2dfa9c974ec4d95beb87780fe223079 : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9da215d9f9c4d6d974018f9fe074632074059921","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-23 23:10:34.000000000","message":"Hashtag added: OSSA-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"eaf7beded9a6fd15adaf6e2956fdbe4f12a7156e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:14:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8437b78cfda41a79daacdc14833809e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3204889530149e79c56eab6a0394d7a : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c432482b0c84856bb8de0dc9287276c : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/836bd213c21f4188a897e8727a0d5d17 : SUCCESS in 1h 03m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a0238c158cc4186a55873c4344888c6 : SUCCESS in 38m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb65288db36641abb1460d5c79e9ca9c : SUCCESS in 47m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00d11a4fbb2f4538aa13d88aa1c5cd32 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18ffb14e4d824ed49c578c41c1b0cc09 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/413280fd57f74701b67f2108b013e5b0 : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/643f2fc6a3ec46fb80bfb257ec56e4fc : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e6f5d297f9f4b019e7ab5e05b26cc78 : SUCCESS in 12m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/364d44509e874322b5d64e034c920fc2 : SUCCESS in 10m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b180af50b97f44c7b9d8fcea10cb563a : SUCCESS in 19m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/17d010b21147492a90a82089763d5706 : SUCCESS in 25m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/373de547d82b4fd49ab57eed20899614 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e0389423fc3b4ffeac4d1d8390131a9a : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2803e2cab8614f92a42305157bcd0822 : SUCCESS in 14m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e360f3e9462416fa30261fae9e1c3b2 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/443a12a20f504514bb63b726f4c10146 : FAILURE in 24m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/75a23703aa6b445a815a4fad5d68890e : SUCCESS in 5m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d30ce199faa54d5d8614869adc4af1f2 : SUCCESS in 45m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad4117b930ac4a72ad7aadcee8a766f7 : SUCCESS in 55m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9938368c4fe94df79babfa58a7c04e47 : SUCCESS in 55m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db9865a61c5b4491b4a0558bba493567 : SUCCESS in 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be24f84dc8bf4547aee4ecafd730c307 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88e9fb4860264b5990e9ce0ba72096cd : SUCCESS in 25m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d34ec414415c42228b79713a5e66377e : SUCCESS in 57m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e95e41e00d5241ffbe7fe585c7244465 : SUCCESS in 32m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"4a940bc567fae9e56e8d61aa2e086543d0131cbf","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 06:43:21.000000000","message":"Hashtags added: CVE-2026-97149, ossa-2026-041","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"037e0f6c692572ce6d68f31abf05f7a943814742","revisions":{"037e0f6c692572ce6d68f31abf05f7a943814742":{"kind":"REWORK","_number":1,"created":"2026-09-23 22:04:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/1007049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/1007049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/1007049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/1007049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/1007049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/1007049/1"}}},"commit":{"parents":[{"commit":"deaa3e4691bb19224f5ef57e7fbcd08b7eff5277","subject":"CHANGELOG for 2.36.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/deaa3e4691bb19224f5ef57e7fbcd08b7eff5277"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-09-10 13:41:44.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-23 22:04:44.000000000","tz":-420},"subject":"tempurl: Disallow X-Copy-From on tempurl requests","message":"tempurl: Disallow X-Copy-From on tempurl requests\n\nDisallow clients using the X-Copy-From header to make a copy of any\nobject within the account using a PUT tempurl by adding the x-copy-from\nheader to DISALLOWED_INCOMING_HEADERS. Requests that use the header now\nget a 400 response, like the requests that use X-Object-Manifest or\nX-Symlink-Target.\n\nX-Copy-From-Account is not in the list, similar to\nX-Symlink-Target-Account. These headers have no effect without\nX-Copy-From or X-Symlink-Target.\n\nCloses-Bug: #2166876\nChange-Id: Ibe98ae41cba42997bbba00c74bb24bee22f32c05\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n(cherry picked from commit 32943ef7b9691f0447c3cfa93bb2f492958283a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/037e0f6c692572ce6d68f31abf05f7a943814742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/037e0f6c692572ce6d68f31abf05f7a943814742"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
