)]}'
{"id":"openstack%2Fswift~1007050","triplet_id":"openstack%2Fswift~stable%2F2025.1~Ibe98ae41cba42997bbba00c74bb24bee22f32c05","project":"openstack/swift","branch":"stable/2025.1","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-29 04:25:23.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":["CVE-2026-97149","OSSA-2026-041","ossa-2026-041"],"change_id":"Ibe98ae41cba42997bbba00c74bb24bee22f32c05","subject":"tempurl: Disallow X-Copy-From on tempurl requests","status":"NEW","created":"2026-09-23 22:05:12.000000000","updated":"2026-09-29 04:25:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1390a14485089f87eba5d9da8cca20dc50d82553","_number":1007050,"virtual_id_number":1007050,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-29 04:25:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":1,"date":"2026-09-25 07:09:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-29 00:13:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-29 00:13:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 22:05:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2026-09-23 22:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-23 23:13:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 07:09:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a275b2a561ad42dc5c3e02a7375c03ae792157ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-23 22:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a901c17a44edca7f8591fd55f2f94a6ad12a0c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 22:47:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb647fb8f39f47b4b96f53ff25288242\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a67a2822611a47ba8c1eb7b8910da8c4 : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b829b4e16195dfd08c6e4633060c89e5dd0620d9","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-23 23:10:34.000000000","message":"Hashtag added: OSSA-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"003a3a627d2770653c7683c516fdd7ef371e3b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 23:13:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71d5d6c966784f349ed10a7ecb321e15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76bf126526d4284b48ce78ee3aa7c5b : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bedd27f1886f4b0c8d27b900449c36bf : SUCCESS in 8m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8be5c8cff4a748c2ae56f19fa4689c9c : FAILURE in 32m 02s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c99761cfca564bc5b8dcad4f184dd512 : SUCCESS in 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e02f828fcf14158a450d3c0c674cdfb : SUCCESS in 1h 02m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43232ba3a71445f1b9e63ee6c04c0f3d : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b23f6b5a9124e5da79a0671c986957d : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8030868332f4a82830b76b859d7990e : SUCCESS in 22m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7bc0c1cb88cb449d9e2ababce03ed87e : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3200ef87c21b4ee78675da4ccb46098f : SUCCESS in 1m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7649d2f2772f41c4a7b8d1aa7beba7e6 : SUCCESS in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7862a0a6e984483392fcab6926d1f075 : SUCCESS in 26m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6cd29bc5e15c46558220c546501b6951 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/77a59895998d4e78ba1432c48711de46 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7174360afb784972926a994b7d6f578c : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b74a2841344473bb3a351417fc5c110 : SUCCESS in 3m 22s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5403ea0230bb4c8e9b3f9a44fb0277d5 : FAILURE in 30m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59ff74ddbc024e019ff17fa0da217f2c : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aef2f0af03e54e0a8fb1b14139116467 : SUCCESS in 38m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/121305537db3405aa6487dfd10a50740 : SUCCESS in 20m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04979b3a66f642818a4cd5ff533d79f0 : SUCCESS in 43m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15ec015169504d29a05c626e8ca3146f : SUCCESS in 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e607fb311c4461989934ed257f97a2 : SUCCESS in 2m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b79551064f49428697feb128e69628b2 : SUCCESS in 18m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fac9929f9bde48eabc7ecdab40619489 : SUCCESS in 51m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/069bd3e1cd1d424a91798a526a4f1af3 : SUCCESS in 57m 17s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"374eeb7191fe900108bd99e6eb7c67c96d9abd60","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 06:43:21.000000000","message":"Hashtags added: CVE-2026-97149, ossa-2026-041","accounts_in_message":[],"_revision_number":1},{"id":"30247a72f2aefe7a05658be5609d3030ca9816d4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-25 07:09:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03b4599c193db0aee25121e08f485a1cad56e910","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-29 00:13:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63b58d68b254f864cc181a188ae858e6d16897cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 00:14:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"026882f69c3776255c953baaed0e8bbfb2e6f57a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 01:27:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdc52d116e404c8a8a2b72e225366e62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0536dd3baa74c3d81cfc134bf7b580d : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/782cd4974b3f4f7db84fd93dc23791b6 : SUCCESS in 9m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/30be69cc7f5f45558f379a1d5413bb61 : FAILURE in 48m 50s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3eec5b2ac1540df856db9374549fb35 : SUCCESS in 57m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e9f8ec889a047189bba21a4321a72c9 : SUCCESS in 1h 04m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/357fefc5c91048b2a62e7eb0f905c47b : RETRY_LIMIT in 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5a66806df0b4a74b58374edb3b72a78 : SUCCESS in 1m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfe5b8e1e7f64024b3d973f25152efc6 : SUCCESS in 22m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/07673557ba3e4997beb92d8aa9a1a6a8 : SUCCESS in 27m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/532a0e3c57814699a3521619ace0336b : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/21ae2837e76d4224ba86b5dbc78b6234 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/751ed31f98fc4375bef24c82791220f3 : SUCCESS in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dbdfdbcc2b954889b981af12d4cd9b29 : SUCCESS in 3m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/159997d9dc22486f8e61261d79c25c98 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4435ac68323848f391049b1591286a1e : SUCCESS in 54m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f1528e8cec941b6be1dea735058ee7e : SUCCESS in 50m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90dc157d979e44388b433427e2fa3fee : SUCCESS in 49m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ffd410940644a7195b87ec629c924b6 : SUCCESS in 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8834a92fdf694f199fd16ccfe8290b75 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df0574375f294575b54ebae516e1e82c : SUCCESS in 12m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fe48a49b872417ca266cb29222ead6f : FAILURE in 44m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5edcfb1049dd495eb45c675b2a02f87c : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"91ecf9b0dc592982becdad5bec63f94a0d55cded","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-29 03:13:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"326783593538ba773f28a533bc0320a6821f302b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 04:17:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91795744921f45da98289770e98aa7f5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0259ed04e6974ff1bec8f6a43f782487 : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1390a14485089f87eba5d9da8cca20dc50d82553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 04:25:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbfb9350e5c0497fb2ffc0ce69b86482\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83804124fb24c099f722ddfda6b9c3b : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/588e9df4f92642c7b02603c2c00fb514 : SUCCESS in 10m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2e8a53e59f2842868384190b54b6b253 : FAILURE in 26m 32s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ddefe70fa2449959c614d8740e3cdfd : SUCCESS in 58m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98b71c17ac914cf5a9b05f5fefb43c53 : SUCCESS in 53m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a87b1e40f41147fbad8e2f76429d7ece : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ded3e5ff1284dc0b94505dd54330a59 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b17ee0a4a77c441f90ff2f72620b4268 : SUCCESS in 12m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fdbdf3b6f70a4af2b01ffa877d9b4fca : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e287704416684d35b562bdeb981c4966 : SUCCESS in 1m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4789ed27a9ee43aabbfab3acd772a97a : SUCCESS in 18m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/941605cbf51c41928bb9b7495fb0badc : SUCCESS in 31m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5dcb7a73f1df4ab49d7dd7aadb1b1644 : SUCCESS in 8m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d3cd0b760bc24b52bed16d22b23e6161 : SUCCESS in 12m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aca58f097a9c49e5a350e2ce572f0de4 : SUCCESS in 13m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30b3ff8339db484198555db5a07caaf8 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83495b91684a4aef9669f87dcf88b430 : FAILURE in 18m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a612781c37a6486abafb9792009a434f : SUCCESS in 6m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9d534cd380a4cb4b7e77d889f5336e5 : SUCCESS in 56m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6dd24b8bd2c54a6d91284e3ef785827f : SUCCESS in 46m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d8f889fa7dc4327ae54bc65897001e2 : SUCCESS in 23m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c17c327be6c84085a4a877b9fe5ce5a0 : SUCCESS in 1m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b75583606e14a03b067bc37e89f6b02 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c5214c41ce945479fce60f9f9a8548b : SUCCESS in 24m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0e3ffec18f94e5fa1d0900ba13d9510 : FAILURE in 40m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70802ca2b9084ba4adf786b1d90d03a8 : SUCCESS in 57m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cca20275be935e3b2296c2f757b11948d0e1fb9","revisions":{"7cca20275be935e3b2296c2f757b11948d0e1fb9":{"kind":"REWORK","_number":1,"created":"2026-09-23 22:05:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/1007050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/1007050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/1007050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/1007050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/1007050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/1007050/1"}}},"commit":{"parents":[{"commit":"fc875f93e82edf98805d0eead3e3c961673c3ef8","subject":"CHANGELOG for 2.35.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc875f93e82edf98805d0eead3e3c961673c3ef8"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-09-10 13:41:44.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-23 22:05:05.000000000","tz":-420},"subject":"tempurl: Disallow X-Copy-From on tempurl requests","message":"tempurl: Disallow X-Copy-From on tempurl requests\n\nDisallow clients using the X-Copy-From header to make a copy of any\nobject within the account using a PUT tempurl by adding the x-copy-from\nheader to DISALLOWED_INCOMING_HEADERS. Requests that use the header now\nget a 400 response, like the requests that use X-Object-Manifest or\nX-Symlink-Target.\n\nX-Copy-From-Account is not in the list, similar to\nX-Symlink-Target-Account. These headers have no effect without\nX-Copy-From or X-Symlink-Target.\n\nCloses-Bug: #2166876\nChange-Id: Ibe98ae41cba42997bbba00c74bb24bee22f32c05\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n(cherry picked from commit 32943ef7b9691f0447c3cfa93bb2f492958283a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cca20275be935e3b2296c2f757b11948d0e1fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cca20275be935e3b2296c2f757b11948d0e1fb9"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
