)]}'
{"id":"openstack%2Fswift~1007964","triplet_id":"openstack%2Fswift~master~Ic5eb2d074790c645d1e167f8261a47f7bc23e489","project":"openstack/swift","branch":"master","topic":"cohanaja-replication-mode-config","hashtags":[],"change_id":"Ic5eb2d074790c645d1e167f8261a47f7bc23e489","subject":"DNM: Add sharder replication benchmark","status":"NEW","created":"2026-09-29 21:05:20.000000000","updated":"2026-09-29 22:18:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd05d52f025782f4a180b3a8b11650783453c5e4","_number":1007964,"virtual_id_number":1007964,"owner":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-29 22:18:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-29 21:57:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-29 22:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fe67f03bd26c9a8986a7afeafe98e03d19ec804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-29 21:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a407e3c52716f984c8782a311a972c7827651f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-29 21:08:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c0726fe347c3ebe1af4f6da0c0b632337290694d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-29 21:13:27.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"19d1c4a96569936fc4504da5e393423a8f73f0bd","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-29 21:18:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a1365e3e79eed11d00613421ddcac06465ba55e2","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-29 21:20:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ab66d4d83c0dc1169425d118de24537d160e939","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 21:57:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7f24039132408d95e5e77781f76399\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b7c112392ecc4f7cb1aa24c01c3611bf : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":3},{"id":"fd05d52f025782f4a180b3a8b11650783453c5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 22:18:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f008f5fb9ae74424b96e75dd68d45669\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7148a3bb516a43c5b2ffe756bfed5438 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97c418633cbc4016a9d46b7afbf17b10 : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/60966aae390c462c9ead9827af28c54b : SUCCESS in 1h 01m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d713fd3c28f1410cb5fb5e58427bdf24 : SUCCESS in 59m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa18c6830e7547228750029bfed99ef7 : SUCCESS in 34m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff9b0c638e26447286a56393a7e053e8 : SUCCESS in 43m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3cf5f1c3b3241c6bca9bf585ed425ac : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6cb056bec0d4651a96c68303d94a9eb : SUCCESS in 22m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/968e2b3e23404b9cacf3bb02fa7233bd : SUCCESS in 11m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/579205e2bfbe44e38b45a3a22cf4bc06 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2f8fe21810b24176a68cb450119398ed : SUCCESS in 13m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3ea74c5cc5f54f959e4219f4bb3b7b73 : SUCCESS in 19m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/98ffb418340346b1b4abc09ef0ace72c : SUCCESS in 22m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/aed90edf05234794912bfac00cc2a733 : SUCCESS in 20m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1b8beb1172d4404491018e33d715515b : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/752a1d5c679c4d42afaa6f6b855547c4 : SUCCESS in 9m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e5164c9f3c01410c905ce113ae06d66b : SUCCESS in 16m 11s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/99900182b9f44d4e863a4419c1d578eb : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5566f9708a294684a11a6d8d7b8c4597 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3d9074f6fcb44d89d6d328f183ce4c9 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6aa7cc2177840f0be225f86e6056116 : SUCCESS in 45m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dbb5b1fe89b4b9c8783fca813b60a2a : SUCCESS in 59m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6017a1e957214411b1527de2a750ef43 : SUCCESS in 56m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2243f86a19a4fd49f4abce86f03b862 : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc462922a854d04933f59077bb70322 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68480630a1d64f6b9dce4536dac2b008 : SUCCESS in 18m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da4948894e7342d4bb270c691ae38abf : SUCCESS in 49m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cdabd059b00788f5ce7e44d1dc9a30c69b2de874","revisions":{"6ad1be5275321ceced72098f5d9358814aeaf3fa":{"kind":"REWORK","_number":1,"created":"2026-09-29 21:05:20.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/64/1007964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/1007964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/1007964/1"}}},"commit":{"parents":[{"commit":"7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b","subject":"Adds rsync/usync/auto replication choice config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-09-29 20:58:15.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-09-29 20:58:15.000000000","tz":-420},"subject":"DNM: Add sharder replication benchmark under live writes","message":"DNM: Add sharder replication benchmark under live writes\n\nRecent work toward improving the container sharder via replacing usync with\nrsync has led to the need for more robust, agent-operable tooling to make it\neasier for experiments onthe sharder to be ran.\n\nThis patch includes the scripts I\u0027ve been messing with to accelerate testing,\nas well as a potential (but currently really messy) rsync_then_merge\nimprovement that\u0027s shown a positive 80-90%ish reduction in total_replicate_time\nwhen ran in a (somewhat beefy) benchmark VSAIO env (8 vCPU, 12 GB).\n\n`tools/sharder-benchmark/run_live.py` runs the whole experiment on an\nall-in-one cluster. This also includes the scripts it depends on,\n`fill_container_db` and `live_load`\n\nYou can run the benchmark easily with 1 command, but you could probably also\nget your agent to look through the logs and better your understanding of them.\n\nCo-Authored-By: Claude Opus 5.5 \u003cnoreply@anthropic.com\u003e\nCo-Authored-By: GPT 5.6 Sol \u003cnoreply@openai.com\u003e\nChange-Id: Ic5eb2d074790c645d1e167f8261a47f7bc23e489\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ad1be5275321ceced72098f5d9358814aeaf3fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ad1be5275321ceced72098f5d9358814aeaf3fa"}]},"branch":"refs/heads/master"},"a400732da4e7f6b45fc0ad1d64530a365d3f04dd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-29 21:08:50.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/64/1007964/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/1007964/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/1007964/2"}}},"commit":{"parents":[{"commit":"7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b","subject":"Adds rsync/usync/auto replication choice config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-09-29 20:58:15.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-09-29 21:08:13.000000000","tz":-420},"subject":"DNM: Add sharder replication benchmark w/ live writes","message":"DNM: Add sharder replication benchmark w/ live writes\n\nRecent work toward improving the container sharder via replacing usync with\nrsync has led to the need for more robust, agent-operable tooling to make it\neasier for experiments onthe sharder to be ran.\n\nThis patch includes the scripts I\u0027ve been messing with to accelerate testing,\nas well as a potential (but currently really messy) rsync_then_merge\nimprovement that\u0027s shown a positive 80-90%ish reduction in total_replicate_time\nwhen ran in a (somewhat beefy) benchmark VSAIO env (8 vCPU, 12 GB).\n\n`tools/sharder-benchmark/run_live.py` runs the whole experiment on an\nall-in-one cluster. This also includes the scripts it depends on,\n`fill_container_db` and `live_load`\n\nYou can run the benchmark easily with 1 command, but you could probably also\nget your agent to look through the logs and better your understanding of them.\n\nCo-Authored-By: Claude Opus 5.5 \u003cnoreply@anthropic.com\u003e\nCo-Authored-By: GPT 5.6 Sol \u003cnoreply@openai.com\u003e\nChange-Id: Ic5eb2d074790c645d1e167f8261a47f7bc23e489\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a400732da4e7f6b45fc0ad1d64530a365d3f04dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a400732da4e7f6b45fc0ad1d64530a365d3f04dd"}]},"branch":"refs/heads/master"},"cdabd059b00788f5ce7e44d1dc9a30c69b2de874":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-29 21:13:27.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/64/1007964/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/1007964/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/1007964/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/1007964/3"}}},"commit":{"parents":[{"commit":"7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b","subject":"Adds rsync/usync/auto replication choice config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-09-29 20:58:15.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-09-29 21:13:03.000000000","tz":-420},"subject":"DNM: Add sharder replication benchmark","message":"DNM: Add sharder replication benchmark\n\nRecent work toward improving the container sharder via replacing usync\nwith rsync has led to the need for more robust, agent-operable tooling\nto make it easier for experiments on the sharder to be run.\n\nThis patch includes the scripts I\u0027ve been messing with to accelerate\ntesting, as well as a potential (but currently really messy)\nrsync_then_merge improvement that\u0027s shown a positive 80-90%ish\nreduction in total_replicate_time when run in a (somewhat beefy)\nbenchmark VSAIO env (8 vCPU, 12 GB).\n\n`tools/sharder-benchmark/run_live.py` runs the whole experiment on an\nall-in-one cluster. This also includes the scripts it depends on,\n`fill_container_db` and `live_load`.\n\nYou can run the benchmark easily with 1 command, but you could probably\nalso get your agent to look through the logs and better your\nunderstanding of them.\n\nCo-Authored-By: Claude Opus 5.5 \u003cnoreply@anthropic.com\u003e\nCo-Authored-By: GPT 5.6 Sol \u003cnoreply@openai.com\u003e\nChange-Id: Ic5eb2d074790c645d1e167f8261a47f7bc23e489\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdabd059b00788f5ce7e44d1dc9a30c69b2de874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdabd059b00788f5ce7e44d1dc9a30c69b2de874"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
