)]}'
{"id":"openstack%2Fswift~1008368","triplet_id":"openstack%2Fswift~master~Ia8829b7e8a8117c02e2a5e0c311e51b6c120ada4","project":"openstack/swift","branch":"master","topic":"codex/fix-996383-object-type","hashtags":[],"change_id":"Ia8829b7e8a8117c02e2a5e0c311e51b6c120ada4","subject":"sq? keep object response labels on client reads","status":"ABANDONED","created":"2026-10-02 01:53:23.000000000","updated":"2026-10-02 21:21:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b241e50851fd60ca62db570b98ddad2fe73ae2ab","_number":1008368,"virtual_id_number":1008368,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-02 03:13:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-02 02:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-10-02 03:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2db15a8962b0dfd99327c2dca98ef5e242732724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-10-02 01:53:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d62fea455b324aa80526d9889b57e8b9280f4401","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 02:54:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2497213b3694542bffa6bac3142022a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/741532409e5144648c6e84cf0787392d : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"71ab69ea56e51e30f1251010907ad0703f397c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 03:13:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b33b2f4c2f7d4a158fd426e066f2a40f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac669b282a04b8ba7ed5751c64a368a : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/703ccc2b38fc48c2ab6c0a363e6da0d9 : SUCCESS in 4m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/dc1129ab9b0d4ba994e924b7ac8f8c74 : SUCCESS in 1h 10m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e44f58a351484658a3cf851a08a88044 : SUCCESS in 1h 13m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa78c620a0444d3d9d10e690104735d5 : SUCCESS in 57m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af7d817978974231b46008e35f708bf0 : SUCCESS in 1h 12m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0880ebf9837c467f862d31bf00f7091a : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/963124a43d714a229a021d61bfffdbc2 : SUCCESS in 11m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb11bde89d754b88bc81e06c1e1309d1 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/98be09479b9f49d999973a65a629a006 : SUCCESS in 11m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af56a8b1103b426f942ceb485e2e9455 : SUCCESS in 26m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fb8f9fcc00d546e788f150854ef540fd : SUCCESS in 9m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e6f4941b7a7e4fbfb7d8d1ad2cbf05be : SUCCESS in 11m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/18761f272598412bbc96437ada6f1710 : SUCCESS in 26m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5a2265b493904da3b78946b8cece151c : SUCCESS in 9m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b054352918b94a96b5f8379b9bc3997a : SUCCESS in 16m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76e7497dc24a4fe3bc5d8f75c84a2b6b : SUCCESS in 8m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/948807e7641447b5bcd7dd84a35867ea : SUCCESS in 19m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/775eb1628320456bba3480747263d29f : SUCCESS in 8m 45s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65adda913ef045aea9e8914ae3040096 : SUCCESS in 9m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6eba6b21054e2c836c98ae895a18e3 : SUCCESS in 58m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51401f2d08ff4d52a8e60d4d1c462770 : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/330f72b371b840f9b349d39070e457a5 : SUCCESS in 1h 03m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2888855862d42bdac88ce4e4c7a0c87 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc6bff334ad94516b61ebef8beaa6eab : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a783772fc93c475488d01c9642c0afd0 : SUCCESS in 23m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/000a7e0507614a08a7492c16dbcc8bf4 : SUCCESS in 57m 48s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b241e50851fd60ca62db570b98ddad2fe73ae2ab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-10-02 21:21:17.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/996383/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40d2a2c1cd3d1f6b639208fdfc746573d6f7e45b","revisions":{"40d2a2c1cd3d1f6b639208fdfc746573d6f7e45b":{"kind":"REWORK","_number":1,"created":"2026-10-02 01:53:23.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/68/1008368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/1008368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/1008368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/1008368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/1008368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/1008368/1"}}},"commit":{"parents":[{"commit":"2cda5ed7607220c28a992016ef0dd5f73034f13e","subject":"proxy_logging: add new label object_type when emitting metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cda5ed7607220c28a992016ef0dd5f73034f13e"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-10-02 01:12:26.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-10-02 01:52:54.000000000","tz":-420},"subject":"sq? keep object response labels on client reads","message":"sq? keep object response labels on client reads\n\nDo not let a COPY source GET set the object type of the client write.\nUse the client resource and method to control response label updates.\nUse the current S3API sysmeta helper in the object tests. Add COPY\nregression coverage and set client methods in the read test fixtures.\n\nAlso, MPU GET/HEAD now sets object_type\u003d\u0027mpu\u0027. Ordinary and unknown\nobjects retain None\n\nChange-Id: Ia8829b7e8a8117c02e2a5e0c311e51b6c120ada4\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40d2a2c1cd3d1f6b639208fdfc746573d6f7e45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40d2a2c1cd3d1f6b639208fdfc746573d6f7e45b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
