)]}'
{"id":"openstack%2Fswift~126371","triplet_id":"openstack%2Fswift~stable%2Ficehouse~I89d7cd27112445893852e62df857c3d5262c27b3","project":"openstack/swift","branch":"stable/icehouse","topic":"bug/1375348","hashtags":[],"change_id":"I89d7cd27112445893852e62df857c3d5262c27b3","subject":"Ensure sys.exit called in fork_child after exception","status":"MERGED","created":"2014-10-06 17:46:09.000000000","updated":"2014-10-28 22:02:08.000000000","submitted":"2014-10-28 22:02:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0d83df5c0f4c1bfc55d44f8d091e441db624bde7","_number":126371,"virtual_id_number":126371,"owner":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":2,"date":"2014-10-28 22:02:08.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":1005,"name":"Eamonn O\u0027Toole","email":"eamonn.otoole@hp.com","username":"eamonn-otoole"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2014-10-24 14:42:23.000000000","_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"value":0,"date":"2014-10-22 18:37:51.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":2,"date":"2014-10-22 18:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-10-24 13:41:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2014-10-22 22:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":1,"date":"2014-10-22 18:36:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1005,"name":"Eamonn O\u0027Toole","email":"eamonn.otoole@hp.com","username":"eamonn-otoole"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":1,"date":"2014-10-22 18:48:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2014-10-22 18:36:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":1,"date":"2014-10-22 21:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2014-10-22 18:36:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":1,"date":"2014-10-28 18:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":1005,"name":"Eamonn O\u0027Toole","email":"eamonn.otoole@hp.com","username":"eamonn-otoole"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1005,"name":"Eamonn O\u0027Toole","email":"eamonn.otoole@hp.com","username":"eamonn-otoole"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-10-07 07:29:01.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2014-10-07 07:47:50.000000000","updated_by":{"_account_id":1005,"name":"Eamonn O\u0027Toole","email":"eamonn.otoole@hp.com","username":"eamonn-otoole"},"reviewer":{"_account_id":1005,"name":"Eamonn O\u0027Toole","email":"eamonn.otoole@hp.com","username":"eamonn-otoole"},"state":"REVIEWER"},{"updated":"2014-10-09 23:20:11.000000000","updated_by":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"reviewer":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"state":"REVIEWER"},{"updated":"2014-10-15 20:32:26.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2014-10-16 17:34:30.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2014-10-22 18:48:29.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2014-10-22 21:10:57.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2014-10-22 22:46:13.000000000","updated_by":{"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},"reviewer":{"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},"state":"REVIEWER"},{"updated":"2014-10-24 13:41:41.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2014-10-24 14:42:23.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-10-28 18:08:29.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2014-10-28 22:02:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"85ab5194980c8a02712a6c0a435d6cd0b02f4f0d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-06 17:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa23f9a12205e587cb3281852982930db212f823","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2014-10-06 17:53:54.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* cluster-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/126371/536/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a13f94588fc00dc1cf5e957abd3b095fecbe62c0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-06 18:44:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-swift-pep8 http://logs.openstack.org/71/126371/1/check/gate-swift-pep8/fa0a9e2 : SUCCESS in 1m 51s\n- gate-swift-docs http://docs-draft.openstack.org/71/126371/1/check/gate-swift-docs/82f8ee1/doc/build/html/ : SUCCESS in 1m 43s\n- gate-swift-python26 http://logs.openstack.org/71/126371/1/check/gate-swift-python26/0511ddb : SUCCESS in 2m 38s\n- gate-swift-python27 http://logs.openstack.org/71/126371/1/check/gate-swift-python27/10899ae : SUCCESS in 2m 26s\n- check-tempest-dsvm-full http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-full/587525c : SUCCESS in 54m 25s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-postgres-full/62c77e6 : SUCCESS in 52m 13s\n- check-tempest-dsvm-neutron http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-neutron/95c2851 : SUCCESS in 54m 58s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-neutron-heat-slow/ce220b4 : SUCCESS in 21m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/71/126371/1/check/gate-tempest-dsvm-large-ops/920278f : SUCCESS in 25m 44s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/71/126371/1/check/gate-tempest-dsvm-neutron-large-ops/d066544 : SUCCESS in 20m 57s\n- check-swift-dsvm-functional http://logs.openstack.org/71/126371/1/check/check-swift-dsvm-functional/116954c : SUCCESS in 14m 51s\n- check-devstack-dsvm-cells http://logs.openstack.org/71/126371/1/check/check-devstack-dsvm-cells/1cd9356 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b04e7ff953e4ce39eee7188b3adba6d590d33d8a","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2014-10-07 07:29:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cbaa710de5d205f92ea8e1ac789be05057584b5a","author":{"_account_id":1005,"name":"Eamonn O\u0027Toole","email":"eamonn.otoole@hp.com","username":"eamonn-otoole"},"date":"2014-10-07 07:47:50.000000000","message":"Patch Set 1: Code-Review+1\n\nWell caught Jay.","accounts_in_message":[],"_revision_number":1},{"id":"2817d993b4c53f93905167545cc3ab9946ab5c6a","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-10-09 23:20:11.000000000","message":"Patch Set 1: Code-Review+1\n\nSeems good to me.\n\nI\u0027m confused as to why I can\u0027t +2 this change.","accounts_in_message":[],"_revision_number":1},{"id":"181bc37dd9a87b61bf5ec859963b90bdf4385d58","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2014-10-09 23:20:34.000000000","message":"Patch Set 1:\n\nOh, stable/icehose. Got it.","accounts_in_message":[],"_revision_number":1},{"id":"276dbde389d8c02e5fd82657c4bc553989cdaf81","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-10 21:21:36.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks like a clean cherry pick for a nasty bug.  The importance of the bug should probably be set to High if you want people from the stable maintenance team to look at this quicker, or nominate it for the Icehouse series in launchpad.","accounts_in_message":[],"_revision_number":1},{"id":"ff73cfc28086bc2205e5529f48a30f96fb69e135","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-10-15 19:42:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM appropriate for stable\nbut I\u0027m confused why stable-maint can +2 here.\n\nSwift is not participating in stable branch point releases, instead Swift is claiming backward compatibility AFAIK.","accounts_in_message":[],"_revision_number":1},{"id":"0262b0fa89254c7dad259cf90e210727caf7362b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-15 19:42:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"274d761380b37e54011dff0d4efe9e146fe54d45","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2014-10-15 20:32:26.000000000","message":"Patch Set 1:\n\nI guess it depends on what you mean by \"participates in stable/*\". We do, and have in the past, backported security issues. To me, that\u0027s participating in stable/* branches. We\u0027re not backporting bugfixes that aren\u0027t security issues.","accounts_in_message":[],"_revision_number":1},{"id":"d18b08badc9205acf86c4c3c1d01a1ca5c26ab47","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-15 20:51:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-swift-docs http://docs-draft.openstack.org/71/126371/1/gate/gate-swift-docs/fe661f1/doc/build/html/ : SUCCESS in 1m 13s\n- gate-swift-pep8 http://logs.openstack.org/71/126371/1/gate/gate-swift-pep8/dd05973 : SUCCESS in 1m 04s\n- gate-swift-python26 http://logs.openstack.org/71/126371/1/gate/gate-swift-python26/3155cf6 : FAILURE in 2m 18s\n- gate-swift-python27 http://logs.openstack.org/71/126371/1/gate/gate-swift-python27/a8b5127 : FAILURE in 2m 04s\n- gate-tempest-dsvm-full http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-full/ef28a3c : SUCCESS in 52m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-postgres-full/93cabb4 : SUCCESS in 49m 22s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-neutron/80d3bd9 : SUCCESS in 1h 02m 37s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-neutron-heat-slow/ce63228 : SUCCESS in 31m 08s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-large-ops/6aa9756 : SUCCESS in 27m 04s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-neutron-large-ops/bfcccc2 : SUCCESS in 21m 47s\n- gate-swift-dsvm-functional http://logs.openstack.org/71/126371/1/gate/gate-swift-dsvm-functional/4103d94 : SUCCESS in 20m 42s\n- gate-devstack-dsvm-cells http://logs.openstack.org/71/126371/1/gate/gate-devstack-dsvm-cells/485b8f0 : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4d4a66c69954b09fbca1913ca8595f80f0b411e5","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-10-15 21:29:26.000000000","message":"Patch Set 1:\n\nI said \"participating in stable branch point releases\" not \"participates in stable/*\"\n\nIn http://lists.openstack.org/pipermail/openstack-stable-maint/2014-October/002832.html I did say that swift stable/* has been used for security fixes only.\n\nFWIW this fix could be seen as a DoS prevention, although it is not remotely exploitable AFAICT.","accounts_in_message":[],"_revision_number":1},{"id":"b0cbf3f808dae8cd150ec64d6656552961ee24f1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-15 21:38:06.000000000","message":"Patch Set 1:\n\nGiven the nature of this issue I think it could be argued it is a security issue that we should probably get fixed.","accounts_in_message":[],"_revision_number":1},{"id":"77a657cbd89ef7a664750137f5e799bc527d73c0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-15 21:51:28.000000000","message":"Patch Set 1:\n\nHmm, is this tickling something in the stable/icehouse swift code?\n\nhttp://logstash.openstack.org/#eyJzZWFyY2giOiJtZXNzYWdlOlwiRkFJTDogdGVzdF9wcmludF9kYl9pbmZvX21ldGFkYXRhICh0ZXN0LnVuaXQuY2xpLnRlc3RfaW5mby5UZXN0Q2xpSW5mbylcIiBBTkQgdGFnczpcImNvbnNvbGVcIiBBTkQgcHJvamVjdDpcIm9wZW5zdGFjay9zd2lmdFwiIiwiZmllbGRzIjpbXSwib2Zmc2V0IjowLCJ0aW1lZnJhbWUiOiI2MDQ4MDAiLCJncmFwaG1vZGUiOiJjb3VudCIsInRpbWUiOnsidXNlcl9pbnRlcnZhbCI6MH0sInN0YW1wIjoxNDEzNDA5ODMyNjA2LCJtb2RlIjoiIiwiYW5hbHl6ZV9maWVsZCI6IiJ9","accounts_in_message":[],"_revision_number":1},{"id":"199be1cfbb8c811e84364cc8ec827feac3f0074f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-15 22:03:40.000000000","message":"Patch Set 1:\n\nI was wondering the same.  It passed when I did UT locally.  Thoughts John?","accounts_in_message":[],"_revision_number":1},{"id":"af41cb2bb1d649813229423be207557a8f32b284","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2014-10-16 17:34:30.000000000","message":"Patch Set 1:\n\nThe versions of everything (according to pip freeze) are identical between the passing and failing runs.\n\nI wonder if Id9fca209c9c1216f1949de7108bbe332808f1045 (which merged between the last check and the current approval) introduced a behavioral change being picked up in the failing test. It doesn\u0027t seem to touch any related code paths though...","accounts_in_message":[],"_revision_number":1},{"id":"3e55d35dbbc6200b5284a54956ce2349f33cb1cd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-16 20:01:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"470af6bcfbab87d7a66a3ca84dfe164feb9b6ade","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-16 20:01:54.000000000","message":"Patch Set 1:\n\nSince this passed before and the test is unrelated as far as I can tell, let\u0027s just run a recheck and see if it happens again.","accounts_in_message":[],"_revision_number":1},{"id":"cd6f0d823c4b1098a063b53c37ee8d84ff6afba1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-16 20:11:48.000000000","message":"Patch Set 1: Code-Review-1\n\nFailed again, unit tests aren\u0027t happy:\n\nhttps://jenkins04.openstack.org/job/gate-swift-python26/750/console","accounts_in_message":[],"_revision_number":1},{"id":"078a67a9d4738c70859652dfccf3b74a3d8b7bbb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-16 20:59:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-swift-pep8 http://logs.openstack.org/71/126371/1/check/gate-swift-pep8/62be622 : SUCCESS in 59s\n- gate-swift-docs http://docs-draft.openstack.org/71/126371/1/check/gate-swift-docs/27b53ee/doc/build/html/ : SUCCESS in 1m 14s\n- gate-swift-python26 http://logs.openstack.org/71/126371/1/check/gate-swift-python26/90bd6fb : FAILURE in 2m 17s\n- gate-swift-python27 http://logs.openstack.org/71/126371/1/check/gate-swift-python27/833d4d8 : SUCCESS in 2m 06s\n- check-tempest-dsvm-full http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-full/03680f3 : SUCCESS in 48m 41s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-postgres-full/baa4f02 : FAILURE in 53m 01s\n- check-tempest-dsvm-neutron http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-neutron/2327211 : SUCCESS in 54m 38s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-neutron-heat-slow/a65199a : SUCCESS in 27m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/71/126371/1/check/gate-tempest-dsvm-large-ops/802b9f7 : SUCCESS in 20m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/71/126371/1/check/gate-tempest-dsvm-neutron-large-ops/e977069 : SUCCESS in 19m 40s\n- check-swift-dsvm-functional http://logs.openstack.org/71/126371/1/check/check-swift-dsvm-functional/909b870 : SUCCESS in 14m 28s\n- check-devstack-dsvm-cells http://logs.openstack.org/71/126371/1/check/check-devstack-dsvm-cells/8ed8aa8 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"848f75694c9d3cf685b6084e52b85b9d45a7e8ef","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-17 01:41:36.000000000","message":"Patch Set 1:\n\nI tried rebasing my change on the latest code to include the change referenced above and still couldn\u0027t recreate it locally.  Given that the failure didn\u0027t happen for both py26 and py27 I wonder if we have a timing issue here somewhere.","accounts_in_message":[],"_revision_number":1},{"id":"2780c166ad9b78e663870b7f21a39cff0bab0c8b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-20 16:38:27.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"697ae4d26770809bfed3784672865d2c9e09da9d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 17:43:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-pep8 http://logs.openstack.org/71/126371/1/check/gate-swift-pep8/11e9bce : SUCCESS in 1m 09s\n- gate-swift-docs http://docs-draft.openstack.org/71/126371/1/check/gate-swift-docs/f5c996d/doc/build/html/ : SUCCESS in 1m 16s\n- gate-swift-python26 http://logs.openstack.org/71/126371/1/check/gate-swift-python26/4cd325e : SUCCESS in 2m 06s\n- gate-swift-python27 http://logs.openstack.org/71/126371/1/check/gate-swift-python27/c6101a7 : SUCCESS in 2m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-full/cf99200 : SUCCESS in 50m 32s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-postgres-full/77e556b : SUCCESS in 55m 02s\n- check-tempest-dsvm-neutron http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-neutron/309163e : SUCCESS in 58m 15s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/71/126371/1/check/check-tempest-dsvm-neutron-heat-slow/1661edc : SUCCESS in 22m 03s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/71/126371/1/check/gate-tempest-dsvm-large-ops/dc6c841 : SUCCESS in 20m 51s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/71/126371/1/check/gate-tempest-dsvm-neutron-large-ops/a5f4e4a : SUCCESS in 20m 04s\n- check-swift-dsvm-functional http://logs.openstack.org/71/126371/1/check/check-swift-dsvm-functional/1514884 : SUCCESS in 15m 16s\n- check-devstack-dsvm-cells http://logs.openstack.org/71/126371/1/check/check-devstack-dsvm-cells/861ce2a : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8866effc127845401afc445f8427ada6b667560e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 17:43:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"3bff27b7f0587837dbed23dff41a74cbcc7cd634","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 19:37:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-swift-docs http://docs-draft.openstack.org/71/126371/1/gate/gate-swift-docs/6463816/doc/build/html/ : SUCCESS in 1m 20s\n- gate-swift-pep8 http://logs.openstack.org/71/126371/1/gate/gate-swift-pep8/3213e23 : SUCCESS in 1m 21s\n- gate-swift-python26 http://logs.openstack.org/71/126371/1/gate/gate-swift-python26/6562fbb : FAILURE in 2m 24s\n- gate-swift-python27 http://logs.openstack.org/71/126371/1/gate/gate-swift-python27/3a62c9a : SUCCESS in 2m 03s\n- gate-tempest-dsvm-full http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-full/e36e129 : SUCCESS in 51m 35s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-postgres-full/ee7dcd5 : SUCCESS in 52m 06s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-neutron/dd0a637 : SUCCESS in 1h 01m 13s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-neutron-heat-slow/029f1ad : SUCCESS in 27m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-large-ops/2b027ef : SUCCESS in 21m 37s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/71/126371/1/gate/gate-tempest-dsvm-neutron-large-ops/473a815 : SUCCESS in 21m 28s\n- gate-swift-dsvm-functional http://logs.openstack.org/71/126371/1/gate/gate-swift-dsvm-functional/199ecb5 : SUCCESS in 16m 14s\n- gate-devstack-dsvm-cells http://logs.openstack.org/71/126371/1/gate/gate-devstack-dsvm-cells/e0d0264 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b8cbfae61aa24d5469c63db431ccb4b30a20bd4e","author":{"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},"date":"2014-10-20 23:41:13.000000000","message":"Patch Set 1:\n\nI think the problem you\u0027re running into is that dictionary order is not guaranteed, so when the dictionary is turned into a \u0027string\u0027 you will not necessarily get the same order of elements.\n\nIn this case look at the order of \u0027Container-Foo\u0027 and \u0027Container-Bar\u0027 to see where they differ.\n\nIf you look at the Kilo version of the test that is failing, they are splitting the output and then sorting it before comparing it to get around this problem.\n\nhttps://github.com/openstack/swift/blob/master/test/unit/cli/test_info.py#L178","accounts_in_message":[],"_revision_number":1},{"id":"432089a3e93a237da613c7a8d81c29f97cba5d2c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-21 03:56:22.000000000","message":"Patch Set 1:\n\nAh, so probably the python hashseed issue that everyone had to deal with at some point, but I guess if stable branches aren\u0027t really a thing for swift, it\u0027s still a thing here?\n\nYou could totally cop out and push a separate change before this that sets PYTHONHASHSEED\u003d0 in tox.ini like nova has:\n\nhttps://github.com/openstack/nova/blob/master/tox.ini#L15\n\nI\u0027m guessing you\u0027d have to track that with a separate bug and fix it on stable/juno first and then backport to stable/icehouse?  I don\u0027t know what state swift is in, so it\u0027s a better question for John.","accounts_in_message":[],"_revision_number":1},{"id":"1c6793bdc309bd09c51d315d4717c973af1332d1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-21 12:27:09.000000000","message":"Patch Set 1:\n\nInteresting.  I was working on this yesterday and couldn\u0027t recreate using RHEL or Ubuntu locally.\n\nI had looked at the output but missed the ordering difference.  Thanks for catching that Jim!  I will open a bug for that and work with John to get the solution they want for Swift.","accounts_in_message":[],"_revision_number":1},{"id":"1de831fa156875d2e3fe71dd9e73b948455f591d","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-10-22 10:13:59.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"c797c20ea386ee6501eb2a19e4ab540aa4a83f9b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-22 18:36:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"855b93d63436045c39ab070bd8be558b61234f8c","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-22 18:37:51.000000000","message":"Patch Set 2:\n\nThis is now dependent upon https://review.openstack.org/#/c/130306 .   That patch should resolve the problem we have been having in the gate.","accounts_in_message":[],"_revision_number":2},{"id":"7f15c3a5fa113d9349002e086ab9f34bdac9d3b8","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2014-10-22 18:43:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* cluster-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/126371/619/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"54e9e24b274cb6c0fd8ce94c8938cb0260da11c8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2014-10-22 18:48:29.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm for stable","accounts_in_message":[],"_revision_number":2},{"id":"d6bedefa25bd14377a3a8b39c3e85503b24b547a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-22 19:35:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-pep8 http://logs.openstack.org/71/126371/2/check/gate-swift-pep8/7e1e791 : SUCCESS in 1m 01s\n- gate-swift-docs http://docs-draft.openstack.org/71/126371/2/check/gate-swift-docs/484fe86/doc/build/html/ : SUCCESS in 1m 26s\n- gate-swift-python27 http://logs.openstack.org/71/126371/2/check/gate-swift-python27/8f14d90 : SUCCESS in 2m 21s\n- check-tempest-dsvm-full http://logs.openstack.org/71/126371/2/check/check-tempest-dsvm-full/7a9e0be : SUCCESS in 53m 23s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/71/126371/2/check/check-tempest-dsvm-postgres-full/ddc1347 : SUCCESS in 49m 19s\n- check-tempest-dsvm-neutron http://logs.openstack.org/71/126371/2/check/check-tempest-dsvm-neutron/8eefa06 : SUCCESS in 55m 21s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/71/126371/2/check/check-tempest-dsvm-neutron-heat-slow/9fbfcaf : SUCCESS in 23m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/71/126371/2/check/gate-tempest-dsvm-large-ops/2980ea3 : SUCCESS in 20m 45s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/71/126371/2/check/gate-tempest-dsvm-neutron-large-ops/f7b5d2c : SUCCESS in 21m 31s\n- gate-swift-python26 http://logs.openstack.org/71/126371/2/check/gate-swift-python26/83eec71 : SUCCESS in 2m 16s\n- check-swift-dsvm-functional http://logs.openstack.org/71/126371/2/check/check-swift-dsvm-functional/0a24d5e : SUCCESS in 15m 09s\n- check-devstack-dsvm-cells http://logs.openstack.org/71/126371/2/check/check-devstack-dsvm-cells/213e454 : SUCCESS in 13m 16s","accounts_in_message":[],"_revision_number":2},{"id":"04b4b3cd211c4f3c4cb7aff7c000650b55e0c0dd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-22 20:44:25.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good now that you have https://review.openstack.org/#/c/130306/.","accounts_in_message":[],"_revision_number":2},{"id":"882ca17d6369697c6f5368614508c2f5b79c64e5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-22 21:10:02.000000000","message":"Patch Set 2: -Code-Review\n\nActually, this is a cherry pick directly from master, why isn\u0027t this a cherry pick from stable/juno first?","accounts_in_message":[],"_revision_number":2},{"id":"be5fcb58ce4cf26ecfff6f2030f8cdc0f7085a3d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-10-22 21:10:57.000000000","message":"Patch Set 2: Code-Review+1\n\nNevermind, duh, because the change on master was when master was Juno:\n\nhttps://review.openstack.org/#/c/125197/","accounts_in_message":[],"_revision_number":2},{"id":"a97ee70c3984cbebd7a5fd63d22ead1761b97dea","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2014-10-22 21:48:34.000000000","message":"Patch Set 2:\n\nOk, good, you answered your own question.  :-)","accounts_in_message":[],"_revision_number":2},{"id":"355341827a79448b54f87cea0f2698a72c986dbf","author":{"_account_id":6601,"name":"James Carey","email":"jecarey@yerac.com","username":"jecarey"},"date":"2014-10-22 22:46:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5ef9915e15f6166ee679d6b4a9dff734de87a415","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2014-10-24 13:41:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"60868322ddef02ca46d73069fbd0aada1f49dcde","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2014-10-24 14:42:23.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* cluster-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/126371/627/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"158d7e2976b9678f025f63adfcddf6d2fa62a675","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-28 18:08:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"50304c231579061ec7fc606a1dad00288efc9225","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-10-28 18:08:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da400a544cbf5332eff386097dd42baf39bd1397","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-28 22:02:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-swift-docs http://docs-draft.openstack.org/71/126371/2/gate/gate-swift-docs/f559197/doc/build/html/ : SUCCESS in 1m 14s\n- gate-swift-pep8 http://logs.openstack.org/71/126371/2/gate/gate-swift-pep8/0760e6c : SUCCESS in 1m 07s\n- gate-swift-python27 http://logs.openstack.org/71/126371/2/gate/gate-swift-python27/30475a9 : SUCCESS in 2m 14s\n- gate-tempest-dsvm-full http://logs.openstack.org/71/126371/2/gate/gate-tempest-dsvm-full/7d410c0 : SUCCESS in 49m 28s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/71/126371/2/gate/gate-tempest-dsvm-postgres-full/dbdb059 : SUCCESS in 50m 06s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/71/126371/2/gate/gate-tempest-dsvm-neutron/20922f3 : SUCCESS in 54m 22s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/71/126371/2/gate/gate-tempest-dsvm-neutron-heat-slow/71de601 : SUCCESS in 23m 23s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/71/126371/2/gate/gate-tempest-dsvm-large-ops/acde72f : SUCCESS in 23m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/71/126371/2/gate/gate-tempest-dsvm-neutron-large-ops/8d7ab66 : SUCCESS in 19m 54s\n- gate-swift-python26 http://logs.openstack.org/71/126371/2/gate/gate-swift-python26/facf8f5 : SUCCESS in 2m 15s\n- gate-swift-dsvm-functional http://logs.openstack.org/71/126371/2/gate/gate-swift-dsvm-functional/a8d92bc : SUCCESS in 15m 46s\n- gate-devstack-dsvm-cells http://logs.openstack.org/71/126371/2/gate/gate-devstack-dsvm-cells/b042e59 : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":2},{"id":"084a0a8627584844265891ee0552c02c15dce317","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-28 22:02:08.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fbe00380b0617476d84d2fde394c6ad347689aac","revisions":{"9fc6ce86ab0240e882ce49614dae9b7100a412a7":{"kind":"REWORK","_number":1,"created":"2014-10-06 17:46:09.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/71/126371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/126371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/126371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/126371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/126371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/126371/1"}}},"commit":{"parents":[{"commit":"b223322ed1ef44f61490f820240aa01f1047ae2e","subject":"properly quote www-authenticate header value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b223322ed1ef44f61490f820240aa01f1047ae2e"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@us.ibm.com","date":"2014-09-30 20:08:59.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@us.ibm.com","date":"2014-10-06 17:32:44.000000000","tz":-300},"subject":"Ensure sys.exit called in fork_child after exception","message":"Ensure sys.exit called in fork_child after exception\n\nCurrently, the fork_child() function in auditor.py does not\nhandle the case where run_audit() encounters an exception\nproperly.\n\nA simple case is where the /srv directory is set\nwith permissions such that the \u0027swift\u0027 user cannot access it.\nSuch a situation causes a os.listdir() to return an OSError\nexception.  When this happens the fork_child() process does not\nrun to completion and sys.exit() is not executed.  The process\nthat was forked off continues to run as a result.  Execution goes\nback up to the audit_loop function which restarts the whole process.  The\nend result is an increasing number of processes on the system\nuntil the parent is terminated.  This can quickly exhaust the\nprocess descriptors on a system.\n\nThis change wraps run_audit() in a try block and adds an\nexception handler that prints what exception was encountered.\nThe sys.exit() was moved to a finally: block so that it will\nalways be run, avoiding the creation of zombies.\n\nChange-Id: I89d7cd27112445893852e62df857c3d5262c27b3\nCloses-bug: 1375348\n(cherry picked from commit 301a96f664d58b4ccad8e3cbf5d5a889cc76790f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fc6ce86ab0240e882ce49614dae9b7100a412a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fc6ce86ab0240e882ce49614dae9b7100a412a7"}]},"branch":"refs/heads/stable/icehouse"},"fbe00380b0617476d84d2fde394c6ad347689aac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2014-10-22 18:36:38.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/71/126371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/126371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/126371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/126371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/126371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/126371/2"}}},"commit":{"parents":[{"commit":"e9d57c726bb8ffef20203bfaf2060d8c0cbb2cf8","subject":"Fix testcase test_print_db_info_metadata()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9d57c726bb8ffef20203bfaf2060d8c0cbb2cf8"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@us.ibm.com","date":"2014-09-30 20:08:59.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@us.ibm.com","date":"2014-10-22 18:26:11.000000000","tz":-300},"subject":"Ensure sys.exit called in fork_child after exception","message":"Ensure sys.exit called in fork_child after exception\n\nCurrently, the fork_child() function in auditor.py does not\nhandle the case where run_audit() encounters an exception\nproperly.\n\nA simple case is where the /srv directory is set\nwith permissions such that the \u0027swift\u0027 user cannot access it.\nSuch a situation causes a os.listdir() to return an OSError\nexception.  When this happens the fork_child() process does not\nrun to completion and sys.exit() is not executed.  The process\nthat was forked off continues to run as a result.  Execution goes\nback up to the audit_loop function which restarts the whole process.  The\nend result is an increasing number of processes on the system\nuntil the parent is terminated.  This can quickly exhaust the\nprocess descriptors on a system.\n\nThis change wraps run_audit() in a try block and adds an\nexception handler that prints what exception was encountered.\nThe sys.exit() was moved to a finally: block so that it will\nalways be run, avoiding the creation of zombies.\n\nChange-Id: I89d7cd27112445893852e62df857c3d5262c27b3\nCloses-bug: 1375348\n(cherry picked from commit 301a96f664d58b4ccad8e3cbf5d5a889cc76790f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbe00380b0617476d84d2fde394c6ad347689aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbe00380b0617476d84d2fde394c6ad347689aac"}]},"branch":"refs/heads/stable/icehouse"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
