)]}'
{"id":"openstack%2Fswift~207656","triplet_id":"openstack%2Fswift~master~I9a1b57aa15663380c45cf783afc8212ab4ffbace","project":"openstack/swift","branch":"master","topic":"bug/1479972","hashtags":[],"change_id":"I9a1b57aa15663380c45cf783afc8212ab4ffbace","subject":"Add a probetest for HUP/reload","status":"MERGED","created":"2015-07-30 22:52:02.000000000","updated":"2015-08-18 18:15:16.000000000","submitted":"2015-08-18 18:15:14.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ea89ef1fa2c0c820f6bccc0b2569495ba57ef84","_number":207656,"virtual_id_number":207656,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2015-08-18 18:15:14.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-07-30 22:58:38.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"date":"2015-07-30 23:07:46.000000000","_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-08-12 19:24:41.000000000","_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2015-08-18 14:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-08-17 18:18:47.000000000","_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2015-08-18 14:52:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-07-30 23:07:46.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-12 19:24:41.000000000","updated_by":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"reviewer":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"state":"REVIEWER"},{"updated":"2015-08-18 14:52:46.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2015-08-18 18:15:14.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"27d689badf44b6d5348da6426f2cfc8f4e10acac","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2015-07-30 22:52:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaf2543d789de0a88afa2b8221105e2efc6dfb59","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2015-07-30 22:58:38.000000000","message":"Patch Set 1:\n\nI thought about trolling around in /proc and finding the worker pids and getting all brittle and down in the weeds - but in this case the basic test that validates the behavior we want seems to fail reliably with out the fix in the dependent patch and works now:\n\n    reload freshens the pids while clients are still connected\n\nSo I *think* even this as simple as it is worth might be worth merging on it\u0027s own (?) - even if folks want to advocate we should add more tests/checks in a follow up patch.  My goal was mainly to give us a base we can extend to what we test/check as we review future changes to the process management code.\n\nyou can run it like this:\n\n  nosetests swift/test/probe/test_wsgi_servers.py","accounts_in_message":[],"_revision_number":1},{"id":"4c81526483a45467d3ec044f69892d277be403d9","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2015-07-30 23:07:46.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/207656/1940/ : SUCCESS\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/207656/598/ssbench : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ffede769dfec2c3daf3891d0beb893da05859e0d","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2015-08-12 19:24:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"caa11136abbb1791c470987527a858de6c0c948e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-12 20:30:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-pep8 http://logs.openstack.org/56/207656/1/check/gate-swift-pep8/e6f8cc5/ : SUCCESS in 2m 08s\n- gate-swift-docs http://docs-draft.openstack.org/56/207656/1/check/gate-swift-docs/1597c61//doc/build/html/ : SUCCESS in 3m 10s\n- gate-swift-python27 http://logs.openstack.org/56/207656/1/check/gate-swift-python27/eab21e7/ : SUCCESS in 6m 23s\n- gate-swift-tox-func http://logs.openstack.org/56/207656/1/check/gate-swift-tox-func/bcc5cd9/ : SUCCESS in 6m 30s (non-voting)\n- gate-swift-python34 http://logs.openstack.org/56/207656/1/check/gate-swift-python34/edc2c0f/ : FAILURE in 3m 08s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/56/207656/1/check/gate-tempest-dsvm-full/9dd3249/ : SUCCESS in 51m 43s\n- gate-grenade-dsvm http://logs.openstack.org/56/207656/1/check/gate-grenade-dsvm/37a9a05/ : SUCCESS in 47m 46s\n- gate-swift-dsvm-functional http://logs.openstack.org/56/207656/1/check/gate-swift-dsvm-functional/0e00861/ : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b5a89b89d530ac56dd4607a11f898143af9d8fd5","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2015-08-17 18:18:47.000000000","message":"Patch Set 1: Code-Review+2\n\nWorks for me.","accounts_in_message":[],"_revision_number":1},{"id":"d8a95c46b9e842972513bcc5e75bfc1eb1f23c1a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2015-08-18 14:52:46.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2\n\nWorking for me too. Nice inital framework too, thanks Clay.","accounts_in_message":[],"_revision_number":1},{"id":"ec5bee4e498beb61c2a18c3afa5271209f0bd450","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-18 14:52:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"c44cc4b6ad73dee73829b2804bf79d745c4848b1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-18 18:15:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-swift-docs http://docs-draft.openstack.org/56/207656/1/gate/gate-swift-docs/1f9e414//doc/build/html/ : SUCCESS in 3m 34s\n- gate-swift-pep8 http://logs.openstack.org/56/207656/1/gate/gate-swift-pep8/4489a4e/ : SUCCESS in 2m 09s\n- gate-swift-python27 http://logs.openstack.org/56/207656/1/gate/gate-swift-python27/3efb9ae/ : SUCCESS in 4m 31s\n- gate-tempest-dsvm-full http://logs.openstack.org/56/207656/1/gate/gate-tempest-dsvm-full/b866166/ : SUCCESS in 42m 25s\n- gate-grenade-dsvm http://logs.openstack.org/56/207656/1/gate/gate-grenade-dsvm/8c93f4e/ : SUCCESS in 44m 55s\n- gate-swift-dsvm-functional http://logs.openstack.org/56/207656/1/gate/gate-swift-dsvm-functional/429b24e/ : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":1},{"id":"197139e3830ed383abb73b88a968913c12fd2652","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-18 18:15:16.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"768d7ab074807175471d589de96d091e3239052e","revisions":{"768d7ab074807175471d589de96d091e3239052e":{"kind":"REWORK","_number":1,"created":"2015-07-30 22:52:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/56/207656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/207656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/207656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/207656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/207656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/207656/1"}}},"commit":{"parents":[{"commit":"207dd9b49d7d53a9faa4849af2c40bb875416fce","subject":"Fix regression in WSGI server SIGHUP behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/207dd9b49d7d53a9faa4849af2c40bb875416fce"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2015-07-30 22:30:35.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2015-07-30 22:49:23.000000000","tz":-420},"subject":"Add a probetest for HUP/reload","message":"Add a probetest for HUP/reload\n\nThis would have been enough to catch the regression, and we can extend\nthem as we work on any future ehancements to our process management.\n\nChange-Id: I9a1b57aa15663380c45cf783afc8212ab4ffbace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/768d7ab074807175471d589de96d091e3239052e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/768d7ab074807175471d589de96d091e3239052e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"207dd9b49d7d53a9faa4849af2c40bb875416fce","is_merged_in_target_branch":true,"change_id":"I95283b8b7cfc2998ab5813e0ad3ca1fa231696c8","change_number":207637,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
