)]}'
{"id":"openstack%2Fswift~209329","triplet_id":"openstack%2Fswift~master~I5145c3fcad1535c6bae5eb258ebb09828cbc3b84","project":"openstack/swift","branch":"master","topic":"bp/changing-policies","hashtags":[],"change_id":"I5145c3fcad1535c6bae5eb258ebb09828cbc3b84","subject":"WIP: Changing Policies","status":"NEW","created":"2015-08-05 03:15:24.000000000","updated":"2017-12-18 03:17:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"45e4e701cc5252f456efd9991c5a0d6ada80ef7b","_number":209329,"virtual_id_number":209329,"owner":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"},{"date":"2017-03-16 00:25:34.000000000","_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2017-03-16 01:03:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-05 07:56:58.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2016-03-03 15:18:50.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2016-03-09 06:44:33.000000000","updated_by":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"reviewer":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"state":"REVIEWER"},{"updated":"2016-07-20 16:19:14.000000000","updated_by":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"reviewer":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"state":"REVIEWER"},{"updated":"2017-03-16 00:25:34.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-03-16 01:03:13.000000000","updated_by":{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"},"reviewer":{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"},"state":"REVIEWER"}],"messages":[{"id":"b2dd395b846e55f8463386dd5d13f8340b26d829","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2015-08-05 03:15:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b18d64e00c724403d6c0e7364e8a39b575bb7bd9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-05 04:09:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/1/check/gate-swift-pep8/c6764a2/ : FAILURE in 1m 49s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/1/check/gate-swift-docs/c783e35//doc/build/html/ : SUCCESS in 3m 42s\n- gate-swift-python27 http://logs.openstack.org/29/209329/1/check/gate-swift-python27/52a0c9a/ : SUCCESS in 4m 53s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/1/check/gate-swift-tox-func/c0775b8/ : SUCCESS in 6m 53s (non-voting)\n- gate-swift-python34 http://logs.openstack.org/29/209329/1/check/gate-swift-python34/69cebc0/ : FAILURE in 3m 16s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/1/check/gate-tempest-dsvm-full/5147d78/ : SUCCESS in 51m 53s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/1/check/gate-grenade-dsvm/8116bf2/ : SUCCESS in 48m 12s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/1/check/gate-swift-dsvm-functional/5e66900/ : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8031633fd62683dfc8c62692067d705e280bb3be","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2015-08-05 04:22:45.000000000","message":"Patch Set 1:\n\nBuild succeeded \n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/662/ssbench : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2020/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/65/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"828c067836fd13e4c0328fbcea4c246615d43671","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2015-08-05 07:56:58.000000000","message":"Patch Set 1:\n\n(12 comments)\n\nWow!  Great progress!  Definitely lots to think about for next week!","accounts_in_message":[],"_revision_number":1},{"id":"7cf7b7eca212a2d9bc5eca9d129e9c0f30230e39","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2015-08-10 21:27:40.000000000","message":"Patch Set 1:\n\n(11 comments)\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"fa745ef899eefaefd7d9931d0a8b3d54cb9f8639","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2015-10-20 21:09:08.000000000","message":"Abandoned\n\nAbanoning based on the lack of activity since the last negative review. If you want to continue working on this, please reopen this patch.","accounts_in_message":[],"_revision_number":1},{"id":"af5ff78fb0e517638892d896a2120d440b904c66","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2015-10-21 11:53:20.000000000","message":"Restored\n\nThis patch is work in progress. I will upload the patch soon.","accounts_in_message":[],"_revision_number":1},{"id":"eb2c68d5391b7efc8d9f85cf3f6c72fc8724a200","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2015-10-21 11:53:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"647e21134b6c555b9d252f154437c12fb81d5dd1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-21 13:05:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/2/check/gate-swift-pep8/06ae63f/ : FAILURE in 2m 01s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/2/check/gate-swift-docs/50721d6//doc/build/html/ : SUCCESS in 2m 02s\n- gate-swift-python27 http://logs.openstack.org/29/209329/2/check/gate-swift-python27/8a52b27/ : FAILURE in 4m 25s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/2/check/gate-swift-tox-func/e0de097/ : FAILURE in 40m 09s\n- gate-swift-python34 http://logs.openstack.org/29/209329/2/check/gate-swift-python34/143c93d/ : FAILURE in 2m 27s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/2/check/gate-tempest-dsvm-full/c7753dc/ : FAILURE in 1h 09m 41s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/2/check/gate-grenade-dsvm/3a5dc27/ : FAILURE in 48m 20s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/2/check/gate-swift-dsvm-functional/83ac55c/ : FAILURE in 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9abe86a206b8f7d9ae2094501921b804b31ae2e1","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2015-10-21 14:15:36.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2844/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/888/ : FAILURE\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/1485/ssbench : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/601/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e9c35c70f0a2d8fbbb4a95f25fc389840f5e0040","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2015-10-23 09:12:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4326055440c6d051a27fc05d8c399db189b77e0d","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2015-10-23 10:01:50.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2870/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/914/ : SUCCESS\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/1511/ssbench : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/627/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"7c4c7ab1957c52864bcab3cd64d9b04bf6051933","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-23 10:10:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/3/check/gate-swift-pep8/6d4175a/ : FAILURE in 2m 16s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/3/check/gate-swift-docs/09917fd//doc/build/html/ : SUCCESS in 3m 01s\n- gate-swift-python27 http://logs.openstack.org/29/209329/3/check/gate-swift-python27/4753d39/ : FAILURE in 4m 33s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/3/check/gate-swift-tox-func/3ab1440/ : SUCCESS in 5m 05s\n- gate-swift-python34 http://logs.openstack.org/29/209329/3/check/gate-swift-python34/3fcbd90/ : FAILURE in 3m 16s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/3/check/gate-tempest-dsvm-full/8393796/ : SUCCESS in 56m 39s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/3/check/gate-grenade-dsvm/8277eae/ : SUCCESS in 53m 00s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/3/check/gate-swift-dsvm-functional/2544cc1/ : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":3},{"id":"e64bd9b4e5fd105c0ccc9cdf4622d1d0bf19139c","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2015-10-26 06:56:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"935eac91880495c7ab2b7cc260984035a0006b12","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-26 07:55:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/4/check/gate-swift-pep8/5180449/ : SUCCESS in 2m 08s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/4/check/gate-swift-docs/6c08bfd//doc/build/html/ : SUCCESS in 3m 14s\n- gate-swift-python27 http://logs.openstack.org/29/209329/4/check/gate-swift-python27/8f71ab1/ : SUCCESS in 5m 12s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/4/check/gate-swift-tox-func/6a039a8/ : SUCCESS in 6m 05s\n- gate-swift-python34 http://logs.openstack.org/29/209329/4/check/gate-swift-python34/3d346dc/ : FAILURE in 2m 55s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/4/check/gate-tempest-dsvm-full/10c6dec/ : SUCCESS in 49m 28s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/4/check/gate-grenade-dsvm/f7716dc/ : SUCCESS in 42m 01s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/4/check/gate-swift-dsvm-functional/887dbfa/ : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":4},{"id":"21a4e740c2c68d346311eb357a60a4a60cfbf490","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2015-10-26 07:55:16.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2899/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/943/ : SUCCESS\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/1540/ssbench : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/656/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"085b2fa6675a4f02498a49f031c338e22725dd73","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-24 18:29:32.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"2ea6079563770d6508d15f1b9999f2643d909d36","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2015-11-26 05:46:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ec4e5795e734c7de35f166136ae2c123a5e0e6c3","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2015-11-26 06:33:31.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3202/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/1246/ : FAILURE\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/1843/ssbench : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/963/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"dc103a55459f94c0c085b3c986552e2d1952efbe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-26 06:50:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/5/check/gate-swift-pep8/d70dae2/ : FAILURE in 1m 50s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/5/check/gate-swift-docs/671a608//doc/build/html/ : SUCCESS in 3m 01s\n- gate-swift-python27 http://logs.openstack.org/29/209329/5/check/gate-swift-python27/a095a1e/ : FAILURE in 4m 54s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/5/check/gate-swift-tox-func/baffa86/ : FAILURE in 2m 00s\n- gate-swift-python34 http://logs.openstack.org/29/209329/5/check/gate-swift-python34/d7b1a13/ : FAILURE in 2m 37s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/5/check/gate-tempest-dsvm-full/1dc5d9f/ : FAILURE in 21m 02s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/5/check/gate-grenade-dsvm/de422dd/ : SUCCESS in 53m 27s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/5/check/gate-swift-dsvm-functional/dec4bcb/ : FAILURE in 27m 50s","accounts_in_message":[],"_revision_number":5},{"id":"dba5d3b6d8161742a097b760d6868b49b773a96f","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-02-22 22:48:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"450a358e2ded8b28adf206507a201087d56991a0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-22 23:41:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/6/check/gate-swift-pep8/933ffa7/ : FAILURE in 58s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/6/check/gate-swift-docs/ffea0a6//doc/build/html/ : SUCCESS in 2m 07s\n- gate-swift-python27 http://logs.openstack.org/29/209329/6/check/gate-swift-python27/1181b77/ : SUCCESS in 4m 04s\n- gate-swift-python34 http://logs.openstack.org/29/209329/6/check/gate-swift-python34/67e039f/ : SUCCESS in 2m 12s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/6/check/gate-swift-tox-func/fb9e807/ : SUCCESS in 4m 16s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/6/check/gate-swift-tox-func-in-process-fast-post/32b8c0d/ : SUCCESS in 4m 02s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/6/check/gate-tempest-dsvm-full/5699f5e/ : SUCCESS in 35m 44s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/6/check/gate-grenade-dsvm/0e912eb/ : SUCCESS in 48m 12s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/6/check/gate-swift-dsvm-functional/bf672cf/ : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":6},{"id":"b9d4082b2d89d3dbbf192dc8f5c963d56b24803a","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-02-23 00:22:21.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2329/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2050/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/4287/ : SUCCESS\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2925/ssbench : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"b6213bbabfa7f1d22d000da23f002efeee018db2","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-02-23 22:19:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8364ee8ccb11435d423a82f3655ced9c9c5eff4c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-24 00:10:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/7/check/gate-swift-pep8/6c8e9c0/ : SUCCESS in 1m 00s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/7/check/gate-swift-docs/41f36ea//doc/build/html/ : SUCCESS in 1m 52s\n- gate-swift-python27 http://logs.openstack.org/29/209329/7/check/gate-swift-python27/0d03fd7/ : SUCCESS in 3m 31s\n- gate-swift-python34 http://logs.openstack.org/29/209329/7/check/gate-swift-python34/c28cf7e/ : SUCCESS in 2m 36s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/7/check/gate-swift-tox-func/d80fe13/ : SUCCESS in 3m 02s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/7/check/gate-swift-tox-func-in-process-fast-post/7904a88/ : SUCCESS in 2m 43s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/7/check/gate-tempest-dsvm-full/0a435f9/ : SUCCESS in 35m 16s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/7/check/gate-grenade-dsvm/e0e7fec/ : SUCCESS in 45m 30s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/7/check/gate-swift-dsvm-functional/71222aa/ : SUCCESS in 24m 42s","accounts_in_message":[],"_revision_number":7},{"id":"01b7631b1340b421af235fcb41e4e284b511a59f","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-02-24 00:55:37.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2343/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2064/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/4301/ : SUCCESS\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2939/ssbench : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"b4184b4bb1c91078abf41e8ae77db034761fe4d9","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2016-03-03 13:02:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f3f8f5530307bbbf1c727201fb37e36d29129171","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-03-03 16:37:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4fdb8b280f8ceef58a790f28c43c2336ba8d3b0a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-03 17:29:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-pep8 http://logs.openstack.org/29/209329/8/check/gate-swift-pep8/b7e059c/ : SUCCESS in 1m 29s\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/8/check/gate-swift-docs/df4a1bc//doc/build/html/ : SUCCESS in 2m 02s\n- gate-swift-python27 http://logs.openstack.org/29/209329/8/check/gate-swift-python27/2909049/ : SUCCESS in 3m 37s\n- gate-swift-python34 http://logs.openstack.org/29/209329/8/check/gate-swift-python34/faf6318/ : SUCCESS in 2m 17s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/8/check/gate-swift-tox-func/e8b390a/ : SUCCESS in 3m 10s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/8/check/gate-swift-tox-func-in-process-fast-post/93f0caa/ : SUCCESS in 4m 53s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/8/check/gate-tempest-dsvm-full/d1ed327/ : SUCCESS in 42m 39s\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/8/check/gate-grenade-dsvm/73c59dd/ : SUCCESS in 44m 45s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/8/check/gate-swift-dsvm-functional/30dfbe7/ : SUCCESS in 25m 11s","accounts_in_message":[],"_revision_number":8},{"id":"39ee94e948da884fa26190113b8499a5970db4ff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-03 19:09:26.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"211b7ec6a28a133575d03c14bf749f07e2e43beb","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-03-04 02:12:49.000000000","message":"Patch Set 8:\n\nBuild succeeded \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/4429/ : SUCCESS\n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3067/ssbench : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2474/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/2216/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"4d123c4659f68c397ee1e63dcb587c49a91f0eb1","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2016-03-09 06:44:33.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1e1674d81081f6b6c83b1817e7cfd666b5212e01","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2016-03-09 11:25:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"540e226a7d342a255a7b120235d6d12042a84485","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-04-06 22:47:49.000000000","message":"Patch Set 8:\n\n(2 comments)\n\nThanks Kazuhiro!","accounts_in_message":[],"_revision_number":8},{"id":"6e84f0bd2a74201c04529f1851083309e7789c53","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2016-04-07 01:28:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a37645be9c212a4069629506e065f820370e27fb","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2016-04-07 02:49:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c7786a65828c74f4a68dba61f4774c00e82483a3","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-04-13 23:51:57.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6faf5e0837d69698b0a5b81560f5d3f1bdb20d8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-14 01:04:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/9/check/gate-swift-docs/936b5ba//doc/build/html/ : SUCCESS in 5m 05s\n- gate-swift-pep8 http://logs.openstack.org/29/209329/9/check/gate-swift-pep8/9108763/ : SUCCESS in 2m 16s\n- gate-swift-python27 http://logs.openstack.org/29/209329/9/check/gate-swift-python27/29517f7/ : FAILURE in 5m 58s\n- gate-swift-python34 http://logs.openstack.org/29/209329/9/check/gate-swift-python34/3434b2f/ : SUCCESS in 3m 23s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/9/check/gate-swift-tox-func/d3133b6/ : SUCCESS in 6m 40s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/9/check/gate-swift-tox-func-in-process-fast-post/5f52931/ : SUCCESS in 7m 24s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/9/check/gate-tempest-dsvm-full/ac0ab7a/ : SUCCESS in 37m 49s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/29/209329/9/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/05210b1/ : SUCCESS in 1h 10m 43s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/9/check/gate-grenade-dsvm/a991e62/ : SUCCESS in 39m 34s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/9/check/gate-swift-dsvm-functional/2cf1272/ : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":9},{"id":"280c3b20838333631e203ab792a1d51f9f646037","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-04-21 02:28:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9b8b93c83f8810449864dc55aa5efd119925ba6a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-21 03:27:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/10/check/gate-swift-docs/31859a6//doc/build/html/ : SUCCESS in 2m 04s\n- gate-swift-pep8 http://logs.openstack.org/29/209329/10/check/gate-swift-pep8/6ee4f76/ : FAILURE in 1m 32s\n- gate-swift-python27 http://logs.openstack.org/29/209329/10/check/gate-swift-python27/0b1d99f/ : FAILURE in 5m 33s\n- gate-swift-python34 http://logs.openstack.org/29/209329/10/check/gate-swift-python34/27bf563/ : SUCCESS in 2m 13s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/10/check/gate-swift-tox-func/e24a99a/ : SUCCESS in 4m 12s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/10/check/gate-swift-tox-func-in-process-fast-post/1222cf1/ : SUCCESS in 3m 47s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/10/check/gate-tempest-dsvm-full/78ec647/ : SUCCESS in 40m 14s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/29/209329/10/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/c59447f/ : SUCCESS in 57m 20s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/10/check/gate-grenade-dsvm/3354fc6/ : SUCCESS in 37m 30s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/10/check/gate-swift-dsvm-functional/d7feb2e/ : SUCCESS in 22m 26s","accounts_in_message":[],"_revision_number":10},{"id":"98a509f9b88e1736eb816d0af33ea089c0fbad66","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-05-26 00:53:12.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"dd135ceb309ef21be0fec07a924e810e7e346b2e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-26 02:37:58.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/11/check/gate-swift-docs/2b75ba5//doc/build/html/ : SUCCESS in 1m 09s\n- gate-swift-pep8 http://logs.openstack.org/29/209329/11/check/gate-swift-pep8/bc66e6f/ : FAILURE in 30m 01s\n- gate-swift-python27 http://logs.openstack.org/29/209329/11/check/gate-swift-python27/9f473ce/ : FAILURE in 2m 27s\n- gate-swift-python34 http://logs.openstack.org/29/209329/11/check/gate-swift-python34/3487f38/ : SUCCESS in 1m 40s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/11/check/gate-swift-tox-func/c266ed3/ : SUCCESS in 3m 28s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/11/check/gate-swift-tox-func-in-process-fast-post/19812c2/ : SUCCESS in 2m 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/11/check/gate-tempest-dsvm-full/440a8fb/ : SUCCESS in 40m 29s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/29/209329/11/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/7f43ab2/ : SUCCESS in 51m 39s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/11/check/gate-grenade-dsvm/38a0c8a/ : SUCCESS in 38m 51s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/11/check/gate-swift-dsvm-functional/6560ef7/ : FAILURE in 1h 40m 03s\n- gate-swift-tox-bandit http://logs.openstack.org/29/209329/11/check/gate-swift-tox-bandit/f38de53/ : SUCCESS in 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d7fbbf42aaecfbc30ada15c3bf6b84ebef6153bf","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-05-26 09:57:59.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3900/ssbench : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3306/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3086/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5262/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"6873154f543ad3c0792bb93a04c0a35bae9c779a","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-06-15 18:50:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3f0b0a44a84573d9ccdc7383c2f230f85a2a619c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-15 20:17:35.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/12/check/gate-swift-docs/975e19f//doc/build/html/ : SUCCESS in 3m 02s\n- gate-swift-pep8 http://logs.openstack.org/29/209329/12/check/gate-swift-pep8/d12aa09/ : SUCCESS in 1m 11s\n- gate-swift-python27 http://logs.openstack.org/29/209329/12/check/gate-swift-python27/2d54069/ : FAILURE in 3m 03s\n- gate-swift-python34 http://logs.openstack.org/29/209329/12/check/gate-swift-python34/3ab78d0/ : SUCCESS in 1m 12s\n- gate-swift-tox-func http://logs.openstack.org/29/209329/12/check/gate-swift-tox-func/64d27a7/ : SUCCESS in 2m 22s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/12/check/gate-swift-tox-func-in-process-fast-post/adc3b98/ : SUCCESS in 4m 39s\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/12/check/gate-tempest-dsvm-full/93d682d/ : SUCCESS in 52m 09s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/29/209329/12/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/be3a57b/ : SUCCESS in 1h 11m 56s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/12/check/gate-grenade-dsvm/7aaa817/ : SUCCESS in 31m 13s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/12/check/gate-swift-dsvm-functional/84be3ad/ : SUCCESS in 26m 06s\n- gate-swift-tox-bandit http://logs.openstack.org/29/209329/12/check/gate-swift-tox-bandit/d1aca6d/ : SUCCESS in 1m 14s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f77358ead2abeb09212eaae10d81aa3047fa9f75","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-06-15 21:05:45.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/4029/ssbench : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3336/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5391/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3435/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"3c2c71d768ba687c725669081c57801ce5b23277","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-07-12 21:42:36.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d8c869030e5a8b2416004b47a8551cb92d2e5a2a","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-07-12 22:46:01.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/4370/ssbench : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3677/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5732/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/3776/ : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"a8332354783e7b8354e317eb411eb0256e663778","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-12 22:54:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs http://docs-draft.openstack.org/29/209329/13/check/gate-swift-docs/0b9ad40//doc/build/html/ : SUCCESS in 3m 15s\n- gate-swift-pep8 http://logs.openstack.org/29/209329/13/check/gate-swift-pep8/ad625ce/ : SUCCESS in 2m 09s\n- gate-swift-python27 http://logs.openstack.org/29/209329/13/check/gate-swift-python27/8b8d192/ : FAILURE in 5m 29s\n- gate-swift-python34 http://logs.openstack.org/29/209329/13/check/gate-swift-python34/bdb949e/ : SUCCESS in 2m 25s\n- gate-swift-python35-nv http://logs.openstack.org/29/209329/13/check/gate-swift-python35-nv/4c16632/ : FAILURE in 4m 07s (non-voting)\n- gate-swift-tox-func http://logs.openstack.org/29/209329/13/check/gate-swift-tox-func/9a525be/ : SUCCESS in 4m 45s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/13/check/gate-swift-tox-func-in-process-fast-post/0271be9/ : SUCCESS in 7m 28s\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/13/check/gate-tempest-dsvm-full/4b0a8b5/ : SUCCESS in 47m 18s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/29/209329/13/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/2b039d3/ : SUCCESS in 1h 08m 42s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/13/check/gate-grenade-dsvm/8ceffdb/ : SUCCESS in 40m 58s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/13/check/gate-swift-dsvm-functional/e10a2dd/ : SUCCESS in 26m 26s\n- gate-swift-tox-bandit http://logs.openstack.org/29/209329/13/check/gate-swift-tox-bandit/15cd4d0/ : SUCCESS in 1m 48s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-nv http://logs.openstack.org/29/209329/13/check/gate-swift-dsvm-functional-identity-v3-only-nv/74cd266/ : SUCCESS in 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d8e90f6b7564136f1331502a2a873e543eabb582","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2016-07-20 16:19:14.000000000","message":"Patch Set 13:\n\nFor the record, at the Swift mid-cycle in San Antonio, the following comments were made:\n\n- Consider using similar logic to container sync in the object   \n  transferrer. i.e., instead of only relocating objects using the\n  primary node, make all all nodes share the work and using two sync\n  points.\n\n  Unlike container sync, I don\u0027t think you need the sync points in\n  a database, but I\u0027d need to think about this a bit more.\n\n- We need some way to track progress (or more accuratly, non-\n  progress). Consider using recon in a similar way to the\n  replicators.\n\nYou have added a number of new user-visible meta data and non-visible sysmeta. Please update overview_policies.html with instructions on how to change the policy and include a description of the user-visible meta data. For the sysmeta, I suggest listing/documenting these in common/middleware/change_policy.py. This will make it easier to review this patch.","accounts_in_message":[],"_revision_number":13},{"id":"3beafd54e9d4f10f7dfbd43ab77b601d3bb506d2","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2016-07-20 16:22:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1b06d87fab8d5f6ba813c89f7e38d24810133024","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-07-20 17:55:29.000000000","message":"Patch Set 13:\n\n(1 comment)\n\nThank you for listing discussion results. I added other comments in hackathon.\n\n- More tests for conditional GET\n- Timing issue for updating metadata","accounts_in_message":[],"_revision_number":13},{"id":"1c0ccd525343b394bfa3ee75b572b6c7f5ffaf11","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-07-28 00:59:10.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"653f7e13a9befda2d5217999602144eac2dd5403","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-07-28 00:59:44.000000000","message":"Patch Set 14:\n\nUpdate DOCs and Fix some bugs","accounts_in_message":[],"_revision_number":14},{"id":"0304b845a4ed90722e6f960fbd1c18df8ac12a95","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-28 02:23:34.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/29/209329/14/check/gate-swift-docs-ubuntu-xenial/61898d0//doc/build/html/ : SUCCESS in 3m 39s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/29/209329/14/check/gate-swift-pep8-ubuntu-xenial/ae07d45/ : SUCCESS in 2m 39s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/29/209329/14/check/gate-swift-python27-ubuntu-xenial/60de7cf/ : SUCCESS in 6m 05s\n- gate-swift-python34 http://logs.openstack.org/29/209329/14/check/gate-swift-python34/cad478f/ : SUCCESS in 2m 11s\n- gate-swift-python35-nv http://logs.openstack.org/29/209329/14/check/gate-swift-python35-nv/b20d5fd/ : FAILURE in 4m 37s (non-voting)\n- gate-swift-tox-func http://logs.openstack.org/29/209329/14/check/gate-swift-tox-func/2e3f4fd/ : SUCCESS in 4m 37s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/29/209329/14/check/gate-swift-tox-func-in-process-fast-post/ba7f096/ : SUCCESS in 4m 09s\n- gate-tempest-dsvm-full http://logs.openstack.org/29/209329/14/check/gate-tempest-dsvm-full/4f882a6/ : SUCCESS in 53m 23s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/29/209329/14/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/94e869c/ : SUCCESS in 1h 12m 56s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/29/209329/14/check/gate-grenade-dsvm/5bcf31e/ : SUCCESS in 49m 19s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/14/check/gate-swift-dsvm-functional/699213f/ : SUCCESS in 34m 22s\n- gate-swift-tox-bandit http://logs.openstack.org/29/209329/14/check/gate-swift-tox-bandit/49b10d7/ : SUCCESS in 1m 18s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-nv http://logs.openstack.org/29/209329/14/check/gate-swift-dsvm-functional-identity-v3-only-nv/9ada212/ : SUCCESS in 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"cf4f5eecb23c1777cbe40a3fc8cab20814f3459b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-18 08:36:20.000000000","message":"Patch Set 14: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"02aa50c3a8cb99ea129044bf37312178a58e1475","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2016-08-18 23:09:49.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"1e9968de3259dad8a184a7f3034e92cb857f54aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-19 00:48:44.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/29/209329/15/check/gate-swift-docs-ubuntu-xenial/98213bc//doc/build/html/ : SUCCESS in 3m 52s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/29/209329/15/check/gate-swift-pep8-ubuntu-xenial/66964cb/ : SUCCESS in 2m 37s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/29/209329/15/check/gate-swift-python27-ubuntu-xenial/ef76600/ : SUCCESS in 5m 20s\n- gate-swift-python34 http://logs.openstack.org/29/209329/15/check/gate-swift-python34/acc852d/ : SUCCESS in 3m 18s\n- gate-swift-python35-nv http://logs.openstack.org/29/209329/15/check/gate-swift-python35-nv/a527297/ : SUCCESS in 3m 27s (non-voting)\n- gate-swift-tox-func-ubuntu-xenial http://logs.openstack.org/29/209329/15/check/gate-swift-tox-func-ubuntu-xenial/efcc173/ : SUCCESS in 5m 06s\n- gate-swift-tox-func-in-process-fast-post-ubuntu-xenial http://logs.openstack.org/29/209329/15/check/gate-swift-tox-func-in-process-fast-post-ubuntu-xenial/5657357/ : SUCCESS in 4m 57s\n- gate-swift-tox-func-in-process-encryption-ubuntu-xenial http://logs.openstack.org/29/209329/15/check/gate-swift-tox-func-in-process-encryption-ubuntu-xenial/132d165/ : SUCCESS in 5m 39s (non-voting)\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/29/209329/15/check/gate-tempest-dsvm-full-ubuntu-xenial/5debe94/ : SUCCESS in 53m 30s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/29/209329/15/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/35e401a/ : FAILURE in 32m 59s (non-voting)\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/29/209329/15/check/gate-grenade-dsvm-ubuntu-trusty/1215dd3/ : SUCCESS in 56m 01s\n- gate-swift-dsvm-functional http://logs.openstack.org/29/209329/15/check/gate-swift-dsvm-functional/1573e60/ : SUCCESS in 37m 29s\n- gate-swift-tox-bandit-ubuntu-xenial http://logs.openstack.org/29/209329/15/check/gate-swift-tox-bandit-ubuntu-xenial/fb16303/ : SUCCESS in 3m 12s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-nv http://logs.openstack.org/29/209329/15/check/gate-swift-dsvm-functional-identity-v3-only-nv/3d41930/ : SUCCESS in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ae3ef2aa87305cb89bbc180ee8826249853507f6","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-08-19 11:45:10.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/4212/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/4080/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/6039/ : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"bda19f05a207a606153fbf8a813fe38a8ab3564d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-13 04:03:09.000000000","message":"Patch Set 15: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"a67ffab76629f6abead8cae39468c72e5599c537","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2017-01-14 03:26:15.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"b55a5d05dc6e0f53546fb95b1c6663dff9a34246","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-14 07:15:24.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/29/209329/16/check/gate-swift-docs-ubuntu-xenial/344d976//doc/build/html/ : SUCCESS in 3m 25s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-pep8-ubuntu-xenial/34589d8/ : FAILURE in 3m 38s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-python27-ubuntu-xenial/46b2835/ : SUCCESS in 5m 43s\n- gate-swift-python35 http://logs.openstack.org/29/209329/16/check/gate-swift-python35/8338572/ : SUCCESS in 2m 46s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/0289027/ : SUCCESS in 6m 45s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/2122b48/ : SUCCESS in 3m 31s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/2f3e169/ : SUCCESS in 5m 17s\n- gate-swift-tox-xfs-tmp-func-fast-post-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-tox-xfs-tmp-func-fast-post-ubuntu-xenial/c5aed1e/ : SUCCESS in 5m 27s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/dc27a68/ : SUCCESS in 5m 31s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/29/209329/16/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/aa3e641/ : FAILURE in 1h 16m 36s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/29/209329/16/check/gate-swift-dsvm-functional-ubuntu-xenial/2fe3c74/ : SUCCESS in 39m 34s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/29/209329/16/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/9a2e565/ : SUCCESS in 40m 47s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ba8c3c128a3fbb0a8142ba96048e7d360a1505fc","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-01-14 10:56:42.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/7234/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5603/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5268/ : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"9da75dbd6157d1d835190b0ade630e712483532a","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2017-02-22 15:16:06.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"2259db3579ed3f0612a21fb99def1f6e0a33c7d8","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-02-22 16:00:18.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5909/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5574/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/7540/ : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"aeff5a1fd054c9a838a8749eb686b5d07229091f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-22 16:45:14.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/29/209329/17/check/gate-swift-docs-ubuntu-xenial/2905f41//doc/build/html/ : SUCCESS in 2m 02s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-pep8-ubuntu-xenial/bea41d8/ : SUCCESS in 2m 32s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-python27-ubuntu-xenial/bac4af8/ : SUCCESS in 6m 13s\n- gate-swift-python35 http://logs.openstack.org/29/209329/17/check/gate-swift-python35/e2bb298/ : SUCCESS in 2m 33s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/e01d2ed/ : SUCCESS in 3m 54s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/d9aa353/ : SUCCESS in 2m 29s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/7226c75/ : SUCCESS in 7m 11s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/041a4f3/ : SUCCESS in 4m 11s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/6af0a14/ : SUCCESS in 3m 21s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/29/209329/17/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/6854e64/ : SUCCESS in 1h 25m 53s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/29/209329/17/check/gate-swift-dsvm-functional-ubuntu-xenial/c0c312a/ : SUCCESS in 1h 01m 50s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/29/209329/17/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/45b43c4/ : SUCCESS in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0705c262000586eef988cc34c916c2eb4a35b154","author":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"date":"2017-03-15 23:43:56.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"73371826b6a75522896ccfe9b228634f56e5fe93","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-03-16 00:25:34.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/6112/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/7743/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/209329/5777/ : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"8d26078fd3df1a19a2a072aea796c29c219b71b1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-16 00:53:54.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/29/209329/18/check/gate-swift-docs-ubuntu-xenial/c736f26//doc/build/html/ : SUCCESS in 1m 50s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-pep8-ubuntu-xenial/ad14df3/ : SUCCESS in 4m 24s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-python27-ubuntu-xenial/83a98cd/ : SUCCESS in 6m 32s\n- gate-swift-python35 http://logs.openstack.org/29/209329/18/check/gate-swift-python35/691b801/ : SUCCESS in 1m 06s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8e66333/ : SUCCESS in 1h 08m 54s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/56764ac/ : SUCCESS in 36m 07s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/4d5cad3/ : SUCCESS in 6m 02s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/3e5ac14/ : SUCCESS in 2m 36s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/9138531/ : SUCCESS in 5m 48s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/72647f4/ : SUCCESS in 3m 24s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/e735cee/ : SUCCESS in 3m 16s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/0e551e5/ : SUCCESS in 3m 33s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/29/209329/18/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/92a3c92/ : SUCCESS in 1h 07m 03s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/29/209329/18/check/gate-swift-dsvm-functional-ubuntu-xenial/c414664/ : SUCCESS in 36m 28s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/29/209329/18/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/2b01b85/ : SUCCESS in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"63009ea741d73c9c8def5d2ebdb10884d7a5e5e9","author":{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"},"date":"2017-03-16 01:03:13.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"d1775528c9a12b674f27494db4b5acddbd971617","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-08 01:32:15.000000000","message":"Patch Set 18: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":18},{"id":"2b420e527ff53269d529bf573ea8f26dddaeaf29","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 03:17:01.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"4437aa061b9ced732fcab48485c045f487409c8a","revisions":{"723063c4b48787cd3a4fac5ae001f00c9be2e473":{"kind":"REWORK","_number":1,"created":"2015-08-05 03:15:24.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/1"}}},"commit":{"parents":[{"commit":"5b24b2249878e25ee702ee4237227c984036c6ec","subject":"Imported Translations from Transifex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b24b2249878e25ee702ee4237227c984036c6ec"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-08-05 02:56:24.000000000","tz":540},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThis is an early version of changing policies functions.\nThis patch ignores failure cases, includes dirty codes and lacks\nsome test cases. However, I would like to show my update to Swift\ndev team in advance of Swift Hackathon held next week.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* Enhance container-server\u0027s GET API and SQL select statements\n  to get a list of objects to be moved\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/723063c4b48787cd3a4fac5ae001f00c9be2e473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/723063c4b48787cd3a4fac5ae001f00c9be2e473"}]},"branch":"refs/heads/master"},"c948423678bccb76abfa033ff53fc6b763288597":{"kind":"REWORK","_number":2,"created":"2015-10-21 11:53:34.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/2"}}},"commit":{"parents":[{"commit":"e0e8d8a2553b396941c47504c2059a912609107a","subject":"Merge \"Port swift.common.utils.backward() to Python 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0e8d8a2553b396941c47504c2059a912609107a"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-10-21 11:25:06.000000000","tz":540},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nTHIS PATCH IS NOT READY FOR REVIEW\n--------------------------------------------------------------\nI do not test this patch and check pep8, but let me push this\npatch to notify \"this patch is active\" and to store working\ndata for backup purpose (actually, yesterday my pc went wrong\nand all my writing patches went away...)\nI will repush a patch soon after testing.\n--------------------------------------------------------------\n\nThis patch reflects the discussion results at Austin Hackathon.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c948423678bccb76abfa033ff53fc6b763288597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c948423678bccb76abfa033ff53fc6b763288597"}]},"branch":"refs/heads/master"},"08c3889ff1a747fa1186934538ce0fff81fa0d37":{"kind":"REWORK","_number":3,"created":"2015-10-23 09:12:17.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/3"}}},"commit":{"parents":[{"commit":"e0e8d8a2553b396941c47504c2059a912609107a","subject":"Merge \"Port swift.common.utils.backward() to Python 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0e8d8a2553b396941c47504c2059a912609107a"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-10-23 09:05:31.000000000","tz":540},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon is reflected to\nthis patch.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08c3889ff1a747fa1186934538ce0fff81fa0d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08c3889ff1a747fa1186934538ce0fff81fa0d37"}]},"branch":"refs/heads/master"},"33fa2bb6bc9ee6ac70ffb564f6a3b463fb284680":{"kind":"REWORK","_number":4,"created":"2015-10-26 06:56:32.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/4"}}},"commit":{"parents":[{"commit":"e0e8d8a2553b396941c47504c2059a912609107a","subject":"Merge \"Port swift.common.utils.backward() to Python 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0e8d8a2553b396941c47504c2059a912609107a"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-10-26 06:55:17.000000000","tz":540},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon is reflected to\nthis patch. Currently, tese cases do not cover all modifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33fa2bb6bc9ee6ac70ffb564f6a3b463fb284680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33fa2bb6bc9ee6ac70ffb564f6a3b463fb284680"}]},"branch":"refs/heads/master"},"c39fd8baaebf3791c2e7b3e49a8770e906a4fdc9":{"kind":"REWORK","_number":5,"created":"2015-11-26 05:46:24.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/5"}}},"commit":{"parents":[{"commit":"ea3a0d38c1b9335f602d89b215983384eca85565","subject":"Merge \"Include params in multipart message part content-type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea3a0d38c1b9335f602d89b215983384eca85565"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@lab.ntt.co.jp","date":"2015-11-26 05:45:42.000000000","tz":540},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\n------------------------------------------------------------------\nTHIS PATCH IS STILL A WORK IN PROGRESS AND IS NOT READY FOR REVIEW\n------------------------------------------------------------------\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c39fd8baaebf3791c2e7b3e49a8770e906a4fdc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c39fd8baaebf3791c2e7b3e49a8770e906a4fdc9"}]},"branch":"refs/heads/master"},"9fd7e7af6f75dfa12bb6b6aa7bc323678b7a04b1":{"kind":"REWORK","_number":6,"created":"2016-02-22 22:48:17.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/6"}}},"commit":{"parents":[{"commit":"d9f500a1289a02bcb47638e4a78b2ac4ec7cfdbc","subject":"Merge \"Make _get_addr() method a function in utils.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9f500a1289a02bcb47638e4a78b2ac4ec7cfdbc"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-02-22 22:45:50.000000000","tz":-480},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fd7e7af6f75dfa12bb6b6aa7bc323678b7a04b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fd7e7af6f75dfa12bb6b6aa7bc323678b7a04b1"}]},"branch":"refs/heads/master"},"a858b422018feb9f1be62f43ad591860899e1824":{"kind":"REWORK","_number":7,"created":"2016-02-23 22:19:58.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/7"}}},"commit":{"parents":[{"commit":"cfd50212358372694ba0713ff8c3bc313c6f1a92","subject":"dry out container unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfd50212358372694ba0713ff8c3bc313c6f1a92"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-02-23 22:18:58.000000000","tz":-480},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a858b422018feb9f1be62f43ad591860899e1824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a858b422018feb9f1be62f43ad591860899e1824"}]},"branch":"refs/heads/master"},"68bdfdb60e7383a3d213fcd630b445c877bda561":{"kind":"REWORK","_number":8,"created":"2016-03-03 16:37:31.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/8"}}},"commit":{"parents":[{"commit":"bd93d44bb46350f716f1dfb66fb43c2ab4ad8c50","subject":"Make sure all temp files get deleted in test_ringbuilder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd93d44bb46350f716f1dfb66fb43c2ab4ad8c50"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-03-03 16:29:59.000000000","tz":-480},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68bdfdb60e7383a3d213fcd630b445c877bda561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68bdfdb60e7383a3d213fcd630b445c877bda561"}]},"branch":"refs/heads/master"},"241890cc45fa00911c2849e43fc31ae8376950f5":{"kind":"REWORK","_number":9,"created":"2016-04-13 23:51:57.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/9"}}},"commit":{"parents":[{"commit":"669c3f1e7ee032c570e35921029087fdb3430667","subject":"Merge \"added javaswift to associated projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/669c3f1e7ee032c570e35921029087fdb3430667"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-04-13 23:51:46.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/241890cc45fa00911c2849e43fc31ae8376950f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/241890cc45fa00911c2849e43fc31ae8376950f5"}]},"branch":"refs/heads/master"},"a2722680b6e930183558f1b33f1f68cb445ed8ff":{"kind":"REWORK","_number":10,"created":"2016-04-21 02:28:54.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/10"}}},"commit":{"parents":[{"commit":"760dcfd53da1eed81b95c04c19fa54de8c947de7","subject":"Merge \"Require account/container metadata be UTF-8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760dcfd53da1eed81b95c04c19fa54de8c947de7"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-04-21 02:23:42.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2722680b6e930183558f1b33f1f68cb445ed8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2722680b6e930183558f1b33f1f68cb445ed8ff"}]},"branch":"refs/heads/master"},"660e5367dd2037915dd38da47932fe18482cecf1":{"kind":"REWORK","_number":11,"created":"2016-05-26 00:53:12.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/11"}}},"commit":{"parents":[{"commit":"b0aef5aed2e8885513cad1cabeb243961124d70b","subject":"Merge \"Fix setup of manifest responses in SLO tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0aef5aed2e8885513cad1cabeb243961124d70b"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-05-26 00:52:57.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/660e5367dd2037915dd38da47932fe18482cecf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/660e5367dd2037915dd38da47932fe18482cecf1"}]},"branch":"refs/heads/master"},"2ce57c0a2ad4a88479f9c609937b86c4acfd04c7":{"kind":"REWORK","_number":12,"created":"2016-06-15 18:50:28.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/12"}}},"commit":{"parents":[{"commit":"ae0149811ddfc550f5e407f16d85d551d10e6f5f","subject":"Merge \"Don\u0027t include holes when reporting how many devices a ring has\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae0149811ddfc550f5e407f16d85d551d10e6f5f"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-06-15 18:50:05.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ce57c0a2ad4a88479f9c609937b86c4acfd04c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ce57c0a2ad4a88479f9c609937b86c4acfd04c7"}]},"branch":"refs/heads/master"},"a67db3efcc8c3c29ed6599e3a3d541dd6ce3a32c":{"kind":"REWORK","_number":13,"created":"2016-07-12 21:42:36.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/13"}}},"commit":{"parents":[{"commit":"2ee1710b689dbc9c100e618dc3951c9a530f1c8d","subject":"Merge \"make 0 be avaiable value of options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ee1710b689dbc9c100e618dc3951c9a530f1c8d"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-07-12 21:26:27.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a67db3efcc8c3c29ed6599e3a3d541dd6ce3a32c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a67db3efcc8c3c29ed6599e3a3d541dd6ce3a32c"}]},"branch":"refs/heads/master"},"4cb00a0403b98b4fd4f0dd808fc30505c7f60ff2":{"kind":"REWORK","_number":14,"created":"2016-07-28 00:59:10.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/14"}}},"commit":{"parents":[{"commit":"2ee1710b689dbc9c100e618dc3951c9a530f1c8d","subject":"Merge \"make 0 be avaiable value of options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ee1710b689dbc9c100e618dc3951c9a530f1c8d"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-07-28 00:58:19.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nThe discussion results at Austin Hackathon and Tokyo Summit is\nreflected to this patch. Currently, tese cases do not cover all\nmodifications.\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  two sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027 and\n  \u0027X-Container-Sysmeta-Objects-Queued\u0027. The former stores the\n  index of source policy and the latter indicates if objects\n  in the container are queued for policy changing.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cb00a0403b98b4fd4f0dd808fc30505c7f60ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cb00a0403b98b4fd4f0dd808fc30505c7f60ff2"}]},"branch":"refs/heads/master"},"2af6da9a26a5597e8c47292e9a5a2bb41e236e86":{"kind":"REWORK","_number":15,"created":"2016-08-18 23:09:49.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/15"}}},"commit":{"parents":[{"commit":"6e900205a93e53ab0d44a2cee24b1a860fcbcb32","subject":"Merge \"monkey_patch_mimetools() now does nothing on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e900205a93e53ab0d44a2cee24b1a860fcbcb32"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2016-08-18 23:07:04.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027. This stores the\n  previous index of storage policy.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2af6da9a26a5597e8c47292e9a5a2bb41e236e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2af6da9a26a5597e8c47292e9a5a2bb41e236e86"}]},"branch":"refs/heads/master"},"39e7ef01167639db78c16e0e5b3f61dc60823d0c":{"kind":"REWORK","_number":16,"created":"2017-01-14 03:26:15.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/16"}}},"commit":{"parents":[{"commit":"aebcc5898c4b7d643b066bd6477fb7b840400e53","subject":"Merge \"Fix object server tests to include content-type headers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aebcc5898c4b7d643b066bd6477fb7b840400e53"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2017-01-14 03:26:07.000000000","tz":-480},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027. This stores the\n  previous index of storage policy.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39e7ef01167639db78c16e0e5b3f61dc60823d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39e7ef01167639db78c16e0e5b3f61dc60823d0c"}]},"branch":"refs/heads/master"},"550d9e80d7deaac05ec91acbe187ccfdff3f03c8":{"kind":"REWORK","_number":17,"created":"2017-02-22 15:16:06.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/17"}}},"commit":{"parents":[{"commit":"413fb2b84fb7de4bc8fe76decedd56b58fa50220","subject":"Merge \"Prevent traceback in object-server on client disconnect\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/413fb2b84fb7de4bc8fe76decedd56b58fa50220"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2017-02-22 15:09:13.000000000","tz":-480},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027. This stores the\n  previous index of storage policy.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/550d9e80d7deaac05ec91acbe187ccfdff3f03c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/550d9e80d7deaac05ec91acbe187ccfdff3f03c8"}]},"branch":"refs/heads/master"},"4437aa061b9ced732fcab48485c045f487409c8a":{"kind":"REWORK","_number":18,"created":"2017-03-15 23:43:56.000000000","uploader":{"_account_id":8859,"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","username":"morrisson"},"ref":"refs/changes/29/209329/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/209329/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/209329/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/209329/18"}}},"commit":{"parents":[{"commit":"2dd835726939f79e70af3c4a60112b421c66be96","subject":"Merge \"refactor some common code from crypto\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dd835726939f79e70af3c4a60112b421c66be96"}]}],"author":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2015-06-25 08:33:04.000000000","tz":540},"committer":{"name":"Daisuke Morita","email":"morita.daisuke@ntti3.com","date":"2017-03-15 23:43:46.000000000","tz":-420},"subject":"WIP: Changing Policies","message":"WIP: Changing Policies\n\nIn this patch, following parts are added or modified\n* Add API to force changing a policy of a container\n* Add object-transferrer daemon to send objects according to\n  the ring of the changed policy\n* A container to be planned to change its policy is tagged by\n  sysmeta \u0027X-Container-Sysmeta-Prev-Index\u0027. This stores the\n  previous index of storage policy.\n* Outline of changing policy process for some container is as\n  follows (it is similar to reconciler\u0027s process)\n  1. User orders to change the storage policy of container by\n     Container POST method\n  2. Container-replicator checks if a container is orderred to\n     change the policy, and if so, replicator create a special\n     container and enqueue objects info to this container so that\n     object-transferrer can really migrate the policy of objects\n  3. Container-update notifies to a system account for policy\n     changing that a special container is created\n  4. Object-transferrer finds objects to be transferred\n* Add headers which show each policy\u0027s stats for Container\n  GET/HEAD to find the progress of policy changing\n\nAbove procedures can minimize the burden of container DB and\nkeep user-side API atomic\n\nImplements: blueprint changing-policies\nChange-Id: I5145c3fcad1535c6bae5eb258ebb09828cbc3b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4437aa061b9ced732fcab48485c045f487409c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4437aa061b9ced732fcab48485c045f487409c8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY"},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23308,"name":"yangjidong","email":"yangjd@inspur.com","username":"yangjidong"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
