)]}'
{"id":"openstack%2Fswift~29327","triplet_id":"openstack%2Fswift~master~I4931add4b8fb7c4f57820e52f182c21040d8db10","project":"openstack/swift","branch":"master","topic":"dead-and-broken","hashtags":[],"change_id":"I4931add4b8fb7c4f57820e52f182c21040d8db10","subject":"Remove dead code.","status":"MERGED","created":"2013-05-15 23:59:17.000000000","updated":"2013-05-16 23:09:56.000000000","submitted":"2013-05-16 23:09:56.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63f438dbcf55a6178c6ce0bfcb56042377597afb","_number":29327,"virtual_id_number":29327,"owner":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":2,"date":"2013-05-16 23:09:56.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"all":[{"value":2,"date":"2013-05-16 21:38:26.000000000","_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-16 01:11:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":2,"date":"2013-05-16 21:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"date":"2013-05-16 04:43:43.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"all":[{"value":1,"date":"2013-05-16 21:38:26.000000000","_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":0,"date":"2013-05-16 21:24:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"date":"2013-05-16 04:43:43.000000000","_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-16 01:11:41.000000000","updated_by":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"reviewer":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"state":"REVIEWER"},{"updated":"2013-05-16 21:24:14.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2013-05-16 21:38:26.000000000","updated_by":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"reviewer":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"state":"REVIEWER"},{"updated":"2013-05-16 23:09:56.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"9c00c12f2039091ec0ad2698b26492961ea7072b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-16 00:42:40.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/29327/1/check/gate-swift-pep8/2369 : SUCCESS in 48s\n- http://docs-draft.openstack.org/29327/1/check/gate-swift-docs/2197/doc/build/html/ : SUCCESS in 1m 08s\n- http://logs.openstack.org/29327/1/check/gate-swift-python26/2445 : FAILURE in 1m 23s\n- http://logs.openstack.org/29327/1/check/gate-swift-python27/2447 : SUCCESS in 1m 11s\n- http://logs.openstack.org/29327/1/check/gate-tempest-devstack-vm-full/16213 : SUCCESS in 41m 13s\n- http://logs.openstack.org/29327/1/check/gate-tempest-devstack-vm-quantum/22715 : SUCCESS in 22m 56s\n- http://logs.openstack.org/29327/1/check/gate-tempest-devstack-vm-postgres-full/15929 : FAILURE in 42m 48s (non-voting)\n- http://logs.openstack.org/29327/1/check/gate-grenade-devstack-vm/3824 : FAILURE in 10m 00s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"3694c099ab50d64a22511de6db07e54fa74daf86","author":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"date":"2013-05-16 01:11:41.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"3ba7ca2c1c1468288db3ba46363c50f4f7e241b5","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2013-05-16 04:43:43.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"4c416b82e55edc38ad17109ec523b94ae7fa6ff6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-16 05:34:10.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/29327/1/check/gate-swift-pep8/2371 : SUCCESS in 45s\n- http://docs-draft.openstack.org/29327/1/check/gate-swift-docs/2199/doc/build/html/ : SUCCESS in 56s\n- http://logs.openstack.org/29327/1/check/gate-swift-python26/2447 : SUCCESS in 1m 12s\n- http://logs.openstack.org/29327/1/check/gate-swift-python27/2449 : SUCCESS in 57s\n- http://logs.openstack.org/29327/1/check/gate-tempest-devstack-vm-full/16239 : SUCCESS in 50m 16s\n- http://logs.openstack.org/29327/1/check/gate-tempest-devstack-vm-quantum/22744 : SUCCESS in 21m 24s\n- http://logs.openstack.org/29327/1/check/gate-tempest-devstack-vm-postgres-full/15955 : FAILURE in 49m 00s (non-voting)\n- http://logs.openstack.org/29327/1/check/gate-grenade-devstack-vm/3849 : FAILURE in 9m 59s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"06711ab7c97ee91556080f7047d5aabe618ca441","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2013-05-16 21:24:14.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"03bc71f2e551b69056c0992cf03bb51ea6e3d60a","author":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"date":"2013-05-16 21:38:26.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"26ea7791ee5277a0868b93b1d530842b80707816","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-16 21:38:28.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"3f8afafa703dfea71251843fd9eddce959b22e9e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-16 23:09:56.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/29327/1/gate/gate-swift-docs/2212/doc/build/html/ : SUCCESS in 1m 00s\n- http://logs.openstack.org/29327/1/gate/gate-swift-pep8/2384 : SUCCESS in 38s\n- http://logs.openstack.org/29327/1/gate/gate-swift-python26/2460 : SUCCESS in 1m 13s\n- http://logs.openstack.org/29327/1/gate/gate-swift-python27/2462 : SUCCESS in 58s\n- http://logs.openstack.org/29327/1/gate/gate-tempest-devstack-vm-full/16419 : SUCCESS in 50m 31s\n- http://logs.openstack.org/29327/1/gate/gate-tempest-devstack-vm-quantum/22947 : SUCCESS in 19m 40s\n- http://logs.openstack.org/29327/1/gate/gate-tempest-devstack-vm-postgres-full/16134 : FAILURE in 41m 50s (non-voting)\n- http://logs.openstack.org/29327/1/gate/gate-grenade-devstack-vm/4013 : FAILURE in 10m 09s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"6b304fdf5e604e3d553bcce884f2021edc438a77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-16 23:09:56.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc10d9993bca14197270b0eb849960272a3e9368","revisions":{"dc10d9993bca14197270b0eb849960272a3e9368":{"kind":"REWORK","_number":1,"created":"2013-05-15 23:59:17.000000000","uploader":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"ref":"refs/changes/27/29327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/29327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/29327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/29327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/29327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/29327/1"}}},"commit":{"parents":[{"commit":"81623cfa2c4142ebf6ba34721269d8f5fc90e058","subject":"Perform mount check after REST params valid on PUT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81623cfa2c4142ebf6ba34721269d8f5fc90e058"}]}],"author":{"name":"Samuel Merritt","email":"sam@swiftstack.com","date":"2013-05-15 23:53:17.000000000","tz":-420},"committer":{"name":"Samuel Merritt","email":"sam@swiftstack.com","date":"2013-05-15 23:53:17.000000000","tz":-420},"subject":"Remove dead code.","message":"Remove dead code.\n\nWhile it is true that one *could* call ContainerBroker.get_info() with\ninclude_metadata\u003dTrue, it is also true that nobody actually does.\n\nIt\u0027s also true that the code path taken when include_metadata\u003dTrue is\nbroken in the case of non-ASCII metadata keys\n(e.g. X-Container-Meta-☃: snowman), so leaving it around is just\nasking for future pain.\n\nChange-Id: I4931add4b8fb7c4f57820e52f182c21040d8db10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc10d9993bca14197270b0eb849960272a3e9368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc10d9993bca14197270b0eb849960272a3e9368"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"81623cfa2c4142ebf6ba34721269d8f5fc90e058","is_merged_in_target_branch":true,"change_id":"I6e5f34cc65b57662b7f88224ca0e1501787e0f43","change_number":29106,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
