)]}'
{"id":"openstack%2Fswift~337861","triplet_id":"openstack%2Fswift~master~I991806227f18f115646b0549a77326617aa2c8fe","project":"openstack/swift","branch":"master","topic":"change-337861-7","hashtags":[],"change_id":"I991806227f18f115646b0549a77326617aa2c8fe","subject":"Permit to bind object-server on replication_port","status":"NEW","created":"2016-07-05 20:46:21.000000000","updated":"2021-02-09 00:24:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a08b712c5ee28df1ff88125bd2459058c23939fb","_number":337861,"virtual_id_number":337861,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-09 00:24:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-10 09:09:07.000000000","updated_by":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"reviewer":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"state":"REVIEWER"},{"updated":"2017-03-10 12:13:20.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-03-10 12:23:34.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2019-10-30 02:55:22.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-12-19 06:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-31 18:15:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-03-19 21:03:10.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"78b70dbd69cd9eb22139664bc9e89722601d98bf","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2016-07-05 20:46:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83f06f13156c156fd37660290e1db0d40b0f36ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-05 21:48:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs http://docs-draft.openstack.org/61/337861/1/check/gate-swift-docs/6f6806f//doc/build/html/ : SUCCESS in 3m 29s\n- gate-swift-pep8 http://logs.openstack.org/61/337861/1/check/gate-swift-pep8/8351519/ : SUCCESS in 3m 16s\n- gate-swift-python27 http://logs.openstack.org/61/337861/1/check/gate-swift-python27/0f054ee/ : SUCCESS in 4m 50s\n- gate-swift-python34 http://logs.openstack.org/61/337861/1/check/gate-swift-python34/aa5623e/ : SUCCESS in 3m 14s\n- gate-swift-python35-nv http://logs.openstack.org/61/337861/1/check/gate-swift-python35-nv/6d7a13f/ : FAILURE in 3m 38s (non-voting)\n- gate-swift-tox-func http://logs.openstack.org/61/337861/1/check/gate-swift-tox-func/2735841/ : SUCCESS in 4m 51s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/61/337861/1/check/gate-swift-tox-func-in-process-fast-post/b51bc1e/ : SUCCESS in 4m 29s\n- gate-tempest-dsvm-full http://logs.openstack.org/61/337861/1/check/gate-tempest-dsvm-full/8164968/ : SUCCESS in 48m 24s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/61/337861/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/8c904c1/ : SUCCESS in 56m 32s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/61/337861/1/check/gate-grenade-dsvm/cb2d963/ : SUCCESS in 31m 43s\n- gate-swift-dsvm-functional http://logs.openstack.org/61/337861/1/check/gate-swift-dsvm-functional/29f815a/ : SUCCESS in 26m 30s\n- gate-swift-tox-bandit http://logs.openstack.org/61/337861/1/check/gate-swift-tox-bandit/2e0a0cd/ : SUCCESS in 3m 17s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-nv http://logs.openstack.org/61/337861/1/check/gate-swift-dsvm-functional-identity-v3-only-nv/1fa7086/ : SUCCESS in 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39d93b1f04ee72a4157b0e5f4326cf863fe47c55","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-07-06 01:05:57.000000000","message":"Patch Set 1:\n\nBuild succeeded \n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/4264/ssbench : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/3571/probetests/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/5626/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/3670/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0aea0820f2245527d8e332b3105ff4edd82e0513","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-31 23:54:09.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"237bff9003f3a2df8c398c90d87cdc2c4a12e919","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-03-02 20:30:12.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"b0c8c8bc6204b4237ded432d7a4e797d65dd308d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-02 20:30:31.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"cc234130bb492242a5354c2811e447470c0509b9","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-03-02 21:10:21.000000000","message":"Patch Set 2:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/5991/probetests/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/5656/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/7622/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e1eb4bf1adccdfc57d780cd19f45ef1e7a32f20b","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-03-09 23:05:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"911f64b66c266f1217d1c79a6c44e3d241391fc8","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-03-09 23:47:00.000000000","message":"Patch Set 3:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/6068/probetests/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/5733/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/7699/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6e2ffe9bf02ad4907474712fe43e1cfe6b29a6e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-10 00:22:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/61/337861/3/check/gate-swift-docs-ubuntu-xenial/4bd318c//doc/build/html/ : SUCCESS in 1m 57s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-pep8-ubuntu-xenial/881ce60/ : SUCCESS in 2m 17s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-python27-ubuntu-xenial/11e397f/ : SUCCESS in 3m 50s\n- gate-swift-python35 http://logs.openstack.org/61/337861/3/check/gate-swift-python35/a22ee81/ : SUCCESS in 2m 05s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3f338e3/ : SUCCESS in 1h 15m 13s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3c78848/ : SUCCESS in 47m 16s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/4cf7c51/ : SUCCESS in 5m 30s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/66ba7ed/ : SUCCESS in 2m 40s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/ccd0090/ : SUCCESS in 6m 07s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/dc1858d/ : SUCCESS in 4m 20s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/f73ccfe/ : SUCCESS in 3m 46s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/61/337861/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/02f2042/ : SUCCESS in 1h 08m 44s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/61/337861/3/check/gate-swift-dsvm-functional-ubuntu-xenial/c121eb8/ : SUCCESS in 21m 04s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/61/337861/3/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/7e437d5/ : SUCCESS in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"272fd7372640c17e082a255bcf99f38b5a180aac","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2017-03-10 01:37:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5e2915d4e44dbb7ae67349a8b55440a6c28b66d0","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-03-10 08:46:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9f6e5cd758f9f5625eaffdb460fc362a1221a66a","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-03-10 08:46:07.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"00d4045e1aad518c7df5da7bef55024f98165699","author":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"date":"2017-03-10 09:09:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nSome nits on first pass","accounts_in_message":[],"_revision_number":4},{"id":"94cf0c135c14b6a2634425d8495faa417d271171","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-03-10 09:35:12.000000000","message":"Patch Set 4:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/6071/probetests/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/5736/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/7702/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"e4153f3a4a6d05d4822e980e62a5e7768b81e7cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-10 10:06:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/61/337861/4/check/gate-swift-docs-ubuntu-xenial/5ea76b0//doc/build/html/ : SUCCESS in 2m 52s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-pep8-ubuntu-xenial/c97b743/ : SUCCESS in 4m 20s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-python27-ubuntu-xenial/e203b64/ : SUCCESS in 5m 41s\n- gate-swift-python35 http://logs.openstack.org/61/337861/4/check/gate-swift-python35/bd3e134/ : SUCCESS in 1m 44s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0f11552/ : SUCCESS in 1h 12m 39s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/9bfc1fc/ : SUCCESS in 37m 54s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/3ddb00d/ : SUCCESS in 5m 55s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/f306ccf/ : SUCCESS in 1m 11s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/e958719/ : SUCCESS in 4m 45s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/88f8ac8/ : SUCCESS in 3m 23s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/3a1fef1/ : SUCCESS in 5m 30s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/61/337861/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/48935ab/ : SUCCESS in 1h 16m 08s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/61/337861/4/check/gate-swift-dsvm-functional-ubuntu-xenial/5863494/ : SUCCESS in 19m 45s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/61/337861/4/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/8669146/ : SUCCESS in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"660b67fbc2fedbf0cb4e377b22ffe3aaabb7d7b9","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-03-10 10:55:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f597024305e9aa8f8c97e514bd2104335c0c7610","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-03-10 10:55:45.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dcc23d9bc1da5dcffed7cc04fa0050013f246174","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-03-10 12:13:20.000000000","message":"Patch Set 5:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/6074/probetests/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/5739/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/337861/7705/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"c36c117042bd847d8b4174b236a17400e712f9d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-10 12:23:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/61/337861/5/check/gate-swift-docs-ubuntu-xenial/ed30c0a//doc/build/html/ : SUCCESS in 3m 06s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-pep8-ubuntu-xenial/2d3bd55/ : SUCCESS in 3m 29s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-python27-ubuntu-xenial/afe8537/ : SUCCESS in 5m 17s\n- gate-swift-python35 http://logs.openstack.org/61/337861/5/check/gate-swift-python35/5c28972/ : SUCCESS in 2m 28s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/110af10/ : SUCCESS in 1h 14m 13s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-grenade-dsvm-neutron-ubuntu-xenial/70b1e8b/ : SUCCESS in 35m 37s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/5e7ebea/ : SUCCESS in 3m 21s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/609da36/ : SUCCESS in 1m 21s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/84d0458/ : SUCCESS in 3m 21s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/e66f450/ : SUCCESS in 5m 25s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/e29aa14/ : SUCCESS in 3m 25s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/61/337861/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/28511de/ : SUCCESS in 1h 26m 16s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/61/337861/5/check/gate-swift-dsvm-functional-ubuntu-xenial/cf725cb/ : SUCCESS in 34m 29s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/61/337861/5/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/06c96b5/ : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b2a4ab2217015d1412457c05ca20f842e6214549","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2019-05-03 16:21:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"036e53851fa38167491e8f323369bafec2322511","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:57:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/61/337861/6/check/openstack-tox-docs/45413c0/html/ : SUCCESS in 4m 06s\n- swift-tox-py27 http://logs.openstack.org/61/337861/6/check/swift-tox-py27/957696e/ : SUCCESS in 9m 51s\n- swift-tox-py35 http://logs.openstack.org/61/337861/6/check/swift-tox-py35/e6cf7d9/ : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/61/337861/6/check/swift-tox-py36/e460633/ : SUCCESS in 7m 45s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/61/337861/6/check/swift-tox-py37/6bd2d56/ : SUCCESS in 7m 58s\n- swift-tox-func http://logs.openstack.org/61/337861/6/check/swift-tox-func/bdd3059/ : SUCCESS in 7m 59s\n- swift-tox-func-encryption http://logs.openstack.org/61/337861/6/check/swift-tox-func-encryption/9c7ad9d/ : SUCCESS in 8m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/61/337861/6/check/swift-tox-func-domain-remap-staticweb/5a365a4/ : SUCCESS in 9m 08s\n- swift-tox-func-ec http://logs.openstack.org/61/337861/6/check/swift-tox-func-ec/0875ee2/ : SUCCESS in 8m 54s\n- swift-tox-func-s3api http://logs.openstack.org/61/337861/6/check/swift-tox-func-s3api/a98db27/ : SUCCESS in 11m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/61/337861/6/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4227720/ : SUCCESS in 11m 55s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/61/337861/6/check/swift-probetests-centos-7/42e6364/ : SUCCESS in 31m 00s\n- swift-dsvm-functional http://logs.openstack.org/61/337861/6/check/swift-dsvm-functional/c78ec06/ : SUCCESS in 28m 46s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/61/337861/6/check/swift-dsvm-functional-ipv6/d5e1a88/ : SUCCESS in 27m 18s\n- swift-tox-lower-constraints http://logs.openstack.org/61/337861/6/check/swift-tox-lower-constraints/e960bc6/ : SUCCESS in 11m 35s\n- openstack-tox-pep8 http://logs.openstack.org/61/337861/6/check/openstack-tox-pep8/2c306be/ : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/61/337861/6/check/swift-multinode-rolling-upgrade/52c419d/ : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":6},{"id":"4baa31ea4a87e43a2f04770e36e9389ec1f3380d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-10-30 02:54:33.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nJust had a quick look, but if this is already being run in prod somewhere, then I think it makes sense to get it in. especially if it 1. makes sense, and 2. makes someone in our community not have to carry it downstream :)\n\nHave a question inline. Which may or may not make sense. I\u0027ll try and find some time to actually do some testing and a proper review.","accounts_in_message":[],"_revision_number":6},{"id":"82b428935dbfc5b29231887df1fdf743d5df0213","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-10-30 02:55:22.000000000","message":"Patch Set 6: Code-Review+1\n\nForgot to give it a score. +1 for now, until I get a chance to properly review","accounts_in_message":[],"_revision_number":6},{"id":"9e7a322eb34ca8e13eb4e3dfb1a9ee07b3eabd5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-19 05:27:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8e4228721e143e1eb88adb4dfb37b0f6497a3807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 06:32:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78109b1733b042d5bdc1bf1a4349884e : SUCCESS in 7m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2423797253d240d9b1abb1afa34aec46 : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/789d4ea53ae44ebc87ab6e518aaca013 : SUCCESS in 56m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18b130758f004cfb853a5b3c87d891c5 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91822134aaf74073af10e3937559765b : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ea41681e9314c689b115778d0b02804 : SUCCESS in 11m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/888d30f41b78495a93112afbc7b5cdb2 : SUCCESS in 11m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fad52e4f4a7e4b65869f7aa562f2b79f : SUCCESS in 11m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3be00938906b4ebca35b4bcf060141a8 : SUCCESS in 9m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e519597c849b444bbe399995412d2cc1 : SUCCESS in 10m 03s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/07ff946660ed43b99c53c7dfea0a17dc : SUCCESS in 12m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/41aac5ab54004d998bbd5e8767753b96 : SUCCESS in 13m 12s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0b7337cc4a3c4eebbb0c900c9bdee781 : SUCCESS in 12m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1060ad9145e24d13ae372b5e611e1d89 : SUCCESS in 9m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/adb829b071bd4e08a0844e91a0a49a29 : SUCCESS in 9m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5d2313f26ed742b4b1fc024405044f57 : SUCCESS in 10m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2c03733b66dc4de09927a2da9cc02b29 : SUCCESS in 10m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/41b5ae934d0c4f72bc5ceac555ab81ce : SUCCESS in 15m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/722ea7aacf434ea1930d1cf7d1ba6131 : SUCCESS in 31m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/63b0b241aaee4007bbd40c8757e2b632 : SUCCESS in 33m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea1021356dd6455abdd2a480b8d3e32e : SUCCESS in 28m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3ef971ab71194a7a91a6c2288c48b6c6 : SUCCESS in 29m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3d164c407fa4f3c95ef94385d5f2858 : SUCCESS in 30m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f8b8757d3424c3195c97af73d85edf3 : SUCCESS in 14m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff87dbcf9ff492da291554dceb045d5 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/beb59b86c8294e90bd1da062362a8388 : SUCCESS in 22m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11a9d43460424c37947cc5faf9b2e12a : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":7},{"id":"713bd849c6867f63fba536a281da00ef3167513c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 18:15:56.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nThis is mainly for when replication_ip \u003d\u003d client-traffic ip but the set of replication_ports is separate from the set of client-traffic ports, yeah?\n\nHow will this behave when replication_ip !\u003d proxy-traffic ip? In particular, if you\u0027ve got separate config files (and separate bind_ips) for your client-traffic vs replication servers?","accounts_in_message":[],"_revision_number":7},{"id":"6cf9857a0dcf080e489725994628754b93b1e82c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 18:19:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0f9707d14c6b6cc76e568defeb70c91e05a7a0a7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-03-19 21:02:52.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nPersonally I find this configuration method too convoluted.\nI would much prefer if servers_per_rep_port were added instead.\nBut since nobody else complains, maybe it\u0027s all right.\nI saw the comments about using separate configs.\n\n-1 for Romain to consider if he wants to abort on config typos.\nI can take it off we agree that it\u0027s a non-issue.","accounts_in_message":[],"_revision_number":7},{"id":"7a7610b5217040323e72967556f6683bc3f12669","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-03-19 21:03:10.000000000","message":"Patch Set 7: Code-Review-1","accounts_in_message":[],"_revision_number":7},{"id":"07ea6062a06d1a92cd897a7453845deba04a03d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2021-02-08 23:24:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a08b712c5ee28df1ff88125bd2459058c23939fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:24:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a58d04a05d764b299cfbabd22fd0d8e0 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/05ffa025ee444971a6680749690a4d02 : SUCCESS in 54m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa10d0a2468446c7ac7b54ae24ae727c : FAILURE in 39m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1487f48cbcb64d6b94e7595c48edbe73 : SUCCESS in 7m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f9632f0f7a645c7bc3aa34e788279fc : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41eccb7c2eca4ea399792aa2a1dd0862 : SUCCESS in 15m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0e1e57c8b0d409ab968e6355a82f509 : SUCCESS in 18m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/271f0731fb8543e996a187b2035514e3 : SUCCESS in 25m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4d1625d84f9a436ebb4cd085c0ede055 : SUCCESS in 14m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/185eb623645c44f7ada2336be4a217ee : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75d7af50680a4619b516e5aeaca65319 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/470dbd3472704a35af8ce285f8d33114 : SUCCESS in 18m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f6ee076bc08d453186d0e9eac79b2ba5 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4e59c45cf2a145c3a85c8609f7f49b60 : SUCCESS in 34m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/93bb2bd571dd45d190ea226e097f4c15 : SUCCESS in 16m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75f453fbb0fd4587942236e1b3226938 : SUCCESS in 20m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/796702f2ecd44aab90735847c0d6fd3d : SUCCESS in 54m 29s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1e7426d09a274ba0b8cd2fcc77f2e773 : SUCCESS in 44m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57e4771a4ad7483e8996506446c0bcd3 : SUCCESS in 39m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd48fc505bdf43b6aea51a75e30f631e : SUCCESS in 35m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88f955d25f784133aeb2a48b18277d9b : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e112b7c2f04bddadc0278cc7365ffc : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77a33a94784c46c29c7dbacdcbb22f03 : FAILURE in 7m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/524ee02c76a14778ae5ed4f1d933cd7e : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"758e53b6ef8432ca66556f237cb86e28d6c6d587","revisions":{"e69281d5772561fa3e9b0992274036cd290b4549":{"kind":"REWORK","_number":1,"created":"2016-07-05 20:46:21.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/61/337861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/1"}}},"commit":{"parents":[{"commit":"9f4a66229680f40164140ca885d38b10ed693b72","subject":"Merge \"Stop digging for publicly_accessible ourselves\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f4a66229680f40164140ca885d38b10ed693b72"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-07-05 20:22:33.000000000","tz":120},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a semicolon.\n\nThe following configuration will create two workers per devices if port\nand replication_port are differents in the ring. One worker for clients\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e69281d5772561fa3e9b0992274036cd290b4549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e69281d5772561fa3e9b0992274036cd290b4549"}]},"branch":"refs/heads/master"},"8851b0c77e93c1a4ddb0496b5422bdee5c223361":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-03-02 20:30:12.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/61/337861/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/2"}}},"commit":{"parents":[{"commit":"9f4a66229680f40164140ca885d38b10ed693b72","subject":"Merge \"Stop digging for publicly_accessible ourselves\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f4a66229680f40164140ca885d38b10ed693b72"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2017-03-02 20:30:09.000000000","tz":0},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a semicolon.\n\nThe following configuration will create two workers per devices if port\nand replication_port are differents in the ring. One worker for clients\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nCloses-Bug: #1669579\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8851b0c77e93c1a4ddb0496b5422bdee5c223361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8851b0c77e93c1a4ddb0496b5422bdee5c223361"}]},"branch":"refs/heads/master"},"9b1c8fd8723b23e22346a95232e568db8f0847f2":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2017-03-09 23:05:36.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/61/337861/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/3"}}},"commit":{"parents":[{"commit":"455827716d7b049fd9253a165fc87875e078d740","subject":"Merge \"Optimize ec duplication and its md5 hashing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/455827716d7b049fd9253a165fc87875e078d740"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2017-03-09 23:03:42.000000000","tz":60},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a semicolon.\n\nThe following configuration will create two workers per devices if port\nand replication_port are differents in the ring. One worker for clients\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nCloses-Bug: #1669579\n\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b1c8fd8723b23e22346a95232e568db8f0847f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b1c8fd8723b23e22346a95232e568db8f0847f2"}]},"branch":"refs/heads/master"},"79a8fd7b7d18ac6dc7e58e43aeda963319e29e1e":{"kind":"REWORK","_number":4,"created":"2017-03-10 08:46:02.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/61/337861/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/4"}}},"commit":{"parents":[{"commit":"455827716d7b049fd9253a165fc87875e078d740","subject":"Merge \"Optimize ec duplication and its md5 hashing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/455827716d7b049fd9253a165fc87875e078d740"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2017-03-10 08:42:38.000000000","tz":60},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a comma.\n\nThe following configuration will create two workers per devices if port\nand replication_port are differents in the ring. One worker for clients\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nCloses-Bug: #1669579\n\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79a8fd7b7d18ac6dc7e58e43aeda963319e29e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79a8fd7b7d18ac6dc7e58e43aeda963319e29e1e"}]},"branch":"refs/heads/master"},"b72f81b1947991433cac7f15369fffbeb8adb32f":{"kind":"REWORK","_number":5,"created":"2017-03-10 10:55:34.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/61/337861/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/5"}}},"commit":{"parents":[{"commit":"455827716d7b049fd9253a165fc87875e078d740","subject":"Merge \"Optimize ec duplication and its md5 hashing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/455827716d7b049fd9253a165fc87875e078d740"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2017-03-10 10:55:24.000000000","tz":60},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a comma.\n\nThe following configuration will create two workers per device if port\nand replication_port are different in the ring. One worker for clients\u0027\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nCloses-Bug: #1669579\n\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b72f81b1947991433cac7f15369fffbeb8adb32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b72f81b1947991433cac7f15369fffbeb8adb32f"}]},"branch":"refs/heads/master"},"cbe5031078cb6b268b726b6d11a9f7195efbad20":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2019-05-03 16:21:00.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/61/337861/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/6"}}},"commit":{"parents":[{"commit":"ce1ba6a51d27271be9cdf9b5915b39bab5498948","subject":"Merge \"Pass logger instances to AccountBroker/ContainerBroker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce1ba6a51d27271be9cdf9b5915b39bab5498948"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-05-03 16:18:32.000000000","tz":-360},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a comma.\n\nThe following configuration will create two workers per device if port\nand replication_port are different in the ring. One worker for clients\u0027\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nIf using different ports in the ring for proxy and replication traffic,\nit allows to dedicate processes to replication, which can be\nIO-intensive and sometime hangs the process for many seconds.\n\nCloses-Bug: #1669579\n\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbe5031078cb6b268b726b6d11a9f7195efbad20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbe5031078cb6b268b726b6d11a9f7195efbad20"}]},"branch":"refs/heads/master"},"47060c889736d2a8fd763d0587180b30ce21c1ad":{"kind":"REWORK","_number":7,"created":"2019-12-19 05:27:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/61/337861/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/7"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-19 05:27:30.000000000","tz":-480},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a comma.\n\nThe following configuration will create two workers per device if port\nand replication_port are different in the ring. One worker for clients\u0027\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nIf using different ports in the ring for proxy and replication traffic,\nit allows to dedicate processes to replication, which can be\nIO-intensive and sometime hangs the process for many seconds.\n\nCloses-Bug: #1669579\n\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47060c889736d2a8fd763d0587180b30ce21c1ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47060c889736d2a8fd763d0587180b30ce21c1ad"}]},"branch":"refs/heads/master"},"758e53b6ef8432ca66556f237cb86e28d6c6d587":{"kind":"REWORK","_number":8,"created":"2021-02-08 23:24:32.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/61/337861/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/337861/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/337861/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/337861/8"}}},"commit":{"parents":[{"commit":"1d34f321ac5d092ec93e7a35bc8406da1eaeda69","subject":"Merge \"Enable shard ranges to be manually shrunk to root container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d34f321ac5d092ec93e7a35bc8406da1eaeda69"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2016-06-26 20:41:40.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2021-02-08 23:24:03.000000000","tz":-300},"subject":"Permit to bind object-server on replication_port","message":"Permit to bind object-server on replication_port\n\nAdd a new configuration option \u0027servers_per_port_type\u0027 to choose on which\nport the object-server must bind when servers_per_port is enabled. It can\nbe \u0027port\u0027 (the default, which does not change the current behavior), or\n\u0027replication_port\u0027, or both separated by a comma.\n\nThe following configuration will create two workers per device if port\nand replication_port are different in the ring. One worker for clients\u0027\ntraffic and one worker for replication traffic:\n    servers_per_port \u003d 1\n    servers_per_port_type \u003d port, replication_port\n\nIf using different ports in the ring for proxy and replication traffic,\nit allows to dedicate processes to replication, which can be\nIO-intensive and sometime hangs the process for many seconds.\n\nCloses-Bug: #1669579\n\nChange-Id: I991806227f18f115646b0549a77326617aa2c8fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/758e53b6ef8432ca66556f237cb86e28d6c6d587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/758e53b6ef8432ca66556f237cb86e28d6c6d587"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
