)]}'
{"id":"openstack%2Fswift~338819","triplet_id":"openstack%2Fswift~master~I7959ab96682f4a459a99c4f92463e1280372f6ef","project":"openstack/swift","branch":"master","topic":"socket-check","attention_set":{},"removed_from_attention_set":{"4608":{"account":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"last_update":"2021-11-19 21:18:53.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-11-10 18:14:30.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I7959ab96682f4a459a99c4f92463e1280372f6ef","subject":"Ensure close socket for memcached if got timeout","status":"MERGED","created":"2016-07-07 10:08:00.000000000","updated":"2021-11-19 21:22:37.000000000","submitted":"2021-11-19 21:18:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"338819-socket-check","meta_rev_id":"366b31a29d8650e381b80184c9936b629111b396","_number":338819,"virtual_id_number":338819,"owner":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":18838,"name":"Paul Dardeau","email":"paul.dardeau@intel.com","username":"pdardeau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-19 21:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18334,"name":"Nandini Tata","email":"nandini.tata@intel.com","username":"ntata"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":18838,"name":"Paul Dardeau","email":"paul.dardeau@intel.com","username":"pdardeau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18334,"name":"Nandini Tata","email":"nandini.tata@intel.com","username":"ntata"},{"value":2,"date":"2021-11-19 20:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":18838,"name":"Paul Dardeau","email":"paul.dardeau@intel.com","username":"pdardeau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18334,"name":"Nandini Tata","email":"nandini.tata@intel.com","username":"ntata"},{"value":1,"date":"2021-11-19 20:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":18334,"name":"Nandini Tata","email":"nandini.tata@intel.com","username":"ntata"},{"_account_id":18838,"name":"Paul Dardeau","email":"paul.dardeau@intel.com","username":"pdardeau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-07 10:56:29.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-07-07 11:17:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-11-28 19:54:14.000000000","updated_by":{"_account_id":18838,"name":"Paul Dardeau","email":"paul.dardeau@intel.com","username":"pdardeau"},"reviewer":{"_account_id":18838,"name":"Paul Dardeau","email":"paul.dardeau@intel.com","username":"pdardeau"},"state":"REVIEWER"},{"updated":"2016-12-14 16:52:29.000000000","updated_by":{"_account_id":18334,"name":"Nandini Tata","email":"nandini.tata@intel.com","username":"ntata"},"reviewer":{"_account_id":18334,"name":"Nandini Tata","email":"nandini.tata@intel.com","username":"ntata"},"state":"REVIEWER"},{"updated":"2021-11-09 23:30:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-11-10 00:26:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e03fe80100bd649c1d545dccc1e2b3585f031c58","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2016-07-07 10:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67edc5fdb439dbde733eb30392d435d279d05625","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-07-07 10:56:29.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* cluster-swift-ssbench https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/338819/4289/ssbench : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/338819/3596/probetests/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/338819/5651/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/338819/3695/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0e4a7491a06d3e96cfbc8f9cf42bb0a7d2479862","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-07 11:17:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs http://docs-draft.openstack.org/19/338819/1/check/gate-swift-docs/044fdf0//doc/build/html/ : SUCCESS in 7m 08s\n- gate-swift-pep8 http://logs.openstack.org/19/338819/1/check/gate-swift-pep8/2d6c042/ : SUCCESS in 3m 07s\n- gate-swift-python27 http://logs.openstack.org/19/338819/1/check/gate-swift-python27/0391458/ : SUCCESS in 6m 37s\n- gate-swift-python34 http://logs.openstack.org/19/338819/1/check/gate-swift-python34/32dc4a9/ : SUCCESS in 7m 07s\n- gate-swift-python35-nv http://logs.openstack.org/19/338819/1/check/gate-swift-python35-nv/b91a784/ : FAILURE in 3m 57s (non-voting)\n- gate-swift-tox-func http://logs.openstack.org/19/338819/1/check/gate-swift-tox-func/912d2b4/ : SUCCESS in 5m 27s\n- gate-swift-tox-func-in-process-fast-post http://logs.openstack.org/19/338819/1/check/gate-swift-tox-func-in-process-fast-post/a1bb466/ : SUCCESS in 7m 19s\n- gate-tempest-dsvm-full http://logs.openstack.org/19/338819/1/check/gate-tempest-dsvm-full/753e7da/ : SUCCESS in 47m 15s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/19/338819/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/ace749a/ : SUCCESS in 1h 07m 37s (non-voting)\n- gate-grenade-dsvm http://logs.openstack.org/19/338819/1/check/gate-grenade-dsvm/892a9d9/ : SUCCESS in 30m 06s\n- gate-swift-dsvm-functional http://logs.openstack.org/19/338819/1/check/gate-swift-dsvm-functional/c51318f/ : SUCCESS in 29m 49s\n- gate-swift-tox-bandit http://logs.openstack.org/19/338819/1/check/gate-swift-tox-bandit/01e4737/ : SUCCESS in 2m 00s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-nv http://logs.openstack.org/19/338819/1/check/gate-swift-dsvm-functional-identity-v3-only-nv/6b85f62/ : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aa34885f804b6f0dda73f21cabdf95b4044f317","author":{"_account_id":18838,"name":"Paul Dardeau","email":"paul.dardeau@intel.com","username":"pdardeau"},"date":"2016-11-28 19:54:14.000000000","message":"Patch Set 1: Code-Review+1\n\nNice work.","accounts_in_message":[],"_revision_number":1},{"id":"2b16b1d3e1aa30b655aa7bb62a5d92106a366d58","author":{"_account_id":18334,"name":"Nandini Tata","email":"nandini.tata@intel.com","username":"ntata"},"date":"2016-12-14 16:52:29.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"d952fcae5e5adb431996ce0b6d051c67867ccf60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-09 23:25:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90aa0b454080040db6fcf688336bc3fcd9808114","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-09 23:30:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"547d259817d745870fb8fe9351cc2a253f3fe80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 00:26:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d7bd28aac54b50afa6eaa02ec80239 : SUCCESS in 15m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f608386c7a6e48ac9f64da2bf32e9c9d : SUCCESS in 52m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c26114d0145e462399321f22b9473fad : SUCCESS in 50m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70098fc712874e6abaabca4debe59ed7 : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f7309a530ece4dc4b1714382d8c777cc : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bdb74b7129a145559dcf237223ef4870 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ef4c8603edc4fa89562222b1b0d2876 : SUCCESS in 17m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8a34a64206546e28569039b44d49e68 : SUCCESS in 15m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/582db28f5b6d45deaf1550e44fcedf63 : SUCCESS in 15m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d07ec434e7c48bba7e06df5b6f3ee47 : SUCCESS in 15m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/adc1179ecfc049eaad5393ebb280ad8f : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/50b4821e207f4bfbb3c450c729706f6d : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22f0a959494f41b5b6dad965e0bc5d30 : SUCCESS in 18m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/598b824e3e574ed59b78379d52a81b83 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/517e5dcedf914c1cbaed7ae1572e351c : SUCCESS in 16m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b9639022898445c2ad66801194dc9df4 : SUCCESS in 17m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/adf85caad00448c699ca07e0cb966b76 : SUCCESS in 14m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a508837721c840b88ad74492add30e17 : SUCCESS in 23m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0253fb59a7a5446bb9d0e59e3bafbd6c : SUCCESS in 51m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c3dccc0adb8740909d327ac574755175 : SUCCESS in 56m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1dac9a1d01fa448caa8d2534fb476cd8 : SUCCESS in 36m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/09d2fd66ec764b1996fd9d9d23518209 : SUCCESS in 42m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ae92c4f9311473cb681daa9e11c27a3 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85738b8bc1a4bdda28fceba2a20ed07 : FAILURE in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9becd47332a74558915e35dba6aacb8d : FAILURE in 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4818a462e00e4db8b864b7ed6e4b9517 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bcf508a28602e171e1c7512bc1813402cefac72e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-10 01:29:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8f675f71c46a472787d6b98494200c793fc4e5a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 02:39:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d630dc0d28d441a481f9524b78df943d : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/cfe2ca15e4264c17a7b6bfe05f21d520 : SUCCESS in 53m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7304856d1cd04872a8e62e4d9db4ca22 : SUCCESS in 42m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9cf143f1f25d4e3295f7f22c9f4a8231 : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c48a1f43310240a696b7d817f29152bf : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8762f85ac9d64d96a112d21e72a76d90 : SUCCESS in 34m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/767c7da865f043ea97e7cdfcee0c5b28 : SUCCESS in 18m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bff82eb2c7824d12932b2cc6dbaf0c10 : SUCCESS in 16m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6d24b4370e14211b15e0edcbb8bac76 : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/20c3d92a329d446cb390a2974421a487 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17610b3e30764913ae6086ca86890c14 : SUCCESS in 24m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d8e67032f3e84f27ba9435d8505eac8a : SUCCESS in 21m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30dd019114e14b05a306226797c32220 : SUCCESS in 36m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a76c4b9e552f442eb13db9c2f4e7e2ec : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f7d84383556f4811902ff6f616349e5f : SUCCESS in 12m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/194a998a62df430aade2673d30b97890 : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6809b38082f744e5b2534469a8dab3c4 : SUCCESS in 12m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3fc188e7881a4db78aafc45c133f2c2e : SUCCESS in 19m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bfa99f4439ba4c638b4eaa8170977163 : SUCCESS in 51m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c458e47bac7e43afa3e4372043cb5a4d : SUCCESS in 57m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/690d87f164ed464594ac11c9bfe381a0 : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/376399f493f341a1b2de041c91f9acf8 : SUCCESS in 33m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc5ec9e2f3bb43b0a7b0c9b50c1b8f16 : TIMED_OUT in 41m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a1ea82b5eb496e8431c541bcf9de80 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54a283039b9a4acfa74d6016b15f830e : FAILURE in 22m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f211acacc34c4cd9837daad4d5a85a1e : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c37e53c46cd4ff934494aae765ced161aee61acf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 04:07:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef4b5554669741d794f45972942180b5 : SUCCESS in 36m 09s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b14806b3aa24c53949c949b5811face : SUCCESS in 35m 59s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/de84d1b603b34be3a9659bec6a3a22b9 : FAILURE in 2h 34m 34s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/40b44b62e145426d9037756b74ffef4f : SUCCESS in 57m 05s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c36a2995ac34c99842bff7dad11d666 : SUCCESS in 1h 09m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"504cb1c7716fe487b8f97c42de8df291f9723758","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-10 18:14:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"19ad74e31995ee84df56b98fab9a262c302c36ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 20:23:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c5dc8a154245b1847ed18065e1dfad : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/7ce27e4911ff4dfb924a927db157d9ca : SUCCESS in 56m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75a992cad4524c82943282fe4a406dd1 : SUCCESS in 42m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec6b0dd318f740b08cd85c17c586ecc6 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/82c5015f426049a4b38fa8c4f62e12be : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59e0c9f13942453c9ace43c72798f47c : SUCCESS in 21m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1215909a3554c7ea0d7c2869de5a13c : SUCCESS in 13m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c8fdc2e85a7a4d44a47ea21eeba9aa40 : SUCCESS in 13m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7f5e5e8fd1574ac0bbc39a968ab5bd61 : SUCCESS in 19m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/38e2f76594ca4004a28e21ede4b3046e : SUCCESS in 18m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d55a698469d64f2eb5d80fe6aa1c55a7 : SUCCESS in 29m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/924b573cc696459e9f66ec961ab3f6ab : SUCCESS in 28m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b856d226f1f453897c7ad3462834846 : SUCCESS in 21m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ba242b8664345d9aa33ba4a0a4ed581 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a562b74ba2e2450e9e0642def7cd2286 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b26cf21e431a45e0850f61727f90b492 : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07e2ddad8dbb4ea7881090d86a0febde : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c7c5d2018c934eaabcac317e274f7583 : SUCCESS in 22m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/981aca044cb340f89cb02c2f7db22e64 : SUCCESS in 1h 00m 36s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7fc1ac84455148af99ebfabcb9c1fcd1 : SUCCESS in 1h 03m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f06d271af68473cbff964c9eb51d14e : SUCCESS in 35m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92d8635f018d40019b7530533541cfb4 : SUCCESS in 40m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b55e2184000145c38131859992b9941c : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76924bf338564bdbb12bdaa3390bbd34 : SUCCESS in 7m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0acaacf30d324a70af2bc663471807e9 : SUCCESS in 33m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5586f8b4fa7d46188c801d448e4fe234 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":3},{"id":"ac57c59cffc19daedac1f2b79ba141eb90f5dbfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 20:25:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fcdffc7f1c14cf2b4726c70063c0f48 : SUCCESS in 50m 59s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99567c6a90094ffdbb36e88c1426ee8c : SUCCESS in 1h 07m 47s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/5af23281897f4c28b9bbaf91809dd83f : FAILURE in 2h 07m 29s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/414ad2e4e44c42c094f42e16d4fb49ac : SUCCESS in 35m 39s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab3cf4d19791441a9e03ea298c6ccf48 : SUCCESS in 1h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7fc0915b5868ec62658abd704d05f06e1ac0435","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-19 20:00:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"47037e26ab6cb3434bdac4c76219d83f0e446703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 20:02:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9795d207b5f9a873c551fa7e980544644125ab29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 21:18:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290f5b7d1c204b6480bea3ef6e8ecf30 : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/2094b9b857dc47fead4f40f92a029404 : SUCCESS in 1h 00m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8547da4ea0df4e99a4e13fd4f68451bb : SUCCESS in 52m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ec73361c02a9406e89ca4ccef40c36ce : SUCCESS in 11m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/88d2c70458254a22b104be4b099312f3 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a29ea79474b64eb0823c26c2c28f3569 : SUCCESS in 16m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95bb7068c33c4d18be85e09a2d7e2329 : SUCCESS in 18m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/feb394a737ba4f4aafc1ba2adc79bee4 : SUCCESS in 17m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e7ae8729d21473793099de6c5467630 : SUCCESS in 17m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11a868a5d60842eb854f0ea09ee95730 : SUCCESS in 21m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d17adba4838493d90d7aeca782c094d : SUCCESS in 22m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1d41ad0753e4f699f3166affb4bb746 : SUCCESS in 28m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e87426396f4419c9874ee02da9f8cee : SUCCESS in 16m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e54543a76a604da9a6e91262337daf15 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c7e93684e43c4be480f49eba3fb9850f : SUCCESS in 19m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/44463667db6c4685aaf6f387114e6374 : SUCCESS in 27m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c6e9c1175f54adeb99e1b4d51596b7b : SUCCESS in 9m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/126c86ebf08e460c9d844fe299ad05c1 : SUCCESS in 59m 32s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5406d79a965f40059a988bf3616bd5b9 : SUCCESS in 56m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b434a02aa3974da083138862089f1e82 : SUCCESS in 49m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/39b8303f485a446184da349c9e9c5006 : SUCCESS in 50m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e60379925a81449db37d6af797fbd558 : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec54a3e456849bab5811617335110aa : SUCCESS in 7m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5903fd5db0224e039b9043fbe186b25d : SUCCESS in 28m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0587e1028af24d719e2ac455a58521bd : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":3},{"id":"84a8c67c858ffb140975507beb93fb6e8b0b3f5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 21:18:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"366b31a29d8650e381b80184c9936b629111b396","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 21:22:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293b0e41d4f94e9eab4f559be3aaebb0 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8e4c3e3c428c4dc483ffb05717caefe5 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2ed9986f0a574e4193292ab7dd12723c : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"439fbbdc7b2821a47d3fb62aec508c36e9e42606","revisions":{"86fff2ff4a12dd0df07dbf5861f2bfac797c9ee7":{"kind":"REWORK","_number":1,"created":"2016-07-07 10:08:00.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/19/338819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/338819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/338819/1"}}},"commit":{"parents":[{"commit":"561284e3d404d3744ce1c669efbbcec0c7050b0c","subject":"Merge \"Prevent down nodes failing PUTs with non-ascii obj names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/561284e3d404d3744ce1c669efbbcec0c7050b0c"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2016-07-06 00:27:31.000000000","tz":-420},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2016-07-07 10:06:32.000000000","tz":-420},"subject":"Ensure close socket for memcached if got timeout","message":"Ensure close socket for memcached if got timeout\n\nCurrentl MemcacheRing attempts to close socket when Timeout occured at\ncreating connection in MemcacheConnPool. However unfortunately it could\nnot because if Timeout happend, the sock variable is not assigned and\nNone value (initial assignment) will be feeded into _exception_occured\nmethod.\n\nThis patch added a unit test to confirm the socket.close() called and\nfixes the behavior with wrapping try/except in MemcacheConnPool.create\nto call close() method.\n\nChange-Id: I7959ab96682f4a459a99c4f92463e1280372f6ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86fff2ff4a12dd0df07dbf5861f2bfac797c9ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86fff2ff4a12dd0df07dbf5861f2bfac797c9ee7"}]},"branch":"refs/heads/master"},"9ca868eb16b69cddf75d22a23decaf63f4b633bf":{"kind":"REWORK","_number":2,"created":"2021-11-09 23:25:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/338819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/338819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/338819/2"}}},"commit":{"parents":[{"commit":"27478f07c3f45067a2644140b7da2b356ca9ac60","subject":"Merge \"proxy: Remove a bunch of logging translations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27478f07c3f45067a2644140b7da2b356ca9ac60"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2016-07-06 00:27:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-11-09 23:18:19.000000000","tz":-480},"subject":"Ensure close socket for memcached if got timeout","message":"Ensure close socket for memcached if got timeout\n\nCurrently MemcacheRing attempts to close socket when Timeout occured at\ncreating connection in MemcacheConnPool. However unfortunately it could\nnot because if Timeout happened, the sock variable is not assigned and\nNone value (initial assignment) will be fed into _exception_occured\nmethod.\n\nThis patch added a unit test to confirm the socket.close() called and\nfixes the behavior with wrapping try/except in MemcacheConnPool.create\nto call close() method.\n\nChange-Id: I7959ab96682f4a459a99c4f92463e1280372f6ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ca868eb16b69cddf75d22a23decaf63f4b633bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ca868eb16b69cddf75d22a23decaf63f4b633bf"}]},"branch":"refs/heads/master"},"439fbbdc7b2821a47d3fb62aec508c36e9e42606":{"kind":"REWORK","_number":3,"created":"2021-11-10 01:29:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/338819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/338819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/338819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/338819/3"}}},"commit":{"parents":[{"commit":"27478f07c3f45067a2644140b7da2b356ca9ac60","subject":"Merge \"proxy: Remove a bunch of logging translations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27478f07c3f45067a2644140b7da2b356ca9ac60"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2016-07-06 00:27:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-11-10 01:29:43.000000000","tz":0},"subject":"Ensure close socket for memcached if got timeout","message":"Ensure close socket for memcached if got timeout\n\nCurrently MemcacheRing attempts to close socket when Timeout occured at\ncreating connection in MemcacheConnPool. However unfortunately it could\nnot because if Timeout happened, the sock variable is not assigned and\nNone value (initial assignment) will be fed into _exception_occured\nmethod.\n\nThis patch added a unit test to confirm the socket.close() called and\nfixes the behavior with wrapping try/except in MemcacheConnPool.create\nto call close() method.\n\nChange-Id: I7959ab96682f4a459a99c4f92463e1280372f6ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/439fbbdc7b2821a47d3fb62aec508c36e9e42606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/439fbbdc7b2821a47d3fb62aec508c36e9e42606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
