)]}'
{"id":"openstack%2Fswift~38014","triplet_id":"openstack%2Fswift~master~Ib2343243f40194d5b2784551a807c7f58970a6e9","project":"openstack/swift","branch":"master","topic":"encode-always-new","hashtags":[],"change_id":"Ib2343243f40194d5b2784551a807c7f58970a6e9","subject":"Encode unicode from JSON before using it as a string.","status":"MERGED","created":"2013-07-20 03:09:16.000000000","updated":"2013-07-30 03:04:43.000000000","submitted":"2013-07-30 03:04:42.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1efb1ee257233e807bf9aed129181d027aecf2c","_number":38014,"virtual_id_number":38014,"owner":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},{"value":2,"date":"2013-07-30 03:04:42.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":0,"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},{"value":0,"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"value":1,"date":"2013-07-29 22:20:37.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":2649,"name":"Zhang Hua","email":"zhuadl@cn.ibm.com","username":"zhang-hare"},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"all":[{"value":0,"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":0,"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},{"value":0,"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-07-30 01:55:26.000000000","_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":2649,"name":"Zhang Hua","email":"zhuadl@cn.ibm.com","username":"zhang-hare"},{"value":2,"date":"2013-07-29 22:14:38.000000000","_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"all":[{"value":0,"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"value":0,"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},{"value":0,"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":1,"date":"2013-07-30 02:15:40.000000000","_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":2649,"name":"Zhang Hua","email":"zhuadl@cn.ibm.com","username":"zhang-hare"},{"value":0,"date":"2013-07-29 22:14:38.000000000","_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"_account_id":2649,"name":"Zhang Hua","email":"zhuadl@cn.ibm.com","username":"zhang-hare"},{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},{"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-22 05:41:17.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2013-07-25 14:27:36.000000000","updated_by":{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},"reviewer":{"_account_id":917,"name":"Chuck Thier","email":"cthier@gmail.com","username":"cthier"},"state":"REVIEWER"},{"updated":"2013-07-28 13:19:21.000000000","updated_by":{"_account_id":2649,"name":"Zhang Hua","email":"zhuadl@cn.ibm.com","username":"zhang-hare"},"reviewer":{"_account_id":2649,"name":"Zhang Hua","email":"zhuadl@cn.ibm.com","username":"zhang-hare"},"state":"REVIEWER"},{"updated":"2013-07-29 07:05:39.000000000","updated_by":{"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},"reviewer":{"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},"state":"REVIEWER"},{"updated":"2013-07-29 15:25:18.000000000","updated_by":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"reviewer":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"state":"REVIEWER"},{"updated":"2013-07-29 22:14:38.000000000","updated_by":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"reviewer":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"state":"REVIEWER"},{"updated":"2013-07-29 22:20:37.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-07-30 02:15:40.000000000","updated_by":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"reviewer":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"state":"REVIEWER"},{"updated":"2013-07-30 03:04:42.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"15eb9f64a6fda5f7e05c9fcbce5b5637d7ec2f86","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-20 03:09:47.000000000","message":"Patch Set 1:\n\nAs a note, I\u0027m not convinced this is the best way to do this, I\u0027d like some feedback to see if anyone has a better approach.","accounts_in_message":[],"_revision_number":1},{"id":"53b57639630815bf06b90deca8f53d5aaa0c987a","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-20 03:17:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0297acf7d8582ce21c46947134225e63f357de3f","author":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"date":"2013-07-20 03:20:11.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n","accounts_in_message":[],"_revision_number":2},{"id":"14a40e8afb861b1c725d0464b284ec01995e1a6b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-20 04:12:52.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/14/38014/2/check/gate-swift-pep8/3166 : SUCCESS in 1m 00s\n- http://docs-draft.openstack.org/14/38014/2/check/gate-swift-docs/2996/doc/build/html/ : SUCCESS in 1m 28s\n- http://logs.openstack.org/14/38014/2/check/gate-swift-python26/3242 : SUCCESS in 2m 24s\n- http://logs.openstack.org/14/38014/2/check/gate-swift-python27/3242 : SUCCESS in 1m 43s\n- http://logs.openstack.org/14/38014/2/check/gate-swift-devstack-vm-functional/77 : SUCCESS in 7m 28s (non-voting)\n- http://logs.openstack.org/14/38014/2/check/gate-tempest-devstack-vm-full/34039 : SUCCESS in 55m 37s\n- http://logs.openstack.org/14/38014/2/check/gate-tempest-devstack-vm-cells-full/1691 : FAILURE in 13m 50s (non-voting)\n- http://logs.openstack.org/14/38014/2/check/gate-tempest-devstack-vm-testr-full/542 : SUCCESS in 29m 56s (non-voting)\n- http://logs.openstack.org/14/38014/2/check/gate-tempest-devstack-vm-neutron/5825 : SUCCESS in 18m 53s (non-voting)\n- http://logs.openstack.org/14/38014/2/check/gate-tempest-devstack-vm-postgres-full/33729 : SUCCESS in 45m 08s\n- http://logs.openstack.org/14/38014/2/check/gate-grenade-devstack-vm/19759 : SUCCESS in 26m 31s\n","accounts_in_message":[],"_revision_number":2},{"id":"d56bc20fff925f94d9c63cbc01bad2cd9e741411","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2013-07-22 22:04:58.000000000","message":"Patch Set 2:\n\n@Peter: is there a particular reason you don\u0027t like this change? I didn\u0027t see any comments on your -1, inline or otherwise.\n\n@Alex: so you\u0027re saying that, if I took the simplejson import out of swift/common/utils.py, I\u0027d start seeing test failures due to the use of stdlib json?","accounts_in_message":[],"_revision_number":2},{"id":"5d5628fc8559912230f6de56a44a4f75d68519db","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-22 22:06:17.000000000","message":"Patch Set 2:\n\nYes, that\u0027s what I\u0027m saying.","accounts_in_message":[],"_revision_number":2},{"id":"1086d1e15ae72226b9efbcfc5211d9f647a0e0ad","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-23 01:04:10.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/102701\n","accounts_in_message":[],"_revision_number":2},{"id":"77c8a08295c8c5501eb35c8128f55b6d59b56652","author":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"date":"2013-07-23 01:50:10.000000000","message":"Patch Set 1: (1 inline comment)\n\nHere is what I was supposed to have submitted with that review, not sure why it did not get posted.","accounts_in_message":[],"_revision_number":1},{"id":"f81ef9955cf7cad6ee775e27a58f02022c21e6bb","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-23 22:43:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63b18a938f9fc087c694c5f59df878d9946c291c","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-23 23:28:27.000000000","message":"Patch Set 3: Works for me\n\nSmokeStack Results (patch set 3):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/103467\n","accounts_in_message":[],"_revision_number":3},{"id":"1fdf144ad3e22093e50a763bff037858a8e62517","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-23 23:53:41.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/14/38014/3/check/gate-swift-pep8/3222 : SUCCESS in 56s\n- http://docs-draft.openstack.org/14/38014/3/check/gate-swift-docs/3052/doc/build/html/ : SUCCESS in 1m 32s\n- http://logs.openstack.org/14/38014/3/check/gate-swift-python26/3297 : SUCCESS in 5m 21s\n- http://logs.openstack.org/14/38014/3/check/gate-swift-python27/3298 : SUCCESS in 1m 52s\n- http://logs.openstack.org/14/38014/3/check/gate-swift-devstack-vm-functional/115 : SUCCESS in 10m 14s (non-voting)\n- http://logs.openstack.org/14/38014/3/check/gate-tempest-devstack-vm-full/34924 : SUCCESS in 1h 00m 14s\n- http://logs.openstack.org/14/38014/3/check/gate-tempest-devstack-vm-cells-full/2394 : FAILURE in 16m 51s (non-voting)\n- http://logs.openstack.org/14/38014/3/check/gate-tempest-devstack-vm-testr-full/1237 : SUCCESS in 32m 37s (non-voting)\n- http://logs.openstack.org/14/38014/3/check/gate-tempest-devstack-vm-neutron/6809 : SUCCESS in 26m 20s (non-voting)\n- http://logs.openstack.org/14/38014/3/check/gate-tempest-devstack-vm-postgres-full/34614 : SUCCESS in 48m 54s\n- http://logs.openstack.org/14/38014/3/check/gate-grenade-devstack-vm/20577 : SUCCESS in 28m 23s\n","accounts_in_message":[],"_revision_number":3},{"id":"00b4352528389eeeb28b288104be609498030529","author":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"date":"2013-07-24 04:09:32.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"6ac4a675b482e47bef57a3a94b6aab58070da63f","author":{"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},"date":"2013-07-24 06:08:13.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"240af7dc49d3d4d1adceac6269efc655b973eb6d","author":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"date":"2013-07-26 21:27:36.000000000","message":"Patch Set 3: I would prefer that you didn\u0027t merge this\n\n(3 inline comments)\n\nlooks good, couple comments. Could you also add a unit test for this?  If we are thinking about switching around simplejson and json the more tests we have dealing with little/important details like this the better.","accounts_in_message":[],"_revision_number":3},{"id":"2e42ba44f2603ef5d2e5aa1b7709f8f1dd5d05dc","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-26 22:17:28.000000000","message":"Patch Set 3:\n\nSo, the existing tests already fail reliably if used with a different JSON library (such as the stdlib one that pypy uses), all you have to do is `pip uninstall simplejson` to reproduce it. It\u0027s not totally obvious to me how to test this condition exterally (I guess I could monkeypatch the module to use the stdlib json, in addition to whatever default tests it runs). Would you like me to pursue that avenue, or is the fact that it fails reliably under a different json encoder enough?","accounts_in_message":[],"_revision_number":3},{"id":"719ba13a82f7d3168f4a731ec0bf402d67a5ba3a","author":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"date":"2013-07-26 22:28:48.000000000","message":"Patch Set 3:\n\ni was hoping you could use something like mock.patch to mock out the json imported from swift.common.utils to maybe use the stdlib json. That should work right?  The only reason being that here would be a good place to have a test to make sure that both work even if the dev does have simplejson installed so that future code changes won\u0027t break all the non-simplejson folks as easily.","accounts_in_message":[],"_revision_number":3},{"id":"b6fe174a8ff2de4284d51720fd943483d4d45ab3","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-26 22:30:18.000000000","message":"Patch Set 3:\n\nYup, I can do that, new patch coming shortly!","accounts_in_message":[],"_revision_number":3},{"id":"4f2d3db320e6fde2ed62ae847e3857975829cf59","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-26 22:37:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"06c5c34972b1ccab65b648660da3ebb43e54c47e","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-26 23:00:14.000000000","message":"Patch Set 4: Works for me\n\nSmokeStack Results (patch set 4):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/104968\n\tLibvirt (Centos 6.4) Success: http://smokestack.openstack.org/?go\u003d/jobs/104969\n","accounts_in_message":[],"_revision_number":4},{"id":"5f37a44b9cd5655b9db26668f3ce2eb5c475c9b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-26 23:30:09.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/14/38014/4/check/gate-swift-pep8/3292 : SUCCESS in 1m 40s\n- http://docs-draft.openstack.org/14/38014/4/check/gate-swift-docs/3123/doc/build/html/ : SUCCESS in 1m 15s\n- http://logs.openstack.org/14/38014/4/check/gate-swift-python26/3370 : SUCCESS in 3m 49s\n- http://logs.openstack.org/14/38014/4/check/gate-swift-python27/3367 : SUCCESS in 1m 47s\n- http://logs.openstack.org/14/38014/4/check/gate-swift-devstack-vm-functional/158 : SUCCESS in 11m 39s (non-voting)\n- http://logs.openstack.org/14/38014/4/check/gate-tempest-devstack-vm-full/36182 : SUCCESS in 52m 48s\n- http://logs.openstack.org/14/38014/4/check/gate-tempest-devstack-vm-cells-full/3326 : FAILURE in 16m 16s (non-voting)\n- http://logs.openstack.org/14/38014/4/check/gate-tempest-devstack-vm-testr-full/2150 : FAILURE in 28m 59s (non-voting)\n- http://logs.openstack.org/14/38014/4/check/gate-tempest-devstack-vm-neutron/8214 : SUCCESS in 20m 40s (non-voting)\n- http://logs.openstack.org/14/38014/4/check/gate-tempest-devstack-vm-postgres-full/35864 : SUCCESS in 51m 26s\n- http://logs.openstack.org/14/38014/4/check/gate-grenade-devstack-vm/21708 : SUCCESS in 22m 29s\n","accounts_in_message":[],"_revision_number":4},{"id":"aabd456dd543d5474e2028c48e43def14d489f8d","author":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"date":"2013-07-27 01:41:05.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"d792bb537a8a772a45a9fdb6f529736b724cb4b7","author":{"_account_id":2649,"name":"Zhang Hua","email":"zhuadl@cn.ibm.com","username":"zhang-hare"},"date":"2013-07-28 13:19:21.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"f93ae0c67afe35a369e382e51472329aa9ad2ad7","author":{"_account_id":6852,"name":"Clifford Baeseman","username":"cbaeseman","inactive":true},"date":"2013-07-29 07:05:39.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"d3909805c78daa63a1b14cb1b9a309a839550afb","author":{"_account_id":995,"name":"David Goetz","email":"david.goetz@rackspace.com","username":"david-goetz"},"date":"2013-07-29 15:25:18.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"7e5a93409e825ca7942b035a5982dac3170d963c","author":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"date":"2013-07-29 21:57:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e492988649a2c3823e78efa9525f764a08e4da1d","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2013-07-29 22:14:38.000000000","message":"Patch Set 5: Looks good to me (core reviewer)\n\nWorks for now.\n\nIn the future, I want to look at replacing simplejson with ujson; the result of ujson.loads() matches json.loads(), while simplejson.loads() doesn\u0027t always (especially around non-ASCII strings, as we see here).","accounts_in_message":[],"_revision_number":5},{"id":"771c4d96db046b69093432300bc95d4edca5bce7","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-29 22:20:37.000000000","message":"Patch Set 5: Works for me\n\nSmokeStack Results (patch set 5):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/105811\n\tLibvirt (Centos 6.4) Success: http://smokestack.openstack.org/?go\u003d/jobs/105812\n","accounts_in_message":[],"_revision_number":5},{"id":"ba1556e2a30d031b3ab748cb9ff4c3c241c3cdf4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-29 22:57:15.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/14/38014/5/check/gate-swift-pep8/3322 : SUCCESS in 1m 43s\n- http://docs-draft.openstack.org/14/38014/5/check/gate-swift-docs/3153/doc/build/html/ : SUCCESS in 4m 07s\n- http://logs.openstack.org/14/38014/5/check/gate-swift-python26/3404 : SUCCESS in 3m 03s\n- http://logs.openstack.org/14/38014/5/check/gate-swift-python27/3397 : SUCCESS in 3m 10s\n- http://logs.openstack.org/14/38014/5/check/gate-swift-devstack-vm-functional/173 : SUCCESS in 11m 40s (non-voting)\n- http://logs.openstack.org/14/38014/5/check/gate-tempest-devstack-vm-full/37130 : SUCCESS in 45m 41s\n- http://logs.openstack.org/14/38014/5/check/gate-tempest-devstack-vm-cells-full/3844 : FAILURE in 19m 58s (non-voting)\n- http://logs.openstack.org/14/38014/5/check/gate-tempest-devstack-vm-testr-full/2661 : SUCCESS in 29m 59s (non-voting)\n- http://logs.openstack.org/14/38014/5/check/gate-tempest-devstack-vm-neutron/9204 : FAILURE in 18m 37s (non-voting)\n- http://logs.openstack.org/14/38014/5/check/gate-tempest-devstack-vm-postgres-full/36806 : SUCCESS in 58m 10s\n- http://logs.openstack.org/14/38014/5/check/gate-grenade-devstack-vm/22530 : SUCCESS in 25m 39s\n","accounts_in_message":[],"_revision_number":5},{"id":"0a8eaff8bb9cd88befd0d7120402a1f06b116e35","author":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"date":"2013-07-30 01:55:26.000000000","message":"Patch Set 5: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"251f501bfefba0d3edfd3195f1cab9b6ff4904d7","author":{"_account_id":6198,"name":"Peter Portante","email":"peter.a.portante@gmail.com","username":"peter-a-portante"},"date":"2013-07-30 02:15:40.000000000","message":"Patch Set 5: Approved\n\n\"Jim, I\u0027m a doctor, not a brick layer!\"","accounts_in_message":[],"_revision_number":5},{"id":"40bbebb5cc96e4e59365b5c63f81f82079ed616f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-30 02:15:41.000000000","message":"Patch Set 5: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"24dd362765dac579d3540905c06b079ed9167456","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-30 03:04:42.000000000","message":"Patch Set 5: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/14/38014/5/gate/gate-swift-docs/3161/doc/build/html/ : SUCCESS in 1m 14s\n- http://logs.openstack.org/14/38014/5/gate/gate-swift-pep8/3330 : SUCCESS in 1m 09s\n- http://logs.openstack.org/14/38014/5/gate/gate-swift-python26/3412 : SUCCESS in 3m 35s\n- http://logs.openstack.org/14/38014/5/gate/gate-swift-python27/3405 : SUCCESS in 1m 43s\n- http://logs.openstack.org/14/38014/5/gate/gate-tempest-devstack-vm-full/37208 : SUCCESS in 42m 10s\n- http://logs.openstack.org/14/38014/5/gate/gate-tempest-devstack-vm-neutron/9289 : FAILURE in 15m 37s (non-voting)\n- http://logs.openstack.org/14/38014/5/gate/gate-tempest-devstack-vm-postgres-full/36884 : SUCCESS in 48m 57s\n- http://logs.openstack.org/14/38014/5/gate/gate-grenade-devstack-vm/22598 : SUCCESS in 22m 29s\n","accounts_in_message":[],"_revision_number":5},{"id":"a4a9ffde59a9adece5c223fa3da3c36c711a8319","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-30 03:04:43.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"be688c31562a5789fba678a0675eff1040308202","revisions":{"1b276adb09bf00ddd0ad6675b6c4ffc2274e362e":{"kind":"REWORK","_number":1,"created":"2013-07-20 03:09:16.000000000","uploader":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"ref":"refs/changes/14/38014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/38014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/38014/1"}}},"commit":{"parents":[{"commit":"0fdad0d9d9e68b00f61171bb2a0dfd840ef5345f","subject":"Close SQLite cursors when creating functions.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdad0d9d9e68b00f61171bb2a0dfd840ef5345f"}]}],"author":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-20 03:07:27.000000000","tz":-420},"committer":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-20 03:07:27.000000000","tz":-420},"subject":"Encode unicode form JSON before using it as a string.","message":"Encode unicode form JSON before using it as a string.\n\nRight now this code fails when used with a JSON\ndecoder that always produces unicode. This isn\u0027t\nusually the case with CPython, where simplejson\nis used most of the time, however with the stdlib\nJSON library (as used on PyPy), this code fails.\n\nChange-Id: Ib2343243f40194d5b2784551a807c7f58970a6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b276adb09bf00ddd0ad6675b6c4ffc2274e362e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b276adb09bf00ddd0ad6675b6c4ffc2274e362e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdad0d9d9e68b00f61171bb2a0dfd840ef5345f","is_merged_in_target_branch":true,"change_id":"I39210616d323691ccb745149f24430a7a61382ec","change_number":37986,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b2063a15ec18abed2433a3cd38cf4ddd5899f22a":{"kind":"REWORK","_number":2,"created":"2013-07-20 03:17:09.000000000","uploader":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"ref":"refs/changes/14/38014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/38014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/38014/2"}}},"commit":{"parents":[{"commit":"0fdad0d9d9e68b00f61171bb2a0dfd840ef5345f","subject":"Close SQLite cursors when creating functions.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fdad0d9d9e68b00f61171bb2a0dfd840ef5345f"}]}],"author":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-20 03:07:27.000000000","tz":-420},"committer":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-20 03:17:00.000000000","tz":-420},"subject":"Encode unicode form JSON before using it as a string.","message":"Encode unicode form JSON before using it as a string.\n\nRight now this code fails when used with a JSON\ndecoder that always produces unicode. This isn\u0027t\nusually the case with CPython, where simplejson\nis used most of the time, however with the stdlib\nJSON library (as used on PyPy), this code fails.\n\nChange-Id: Ib2343243f40194d5b2784551a807c7f58970a6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2063a15ec18abed2433a3cd38cf4ddd5899f22a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2063a15ec18abed2433a3cd38cf4ddd5899f22a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdad0d9d9e68b00f61171bb2a0dfd840ef5345f","is_merged_in_target_branch":true,"change_id":"I39210616d323691ccb745149f24430a7a61382ec","change_number":37986,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6ecaf0bf983bea1581857cd8a2f558781165b4e5":{"kind":"REWORK","_number":3,"created":"2013-07-23 22:43:50.000000000","uploader":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"ref":"refs/changes/14/38014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/38014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/38014/3"}}},"commit":{"parents":[{"commit":"7a2b0fb000d99a3c692631b52b173bb0ffa30ea3","subject":"Merge \"Add \u0027Z\u0027 into isoformat for UTC time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a2b0fb000d99a3c692631b52b173bb0ffa30ea3"}]}],"author":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-20 03:07:27.000000000","tz":-420},"committer":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-23 22:42:16.000000000","tz":-420},"subject":"Encode unicode form JSON before using it as a string.","message":"Encode unicode form JSON before using it as a string.\n\nRight now this code fails when used with a JSON\ndecoder that always produces unicode. This isn\u0027t\nusually the case with CPython, where simplejson\nis used most of the time, however with the stdlib\nJSON library (as used on PyPy), this code fails.\n\nChange-Id: Ib2343243f40194d5b2784551a807c7f58970a6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ecaf0bf983bea1581857cd8a2f558781165b4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ecaf0bf983bea1581857cd8a2f558781165b4e5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7a2b0fb000d99a3c692631b52b173bb0ffa30ea3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"62b39fa12120e7ec45f5acf3d04b70c9a455ccaa":{"kind":"REWORK","_number":4,"created":"2013-07-26 22:37:16.000000000","uploader":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"ref":"refs/changes/14/38014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/38014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/38014/4"}}},"commit":{"parents":[{"commit":"741c3166d10d15a5975499f60151be580b51cf2a","subject":"fix name \u0027recon_container\u0027 to \u0027rcache\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/741c3166d10d15a5975499f60151be580b51cf2a"}]}],"author":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-20 03:07:27.000000000","tz":-420},"committer":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-26 22:36:29.000000000","tz":-420},"subject":"Encode unicode from JSON before using it as a string.","message":"Encode unicode from JSON before using it as a string.\n\nRight now this code fails when used with a JSON\ndecoder that always produces unicode. This isn\u0027t\nusually the case with CPython, where simplejson\nis used most of the time, however with the stdlib\nJSON library (as used on PyPy), this code fails.\n\nChange-Id: Ib2343243f40194d5b2784551a807c7f58970a6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62b39fa12120e7ec45f5acf3d04b70c9a455ccaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62b39fa12120e7ec45f5acf3d04b70c9a455ccaa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"741c3166d10d15a5975499f60151be580b51cf2a","is_merged_in_target_branch":true,"change_id":"I3a6e3d22ba1dbffc4309bc22ff37873b4a3f09b3","change_number":37390,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"be688c31562a5789fba678a0675eff1040308202":{"kind":"REWORK","_number":5,"created":"2013-07-29 21:57:35.000000000","uploader":{"_account_id":7680,"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","username":"alex_gaynor"},"ref":"refs/changes/14/38014/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/38014/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/38014/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/38014/5"}}},"commit":{"parents":[{"commit":"87ab2f6e8c771e2c88ae59783aa7c834b1300ce6","subject":"Merge \"Fix incorrect status handling at staticweb\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87ab2f6e8c771e2c88ae59783aa7c834b1300ce6"}]}],"author":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-20 03:07:27.000000000","tz":-420},"committer":{"name":"Alex Gaynor","email":"alex.gaynor@gmail.com","date":"2013-07-29 21:56:50.000000000","tz":-420},"subject":"Encode unicode from JSON before using it as a string.","message":"Encode unicode from JSON before using it as a string.\n\nRight now this code fails when used with a JSON\ndecoder that always produces unicode. This isn\u0027t\nusually the case with CPython, where simplejson\nis used most of the time, however with the stdlib\nJSON library (as used on PyPy), this code fails.\n\nChange-Id: Ib2343243f40194d5b2784551a807c7f58970a6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be688c31562a5789fba678a0675eff1040308202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be688c31562a5789fba678a0675eff1040308202"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"87ab2f6e8c771e2c88ae59783aa7c834b1300ce6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
