)]}'
{"id":"openstack%2Fswift~395582","triplet_id":"openstack%2Fswift~master~I4a4a8be9b631598ca1cd52c53885c68c3fbdfc4a","project":"openstack/swift","branch":"master","topic":"add_unit_test","hashtags":[],"change_id":"I4a4a8be9b631598ca1cd52c53885c68c3fbdfc4a","subject":"Handle deleted files with if-none-match","status":"MERGED","created":"2016-11-09 13:11:42.000000000","updated":"2016-11-24 08:11:28.000000000","submitted":"2016-11-24 08:11:28.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"095265cd57ad6f3a82a41123bd2eb51d2c74752f","_number":395582,"virtual_id_number":395582,"owner":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-11-24 08:11:28.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-11-23 02:38:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"value":2,"date":"2016-11-24 07:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":2,"date":"2016-11-23 02:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2016-11-23 02:38:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"value":1,"date":"2016-11-24 07:32:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-11-09 14:16:29.000000000","updated_by":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"reviewer":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"state":"REVIEWER"},{"updated":"2016-11-09 15:06:40.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 17:37:12.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2016-11-23 02:39:48.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2016-11-24 07:32:45.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2016-11-24 08:11:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"347b3571c7bef8b5caf81a11a31609030167f604","author":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"date":"2016-11-09 13:11:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8a16301174efd798f46a64f9805994d10f3e768","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-09 14:00:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/82/395582/1/check/gate-swift-docs-ubuntu-xenial/8abfd39//doc/build/html/ : SUCCESS in 2m 08s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-swift-pep8-ubuntu-xenial/808f269/ : SUCCESS in 1m 59s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-swift-python27-ubuntu-xenial/54189e5/ : SUCCESS in 3m 48s\n- gate-swift-python34 http://logs.openstack.org/82/395582/1/check/gate-swift-python34/c9d28e9/ : SUCCESS in 1m 15s\n- gate-swift-python35-nv http://logs.openstack.org/82/395582/1/check/gate-swift-python35-nv/3e59a27/ : SUCCESS in 1m 21s (non-voting)\n- gate-swift-tox-func-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-swift-tox-func-ubuntu-xenial/e62a702/ : SUCCESS in 3m 56s\n- gate-swift-tox-func-in-process-fast-post-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-swift-tox-func-in-process-fast-post-ubuntu-xenial/f7404ed/ : SUCCESS in 3m 30s\n- gate-swift-tox-func-in-process-encryption-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-swift-tox-func-in-process-encryption-ubuntu-xenial/0b8a77f/ : SUCCESS in 3m 29s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/82/395582/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/fcd7c35/ : SUCCESS in 46m 36s (non-voting)\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-grenade-dsvm-ubuntu-xenial/4345686/ : SUCCESS in 28m 11s\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-swift-dsvm-functional-ubuntu-xenial/e175d1d/ : SUCCESS in 18m 17s\n- gate-swift-tox-bandit-ubuntu-xenial http://logs.openstack.org/82/395582/1/check/gate-swift-tox-bandit-ubuntu-xenial/a2abc82/ : SUCCESS in 1m 31s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/82/395582/1/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/39cee2d/ : SUCCESS in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d69804140c8b8c0c5a827a13000851d241f81c3","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 14:01:04.000000000","message":"Patch Set 1:\n\nBuild succeeded \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/395582/4761/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/395582/6722/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/395582/5068/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"094e6a865c49a11de594e23fd50b483e09482977","author":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"date":"2016-11-09 14:16:01.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"488622f8eca67aa3504f9959741d7044c2ff18bb","author":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"date":"2016-11-09 14:16:29.000000000","message":"Patch Set 2: Code-Review+1\n\nUpdated commit message. Looks good.","accounts_in_message":[],"_revision_number":2},{"id":"e61d4ba4fd0e6bc71319b4119c6f342ee84d8608","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 15:06:40.000000000","message":"Patch Set 2:\n\nBuild succeeded \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/395582/4762/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/395582/6723/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/395582/5069/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"58a134552a9b8f8bf75d55394fa19d263dcc4298","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-09 15:20:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/82/395582/2/check/gate-swift-docs-ubuntu-xenial/67711bd//doc/build/html/ : SUCCESS in 2m 09s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-swift-pep8-ubuntu-xenial/5fa2438/ : SUCCESS in 1m 50s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-swift-python27-ubuntu-xenial/b7b657b/ : SUCCESS in 3m 33s\n- gate-swift-python34 http://logs.openstack.org/82/395582/2/check/gate-swift-python34/02b267f/ : SUCCESS in 2m 12s\n- gate-swift-python35-nv http://logs.openstack.org/82/395582/2/check/gate-swift-python35-nv/1c6b298/ : SUCCESS in 1m 21s (non-voting)\n- gate-swift-tox-func-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-swift-tox-func-ubuntu-xenial/c83f77b/ : SUCCESS in 3m 46s\n- gate-swift-tox-func-in-process-fast-post-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-swift-tox-func-in-process-fast-post-ubuntu-xenial/d39678e/ : SUCCESS in 3m 20s\n- gate-swift-tox-func-in-process-encryption-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-swift-tox-func-in-process-encryption-ubuntu-xenial/e81fa6b/ : SUCCESS in 3m 28s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/82/395582/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/ca4a46c/ : SUCCESS in 49m 29s (non-voting)\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-grenade-dsvm-ubuntu-xenial/66d92e4/ : SUCCESS in 29m 10s\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-swift-dsvm-functional-ubuntu-xenial/f705616/ : SUCCESS in 17m 31s\n- gate-swift-tox-bandit-ubuntu-xenial http://logs.openstack.org/82/395582/2/check/gate-swift-tox-bandit-ubuntu-xenial/61f95c5/ : SUCCESS in 1m 33s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/82/395582/2/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/b3b3a15/ : SUCCESS in 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b8808a4a4a33a6d562f02215b3bc7a0f7cab60d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2016-11-09 17:37:12.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\ngreat turn around - we should try and get this merged\n\nI\u0027m still clearing out emails, I\u0027ll check out this change shortly\n\nI wouldn\u0027t mind adding a unittest","accounts_in_message":[],"_revision_number":2},{"id":"e287992d53c7d20284fbcfaa38d8c39273426eaf","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2016-11-23 02:38:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"84e42fef8f846bff751dd83be6127bc8859014c4","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2016-11-23 02:39:48.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM. I tested on master and was able to easily verify the bug, also tested the functional test on master and saw it fail. I submitted a follow-on patch with unit tests, feel free to squash it you would like","accounts_in_message":[],"_revision_number":3},{"id":"52ef5cd5d300a8b4728f17f36ce7b53022383c40","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2016-11-23 03:14:17.000000000","message":"Patch Set 3:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/395582/5212/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"39b46122c6350b055c6965ebd529e278fb8ceaf6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-23 03:59:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/82/395582/3/check/gate-swift-docs-ubuntu-xenial/d32c14c//doc/build/html/ : SUCCESS in 2m 41s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-pep8-ubuntu-xenial/d7c3a49/ : SUCCESS in 2m 13s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-python27-ubuntu-xenial/8b40879/ : SUCCESS in 3m 38s\n- gate-swift-python34 http://logs.openstack.org/82/395582/3/check/gate-swift-python34/a0ca959/ : SUCCESS in 1m 16s\n- gate-swift-python35 http://logs.openstack.org/82/395582/3/check/gate-swift-python35/c3e5758/ : SUCCESS in 3m 15s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/b868d8b/ : SUCCESS in 3m 58s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/95a1a33/ : SUCCESS in 1m 34s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/083426a/ : SUCCESS in 3m 15s\n- gate-swift-tox-xfs-tmp-func-fast-post-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-tox-xfs-tmp-func-fast-post-ubuntu-xenial/c814256/ : SUCCESS in 3m 27s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/6525fa1/ : SUCCESS in 4m 21s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/82/395582/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/a2a228c/ : SUCCESS in 1h 12m 43s (non-voting)\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-grenade-dsvm-ubuntu-xenial/98e17ef/ : SUCCESS in 27m 50s\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-dsvm-functional-ubuntu-xenial/5bee12d/ : SUCCESS in 29m 51s\n- gate-swift-tox-xfs-tmp-bandit-ubuntu-xenial http://logs.openstack.org/82/395582/3/check/gate-swift-tox-xfs-tmp-bandit-ubuntu-xenial/9bc6459/ : SUCCESS in 2m 35s (non-voting)\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/82/395582/3/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/ef505bd/ : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c6b843c8808fa14822950d8820a8d275d7a96ff7","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2016-11-24 07:32:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":3},{"id":"6727d329d5866316ceb1bf370c20f140a890af14","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-24 07:32:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"979c276f8a70e607a9bd3aae5000b3e24b205f5e","date":"2016-11-24 08:11:28.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"98bf7ec9c80095dd3a7ca3555fdd35b5d881cc0b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-24 08:11:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/82/395582/3/gate/gate-swift-docs-ubuntu-xenial/0248649//doc/build/html/ : SUCCESS in 2m 31s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-pep8-ubuntu-xenial/30bec7a/ : SUCCESS in 1m 58s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-python27-ubuntu-xenial/e13bf59/ : SUCCESS in 3m 33s\n- gate-swift-python34 http://logs.openstack.org/82/395582/3/gate/gate-swift-python34/e98fe7d/ : SUCCESS in 1m 09s\n- gate-swift-python35 http://logs.openstack.org/82/395582/3/gate/gate-swift-python35/8cd72a5/ : SUCCESS in 1m 58s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/66f0724/ : SUCCESS in 3m 21s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/ceaa39d/ : SUCCESS in 1m 20s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/c85d50e/ : SUCCESS in 3m 17s\n- gate-swift-tox-xfs-tmp-func-fast-post-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-tox-xfs-tmp-func-fast-post-ubuntu-xenial/084b9d1/ : SUCCESS in 4m 43s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/0986179/ : SUCCESS in 3m 31s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-grenade-dsvm-ubuntu-xenial/0f9d325/ : SUCCESS in 26m 28s\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/82/395582/3/gate/gate-swift-dsvm-functional-ubuntu-xenial/76e6ceb/ : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2bdf61fadda655323a36185f921f8a3e183bcdf9","revisions":{"f43577acf0b3f56dfc31e1b6cab8dcb296f33aae":{"kind":"REWORK","_number":1,"created":"2016-11-09 13:11:42.000000000","uploader":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"ref":"refs/changes/82/395582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/395582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/395582/1"}}},"commit":{"parents":[{"commit":"a30351bdfc60f827979c5afe76ad40bf3a2b05c3","subject":"Merge \"Fix ZeroDivisionError in reconstructor.stats_line\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a30351bdfc60f827979c5afe76ad40bf3a2b05c3"}]}],"author":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2016-11-09 13:02:23.000000000","tz":60},"committer":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2016-11-09 13:11:40.000000000","tz":60},"subject":"Handle deleted files with if-none-match","message":"Handle deleted files with if-none-match\n\nWhen an object is deleted, we retrieve its metadata on PUT requests, which\nmade if-none-match requests fail while the object shouldn\u0027t be here. It\nseems we\u0027re only interested in the timestamp of the deleted object, so\nget that but forget the metadata.\n\nChange-Id: I4a4a8be9b631598ca1cd52c53885c68c3fbdfc4a\nCloses-Bug: #1640448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f43577acf0b3f56dfc31e1b6cab8dcb296f33aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f43577acf0b3f56dfc31e1b6cab8dcb296f33aae"}]},"branch":"refs/heads/master"},"ecf70489f8832f3d7ea49721965565bf682a6b2c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2016-11-09 14:16:01.000000000","uploader":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"ref":"refs/changes/82/395582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/395582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/395582/2"}}},"commit":{"parents":[{"commit":"a30351bdfc60f827979c5afe76ad40bf3a2b05c3","subject":"Merge \"Fix ZeroDivisionError in reconstructor.stats_line\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a30351bdfc60f827979c5afe76ad40bf3a2b05c3"}]}],"author":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2016-11-09 13:02:23.000000000","tz":60},"committer":{"name":"Mahati Chamarthy","email":"mahati.chamarthy@gmail.com","date":"2016-11-09 14:15:52.000000000","tz":0},"subject":"Handle deleted files with if-none-match","message":"Handle deleted files with if-none-match\n\nWhen an object is deleted, we retrieve its metadata on the next PUT\nrequests, which make if-none-match requests fail while the object\nshouldn\u0027t be here. It seems we\u0027re only interested in the timestamp of\nthe deleted object, so get that but forget the metadata.\n\nChange-Id: I4a4a8be9b631598ca1cd52c53885c68c3fbdfc4a\nCloses-Bug: #1640448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecf70489f8832f3d7ea49721965565bf682a6b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecf70489f8832f3d7ea49721965565bf682a6b2c"}]},"branch":"refs/heads/master"},"2bdf61fadda655323a36185f921f8a3e183bcdf9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-11-23 02:38:18.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/82/395582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/395582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/395582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/395582/3"}}},"commit":{"parents":[{"commit":"0873b7c03eddfa02c883a8b553abc6f2d5a26b1f","subject":"Merge \"Document access control lists (ACLs)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0873b7c03eddfa02c883a8b553abc6f2d5a26b1f"}]}],"author":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2016-11-09 13:02:23.000000000","tz":60},"committer":{"name":"Thiago da Silva","email":"thiago@redhat.com","date":"2016-11-23 02:06:24.000000000","tz":-300},"subject":"Handle deleted files with if-none-match","message":"Handle deleted files with if-none-match\n\nWhen an object is deleted, we retrieve its metadata on the next PUT\nrequests, which make if-none-match requests fail while the object\nshouldn\u0027t be here. It seems we\u0027re only interested in the timestamp of\nthe deleted object, so get that but forget the metadata.\n\nChange-Id: I4a4a8be9b631598ca1cd52c53885c68c3fbdfc4a\nCloses-Bug: #1640448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bdf61fadda655323a36185f921f8a3e183bcdf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bdf61fadda655323a36185f921f8a3e183bcdf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
