)]}'
{"id":"openstack%2Fswift~39740","triplet_id":"openstack%2Fswift~feature%2Fec~I5282fdb552aadeaca1ba85bd57c0ef23d5d80fb2","project":"openstack/swift","branch":"feature/ec","topic":"bug/1004114","hashtags":[],"change_id":"I5282fdb552aadeaca1ba85bd57c0ef23d5d80fb2","subject":"Merge master into ec feature branch","status":"MERGED","created":"2013-08-01 16:05:38.000000000","updated":"2013-08-01 17:05:52.000000000","submitted":"2013-08-01 17:05:52.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6429cfdbd6efc11203f893aaf1a212338eabe557","_number":39740,"virtual_id_number":39740,"owner":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-08-01 17:05:52.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-08-01 16:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-08-01 16:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-08-01 17:05:52.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"613c7ab18209ed77cb4ecbcb7f86961d2f2c8b73","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2013-08-01 16:06:05.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"8ff5c88630a10fe30a218a4b7d3fdf0dafbe1a38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-01 16:06:08.000000000","message":"Patch Set 1:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"30bb90d0fdca5877b5e35471c5bd4a53168ca259","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-01 16:52:13.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/40/39740/1/check/gate-swift-pep8/c6ccd9c : SUCCESS in 1m 51s\n- http://docs-draft.openstack.org/40/39740/1/check/gate-swift-docs/9c7dd57/doc/build/html/ : SUCCESS in 1m 46s\n- http://logs.openstack.org/40/39740/1/check/gate-swift-python26/6ec443a : SUCCESS in 3m 36s\n- http://logs.openstack.org/40/39740/1/check/gate-swift-python27/728240f : SUCCESS in 2m 16s\n- http://logs.openstack.org/40/39740/1/check/gate-swift-devstack-vm-functional/bc6a06a : SUCCESS in 11m 16s (non-voting)\n- http://logs.openstack.org/40/39740/1/check/gate-tempest-devstack-vm-full/781c532 : SUCCESS in 44m 10s\n- http://logs.openstack.org/40/39740/1/check/gate-tempest-devstack-vm-cells-full/7bb6cbd : FAILURE in 26m 34s (non-voting)\n- http://logs.openstack.org/40/39740/1/check/gate-tempest-devstack-vm-testr-full/ff6f948 : FAILURE in 27m 45s (non-voting)\n- http://logs.openstack.org/40/39740/1/check/gate-tempest-devstack-vm-neutron/1906b9f : FAILURE in 9m 15s (non-voting)\n- http://logs.openstack.org/40/39740/1/check/gate-tempest-devstack-vm-postgres-full/8eaa68a : SUCCESS in 46m 29s\n- http://logs.openstack.org/40/39740/1/check/gate-grenade-devstack-vm/c4e3422 : SUCCESS in 23m 05s\n","accounts_in_message":[],"_revision_number":1},{"id":"ce98d6256550ad99208fa0beccf024c7deb684c9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-01 17:05:52.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1},{"id":"671be404aa5a85795be3357befd8988136b77bbc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-01 17:05:52.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/40/39740/1/gate/gate-swift-docs/e7798d7/doc/build/html/ : SUCCESS in 1m 45s\n- http://logs.openstack.org/40/39740/1/gate/gate-swift-pep8/de8257a : SUCCESS in 1m 05s\n- http://logs.openstack.org/40/39740/1/gate/gate-swift-python26/7f813d4 : SUCCESS in 3m 09s\n- http://logs.openstack.org/40/39740/1/gate/gate-swift-python27/c7d75aa : SUCCESS in 1m 55s\n- http://logs.openstack.org/40/39740/1/gate/gate-tempest-devstack-vm-full/0c9b1f9 : SUCCESS in 44m 34s\n- http://logs.openstack.org/40/39740/1/gate/gate-tempest-devstack-vm-neutron/8897c91 : FAILURE in 9m 17s (non-voting)\n- http://logs.openstack.org/40/39740/1/gate/gate-tempest-devstack-vm-postgres-full/e177762 : SUCCESS in 59m 39s\n- http://logs.openstack.org/40/39740/1/gate/gate-grenade-devstack-vm/dc77022 : SUCCESS in 23m 59s\n","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a404cdea400e12d1ee8660c9f9477a87e419602","revisions":{"7a404cdea400e12d1ee8660c9f9477a87e419602":{"kind":"REWORK","_number":1,"created":"2013-08-01 16:05:38.000000000","uploader":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"ref":"refs/changes/40/39740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/39740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/39740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/39740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/39740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/39740/1"}}},"commit":{"parents":[{"commit":"f3ba55039c4caa91b71b453f2dc1f5310c2ac801","subject":"Merge master into ec branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3ba55039c4caa91b71b453f2dc1f5310c2ac801"}]},{"commit":"def2b850f340042e8648653a47666e8489f84ca1","subject":"Merge \"Return name of header larger than MAX_HEADER_SIZE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/def2b850f340042e8648653a47666e8489f84ca1"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2013-08-01 16:04:42.000000000","tz":-420},"committer":{"name":"John Dickinson","email":"me@not.mn","date":"2013-08-01 16:05:15.000000000","tz":-420},"subject":"Merge master into ec feature branch","message":"Merge master into ec feature branch\n\nChange-Id: I5282fdb552aadeaca1ba85bd57c0ef23d5d80fb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a404cdea400e12d1ee8660c9f9477a87e419602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a404cdea400e12d1ee8660c9f9477a87e419602"}]},"parents_data":[{"branch_name":"refs/heads/feature/ec","commit_id":"f3ba55039c4caa91b71b453f2dc1f5310c2ac801","is_merged_in_target_branch":true,"change_id":"Idf3b14a636580107c379771a1ad578ca9dacd200","change_number":38552,"patch_set_number":1,"change_status":"MERGED"},{"branch_name":"refs/heads/feature/ec","commit_id":"def2b850f340042e8648653a47666e8489f84ca1","is_merged_in_target_branch":false}],"branch":"refs/heads/feature/ec"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
