)]}'
{"id":"openstack%2Fswift~427911","triplet_id":"openstack%2Fswift~master~I64b0d8fdb2ffce786f56665a74ed7eb2603abfda","project":"openstack/swift","branch":"master","topic":"ecputpost","hashtags":[],"change_id":"I64b0d8fdb2ffce786f56665a74ed7eb2603abfda","subject":"Replace MIME with PUT+POST for EC and Encryption","status":"NEW","created":"2017-02-01 20:42:12.000000000","updated":"2024-10-07 23:46:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":187,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19a2000fb6fb131516f21c11eda92ee816c17566","_number":427911,"virtual_id_number":427911,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26240,"name":"libaoan","email":"baoan_008@163.com","username":"baoan"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-07 23:46:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26240,"name":"libaoan","email":"baoan_008@163.com","username":"baoan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26240,"name":"libaoan","email":"baoan_008@163.com","username":"baoan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26240,"name":"libaoan","email":"baoan_008@163.com","username":"baoan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-05-03 15:19:50.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2017-09-09 04:36:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-12-21 07:45:09.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-19 14:57:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2018-06-27 21:32:35.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2018-08-02 20:50:49.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2018-08-03 06:27:10.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2018-08-12 23:33:51.000000000","updated_by":{"_account_id":26240,"name":"libaoan","email":"baoan_008@163.com","username":"baoan"},"reviewer":{"_account_id":26240,"name":"libaoan","email":"baoan_008@163.com","username":"baoan"},"state":"REVIEWER"},{"updated":"2018-09-10 22:38:27.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2018-09-11 05:16:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2018-10-19 12:53:10.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2019-05-03 22:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59113386e141438bb14ec1131f234395b77d9d79","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-02-01 20:42:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c46fdc22a5e3443b007fe9946cabaca3c8c2a6f","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-02-01 21:29:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5748/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5414/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7380/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e17093fdd76a120cc161d37d2e57e30f407fed24","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-01 21:43:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/1/check/gate-swift-docs-ubuntu-xenial/bf60488//doc/build/html/ : SUCCESS in 2m 58s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-pep8-ubuntu-xenial/e04a3c1/ : FAILURE in 2m 39s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-python27-ubuntu-xenial/199ad22/ : FAILURE in 3m 11s\n- gate-swift-python35 http://logs.openstack.org/11/427911/1/check/gate-swift-python35/99a1bfc/ : SUCCESS in 1m 27s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/4d1cb27/ : FAILURE in 3m 08s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/972c4d9/ : SUCCESS in 2m 25s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/45346af/ : SUCCESS in 3m 12s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/e5ce8c9/ : SUCCESS in 4m 07s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/7fdfb55/ : SUCCESS in 4m 07s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/54b747d/ : SUCCESS in 1h 00m 07s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/1/check/gate-swift-dsvm-functional-ubuntu-xenial/db50a5e/ : SUCCESS in 32m 35s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/1/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/c4b4414/ : SUCCESS in 51m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cebc4f2009f5c91a67d19103054b32c6c1ae8589","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-02-02 03:50:09.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nI don\u0027t think I really grok this yet - but keep going!","accounts_in_message":[],"_revision_number":1},{"id":"3405675bc953402c5b3e99ccad1d4f8c0526dec0","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-02-04 03:07:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b67aab673285b38cce41ca215eb8bd8a06c4841","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-02-04 03:54:14.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5765/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5431/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7397/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"29bec840b2925f842a95571a1271b76412981655","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-04 04:56:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/2/check/gate-swift-docs-ubuntu-xenial/6f13f64//doc/build/html/ : SUCCESS in 1m 45s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-pep8-ubuntu-xenial/52f498e/ : FAILURE in 3m 09s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-python27-ubuntu-xenial/80818d9/ : FAILURE in 7m 07s\n- gate-swift-python35 http://logs.openstack.org/11/427911/2/check/gate-swift-python35/d1bfd4c/ : SUCCESS in 58s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/38ed5f3/ : FAILURE in 4m 50s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/fe6ca51/ : SUCCESS in 1m 05s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/73dd2c4/ : SUCCESS in 3m 29s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/4ef93f6/ : SUCCESS in 4m 22s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/18927fc/ : SUCCESS in 5m 26s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/1d00ec1/ : SUCCESS in 1h 48m 24s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/2/check/gate-swift-dsvm-functional-ubuntu-xenial/56eeb54/ : SUCCESS in 19m 49s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/2/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/a06b96c/ : SUCCESS in 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"988e6c0ef04966b52229d270d580bd80bdbbe473","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-02-11 04:00:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee8531dffe5eadb9e46cc28299817c9ebc16da87","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-02-11 04:47:12.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5816/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5482/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7448/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"46d6034c11e690abe976432b440de8290fb5b9d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-11 05:03:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/3/check/gate-swift-docs-ubuntu-xenial/47bfe17//doc/build/html/ : SUCCESS in 1m 44s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-pep8-ubuntu-xenial/010714d/ : SUCCESS in 2m 12s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-python27-ubuntu-xenial/7dbaa1d/ : FAILURE in 4m 52s\n- gate-swift-python35 http://logs.openstack.org/11/427911/3/check/gate-swift-python35/9a1bdaf/ : SUCCESS in 1m 48s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/413e4f4/ : FAILURE in 2m 52s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/46cd41c/ : SUCCESS in 1m 15s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/34c136a/ : SUCCESS in 3m 01s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/ac97ffc/ : SUCCESS in 3m 01s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/94de832/ : SUCCESS in 5m 09s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/af1b54c/ : SUCCESS in 1h 02m 47s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/3/check/gate-swift-dsvm-functional-ubuntu-xenial/18c0c6c/ : SUCCESS in 20m 09s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/3/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/3ae3ac5/ : SUCCESS in 19m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"91ed5aec994b90edb1431646bfdf7663e77fa4d9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-02-16 05:04:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8100999dd779f63d1993c29a0206e7b85c475c5a","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-02-16 05:45:10.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5854/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/5519/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7485/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0bffe5e9fea7b7fb1c8f05f0a7bb01c18a2cc418","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-16 06:15:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/4/check/gate-swift-docs-ubuntu-xenial/ca557f9//doc/build/html/ : SUCCESS in 3m 35s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-pep8-ubuntu-xenial/423eee8/ : SUCCESS in 2m 15s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-python27-ubuntu-xenial/d2af7bd/ : SUCCESS in 3m 14s\n- gate-swift-python35 http://logs.openstack.org/11/427911/4/check/gate-swift-python35/f41f83e/ : SUCCESS in 1m 03s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/d81c6dc/ : SUCCESS in 5m 59s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/878222a/ : SUCCESS in 1m 04s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/e32b3be/ : SUCCESS in 4m 27s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/6d13cde/ : SUCCESS in 4m 53s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/e98f342/ : SUCCESS in 4m 33s\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/ff82607/ : SUCCESS in 1h 10m 07s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/4/check/gate-swift-dsvm-functional-ubuntu-xenial/0253250/ : SUCCESS in 20m 37s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/4/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/4484bd9/ : SUCCESS in 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"76ad8974791ef5a97e0d60bd1c522322adefb66a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-26 06:28:28.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"64dfc5815c977e403ce85db1a04fb1d901f334ab","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-04-20 21:30:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b4ee5d326db4fdad20547add328345e6bb76acc5","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-04-20 22:45:17.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6099/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6436/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8069/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"d85cb25ede3db92755d060b7bce96c8dd51b3862","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-20 22:52:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/5/check/gate-swift-docs-ubuntu-xenial/5872bf4//doc/build/html/ : SUCCESS in 2m 59s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-pep8-ubuntu-xenial/165f699/ : SUCCESS in 3m 25s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-python27-ubuntu-xenial/3ab9ee6/ : FAILURE in 3m 46s\n- gate-swift-python35 http://logs.openstack.org/11/427911/5/check/gate-swift-python35/36835b7/ : SUCCESS in 1m 56s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/312e503/ : SUCCESS in 1h 17m 49s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-grenade-dsvm-neutron-ubuntu-xenial/37a277d/ : SUCCESS in 54m 04s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/f381015/ : FAILURE in 4m 00s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/1c97c90/ : SUCCESS in 1m 53s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/a082be0/ : SUCCESS in 4m 47s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/5fc15bc/ : SUCCESS in 3m 35s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/87b2393/ : SUCCESS in 3m 34s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/8549993/ : FAILURE in 16m 21s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/6a83e76/ : SUCCESS in 1h 19m 54s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/5/check/gate-swift-dsvm-functional-ubuntu-xenial/3941df8/ : SUCCESS in 28m 04s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/5/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/5a0b885/ : SUCCESS in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd82c447461c6ce389a927044cfb63a00dd6afc3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-04-24 01:49:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"568d672b5452615d575c943e4af82db8e22ff4f9","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-04-24 02:33:55.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6108/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8078/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6445/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"465d1d140ef88a3985368e30c4cc08998dccee7e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-24 02:57:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/6/check/gate-swift-docs-ubuntu-xenial/285ada3//doc/build/html/ : SUCCESS in 3m 27s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-pep8-ubuntu-xenial/3bc776e/ : SUCCESS in 2m 05s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-python27-ubuntu-xenial/59d0e11/ : SUCCESS in 5m 54s\n- gate-swift-python35 http://logs.openstack.org/11/427911/6/check/gate-swift-python35/8f33cfc/ : SUCCESS in 2m 30s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/aadc1a2/ : SUCCESS in 1h 07m 11s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/107bb6e/ : SUCCESS in 32m 41s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/3a9beb7/ : SUCCESS in 3m 20s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/7f0e2d4/ : SUCCESS in 59s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/54fa54a/ : SUCCESS in 3m 09s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/3dc5806/ : SUCCESS in 3m 10s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/4dd6621/ : SUCCESS in 3m 12s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/f913e06/ : SUCCESS in 3m 22s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/6/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/dada627/ : SUCCESS in 1h 03m 37s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/6/check/gate-swift-dsvm-functional-ubuntu-xenial/9caa923/ : SUCCESS in 18m 17s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/6/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/abd8e09/ : SUCCESS in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"33338b55fce0ea44439a95b464d51720c88440a8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-04-28 00:43:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"45f5f904cf98e23f273eaf2342427c74fee50f5c","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-04-28 01:26:18.000000000","message":"Patch Set 7:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6492/probetests/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8128/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6153/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"abb5c13a917bd94d947d7a14aeebb7d32291a9ca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-28 02:01:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/7/check/gate-swift-docs-ubuntu-xenial/0d16bc1//doc/build/html/ : SUCCESS in 3m 10s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-pep8-ubuntu-xenial/8eae01b/ : SUCCESS in 2m 14s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-python27-ubuntu-xenial/b8285a0/ : FAILURE in 3m 30s\n- gate-swift-python35 http://logs.openstack.org/11/427911/7/check/gate-swift-python35/e8360af/ : SUCCESS in 1m 05s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/2e63d8a/ : SUCCESS in 1h 04m 53s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/485a91a/ : SUCCESS in 48m 53s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/c420f8d/ : FAILURE in 3m 16s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/1de1c5a/ : SUCCESS in 1m 03s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/e2b660c/ : SUCCESS in 3m 21s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/c670eaa/ : SUCCESS in 3m 14s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/60e821c/ : SUCCESS in 5m 28s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/43ef4f9/ : SUCCESS in 5m 00s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/7/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/5c505d8/ : SUCCESS in 1h 16m 21s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/7/check/gate-swift-dsvm-functional-ubuntu-xenial/93e040a/ : SUCCESS in 19m 39s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/7/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/c1f8080/ : SUCCESS in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"65b596de9083895ff78f02accd1e9f47080c630f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-05-02 07:46:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5eb8cfcf816c5b6504a4ee74bd27e71615b4e99e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-02 08:53:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/8/check/gate-swift-docs-ubuntu-xenial/03d66d8//doc/build/html/ : SUCCESS in 3m 29s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-pep8-ubuntu-xenial/52118bd/ : SUCCESS in 3m 21s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-python27-ubuntu-xenial/593d830/ : SUCCESS in 5m 23s\n- gate-swift-python35 http://logs.openstack.org/11/427911/8/check/gate-swift-python35/ab5f179/ : SUCCESS in 1m 46s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6ead061/ : SUCCESS in 1h 05m 47s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-grenade-dsvm-neutron-ubuntu-xenial/01780cc/ : SUCCESS in 34m 40s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/30c4034/ : SUCCESS in 5m 20s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/f2e0186/ : SUCCESS in 1m 18s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/1c23ffc/ : SUCCESS in 3m 54s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/aeadacb/ : SUCCESS in 5m 40s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/87a551e/ : SUCCESS in 4m 23s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/69d83ee/ : SUCCESS in 4m 43s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/8/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/0e6bde5/ : FAILURE in 59m 38s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/8/check/gate-swift-dsvm-functional-ubuntu-xenial/27e237a/ : SUCCESS in 19m 50s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/8/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/e566e3c/ : SUCCESS in 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6312fd493eb1babad6169d6af35ed431a710f2ee","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-05-02 08:55:38.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6514/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8130/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6155/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"8898ed7e81094158d38b5056b6697345f6712ea5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-05-02 17:16:37.000000000","message":"Patch Set 8:\n\n(19 comments)\n\nQuick first pass. What (if anything) does this mean for encryption? What does that back-and-forth look like, both for replicated and for EC?\n\nDo we still have the proxy \u003c-\u003e object ETag validation from https://github.com/openstack/swift/blob/2.14.0/swift/common/middleware/crypto/encrypter.py#L120-L121 ? If we wanted to add something similar for EC (and I\u0027ve been meaning to find the time...) will it be as simple as it is on master, where we just set it in footers?","accounts_in_message":[],"_revision_number":8},{"id":"c0b267d3eed42ceed4607a7a41993714a7a727ba","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-05-03 21:12:18.000000000","message":"Patch Set 8:\n\n(4 comments)\n\n(answered some of Tim\u0027s comments inline -- thanks a lot for reviewing -- but not all, not those that required code changes)","accounts_in_message":[],"_revision_number":8},{"id":"9c98977a75d350825c281216b4e93da4986b1b00","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-05-04 03:29:09.000000000","message":"Patch Set 8:\n\n(9 comments)\n\n(inline)","accounts_in_message":[],"_revision_number":8},{"id":"56d145c0a1bd98fe5abd58a2c17ec48d8c1a32e1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-12 21:27:19.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"043c132f34c41325dfd273c3c3869734eaa5156e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-05-17 22:28:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"318e4d1cc6fe3110c3685de4c669b02bd5a8c8bd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-18 00:15:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/9/check/gate-swift-docs-ubuntu-xenial/4b068b0//doc/build/html/ : SUCCESS in 4m 18s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-pep8-ubuntu-xenial/d8abf5c/ : SUCCESS in 5m 27s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-python27-ubuntu-xenial/5c661d2/ : SUCCESS in 6m 55s\n- gate-swift-python35 http://logs.openstack.org/11/427911/9/check/gate-swift-python35/7c5c787/ : SUCCESS in 2m 59s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/56f706c/ : SUCCESS in 1h 28m 29s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-grenade-dsvm-neutron-ubuntu-xenial/1a6e75a/ : SUCCESS in 55m 15s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/5d76c4b/ : SUCCESS in 6m 46s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/06765a4/ : SUCCESS in 3m 36s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/3807689/ : SUCCESS in 6m 06s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/1dbf7e9/ : SUCCESS in 6m 15s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/fad1365/ : SUCCESS in 6m 04s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/f291073/ : SUCCESS in 6m 22s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/11/427911/9/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/613514a/ : SUCCESS in 1h 35m 52s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/9/check/gate-swift-dsvm-functional-ubuntu-xenial/d37c3d9/ : SUCCESS in 36m 04s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/9/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/daf9900/ : SUCCESS in 36m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8079d4038e1854907fe3ebd25b475bcaf92e5241","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-05-18 02:07:00.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6634/probetests/ : ABORTED\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8250/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6273/ : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"d530d7d387064ca77d05e0a9940295fa59989c53","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-05-24 01:52:04.000000000","message":"Patch Set 8:\n\n(1 comment)\n\n\"Goodness gratious, not that nightmare again\" caught my eye :)\n\nYou\u0027re doing great Pete - this is a killer diff!\n\nI love all the XXX :P","accounts_in_message":[],"_revision_number":8},{"id":"bcbeb95db08cb60554b98b2f17c54c1c03947afd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 18:46:22.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"5cf24967322020c985c4b7d768b04406966ebaf9","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2017-06-30 13:23:25.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e63e8d024fb3ef5365d8b5bc9b4c607e8fc0675d","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2017-06-30 13:23:58.000000000","message":"Patch Set 10:\n\npatchset10 was just a rebase","accounts_in_message":[],"_revision_number":10},{"id":"f56726bc5eb5713a111a0375599607d495b440f8","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-06-30 14:06:45.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6966/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8580/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6602/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"9041e4c9289c5739c742bd3dde1c8f315c3f6fd4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-30 15:03:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/10/check/gate-swift-docs-ubuntu-xenial/759085f//doc/build/html/ : SUCCESS in 3m 29s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-pep8-ubuntu-xenial/17ed0dc/ : SUCCESS in 2m 30s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-python27-ubuntu-xenial/a8d633b/ : SUCCESS in 3m 37s\n- gate-swift-python35 http://logs.openstack.org/11/427911/10/check/gate-swift-python35/1ca4fc7/ : SUCCESS in 1m 25s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/59d7955/ : SUCCESS in 1h 38m 08s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-grenade-dsvm-neutron-ubuntu-xenial/dcdd636/ : SUCCESS in 49m 14s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/1ca33e9/ : SUCCESS in 4m 03s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/cc75eed/ : SUCCESS in 2m 06s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/e16b9f6/ : SUCCESS in 6m 47s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/338d502/ : SUCCESS in 4m 12s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/0f00c99/ : SUCCESS in 3m 35s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/5e796bb/ : SUCCESS in 7m 12s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/10/check/gate-swift-dsvm-functional-ubuntu-xenial/eca9a24/ : SUCCESS in 17m 52s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/10/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/e74ba7e/ : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"913a0f401a0e06798101d40a992ec25f4b245214","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-05 14:52:58.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"7ad6815e46a6e0404c284a201dec26cd0f1c469a","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-07-07 20:23:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"de5feb7e12e180406b9008070d294c783a3f2b72","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-07-07 21:07:10.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7003/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8617/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6639/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"4c5856c5a0ba0802dea28ffe6f5bad3cb6400866","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-07 21:52:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/11/check/gate-swift-docs-ubuntu-xenial/c1ef768//doc/build/html/ : SUCCESS in 3m 07s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-pep8-ubuntu-xenial/4315827/ : SUCCESS in 3m 35s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-python27-ubuntu-xenial/6e9085d/ : SUCCESS in 5m 48s\n- gate-swift-python35 http://logs.openstack.org/11/427911/11/check/gate-swift-python35/be80504/ : SUCCESS in 1m 57s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6db1f6e/ : SUCCESS in 1h 29m 00s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-grenade-dsvm-neutron-ubuntu-xenial/731cdc4/ : SUCCESS in 35m 20s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/8e7059b/ : SUCCESS in 3m 33s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/cfb0656/ : SUCCESS in 1m 24s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/b7f6c72/ : SUCCESS in 3m 23s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/f968b02/ : SUCCESS in 3m 27s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/15c4763/ : SUCCESS in 3m 29s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/e574f66/ : SUCCESS in 4m 59s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/11/check/gate-swift-dsvm-functional-ubuntu-xenial/ae699bb/ : SUCCESS in 18m 42s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/11/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/ee75cea/ : SUCCESS in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8368266c22ddf5a8c97d94f55e4b1e5534a8886f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-11 05:37:25.000000000","message":"Patch Set 11: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"95d420e1b865a6bc618512b30a2920b5859ebaed","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-07-12 00:00:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"51271cd366da46bd842eff8b92defd5f370717d9","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-07-12 00:43:43.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7022/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8636/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6658/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"07c2ea014e7177f8a0a90f1106420f2c9ce63fb8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-12 01:20:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/12/check/gate-swift-docs-ubuntu-xenial/47abf32//doc/build/html/ : SUCCESS in 3m 30s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-pep8-ubuntu-xenial/0cc5190/ : SUCCESS in 3m 35s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-python27-ubuntu-xenial/22f1747/ : SUCCESS in 5m 34s\n- gate-swift-python35 http://logs.openstack.org/11/427911/12/check/gate-swift-python35/b64a1aa/ : SUCCESS in 3m 04s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8029dba/ : SUCCESS in 1h 19m 13s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3b36413/ : SUCCESS in 38m 08s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/50502b7/ : SUCCESS in 5m 29s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/07d00d3/ : SUCCESS in 1m 08s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/f3f560b/ : SUCCESS in 5m 07s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/efd90bd/ : SUCCESS in 5m 15s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/d2fc422/ : SUCCESS in 4m 16s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/f45fac6/ : SUCCESS in 5m 30s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/12/check/gate-swift-dsvm-functional-ubuntu-xenial/e3c9fd4/ : SUCCESS in 18m 27s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/12/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/5d7423a/ : SUCCESS in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fbd95314378100564f87bfd312206d7f61240f9c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-25 23:18:05.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"24ec1441061198552105690f26267e41d3fbee80","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-07-27 00:16:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0bc0144cc5dd461e3091f00b77cfc4c886f45051","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-27 00:49:35.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/13/check/gate-swift-docs-ubuntu-xenial/1731d92//doc/build/html/ : SUCCESS in 2m 17s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-pep8-ubuntu-xenial/3a97718/ : SUCCESS in 4m 28s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-python27-ubuntu-xenial/66325a0/ : SUCCESS in 3m 30s\n- gate-swift-python35 http://logs.openstack.org/11/427911/13/check/gate-swift-python35/533288a/ : SUCCESS in 1m 58s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a9f9735/ : FAILURE in 22m 20s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-grenade-dsvm-neutron-ubuntu-xenial/524470a/ : FAILURE in 28m 45s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/bb22943/ : SUCCESS in 3m 46s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/d1cf960/ : SUCCESS in 2m 39s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/118c9f5/ : SUCCESS in 5m 11s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/3ef3743/ : SUCCESS in 3m 45s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/9b38081/ : SUCCESS in 4m 18s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/0409f0c/ : SUCCESS in 8m 44s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/13/check/gate-swift-dsvm-functional-ubuntu-xenial/25e5be8/ : FAILURE in 30m 40s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/13/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/37fc8f3/ : FAILURE in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"04a77cbe19a48a9a85b7fac69fd7c8b60ab60cb4","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-07-27 01:03:42.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7142/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8756/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6778/ : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"5654a136ce3c47cf95497043c9ad574432a671fb","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-09-02 05:18:33.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2de61ed5d3a76e5e1b09a7a2f58f966c6e29974e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-02 06:00:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/14/check/gate-swift-docs-ubuntu-xenial/c9637a0//doc/build/html/ : SUCCESS in 3m 05s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-pep8-ubuntu-xenial/ac5a728/ : SUCCESS in 3m 48s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-python27-ubuntu-xenial/284d8af/ : SUCCESS in 5m 48s\n- gate-swift-python35 http://logs.openstack.org/11/427911/14/check/gate-swift-python35/bbe576c/ : SUCCESS in 1m 27s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e2c711e/ : FAILURE in 25m 26s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ac6f412/ : FAILURE in 41m 37s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/8cf537a/ : SUCCESS in 5m 48s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/439adb4/ : SUCCESS in 2m 13s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/1a7c211/ : SUCCESS in 4m 51s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/34b7fd9/ : SUCCESS in 6m 35s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/d9bde88/ : SUCCESS in 4m 43s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/ad41f8c/ : SUCCESS in 4m 48s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/14/check/gate-swift-dsvm-functional-ubuntu-xenial/262e1dd/ : FAILURE in 31m 50s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/14/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/1cb6a14/ : FAILURE in 38m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c979452dcdabbae0f544ddc2021cb93782d33051","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-02 06:04:25.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6964/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8942/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7328/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"4bce7df40b8f0e1557c13e1ade6ca45a3b26c6c8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-09-03 23:08:55.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"16cf06e4c1f5538cde2b94390b17960cd1f0afc1","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-04 00:01:51.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7331/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6967/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8945/ : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"063e9d16ec22948a57c28078b5f38dd5b2ec4869","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-04 00:40:32.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/15/check/gate-swift-docs-ubuntu-xenial/2b57823//doc/build/html/ : SUCCESS in 3m 12s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-pep8-ubuntu-xenial/7c9a332/ : SUCCESS in 3m 57s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-python27-ubuntu-xenial/0bdf74c/ : SUCCESS in 6m 25s\n- gate-swift-python35 http://logs.openstack.org/11/427911/15/check/gate-swift-python35/81f8f1b/ : SUCCESS in 4m 34s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e32492f/ : SUCCESS in 1h 31m 20s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c854eab/ : SUCCESS in 48m 55s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/09c9076/ : SUCCESS in 6m 00s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/528f255/ : SUCCESS in 1m 44s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/6db913c/ : SUCCESS in 4m 30s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/7c72ca1/ : SUCCESS in 5m 02s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/f7fea07/ : SUCCESS in 3m 58s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/94d8863/ : SUCCESS in 4m 37s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/15/check/gate-swift-dsvm-functional-ubuntu-xenial/957d4b4/ : SUCCESS in 20m 25s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/15/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/c38b0bc/ : SUCCESS in 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a6d0f64ce1d17a41df4152c626c2855eaff2af1e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-09-05 03:37:40.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2da97ab636ff1010624d6b4b3d97def2ab9228a2","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-05 04:22:50.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7332/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/6968/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8946/ : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"e572cae97c6e4a9d76dcca340b895335d408caf9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-05 05:12:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/16/check/gate-swift-docs-ubuntu-xenial/256b45d//doc/build/html/ : SUCCESS in 3m 00s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-pep8-ubuntu-xenial/1d659bc/ : SUCCESS in 2m 59s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-python27-ubuntu-xenial/b344b05/ : SUCCESS in 4m 50s\n- gate-swift-python35 http://logs.openstack.org/11/427911/16/check/gate-swift-python35/c9f7034/ : SUCCESS in 1m 18s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/5531666/ : SUCCESS in 1h 34m 19s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3341df8/ : SUCCESS in 50m 37s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/377dc23/ : SUCCESS in 5m 18s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/c879346/ : SUCCESS in 1m 30s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/254fece/ : SUCCESS in 4m 10s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/36c7961/ : SUCCESS in 4m 08s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/6888b0f/ : SUCCESS in 4m 43s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/1343050/ : SUCCESS in 4m 56s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/16/check/gate-swift-dsvm-functional-ubuntu-xenial/182b4ed/ : SUCCESS in 23m 43s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/16/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/8d7b312/ : SUCCESS in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7934163e501862647bab9282f2b15458f03598f7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-09-07 23:32:53.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"e672c67e74fcc56d76ee5f1fb7442b624d6c668e","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-08 00:16:39.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7374/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7010/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8988/ : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"9500aa2b43d7ee4d4e2775bdb73af1dbb64375fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-08 01:01:55.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/17/check/gate-swift-docs-ubuntu-xenial/d048a80//doc/build/html/ : SUCCESS in 3m 16s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-pep8-ubuntu-xenial/c586f09/ : SUCCESS in 3m 10s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-python27-ubuntu-xenial/3b56622/ : SUCCESS in 5m 20s\n- gate-swift-python35 http://logs.openstack.org/11/427911/17/check/gate-swift-python35/a4a2714/ : SUCCESS in 2m 18s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/09e0cae/ : SUCCESS in 1h 28m 41s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-grenade-dsvm-neutron-ubuntu-xenial/2c7958a/ : SUCCESS in 54m 18s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/bb038b4/ : SUCCESS in 5m 13s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/e00e14b/ : SUCCESS in 1m 43s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/8143798/ : SUCCESS in 4m 28s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/02a7264/ : SUCCESS in 4m 42s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/e9ddd70/ : SUCCESS in 4m 22s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/955cc8f/ : SUCCESS in 4m 52s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/17/check/gate-swift-dsvm-functional-ubuntu-xenial/0797996/ : SUCCESS in 19m 33s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/17/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/80a068d/ : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"67e7cf453546a9aa5171a0df0f5fb6375473b33a","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-09-09 03:13:38.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"85a7cc0e0621776333167a53272ec6552eac2be3","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-09 03:57:50.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7016/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8994/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7380/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"591b60ba225cda367c80b7560dc7dd3ef6ee349f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-09 04:36:41.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/11/427911/18/check/gate-swift-docs-ubuntu-xenial/f95368e//doc/build/html/ : SUCCESS in 3m 31s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-pep8-ubuntu-xenial/bc15c0e/ : SUCCESS in 3m 33s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-python27-ubuntu-xenial/3253a0b/ : SUCCESS in 5m 33s\n- gate-swift-python35 http://logs.openstack.org/11/427911/18/check/gate-swift-python35/2323d2c/ : SUCCESS in 2m 02s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d2ecbb1/ : SUCCESS in 1h 20m 19s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/36674a2/ : SUCCESS in 58m 05s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/fd79ba7/ : SUCCESS in 7m 16s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/63657ab/ : SUCCESS in 1m 43s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/a506512/ : SUCCESS in 4m 46s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/29cf2df/ : SUCCESS in 4m 56s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/d18baa6/ : SUCCESS in 5m 10s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/81af79b/ : SUCCESS in 4m 35s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/11/427911/18/check/gate-swift-dsvm-functional-ubuntu-xenial/35e9511/ : SUCCESS in 21m 47s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/11/427911/18/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/ed3aaa8/ : SUCCESS in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"192eab1ab5c1f33b952011933f016b45197656d9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2017-09-27 17:16:26.000000000","message":"Patch Set 18:\n\n(1 comment)\n\njust one minor comment that came up during the video call","accounts_in_message":[],"_revision_number":18},{"id":"69e3c053e721241daf71d7118ec81ea161e5d7b8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-09-27 17:22:07.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f5bc98d02b65211b691a9386469c41858a383969","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-10-03 17:44:41.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"397ab18d589ebf3e33169837317637c6dc4d32a4","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-03 18:50:41.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7699/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7335/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/9313/ : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"46c4035a8817184b73b4a0e75e6fd2c9fdbb0af5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-10-11 01:39:37.000000000","message":"Patch Set 19: Code-Review-1\n\n(9 comments)\n\nI think this is coming along nicely!  But I don\u0027t think this works yet.\n\nFirst and foremost, let\u0027s drop out the auto upgrade fail_v1 shenanigans - that stuff is wasting our time for now. \n I would much prefer an explicit config option \"use_legacy_eventlet_object_server_api\" which defaults to True for transparent upgrade, and then we can just turn it to false to test and in the lab - and later *after* we merge it - then we can decide if/how we need/want to take on automatic and/or cross-version rolling upgrades.\n\nI managed to get the proxy to upload an encrypted EC object with PUT+POST+POST but I had to do... horrible... horrible things [1]\n\nI\u0027d like to see test-ecput split up a little bit so you can pause the script at various phases of the process and inspect the on-disk artifacts (i just inserted a bunch of\n\n    raw_input(\u0027press enter to continue to XXX phase\u0027)\n\nI always seemed to end up with a .data + .meta with this patch.  We want the EC/trailing metadata *on* the .datafile I\u0027m pretty sure... I\u0027m not sure if high-jacking the POST verb to do three different things is really ideal anymore.\n\nIf we want to use POST we should at least have the top-level method looking more like a router:\n\n    def POST(req):\n        if is_trailing_metadata(req):\n            return self._update_datafile_with_metadata(req)\n        elif is_final_commit_message(req):\n            return self._do_commit(req)\n        else:\n            # regular object metadata POST\n            return self._do_obj_post(req)\n\nTrying to weave three wildly different things through a long winding series of if branches is an unmaintainable spaghetti disaster.\n\nBut I think in general this is very much on the right track!  Keep after it Pete!  Hope you survived your vacation thing... for some reason I imagine you jumping out of a plane!  :P\n\n1. and then I \u0026^%$ing lost it in an ill fated git stash pop :\u0027(  I think I managed to get inline comments on all the bits I had to change tho!","accounts_in_message":[],"_revision_number":19},{"id":"1c736b4fd62f145cee3cb2d6098812ddb2faf3dc","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-10-17 01:51:27.000000000","message":"Patch Set 19:\n\n(3 comments)\n\n(reply to comments)","accounts_in_message":[],"_revision_number":19},{"id":"bd5632b6780c1c4a23977eda420215ff05a35de6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-10-30 03:17:33.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"1b13c9b82f530379fdaa422a13dd2cd13522bbcc","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-10-30 04:01:26.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"12ab7d82ddf2e14e39eafe514f06fb3069123ce9","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-30 04:50:56.000000000","message":"Patch Set 20:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7954/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7590/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/9568/ : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"74f6591893c68190399f033344af1d6eb912cdbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-30 05:14:21.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/20/check/openstack-tox-pep8/9df686f/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/11/427911/20/check/openstack-tox-py27/a65cb48/ : SUCCESS in 6m 15s\n- openstack-tox-py35 http://logs.openstack.org/11/427911/20/check/openstack-tox-py35/c05b818/ : SUCCESS in 3m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/20/check/build-openstack-sphinx-docs/4a1327c/html/ : SUCCESS in 5m 07s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/11/427911/20/check/legacy-swift-tox-xfs-tmp-py27/28632d0/ : SUCCESS in 7m 13s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/11/427911/20/check/legacy-swift-tox-xfs-tmp-py35/709ab49/ : SUCCESS in 2m 37s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/11/427911/20/check/legacy-swift-tox-xfs-tmp-func/9f4a5a4/ : SUCCESS in 5m 39s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/11/427911/20/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/1217efb/ : SUCCESS in 6m 28s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/11/427911/20/check/legacy-swift-tox-xfs-tmp-func-encryption/0d55c7f/ : FAILURE in 6m 22s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/11/427911/20/check/legacy-swift-tox-xfs-tmp-func-ec/8f284a4/ : FAILURE in 18m 00s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/20/check/legacy-swift-dsvm-functional/898f784/ : SUCCESS in 22m 00s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/20/check/legacy-swift-dsvm-functional-identity-v3-only/44b0de2/ : SUCCESS in 21m 36s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/20/check/legacy-tempest-dsvm-neutron-full/30adba3/ : SUCCESS in 1h 11m 45s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/11/427911/20/check/legacy-grenade-dsvm-neutron/28e985f/ : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":20},{"id":"1108da92711c7ca635bbe9354b27cc53eb83d1cc","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-12-07 03:32:22.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"a45124da6597275ca44d269dcba971331eea5a45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-07 05:02:06.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/21/check/openstack-tox-pep8/05168b0/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/11/427911/21/check/openstack-tox-py27/9594c12/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/11/427911/21/check/openstack-tox-py35/d2bde73/ : SUCCESS in 3m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/21/check/build-openstack-sphinx-docs/3995cda/html/ : SUCCESS in 5m 29s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/11/427911/21/check/legacy-swift-tox-xfs-tmp-py27/2d88415/ : SUCCESS in 12m 40s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/11/427911/21/check/legacy-swift-tox-xfs-tmp-py35/1fb7a4b/ : SUCCESS in 2m 38s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/11/427911/21/check/legacy-swift-tox-xfs-tmp-func/291bcec/ : SUCCESS in 10m 04s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/11/427911/21/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/a92c9eb/ : SUCCESS in 6m 06s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/11/427911/21/check/legacy-swift-tox-xfs-tmp-func-encryption/9e66ef7/ : FAILURE in 8m 44s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/11/427911/21/check/legacy-swift-tox-xfs-tmp-func-ec/fae532c/ : FAILURE in 16m 15s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/21/check/legacy-swift-dsvm-functional/1e1183c/ : SUCCESS in 26m 45s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/21/check/legacy-swift-dsvm-functional-identity-v3-only/7aa1803/ : SUCCESS in 20m 25s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/21/check/legacy-tempest-dsvm-neutron-full/5c2297a/ : SUCCESS in 1h 27m 34s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/11/427911/21/check/legacy-grenade-dsvm-neutron/2bab92f/ : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":21},{"id":"c959b33e9a5bb0213bef46fcc2de3f4f41b84743","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-12-16 04:34:55.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"43a4ff86e76e147bba890411b06966f43e13b878","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-16 06:25:24.000000000","message":"Patch Set 22:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7686/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/9665/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8099/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"f4c51a736effe91e43b8cbc6df08f332051e5e4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-16 06:37:33.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/22/check/openstack-tox-pep8/0b4fb74/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/11/427911/22/check/openstack-tox-py27/d11b48e/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/11/427911/22/check/openstack-tox-py35/230cb86/ : SUCCESS in 2m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/22/check/build-openstack-sphinx-docs/c0a3fe0/html/ : SUCCESS in 4m 10s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/11/427911/22/check/legacy-swift-tox-xfs-tmp-py27/ee97b68/ : SUCCESS in 6m 29s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/11/427911/22/check/legacy-swift-tox-xfs-tmp-py35/3d9361f/ : SUCCESS in 3m 27s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/11/427911/22/check/legacy-swift-tox-xfs-tmp-func/3235d3f/ : TIMED_OUT in 2h 01m 43s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/11/427911/22/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/278e5da/ : SUCCESS in 10m 19s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/11/427911/22/check/legacy-swift-tox-xfs-tmp-func-encryption/319782f/ : FAILURE in 6m 54s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/11/427911/22/check/legacy-swift-tox-xfs-tmp-func-ec/5b0b680/ : FAILURE in 19m 09s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/22/check/legacy-swift-dsvm-functional/60a2f15/ : SUCCESS in 41m 12s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/22/check/legacy-swift-dsvm-functional-identity-v3-only/91db772/ : SUCCESS in 25m 32s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/22/check/legacy-tempest-dsvm-neutron-full/47cad86/ : SUCCESS in 1h 35m 55s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/11/427911/22/check/legacy-grenade-dsvm-neutron/07b8b5c/ : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":22},{"id":"9c6ab7ba55a4b64fc42ccad7e4bf6a1cb8bed3ef","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2017-12-21 05:20:07.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"3aa379ee87b7efd10444437a9046a91d53788d8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-21 06:39:21.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/23/check/openstack-tox-pep8/67c72de/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/11/427911/23/check/openstack-tox-py27/8200db8/ : SUCCESS in 4m 39s\n- openstack-tox-py35 http://logs.openstack.org/11/427911/23/check/openstack-tox-py35/1155508/ : SUCCESS in 4m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/23/check/build-openstack-sphinx-docs/070b1c4/html/ : SUCCESS in 4m 22s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/11/427911/23/check/legacy-swift-tox-xfs-tmp-py27/94534d5/ : SUCCESS in 7m 47s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/11/427911/23/check/legacy-swift-tox-xfs-tmp-py35/7058e89/ : SUCCESS in 3m 00s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/11/427911/23/check/legacy-swift-tox-xfs-tmp-func/323766b/ : SUCCESS in 15m 27s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/11/427911/23/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/01e211f/ : SUCCESS in 9m 00s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/11/427911/23/check/legacy-swift-tox-xfs-tmp-func-encryption/9ba5ca0/ : FAILURE in 6m 23s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/11/427911/23/check/legacy-swift-tox-xfs-tmp-func-ec/388f51f/ : FAILURE in 19m 34s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/23/check/legacy-swift-dsvm-functional/44ef797/ : SUCCESS in 24m 05s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/23/check/legacy-swift-dsvm-functional-identity-v3-only/b385db9/ : SUCCESS in 21m 28s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/23/check/legacy-tempest-dsvm-neutron-full/2da6165/ : SUCCESS in 1h 15m 12s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/11/427911/23/check/legacy-grenade-dsvm-neutron/1076dbd/ : SUCCESS in 47m 02s\n- swift-tox-py27 http://logs.openstack.org/11/427911/23/check/swift-tox-py27/319d92e/ : SUCCESS in 4m 50s\n- swift-tox-py35 http://logs.openstack.org/11/427911/23/check/swift-tox-py35/5cb090f/ : SUCCESS in 4m 26s\n- swift-tox-func http://logs.openstack.org/11/427911/23/check/swift-tox-func/8d8f122/ : SUCCESS in 3m 31s\n- swift-tox-func-post-as-copy http://logs.openstack.org/11/427911/23/check/swift-tox-func-post-as-copy/5cdfb51/ : SUCCESS in 3m 14s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/23/check/swift-tox-func-encryption/0b34675/ : SUCCESS in 3m 54s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/23/check/swift-tox-func-ec/8902ae1/ : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":23},{"id":"3637d7147aef84347db22e0aab13d134d2e36020","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-21 07:45:09.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/7727/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/9706/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/427911/8122/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"b7654e2c478406bb4f2c420ca84ac46263b558d1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-01-23 05:57:00.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"1e35b97d5aa1a631008aed523382044f5eed9e61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 07:24:46.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/24/check/build-openstack-sphinx-docs/affdc76/html/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/24/check/openstack-tox-pep8/72d6c09/ : SUCCESS in 3m 43s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/24/check/legacy-swift-dsvm-functional/f61563a/ : SUCCESS in 22m 23s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/24/check/legacy-swift-dsvm-functional-identity-v3-only/a4d0208/ : SUCCESS in 27m 49s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/24/check/legacy-tempest-dsvm-neutron-full/efc2544/ : SUCCESS in 1h 25m 44s\n- neutron-grenade http://logs.openstack.org/11/427911/24/check/neutron-grenade/ba000f3/ : SUCCESS in 46m 46s\n- swift-tox-py27 http://logs.openstack.org/11/427911/24/check/swift-tox-py27/4d57719/ : SUCCESS in 6m 42s\n- swift-tox-py35 http://logs.openstack.org/11/427911/24/check/swift-tox-py35/2e98500/ : SUCCESS in 3m 32s\n- swift-tox-func http://logs.openstack.org/11/427911/24/check/swift-tox-func/8d8fd62/ : SUCCESS in 11m 22s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/24/check/swift-tox-func-encryption/9065de5/ : FAILURE in 7m 29s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/24/check/swift-tox-func-ec/d59a731/ : FAILURE in 21m 00s","accounts_in_message":[],"_revision_number":24},{"id":"17dcf7cc10488e42e13bbdb5056902ebd1f40c22","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-01-25 00:59:15.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"cbe0832a2646138cdbfe1d83c55ea48b23c1676a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-25 03:07:04.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/25/check/build-openstack-sphinx-docs/572ef33/html/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/25/check/openstack-tox-pep8/ae6f7a3/ : SUCCESS in 4m 26s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/25/check/legacy-swift-dsvm-functional/c25935c/ : SUCCESS in 23m 25s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/25/check/legacy-swift-dsvm-functional-identity-v3-only/6513445/ : SUCCESS in 25m 06s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/25/check/legacy-tempest-dsvm-neutron-full/d9bdeae/ : SUCCESS in 1h 40m 09s\n- neutron-grenade http://logs.openstack.org/11/427911/25/check/neutron-grenade/7231fdd/ : SUCCESS in 55m 22s\n- swift-tox-py27 http://logs.openstack.org/11/427911/25/check/swift-tox-py27/3d3e3a3/ : SUCCESS in 7m 22s\n- swift-tox-py35 http://logs.openstack.org/11/427911/25/check/swift-tox-py35/eb56c2b/ : SUCCESS in 4m 34s\n- swift-tox-func http://logs.openstack.org/11/427911/25/check/swift-tox-func/e2a8e8e/ : SUCCESS in 14m 46s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/25/check/swift-tox-func-encryption/29452ae/ : SUCCESS in 7m 34s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/25/check/swift-tox-func-ec/8ba9915/ : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":25},{"id":"cf5190bda49a26788552ee732aefcf3d11eb261a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-03-03 01:37:58.000000000","message":"Patch Set 25:\n\n(21 comments)\n\nFirst pass. Getting it loaded into my head...","accounts_in_message":[],"_revision_number":25},{"id":"3046d457b4a640565fe4ce39e867363aa4ac0b87","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-03-08 15:22:30.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"f68c76b65fce6188a5bbdf1e71b7ede67c6b971a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-08 16:45:52.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/25/check/build-openstack-sphinx-docs/5f91cdb/html/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/25/check/openstack-tox-pep8/d1b5515/ : SUCCESS in 3m 50s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/25/check/legacy-swift-dsvm-functional/2e3ee04/ : SUCCESS in 26m 14s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/25/check/legacy-swift-dsvm-functional-identity-v3-only/72b9c05/ : SUCCESS in 21m 43s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/25/check/legacy-tempest-dsvm-neutron-full/3fe77b5/ : SUCCESS in 1h 19m 30s\n- neutron-grenade http://logs.openstack.org/11/427911/25/check/neutron-grenade/3cae9fe/ : SUCCESS in 57m 41s\n- swift-tox-py27 http://logs.openstack.org/11/427911/25/check/swift-tox-py27/440d0e3/ : SUCCESS in 6m 49s\n- swift-tox-py35 http://logs.openstack.org/11/427911/25/check/swift-tox-py35/c8276c4/ : SUCCESS in 3m 52s\n- swift-tox-func http://logs.openstack.org/11/427911/25/check/swift-tox-func/f7a3b25/ : SUCCESS in 7m 15s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/25/check/swift-tox-func-encryption/bbfd5e3/ : SUCCESS in 6m 51s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/25/check/swift-tox-func-ec/a217c57/ : SUCCESS in 7m 54s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/25/check/swift-probetests-centos-7/eea9460/ : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"2916044a01da4770a045ad09d15287c44263d3c1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-03-12 04:07:16.000000000","message":"Patch Set 25:\n\n(7 comments)","accounts_in_message":[],"_revision_number":25},{"id":"2fe2f238ca224998ba3971bf97e6360820f886e2","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-03-20 22:37:33.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"0cd537468ab3fb917bd3f64a997e5bc6f4fc3fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-21 00:10:13.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/26/check/build-openstack-sphinx-docs/f3e8acb/html/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/26/check/openstack-tox-pep8/16b679c/ : SUCCESS in 4m 02s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/26/check/legacy-swift-dsvm-functional/7f189f6/ : SUCCESS in 26m 13s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/26/check/legacy-swift-dsvm-functional-identity-v3-only/e3effd1/ : SUCCESS in 22m 02s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/26/check/legacy-tempest-dsvm-neutron-full/72711ee/ : SUCCESS in 1h 31m 16s\n- neutron-grenade http://logs.openstack.org/11/427911/26/check/neutron-grenade/3965b62/ : SUCCESS in 52m 37s\n- swift-tox-py27 http://logs.openstack.org/11/427911/26/check/swift-tox-py27/2ee8433/ : SUCCESS in 6m 45s\n- swift-tox-py35 http://logs.openstack.org/11/427911/26/check/swift-tox-py35/5997bd2/ : SUCCESS in 3m 57s\n- swift-tox-func http://logs.openstack.org/11/427911/26/check/swift-tox-func/ff3b081/ : SUCCESS in 8m 33s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/26/check/swift-tox-func-encryption/dcfe015/ : SUCCESS in 7m 36s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/26/check/swift-tox-func-domain-remap-staticweb/2c0fe7d/ : SUCCESS in 6m 59s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/26/check/swift-tox-func-ec/ac9af8b/ : SUCCESS in 8m 18s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/26/check/swift-probetests-centos-7/a4a8751/ : SUCCESS in 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"c83e290b70e8240f3d685e4f9336d466bad21fbe","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-03-21 01:21:14.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"ac073ce7e6e4338a51a5c99cf18d99f32b7af7dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-21 03:00:39.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/27/check/build-openstack-sphinx-docs/dc40b1f/html/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/27/check/openstack-tox-pep8/0f6b445/ : SUCCESS in 4m 07s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/27/check/legacy-swift-dsvm-functional/b0ab2fd/ : SUCCESS in 25m 51s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/27/check/legacy-swift-dsvm-functional-identity-v3-only/e464b91/ : SUCCESS in 23m 24s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/27/check/legacy-tempest-dsvm-neutron-full/292feb0/ : SUCCESS in 1h 38m 06s\n- neutron-grenade http://logs.openstack.org/11/427911/27/check/neutron-grenade/1545b3f/ : SUCCESS in 54m 07s\n- swift-tox-py27 http://logs.openstack.org/11/427911/27/check/swift-tox-py27/0f3b7fe/ : SUCCESS in 7m 48s\n- swift-tox-py35 http://logs.openstack.org/11/427911/27/check/swift-tox-py35/cb8afca/ : SUCCESS in 14m 08s\n- swift-tox-func http://logs.openstack.org/11/427911/27/check/swift-tox-func/560a1e9/ : SUCCESS in 6m 46s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/27/check/swift-tox-func-encryption/33c34a1/ : SUCCESS in 8m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/27/check/swift-tox-func-domain-remap-staticweb/3ed7d28/ : SUCCESS in 7m 53s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/27/check/swift-tox-func-ec/8fc45b1/ : SUCCESS in 9m 00s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/27/check/swift-probetests-centos-7/4f6f518/ : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"c842cb5ed37538269282bb08f3652c0bc749970a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2018-06-14 17:43:39.000000000","message":"Patch Set 27:\n\n(17 comments)\n\nI spent some time getting familiar with the patch. I think there may still be some hangover from previous versions. This is by no means a complete review. So far, I see opportunities to perhaps simplify, and I will push some follow on patches to gerrit with ideas - they are meant as suggestions, could be squashed but I have not thoroughly tested the ideas.\n\nIn principle I think the approach will work. I suspect we need more tests because I have made a few changes while playing with this without seeing any tests break.","accounts_in_message":[],"_revision_number":27},{"id":"2eb5a020902e96d3e855de9bec34cd087e119bc8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2018-06-14 17:49:39.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"3b88504011a4c726186c22699f2bbbdeda76e89b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2018-06-14 17:51:22.000000000","message":"Patch Set 28:\n\nrebased on master to resolve a conflict in diskfile.py and bring up to date, no other changes","accounts_in_message":[],"_revision_number":28},{"id":"e0e24532006f8ec0b91291c970804286f00f0a45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-14 21:28:24.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/28/check/build-openstack-sphinx-docs/93cc4d7/html/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/28/check/openstack-tox-pep8/7fb29a1/ : SUCCESS in 5m 06s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/28/check/legacy-swift-dsvm-functional/01c6f40/ : SUCCESS in 29m 03s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/28/check/legacy-swift-dsvm-functional-identity-v3-only/588bfcd/ : SUCCESS in 24m 50s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/28/check/legacy-tempest-dsvm-neutron-full/1418fe0/ : SUCCESS in 1h 23m 45s\n- neutron-grenade http://logs.openstack.org/11/427911/28/check/neutron-grenade/556be8a/ : SUCCESS in 48m 10s\n- swift-tox-py27 http://logs.openstack.org/11/427911/28/check/swift-tox-py27/d45f12d/ : FAILURE in 10m 14s\n- swift-tox-py35 http://logs.openstack.org/11/427911/28/check/swift-tox-py35/760f730/ : SUCCESS in 5m 11s\n- swift-tox-func http://logs.openstack.org/11/427911/28/check/swift-tox-func/be8ca66/ : SUCCESS in 9m 26s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/28/check/swift-tox-func-encryption/52e3cc0/ : SUCCESS in 8m 05s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/28/check/swift-tox-func-domain-remap-staticweb/86a9fc1/ : SUCCESS in 9m 14s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/28/check/swift-tox-func-ec/285a884/ : SUCCESS in 9m 45s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/28/check/swift-tox-func-s3api/f1dfbe8/ : SUCCESS in 10m 11s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/28/check/swift-probetests-centos-7/678ee22/ : SUCCESS in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"adc80765639e411e7ccdc8cd708fc549e0803eee","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2018-06-19 14:57:43.000000000","message":"Patch Set 28: Code-Review-1\n\n(2 comments)\n\nMy follow on patch [1] fixes a few bugs I found, and plumbs through the replicated policy path for encrypted objects. I think I saw all func tests pass with that!\n\n[1] https://review.openstack.org/#/c/575512a","accounts_in_message":[],"_revision_number":28},{"id":"c492dc9f4e7cc1ce5e7ba1e00fcdb6589b351be1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-06-20 01:09:29.000000000","message":"Patch Set 27:\n\n(6 comments)","accounts_in_message":[],"_revision_number":27},{"id":"fcb9d04308965f7aad638904c84dbefd731ad999","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-06-26 05:03:43.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"8274b7c8da938fe1cb3dbadec0312d8beb03079b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-26 06:47:47.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/29/check/build-openstack-sphinx-docs/6169a91/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/29/check/openstack-tox-pep8/5bcce0a/ : SUCCESS in 4m 58s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/29/check/legacy-swift-dsvm-functional/63a6781/ : SUCCESS in 25m 29s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/29/check/legacy-swift-dsvm-functional-identity-v3-only/135c14b/ : SUCCESS in 31m 09s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/29/check/legacy-tempest-dsvm-neutron-full/6993e5b/ : SUCCESS in 1h 42m 14s\n- neutron-grenade http://logs.openstack.org/11/427911/29/check/neutron-grenade/f50a09b/ : SUCCESS in 49m 29s\n- swift-tox-py27 http://logs.openstack.org/11/427911/29/check/swift-tox-py27/f7781a1/ : SUCCESS in 11m 15s\n- swift-tox-py35 http://logs.openstack.org/11/427911/29/check/swift-tox-py35/7a87eaa/ : SUCCESS in 4m 49s\n- swift-tox-func http://logs.openstack.org/11/427911/29/check/swift-tox-func/292de4f/ : SUCCESS in 7m 58s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/29/check/swift-tox-func-encryption/5ef7a38/ : SUCCESS in 10m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/29/check/swift-tox-func-domain-remap-staticweb/8cad5e3/ : SUCCESS in 9m 52s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/29/check/swift-tox-func-ec/afe584f/ : SUCCESS in 9m 14s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/29/check/swift-tox-func-s3api/83f4041/ : SUCCESS in 10m 11s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/29/check/swift-probetests-centos-7/3a150c9/ : SUCCESS in 29m 17s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"8e312b6ad569eae538e46c67b347aae90712c101","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-06-27 21:32:35.000000000","message":"Patch Set 29:\n\nmigration choices described at https://etherpad.openstack.org/p/swift-put-post","accounts_in_message":[],"_revision_number":29},{"id":"1550bdd51016ad95d0836264df818bda81eedfb3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-07-03 04:02:49.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"2b2a5d773b36b7da082f336c887602711a10d775","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-07-03 04:04:11.000000000","message":"Patch Set 30:\n\n\u003cacoles\u003e zaitcev__: we might want to consider doing the test restructuring in p 579227 as a separate patch *ahead* of the PUT+POST patch because it will be much easier to review those structural changes on their own. That would mean moving the DoublePutter related tests out of  579227.\n\nSee patch #30, depends on the change 579227 now.","accounts_in_message":[],"_revision_number":30},{"id":"82b0ce0fa4fb440d639e9980078d8556dcaf0668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-03 05:42:15.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/30/check/build-openstack-sphinx-docs/fa2382d/html/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/30/check/openstack-tox-pep8/db9bead/ : SUCCESS in 5m 45s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/30/check/legacy-swift-dsvm-functional/2fab1a6/ : SUCCESS in 28m 49s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/30/check/legacy-swift-dsvm-functional-identity-v3-only/981206a/ : SUCCESS in 26m 38s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/30/check/legacy-tempest-dsvm-neutron-full/019672e/ : SUCCESS in 1h 36m 32s\n- neutron-grenade http://logs.openstack.org/11/427911/30/check/neutron-grenade/d2bcbd0/ : SUCCESS in 52m 56s\n- swift-tox-py27 http://logs.openstack.org/11/427911/30/check/swift-tox-py27/ae52129/ : SUCCESS in 9m 53s\n- swift-tox-py35 http://logs.openstack.org/11/427911/30/check/swift-tox-py35/2dc6b5e/ : SUCCESS in 5m 28s\n- swift-tox-func http://logs.openstack.org/11/427911/30/check/swift-tox-func/65a1ca3/ : SUCCESS in 8m 04s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/30/check/swift-tox-func-encryption/a9d577a/ : SUCCESS in 10m 04s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/30/check/swift-tox-func-domain-remap-staticweb/a5faf04/ : SUCCESS in 8m 38s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/30/check/swift-tox-func-ec/60a1f98/ : SUCCESS in 8m 36s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/30/check/swift-tox-func-s3api/403034d/ : SUCCESS in 10m 31s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/30/check/swift-probetests-centos-7/9937080/ : SUCCESS in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"d7f6846f9db1bfdfd82d197dccf39655c3df2207","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2018-07-05 17:10:05.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"bacba3773c0abd0d11e0a3a81a0464b97848a412","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-05 20:50:57.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/31/check/build-openstack-sphinx-docs/d4083e0/html/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/31/check/openstack-tox-pep8/2ce1c01/ : SUCCESS in 5m 51s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/31/check/legacy-swift-dsvm-functional/f4dcd0f/ : SUCCESS in 25m 59s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/31/check/legacy-swift-dsvm-functional-identity-v3-only/b371264/ : SUCCESS in 28m 38s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/31/check/legacy-tempest-dsvm-neutron-full/91eed01/ : SUCCESS in 1h 49m 59s\n- neutron-grenade http://logs.openstack.org/11/427911/31/check/neutron-grenade/00ff777/ : SUCCESS in 47m 45s\n- swift-tox-py27 http://logs.openstack.org/11/427911/31/check/swift-tox-py27/6c409a1/ : SUCCESS in 10m 09s\n- swift-tox-py35 http://logs.openstack.org/11/427911/31/check/swift-tox-py35/ab011c1/ : SUCCESS in 6m 32s\n- swift-tox-func http://logs.openstack.org/11/427911/31/check/swift-tox-func/1a3f5f7/ : SUCCESS in 9m 07s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/31/check/swift-tox-func-encryption/c864803/ : SUCCESS in 10m 25s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/31/check/swift-tox-func-domain-remap-staticweb/499db4f/ : SUCCESS in 10m 08s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/31/check/swift-tox-func-ec/54d0574/ : SUCCESS in 10m 15s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/31/check/swift-tox-func-s3api/b20f8d5/ : SUCCESS in 10m 31s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/31/check/swift-probetests-centos-7/075ea9d/ : SUCCESS in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"aa30145621501dad9fca0251882df57416f466ef","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-07-18 21:36:49.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"67c2dd0c45b9f92f6210fa70956f4217c18b9419","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-19 00:16:50.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/427911/32/check/build-openstack-sphinx-docs/5055766/html/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/32/check/openstack-tox-pep8/f1a669e/ : SUCCESS in 6m 35s\n- legacy-swift-dsvm-functional http://logs.openstack.org/11/427911/32/check/legacy-swift-dsvm-functional/6a63348/ : SUCCESS in 25m 22s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/11/427911/32/check/legacy-swift-dsvm-functional-identity-v3-only/9f85b75/ : SUCCESS in 25m 47s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/11/427911/32/check/legacy-tempest-dsvm-neutron-full/56f6abb/ : SUCCESS in 1h 26m 48s\n- neutron-grenade http://logs.openstack.org/11/427911/32/check/neutron-grenade/ba19b63/ : SUCCESS in 48m 46s\n- swift-tox-py27 http://logs.openstack.org/11/427911/32/check/swift-tox-py27/e6c3afb/ : SUCCESS in 11m 53s\n- swift-tox-py35 http://logs.openstack.org/11/427911/32/check/swift-tox-py35/f0ff296/ : SUCCESS in 5m 59s\n- swift-tox-func http://logs.openstack.org/11/427911/32/check/swift-tox-func/0a95e88/ : SUCCESS in 9m 36s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/32/check/swift-tox-func-encryption/fe554a9/ : SUCCESS in 9m 42s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/32/check/swift-tox-func-domain-remap-staticweb/7d1722c/ : SUCCESS in 10m 30s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/32/check/swift-tox-func-ec/87e2e2c/ : SUCCESS in 10m 18s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/32/check/swift-tox-func-s3api/60f3d53/ : SUCCESS in 11m 39s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/32/check/swift-probetests-centos-7/fef8d27/ : FAILURE in 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"2f7cd3091471486155fa7aecc3fc6d56988f5933","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-02 07:32:10.000000000","message":"Patch Set 32:\n\n(3 comments)\n\nJust a few comment/doc string fix ups that are just NITs while I\u0027m looking through and understanding the code.\n\nI think I understand how the code works, tomorrow I\u0027ll work through tests and run locally. Hope to finish the review by the end of the day. Sorry it\u0027s taken so long to get around to.","accounts_in_message":[],"_revision_number":32},{"id":"04a3c3dd769d7521372336fffcda15b841e5e8f5","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-08-02 20:50:49.000000000","message":"Patch Set 32: Code-Review-1\n\n(4 comments)\n\nI\u0027m sorta in the same place as Matt, I\u0027ve gone throught the code once, posted a few nits comments and ran tests on my machine and it is all passing fine. I think I have a somewhat good understanding of the new protocol, but I need to spend more time on proxy/controllers/obj.py and diskfile.\n\nCouple things that I noticed while going through the code:\n1. We would benefit greatly by documenting this new protocol somewhere, the challenge is that we don\u0027t really have design docs in git, so not sure how to go about that, but at the very least, we should definetely update this section: https://docs.openstack.org/swift/latest/overview_erasure_code.html#under-the-hood\n2. Got a feeling we need more tests. again I need to spend more time on our tests too.\n\nLeaving a -1 for now, not because of something that is terribly wrong, but just realizing this still needs some more work before it merges.","accounts_in_message":[],"_revision_number":32},{"id":"5d5354c06967a495b5bff02ba2c161bcab1f1873","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-08-02 23:19:01.000000000","message":"Patch Set 32:\n\n(3 comments)","accounts_in_message":[],"_revision_number":32},{"id":"a37021642dfe43c1b138023ff8b0bd2616bca57e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-03 06:27:10.000000000","message":"Patch Set 32: Code-Review-1\n\n(5 comments)\n\nYou can tell I spent a part of today looking into Al\u0027s patch. Not saying this one isn\u0027t good enough as it is yet, just determining the differences.\n\n-1 For the fallback, that didn\u0027t seem to fall back for me. My object server running the older code just failed and the write failed (the server was holding 2 of the fragment location drives). \n\nGetting closer, thanks Pete for all the work with this, it\u0027s awesome. So much easier to understand then all that Mime stuff :)","accounts_in_message":[],"_revision_number":32},{"id":"3a4e11c77137716276fc6c095aa1ad508299c7a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-08-04 00:48:58.000000000","message":"Patch Set 32:\n\n(21 comments)\n\nStill getting myself oriented... but I ought to post my thoughts so far.","accounts_in_message":[],"_revision_number":32},{"id":"6449a1bd6604153b4b42cab2871e1732bffe880d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-08-04 05:42:15.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"76782e5bcd51d06b67169efe197ae16bbf2b57ff","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-09 07:24:06.000000000","message":"Patch Set 32:\n\n(3 comments)\n\nNice review Tim,\n\nKeeping it a -1, especially for the loss of the transaction IDs in the subsequent POSTs. We need to find a way to plumb the IDs through so requests can be tracked the entire way.","accounts_in_message":[],"_revision_number":32},{"id":"2cf6bb3adb999e06dc371db9111998134a1661f7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-10 01:37:12.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"ba767c6e5265dddc9888ddc3c34d0ce11c4c3ef8","author":{"_account_id":26240,"name":"libaoan","email":"baoan_008@163.com","username":"baoan"},"date":"2018-08-12 23:33:51.000000000","message":"Patch Set 32:\n\nfollow","accounts_in_message":[],"_revision_number":32},{"id":"c4b0816434ae5f4b7237d7f53180dfcc86190c68","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-09-10 15:51:15.000000000","message":"Patch Set 32:\n\n(3 comments)\n\nOh, looks like I had some unpushed comments...","accounts_in_message":[],"_revision_number":32},{"id":"a56346d93124dea05e7aded037ac4fce3fad9eb3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-10 22:38:27.000000000","message":"Patch Set 32: Code-Review-1\n\n(7 comments)\n\nOk, I have this loaded back in my head.  We need to fix a few things.\n\nFirst of all uploading to EC causes some HTTP protocol funkyness:\n\n    Sep 10 21:17:50 saio object-6030: ERROR WSGI: code 400, message Bad request syntax (\u00270\u0027)\n\nThis does not seem to cause the client upload to fail - I think maybe POST is sending an empty chunked *body* (!?)\n\nI\u0027m super uncomfortable with how overloaded the POST api verb is on the object server.\n\nIn pseudo code I see two options:\n\n\n    def PUT():\n        initial_metadata \u003d source.read_initial_metadata()\n        fh \u003d create(initial_metadata)\n        for chunk in source:\n            fh.write(chunk)\n        final_metadata \u003d source.read_final_metadata()\n        fh._store_metadata(final_metadata)\n        fh._commit()\n        make_container_updates()\n        return CREATED\n\n\n    def POST():\n        new_metadata \u003d source.read_all_metadata()\n        create_new_metafile(new_metadata)\n        make_container_updates()\n        return ACCEPTED\n\n\n    def STAGE():\n        initial_metadata \u003d source.read_initial_metadata()\n        fh \u003d create(initial_metadata)\n        for chunk in source:\n            fh.write(chunk)\n        temp_id \u003d stash(fh)\n        return ACCEPTED(temp_id)\n\n\n    def UPDATE():\n        fh \u003d unstash(fh)\n        final_metadata \u003d source.read_final_metadata()\n        fh._store_metadata(final_metadata)\n        temp_id \u003d stash(fh)\n        return ACCEPTED(temp_id)\n\n\n    def COMMIT():\n        fh \u003d unstash(fh)\n        # N.B. only encrypted replicated will send final_metadata w/ COMMIT\n        final_metadata \u003d source.read_final_metadata()\n        fh._store_metadata(final_metadata)\n        fh._commit()\n        make_container_updates()\n        return CREATED\n\n\nor if we\u0027re scared of VERBs\n\n    def PUT():\n        if x-commit:\n            return _handle_put()\n        else:\n            return _handle_stage()\n\n    def _handle_put():\n        initial_metadata \u003d source.read_initial_metadata()\n        fh \u003d create(initial_metadata)\n        for chunk in source:\n            fh.write(chunk)\n        final_metadata \u003d source.read_final_metadata()\n        fh._store_metadata(final_metadata)\n        fh._commit()\n        make_container_updates()\n        return CREATED\n\n\n    def POST():\n        if x-temp:\n            if x-commit:\n                return _handle_commit()\n            else:\n                return _handle_update()\n        else:\n            return _handle_post()\n\n    def _handle_post()\n        new_metadata \u003d source.read_all_metadata()\n        create_new_metafile(new_metadata)\n        make_container_updates()\n        return ACCEPTED\n\n\n    def _handle_stage():\n        initial_metadata \u003d source.read_initial_metadata()\n        fh \u003d create(initial_metadata)\n        for chunk in source:\n            fh.write(chunk)\n        temp_id \u003d stash(fh)\n        return ACCEPTED(temp_id)\n\n\n    def _handle_update():\n        fh \u003d unstash(fh)\n        final_metadata \u003d source.read_final_metadata()\n        fh._store_metadata(final_metadata)\n        temp_id \u003d stash(fh)\n        return ACCEPTED(temp_id)\n\n\n    def _handle_commit():\n        fh \u003d unstash(fh)\n        # N.B. only encrypted replicated will send final_metadata w/ COMMIT\n        final_metadata \u003d source.read_final_metadata()\n        fh._store_metadata(final_metadata)\n        fh._commit()\n        make_container_updates()\n        return CREATED","accounts_in_message":[],"_revision_number":32},{"id":"75867a9c210c01e910797df485d336bb97fd2337","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-10 22:56:26.000000000","message":"Patch Set 32:\n\n(4 comments)\n\n(answers)","accounts_in_message":[],"_revision_number":32},{"id":"4483570c3af2600c4a19fef1befd37785ec793a5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-09-11 05:16:03.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"ade2f0775d29c70a287a89a39524a57cdc5bb005","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-11 18:11:40.000000000","message":"Patch Set 32:\n\n(11 comments)\n\nI have a pretty good idea of how I want to move forward with the DiskFile stuff, and I\u0027m going to try to move to new STAGE-UPDATE-COMMIT verbs in the obj-server.\n\nI *think* like the DoulbePutter stuff in the proxy...\n\nI\u0027ll probably put up some code before I review this again.","accounts_in_message":[],"_revision_number":32},{"id":"1ed0dd252026e225621493b8cc2d4493c675ee73","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-11 21:05:09.000000000","message":"Patch Set 32:\n\n(8 comments)\n\nOkay. I\u0027m going to push a new patch that addresses Tim\u0027s and Matthew comments. Also, addresses Clay\u0027s comments on the Proxy, but _not_ on Diskfile side yet. Also, no verbs. But! fixes the weird error message about \"0\". I\u0027ll keep a -1 on this.","accounts_in_message":[],"_revision_number":32},{"id":"cfb7748d70412b6df189afd2ed0925c4cbd87cd7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-11 21:08:00.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"357c66170b9ae81891d17b2fcc641c2d61e8b525","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-11 21:26:41.000000000","message":"Patch Set 33: Workflow-1","accounts_in_message":[],"_revision_number":33},{"id":"2f5bc69bbf82d33a3ace204ad6542398cd3d5b09","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-11 21:37:51.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"7c225611f697778999853150580416f0fb6c05ed","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-11 21:38:22.000000000","message":"Patch Set 34: Code-Review-1\n\n(-1 while I\u0027m looking at the diskfile)","accounts_in_message":[],"_revision_number":34},{"id":"15157f07ee0caffa1281eb922e8fecac6a7ffd95","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-11 21:58:05.000000000","message":"Patch Set 32:\n\n(5 comments)\n\n(answers)","accounts_in_message":[],"_revision_number":32},{"id":"7d4588b159db988b161a45d54b1498ac3077955a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-11 22:36:22.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/11/427911/34/check/openstack-tox-docs/5186ab9/html/ : SUCCESS in 4m 00s\n- swift-tox-py27 http://logs.openstack.org/11/427911/34/check/swift-tox-py27/ac44cde/ : SUCCESS in 11m 48s\n- swift-tox-py35 http://logs.openstack.org/11/427911/34/check/swift-tox-py35/f0eac8e/ : SUCCESS in 5m 19s\n- swift-tox-func http://logs.openstack.org/11/427911/34/check/swift-tox-func/e2f8b44/ : SUCCESS in 8m 59s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/34/check/swift-tox-func-encryption/41b5722/ : SUCCESS in 10m 20s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/34/check/swift-tox-func-domain-remap-staticweb/555e97a/ : SUCCESS in 11m 29s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/34/check/swift-tox-func-ec/1e82861/ : SUCCESS in 10m 26s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/34/check/swift-tox-func-s3api/ddadb8c/ : SUCCESS in 19m 36s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/34/check/swift-probetests-centos-7/26ea6b5/ : FAILURE in 28m 48s\n- swift-dsvm-functional http://logs.openstack.org/11/427911/34/check/swift-dsvm-functional/40fccef/ : SUCCESS in 23m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/427911/34/check/swift-dsvm-functional-ipv6/392d4c2/ : SUCCESS in 21m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/427911/34/check/openstack-tox-lower-constraints/e1fe312/ : SUCCESS in 8m 02s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/34/check/openstack-tox-pep8/27bd450/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":34},{"id":"1b9fb0ebe3b679ca5f9e19fc1f3ae5e7f793257d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-12 20:03:22.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"87a2282f38bc9c62ba35071d0ac155f5b1fbbac8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 20:36:28.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/427911/35/check/openstack-tox-docs/9a445e6/html/ : SUCCESS in 3m 45s\n- swift-tox-py27 http://logs.openstack.org/11/427911/35/check/swift-tox-py27/7b36045/ : SUCCESS in 11m 39s\n- swift-tox-py35 http://logs.openstack.org/11/427911/35/check/swift-tox-py35/5903e23/ : SUCCESS in 4m 36s\n- swift-tox-py36 http://logs.openstack.org/11/427911/35/check/swift-tox-py36/69428b9/ : SUCCESS in 4m 46s (non-voting)\n- swift-tox-func http://logs.openstack.org/11/427911/35/check/swift-tox-func/1710bb9/ : SUCCESS in 11m 19s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/35/check/swift-tox-func-encryption/3f26917/ : SUCCESS in 8m 41s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/35/check/swift-tox-func-domain-remap-staticweb/da2735a/ : SUCCESS in 8m 22s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/35/check/swift-tox-func-ec/3facae5/ : SUCCESS in 9m 54s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/35/check/swift-tox-func-s3api/e4b34f9/ : SUCCESS in 11m 37s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/35/check/swift-probetests-centos-7/19b8467/ : SUCCESS in 26m 05s\n- swift-dsvm-functional http://logs.openstack.org/11/427911/35/check/swift-dsvm-functional/47a243d/ : SUCCESS in 25m 39s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/427911/35/check/swift-dsvm-functional-ipv6/0a4bee2/ : SUCCESS in 26m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/427911/35/check/openstack-tox-lower-constraints/f231f79/ : SUCCESS in 7m 29s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/35/check/openstack-tox-pep8/0b3c69a/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":35},{"id":"67eec582e8fefc4902e88ffdcde3e40c16cfee03","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-13 05:03:41.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"02b4bf3cbb36b94133ab82f970836a6742e6c15a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 05:37:32.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/427911/36/check/openstack-tox-docs/fe367aa/html/ : SUCCESS in 3m 48s\n- swift-tox-py27 http://logs.openstack.org/11/427911/36/check/swift-tox-py27/587d4a9/ : SUCCESS in 13m 02s\n- swift-tox-py35 http://logs.openstack.org/11/427911/36/check/swift-tox-py35/d0250d6/ : SUCCESS in 4m 29s\n- swift-tox-py36 http://logs.openstack.org/11/427911/36/check/swift-tox-py36/a9c7800/ : SUCCESS in 5m 05s (non-voting)\n- swift-tox-func http://logs.openstack.org/11/427911/36/check/swift-tox-func/7b51fea/ : SUCCESS in 9m 23s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/36/check/swift-tox-func-encryption/bf91617/ : SUCCESS in 14m 23s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/36/check/swift-tox-func-domain-remap-staticweb/8475cf2/ : SUCCESS in 8m 11s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/36/check/swift-tox-func-ec/75e8593/ : SUCCESS in 9m 42s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/36/check/swift-tox-func-s3api/c859550/ : SUCCESS in 8m 56s\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/36/check/swift-probetests-centos-7/072909a/ : SUCCESS in 25m 03s\n- swift-dsvm-functional http://logs.openstack.org/11/427911/36/check/swift-dsvm-functional/2ac1717/ : SUCCESS in 22m 22s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/427911/36/check/swift-dsvm-functional-ipv6/56a3fc1/ : SUCCESS in 27m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/427911/36/check/openstack-tox-lower-constraints/8f7fbd2/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/36/check/openstack-tox-pep8/a1c59ff/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":36},{"id":"823cbad5ad77ff8b73df0015db9260cce2a16f7b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-03 21:20:07.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"dfb029fa56804ea633cdc85eafb33ed6d4673f38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 22:59:40.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/427911/37/check/openstack-tox-docs/404eee2/html/ : SUCCESS in 4m 02s\n- swift-tox-py27 http://logs.openstack.org/11/427911/37/check/swift-tox-py27/b78ed48/ : SUCCESS in 10m 23s\n- swift-tox-py35 http://logs.openstack.org/11/427911/37/check/swift-tox-py35/0ed01dd/ : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/11/427911/37/check/swift-tox-py36/ed4f8f3/ : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/11/427911/37/check/swift-tox-py37/fd8655c/ : SUCCESS in 8m 50s\n- swift-tox-func http://logs.openstack.org/11/427911/37/check/swift-tox-func/0b58081/ : SUCCESS in 9m 23s\n- swift-tox-func-encryption http://logs.openstack.org/11/427911/37/check/swift-tox-func-encryption/917ee5c/ : SUCCESS in 9m 56s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/427911/37/check/swift-tox-func-domain-remap-staticweb/d4cc87f/ : SUCCESS in 13m 02s\n- swift-tox-func-ec http://logs.openstack.org/11/427911/37/check/swift-tox-func-ec/a69bd5e/ : SUCCESS in 9m 42s\n- swift-tox-func-s3api http://logs.openstack.org/11/427911/37/check/swift-tox-func-s3api/3af5bf4/ : SUCCESS in 10m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/11/427911/37/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8d3a88c/ : SUCCESS in 13m 56s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/11/427911/37/check/swift-probetests-centos-7/da0bb6b/ : SUCCESS in 31m 44s\n- swift-dsvm-functional http://logs.openstack.org/11/427911/37/check/swift-dsvm-functional/3e3cb92/ : SUCCESS in 27m 58s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/427911/37/check/swift-dsvm-functional-ipv6/fa297fa/ : SUCCESS in 26m 23s\n- swift-tox-lower-constraints http://logs.openstack.org/11/427911/37/check/swift-tox-lower-constraints/3a2eba8/ : SUCCESS in 11m 44s\n- openstack-tox-pep8 http://logs.openstack.org/11/427911/37/check/openstack-tox-pep8/4a813c4/ : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/11/427911/37/check/swift-multinode-rolling-upgrade/321062c/ : SUCCESS in 18m 38s","accounts_in_message":[],"_revision_number":37},{"id":"364c9bc859565dc72820c5b32fe044068b5511c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-07 22:11:14.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"c29f603867caca03269453241c5e829df9ae1213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-07 22:44:58.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"597722976221e56c4962a6b1f1736a2311a3364d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 23:45:58.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca81413f10e941a6811a6877ae6bdb4a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/690f0c7dff014ad6bee4c2db81b68fab : SUCCESS in 58m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dabb0eea7ccc4e1c9c059f9c43b6ea1e : SUCCESS in 48m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d28172897757410a93b9b1c124c9a342 : SUCCESS in 46m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe2ef2d2d5a6462691814136ce27da07 : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":39},{"id":"19a2000fb6fb131516f21c11eda92ee816c17566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 23:46:18.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e7f64b698944a0a80e9ba20c0ee7bbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f8cecfc473487caa6b9323a3bfaa6d : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/2349e084279b4f0d9bc142f47cac92dc : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6a40ecb78ee4e6ea1301ad64ea889cc : SUCCESS in 58m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7bb384f4bd714f45bb74add24d551bd8 : SUCCESS in 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78edd27bfc7341c7be9324bcfcd6ff58 : SUCCESS in 46m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f564308b88ea479aa1caa9631aa74a9b : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f978a4c779c4d708184bd0940b4ee67 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b17146ae35be494988168147b08c9ffd : SUCCESS in 17m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/812b90bc648d400fa131f143dc4c02f5 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23029285ad5f4455826c9761238f8599 : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac2e0e6ecf4749b2b61781281b632b05 : SUCCESS in 19m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c87ebcc481494243817cdf53e69f7f32 : SUCCESS in 17m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6dbb5fe2aa41476d9b822e52cc80ffcb : SUCCESS in 21m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6bfd0286d42e4cb5b4186f430d9ab26c : SUCCESS in 19m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b891d8e9d632428aae489eaae6b17ced : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b2a5a68b5d849628ff94a1453929026 : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f64ac32128dc4955a286ba3f71efada2 : SUCCESS in 11m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/97faf02ef69e4252a2775fbf530c449d : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54f3a0d7976e487a94fd2740208945b1 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3067b05782164b67a176cbc8cd1ae108 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b745cb5a6caa4cf4aac917863ef706c0 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0031aea0b36a4f9aa5d9e12fd42a388e : SUCCESS in 13m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/389792101a99445880194332e3f5fe2e : SUCCESS in 5m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f5d1282ef234d62b86fd1114449d524 : SUCCESS in 38m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cea8f3605c12483aa744121560f4067e : SUCCESS in 49m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d572773731f4c96a11717c4df26856b : SUCCESS in 35m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ceee016517434eb2bb0a2b0e9bfd168f : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/031ef6b8aa4a4c9096a7931ea82690c1 : SUCCESS in 3m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbe25857682f47968f0239436a226adb : SUCCESS in 11m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f68a0fd913f4c2589e439d91e4d22a5 : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":39}],"current_revision_number":39,"current_revision":"14408ad9c0860fb00dc4912849f4e4f97f0cc043","revisions":{"442f483da525b9f7c9f6bdd35f412a1fb92c8974":{"kind":"REWORK","_number":1,"created":"2017-02-01 20:42:12.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/1"}}},"commit":{"parents":[{"commit":"65744c8448cc5c00287dd75d91997b5874d8cd3c","subject":"Merge \"Shuffle disks and parts in reconstructor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65744c8448cc5c00287dd75d91997b5874d8cd3c"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-01 20:41:19.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-01 20:41:52.000000000","tz":-420},"subject":"WIP PUT+POST and its development test","message":"WIP PUT+POST and its development test\n\n- test-ecput.py\n  N.B. The test adds -s and -n to object-server.\n- putter_map to resolve index needed at connection time\n\n(WIP)\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/442f483da525b9f7c9f6bdd35f412a1fb92c8974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/442f483da525b9f7c9f6bdd35f412a1fb92c8974"}]},"branch":"refs/heads/master"},"ab655e0536a224d0b97e6bf68579b53a61651730":{"kind":"REWORK","_number":2,"created":"2017-02-04 03:07:17.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/2"}}},"commit":{"parents":[{"commit":"4125e8c1df7ef534bbd643064acc8e071ee58d69","subject":"Merge \"Handle log message interpolation by the logger\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4125e8c1df7ef534bbd643064acc8e071ee58d69"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-04 03:04:38.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-04 03:06:58.000000000","tz":-420},"subject":"WIP PUT+POST and its development test","message":"WIP PUT+POST and its development test\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server.\n- putter_map to resolve index needed at connection time\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n\nTODO:\n\n- POST ends in 500\n    metadata \u003d dict(disk_file.get_metafile_metadata())\n- cludged headers into commit... and other XXX\n- removing Connection: close breaks things, so where is the pipeline?\n\n(WIP)\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab655e0536a224d0b97e6bf68579b53a61651730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab655e0536a224d0b97e6bf68579b53a61651730"}]},"branch":"refs/heads/master"},"f5c4310bfbd7b0550853ed69fb96c6799818c9f2":{"kind":"REWORK","_number":3,"created":"2017-02-11 04:00:37.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/3"}}},"commit":{"parents":[{"commit":"6249945a4f3b634b1e298b5a730853d1c1ee8dea","subject":"Fix misleading hash invalidations test comments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6249945a4f3b634b1e298b5a730853d1c1ee8dea"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-11 03:58:31.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-11 03:58:31.000000000","tz":-420},"subject":"WIP PUT+POST and its development test","message":"WIP PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- putter_map to resolve index needed at connection time\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers saved too.\n\nNOTE:\n\n- POST creates .meta with diskfile.write_metadata(), which is different\n  from what PUT does using writer.put().\n  Also, we add metadata with the same timestamp as data.\n\nTODO:\n\n- resolve XXX\n  such as PUT always calls writer.commit(), no way to create non-durable\n- verify that connection is kept open between PUT and POST\n- container updates in PUT and POST occur twice\n- potential issue with visibility between PUT and POST\n  hopefuly going to be ameliorated by reaving non-durable after PUT\n- PUT returns 201, POST returns 202 - may get visible by client\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5c4310bfbd7b0550853ed69fb96c6799818c9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5c4310bfbd7b0550853ed69fb96c6799818c9f2"}]},"branch":"refs/heads/master"},"054ad491a2cf5fd8407e99a04485a11702f9952b":{"kind":"REWORK","_number":4,"created":"2017-02-16 05:04:46.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/4"}}},"commit":{"parents":[{"commit":"801d281a4aebb2ce8fb13bae522cfee1353938af","subject":"Merge \"Fix swift-get-nodes arg parsing for missing ring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/801d281a4aebb2ce8fb13bae522cfee1353938af"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-16 05:03:55.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-02-16 05:03:55.000000000","tz":-420},"subject":"WIP PUT+POST and its development test","message":"WIP PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- putter_map to resolve index needed at connection time\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers saved too.\n\nNOTE:\n\n- POST creates .meta with diskfile.write_metadata(), which is different\n  from what PUT does using writer.put().\n  Also, we add metadata with the same timestamp as data.\n\nTODO:\n\n- resolve XXX\n  such as PUT always calls writer.commit(), no way to create non-durable\n- container updates in PUT and POST occur twice\n- potential issue with visibility between PUT and POST\n  hopefuly going to be ameliorated by reaving non-durable after PUT\n- PUT returns 201, POST returns 202 - may get visible by client\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054ad491a2cf5fd8407e99a04485a11702f9952b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054ad491a2cf5fd8407e99a04485a11702f9952b"}]},"branch":"refs/heads/master"},"fbb0a3222e517fb8306dfe794c6bd30483b06582":{"kind":"REWORK","_number":5,"created":"2017-04-20 21:30:42.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/5"}}},"commit":{"parents":[{"commit":"9627bcbc9440909bacfe0d735562a5be4f125fda","subject":"Merge \"Fixed get ring name from recon cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9627bcbc9440909bacfe0d735562a5be4f125fda"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-04-20 21:18:26.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-04-20 21:30:27.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n  Saves connection setup times, but PUT and POST are not pipelined.\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other necessary headers are saved too, such as Content-Length.\n- Saving the file into temporary location at PUT, not a visible object\n- No double updates to container server anymore\n\nTODO:\n\n* unit tests are busted in the switch to the tempfile-based PUT+POST\n- resolve XXX\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbb0a3222e517fb8306dfe794c6bd30483b06582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbb0a3222e517fb8306dfe794c6bd30483b06582"}]},"branch":"refs/heads/master"},"c767de6e3a45bebb279f1cde1fe6826730cec4b6":{"kind":"REWORK","_number":6,"created":"2017-04-24 01:49:54.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/6"}}},"commit":{"parents":[{"commit":"b2a22c1097fb2ecdc85a0014e7ee75b9fedace39","subject":"Merge \"Remove redundant assignments of self._datadir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2a22c1097fb2ecdc85a0014e7ee75b9fedace39"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-04-24 01:49:03.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-04-24 01:49:03.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers saved too.\n- Saving the file into temporary location at PUT\n- No double updates to container server anymore\n\nTODO:\n\n- resolve XXX\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c767de6e3a45bebb279f1cde1fe6826730cec4b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c767de6e3a45bebb279f1cde1fe6826730cec4b6"}]},"branch":"refs/heads/master"},"f5f18ccf80bc90ad9cf34e2580809d3e047c87ee":{"kind":"REWORK","_number":7,"created":"2017-04-28 00:43:58.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/7"}}},"commit":{"parents":[{"commit":"20072570d9a58416fbd30a71cf0719c1db3bb686","subject":"Fix sporadic failure in test/unit/obj/test_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20072570d9a58416fbd30a71cf0719c1db3bb686"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-04-28 00:42:52.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-04-28 00:43:35.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n\nTODO:\n\n- Make it so we actually fall back once in 20 minutes\n- resolve XXX\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5f18ccf80bc90ad9cf34e2580809d3e047c87ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5f18ccf80bc90ad9cf34e2580809d3e047c87ee"}]},"branch":"refs/heads/master"},"78f47b42db89987c611761d87ef3ff2e2c28b491":{"kind":"REWORK","_number":8,"created":"2017-05-02 07:46:56.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/8"}}},"commit":{"parents":[{"commit":"4f3ce4db31d50de2c93e78543002bb7aac98efc7","subject":"Merge \"Fix install guide filter:authtoken config example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f3ce4db31d50de2c93e78543002bb7aac98efc7"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-05-02 07:44:53.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-05-02 07:44:53.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n\nTODO:\n\n- Add tests\n- Resolve XXX\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78f47b42db89987c611761d87ef3ff2e2c28b491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78f47b42db89987c611761d87ef3ff2e2c28b491"}]},"branch":"refs/heads/master"},"ee26b469b4fb5c6d22b6ae3394ae7396869764d2":{"kind":"REWORK","_number":9,"created":"2017-05-17 22:28:42.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/9"}}},"commit":{"parents":[{"commit":"6e893e228840bc42cfd13546245438832bc2bb46","subject":"Merge \"Drop unnecessary dereferencing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e893e228840bc42cfd13546245438832bc2bb46"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-05-02 07:44:53.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-05-17 22:28:20.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n\nTODO:\n\n- Add tests\n- Resolve XXX\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee26b469b4fb5c6d22b6ae3394ae7396869764d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee26b469b4fb5c6d22b6ae3394ae7396869764d2"}]},"branch":"refs/heads/master"},"78ebbecdf917b461c4c954118cf9d33f9e2e593a":{"kind":"REWORK","_number":10,"created":"2017-06-30 13:23:25.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/11/427911/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/10"}}},"commit":{"parents":[{"commit":"3361cd083ea9b242baee7f44f1570793ff681d73","subject":"Merge \"Order devices in the output of swift-ring-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3361cd083ea9b242baee7f44f1570793ff681d73"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-05-02 07:44:53.000000000","tz":-360},"committer":{"name":"Thiago da Silva","email":"thiago@redhat.com","date":"2017-06-30 13:23:21.000000000","tz":-240},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n\nTODO:\n\n- Add tests\n- Resolve XXX\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78ebbecdf917b461c4c954118cf9d33f9e2e593a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78ebbecdf917b461c4c954118cf9d33f9e2e593a"}]},"branch":"refs/heads/master"},"a1c2a2441045df33277665323ac0850d3a0579f3":{"kind":"REWORK","_number":11,"created":"2017-07-07 20:23:28.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/11"}}},"commit":{"parents":[{"commit":"37a1935198fd9bbbbf592760e31cc60774befefb","subject":"Merge \"Switch from oslosphinx to openstackdocstheme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37a1935198fd9bbbbf592760e31cc60774befefb"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-07-07 20:21:40.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-07-07 20:22:52.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n\nTODO:\n\n- do something about the relocation of _finalize_durable, too much trouble\n  - it dragged _part_power_cleanup in its wake even\n- Add tests\n- Resolve XXX\n  - move hold() to EC putter\n  - make sure etag is returned out of proxy PUT as before\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- rename \"hold\" into \"stage\" (from tburke)\n- Make tmp IDs more opaque, remove the path. Sanitize them when supplied.\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Reduce X-Backend-Post-To-Durable and X-Backend-Temp-Id to one\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1c2a2441045df33277665323ac0850d3a0579f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1c2a2441045df33277665323ac0850d3a0579f3"}]},"branch":"refs/heads/master"},"21a5332322ab2bcbe4e77adb3f9dd122ea1b5b27":{"kind":"REWORK","_number":12,"created":"2017-07-12 00:00:23.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/12"}}},"commit":{"parents":[{"commit":"5cb1303d5baed0a5beab7cf6de8e2ec88f8ab49f","subject":"Merge \"Add small note for handoffs_only acceleration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb1303d5baed0a5beab7cf6de8e2ec88f8ab49f"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-07-11 23:52:03.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-07-11 23:59:50.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -s and -n to object-server and fixes swift-get-info.\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n\nTODO:\n\n- Add tests\n- Resolve XXX\n  - move hold() to EC putter\n  - make sure etag is returned out of proxy PUT as before\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- rename \"hold\" into \"stage\" (from tburke)\n- Make tmp IDs more opaque, remove the path. Sanitize them when supplied.\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Reduce X-Backend-Post-To-Durable and X-Backend-Temp-Id to one\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a5332322ab2bcbe4e77adb3f9dd122ea1b5b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a5332322ab2bcbe4e77adb3f9dd122ea1b5b27"}]},"branch":"refs/heads/master"},"b69978dbe94d272478ec4c8fa0b75cf79d6196f6":{"kind":"REWORK","_number":13,"created":"2017-07-27 00:16:03.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/13"}}},"commit":{"parents":[{"commit":"4ac456882eade2c5e5a51a5bc64cba5281971771","subject":"Merge \"Add ever_rebalanced property to RingBuilder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ac456882eade2c5e5a51a5bc64cba5281971771"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-07-27 00:08:26.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-07-27 00:08:26.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n\nTODO:\n\n- Add tests\n- Resolve XXX\n  - move hold() to EC putter\n  - make sure etag is returned out of proxy PUT as before\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- rename \"hold\" into \"stage\" (from tburke)\n- Make tmp IDs more opaque, remove the path. Sanitize them when supplied.\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Reduce X-Backend-Post-To-Durable and X-Backend-Temp-Id to one\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b69978dbe94d272478ec4c8fa0b75cf79d6196f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b69978dbe94d272478ec4c8fa0b75cf79d6196f6"}]},"branch":"refs/heads/master"},"8021363b7f9c61594c07cc4bd706d29fa1e7dc8c":{"kind":"REWORK","_number":14,"created":"2017-09-02 05:18:33.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/14"}}},"commit":{"parents":[{"commit":"6da17e992375cba37035527e69ef17aa6a5d6f28","subject":"Merge \"Cleanup test tempdirs in tearDown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6da17e992375cba37035527e69ef17aa6a5d6f28"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-02 05:15:13.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-02 05:15:13.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n\nTODO:\n\n- Add tests\n- Resolve XXX\n  - move hold() to EC putter\n  - make sure etag is returned out of proxy PUT as before\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- rename \"hold\" into \"stage\" (from tburke)\n- Make tmp IDs more opaque, remove the path. Sanitize them when supplied.\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Reduce X-Backend-Post-To-Durable and X-Backend-Temp-Id to one\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8021363b7f9c61594c07cc4bd706d29fa1e7dc8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8021363b7f9c61594c07cc4bd706d29fa1e7dc8c"}]},"branch":"refs/heads/master"},"09e8fe563a98fa592a0b764b175d4043471fd174":{"kind":"REWORK","_number":15,"created":"2017-09-03 23:08:55.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/15"}}},"commit":{"parents":[{"commit":"6da17e992375cba37035527e69ef17aa6a5d6f28","subject":"Merge \"Cleanup test tempdirs in tearDown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6da17e992375cba37035527e69ef17aa6a5d6f28"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-03 22:56:51.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-03 22:56:51.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n\nTODO:\n\n- Add tests\n- Resolve XXX\n  - make sure etag is returned out of proxy PUT as before\n- PUT returns 201, POST returns 202 - may get visible by client\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Make tmp IDs more opaque, remove the path. Sanitize them when supplied.\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Reduce X-Backend-Post-To-Durable and X-Backend-Temp-Id to one\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09e8fe563a98fa592a0b764b175d4043471fd174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09e8fe563a98fa592a0b764b175d4043471fd174"}]},"branch":"refs/heads/master"},"f5a2732a764241a3179a6d1dddbe3927d70a3696":{"kind":"REWORK","_number":16,"created":"2017-09-05 03:37:40.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/16"}}},"commit":{"parents":[{"commit":"6da17e992375cba37035527e69ef17aa6a5d6f28","subject":"Merge \"Cleanup test tempdirs in tearDown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6da17e992375cba37035527e69ef17aa6a5d6f28"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-05 03:27:36.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-05 03:27:36.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from POST, for testing reasons\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n\nTODO:\n\n- Add more tests\n- Resolve XXX\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Make tmp IDs more opaque, remove the path. Sanitize them when supplied.\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Reduce X-Backend-Post-To-Durable and X-Backend-Temp-Id to one\n- Catch ChunkWriteTimeout from send_commit_confirmation()\n- Decide if the loss of transient sysmeta is bad, see the unit test\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5a2732a764241a3179a6d1dddbe3927d70a3696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5a2732a764241a3179a6d1dddbe3927d70a3696"}]},"branch":"refs/heads/master"},"02d08298b1718dd56c9a2ac6b392e15fd071496b":{"kind":"REWORK","_number":17,"created":"2017-09-07 23:32:53.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/17"}}},"commit":{"parents":[{"commit":"834b733d297d08ef8bc5e09c52653f9292d61f75","subject":"Merge \"Remove vestigate HUDSON_PUBLISH_DOCS reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/834b733d297d08ef8bc5e09c52653f9292d61f75"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-07 23:20:42.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-07 23:20:42.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from POST, for testing reasons\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n\nTODO:\n\n- Add more tests\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Make tmp IDs more opaque, remove the path. Sanitize them when supplied.\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Catch ChunkWriteTimeout from send_commit_confirmation()\n- Decide if the loss of transient sysmeta is bad, see the unit test\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02d08298b1718dd56c9a2ac6b392e15fd071496b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02d08298b1718dd56c9a2ac6b392e15fd071496b"}]},"branch":"refs/heads/master"},"ad7a81a738ce86371535bbdf94b6c359ae0bb2cd":{"kind":"REWORK","_number":18,"created":"2017-09-09 03:13:38.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/18"}}},"commit":{"parents":[{"commit":"834b733d297d08ef8bc5e09c52653f9292d61f75","subject":"Merge \"Remove vestigate HUDSON_PUBLISH_DOCS reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/834b733d297d08ef8bc5e09c52653f9292d61f75"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-09 03:11:01.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-09-09 03:11:01.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n  Other headers are saved too.\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from POST, for testing reasons\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n\nTODO:\n\n- Add more tests\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Verify that we reject POST with tmp ID if the temporary file is missing\n- Catch ChunkWriteTimeout from send_commit_confirmation()\n- Decide if the loss of transient sysmeta is bad, see the unit test\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad7a81a738ce86371535bbdf94b6c359ae0bb2cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad7a81a738ce86371535bbdf94b6c359ae0bb2cd"}]},"branch":"refs/heads/master"},"69ef1ba4da42fb6b3d2a6feca8ec39215db6b0a9":{"kind":"REWORK","_number":19,"created":"2017-10-03 17:44:41.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/19"}}},"commit":{"parents":[{"commit":"39dc227f3949f9dfd597109e28d721de6aef1892","subject":"Merge \"Fix functest for IPv6 endpoints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39dc227f3949f9dfd597109e28d721de6aef1892"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-10-03 17:43:12.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-10-03 17:43:12.000000000","tz":-300},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST).\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from 2nd POST\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n- Made sure \"v2\" returns 400 (future-proof) - a review comment by acoles\n- Made the final_phase depend on the underlying putter\n- Introduced frag_prefs, or else POST cannot open non-durable and 404\n- Preserved our meta from being filtered out\n\nTODO:\n\n- Incorporate torgomatic\u0027s patch that allows us to return to linkat()\n- Run functests\n- Add more unit tests - acoles volunteered to help with the framework issues\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Catch newly-added ChunkWriteTimeout in proxy\n- [Decide if the loss of transient sysmeta is bad, see the unit test]\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69ef1ba4da42fb6b3d2a6feca8ec39215db6b0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69ef1ba4da42fb6b3d2a6feca8ec39215db6b0a9"}]},"branch":"refs/heads/master"},"6026560216c751abfe95db7e8d32f95b2c92663a":{"kind":"REWORK","_number":20,"created":"2017-10-30 04:01:26.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/20"}}},"commit":{"parents":[{"commit":"ff2acfe393afeb492ffa29f9d1c34c4589fd50fe","subject":"Merge \"Added the man page for container-reconciler.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff2acfe393afeb492ffa29f9d1c34c4589fd50fe"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-10-30 03:25:38.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-10-30 03:25:38.000000000","tz":-300},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST).\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from 2nd POST\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n- Made sure \"v2\" returns 400 (future-proof) - a review comment by acoles\n- Made the final_phase depend on the underlying putter\n- Introduced frag_prefs, or else POST cannot open non-durable and 404\n- Preserved our meta from being filtered out\n- Removed the timer-based fallback, added the use_put_v1 knob\n- re-factored object POST to reduce technical debt w/ _post_metadata()\n- let DoublePutter to pipe chunked, pushed treatment of length into putter\n- stored the metadata in the .data like PUT, not .meta as a normal POST\n- Allow successes as well as informational responses from putters\n\nTODO:\n\n- Incorporate torgomatic\u0027s patch that allows us to return to linkat()\n- Add more unit tests - acoles volunteered to help with the framework issues\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Catch newly-added ChunkWriteTimeout in proxy\n- [Decide if the loss of transient sysmeta is bad, see the unit test]\n- Find why proxy adds \"Connection: Close\"\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6026560216c751abfe95db7e8d32f95b2c92663a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6026560216c751abfe95db7e8d32f95b2c92663a"}]},"branch":"refs/heads/master"},"861549ebd2ab6c8a57352a1948055537fe7cfe24":{"kind":"REWORK","_number":21,"created":"2017-12-07 03:32:22.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/21"}}},"commit":{"parents":[{"commit":"c53d5bfe9f9b57a35ef8c3dfb5fcfe6f3608f13c","subject":"Merge \"Better handle missing files in _construct_from_data_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c53d5bfe9f9b57a35ef8c3dfb5fcfe6f3608f13c"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-12-07 01:21:54.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-12-07 01:21:54.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST).\nAlso, adds Sam\u0027s hanging writer patch.\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from 2nd POST\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n- Made sure \"v2\" returns 400 (future-proof) - a review comment by acoles\n- Made the final_phase depend on the underlying putter\n- Introduced frag_prefs, or else POST cannot open non-durable and 404\n- Preserved our meta from being filtered out\n- Removed the timer-based fallback, added the use_put_v1 knob\n- re-factored object POST to reduce technical debt w/ _post_metadata()\n- let DoublePutter to pipe chunked, pushed treatment of length into putter\n- stored the metadata in the .data like PUT, not .meta as a normal POST\n- Allow successes as well as informational responses from putters\n\nTODO:\n\n- Incorporate torgomatic\u0027s patch that allows us to return to linkat()\n- Add more unit tests - acoles volunteered to help with the framework issues\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Catch newly-added ChunkWriteTimeout in proxy\n- [Decide if the loss of transient sysmeta is bad, see the unit test]\n- Find why proxy adds \"Connection: Close\"\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/861549ebd2ab6c8a57352a1948055537fe7cfe24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/861549ebd2ab6c8a57352a1948055537fe7cfe24"}]},"branch":"refs/heads/master"},"b732d251df8ea58d43e0944781f72c1e48896a96":{"kind":"REWORK","_number":22,"created":"2017-12-16 04:34:55.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/22"}}},"commit":{"parents":[{"commit":"d4e62830c93fd614db79b109ed611de3c665124a","subject":"Merge \"Remove symlink from xml listing response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4e62830c93fd614db79b109ed611de3c665124a"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-12-16 04:33:57.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-12-16 04:33:57.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST).\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from 2nd POST\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n- Made sure \"v2\" returns 400 (future-proof) - a review comment by acoles\n- Made the final_phase depend on the underlying putter\n- Introduced frag_prefs, or else POST cannot open non-durable and 404\n- Preserved our meta from being filtered out\n- Removed the timer-based fallback, added the use_put_v1 knob\n- re-factored object POST to reduce technical debt w/ _post_metadata()\n- let DoublePutter to pipe chunked, pushed treatment of length into putter\n- stored the metadata in the .data like PUT, not .meta as a normal POST\n- Allow successes as well as informational responses from putters\n- Added a persisting DiskFileWriter (in order to regain linkat())\n\nTODO:\n\n- Incorporate torgomatic\u0027s patch that allows us to return to linkat()\n  - almost done! just needs to make use of the persisting DiskFileWriter\n- Add more unit tests - acoles volunteered to help with the framework issues\n- Ensure a cleanup of tmp*/ that does not sweep fresh temporary files\n- Catch newly-added ChunkWriteTimeout in proxy\n- [Decide if the loss of transient sysmeta is bad, see the unit test]\n- Find why proxy adds \"Connection: Close\"\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b732d251df8ea58d43e0944781f72c1e48896a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b732d251df8ea58d43e0944781f72c1e48896a96"}]},"branch":"refs/heads/master"},"a6b97ff04b6aee1fa4acb1661bc97ef614f6aec5":{"kind":"REWORK","_number":23,"created":"2017-12-21 05:20:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/23"}}},"commit":{"parents":[{"commit":"8dcef649757dc029d2fc7d776b9e7f8926f2bbd5","subject":"Merge \"Move symlink versioning functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8dcef649757dc029d2fc7d776b9e7f8926f2bbd5"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-12-21 05:19:20.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2017-12-21 05:19:20.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nDONE:\n\n- Added (temporarily) test-ecput.py\n  N.B. The test adds -n to object-server\n- Proxy conection is kept open between PUT and POST\n- X-Timestamp same for POST as for PUT (with X-Backend-To-Durable)\n- Saving the file into a temporary location at PUT\n- No double updates to container server anymore\n- Using conn.path\n- Falls back to legacy and attempts to return after 20 minutes\n- Since POST does not have a writer, relocated commit() to DiskFile\n- Cleaned up useless *kw in make_on_disk_filename()\n- Return etag from 2nd POST\n- Return 201 from POST, even if it\u0027s not very correct\n- Patched node_timeout through DoublePutter() - unlike conn.path\n- Made sure \"v2\" returns 400 (future-proof) - a review comment by acoles\n- Made the final_phase depend on the underlying putter\n- Introduced frag_prefs, or else POST cannot open non-durable and 404\n- Preserved our meta from being filtered out\n- Removed the timer-based fallback, added the use_put_v1 knob\n- re-factored object POST to reduce technical debt w/ _post_metadata()\n- let DoublePutter to pipe chunked, pushed treatment of length into putter\n- stored the metadata in the .data like PUT, not .meta as a normal POST\n- Allow successes as well as informational responses from putters\n- Added a persisting DiskFileWriter (in order to regain linkat())\n\nTODO:\n\n- Add more unit tests - acoles volunteered to help with the framework issues\n- Catch newly-added ChunkWriteTimeout in proxy\n- [Decide if the loss of transient sysmeta is bad, see the unit test]\n- Find why proxy adds \"Connection: Close\"\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6b97ff04b6aee1fa4acb1661bc97ef614f6aec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6b97ff04b6aee1fa4acb1661bc97ef614f6aec5"}]},"branch":"refs/heads/master"},"e4eabb8e31a2038438003892d5823947a3afa0bd":{"kind":"REWORK","_number":24,"created":"2018-01-23 05:57:00.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/24"}}},"commit":{"parents":[{"commit":"da5c71b7cb47a52ccd3a6bbb58afe4e922977e31","subject":"Merge \"Remove redundant blank space in README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da5c71b7cb47a52ccd3a6bbb58afe4e922977e31"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-01-23 05:54:39.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-01-23 05:54:39.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nN.B. This version adds test-ecput.py, which needs to add -n to the\nobject server. These changes are only necessary for the development.\n\nTODO:\n\n- Add more unit tests - acoles volunteered to help with the framework issues\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4eabb8e31a2038438003892d5823947a3afa0bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4eabb8e31a2038438003892d5823947a3afa0bd"}]},"branch":"refs/heads/master"},"79fec861768bb219815e6cab80d3b1320e393b1e":{"kind":"REWORK","_number":25,"created":"2018-01-25 00:59:15.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/25"}}},"commit":{"parents":[{"commit":"5cf96230c82d4fcbac297775997a7e0abe3e9ff9","subject":"Merge \"Add pipeline modification test for previously-recommended pipelines\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cf96230c82d4fcbac297775997a7e0abe3e9ff9"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-01-25 00:58:19.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-01-25 00:58:19.000000000","tz":-360},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nN.B. This version adds test-ecput.py, which needs to add -n to the\nobject server. These changes are only necessary for the development.\n\nTODO:\n\n- Add more unit tests - acoles volunteered to help with the framework issues\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79fec861768bb219815e6cab80d3b1320e393b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79fec861768bb219815e6cab80d3b1320e393b1e"}]},"branch":"refs/heads/master"},"c71f8f231c4ab67628c2bb828a0e6bb929bd54b7":{"kind":"REWORK","_number":26,"created":"2018-03-20 22:37:33.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/26"}}},"commit":{"parents":[{"commit":"0b649d538749c9e80d2a7cf3efea06d0b198e4d6","subject":"Merge \"Refactor tempurl functional test\u0027s try-finally\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b649d538749c9e80d2a7cf3efea06d0b198e4d6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-03-20 22:35:16.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-03-20 22:35:16.000000000","tz":-300},"subject":"PUT+POST and its development test","message":"PUT+POST and its development test\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nN.B. This version adds test-ecput.py, which needs to add -n to the\nobject server. These changes are only necessary for the development.\n\nTODO:\n\n- get rid of test-ecput.py\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c71f8f231c4ab67628c2bb828a0e6bb929bd54b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c71f8f231c4ab67628c2bb828a0e6bb929bd54b7"}]},"branch":"refs/heads/master"},"061690bbc467be6a017aaddd5c3cc79632ec55b8":{"kind":"REWORK","_number":27,"created":"2018-03-21 01:21:14.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/27"}}},"commit":{"parents":[{"commit":"0b649d538749c9e80d2a7cf3efea06d0b198e4d6","subject":"Merge \"Refactor tempurl functional test\u0027s try-finally\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b649d538749c9e80d2a7cf3efea06d0b198e4d6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-03-21 00:53:30.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-03-21 00:58:37.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/061690bbc467be6a017aaddd5c3cc79632ec55b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/061690bbc467be6a017aaddd5c3cc79632ec55b8"}]},"branch":"refs/heads/master"},"46ee580e707c3b8bcdcd5de49391f025a2016433":{"kind":"REWORK","_number":28,"created":"2018-06-14 17:49:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/427911/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/28"}}},"commit":{"parents":[{"commit":"d3368521933022e865302978a01ab82a4083fd9d","subject":"Merge \"rectify \u0027a integer number\u0027 to \u0027an integer number\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3368521933022e865302978a01ab82a4083fd9d"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-03-21 00:53:30.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2018-06-14 17:47:08.000000000","tz":60},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46ee580e707c3b8bcdcd5de49391f025a2016433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46ee580e707c3b8bcdcd5de49391f025a2016433"}]},"branch":"refs/heads/master"},"cae53f566a2a0da91013bb83518c4901bb929111":{"kind":"REWORK","_number":29,"created":"2018-06-26 05:03:43.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/29"}}},"commit":{"parents":[{"commit":"1fa4b9e3feba75e56293aaf57352ec5add3551a4","subject":"Remove unnecessary loop in test_keymaster.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fa4b9e3feba75e56293aaf57352ec5add3551a4"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-06-26 05:02:40.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-06-26 05:02:40.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cae53f566a2a0da91013bb83518c4901bb929111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cae53f566a2a0da91013bb83518c4901bb929111"}]},"branch":"refs/heads/master"},"4428ce434517b95e3ccdf91dee6f5f4df344c5c4":{"kind":"REWORK","_number":30,"created":"2018-07-03 04:02:49.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/30"}}},"commit":{"parents":[{"commit":"91a8cd2952a25b340f5778cace95910e43cbcbb6","subject":"PUT+POST: break out putter specific test classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91a8cd2952a25b340f5778cace95910e43cbcbb6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-07-03 03:58:16.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-07-03 04:02:13.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4428ce434517b95e3ccdf91dee6f5f4df344c5c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4428ce434517b95e3ccdf91dee6f5f4df344c5c4"}]},"branch":"refs/heads/master"},"87254cfccc5497106f6ecea6c4445c3035eb4e2a":{"kind":"REWORK","_number":31,"created":"2018-07-05 17:10:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/427911/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/31"}}},"commit":{"parents":[{"commit":"91a8cd2952a25b340f5778cace95910e43cbcbb6","subject":"PUT+POST: break out putter specific test classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91a8cd2952a25b340f5778cace95910e43cbcbb6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-07-03 03:58:16.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2018-07-05 17:09:00.000000000","tz":60},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87254cfccc5497106f6ecea6c4445c3035eb4e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87254cfccc5497106f6ecea6c4445c3035eb4e2a"}]},"branch":"refs/heads/master"},"83256db01ae98b902df7691cad155bbdc6024080":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2018-07-18 21:36:49.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/32"}}},"commit":{"parents":[{"commit":"791009a6ff9c85556ccd1a7e7fee312f1d32abd7","subject":"Merge \"PUT+POST: break out putter specific test classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/791009a6ff9c85556ccd1a7e7fee312f1d32abd7"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-07-03 03:58:16.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-07-18 21:32:07.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83256db01ae98b902df7691cad155bbdc6024080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83256db01ae98b902df7691cad155bbdc6024080"}]},"branch":"refs/heads/master"},"e1e2274a9a4236c43c5ffebe4d246b964095326c":{"kind":"REWORK","_number":33,"created":"2018-09-11 21:08:00.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/33"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-11 21:07:29.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-11 21:07:29.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1e2274a9a4236c43c5ffebe4d246b964095326c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1e2274a9a4236c43c5ffebe4d246b964095326c"}]},"branch":"refs/heads/master"},"6affdc190ab0fba68fa29cb4fe71c8c72a7985cc":{"kind":"REWORK","_number":34,"created":"2018-09-11 21:37:51.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/34"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-11 21:37:17.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-11 21:37:17.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6affdc190ab0fba68fa29cb4fe71c8c72a7985cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6affdc190ab0fba68fa29cb4fe71c8c72a7985cc"}]},"branch":"refs/heads/master"},"7393798d07927fd979a8240835484acc72beb79c":{"kind":"REWORK","_number":35,"created":"2018-09-12 20:03:22.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/35"}}},"commit":{"parents":[{"commit":"33c76507531073e431148a4f65da0a40e5ba88eb","subject":"Add writer method to DiskFile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33c76507531073e431148a4f65da0a40e5ba88eb"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-12 19:43:49.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-12 19:51:23.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7393798d07927fd979a8240835484acc72beb79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7393798d07927fd979a8240835484acc72beb79c"}]},"branch":"refs/heads/master"},"d9f04fe6f5f30597270ac980aea0ce7150186ded":{"kind":"REWORK","_number":36,"created":"2018-09-13 05:03:41.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/36"}}},"commit":{"parents":[{"commit":"aeeec5c1ae3bca955ad36af450795064596e7173","subject":"Merge \"DiskFile(Writer) refactor cleanups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeeec5c1ae3bca955ad36af450795064596e7173"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-12 19:43:49.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-13 05:03:31.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nThe Id7619a7ea356be33d3e5b3ad70d6880608320740 is not used anymore.\nInstead, we rely of DiskFileWriter.open().\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9f04fe6f5f30597270ac980aea0ce7150186ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9f04fe6f5f30597270ac980aea0ce7150186ded"}]},"branch":"refs/heads/master"},"c569071aec8369f3fe1e6d187a33e3e1d9e87431":{"kind":"REWORK","_number":37,"created":"2019-05-03 21:20:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/11/427911/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/37"}}},"commit":{"parents":[{"commit":"f55167a735632cdc89fb6f83f97e4fe3ed09b7cc","subject":"Merge \"Increase node_timeout in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f55167a735632cdc89fb6f83f97e4fe3ed09b7cc"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-12 19:43:49.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-05-03 21:20:01.000000000","tz":-300},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nThe Id7619a7ea356be33d3e5b3ad70d6880608320740 is not used anymore.\nInstead, we rely of DiskFileWriter.open().\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c569071aec8369f3fe1e6d187a33e3e1d9e87431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c569071aec8369f3fe1e6d187a33e3e1d9e87431"}]},"branch":"refs/heads/master"},"adc2a55caed61aced08f07231fc01ccc2c36b60a":{"kind":"REWORK","_number":38,"created":"2024-10-07 22:11:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/427911/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/38"}}},"commit":{"parents":[{"commit":"44653d8efbae369ac6127a27648405ccaf60ccc4","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44653d8efbae369ac6127a27648405ccaf60ccc4"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-12 19:43:49.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-07 22:03:06.000000000","tz":-420},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nThe Id7619a7ea356be33d3e5b3ad70d6880608320740 is not used anymore.\nInstead, we rely of DiskFileWriter.open().\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCloses-Bug: #1496636\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adc2a55caed61aced08f07231fc01ccc2c36b60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adc2a55caed61aced08f07231fc01ccc2c36b60a"}]},"branch":"refs/heads/master"},"14408ad9c0860fb00dc4912849f4e4f97f0cc043":{"kind":"REWORK","_number":39,"created":"2024-10-07 22:44:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/427911/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/427911/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/427911/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/427911/39"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-09-12 19:43:49.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-07 22:44:51.000000000","tz":-420},"subject":"Replace MIME with PUT+POST for EC and Encryption","message":"Replace MIME with PUT+POST for EC and Encryption\n\nActually, we do not force a replacement outright, for compatibility\nand safety reasons. Instead, a configuration option is provided to\nenable the new protocol.\n\nThis is the 3-phase version (PUT+POST+POST) with a persisting DiskFile.\n\nThe Id7619a7ea356be33d3e5b3ad70d6880608320740 is not used anymore.\nInstead, we rely of DiskFileWriter.open().\n\nCo-authored-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCloses-Bug: #1496636\nChange-Id: I64b0d8fdb2ffce786f56665a74ed7eb2603abfda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14408ad9c0860fb00dc4912849f4e4f97f0cc043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14408ad9c0860fb00dc4912849f4e4f97f0cc043"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
