)]}'
{"id":"openstack%2Fswift~468105","triplet_id":"openstack%2Fswift~master~I8f9de0bec01032d9d9b58848e2a76ac92e65ab09","project":"openstack/swift","branch":"master","topic":"bug/1639691","hashtags":[],"change_id":"I8f9de0bec01032d9d9b58848e2a76ac92e65ab09","subject":"Require that known-bad EC schemes be deprecated","status":"MERGED","created":"2017-05-25 16:56:19.000000000","updated":"2017-06-22 01:11:03.000000000","submitted":"2017-06-22 01:11:03.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"12343aef0fbde2f2660e11a9d02d1b1e6400cae0","_number":468105,"virtual_id_number":468105,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2017-06-22 01:11:03.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2017-06-19 01:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-06-19 04:19:37.000000000","_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2017-06-21 21:05:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-05-25 16:59:44.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2017-05-25 21:27:10.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-06-07 21:34:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2017-06-19 01:05:48.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2017-06-19 04:19:37.000000000","updated_by":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"reviewer":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"state":"REVIEWER"},{"updated":"2017-06-21 21:05:07.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2017-06-22 01:11:03.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c8409279c4b1c81eee060a5e8007e6149714a321","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-05-25 16:56:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5a9ad7d42a7405bfd8a1bb0a85ab63a61ebd351","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2017-05-25 16:59:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nmakes sense, let\u0027s do it.","accounts_in_message":[],"_revision_number":1},{"id":"ba4adc9a8e60360f5f1285c6147ee74977dca8b6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-05-25 17:16:34.000000000","message":"Patch Set 1:\n\nPlease don\u0027t forget this needs an UpgradeImpact tag - provisioning systems that deploy swift will need to be updated to have this *same* validation and either prevent deployment/upgrade or just enforce the deprecation automatically.","accounts_in_message":[],"_revision_number":1},{"id":"3855dcba7c60c4d68d44f599e9b025ef29b53bf4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-05-25 17:42:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0f3bd9e5d64b060387e5ef99d3917d5f1bd191b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-05-25 17:42:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e7729d602fe7d0980e6d81183b4f0af5e8ad7e2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-25 19:25:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/05/468105/2/check/gate-swift-docs-ubuntu-xenial/d989b5f//doc/build/html/ : SUCCESS in 2m 45s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-pep8-ubuntu-xenial/0cb32d5/ : SUCCESS in 5m 18s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-python27-ubuntu-xenial/1deb8ee/ : SUCCESS in 5m 43s\n- gate-swift-python35 http://logs.openstack.org/05/468105/2/check/gate-swift-python35/cf4a0fc/ : SUCCESS in 2m 17s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/da97205/ : SUCCESS in 1h 32m 50s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3136622/ : SUCCESS in 49m 05s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/3dccaf1/ : SUCCESS in 3m 40s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/385acd1/ : SUCCESS in 1m 14s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/872f716/ : SUCCESS in 4m 07s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/6009b4b/ : SUCCESS in 5m 41s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/b836eda/ : SUCCESS in 3m 50s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/a8e4a95/ : SUCCESS in 7m 02s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/05/468105/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/4cf80c3/ : SUCCESS in 1h 40m 40s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/05/468105/2/check/gate-swift-dsvm-functional-ubuntu-xenial/1b3c5d4/ : SUCCESS in 37m 34s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/05/468105/2/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/1a8fef6/ : SUCCESS in 30m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98aee75b3b56a5e82284c40b906d5091cffe8ad7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-05-25 19:36:33.000000000","message":"Patch Set 2: Code-Review+1\n\nI\u0027m sure this is great.  I added it on the upstream agenda for next week [1]\n\nwe can merge this orthogonally - but I\u0027d like to try and con someone into sending something to the openstack operators ML or blog or something (ideas?).  We don\u0027t break backwards compat often - but when we *need to* - we do it LOUDLY ;)\n\n1. https://wiki.openstack.org/wiki/Meetings/Swift","accounts_in_message":[],"_revision_number":2},{"id":"10cf4837aa7ddfed4d8fc16bebe7b1b46aecb4bb","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-05-25 21:06:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/6677/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/8293/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/6316/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"91c91f85cc96d23c9baa8afc262bf84ba50aa4db","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-05-25 21:27:10.000000000","message":"Patch Set 2:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/6678/probetests/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/8294/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/6317/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"bde15f8225585852fc1adaab8c8013d6f3efe3eb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-06-07 21:34:09.000000000","message":"Patch Set 2: Code-Review+2\n\nUpgradeImpact LGTM, code looks tight.\n\nWOMM\n\n    ubuntu@saio:~$ swift-init proxy status\n    proxy-server running (7731 - /etc/swift/proxy-server/proxy-noauth.conf.d)\n    proxy-server running (7732 - /etc/swift/proxy-server/proxy-server.conf.d)\n    ubuntu@saio:~$ swift-init proxy restart\n    Signal proxy-server  pid: 7731  signal: 15\n    Signal proxy-server  pid: 7732  signal: 15\n    proxy-server (7731) appears to have stopped\n    proxy-server (7732) appears to have stopped\n    WARNING: Unable to modify max process limit.  Running as non-root?\n    Starting proxy-server...(/etc/swift/proxy-server/proxy-noauth.conf.d)\n    Starting proxy-server...(/etc/swift/proxy-server/proxy-server.conf.d)\n    Storage policy ec uses an EC configuration known to harm data durability. Any data in this policy should be migrated. See https://bugs.launchpad.net/swift/+bug/1639691 for more information.\n    ERROR: Invalid Storage Policy Configuration in /etc/swift/swift.conf (Storage policy ec uses an EC configuration known to harm data durability. This policy MUST be deprecated.)\n\n    Storage policy ec uses an EC configuration known to harm data durability. Any data in this policy should be migrated. See https://bugs.launchpad.net/swift/+bug/1639691 for more information.\n    ERROR: Invalid Storage Policy Configuration in /etc/swift/swift.conf (Storage policy ec uses an EC configuration known to harm data durability. This policy MUST be deprecated.)\n\n    ubuntu@saio:~$ echo $?\n    1\n    ubuntu@saio:~$ swift-init proxy status\n    No proxy-server running","accounts_in_message":[],"_revision_number":2},{"id":"1fb89a4612ff34f0c84cc40fbed2914b210ffe68","author":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"date":"2017-06-14 07:39:42.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Tim, looks good.\n\nIt doesn\u0027t seem like there is a mention in any of the docs about the change impact, it\u0027d be good to have a note about this. Perhaps, here doc/source/overview_erasure_code.rst?","accounts_in_message":[],"_revision_number":2},{"id":"0662b902d3edc12710676749c903e4be281fc004","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-06-16 17:56:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5d2ec61dd911cf81e3ac5fc634604e8660fc89b4","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-06-16 18:39:52.000000000","message":"Patch Set 3:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/6855/probetests/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/8469/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/468105/6492/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b21b4b16a1778fb84cf2be735ee32fa66305db50","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 19:23:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/05/468105/3/check/gate-swift-docs-ubuntu-xenial/c6b0db2//doc/build/html/ : SUCCESS in 3m 43s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-pep8-ubuntu-xenial/70edfa1/ : SUCCESS in 2m 19s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-python27-ubuntu-xenial/7c3dc69/ : SUCCESS in 3m 40s\n- gate-swift-python35 http://logs.openstack.org/05/468105/3/check/gate-swift-python35/69b464c/ : SUCCESS in 3m 05s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/5332b0f/ : SUCCESS in 1h 11m 21s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/2f27f7d/ : SUCCESS in 50m 52s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/b2f183b/ : SUCCESS in 5m 39s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/8325dbd/ : SUCCESS in 1m 07s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/47a80a5/ : SUCCESS in 5m 36s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/18b916d/ : SUCCESS in 4m 43s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/437729b/ : SUCCESS in 3m 23s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/73b6b48/ : SUCCESS in 3m 26s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/05/468105/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/cd27edd/ : SUCCESS in 1h 25m 25s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/05/468105/3/check/gate-swift-dsvm-functional-ubuntu-xenial/1857d1f/ : SUCCESS in 18m 33s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/05/468105/3/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/0dd5ba2/ : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4093bd18dc177ad7c3deeabfef0cce287d8e06c6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2017-06-19 01:05:48.000000000","message":"Patch Set 3: Code-Review+2\n\nYeah, think we\u0027ve talked about it enough. The warning in the docs looks good.\n\nWhen testing, things fail to start if I don\u0027t have it deprecated, but do (but warn) when i do.\n\nEnough talk, mergy mergy","accounts_in_message":[],"_revision_number":3},{"id":"8f69c6906b605f9ee5a41cd2ea5fe028564774de","author":{"_account_id":12261,"name":"Mahati","email":"mahati.chamarthy@gmail.com","username":"Mahati"},"date":"2017-06-19 04:19:37.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks, Looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"c34d78430e14b5a04f9500b8d50e8a33a780e4c0","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2017-06-21 21:05:07.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8ce1cee4b122cc506df488ea58431afd817688e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-21 21:05:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"a13421ef4d974afe61f1943c616ee9f1d9432797","date":"2017-06-22 01:11:03.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"9b4e615dee0cef825448c85d89a1c88d7cbf6bb8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-22 01:11:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/05/468105/3/gate/gate-swift-docs-ubuntu-xenial/ec31d8d//doc/build/html/ : SUCCESS in 2m 51s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-pep8-ubuntu-xenial/e58d78e/ : SUCCESS in 2m 58s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-python27-ubuntu-xenial/95e0b26/ : SUCCESS in 3m 14s\n- gate-swift-python35 http://logs.openstack.org/05/468105/3/gate/gate-swift-python35/db50b9c/ : SUCCESS in 1m 28s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3529854/ : SUCCESS in 1h 19m 21s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/c336eb1/ : SUCCESS in 48m 08s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/e98e000/ : SUCCESS in 3m 30s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/2e5e525/ : SUCCESS in 1m 50s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/a3bdc63/ : SUCCESS in 4m 56s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/dddd350/ : SUCCESS in 4m 43s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/a5c34c4/ : SUCCESS in 4m 43s\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/05/468105/3/gate/gate-swift-dsvm-functional-ubuntu-xenial/ad73401/ : SUCCESS in 16m 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2c3ac543f4106171997752d5784732bda6e6bf3e","revisions":{"e04cf3c4be4967c5c9acad74f44328a7d99bd257":{"kind":"REWORK","_number":1,"created":"2017-05-25 16:56:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/468105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/468105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/468105/1"}}},"commit":{"parents":[{"commit":"96de9ad126fb73dae30e999692b60145bc6a2009","subject":"Merge \"Clean up how PatchPolicies works\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96de9ad126fb73dae30e999692b60145bc6a2009"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-05-25 16:40:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-05-25 16:47:44.000000000","tz":-420},"subject":"Require that known-bad EC schemes be deprecated","message":"Require that known-bad EC schemes be deprecated\n\nWe said we were going to do it, we\u0027ve had two releases saying we\u0027d do\nit, we\u0027ve even backported our saying it to Newton -- let\u0027s actually do\nit.\n\nRelated-Change: I50159c9d19f2385d5f60112e9aaefa1a68098313\nChange-Id: I8f9de0bec01032d9d9b58848e2a76ac92e65ab09\nCloses-Bug: 1639691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e04cf3c4be4967c5c9acad74f44328a7d99bd257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e04cf3c4be4967c5c9acad74f44328a7d99bd257"}]},"branch":"refs/heads/master"},"620942a4eeffa1be1fdc24dd27e57141ab2b6d9c":{"kind":"REWORK","_number":2,"created":"2017-05-25 17:42:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/468105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/468105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/468105/2"}}},"commit":{"parents":[{"commit":"96de9ad126fb73dae30e999692b60145bc6a2009","subject":"Merge \"Clean up how PatchPolicies works\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96de9ad126fb73dae30e999692b60145bc6a2009"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-05-25 16:40:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-05-25 17:41:26.000000000","tz":-420},"subject":"Require that known-bad EC schemes be deprecated","message":"Require that known-bad EC schemes be deprecated\n\nWe said we were going to do it, we\u0027ve had two releases saying we\u0027d do\nit, we\u0027ve even backported our saying it to Newton -- let\u0027s actually do\nit.\n\nUpgrade Consideration\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nErasure-coded storage policies using isa_l_rs_vand and nparity \u003e\u003d 5 must\nbe configured as deprecated, preventing any new containers from being\ncreated with such a policy. This configuration is known to harm data\ndurability. Any data in such policies should be migrated to a new\npolicy. See https://bugs.launchpad.net/swift/+bug/1639691 for more\ninformation.\n\nUpgradeImpact\nRelated-Change: I50159c9d19f2385d5f60112e9aaefa1a68098313\nChange-Id: I8f9de0bec01032d9d9b58848e2a76ac92e65ab09\nCloses-Bug: 1639691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/620942a4eeffa1be1fdc24dd27e57141ab2b6d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/620942a4eeffa1be1fdc24dd27e57141ab2b6d9c"}]},"branch":"refs/heads/master"},"2c3ac543f4106171997752d5784732bda6e6bf3e":{"kind":"REWORK","_number":3,"created":"2017-06-16 17:56:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/468105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/468105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/468105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/468105/3"}}},"commit":{"parents":[{"commit":"96de9ad126fb73dae30e999692b60145bc6a2009","subject":"Merge \"Clean up how PatchPolicies works\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96de9ad126fb73dae30e999692b60145bc6a2009"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-05-25 16:40:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-06-16 17:58:43.000000000","tz":0},"subject":"Require that known-bad EC schemes be deprecated","message":"Require that known-bad EC schemes be deprecated\n\nWe said we were going to do it, we\u0027ve had two releases saying we\u0027d do\nit, we\u0027ve even backported our saying it to Newton -- let\u0027s actually do\nit.\n\nUpgrade Consideration\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nErasure-coded storage policies using isa_l_rs_vand and nparity \u003e\u003d 5 must\nbe configured as deprecated, preventing any new containers from being\ncreated with such a policy. This configuration is known to harm data\ndurability. Any data in such policies should be migrated to a new\npolicy. See https://bugs.launchpad.net/swift/+bug/1639691 for more\ninformation.\n\nUpgradeImpact\nRelated-Change: I50159c9d19f2385d5f60112e9aaefa1a68098313\nChange-Id: I8f9de0bec01032d9d9b58848e2a76ac92e65ab09\nCloses-Bug: 1639691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c3ac543f4106171997752d5784732bda6e6bf3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c3ac543f4106171997752d5784732bda6e6bf3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
