)]}'
{"id":"openstack%2Fswift~499260","triplet_id":"openstack%2Fswift~master~I3787200dd89831854ee8ef04dd6c3b8f8404f576","project":"openstack/swift","branch":"master","topic":"tlo","hashtags":[],"change_id":"I3787200dd89831854ee8ef04dd6c3b8f8404f576","subject":"WIP: Add TLO middleware","status":"NEW","created":"2017-08-30 19:05:17.000000000","updated":"2026-06-03 11:52:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"375dea132b74295eeb08d9c6d3726df9d4b3d6fe","_number":499260,"virtual_id_number":499260,"owner":{"_account_id":9216,"name":"Joel Wright","email":"joel.wright@gmail.com","username":"jjw-sohonet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-05 11:07:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"date":"2019-11-05 09:50:58.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-31 11:22:26.000000000","updated_by":{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},"reviewer":{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},"state":"REVIEWER"},{"updated":"2017-12-31 12:39:40.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 09:50:58.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-11-05 11:07:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-07 07:45:38.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-06-03 11:52:19.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REMOVED"}],"messages":[{"id":"386bf7bb67e4ee6105c78cb31d87d23bf8214581","author":{"_account_id":9216,"name":"Joel Wright","email":"joel.wright@gmail.com","username":"jjw-sohonet"},"date":"2017-08-30 19:05:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90e2224490c25cbe9015881700b4009d4990a179","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-08-30 19:50:23.000000000","message":"Patch Set 1:\n\nBuild succeeded \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/6941/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/8919/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/7305/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c494b1c69bf7f955071ac3768f91e77a3d1e8d40","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 20:27:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/60/499260/1/check/gate-swift-docs-ubuntu-xenial/974c801//doc/build/html/ : SUCCESS in 3m 31s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-pep8-ubuntu-xenial/13059a9/ : SUCCESS in 3m 34s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-python27-ubuntu-xenial/895fe67/ : SUCCESS in 6m 05s\n- gate-swift-python35 http://logs.openstack.org/60/499260/1/check/gate-swift-python35/f6aa4d0/ : SUCCESS in 3m 19s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/fc9142c/ : SUCCESS in 1h 20m 11s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4a23c00/ : SUCCESS in 1h 01m 59s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/142beec/ : SUCCESS in 6m 37s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/831a37c/ : SUCCESS in 2m 29s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/5a052f3/ : SUCCESS in 4m 36s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/e0d33a1/ : SUCCESS in 5m 44s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/1eed8e3/ : SUCCESS in 4m 39s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/ede5cb5/ : SUCCESS in 12m 18s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/60/499260/1/check/gate-swift-dsvm-functional-ubuntu-xenial/c20f31c/ : SUCCESS in 21m 43s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/60/499260/1/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/138c409/ : SUCCESS in 20m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ce24576fb8a88246e569f41d01ffd048586586e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-20 03:05:04.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"49183d40101cbe1d9cb6b97c856b559e3b8c7295","author":{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},"date":"2017-09-21 13:29:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fa16d16c4e7a02e0455c6daa40de420b03121efa","author":{"_account_id":9216,"name":"Joel Wright","email":"joel.wright@gmail.com","username":"jjw-sohonet"},"date":"2017-09-21 13:36:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"78f56284d37438f8d7741561ca80a93ec5c5d0b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 15:40:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/60/499260/3/check/gate-swift-docs-ubuntu-xenial/f3350aa//doc/build/html/ : SUCCESS in 4m 12s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-pep8-ubuntu-xenial/1204792/ : SUCCESS in 4m 02s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-python27-ubuntu-xenial/1527eec/ : SUCCESS in 6m 30s\n- gate-swift-python35 http://logs.openstack.org/60/499260/3/check/gate-swift-python35/24eb214/ : SUCCESS in 3m 00s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d90a8ec/ : SUCCESS in 1h 27m 42s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/cd229c8/ : SUCCESS in 49m 29s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/fde2611/ : SUCCESS in 7m 00s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/630126f/ : SUCCESS in 2m 18s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/6b96290/ : SUCCESS in 5m 36s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/10e4fbe/ : SUCCESS in 5m 28s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/3d96aa3/ : SUCCESS in 5m 41s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/7bb748e/ : SUCCESS in 7m 22s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/60/499260/3/check/gate-swift-dsvm-functional-ubuntu-xenial/96fe7c2/ : SUCCESS in 22m 33s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/60/499260/3/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/1d3f989/ : SUCCESS in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e2792eff34da29e9accbd75a03b805313d3bc8f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-06 23:01:38.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"699b26135aacd87f5a16c56cccbd822bbe80a782","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-10-10 18:45:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"fc0fa36473b3c53f3a17983e186ceeb2c862c90b","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 23:08:38.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/7798/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/7434/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/9412/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ecf153d2d762ce9ad06e502969bceb4930be8c20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-11 00:45:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-docs-ubuntu-xenial/20a043b/ : FAILURE in 3m 29s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-pep8-ubuntu-xenial/b1b1e71/ : SUCCESS in 3m 17s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-python27-ubuntu-xenial/c5f8528/ : FAILURE in 5m 32s\n- gate-swift-python35 http://logs.openstack.org/60/499260/4/check/gate-swift-python35/3060a5a/ : SUCCESS in 2m 42s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f5058f7/ : SUCCESS in 1h 28m 37s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/1034044/ : SUCCESS in 1h 06m 55s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/3eb0948/ : FAILURE in 6m 03s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/8f5fad3/ : SUCCESS in 2m 32s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/9696988/ : SUCCESS in 4m 29s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/6ff49df/ : SUCCESS in 5m 02s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/56b6975/ : SUCCESS in 7m 07s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/881ed8c/ : SUCCESS in 6m 54s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/60/499260/4/check/gate-swift-dsvm-functional-ubuntu-xenial/4b87eb1/ : SUCCESS in 28m 37s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/60/499260/4/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/9365c25/ : SUCCESS in 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0414eeb0f90574b954bdccfa91f9ffe3c5104216","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 01:35:16.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":4},{"id":"6805a56c35ed9d2f86b9030fb6c1645d3f8e2ac5","author":{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},"date":"2017-12-31 11:20:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"13fe50b640d97a8d5de9d2bbbf56d97a509048d4","author":{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},"date":"2017-12-31 11:22:26.000000000","message":"Patch Set 5:\n\nJust update concept to use data segments rather than pre/post-ambles. Addressing TODOs next.","accounts_in_message":[],"_revision_number":5},{"id":"1bc314aafd4921928efaa9ac40af03431c0d21e1","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-31 12:39:40.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/7764/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/9743/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/499260/8158/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c8866ea71cd0a4ec7f8dd74c52408f0415322d62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-31 12:49:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/60/499260/5/check/build-openstack-sphinx-docs/19471e8/html/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/60/499260/5/check/openstack-tox-pep8/72fc8f8/ : SUCCESS in 4m 32s\n- legacy-swift-dsvm-functional http://logs.openstack.org/60/499260/5/check/legacy-swift-dsvm-functional/9409997/ : SUCCESS in 28m 01s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/60/499260/5/check/legacy-swift-dsvm-functional-identity-v3-only/96a6559/ : SUCCESS in 27m 19s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/60/499260/5/check/legacy-tempest-dsvm-neutron-full/cd9bbb3/ : SUCCESS in 1h 26m 49s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/60/499260/5/check/legacy-grenade-dsvm-neutron/a04138a/ : SUCCESS in 56m 03s\n- swift-tox-py27 http://logs.openstack.org/60/499260/5/check/swift-tox-py27/7b1568d/ : SUCCESS in 6m 40s\n- swift-tox-py35 http://logs.openstack.org/60/499260/5/check/swift-tox-py35/83d0077/ : SUCCESS in 2m 42s\n- swift-tox-func http://logs.openstack.org/60/499260/5/check/swift-tox-func/85d23e9/ : SUCCESS in 7m 04s\n- swift-tox-func-post-as-copy http://logs.openstack.org/60/499260/5/check/swift-tox-func-post-as-copy/d28aa21/ : SUCCESS in 6m 57s\n- swift-tox-func-encryption http://logs.openstack.org/60/499260/5/check/swift-tox-func-encryption/12f3a2e/ : SUCCESS in 7m 32s\n- swift-tox-func-ec http://logs.openstack.org/60/499260/5/check/swift-tox-func-ec/75e8112/ : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":5},{"id":"e97ff080c4c4eb2ec8f75e42e121a2266e9982aa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-02 20:44:17.000000000","message":"Patch Set 5:\n\n(13 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dbd7d9f54edf5389ffc02acc8a4f85c1efd70b08","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-19 20:50:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"428a6c8df9f8eddc9d82236e87464e187d2ff40a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-19 20:51:21.000000000","message":"Patch Set 6:\n\n(10 comments)\n\nI\u0027m totally digging this :-)","accounts_in_message":[],"_revision_number":6},{"id":"e3e741ccd11889339364bda6c8a8c54a29e05ba8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-19 22:27:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ff765c4b12a1682d82e8092fb52cda20b4cd2262","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-20 02:27:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/60/499260/7/check/build-openstack-sphinx-docs/18ac12b/html/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/60/499260/7/check/openstack-tox-pep8/6983024/ : SUCCESS in 7m 04s\n- legacy-swift-dsvm-functional http://logs.openstack.org/60/499260/7/check/legacy-swift-dsvm-functional/96e392e/ : SUCCESS in 31m 57s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/60/499260/7/check/legacy-swift-dsvm-functional-identity-v3-only/de55d91/ : SUCCESS in 22m 58s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/60/499260/7/check/legacy-tempest-dsvm-neutron-full/24897d1/ : SUCCESS in 1h 34m 02s\n- neutron-grenade http://logs.openstack.org/60/499260/7/check/neutron-grenade/4aab0c1/ : SUCCESS in 56m 53s\n- swift-tox-py27 http://logs.openstack.org/60/499260/7/check/swift-tox-py27/af7c19b/ : SUCCESS in 18m 20s\n- swift-tox-py35 http://logs.openstack.org/60/499260/7/check/swift-tox-py35/0c12f31/ : SUCCESS in 3m 27s\n- swift-tox-func http://logs.openstack.org/60/499260/7/check/swift-tox-func/c13314e/ : SUCCESS in 10m 03s\n- swift-tox-func-encryption http://logs.openstack.org/60/499260/7/check/swift-tox-func-encryption/c8bd2d5/ : SUCCESS in 9m 04s\n- swift-tox-func-ec http://logs.openstack.org/60/499260/7/check/swift-tox-func-ec/15bd487/ : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":7},{"id":"7d3ab6439c0fb2f9624860084036c0dbfaa27f3a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-10 05:30:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f2650e6a648a36d59d98859e39f691e82cc12e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 07:00:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5c3fbff3575e483cb8ffd40d5810a17a : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb84d72aeeff49f9a872875f08948a43 : SUCCESS in 7m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2563ea2192d842edafa4e86fa43fce82 : SUCCESS in 52m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/022f65357a5846f0880dd5ea0f72f6ec : SUCCESS in 53m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb6e7fe2c8474e2590a18f8e8a5e7bce : SUCCESS in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0b5eee4cdaa24798a8d7cb1916898d2f : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/48a5551c1d084f4b82c8007fa5bd19bb : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ded90e5dc5c41f7877b3efe16a58f1b : SUCCESS in 13m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a9d95f81634463cac4143a431a54b57 : SUCCESS in 17m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/90021ce9b280427dadf12611bb16706b : FAILURE in 11m 20s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/09d11d6febe540c68296b4c032bb70db : SUCCESS in 24m 30s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1291cb5fe648420dafb7e7027d16e3d6 : SUCCESS in 11m 28s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d9a92857a2d243dbbadb653b68d82b27 : SUCCESS in 12m 21s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/28828af4c87d4204aaf05570292faf74 : FAILURE in 15m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ef95506d5e6643a4ab9b4e2b9d3c190c : SUCCESS in 9m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9d9e72a47d544a5ab83c71032ca7b795 : SUCCESS in 11m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7489c2be44d84f1fa0be702cd0cb2bbb : SUCCESS in 11m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/93208a80d55047698a03d7f9a26c3fbd : SUCCESS in 12m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1e61702bea4e4dc287f59d0ca83dac41 : FAILURE in 13m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/474d26fca75447cba801e4813fa3018c : SUCCESS in 15m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/89e406f18df34058814b043d9fc33ebe : SUCCESS in 31m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5da91a3cfaf44a9eb17e3a8d9df2422a : SUCCESS in 28m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/edda21bff3224ca6933c9b7f4fe7cc62 : SUCCESS in 30m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc8920ee93874aca8942a035db77d64f : FAILURE in 29m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16210ecf3e0e4d51bb5d5568e317ca5b : SUCCESS in 26m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b5144b8566422f9bad5161c4d5b9d6 : FAILURE in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d9eff275e78452cacdcccd0d47a873b : SUCCESS in 16m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5e0375ece1c465bad474f578f295493 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":8},{"id":"1866fa74951a5b4845c0a96cfdef238b1646ec8b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 07:17:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3b0cf9d142819fecc0b23f3da0af80521815a8e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 08:21:22.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/612b08e48fbd47e5bad0e583a530075e : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd1bb1c312a4014a6e919cd16c9d092 : SUCCESS in 9m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/463f63d313404110a066e3566b6f356c : SUCCESS in 57m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8e50c011af5495ba08491aa72e63344 : SUCCESS in 50m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00a38b0290c44954b51ba2644baa631a : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c472fe84146940f0bcd6c2cf78021a90 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b99a2b0d863a4f229902b31215629929 : SUCCESS in 12m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c546035719c44e4b524f0954b6ef23c : SUCCESS in 11m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1faef883354f446fb653d74ef9b09e3c : SUCCESS in 12m 58s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/2315f7208132466389d0e2d481607001 : SUCCESS in 9m 08s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b3935b02a076495aa3a620a8753a5c44 : SUCCESS in 10m 17s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/15b6e5d912d049f4a6b17492e2037238 : SUCCESS in 11m 16s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/cb67976c68f744c3b2dc3ff88bd595ae : SUCCESS in 11m 23s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e80a27b61acf41b6859b4387130cb0fd : FAILURE in 12m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/aa81a23eea734dada25eea2354e3d9d0 : SUCCESS in 10m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b6d422fb2a1442c7b80105b5b112c052 : SUCCESS in 10m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/39db9c75f9f7492da123a49454cb55db : SUCCESS in 10m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/652b595d1302472dbfb3d11c541af566 : SUCCESS in 12m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5a64787aef934510b25ac110754a4b41 : SUCCESS in 13m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed727d1009e0427ab4a31a30525acbbd : SUCCESS in 14m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/12960c480e634e328a93a1ff78a30c9d : SUCCESS in 31m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdd462800ffa4403b7f915f3076a4c4c : SUCCESS in 30m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9d67ca14a17644baa26905ca7ab4d2a5 : SUCCESS in 30m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78ed60bd6fc3417b8c104b39e49b3292 : SUCCESS in 27m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21e73e9619ae451ab5c09b516a349702 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56ada6ef99a453fae494161edbe9984 : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b51665ec78e4d2393ac1d32a26f8eb5 : SUCCESS in 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6061bfd75a7840c4b6d8dc4b3dda29bb : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":9},{"id":"98a96ee25c99469fe0e3d752286d3da7444c389c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 09:50:58.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"49801f48c9bd2a93d0cda85256309cf9f53e5ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:07:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3154d03543b04f30a1a131843d258ac3 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb1c7c1c54d4cef9f24d294b15f84bd : SUCCESS in 8m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e777270657ce408b9efd8c9aaaaa7f1e : SUCCESS in 52m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/984e3621b6b0426f9408140c4cbc6e98 : SUCCESS in 51m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/086923a6ecd3481d865343323d399a42 : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d9e3dc929aa741df95806bc69015ddc5 : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95601260d65741989d17c9b5664d13f6 : SUCCESS in 11m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e7bde6bcd9a44f88a476bb64babf92d6 : SUCCESS in 16m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/75db3457de054ccb84345ce47b197d78 : SUCCESS in 12m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c106ca14238a40578c623261cfa5defa : SUCCESS in 10m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4b9653cd4f74c98bd7d113687a1c694 : SUCCESS in 18m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/71484db9bc4d47f49c74ed3db1a53628 : SUCCESS in 10m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f549b0de4aa34c929466c99b780a970d : SUCCESS in 10m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7ac5fd27e9e7448f9c4a4eb89de5fee4 : SUCCESS in 12m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8f3b14d2d0f14145af6c756d8af46d4b : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1500f30fc74d41f9a7125d25dd1b79eb : SUCCESS in 10m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5846ca8f6ea547d3a74f78fbc2cd50ec : SUCCESS in 18m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/461a2400cc1c4293bbb32cbca0868937 : SUCCESS in 13m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d5dde161371b4d79b74fe14d2a688df9 : SUCCESS in 13m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dedcfc45d7e44fb483445b023fd99e4b : SUCCESS in 13m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/744c3e19f03b45cab2566f6e233447d5 : SUCCESS in 29m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/88e993f9873c4237a84751048e5db0c4 : SUCCESS in 27m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/69d54f2a563f42b1b6fbec8423471c77 : SUCCESS in 28m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ec813ce1ea248ee8110cc0870fc9095 : SUCCESS in 28m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d92ca2b7d2b74d9c9f54e958e5d13a46 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00234be5a7b94c83a0bdd1ecebca537d : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bba5e32219449909242aa46b8504987 : SUCCESS in 21m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d305baf05ee8410093e9f463bd844c2c : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":9},{"id":"375dea132b74295eeb08d9c6d3726df9d4b3d6fe","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-03 11:52:19.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_6968\u003e.","accounts_in_message":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}],"_revision_number":9}],"current_revision_number":9,"current_revision":"2f6685a53de54c67897a6768d41e1e8651bd91f1","revisions":{"4b16fd46f1af7c3f1ab178ce2dcac4c6297a4a51":{"kind":"REWORK","_number":1,"created":"2017-08-30 19:05:17.000000000","uploader":{"_account_id":9216,"name":"Joel Wright","email":"joel.wright@gmail.com","username":"jjw-sohonet"},"ref":"refs/changes/60/499260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/1"}}},"commit":{"parents":[{"commit":"6c08567c9a8d2f004de23db2899289653ddb3c71","subject":"Add Preamble and Postamble to SLO and SegmentedIterable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c08567c9a8d2f004de23db2899289653ddb3c71"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-30 19:00:52.000000000","tz":60},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b16fd46f1af7c3f1ab178ce2dcac4c6297a4a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b16fd46f1af7c3f1ab178ce2dcac4c6297a4a51"}]},"branch":"refs/heads/master"},"f1b9def838e53786a8774fdb87df448af0f54209":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2017-09-21 13:29:46.000000000","uploader":{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},"ref":"refs/changes/60/499260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/2"}}},"commit":{"parents":[{"commit":"c7d068ad47ec5d8a15a90be1e99935aa3b76a595","subject":"Merge \"Log deprecation warning for allow_versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7d068ad47ec5d8a15a90be1e99935aa3b76a595"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Joel Wright","email":"joel.wright@gmail.com","date":"2017-09-21 13:29:46.000000000","tz":0},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1b9def838e53786a8774fdb87df448af0f54209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1b9def838e53786a8774fdb87df448af0f54209"}]},"branch":"refs/heads/master"},"5376a9555e18269693551a7836c6dd5ada19aea4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-09-21 13:36:49.000000000","uploader":{"_account_id":9216,"name":"Joel Wright","email":"joel.wright@gmail.com","username":"jjw-sohonet"},"ref":"refs/changes/60/499260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/3"}}},"commit":{"parents":[{"commit":"04349ed8e26da10a8187d6f4dfe3a1fa6ccfbcb5","subject":"Add Preamble and Postamble to SLO and SegmentedIterable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04349ed8e26da10a8187d6f4dfe3a1fa6ccfbcb5"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-09-21 13:36:26.000000000","tz":60},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5376a9555e18269693551a7836c6dd5ada19aea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5376a9555e18269693551a7836c6dd5ada19aea4"}]},"branch":"refs/heads/master"},"4b7c9a65d6b9a7bb07a679b964fcaddc13b56454":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-10-10 18:45:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/499260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/4"}}},"commit":{"parents":[{"commit":"83ad470ff5382b3dc3423488986e9e9ceebd89ec","subject":"Add Preamble and Postamble to SLO and SegmentedIterable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83ad470ff5382b3dc3423488986e9e9ceebd89ec"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-10-10 18:45:31.000000000","tz":-420},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b7c9a65d6b9a7bb07a679b964fcaddc13b56454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b7c9a65d6b9a7bb07a679b964fcaddc13b56454"}]},"branch":"refs/heads/master"},"551a347df2560627a4ec4b515429b7ef499b4593":{"kind":"REWORK","_number":5,"created":"2017-12-31 11:20:04.000000000","uploader":{"_account_id":26946,"name":"Joel Wright","username":"joeljwright","inactive":true},"ref":"refs/changes/60/499260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/5"}}},"commit":{"parents":[{"commit":"734c8a18acf2fb0de1395e509cd8e643a35a4367","subject":"Add support for data segments to SLO and SegmentedIterable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/734c8a18acf2fb0de1395e509cd8e643a35a4367"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Joel Wright","email":"joel.wright@gmail.com","date":"2017-12-31 11:17:52.000000000","tz":0},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/551a347df2560627a4ec4b515429b7ef499b4593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/551a347df2560627a4ec4b515429b7ef499b4593"}]},"branch":"refs/heads/master"},"f335d8b5d5fd2bc0d0d3aa756364ed3027caf084":{"kind":"REWORK","_number":6,"created":"2018-01-19 20:50:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/499260/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/6"}}},"commit":{"parents":[{"commit":"d740a9a3d829c276be5cfc2d97d1827e9f50faa5","subject":"Add support for data segments to SLO and SegmentedIterable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d740a9a3d829c276be5cfc2d97d1827e9f50faa5"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-19 20:49:29.000000000","tz":0},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f335d8b5d5fd2bc0d0d3aa756364ed3027caf084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f335d8b5d5fd2bc0d0d3aa756364ed3027caf084"}]},"branch":"refs/heads/master"},"655e000cde63f428d83ac6a1f92316ef19f6bb16":{"kind":"REWORK","_number":7,"created":"2018-01-19 22:27:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/499260/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/7"}}},"commit":{"parents":[{"commit":"d740a9a3d829c276be5cfc2d97d1827e9f50faa5","subject":"Add support for data segments to SLO and SegmentedIterable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d740a9a3d829c276be5cfc2d97d1827e9f50faa5"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-19 22:27:12.000000000","tz":0},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/655e000cde63f428d83ac6a1f92316ef19f6bb16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/655e000cde63f428d83ac6a1f92316ef19f6bb16"}]},"branch":"refs/heads/master"},"9c502eb3326be79e6e972fafdd571be9c1f04e70":{"kind":"REWORK","_number":8,"created":"2019-10-10 05:30:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/499260/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/8"}}},"commit":{"parents":[{"commit":"6374a5f92e1f50e7d321312dedcafaebeef2bb13","subject":"slo_manifest_hook follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6374a5f92e1f50e7d321312dedcafaebeef2bb13"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-10 05:30:20.000000000","tz":-420},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c502eb3326be79e6e972fafdd571be9c1f04e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c502eb3326be79e6e972fafdd571be9c1f04e70"}]},"branch":"refs/heads/master"},"2f6685a53de54c67897a6768d41e1e8651bd91f1":{"kind":"REWORK","_number":9,"created":"2019-11-05 07:17:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/499260/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/499260/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/499260/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/499260/9"}}},"commit":{"parents":[{"commit":"93d1842dcd043bb804d5c9f175ebac4c3bb88a38","subject":"slo_manifest_hook follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93d1842dcd043bb804d5c9f175ebac4c3bb88a38"}]}],"author":{"name":"Joel Wright","email":"joel.wright@sohonet.com","date":"2017-08-09 08:44:05.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-05 07:15:52.000000000","tz":0},"subject":"WIP: Add TLO middleware","message":"WIP: Add TLO middleware\n\nThis patch introduces a middleware for creating tar objects\nwith a middleware in a similar style to SLO. The only\nactions this middleware performs are to transform a tar\nmanifest into an SLO manifest that makes use of preamble\nand postamble data for the tar headers and padding.\n\nChange-Id: I3787200dd89831854ee8ef04dd6c3b8f8404f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f6685a53de54c67897a6768d41e1e8651bd91f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f6685a53de54c67897a6768d41e1e8651bd91f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
