)]}'
{"id":"openstack%2Fswift~507808","triplet_id":"openstack%2Fswift~master~Ib537065a88ba3b7d97235d74cf8bcae465ccf055","project":"openstack/swift","branch":"master","topic":"bug/1521351","hashtags":[],"change_id":"Ib537065a88ba3b7d97235d74cf8bcae465ccf055","subject":"Add ability to undelete an account.","status":"NEW","created":"2017-09-27 09:47:33.000000000","updated":"2020-01-29 04:50:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":174,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6891dd2f77843d5b08daee87b9775bd21dc30f33","_number":507808,"virtual_id_number":507808,"owner":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"date":"2020-01-29 04:50:56.000000000","_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"date":"2019-11-08 03:22:31.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2019-07-10 07:36:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"date":"2019-11-08 04:05:23.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-04 21:32:54.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2017-10-11 15:44:21.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-25 08:44:10.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-26 08:34:16.000000000","updated_by":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"reviewer":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"state":"REVIEWER"},{"updated":"2017-11-02 12:51:02.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2018-08-08 14:33:52.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2018-08-23 06:45:06.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-07-10 07:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-08 03:22:31.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-11-08 04:05:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"ed2704bc742f448644dceb1dbc0779585bed04a0","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-09-27 09:47:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ad9dc72ee7b6f201fd71846ea9f7d7881e870f8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2017-09-27 09:56:49.000000000","message":"Patch Set 1:\n\nhow does this relate to this path https://review.openstack.org/#/c/445160 ?","accounts_in_message":[],"_revision_number":1},{"id":"f8dd3f1322d1f225db07fc00823f5801b7330926","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-09-27 10:05:35.000000000","message":"Patch Set 1:\n\nThis might be related with the path  https://review.openstack.org/#/c/445160 because the bug id is same.","accounts_in_message":[],"_revision_number":1},{"id":"73928f881a51ac4cf4de3a146b759e5a4c801d9d","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2017-09-27 10:37:22.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI copied this comment from your abandoned patch [1]:\n\nPlease update the \"revive a deleted account\" procedure in: doc/source/ops_runbook/procedures.rst as part of this patch. End-users won\u0027t have the commit message available to them, so won\u0027t know how (and when) to run the curl command to undelete the account.\n\n[1] https://review.openstack.org/#/c/507785","accounts_in_message":[],"_revision_number":1},{"id":"ea37a40b48010d9628fa4c65ba72ae0d8e3fde86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-27 10:40:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-docs-ubuntu-xenial/87fdeb0/ : FAILURE in 3m 20s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-pep8-ubuntu-xenial/1cf5985/ : FAILURE in 3m 29s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-python27-ubuntu-xenial/fabf37c/ : FAILURE in 4m 00s\n- gate-swift-python35 http://logs.openstack.org/08/507808/1/check/gate-swift-python35/16ee9ec/ : SUCCESS in 2m 05s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0ea465f/ : FAILURE in 23m 14s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7b20fe4/ : FAILURE in 49m 54s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/94473b9/ : FAILURE in 7m 30s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/1d911ee/ : SUCCESS in 2m 59s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/43926a0/ : FAILURE in 1m 53s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/6706c18/ : FAILURE in 2m 11s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/77e5f30/ : FAILURE in 2m 31s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/bb7f461/ : FAILURE in 2m 27s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/08/507808/1/check/gate-swift-dsvm-functional-ubuntu-xenial/c8f24fc/ : FAILURE in 11m 59s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/08/507808/1/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/af8448a/ : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ca85000abde812afd9465da0c608fe09983f0ba","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-27 12:49:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7272/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9250/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7636/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"06776975461db9b0a7711ca2e849849f6024ee3d","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-09-28 08:53:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f41648569c77fdebb2310cfaee05753db8a894a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-28 09:37:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-docs-ubuntu-xenial/5482ec1/ : FAILURE in 3m 28s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-pep8-ubuntu-xenial/99a1e2c/ : FAILURE in 3m 28s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-python27-ubuntu-xenial/7dc70e8/ : FAILURE in 3m 55s\n- gate-swift-python35 http://logs.openstack.org/08/507808/2/check/gate-swift-python35/43a32c9/ : SUCCESS in 2m 41s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/eae2d97/ : FAILURE in 20m 35s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ceb54f0/ : FAILURE in 41m 17s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/a525ad2/ : FAILURE in 5m 13s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/0be5198/ : SUCCESS in 3m 03s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/9e480aa/ : FAILURE in 2m 00s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/5cc9157/ : FAILURE in 2m 51s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/096b12d/ : FAILURE in 2m 07s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/a08d58f/ : FAILURE in 2m 28s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/08/507808/2/check/gate-swift-dsvm-functional-ubuntu-xenial/53f8767/ : FAILURE in 12m 18s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/08/507808/2/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/b2c34e6/ : FAILURE in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6701f53251ae2910e7a13b32997fcb89dd60c481","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-09-28 10:17:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5420ba48838fc961a6851a179a3f283aa186f884","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-28 10:59:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-docs-ubuntu-xenial/0ccac21/ : FAILURE in 3m 10s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-pep8-ubuntu-xenial/86521f4/ : SUCCESS in 3m 04s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-python27-ubuntu-xenial/91e5933/ : FAILURE in 4m 08s\n- gate-swift-python35 http://logs.openstack.org/08/507808/3/check/gate-swift-python35/c02d4ec/ : SUCCESS in 2m 16s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f0c0157/ : FAILURE in 24m 45s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a21e5a2/ : FAILURE in 39m 52s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/d3d8e0b/ : FAILURE in 3m 42s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/8e10841/ : SUCCESS in 2m 33s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/8774c91/ : FAILURE in 2m 24s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/cf75d0f/ : FAILURE in 2m 49s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/f1ff2c3/ : FAILURE in 2m 45s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/9c48741/ : FAILURE in 1m 33s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/08/507808/3/check/gate-swift-dsvm-functional-ubuntu-xenial/0d479ea/ : FAILURE in 12m 47s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/08/507808/3/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/00cd9ee/ : FAILURE in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2b6e2acdeb4e7b213ff00739e00a40c7a4f0173","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-28 14:37:45.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7307/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9285/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7671/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"83b6a83291feaec045979c0c1a85dd004855c486","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-09-28 15:02:20.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7308/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9286/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7672/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"41a7013c3349f9043155b9882af3c7b6c573b68e","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2017-09-28 15:52:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8249599ed0ae1ec7e5061bc4ba590c3669df86dc","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-03 09:13:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"26cf93a24bff6a5343a43e37f490972afe2c5b18","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-03 11:57:41.000000000","message":"Patch Set 4:\n\nBuild succeeded \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7331/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9309/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7695/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6871eac9341239da1e17c82e8bd12ffb6070945a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-03 13:18:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/4/check/build-openstack-sphinx-docs/e114eb5/html/latest/ : SUCCESS in 6m 56s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/4/check/openstack-tox-pep8/b3dafde/ : SUCCESS in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/4/check/openstack-tox-py27/62c4f4c/ : SUCCESS in 10m 43s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/4/check/openstack-tox-py35/211d144/ : SUCCESS in 6m 59s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/4/check/legacy-swift-tox-xfs-tmp-py27/7afe5ee/ : SUCCESS in 10m 43s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/4/check/legacy-swift-tox-xfs-tmp-py35/455f53b/ : SUCCESS in 7m 06s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/4/check/legacy-swift-tox-xfs-tmp-func/46c50f2/ : SUCCESS in 15m 09s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/4/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/5f1ada9/ : SUCCESS in 14m 00s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/4/check/legacy-swift-tox-xfs-tmp-func-encryption/a0ce330/ : SUCCESS in 10m 29s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/4/check/legacy-swift-tox-xfs-tmp-func-ec/9b7c13e/ : SUCCESS in 15m 22s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/4/check/legacy-swift-dsvm-functional/5e7c0a4/ : SUCCESS in 27m 47s (non-voting)\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/4/check/legacy-swift-dsvm-functional-identity-v3-only/1a3105b/ : SUCCESS in 24m 41s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/4/check/legacy-tempest-dsvm-neutron-full/4f7595c/ : POST_FAILURE in 2h 25m 51s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/4/check/legacy-grenade-dsvm-neutron/f912e32/ : POST_FAILURE in 2h 19m 13s","accounts_in_message":[],"_revision_number":4},{"id":"5456a7cf754f0e89395ab91e52765b85ec1fa389","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-10-04 18:17:12.000000000","message":"Patch Set 4:\n\nrecheck\n\nI don\u0027t trust that gate result at all.","accounts_in_message":[],"_revision_number":4},{"id":"94a76763315bd43a837cc9ad129a2389975d1a2c","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-10-04 20:43:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nIMHO, the default delay_reaping must be updated to make this feature useful on a default deployment.","accounts_in_message":[],"_revision_number":4},{"id":"904d1fbc4680dc3a0ddded03ca709fcb47893092","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-10-04 21:13:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d370212eca26609dbb799fce155b0bf96f25ad3b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2017-10-04 21:32:54.000000000","message":"Patch Set 4:\n\nI wonder if we shouldn\u0027t be reviewing https://review.openstack.org/#/c/445160 which pre-dates this","accounts_in_message":[],"_revision_number":4},{"id":"6dbb03e5e70d27d4c755c4ed28b5fded80c2fb5f","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2017-10-04 21:34:43.000000000","message":"Patch Set 4:\n\nWe discussed this patch in the Swift team meeting today, and we\u0027re pretty excited about it. It looks like you\u0027re getting some reviews.\n\nIt would probably be helpful for you (the author) and us (the reviewers) to chat in IRC about it, but we can\u0027t figure out your IRC nick based on the patch author. Can you drop by #openstack-swift on freenode and say \"hi\"?","accounts_in_message":[],"_revision_number":4},{"id":"18c86f9e0852f00d18ab6c9ed1b86b786784ea8f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-04 22:18:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/08/507808/4/check/gate-swift-docs-ubuntu-xenial/26d5af9//doc/build/html/ : SUCCESS in 4m 08s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-pep8-ubuntu-xenial/6d28240/ : SUCCESS in 4m 24s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-python27-ubuntu-xenial/ad4dd9c/ : SUCCESS in 6m 23s\n- gate-swift-python35 http://logs.openstack.org/08/507808/4/check/gate-swift-python35/da04439/ : SUCCESS in 3m 20s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6a6c6a7/ : FAILURE in 1h 16m 15s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/bc1e63d/ : SUCCESS in 1h 14m 43s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/98fb1eb/ : SUCCESS in 5m 41s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/47c391f/ : SUCCESS in 2m 34s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/49773fd/ : SUCCESS in 5m 05s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/b9c0413/ : SUCCESS in 5m 18s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/18442d1/ : SUCCESS in 5m 03s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/91a4eb0/ : SUCCESS in 5m 50s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/08/507808/4/check/gate-swift-dsvm-functional-ubuntu-xenial/ab983a1/ : SUCCESS in 32m 03s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/08/507808/4/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/f6fc998/ : SUCCESS in 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"700618987a9bfbf16fe6a39d7893a87140041e71","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-05 08:34:57.000000000","message":"Patch Set 4:\n\n(1 comment)\n\n\u003e recheck\n \u003e \n \u003e I don\u0027t trust that gate result at all.\n\nThanks for the review suggestion.\nCan you bit elaborate us. What exactly you want here?","accounts_in_message":[],"_revision_number":4},{"id":"1c5ffb02f0b01b8ba4654ee6b5f579e9936dc482","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-05 08:55:10.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\u003eAlso concerned about setting delete_ts to 0, which would lost the \n\u003einformation that the account was undeleted. checking the code at \u003e[1], it seems we can safely keep the delete timestamp untouched.\n\nAs we have analyzed the source code behaviour for Account Creation and Account Deletion the fields for \"time stamp\" and \"status\" are getting changed:\n1) During Account Creation the following fields updated:\na) delete_timestamp with 0\nb) created_at with timestamp\nc) put_timestamp with timestamp\nd) status_changed_at with timestamp\ne) status with blank\n\n2) During Account Deletion the following fields updated:\na) delete_timestamp with timestamp\nb) status with \"DELETED\"\nc) status_changed_at with timestamp\n\nWe thought that if we are undeleting the account then the following fields should be updated: \na) deleted_timestamp with 0\nb) status_changed_at with timestamp\nc) put_timestamp with timestamp\nd) status with blank","accounts_in_message":[],"_revision_number":4},{"id":"6a83f4c337fc0fe8283669cb26b6f36af044ef70","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-05 10:46:48.000000000","message":"Patch Set 4:\n\n\u003e (2 comments)\n \u003e \n \u003e IMHO, the default delay_reaping must be updated to make this\n \u003e feature useful on a default deployment.\n\n\u003e I\u0027m expecting that a meta will be stored. So having a non-meta\n\u003e header would feel more appropriate to me.\n\nWe have introduced the meta in header to differentiate between the normal operation and the undelete operation.\n\nIf we would store this meta rather than providing into the header then where we need store it?","accounts_in_message":[],"_revision_number":4},{"id":"1a395c0ce2ff214ef8cf7067d9c81919b5d07dfd","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-05 10:48:04.000000000","message":"Patch Set 4:\n\n\u003e (8 comments)\n\n\u003e I think it\u0027s safe to *replace* most of the above steps with your \u003e procedure.\n\nDoes it means to replace the entire stuff mentioned in Procedure.rst for \"Procedure: Revive a deleted account\"\nfrom line number 262 to 359?","accounts_in_message":[],"_revision_number":4},{"id":"6de09e49cc45729ab9fc5b8385fb308bf197093f","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2017-10-05 13:36:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1790cf97c0ca954f7f765ae15d065069da19d55a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2017-10-05 23:05:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)\n\ni agree with Romain!\n\nSee some inline comments. But other then that, thanks heaps for this work.","accounts_in_message":[],"_revision_number":4},{"id":"2d27ec9407cbac1784eb018ba322423b78ed40f0","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-06 05:48:22.000000000","message":"Patch Set 4:\n\n\u003e (6 comments)\n \u003e \n \u003e i agree with Romain!\n \u003e \n \u003e See some inline comments. But other then that, thanks heaps for\n \u003e this work.\n \u003e swift/account/backend.py\n \u003e Line 217:\n \u003e +1 to this. Updating the put_timestamp is all that is needed, \n \u003e don\u0027t nuke the \u003edelete_timestamp.\n\nAs the reviewers suggested to leave the delete_timestamp to untouched (i.e. not set to 0). We will update it and provide it in the Patch Set 5.\n\n \u003e swift/account/server.py\n \u003e Line 149:\n \u003e yeah, depending on what value you supply.. which seems to be a \n \u003e boolean means on true, \u003ethis will be stored in the account\u0027s  \n \u003e metadata.\n \u003e We don\u0027t need to store it if the header is just used as a  \n \u003e trigger to undelete the container.\n\nThis header is used to trigger the undelete the account, as the \nreviewers confirmed we will not be storing the header. The code remains as it is in Patch Set 5.\n\n \u003e swift/common/db.py\n \u003e Line 316:\n \u003e do we need to have an extra line here?\n\nWe will update it and provide it in the Patch Set 5.\n\n \u003e Line 320:\n \u003e This has nothing to do with this patch, but do we really need \n \u003e to clear db metadata when \u003emarking deleted? This is a problem \n \u003e we\u0027re facing in sharding, where there is a case where we \u003emay \n \u003e need to undelete shard containers, which would be much simpler \n \u003e if all the metadata \u003ewasn\u0027t removed.\n \u003e But the same problem persists here with account undeletion, \n \u003e what about all the metadata \u003eattached to the account before it \n \u003e was deleted? Maybe this is something we need to raise and \u003eask \n \u003e the community about.\n\nAs this is not related with the Patch which we have provided here, so do we need to work on this or this can be treated as the separate point of investigation or task?\n\n \u003e swift/container/backend.py\n \u003e Line 375:\n \u003e Same as Romain mentioned earlier, let\u0027s not set \n \u003e delete_timestamp to 0. Like in delete_db, \u003ewe don\u0027t 0 out the \n \u003e put_timestamp.\n \u003e test/unit/common/test_db.py\n \u003e Line 291:\n \u003e here to\n\nCan you please elaborate more that for which field you are talking about, as we will not be setting the delete_timestamp to 0? Then what change you want here for put_timestamp. Please confirm.","accounts_in_message":[],"_revision_number":4},{"id":"3ce7854ae9810a47561f62608fa87b91fa3a930c","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-06 05:54:05.000000000","message":"Patch Set 4:\n\n\u003e swift/account/server.py\n \u003e Line 150:\n \u003e should it check delay_reaping and refuse to undelete if it\u0027s \n \u003e too late? once the \u003ereaper started, i\u0027m not sure it will stop \n \u003e if account is undeleted, which can create \u003econfusion for users \n \u003e that would see objects disapearing after undeletion\n\nWe will analyse the required changes here to accommodate the delay_reaping.","accounts_in_message":[],"_revision_number":4},{"id":"faef0e7661c501c7ac371d0f142a07b3e1c73cb1","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-06 08:12:27.000000000","message":"Patch Set 4:\n\n\u003e \u003e swift/account/server.py\n \u003e \u003e Line 150:\n \u003e \u003e should it check delay_reaping and refuse to undelete if it\u0027s\n \u003e \u003e too late? once the \u003ereaper started, i\u0027m not sure it will stop\n \u003e \u003e if account is undeleted, which can create \u003econfusion for users\n \u003e \u003e that would see objects disapearing after undeletion\n\nAs we have analysed the account reaper behaviour, we have verified it by the changing the following key in account-server.conf file.\n\n1) Changed the value of interval to 60 seconds under [account-reaper] section In account-server.conf .\n\n2) Keep the default value of delay_reaper to 0 under [account-reaper] section In account-server.conf .\n\nSteps executed:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nStep1: Executed the account-reaper service.\nStep2: Delete the account by executing the curl command.\nStep3: Verified that reaper has deleted all the containers and objects from that account.\nStep4: Execute the curl command to undelete the account, account get successfully undeleted with no containers and objects.\n\nPlease confirm what exactly you want here to handle the delay_reaper?","accounts_in_message":[],"_revision_number":4},{"id":"8b9214c58325984ef01766e77b35a0d8f8604b45","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-10-09 08:04:24.000000000","message":"Patch Set 4: -Code-Review\n\n\u003e Step4: Execute the curl command to undelete the account, account\n \u003e get successfully undeleted with no containers and objects.\n \nThat\u0027s the issue I think. There is no point in having undeletion if some or all objects are lost. And, once undeletion is started, it won\u0027t be stopped before all objects are deleted, so even if undeleted, user would lost all objects.\n\nI feel like the delay_reaper should be set to a reasonable default (7 days?), and that undeletion should only be allowed before that delay is passed. So, when user undeleted an account, no objects were already lost.","accounts_in_message":[],"_revision_number":4},{"id":"fb5778f69e00e955cf166b17dbd1e3275895ceee","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-10 10:23:12.000000000","message":"Patch Set 4:\n\n\u003e \u003e Step4: Execute the curl command to undelete the account, account\n \u003e \u003e get successfully undeleted with no containers and objects.\n \u003e \n \u003e That\u0027s the issue I think. There is no point in having undeletion if\n \u003e some or all objects are lost. And, once undeletion is started, it\n \u003e won\u0027t be stopped before all objects are deleted, so even if\n \u003e undeleted, user would lost all objects.\n \u003e \n \u003e I feel like the delay_reaper should be set to a reasonable default\n \u003e (7 days?), and that undeletion should only be allowed before that\n \u003e delay is passed. So, when user undeleted an account, no objects\n \u003e were already lost.\n\nThanks for the suggestion.\n\nAfter analyzing your suggestion for delay_reaper, we have derived the following understanding:\n\nWe will introduce some checks before going to undelete the account and the check would validate as like following:\n\n1)Calculate current time (e.g. 1507586442.98)\n2)Extract the delete_timestamp of the account (e.g. 1506983453.43)\n3)Extract the value of delay_reaping from configuration file (e.g. 604800 (7 days in seconds))\n4)Compare it as like below:\n(1)– (2) \u003c (3) \n\ne.g. 1507586442.98 -  1506983453.43 \u003c  604800\ne.g. 602991.86 \u003c 604800\ni.e. TRUE, then the deleted account will be undeleted otherwise it won’t.","accounts_in_message":[],"_revision_number":4},{"id":"36b327d4d6ed63a9112cfb10191378fdfa32e77c","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-11 10:43:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0ca61eb28eac1253a361a8746af764e73516446a","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-11 11:52:58.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7450/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9428/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7814/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"22bbab38abf803f24e9c0ce364b487b0d9754b72","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-11 15:44:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-swift-docs-ubuntu-xenial http://docs-draft.openstack.org/08/507808/5/check/gate-swift-docs-ubuntu-xenial/83a365d//doc/build/html/ : SUCCESS in 4m 22s\n- gate-swift-pep8-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-pep8-ubuntu-xenial/351cdfc/ : SUCCESS in 4m 08s\n- gate-swift-python27-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-python27-ubuntu-xenial/d53a47c/ : FAILURE in 4m 39s\n- gate-swift-python35 http://logs.openstack.org/08/507808/5/check/gate-swift-python35/d197eca/ : SUCCESS in 2m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c75b3e1/ : SUCCESS in 1h 27m 27s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-grenade-dsvm-neutron-ubuntu-xenial/0bce85d/ : SUCCESS in 49m 43s\n- gate-swift-tox-xfs-tmp-py27-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-tox-xfs-tmp-py27-ubuntu-xenial/0c35d1e/ : FAILURE in 4m 34s\n- gate-swift-tox-xfs-tmp-py35-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-tox-xfs-tmp-py35-ubuntu-xenial/ef02076/ : SUCCESS in 2m 22s\n- gate-swift-tox-xfs-tmp-func-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-tox-xfs-tmp-func-ubuntu-xenial/6e6e976/ : FAILURE in 2m 21s\n- gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-tox-xfs-tmp-func-post-as-copy-ubuntu-xenial/2073a74/ : FAILURE in 1m 37s\n- gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-tox-xfs-tmp-func-encryption-ubuntu-xenial/2cc3b2f/ : FAILURE in 2m 15s\n- gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-tox-xfs-tmp-func-ec-ubuntu-xenial/4660c6a/ : FAILURE in 3m 34s (non-voting)\n- gate-swift-dsvm-functional-ubuntu-xenial http://logs.openstack.org/08/507808/5/check/gate-swift-dsvm-functional-ubuntu-xenial/928e07f/ : SUCCESS in 23m 33s\n- gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv http://logs.openstack.org/08/507808/5/check/gate-swift-dsvm-functional-identity-v3-only-ubuntu-xenial-nv/856a737/ : SUCCESS in 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"06564adcbed2b243138c90e6f4fc58517db03aa9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-10-11 18:19:44.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\ntests failures are real... the problem is the account server can\u0027t always read the reaper config... take a look at how the object-auditor handles ValueError and IOError when the config-file/section isn\u0027t available.\n\nIdeally you want to have an explicit option for the account server, preferably lifting the single documented option up to DEFAULT where it\u0027s available to all the services that need it.  Then the access to the other config section implicitly is just a best-effort for transparent upgrade.\n\nIf you need some more details try to catch up with me via IRC in #openstack-swift on freenode.","accounts_in_message":[],"_revision_number":5},{"id":"a1e35380ac963a3ed7f8402f5506315bbe492b24","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-12 06:05:58.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n \u003e \n \u003e tests failures are real... the problem is the account server can\u0027t\n \u003e always read the reaper config... take a look at how the\n \u003e object-auditor handles ValueError and IOError when the\n \u003e config-file/section isn\u0027t available.\n \u003e \n \u003e Ideally you want to have an explicit option for the account server,\n \u003e preferably lifting the single documented option up to DEFAULT where\n \u003e it\u0027s available to all the services that need it.  Then the access\n \u003e to the other config section implicitly is just a best-effort for\n \u003e transparent upgrade.\n \u003e \n \u003e If you need some more details try to catch up with me via IRC in\n \u003e #openstack-swift on freenode.\n\nThanks Clay for your valuable inputs.\n\nWe would like to ask the following:\n\n1) If we lift up the delay_reaping to DEFAULT section of the configuration file then it would be accessible from the DEFAULT section always.\n\n2) The other section named [account-reaper] which consist of this key (i.e. delay_reaping) in account-server.conf but not in account-server\\1.conf to 4.conf (these files only has the empty section named as [account-reaper]). \n\n3) If we would add this key in DEFAULT section, then this will lead the duplicate key in configuration files and user need to set the value on two places. If we go with this approach, then what will be name of key and the following configuration files should also be updated here:\na) /etc/account-server.conf-sample (this will be used as account-server.conf in actual)\nb) /doc/saio/swift/account-server/1.conf\nc) /doc/saio/swift/account-server/2.conf\nd) /doc/saio/swift/account-server/3.conf\ne) /doc/saio/swift/account-server/4.conf\n\n4) If we would be adding this key in DEFAULT section of all the configuration files, then where to update in the documentation side.\n\nPlease confirm.","accounts_in_message":[],"_revision_number":5},{"id":"4920d79164f2050dd524bcc5671ba6b344ba4bd4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-10-12 18:35:29.000000000","message":"Patch Set 5:\n\nHope this clears things up!  Sorry i missed you on IRC, but I\u0027m sure anyone who\u0027s around in #openstack-swift could help explain the pattern we use for moving config options up to DEFAULT (e.g. reclaim_age, rsync_timeout) - just point them at this change.\n\n\u003e 1) If we lift up the delay_reaping to DEFAULT section of the configuration file\n\u003e then it would be accessible from the DEFAULT section always.\nThat sounds great!  Anytime we need a single option in multiple services we\nmove it to the DEFAULT, e.g. bind_port.\n\n\u003e 2) The other section named [account-reaper] which consist of this key (i.e.\n\u003e delay_reaping) in account-server.conf but not in account-server\\1.conf to\n\u003e 4.conf (these files only has the empty section named as [account-reaper]). \nThe example configs document all the options, the saio configs are minimal\nconfigs - any options ommited use the default docuemnted values.\n\n\u003e 3) If we would add this key in DEFAULT section, then this will lead the\n\u003e duplicate key in configuration files and user need to set the value on two\n\u003e places. If we go with this approach, then what will be name of key and the\n\u003e following configuration files should also be updated here:\n\u003e     a) /etc/account-server.conf-sample (this will be used as account-server.conf in actual)\n\u003e     b) /doc/saio/swift/account-server/1.conf\n\u003e     c) /doc/saio/swift/account-server/2.conf\n\u003e     d) /doc/saio/swift/account-server/3.conf\n\u003e     e) /doc/saio/swift/account-server/4.conf\nYou should *move* the option from [account-reaper] to [DEFAULT]\n/etc/account-server.conf-sample - changing the example config won\u0027t have any\neffect on deployed clusters..  Since the saio configs currently omit the option\nyou don\u0027t need to change anything in the example saio configs.\n\n\u003e 4) If we would be adding this key in DEFAULT section of all the configuration\n\u003e files, then where to update in the documentation side.\nYou would document the change in multiple places, the commit message should\nexplain the change and be flagged UpgradeImpact.  And any docs for the\naccount-server.conf should reflect the change:\n\n * /etc/account-server.conf-sample\n * doc/source/deployment_guide.rst\n * doc/manpages/account-server.conf.5\n\nBut you can *also* add some code for backwards compatibility.  But the code\nmust be robust against failure.  The account server should first and foremost\ntry to get the value of the option from it\u0027s config (probably coming from\n[DEFAULT] section if the deployment is following the docs).  If it can\u0027t find\nthe value in it\u0027s config, only then would it *try* to read the [account-reaper]\nsection, but if that fails for any reason it should assume the default value.","accounts_in_message":[],"_revision_number":5},{"id":"3ce45ab6f1e9e0661994157794d04109a4fb4bb3","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-16 08:16:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"81ab2e5c2b409e3cef421ff989413e4abedf3473","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-16 09:13:42.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7504/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9482/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7868/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"729ca6e8de6dbb97ab563eba75cb0154e1bdcb28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-16 09:44:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/6/check/openstack-tox-pep8/22d29f5/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/6/check/openstack-tox-py27/8e9a546/ : FAILURE in 5m 39s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/6/check/openstack-tox-py35/c200d7e/ : SUCCESS in 2m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/6/check/build-openstack-sphinx-docs/f880259/html/ : SUCCESS in 4m 12s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/6/check/legacy-swift-tox-xfs-tmp-py27/fec10bf/ : FAILURE in 4m 53s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/6/check/legacy-swift-tox-xfs-tmp-py35/b36d883/ : SUCCESS in 2m 53s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/6/check/legacy-swift-tox-xfs-tmp-func/04b4d3a/ : FAILURE in 2m 04s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/6/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/afeecf3/ : FAILURE in 2m 28s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/6/check/legacy-swift-tox-xfs-tmp-func-encryption/3b42160/ : FAILURE in 3m 42s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/6/check/legacy-swift-tox-xfs-tmp-func-ec/85eb6fc/ : FAILURE in 4m 48s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/6/check/legacy-swift-dsvm-functional/f554c6a/ : SUCCESS in 21m 54s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/6/check/legacy-swift-dsvm-functional-identity-v3-only/70136c6/ : SUCCESS in 21m 48s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/6/check/legacy-tempest-dsvm-neutron-full/ccd9118/ : SUCCESS in 1h 22m 42s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/6/check/legacy-grenade-dsvm-neutron/7aae94f/ : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":6},{"id":"ae188515dc749066350fc624dc2ef7584e758818","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-17 09:42:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3cf17af0d502cc92056310938059b0444d555bdc","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-17 10:21:11.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9501/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7523/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7887/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"2f995733645e481856ee9a4178869f06fef18902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-17 11:06:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/7/check/openstack-tox-pep8/bd53f19/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/7/check/openstack-tox-py27/4f2ef63/ : SUCCESS in 5m 44s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/7/check/openstack-tox-py35/d7b0dd7/ : SUCCESS in 2m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/7/check/build-openstack-sphinx-docs/59cd68f/html/ : SUCCESS in 4m 09s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-py27/b9a40ae/ : FAILURE in 5m 55s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-py35/158fb58/ : SUCCESS in 3m 17s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func/547bcb0/ : SUCCESS in 5m 08s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/caf4005/ : SUCCESS in 6m 35s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func-encryption/b6d8662/ : SUCCESS in 7m 41s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func-ec/aa71594/ : SUCCESS in 5m 39s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/7/check/legacy-swift-dsvm-functional/cce0467/ : SUCCESS in 21m 18s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/7/check/legacy-swift-dsvm-functional-identity-v3-only/d919fba/ : SUCCESS in 26m 43s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/7/check/legacy-tempest-dsvm-neutron-full/ed4c7d3/ : SUCCESS in 1h 21m 37s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/7/check/legacy-grenade-dsvm-neutron/cc9127f/ : SUCCESS in 51m 08s","accounts_in_message":[],"_revision_number":7},{"id":"e8c1e88835b74bb80e62641757c1ef15fe52d022","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-10-17 19:44:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"96dbc87ceabd5d04d05dcbd0f40229aaa2d8eb18","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-10-17 19:58:22.000000000","message":"Patch Set 7: Code-Review+1\n\n(5 comments)\n\nok, i\u0027m starting to get a sense for how this will work - thanks for putting so much time into it!  Hopefully we can get that test passing, it looks like the change could use some more tests in general (esp exercising the new account server api, i.e. test/unit/account/test_server)","accounts_in_message":[],"_revision_number":7},{"id":"1301e2954ee6335c0c6928e48699df32ed7c7bcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-17 21:51:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/7/check/openstack-tox-pep8/cdd982b/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/7/check/openstack-tox-py27/f007284/ : SUCCESS in 6m 23s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/7/check/openstack-tox-py35/e20c774/ : SUCCESS in 3m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/7/check/build-openstack-sphinx-docs/00b0a87/html/ : SUCCESS in 4m 39s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-py27/198484a/ : SUCCESS in 6m 22s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-py35/c610868/ : SUCCESS in 3m 01s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func/339e59d/ : SUCCESS in 5m 40s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/8429729/ : SUCCESS in 6m 34s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func-encryption/21a2a7d/ : SUCCESS in 6m 13s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/7/check/legacy-swift-tox-xfs-tmp-func-ec/7b4a8d0/ : SUCCESS in 6m 07s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/7/check/legacy-swift-dsvm-functional/8555ce7/ : SUCCESS in 21m 56s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/7/check/legacy-swift-dsvm-functional-identity-v3-only/cdad726/ : SUCCESS in 21m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/7/check/legacy-tempest-dsvm-neutron-full/255ff38/ : SUCCESS in 1h 34m 02s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/7/check/legacy-grenade-dsvm-neutron/da66eb6/ : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":7},{"id":"ac369bb48ce2181bb41d2bfcda55c31d516e2834","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-18 07:03:07.000000000","message":"Patch Set 5:\n\n\u003e (5 comments)\n \u003e \n \u003e ok, i\u0027m starting to get a sense for how this will work - thanks for\n \u003e putting so much time into it!  Hopefully we can get that test\n \u003e passing, it looks like the change could use some more tests in\n \u003e general (esp exercising the new account server api, i.e.\n \u003e test/unit/account/test_server)\n\u003e swift/account/backend.py\n\u003e Line 58:\n\u003e those backend queries are *so* similar - I wonder if it would \n\u003e make sense to grow a stat_table class attribute\n\u003e swift/account/server.py\nPlease bit detail what is the point of benefit to add stat_table class attribute in scope of this fix.\n\n\u003e Line 68:\n\u003e this line can\u0027t  raise KeyError or IOError (that I can see?) -  \u003e the ValueError is possible if the conf dict has \n\u003e {\u0027delay_reaping\u0027: \u0027foo\u0027}... but I\u0027m not sure the end result is  \u003e the correct failure mode fallback behavior.\nWe have validate the below sample cases for Value and Key Error:\n\nExample Cases verified:\n\nCase 1: Wrong configuration file path specified in read_conf\n[DEFAULT]\ndelay_reaping \u003d \n[account-reaper]\ndelay_reaping \u003d 2\nResult: This is a case of KeyError and delay_reaping gets value of constant.\n\nCase 2: Configuration FilePath is correct\n[DEFAULT]\ndelay_reaping \u003d \n[account-reaper]\ndelay_reaping \u003d \nResult: This is a case of ValueError and delay_reaping gets value of constant.\n\nCase 3: Wrong configuration file path specified in read_conf\n[DEFAULT]\ndelay_reaping \u003d 1\n[account-reaper]\ndelay_reaping \u003d 2\nResult: The value of delay_reaping will be 1.\n\nCase 4: Wrong configuration file path specified in read_conf\n[DEFAULT]\n[account-reaper]\ndelay_reaping \u003d 2\nResult: The value of delay_reaping will be constant.\n\nCase 5: Configuration FilePath is correct\n[DEFAULT]\ndelay_reaping \u003d \n[account-reaper]\ndelay_reaping \u003d 2\nResult: The value of delay_reaping will be 2.\n\nFor IO Error: It can be thrown while there is some internal system error during reading the configuration file.\n\n\u003e Line 163:\n\u003e I don\u0027t love this API - this is in the user meta name space for \u003e one... I\u0027m not even sure a header is the right interface (as \n\u003e opposed to a different method or a query string..\n\u003e I need to doublecheck ownership permissions for swift_owner PUT \u003e to their own account... it\u0027s probably safe to assume you can\u0027t  \u003e get a valid token once your account has been marked for \n\u003e deletion; but idk\nPlease detail bit more on this.\n\n\u003e test/unit/account/test_backend.py\n\u003e Line 130:\n\u003e should container/test_backend have a similar test?\nWe don\u0027t think, that test should be in container test backend to undelete the account because the delete account is also not part of the container test backend.","accounts_in_message":[],"_revision_number":5},{"id":"a01737ca34562b9405ab26d18184c2129a31e9fe","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-18 07:08:30.000000000","message":"Patch Set 7:\n\n\u003e (5 comments)\n\u003e\n\u003e ok, i\u0027m starting to get a sense for how this will work - thanks for\n\u003e putting so much time into it!  Hopefully we can get that test\n\u003e passing, it looks like the change could use some more tests in\n\u003e general (esp exercising the new account server api, i.e.\n\u003e test/unit/account/test_server)\n\u003e swift/account/backend.py\n\u003e Line 58:\n\u003e those backend queries are *so* similar - I wonder if it would\n\u003e make sense to grow a stat_table class attribute\n\u003e swift/account/server.py\nPlease bit detail what is the point of benefit to add stat_table\nclass attribute in scope of this fix.\n \n\u003e Line 68:\n\u003e this line can\u0027t  raise KeyError or IOError (that I can see?) -  \n\u003e the ValueError is possible if the conf dict has\n\u003e {\u0027delay_reaping\u0027: \u0027foo\u0027}... but I\u0027m not sure the end result is  \n\u003e the correct failure mode fallback behavior.\nWe have validate the below sample cases for Value and Key Error:\n \nExample Cases verified:\n \nCase 1: Wrong configuration file path specified in read_conf\n[DEFAULT]\ndelay_reaping \u003d\n[account-reaper]\ndelay_reaping \u003d 2\nResult: This is a case of KeyError and delay_reaping gets value of\nconstant.\n\nCase 2: Configuration FilePath is correct\n[DEFAULT]\ndelay_reaping \u003d\n[account-reaper]\ndelay_reaping \u003d\nResult: This is a case of ValueError and delay_reaping gets value\nof constant.\n \nCase 3: Wrong configuration file path specified in read_conf\n[DEFAULT]\ndelay_reaping \u003d 1\n[account-reaper]\ndelay_reaping \u003d 2\nResult: The value of delay_reaping will be 1.\n \nCase 4: Wrong configuration file path specified in read_conf\n[DEFAULT]\n[account-reaper]\ndelay_reaping \u003d 2\nResult: The value of delay_reaping will be constant.\n \nCase 5: Configuration FilePath is correct\n[DEFAULT]\ndelay_reaping \u003d\n[account-reaper]\ndelay_reaping \u003d 2\nResult: The value of delay_reaping will be 2.\n \nFor IO Error: It can be thrown while there is some internal system\nerror during reading the configuration file.\n \n\u003e Line 163:\n\u003e I don\u0027t love this API - this is in the user meta name space for\n\u003e one... I\u0027m not even sure a header is the right interface (as\n\u003e opposed to a different method or a query string..\n\u003e I need to doublecheck ownership permissions for swift_owner PUT\n\u003e to their own account... it\u0027s probably safe to assume you can\u0027t\n\u003e get a valid token once your account has been marked for\n\u003e deletion; but idk\nPlease detail bit more on this.\n \n\u003e test/unit/account/test_backend.py\n\u003e Line 130:\n\u003e should container/test_backend have a similar test?\nWe don\u0027t think, that test should be in container test backend to\nundelete the account because the delete account is also not part of\nthe container test backend.","accounts_in_message":[],"_revision_number":7},{"id":"5c90064f9c262474fa50da1b7cc6a56baa86b933","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-10-18 23:36:36.000000000","message":"Patch Set 7: Code-Review-1\n\nit\u0027s easier to track discussion on specific inline comments if you leave click reply on the inline comment.\n\nI\u0027ll consider your rebuttals and try to get you some diffs to demonstrate the issues that I see as needing to be addressed (if any); in the meantime please add some code to test/unit/account/test_server that exercises the new api","accounts_in_message":[],"_revision_number":7},{"id":"293048aeac7921f07554c89daa64a168ddf264fe","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2017-10-19 08:56:43.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)\n\nBTW: you can view the formatted documents by clicking on the build-openstack-sphinx-docs zuul check job. For example, in this patch, the link is http://logs.openstack.org/08/507808/7/check/build-openstack-sphinx-docs/00b0a87/html/","accounts_in_message":[],"_revision_number":7},{"id":"8601c153b471b4bf93223d19939e4d23dfa2af85","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-23 10:36:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bed6dec407270efefe83c9c17cecfa10ccb3fe94","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 11:23:16.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7900/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9514/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7536/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"db2d3f5cad8bbbefdc8081ce9eabbcbc70792e98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-23 11:59:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/8/check/openstack-tox-pep8/f7d66a9/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/8/check/openstack-tox-py27/415192a/ : SUCCESS in 8m 35s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/8/check/openstack-tox-py35/3dd1891/ : SUCCESS in 3m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/8/check/build-openstack-sphinx-docs/cc71abb/html/ : SUCCESS in 5m 15s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/8/check/legacy-swift-tox-xfs-tmp-py27/8c411e1/ : SUCCESS in 8m 50s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/8/check/legacy-swift-tox-xfs-tmp-py35/3ce2983/ : SUCCESS in 3m 15s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/8/check/legacy-swift-tox-xfs-tmp-func/59fb687/ : SUCCESS in 10m 40s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/8/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/0742de7/ : SUCCESS in 6m 24s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/8/check/legacy-swift-tox-xfs-tmp-func-encryption/d254dae/ : SUCCESS in 8m 34s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/8/check/legacy-swift-tox-xfs-tmp-func-ec/0c47c7d/ : SUCCESS in 5m 19s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/8/check/legacy-swift-dsvm-functional/50c7829/ : SUCCESS in 25m 29s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/8/check/legacy-swift-dsvm-functional-identity-v3-only/055b490/ : SUCCESS in 21m 48s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/8/check/legacy-tempest-dsvm-neutron-full/0e1d60a/ : SUCCESS in 1h 21m 35s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/8/check/legacy-grenade-dsvm-neutron/9297a03/ : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":8},{"id":"240b10d41931c620b3e213294f67c73dd02d9f7b","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-23 15:56:21.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c2d26775624ad3dc40dac9cd9f238d60a73a218d","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-23 16:08:29.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f9710d0982974c0f97cd7060d0504a40db3c2fb0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2017-10-23 20:02:32.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)\n\nOk, i was really confused about why we needed the undelete support for the container - but it\u0027s because we have some test infrastructure trying to keep us honest about api parity of Account/Container/Example brokers!!!\n\nThat\u0027s probably good - this feature doesn\u0027t strictly require the method on the container broker but it\u0027s easier to just add it.  Nothing in tree consumes it, but if we need it later as a starting point it\u0027ll work just like the account brokers\u0027\n\nThat was my major confusion - I personally would want to do some functional validation to put my +2 on it - but as near as I can tell there\u0027s no exposed harm unless you try to use the feature.  The design, docs and tests look good!","accounts_in_message":[],"_revision_number":8},{"id":"12d90322718b3fa0783e3da46b4e0c661d489be4","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-24 08:00:48.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9353a97ad1a35158618956f5c74e3300ff6c3bd6","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2017-10-24 13:05:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3a4e4a6ac1ce247b8c315ae692765d7d8924f1f0","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-24 16:06:54.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"6339a17d72c7064b1995d5e72b6818f454f18570","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-24 16:07:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4ba60f1477fc6aefae3035a369ac912ee11e7902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-24 17:47:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/9/check/openstack-tox-pep8/a9cbdba/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/9/check/openstack-tox-py27/2b4df41/ : SUCCESS in 8m 18s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/9/check/openstack-tox-py35/46bcadb/ : SUCCESS in 5m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/9/check/build-openstack-sphinx-docs/e66fdc6/ : FAILURE in 6m 01s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/9/check/legacy-swift-tox-xfs-tmp-py27/1af7b8b/ : SUCCESS in 7m 31s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/9/check/legacy-swift-tox-xfs-tmp-py35/31ffa7c/ : SUCCESS in 3m 37s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/9/check/legacy-swift-tox-xfs-tmp-func/367b670/ : SUCCESS in 7m 29s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/9/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/be620e4/ : SUCCESS in 7m 12s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/9/check/legacy-swift-tox-xfs-tmp-func-encryption/78d912c/ : SUCCESS in 7m 00s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/9/check/legacy-swift-tox-xfs-tmp-func-ec/e277649/ : SUCCESS in 18m 25s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/9/check/legacy-swift-dsvm-functional/a3901df/ : SUCCESS in 19m 48s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/9/check/legacy-swift-dsvm-functional-identity-v3-only/d002825/ : SUCCESS in 19m 56s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/9/check/legacy-tempest-dsvm-neutron-full/df133a6/ : SUCCESS in 1h 31m 35s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/9/check/legacy-grenade-dsvm-neutron/5028a02/ : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":9},{"id":"67a9693e3efeb75ad9300158ad17f5807bf4a695","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-24 19:28:47.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7917/probetests/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9531/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7553/ : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"215c3c26fd5f04ed40dd601b0691ee23e273d102","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-25 05:17:07.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"98512bebcda3eac750e59de4c8a8810d0d20a6b8","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-25 06:41:14.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7926/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7562/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9540/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"9c2a667bb74afb1b4b5bc7053c1a971edd88eed3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 06:46:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/10/check/openstack-tox-pep8/4e05bbe/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/10/check/openstack-tox-py27/e98637d/ : SUCCESS in 8m 01s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/10/check/openstack-tox-py35/9296361/ : SUCCESS in 2m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/10/check/build-openstack-sphinx-docs/25e1756/html/ : SUCCESS in 5m 25s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/10/check/legacy-swift-tox-xfs-tmp-py27/7e552d7/ : SUCCESS in 7m 17s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/10/check/legacy-swift-tox-xfs-tmp-py35/1e077e1/ : SUCCESS in 2m 59s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/10/check/legacy-swift-tox-xfs-tmp-func/904d7ef/ : SUCCESS in 5m 47s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/10/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/1f7d8df/ : SUCCESS in 7m 33s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/10/check/legacy-swift-tox-xfs-tmp-func-encryption/0852003/ : SUCCESS in 9m 15s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/10/check/legacy-swift-tox-xfs-tmp-func-ec/1723bad/ : SUCCESS in 6m 53s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/10/check/legacy-swift-dsvm-functional/cd6cbff/ : SUCCESS in 18m 29s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/10/check/legacy-swift-dsvm-functional-identity-v3-only/4622ea9/ : SUCCESS in 21m 28s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/10/check/legacy-tempest-dsvm-neutron-full/2700e6f/ : SUCCESS in 1h 26m 38s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/10/check/legacy-grenade-dsvm-neutron/61bb998/ : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":10},{"id":"62bacaadc90ce074d648f34ebf1ca6a48d26912d","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2017-10-25 07:47:43.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"134c629b4a8e441f0278966924d9a511121661b8","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-25 07:58:45.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"991f8ed8636ae6f377580449428308bd049592b5","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-10-25 08:00:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a749f3e70a798e19cbd9f8ffcd52bcbb955ec8c9","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-10-25 08:44:10.000000000","message":"Patch Set 11:\n\nBuild succeeded \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7927/probetests/ : SUCCESS\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/9541/ : SUCCESS\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/507808/7563/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"e11debba8c7d536385c33f3e47e4e7a6f2626ef5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 09:40:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/11/check/openstack-tox-pep8/4bb744e/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/11/check/openstack-tox-py27/3c81ed2/ : SUCCESS in 8m 34s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/11/check/openstack-tox-py35/5e2a606/ : SUCCESS in 4m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/11/check/build-openstack-sphinx-docs/2d8f9b7/html/ : SUCCESS in 5m 30s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/11/check/legacy-swift-tox-xfs-tmp-py27/efff344/ : SUCCESS in 8m 52s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/11/check/legacy-swift-tox-xfs-tmp-py35/2a4014b/ : SUCCESS in 4m 03s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/11/check/legacy-swift-tox-xfs-tmp-func/49d7c38/ : SUCCESS in 7m 48s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/11/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/d18f46d/ : SUCCESS in 6m 18s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/11/check/legacy-swift-tox-xfs-tmp-func-encryption/8fbde37/ : SUCCESS in 6m 41s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/11/check/legacy-swift-tox-xfs-tmp-func-ec/cba47f0/ : SUCCESS in 7m 42s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/11/check/legacy-swift-dsvm-functional/782b3f0/ : SUCCESS in 27m 27s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/11/check/legacy-swift-dsvm-functional-identity-v3-only/df964d0/ : SUCCESS in 21m 45s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/11/check/legacy-tempest-dsvm-neutron-full/ff037bb/ : SUCCESS in 1h 32m 14s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/11/check/legacy-grenade-dsvm-neutron/b9da646/ : SUCCESS in 47m 13s","accounts_in_message":[],"_revision_number":11},{"id":"6f670bf5f94cc0114c9f45208fd58c1999a03685","author":{"_account_id":860,"name":"Donagh McCabe","email":"donagh.mccabe@gmail.com","username":"donagh-mccabe"},"date":"2017-10-26 08:34:16.000000000","message":"Patch Set 11: Code-Review+1\n\n+1 docs only; I\u0027ve not reviewed the code.","accounts_in_message":[],"_revision_number":11},{"id":"a4bcf95c509f75d8f77f47876639e8c9f7bf4a59","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-11-02 12:51:02.000000000","message":"Patch Set 11: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e456265960660bfffe0b0bbaf8fcb037a6f38996","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-11-03 09:57:56.000000000","message":"Patch Set 11:\n\n(11 comments)\n\nFixed comments as provided by Romain over PS7, and added few observations to confirm.","accounts_in_message":[],"_revision_number":11},{"id":"ae41749f75a2b9dd6fcd9210a0b38095c9add358","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-11-03 10:39:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"63459f3dd3064079a707dc827465d080f2df2149","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-03 12:16:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/12/check/openstack-tox-pep8/23e8c23/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/12/check/openstack-tox-py27/f6c05e3/ : SUCCESS in 7m 00s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/12/check/openstack-tox-py35/c625ccd/ : SUCCESS in 2m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/12/check/build-openstack-sphinx-docs/7111d4d/html/ : SUCCESS in 4m 02s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/12/check/legacy-swift-tox-xfs-tmp-py27/eca0d7a/ : SUCCESS in 7m 47s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/12/check/legacy-swift-tox-xfs-tmp-py35/f30f2eb/ : SUCCESS in 2m 47s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/12/check/legacy-swift-tox-xfs-tmp-func/16d9d7b/ : SUCCESS in 5m 08s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/12/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/91acacf/ : SUCCESS in 6m 27s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/12/check/legacy-swift-tox-xfs-tmp-func-encryption/b6f8eb7/ : SUCCESS in 5m 23s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/12/check/legacy-swift-tox-xfs-tmp-func-ec/0235ca4/ : SUCCESS in 5m 28s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/12/check/legacy-swift-dsvm-functional/b7bbe8f/ : SUCCESS in 21m 26s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/12/check/legacy-swift-dsvm-functional-identity-v3-only/38ea2e7/ : SUCCESS in 21m 54s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/12/check/legacy-tempest-dsvm-neutron-full/148f268/ : SUCCESS in 1h 35m 13s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/12/check/legacy-grenade-dsvm-neutron/a763fe0/ : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":12},{"id":"2d0af7536086b888d8942b3b9e9d64ec8b0c498d","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-11-16 10:33:35.000000000","message":"Patch Set 12:\n\nReviewers, Please provide the suggestions (if any).","accounts_in_message":[],"_revision_number":12},{"id":"62b224fa5209275c2a3e904405c4eb035db1722c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2017-11-22 05:17:20.000000000","message":"Patch Set 12: Code-Review-1\n\n(8 comments)\n\nNice, thanks for working on this. Some inline comments. Happy to attach a more detailed patch to show you want I mean if this doesn\u0027t make sense.\n\nAlso, adding grabbing the delay_reaping to the account server, which is cool. However, in the account if there it\u0027s not defined it\u0027ll default to 7 days.. however the reaper will default to 0. Shouldn\u0027t they be the same. otherwise the reaper may reap while the account server thinks it\u0027s ok?","accounts_in_message":[],"_revision_number":12},{"id":"1066b829b9a73a2059521fa84fecebd2fc5c4bd8","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-11-22 09:23:50.000000000","message":"Patch Set 12:\n\n(2 comments)\n\nThanks for the review suggestions, we have added our few inline comments for more clarity on it.\n\n\u003eAlso, adding grabbing the delay_reaping to the account server, \u003ewhich is cool. However, in the account if there it\u0027s not defined \u003eit\u0027ll default to 7 days.. however the reaper will default to 0. \u003eShouldn\u0027t they be the same. otherwise the reaper may reap while \u003ethe account server thinks it\u0027s ok?\n\nIn documentation side we have updated the value of dealy_reaping \u003d 604800 (i.e. 7 days) in the following files:\n\n1. deployment_guide.rst\n2. account-server.conf.5\n\nYou are saying something different here? Please confirm.","accounts_in_message":[],"_revision_number":12},{"id":"2a5fbfaf370d3e274f58fe2467b85c2cda546c18","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2017-11-23 06:30:16.000000000","message":"Patch Set 12:\n\nI was thinking something like: http://paste.openstack.org/show/627173/\n\nWhich is a patch to this change set. Although I haven\u0027t had a chance to test etc.","accounts_in_message":[],"_revision_number":12},{"id":"4da3612898383ab9647444ecc84929fa751d4805","author":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"date":"2017-11-23 09:03:31.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d0c5fd45d9952d46df42162f0d96cf2ed0f316c8","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-11-24 08:40:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"206177be08a7b94d9e0ac8ff6c704fd2e21092d5","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-11-24 08:40:53.000000000","message":"Patch Set 12:\n\n(10 comments)","accounts_in_message":[],"_revision_number":12},{"id":"375e4300efb89e4117066215bf11ed3f7a4acb55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-24 09:54:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/13/check/openstack-tox-pep8/af02027/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/08/507808/13/check/openstack-tox-py27/3d94522/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/08/507808/13/check/openstack-tox-py35/5489753/ : SUCCESS in 2m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/13/check/build-openstack-sphinx-docs/b55a5db/html/ : SUCCESS in 4m 04s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/08/507808/13/check/legacy-swift-tox-xfs-tmp-py27/5988a08/ : SUCCESS in 6m 51s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/08/507808/13/check/legacy-swift-tox-xfs-tmp-py35/c381b06/ : SUCCESS in 2m 49s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/08/507808/13/check/legacy-swift-tox-xfs-tmp-func/d45ab03/ : SUCCESS in 5m 31s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/08/507808/13/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/bba8d11/ : SUCCESS in 5m 40s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/08/507808/13/check/legacy-swift-tox-xfs-tmp-func-encryption/09643fb/ : SUCCESS in 5m 19s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/08/507808/13/check/legacy-swift-tox-xfs-tmp-func-ec/9a112fb/ : SUCCESS in 6m 32s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/13/check/legacy-swift-dsvm-functional/e260300/ : SUCCESS in 20m 03s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/13/check/legacy-swift-dsvm-functional-identity-v3-only/97cce12/ : SUCCESS in 23m 20s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/13/check/legacy-tempest-dsvm-neutron-full/b9ebdf5/ : SUCCESS in 1h 10m 01s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/08/507808/13/check/legacy-grenade-dsvm-neutron/9c4a83b/ : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":13},{"id":"aee4d7436c9a3b0da93dfb8a8e27dacbdbee6b27","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-12-07 09:53:01.000000000","message":"Patch Set 13:\n\nReviewers, Please provide the suggestions (if any).","accounts_in_message":[],"_revision_number":13},{"id":"b713f768bc5db1ca00d2895fae49503d14bea0b7","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2017-12-15 05:11:05.000000000","message":"Patch Set 13:\n\nReviewers, Please provide the suggestions (if any).","accounts_in_message":[],"_revision_number":13},{"id":"f6f3431e2f821f8588020ad40ddb2721fb1a6e14","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-28 01:16:14.000000000","message":"Patch Set 13: Code-Review-1\n\n(11 comments)\n\nWorks like a champ! As an admin, I do a\n\n $ curl -H \u0027x-auth-token: \u003cadmin token\u003e\u0027 $OS_STORAGE_URL -X DELETE\n \u003e DELETE /v1/AUTH_test HTTP/1.1\n ...\n \u003c HTTP/1.1 204 No Content\n ...\n\n...and then as a user all I get back is\n\n $ curl -H \u0027x-auth-token: \u003cuser token\u003e\u0027 $OS_STORAGE_URL\n \u003e GET /v1/AUTH_test HTTP/1.1\n ...\n \u003c HTTP/1.1 410 Gone\n ...\n\nUser can\u0027t revive it:\n\n $ curl -H \u0027x-auth-token: \u003cuser token\u003e\u0027 $OS_STORAGE_URL -X PUT -H \u0027x-undelete-account: true\u0027\n \u003e PUT /v1/AUTH_test HTTP/1.1\n ...\n \u003c HTTP/1.1 403 Forbidden\n ...\n\nBut the admin can!\n\n $ curl -H \u0027x-auth-token: \u003cadmin token\u003e\u0027 $OS_STORAGE_URL -X PUT -H \u0027x-undelete-account: true\u0027\n \u003e PUT /v1/AUTH_test HTTP/1.1\n ...\n \u003c HTTP/1.1 202 Accepted\n ...\n\nAnd then everything\u0027s back to normal:\n\n $ curl -H \u0027x-auth-token: \u003cuser token\u003e\u0027 $OS_STORAGE_URL\n \u003e GET /v1/AUTH_test HTTP/1.1\n ...\n \u003c HTTP/1.1 200 OK\n ...\n\nA few more thoughts, though:\n\nIt might be worth including an UpgradeImpact, since we\u0027re both relocating a config option and updating its default. For operators that currently explicitly set the option, what should they do here? Update config and then upgrade, I\u0027m assuming? Out of curiosity, what are the consequences of doing it in the wrong order?\n\nWhat happens when at least one account server misses the X-Undelete-Account request? Are we relying on replication to make sure it eventually gets in sync before it starts reaping? Or does the reaper try to get consensus and refuse to do anything unless it knows that the other replicas agree that the account should be deleted? This might be a good opportunity for a probe test...\n\nI think once we resolve\n\n- the deployment guide comment (pretty sure that was just choosing the wrong side of a merge conflict),\n- what to tell operators who are upgrading, and\n- better restrictions on who can revive an account\n\n...I\u0027ll be at least +1; I\u0027ll feel even better once we\n\n- have a probe test,\n- refuse to start with bad configs, and\n- recommend GET instead of PUT for checking the recently-deleted status\n\n...but maybe we could do those as follow-ups?","accounts_in_message":[],"_revision_number":13},{"id":"9e6e82e6b63e5add9a994bd82a841e53edb7b578","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-01-08 06:25:01.000000000","message":"Patch Set 13:\n\n(11 comments)\n\nPlease provide the views over the added inline comments.","accounts_in_message":[],"_revision_number":13},{"id":"99e717376cbf237d985f34837c8e962b6d1c0e92","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-01-30 10:32:40.000000000","message":"Patch Set 13:\n\nReviewers, Please provide the suggestions on the follow-up questions asked by us.","accounts_in_message":[],"_revision_number":13},{"id":"b5e2c075da8ad737748d17cdae57b9d0e2a8643a","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-02-05 06:17:40.000000000","message":"Patch Set 13:\n\nReviewers, Please provide the suggestions on the follow-up questions asked by us.","accounts_in_message":[],"_revision_number":13},{"id":"bd74f817a71b8cb9e155e4559f17a1f17fabbbae","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-02-05 06:17:50.000000000","message":"Patch Set 13:\n\nReviewers, Please provide the suggestions on the follow-up questions asked by us.","accounts_in_message":[],"_revision_number":13},{"id":"326e89f9565e0732ee097e90f589bf3576f2ddfe","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-02-20 08:17:27.000000000","message":"Patch Set 13:\n\n(2 comments)\n\nFixed 2 Comments.","accounts_in_message":[],"_revision_number":13},{"id":"be4deff23c62f8aa9ec5e252cbc7a2f301621828","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-02-20 08:40:07.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1bb172ea761c4f838f04f79a9d8859c19ca8e90e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-20 10:13:37.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/14/check/build-openstack-sphinx-docs/5fdea05/html/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/14/check/openstack-tox-pep8/5a1afbe/ : SUCCESS in 3m 43s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/14/check/legacy-swift-dsvm-functional/9c65b5d/ : SUCCESS in 21m 56s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/14/check/legacy-swift-dsvm-functional-identity-v3-only/0ccb219/ : SUCCESS in 23m 34s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/14/check/legacy-tempest-dsvm-neutron-full/629ded6/ : SUCCESS in 1h 27m 58s\n- neutron-grenade http://logs.openstack.org/08/507808/14/check/neutron-grenade/a7595ce/ : SUCCESS in 45m 59s\n- swift-tox-py27 http://logs.openstack.org/08/507808/14/check/swift-tox-py27/e5de83a/ : SUCCESS in 6m 29s\n- swift-tox-py35 http://logs.openstack.org/08/507808/14/check/swift-tox-py35/4c64b62/ : SUCCESS in 3m 04s\n- swift-tox-func http://logs.openstack.org/08/507808/14/check/swift-tox-func/9829d63/ : SUCCESS in 8m 36s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/14/check/swift-tox-func-encryption/52862bd/ : SUCCESS in 7m 09s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/14/check/swift-tox-func-ec/7e5fba6/ : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":14},{"id":"8b1432c7a9be8bf3d59c71b2e6d1f4a39fdf3c5a","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-03-05 05:58:02.000000000","message":"Patch Set 14:\n\nReviewers, Please provide the suggestions (if any).","accounts_in_message":[],"_revision_number":14},{"id":"9e27e146df4afb36f534591a136e14a728008785","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-03-16 08:20:00.000000000","message":"Patch Set 14:\n\nReviewers, Please provide the suggestions (if any).","accounts_in_message":[],"_revision_number":14},{"id":"4d50509a0825e9a21d69334a7d3ee530b9d436b9","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-03-19 16:21:33.000000000","message":"Patch Set 14:\n\nThanks for being patient with us on this. There\u0027s quite a bit of other work going on (getting feature/deep ready), and nearly everything else gets pushed aside when that happens.\n\nWhile not a review on that patch, I do love this feature. Thank you for working on it.","accounts_in_message":[],"_revision_number":14},{"id":"30460645a18549dad39ca07dd93724ecd1940e04","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-04-19 06:28:39.000000000","message":"Patch Set 14:\n\nReviewers, Please provide the suggestions (if any).","accounts_in_message":[],"_revision_number":14},{"id":"f1b0cbfdeb5e17ae53f34486eed9c0c1871daf2f","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-05-10 09:06:33.000000000","message":"Patch Set 14:\n\nReviewers, Please provide the suggestions (if any).","accounts_in_message":[],"_revision_number":14},{"id":"01b6e480bf2c795b8a2c2021f3d6c6505cf76897","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-07-06 05:14:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e2203ef607cda0f89c683a5cefd8b0218fe2d806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 06:09:42.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/15/check/build-openstack-sphinx-docs/8bbced3/ : FAILURE in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/15/check/openstack-tox-pep8/2c7abd9/ : FAILURE in 4m 24s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/15/check/legacy-swift-dsvm-functional/d438aa3/ : FAILURE in 19m 18s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/15/check/legacy-swift-dsvm-functional-identity-v3-only/90478fb/ : FAILURE in 16m 06s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/15/check/legacy-tempest-dsvm-neutron-full/8f1f072/ : FAILURE in 24m 46s\n- neutron-grenade http://logs.openstack.org/08/507808/15/check/neutron-grenade/871c42c/ : FAILURE in 49m 33s\n- swift-tox-py27 http://logs.openstack.org/08/507808/15/check/swift-tox-py27/a8e7f3a/ : FAILURE in 5m 04s\n- swift-tox-py35 http://logs.openstack.org/08/507808/15/check/swift-tox-py35/c8c09ac/ : FAILURE in 4m 19s\n- swift-tox-func http://logs.openstack.org/08/507808/15/check/swift-tox-func/6667e97/ : FAILURE in 4m 10s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/15/check/swift-tox-func-encryption/364d846/ : FAILURE in 3m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/15/check/swift-tox-func-domain-remap-staticweb/248675b/ : FAILURE in 4m 16s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/15/check/swift-tox-func-ec/990b1db/ : FAILURE in 3m 33s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/15/check/swift-tox-func-s3api/d622227/ : FAILURE in 3m 27s\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/15/check/swift-probetests-centos-7/9204826/ : FAILURE in 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5bec5e9114f6451da86876aa8fb9850afa74f680","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-07-09 05:41:47.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"b9bd731d2a6ece07086ee09f396da7293fa485af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-09 07:08:18.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/16/check/build-openstack-sphinx-docs/6729f3c/html/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/16/check/openstack-tox-pep8/35d966e/ : SUCCESS in 4m 23s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/16/check/legacy-swift-dsvm-functional/3d3b0b8/ : SUCCESS in 23m 45s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/16/check/legacy-swift-dsvm-functional-identity-v3-only/c803728/ : SUCCESS in 40m 25s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/16/check/legacy-tempest-dsvm-neutron-full/6fc2bf3/ : SUCCESS in 1h 25m 26s\n- neutron-grenade http://logs.openstack.org/08/507808/16/check/neutron-grenade/55e1063/ : SUCCESS in 53m 33s\n- swift-tox-py27 http://logs.openstack.org/08/507808/16/check/swift-tox-py27/dfc4890/ : SUCCESS in 9m 53s\n- swift-tox-py35 http://logs.openstack.org/08/507808/16/check/swift-tox-py35/9e54031/ : SUCCESS in 4m 14s\n- swift-tox-func http://logs.openstack.org/08/507808/16/check/swift-tox-func/8a09500/ : SUCCESS in 8m 08s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/16/check/swift-tox-func-encryption/3b778f4/ : SUCCESS in 9m 05s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/16/check/swift-tox-func-domain-remap-staticweb/abbf227/ : SUCCESS in 7m 53s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/16/check/swift-tox-func-ec/375dba5/ : SUCCESS in 8m 23s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/16/check/swift-tox-func-s3api/df31a17/ : SUCCESS in 8m 42s\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/16/check/swift-probetests-centos-7/b7e7a5c/ : FAILURE in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4e7c869d4c83b6d50e2d4595ecbe7247c0a4cd1b","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-07-31 06:32:04.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"4aa0b8d88934ad4d4aeca02921f7db7d450cb76b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-31 08:06:16.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/17/check/build-openstack-sphinx-docs/229cf0e/html/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/17/check/openstack-tox-pep8/466d773/ : SUCCESS in 5m 01s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/17/check/legacy-swift-dsvm-functional/5d8148b/ : SUCCESS in 23m 23s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/17/check/legacy-swift-dsvm-functional-identity-v3-only/b1392d3/ : SUCCESS in 28m 44s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/17/check/legacy-tempest-dsvm-neutron-full/962af82/ : SUCCESS in 1h 31m 27s\n- neutron-grenade http://logs.openstack.org/08/507808/17/check/neutron-grenade/9058a2f/ : SUCCESS in 1h 00m 24s\n- swift-tox-py27 http://logs.openstack.org/08/507808/17/check/swift-tox-py27/92367c4/ : SUCCESS in 10m 20s\n- swift-tox-py35 http://logs.openstack.org/08/507808/17/check/swift-tox-py35/4a18624/ : SUCCESS in 5m 05s\n- swift-tox-func http://logs.openstack.org/08/507808/17/check/swift-tox-func/9b19de6/ : SUCCESS in 8m 20s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/17/check/swift-tox-func-encryption/4589574/ : SUCCESS in 8m 39s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/17/check/swift-tox-func-domain-remap-staticweb/0393563/ : SUCCESS in 9m 05s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/17/check/swift-tox-func-ec/1feacfd/ : SUCCESS in 9m 03s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/17/check/swift-tox-func-s3api/155ab76/ : SUCCESS in 9m 22s\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/17/check/swift-probetests-centos-7/21eb953/ : RETRY_LIMIT in 3m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/507808/17/check/openstack-tox-lower-constraints/186daab/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":17},{"id":"4b1cadd5ac6028b03a3636813f879eebaa7bf0fc","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-08-02 06:55:05.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"bdd499056e5ec175d74e9ce917cbbad5f818099c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-02 08:28:13.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/18/check/build-openstack-sphinx-docs/8743d56/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/18/check/openstack-tox-pep8/c982525/ : SUCCESS in 6m 19s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/18/check/legacy-swift-dsvm-functional/60c962c/ : SUCCESS in 28m 49s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/18/check/legacy-swift-dsvm-functional-identity-v3-only/9c8fec5/ : SUCCESS in 24m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/18/check/legacy-tempest-dsvm-neutron-full/29efc15/ : SUCCESS in 1h 31m 44s\n- neutron-grenade http://logs.openstack.org/08/507808/18/check/neutron-grenade/afa2483/ : SUCCESS in 46m 52s\n- swift-tox-py27 http://logs.openstack.org/08/507808/18/check/swift-tox-py27/adcbcb5/ : SUCCESS in 9m 59s\n- swift-tox-py35 http://logs.openstack.org/08/507808/18/check/swift-tox-py35/39874e5/ : SUCCESS in 5m 07s\n- swift-tox-func http://logs.openstack.org/08/507808/18/check/swift-tox-func/f7151d4/ : SUCCESS in 8m 54s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/18/check/swift-tox-func-encryption/d4143ac/ : SUCCESS in 9m 30s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/18/check/swift-tox-func-domain-remap-staticweb/b71b805/ : SUCCESS in 8m 03s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/18/check/swift-tox-func-ec/1c0d782/ : SUCCESS in 8m 42s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/18/check/swift-tox-func-s3api/0018fcd/ : SUCCESS in 9m 58s\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/18/check/swift-probetests-centos-7/2d6b0ab/ : FAILURE in 28m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/507808/18/check/openstack-tox-lower-constraints/a43d6c7/ : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":18},{"id":"117c5b35c0486012c221d9e4c54c789e08860753","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-08-02 09:28:58.000000000","message":"Patch Set 18:\n\nrecheck.","accounts_in_message":[],"_revision_number":18},{"id":"8bb311663ca3fbcba1ec716a8426f285845a3f08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-02 11:01:15.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/18/check/build-openstack-sphinx-docs/d8353d9/html/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/18/check/openstack-tox-pep8/0af9951/ : SUCCESS in 4m 55s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/18/check/legacy-swift-dsvm-functional/21d1a8a/ : SUCCESS in 25m 52s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/18/check/legacy-swift-dsvm-functional-identity-v3-only/c2410bb/ : SUCCESS in 30m 36s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/18/check/legacy-tempest-dsvm-neutron-full/862cfc1/ : SUCCESS in 1h 29m 53s\n- neutron-grenade http://logs.openstack.org/08/507808/18/check/neutron-grenade/7af2fb9/ : SUCCESS in 56m 05s\n- swift-tox-py27 http://logs.openstack.org/08/507808/18/check/swift-tox-py27/1a94f60/ : SUCCESS in 9m 53s\n- swift-tox-py35 http://logs.openstack.org/08/507808/18/check/swift-tox-py35/481a07e/ : SUCCESS in 5m 16s\n- swift-tox-func http://logs.openstack.org/08/507808/18/check/swift-tox-func/b0c03de/ : SUCCESS in 8m 03s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/18/check/swift-tox-func-encryption/20c8e14/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/18/check/swift-tox-func-domain-remap-staticweb/1206294/ : SUCCESS in 9m 08s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/18/check/swift-tox-func-ec/c6e1109/ : SUCCESS in 9m 14s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/18/check/swift-tox-func-s3api/e9eee34/ : SUCCESS in 9m 19s\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/18/check/swift-probetests-centos-7/80fdbff/ : FAILURE in 25m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/507808/18/check/openstack-tox-lower-constraints/812fa8a/ : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":18},{"id":"8f3c01b1f8518fe43aa3bb29ede5419cfa2efa08","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-08-08 10:18:23.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"e72754c983c7bab4a5001437ac370097be4f07e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-08 12:17:08.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/19/check/build-openstack-sphinx-docs/98e1281/html/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/19/check/openstack-tox-pep8/b94611d/ : SUCCESS in 5m 29s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/19/check/legacy-swift-dsvm-functional/cbd6b22/ : SUCCESS in 29m 36s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/19/check/legacy-swift-dsvm-functional-identity-v3-only/380ce5f/ : SUCCESS in 27m 03s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/19/check/legacy-tempest-dsvm-neutron-full/d5e6e34/ : SUCCESS in 1h 33m 54s\n- neutron-grenade http://logs.openstack.org/08/507808/19/check/neutron-grenade/87cfba3/ : SUCCESS in 50m 13s\n- swift-tox-py27 http://logs.openstack.org/08/507808/19/check/swift-tox-py27/875b21c/ : SUCCESS in 10m 22s\n- swift-tox-py35 http://logs.openstack.org/08/507808/19/check/swift-tox-py35/5844a87/ : TIMED_OUT in 40m 58s\n- swift-tox-func http://logs.openstack.org/08/507808/19/check/swift-tox-func/74bdea1/ : SUCCESS in 9m 24s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/19/check/swift-tox-func-encryption/a36aebe/ : SUCCESS in 12m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/19/check/swift-tox-func-domain-remap-staticweb/04a596d/ : SUCCESS in 8m 29s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/19/check/swift-tox-func-ec/1d94dd2/ : SUCCESS in 12m 00s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/19/check/swift-tox-func-s3api/09cd6c4/ : SUCCESS in 9m 16s\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/19/check/swift-probetests-centos-7/9826a13/ : SUCCESS in 26m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/507808/19/check/openstack-tox-lower-constraints/9dfa2e5/ : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":19},{"id":"e0b65da3f3a6e72f177ab5ddd9ec770fb04e13c6","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-08-08 14:33:52.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"84b7fe22e4cd40cb44c134b26b390f00b2fc0db8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-08 18:43:13.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/507808/19/check/build-openstack-sphinx-docs/16d4cdf/html/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/19/check/openstack-tox-pep8/1f7e8b6/ : SUCCESS in 5m 46s\n- legacy-swift-dsvm-functional http://logs.openstack.org/08/507808/19/check/legacy-swift-dsvm-functional/029186b/ : SUCCESS in 34m 21s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/08/507808/19/check/legacy-swift-dsvm-functional-identity-v3-only/66239e9/ : SUCCESS in 31m 37s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/08/507808/19/check/legacy-tempest-dsvm-neutron-full/c242211/ : SUCCESS in 1h 58m 23s\n- neutron-grenade http://logs.openstack.org/08/507808/19/check/neutron-grenade/090d960/ : SUCCESS in 58m 21s\n- swift-tox-py27 http://logs.openstack.org/08/507808/19/check/swift-tox-py27/ff50ee1/ : SUCCESS in 10m 43s\n- swift-tox-py35 http://logs.openstack.org/08/507808/19/check/swift-tox-py35/03b5f81/ : SUCCESS in 5m 24s\n- swift-tox-func http://logs.openstack.org/08/507808/19/check/swift-tox-func/4888a1c/ : SUCCESS in 8m 17s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/19/check/swift-tox-func-encryption/ea778d2/ : SUCCESS in 8m 38s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/19/check/swift-tox-func-domain-remap-staticweb/b2e15ab/ : SUCCESS in 11m 56s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/19/check/swift-tox-func-ec/f107484/ : SUCCESS in 9m 51s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/19/check/swift-tox-func-s3api/08b9708/ : SUCCESS in 15m 14s\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/19/check/swift-probetests-centos-7/528b07d/ : SUCCESS in 29m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/507808/19/check/openstack-tox-lower-constraints/2f86a19/ : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":19},{"id":"3fbab63e851ef60e9ec6ba5753fc359c1a1c7cee","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-13 00:08:08.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"30f04772553189d87decc83efa0b7bb1c062cd2a","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-08-21 10:11:11.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"4c5d6df0741431da21287545f8adf3f22e64a6f0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-23 06:30:38.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"9d5e2bfcd7a0a80f26559d5750478a1b802b3d53","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-23 06:43:32.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9874aa3e4cc56c4d265ebf342773ba77234e1587","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-23 06:45:06.000000000","message":"Patch Set 19: Code-Review-1\n\nFor others interesting in reviewing this, the dicsussion is happening on patchset 13. We should move it forward, but it was eaiser to reply then to jump around patchsets :)","accounts_in_message":[],"_revision_number":19},{"id":"608d28838a08e2685c3599092f13dcd76c65e7c0","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-10-01 06:30:49.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"d41ac2f0ae073b42be73a7e84321d0088e19ce67","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-11-01 05:38:04.000000000","message":"Patch Set 19:\n\nReviewers, Please provide the suggestions.","accounts_in_message":[],"_revision_number":19},{"id":"e24b66595a685c6ec39e497277cccb97af69cfb6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-11-28 03:29:46.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"75fe439010778adcc28e7014fe6adbc820566dba","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-12-05 05:49:24.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"56a81da9b9f944cec8e853546371d0c47df21f89","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2018-12-19 05:29:05.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"0ff60c4ac5ab75a2d7856b24204b20451cbfd36e","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-01-10 06:01:07.000000000","message":"Patch Set 19:\n\nHi Matthew,\n\nWhen you are planning to provide a sample patch, which can be used as a reference to implement quorum check here?","accounts_in_message":[],"_revision_number":19},{"id":"b50e3d13b3047a4e75bfeca4c491c23689be86c6","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-03-18 06:06:20.000000000","message":"Patch Set 19:\n\nHi Matthew,\n\nWhen you are planning to provide a sample patch, which can be used as a reference to implement quorum check here?","accounts_in_message":[],"_revision_number":19},{"id":"2266f8a1f63cdc0c83a5c0f165d1d9c124319802","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-04-15 11:11:57.000000000","message":"Patch Set 19:\n\nReviewers, Please provide the suggestions.\n\nHi Matthew,\n\nWhen you are planning to provide a sample patch, which can be used as a reference to implement quorum check here?","accounts_in_message":[],"_revision_number":19},{"id":"097c1bfe664825beae77fe4066b0b812c23dbecc","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-06-26 05:29:55.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ae0b80c728c29442558379d43d9ef499eb8d855d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 13:31:34.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/08/507808/20/check/openstack-tox-docs/ed5041c/ : FAILURE in 5m 08s\n- swift-build-image http://logs.openstack.org/08/507808/20/check/swift-build-image/dd6653d/ : SUCCESS in 10m 01s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/08/507808/20/check/swift-tox-py27/e44e70a/ : FAILURE in 5m 39s\n- swift-tox-py36 http://logs.openstack.org/08/507808/20/check/swift-tox-py36/78acc63/ : FAILURE in 9m 45s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/08/507808/20/check/swift-tox-py37/d469e5a/ : FAILURE in 6m 12s\n- swift-tox-func http://logs.openstack.org/08/507808/20/check/swift-tox-func/ee15af1/ : FAILURE in 3m 40s\n- swift-tox-func-py37 http://logs.openstack.org/08/507808/20/check/swift-tox-func-py37/839c7ee/ : FAILURE in 4m 09s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/20/check/swift-tox-func-encryption/9dd14b5/ : FAILURE in 4m 07s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/20/check/swift-tox-func-domain-remap-staticweb/42d7a21/ : FAILURE in 3m 14s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/20/check/swift-tox-func-ec/a046479/ : FAILURE in 3m 49s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/20/check/swift-tox-func-s3api/86e83e4/ : FAILURE in 3m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/08/507808/20/check/swift-tox-func-s3api-ceph-s3tests-tempauth/96b367c/ : POST_FAILURE in 6m 04s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/20/check/swift-probetests-centos-7/90b68f1/ : FAILURE in 41m 02s\n- swift-dsvm-functional http://logs.openstack.org/08/507808/20/check/swift-dsvm-functional/d14a1d8/ : FAILURE in 19m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/507808/20/check/swift-dsvm-functional-ipv6/4da01cc/ : FAILURE in 19m 26s\n- swift-tox-lower-constraints http://logs.openstack.org/08/507808/20/check/swift-tox-lower-constraints/61c0974/ : FAILURE in 7m 02s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/20/check/openstack-tox-pep8/1a2188c/ : FAILURE in 5m 06s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/08/507808/20/check/swift-multinode-rolling-upgrade/dde097b/ : FAILURE in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6c5c3bb5ef363e618083259a6fc2f33e11a7a2e0","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-07-03 09:29:14.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"015dd602425cd16808c6ff185945888a2c6ec2c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 11:26:55.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/08/507808/21/check/openstack-tox-docs/6398897/ : FAILURE in 3m 52s\n- swift-build-image http://logs.openstack.org/08/507808/21/check/swift-build-image/efd2daf/ : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/08/507808/21/check/swift-tox-py27/9ee5ab2/ : FAILURE in 10m 29s\n- swift-tox-py36 http://logs.openstack.org/08/507808/21/check/swift-tox-py36/1055c4d/ : FAILURE in 11m 23s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/08/507808/21/check/swift-tox-py37/809da5e/ : FAILURE in 11m 28s\n- swift-tox-func http://logs.openstack.org/08/507808/21/check/swift-tox-func/fbfbb22/ : SUCCESS in 9m 25s\n- swift-tox-func-py37 http://logs.openstack.org/08/507808/21/check/swift-tox-func-py37/4782d64/ : SUCCESS in 9m 11s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/21/check/swift-tox-func-encryption/4864989/ : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/21/check/swift-tox-func-domain-remap-staticweb/73cac7a/ : SUCCESS in 8m 57s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/21/check/swift-tox-func-ec/4e83053/ : SUCCESS in 9m 39s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/21/check/swift-tox-func-s3api/32ee728/ : SUCCESS in 13m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/08/507808/21/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e16c91e/ : SUCCESS in 11m 48s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/21/check/swift-probetests-centos-7/e71e85f/ : SUCCESS in 33m 13s\n- swift-dsvm-functional http://logs.openstack.org/08/507808/21/check/swift-dsvm-functional/3199ea6/ : SUCCESS in 30m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/507808/21/check/swift-dsvm-functional-ipv6/21dfce0/ : SUCCESS in 32m 48s\n- swift-tox-lower-constraints http://logs.openstack.org/08/507808/21/check/swift-tox-lower-constraints/e859f85/ : FAILURE in 19m 48s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/21/check/openstack-tox-pep8/96abf4c/ : FAILURE in 5m 09s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/08/507808/21/check/swift-multinode-rolling-upgrade/15de1ab/ : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"09c045d152b1f576e132d32cf1a3bf5228d0b891","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-07-04 08:04:40.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"603d078883f962bffe99803da6047aae79ccb35f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:46:13.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/08/507808/22/check/openstack-tox-docs/354a99c/ : FAILURE in 4m 02s\n- swift-build-image http://logs.openstack.org/08/507808/22/check/swift-build-image/c3de8b6/ : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/08/507808/22/check/swift-tox-py27/149f0cf/ : SUCCESS in 10m 32s\n- swift-tox-py36 http://logs.openstack.org/08/507808/22/check/swift-tox-py36/dcf9c64/ : FAILURE in 11m 09s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/08/507808/22/check/swift-tox-py37/afc218e/ : FAILURE in 10m 40s\n- swift-tox-func http://logs.openstack.org/08/507808/22/check/swift-tox-func/ec0ab22/ : SUCCESS in 9m 23s\n- swift-tox-func-py37 http://logs.openstack.org/08/507808/22/check/swift-tox-func-py37/68acdb3/ : SUCCESS in 9m 45s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/22/check/swift-tox-func-encryption/b69c7a9/ : SUCCESS in 8m 43s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/22/check/swift-tox-func-domain-remap-staticweb/c543923/ : SUCCESS in 11m 17s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/22/check/swift-tox-func-ec/b0f2aa2/ : SUCCESS in 10m 42s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/22/check/swift-tox-func-s3api/9d1c9cd/ : SUCCESS in 11m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/08/507808/22/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f81c8ca/ : SUCCESS in 12m 11s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/22/check/swift-probetests-centos-7/386efe6/ : SUCCESS in 38m 41s\n- swift-dsvm-functional http://logs.openstack.org/08/507808/22/check/swift-dsvm-functional/49fe20d/ : SUCCESS in 27m 00s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/507808/22/check/swift-dsvm-functional-ipv6/5fa93c4/ : SUCCESS in 28m 14s\n- swift-tox-lower-constraints http://logs.openstack.org/08/507808/22/check/swift-tox-lower-constraints/bf4dd55/ : FAILURE in 12m 16s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/22/check/openstack-tox-pep8/2c0174b/ : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/08/507808/22/check/swift-multinode-rolling-upgrade/b1963a5/ : FAILURE in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"f9a05df6b47a74e7176e9ff4f5339ac08daf415d","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-07-09 06:52:46.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"ba2a4df3e54e45a47b40ac676de7b4a6afc96802","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 08:39:16.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/08/507808/23/check/openstack-tox-docs/79a9957/html/ : SUCCESS in 4m 37s\n- swift-build-image http://logs.openstack.org/08/507808/23/check/swift-build-image/4061e8a/ : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/08/507808/23/check/swift-tox-py27/dcf0ad2/ : SUCCESS in 11m 02s\n- swift-tox-py36 http://logs.openstack.org/08/507808/23/check/swift-tox-py36/d549c58/ : FAILURE in 10m 51s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/08/507808/23/check/swift-tox-py37/a46734e/ : FAILURE in 13m 59s\n- swift-tox-func http://logs.openstack.org/08/507808/23/check/swift-tox-func/d06db18/ : SUCCESS in 8m 49s\n- swift-tox-func-py37 http://logs.openstack.org/08/507808/23/check/swift-tox-func-py37/a7fd915/ : SUCCESS in 8m 26s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/23/check/swift-tox-func-encryption/ea34f32/ : SUCCESS in 9m 25s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/23/check/swift-tox-func-domain-remap-staticweb/df30ab5/ : SUCCESS in 9m 35s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/23/check/swift-tox-func-ec/487a32d/ : SUCCESS in 10m 59s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/23/check/swift-tox-func-s3api/9454cec/ : SUCCESS in 15m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/08/507808/23/check/swift-tox-func-s3api-ceph-s3tests-tempauth/604a589/ : SUCCESS in 14m 46s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/23/check/swift-probetests-centos-7/45eb9f7/ : SUCCESS in 34m 30s\n- swift-dsvm-functional http://logs.openstack.org/08/507808/23/check/swift-dsvm-functional/d9d5a59/ : SUCCESS in 28m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/507808/23/check/swift-dsvm-functional-ipv6/561c93f/ : SUCCESS in 27m 48s\n- swift-tox-lower-constraints http://logs.openstack.org/08/507808/23/check/swift-tox-lower-constraints/c3f3db2/ : SUCCESS in 13m 21s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/23/check/openstack-tox-pep8/7ad5c59/ : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/08/507808/23/check/swift-multinode-rolling-upgrade/415d150/ : FAILURE in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"33e6cf0d462b7c9f3e10e8d1caea856f547ddb8a","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-07-09 09:46:12.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"887688c312a3d90bbc9406593779dc1edfa52165","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 12:09:26.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/08/507808/23/check/openstack-tox-docs/407a925/html/ : SUCCESS in 4m 53s\n- swift-build-image http://logs.openstack.org/08/507808/23/check/swift-build-image/4bf8306/ : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/08/507808/23/check/swift-tox-py27/e057789/ : SUCCESS in 20m 27s\n- swift-tox-py36 http://logs.openstack.org/08/507808/23/check/swift-tox-py36/70cf4af/ : FAILURE in 12m 35s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/08/507808/23/check/swift-tox-py37/3e9e532/ : FAILURE in 11m 33s\n- swift-tox-func http://logs.openstack.org/08/507808/23/check/swift-tox-func/2d0f6ee/ : SUCCESS in 8m 58s\n- swift-tox-func-py37 http://logs.openstack.org/08/507808/23/check/swift-tox-func-py37/3749b3f/ : SUCCESS in 12m 58s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/23/check/swift-tox-func-encryption/2a4b318/ : SUCCESS in 12m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/23/check/swift-tox-func-domain-remap-staticweb/fa85a13/ : SUCCESS in 9m 01s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/23/check/swift-tox-func-ec/d478d66/ : SUCCESS in 9m 38s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/23/check/swift-tox-func-s3api/101d7de/ : SUCCESS in 13m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/08/507808/23/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9ce9d79/ : SUCCESS in 12m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/23/check/swift-probetests-centos-7/94a1e05/ : SUCCESS in 31m 21s\n- swift-dsvm-functional http://logs.openstack.org/08/507808/23/check/swift-dsvm-functional/5565e77/ : SUCCESS in 26m 40s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/507808/23/check/swift-dsvm-functional-ipv6/15bc9ee/ : SUCCESS in 29m 41s\n- swift-tox-lower-constraints http://logs.openstack.org/08/507808/23/check/swift-tox-lower-constraints/cf888f7/ : SUCCESS in 13m 18s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/23/check/openstack-tox-pep8/d98de1a/ : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/08/507808/23/check/swift-multinode-rolling-upgrade/535ae66/ : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"6e9c71264342f25d7247939d161ed1511943da9f","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-07-10 06:56:58.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"126cb60322e39bb514f260a8ebb609a4fe27cac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 07:36:15.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/507808/24/check/openstack-tox-docs/e445106/html/ : SUCCESS in 4m 19s\n- swift-build-image http://logs.openstack.org/08/507808/24/check/swift-build-image/d1e7ca2/ : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/08/507808/24/check/swift-tox-py27/e4a067f/ : SUCCESS in 11m 03s\n- swift-tox-py36 http://logs.openstack.org/08/507808/24/check/swift-tox-py36/0bbc478/ : SUCCESS in 10m 54s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/08/507808/24/check/swift-tox-py37/06dce1c/ : SUCCESS in 10m 32s\n- swift-tox-func http://logs.openstack.org/08/507808/24/check/swift-tox-func/607f1ca/ : SUCCESS in 9m 11s\n- swift-tox-func-py37 http://logs.openstack.org/08/507808/24/check/swift-tox-func-py37/5c3850e/ : SUCCESS in 11m 19s\n- swift-tox-func-encryption http://logs.openstack.org/08/507808/24/check/swift-tox-func-encryption/168ba2a/ : SUCCESS in 9m 08s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/507808/24/check/swift-tox-func-domain-remap-staticweb/f200843/ : SUCCESS in 8m 07s\n- swift-tox-func-ec http://logs.openstack.org/08/507808/24/check/swift-tox-func-ec/7d26cf6/ : SUCCESS in 9m 45s\n- swift-tox-func-s3api http://logs.openstack.org/08/507808/24/check/swift-tox-func-s3api/683460a/ : SUCCESS in 13m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/08/507808/24/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1c6fc63/ : SUCCESS in 14m 17s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/08/507808/24/check/swift-probetests-centos-7/419d378/ : SUCCESS in 32m 34s\n- swift-dsvm-functional http://logs.openstack.org/08/507808/24/check/swift-dsvm-functional/b35214b/ : SUCCESS in 31m 29s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/507808/24/check/swift-dsvm-functional-ipv6/dd4218b/ : SUCCESS in 33m 42s\n- swift-tox-lower-constraints http://logs.openstack.org/08/507808/24/check/swift-tox-lower-constraints/a2b4094/ : SUCCESS in 12m 08s\n- openstack-tox-pep8 http://logs.openstack.org/08/507808/24/check/openstack-tox-pep8/7f4def6/ : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/08/507808/24/check/swift-multinode-rolling-upgrade/ef269eb/ : FAILURE in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"77665e071591136d4541eecd0c005a295e1fceaa","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-07-22 08:00:03.000000000","message":"Patch Set 24:\n\nReviewers, Please provide the suggestions.","accounts_in_message":[],"_revision_number":24},{"id":"96ff98c892c737d512f2eb9b874dd5702b3ee103","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-08-13 05:03:19.000000000","message":"Patch Set 24:\n\nReviewers, Please provide the suggestions.","accounts_in_message":[],"_revision_number":24},{"id":"85a8d87918167ba84d89dcf09f1c85a2eb1a28e2","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-09-20 08:43:17.000000000","message":"Patch Set 24:\n\nReviewers, Please provide the suggestions.","accounts_in_message":[],"_revision_number":24},{"id":"18eed51d4fb340dee7fc3a1b6c08197d8b92bcca","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-10-15 12:19:28.000000000","message":"Patch Set 24:\n\nReviewers, Please provide the suggestions.","accounts_in_message":[],"_revision_number":24},{"id":"4b3105a63c96b08369460e1b871dc008f476d117","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-11-08 03:22:31.000000000","message":"Patch Set 24:\n\n(1 comment)\n\nif we get a create_account call added to InternalClient - it might be handy to have a little bin/recover-swift-account script","accounts_in_message":[],"_revision_number":24},{"id":"758e01e700a9e1526a452bc01092bbd1ed20e596","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-08 04:05:23.000000000","message":"Patch Set 24:\n\n(16 comments)\n\nI\u0027m working on getting status to replicate properly over in https://review.opendev.org/#/c/693299/ FWIW. Sorry, the more we poke at this, the more I realize that \"low-hanging-fruit\" really doesn\u0027t describe this bug!","accounts_in_message":[],"_revision_number":24},{"id":"6f8ee94887f581ca2f2082ce41b030f5f4f1a6a7","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2019-12-30 12:04:30.000000000","message":"Patch Set 24:\n\n(17 comments)\n\nReviewers, we have updated out comments inline to your valuable suggestions/comments. Kindly provide your valuable inputs.","accounts_in_message":[],"_revision_number":24},{"id":"8c47a1af7c668ad1882e9e59f41d47461fe28ddd","author":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"date":"2020-01-29 04:50:56.000000000","message":"Patch Set 24:\n\nReviewers, we have added our findings/observations and seeking for your valuable inputs. Kindly provide.","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"ac74dc38668b965bbab7d777ad67c525fb358bd1","revisions":{"c31f1f7a82f22d18c3278d5871a308d5fa80ad4e":{"kind":"REWORK","_number":1,"created":"2017-09-27 09:47:33.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/1"}}},"commit":{"parents":[{"commit":"ff23dcdf3aaef63b85ed3e762cecfc9432206fdc","subject":"Merge \"Stop reloading swift.common.utils in test_daemon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff23dcdf3aaef63b85ed3e762cecfc9432206fdc"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-27 09:46:15.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-27 09:46:15.000000000","tz":-420},"subject":"Implemented the fix to support the feature of undeleting the account which was marked for deletion for the reaper.","message":"Implemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c31f1f7a82f22d18c3278d5871a308d5fa80ad4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c31f1f7a82f22d18c3278d5871a308d5fa80ad4e"}]},"branch":"refs/heads/master"},"d0466992ae16c8478f86866dc7a6935df2dc6949":{"kind":"REWORK","_number":2,"created":"2017-09-28 08:53:52.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/2"}}},"commit":{"parents":[{"commit":"ff23dcdf3aaef63b85ed3e762cecfc9432206fdc","subject":"Merge \"Stop reloading swift.common.utils in test_daemon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff23dcdf3aaef63b85ed3e762cecfc9432206fdc"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-27 09:46:15.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-28 08:42:17.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated the \"Revive a deleted account\" procedure in procedures.rst file for\nend-users point of view.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0466992ae16c8478f86866dc7a6935df2dc6949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0466992ae16c8478f86866dc7a6935df2dc6949"}]},"branch":"refs/heads/master"},"7a56381b02de207e0efb9e37bcd9e22f9f50dcb9":{"kind":"REWORK","_number":3,"created":"2017-09-28 10:17:55.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/3"}}},"commit":{"parents":[{"commit":"ff23dcdf3aaef63b85ed3e762cecfc9432206fdc","subject":"Merge \"Stop reloading swift.common.utils in test_daemon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff23dcdf3aaef63b85ed3e762cecfc9432206fdc"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-27 09:46:15.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-28 10:15:57.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated the \"Revive a deleted account\" procedure in procedures.rst file for\nend-users point of view.\n\nUpdated the code indentation pep8 issues.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a56381b02de207e0efb9e37bcd9e22f9f50dcb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a56381b02de207e0efb9e37bcd9e22f9f50dcb9"}]},"branch":"refs/heads/master"},"ad7beb636ee0fa58aa3e02c902a83a0a913cb8d3":{"kind":"REWORK","_number":4,"created":"2017-10-03 09:13:57.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/4"}}},"commit":{"parents":[{"commit":"ff23dcdf3aaef63b85ed3e762cecfc9432206fdc","subject":"Merge \"Stop reloading swift.common.utils in test_daemon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff23dcdf3aaef63b85ed3e762cecfc9432206fdc"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-27 09:46:15.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-03 09:12:20.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated the \"Revive a deleted account\" procedure in procedures.rst file for\nend-users point of view.\n\nUpdated the Jenkins issues.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad7beb636ee0fa58aa3e02c902a83a0a913cb8d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad7beb636ee0fa58aa3e02c902a83a0a913cb8d3"}]},"branch":"refs/heads/master"},"7bb90382ce046982072c22a21b51195cb7ef921b":{"kind":"REWORK","_number":5,"created":"2017-10-11 10:43:17.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/5"}}},"commit":{"parents":[{"commit":"ff23dcdf3aaef63b85ed3e762cecfc9432206fdc","subject":"Merge \"Stop reloading swift.common.utils in test_daemon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff23dcdf3aaef63b85ed3e762cecfc9432206fdc"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-27 09:46:15.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-11 10:38:19.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated for the reviews received on Patch Set 4.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb90382ce046982072c22a21b51195cb7ef921b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb90382ce046982072c22a21b51195cb7ef921b"}]},"branch":"refs/heads/master"},"a4618131f4cf4b8b07b3e869fef3b8772dfe96d4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2017-10-16 08:16:47.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/6"}}},"commit":{"parents":[{"commit":"98e7014aafbd52bd5e388314f3c2c1e1d2375806","subject":"Merge \"Remove some leftover threadpool cruft.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98e7014aafbd52bd5e388314f3c2c1e1d2375806"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-09-27 09:46:15.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-16 08:16:47.000000000","tz":0},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated for the reviews received on Patch Set 4.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4618131f4cf4b8b07b3e869fef3b8772dfe96d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4618131f4cf4b8b07b3e869fef3b8772dfe96d4"}]},"branch":"refs/heads/master"},"bcc01e7165b6c53f0141734a8b537900312d54ff":{"kind":"REWORK","_number":7,"created":"2017-10-17 09:42:37.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/7"}}},"commit":{"parents":[{"commit":"d52987bf6cd2d0aa996611d228b2031252cbd9ae","subject":"Merge \"Respond 400 Bad Request when Accept headers fail to parse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d52987bf6cd2d0aa996611d228b2031252cbd9ae"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-17 09:39:42.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-17 09:39:42.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated for the reviews received on Patch Set 5.\nUpdated the conf file and documentation for delay_reaping.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcc01e7165b6c53f0141734a8b537900312d54ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcc01e7165b6c53f0141734a8b537900312d54ff"}]},"branch":"refs/heads/master"},"835a534c23ef970bfb6b11f93d981fb28188c226":{"kind":"REWORK","_number":8,"created":"2017-10-23 10:36:04.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/8"}}},"commit":{"parents":[{"commit":"1da1bfb3816e4570523b7b14b72dbc09639a4cb3","subject":"Merge \"Stop logging tracebacks on bad xLOs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1da1bfb3816e4570523b7b14b72dbc09639a4cb3"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-23 10:34:03.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-23 10:34:03.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated for the reviews received till Patch Set 7.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/835a534c23ef970bfb6b11f93d981fb28188c226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/835a534c23ef970bfb6b11f93d981fb28188c226"}]},"branch":"refs/heads/master"},"f3c0717aef96220327dd140c266c2d9265f83aa8":{"kind":"REWORK","_number":9,"created":"2017-10-24 16:06:54.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/9"}}},"commit":{"parents":[{"commit":"1da1bfb3816e4570523b7b14b72dbc09639a4cb3","subject":"Merge \"Stop logging tracebacks on bad xLOs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1da1bfb3816e4570523b7b14b72dbc09639a4cb3"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-23 10:34:03.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-24 16:04:48.000000000","tz":0},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated for the reviews received till Patch Set 7.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3c0717aef96220327dd140c266c2d9265f83aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3c0717aef96220327dd140c266c2d9265f83aa8"}]},"branch":"refs/heads/master"},"cb2e02505394ca2cbff143144856043e3c6b0fd3":{"kind":"REWORK","_number":10,"created":"2017-10-25 05:17:07.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/10"}}},"commit":{"parents":[{"commit":"1da1bfb3816e4570523b7b14b72dbc09639a4cb3","subject":"Merge \"Stop logging tracebacks on bad xLOs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1da1bfb3816e4570523b7b14b72dbc09639a4cb3"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-23 10:34:03.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-25 05:16:49.000000000","tz":0},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated for the reviews received till Patch Set 7.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb2e02505394ca2cbff143144856043e3c6b0fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb2e02505394ca2cbff143144856043e3c6b0fd3"}]},"branch":"refs/heads/master"},"0090e8dadb9b32e1c0d750ae629cb4004b2f6eb3":{"kind":"REWORK","_number":11,"created":"2017-10-25 07:58:45.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/11"}}},"commit":{"parents":[{"commit":"1da1bfb3816e4570523b7b14b72dbc09639a4cb3","subject":"Merge \"Stop logging tracebacks on bad xLOs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1da1bfb3816e4570523b7b14b72dbc09639a4cb3"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-23 10:34:03.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-10-25 07:58:15.000000000","tz":0},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nUpdated for the reviews received till Patch Set 7.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0090e8dadb9b32e1c0d750ae629cb4004b2f6eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0090e8dadb9b32e1c0d750ae629cb4004b2f6eb3"}]},"branch":"refs/heads/master"},"b634f35b12f093656ae62c51b4c3ce1442840159":{"kind":"REWORK","_number":12,"created":"2017-11-03 10:39:42.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/12"}}},"commit":{"parents":[{"commit":"70a47b318760387df5df105d6bc028aa7323e7fd","subject":"Merge \"Return 404 on a GET if tombstone is newer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70a47b318760387df5df105d6bc028aa7323e7fd"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-11-03 10:37:56.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-11-03 10:37:56.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nTo undelete the account, the following cURL command would\nbe invoked to undelete the account.\ncurl -v -X PUT -H \u0027X-account-meta-undelete: true\u0027 -H \u0027X-Auth-Token:\n\u003cadmin_authentication_token\u003e\u0027 http://\u003cip_address\u003e:8080/v1/\u003cname_of_deleted_account\u003e\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b634f35b12f093656ae62c51b4c3ce1442840159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b634f35b12f093656ae62c51b4c3ce1442840159"}]},"branch":"refs/heads/master"},"26dcececc82dd1a7c6511ed9d1da602a806da8b9":{"kind":"REWORK","_number":13,"created":"2017-11-24 08:40:52.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/13"}}},"commit":{"parents":[{"commit":"3135878d2fe9909f49fcadeeb9cc6c6933d06127","subject":"Fix small formatting issues in middleware docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3135878d2fe9909f49fcadeeb9cc6c6933d06127"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-11-24 08:38:02.000000000","tz":-480},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2017-11-24 08:38:02.000000000","tz":-480},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26dcececc82dd1a7c6511ed9d1da602a806da8b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26dcececc82dd1a7c6511ed9d1da602a806da8b9"}]},"branch":"refs/heads/master"},"f4e3cdfe021623517542db9ceb2294ceb2a5c247":{"kind":"REWORK","_number":14,"created":"2018-02-20 08:40:07.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/14"}}},"commit":{"parents":[{"commit":"ee282c1166e0a9e717364d08fd471ca262aa58f4","subject":"Merge \"Fix suffix-byte-range responses for zero-byte replicated objects.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee282c1166e0a9e717364d08fd471ca262aa58f4"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-02-20 08:38:04.000000000","tz":-480},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-02-20 08:38:04.000000000","tz":-480},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4e3cdfe021623517542db9ceb2294ceb2a5c247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4e3cdfe021623517542db9ceb2294ceb2a5c247"}]},"branch":"refs/heads/master"},"0bd36d6c2d655de705658846063a93d15d46bc77":{"kind":"REWORK","_number":15,"created":"2018-07-06 05:14:27.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/15"}}},"commit":{"parents":[{"commit":"ec7d60ebd2fc126d28b1cd575aae8612d93fe5c7","subject":"Merge \"Add Alpine Linux suuprt to bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec7d60ebd2fc126d28b1cd575aae8612d93fe5c7"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-07-06 05:12:19.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-07-06 05:12:19.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bd36d6c2d655de705658846063a93d15d46bc77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bd36d6c2d655de705658846063a93d15d46bc77"}]},"branch":"refs/heads/master"},"62ed723a48a1e1a4d8d093ff0df449d24d1b56f9":{"kind":"REWORK","_number":16,"created":"2018-07-09 05:41:47.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/16"}}},"commit":{"parents":[{"commit":"4409da697f0a563a5e1fde923fadb98322f2ea99","subject":"Update doc building instructions in Development Guidelines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4409da697f0a563a5e1fde923fadb98322f2ea99"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-07-09 05:40:14.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-07-09 05:40:14.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62ed723a48a1e1a4d8d093ff0df449d24d1b56f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62ed723a48a1e1a4d8d093ff0df449d24d1b56f9"}]},"branch":"refs/heads/master"},"82505316bba1483f934cb5571c1bb9ad7abbc979":{"kind":"REWORK","_number":17,"created":"2018-07-31 06:32:04.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/17"}}},"commit":{"parents":[{"commit":"d97673cf5431daf276a2395c449be48951cfd6f0","subject":"Merge \"Define keymaster log routes on the class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97673cf5431daf276a2395c449be48951cfd6f0"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-07-31 06:30:00.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-07-31 06:30:00.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82505316bba1483f934cb5571c1bb9ad7abbc979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82505316bba1483f934cb5571c1bb9ad7abbc979"}]},"branch":"refs/heads/master"},"f70a6b2157a97c77bcb011e0907177774017552e":{"kind":"REWORK","_number":18,"created":"2018-08-02 06:55:05.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/18"}}},"commit":{"parents":[{"commit":"86d78f6667392b2ee0ca017da8f54ff153aca491","subject":"imported some docs from the old user-guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86d78f6667392b2ee0ca017da8f54ff153aca491"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-08-02 06:51:59.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-08-02 06:51:59.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f70a6b2157a97c77bcb011e0907177774017552e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f70a6b2157a97c77bcb011e0907177774017552e"}]},"branch":"refs/heads/master"},"541b9d4dd80a5beb7be28daca30f74fb52a629c1":{"kind":"REWORK","_number":19,"created":"2018-08-08 10:18:23.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/19"}}},"commit":{"parents":[{"commit":"7f7482c096d6c587c64a75e207842ab4dee6e8dd","subject":"Merge \"Move keymaster_config_path parsing out of _get_root_secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f7482c096d6c587c64a75e207842ab4dee6e8dd"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-08-08 10:15:45.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2018-08-08 10:15:45.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/541b9d4dd80a5beb7be28daca30f74fb52a629c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/541b9d4dd80a5beb7be28daca30f74fb52a629c1"}]},"branch":"refs/heads/master"},"22e0ffdd26b79d44c401862db05b42a227119db2":{"kind":"REWORK","_number":20,"created":"2019-06-26 05:29:55.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/20"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-06-26 05:28:37.000000000","tz":-420},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-06-26 05:28:37.000000000","tz":-420},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22e0ffdd26b79d44c401862db05b42a227119db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22e0ffdd26b79d44c401862db05b42a227119db2"}]},"branch":"refs/heads/master"},"a974e134d2b1c51e30d9eda1ec33c3066a8b6b2d":{"kind":"REWORK","_number":21,"created":"2019-07-03 09:29:14.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/21"}}},"commit":{"parents":[{"commit":"dc84c69c652d899bf6b8177fa352cf980ae5ff8a","subject":"Merge \"Add information about secret_cache_duration to sample config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc84c69c652d899bf6b8177fa352cf980ae5ff8a"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-03 09:27:44.000000000","tz":330},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-03 09:27:44.000000000","tz":330},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a974e134d2b1c51e30d9eda1ec33c3066a8b6b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a974e134d2b1c51e30d9eda1ec33c3066a8b6b2d"}]},"branch":"refs/heads/master"},"b613c75164d7ba11c2fb0240e64a1f07926057e3":{"kind":"REWORK","_number":22,"created":"2019-07-04 08:04:40.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/22"}}},"commit":{"parents":[{"commit":"345f577ff191ee01c3cb4626805338028815c2b4","subject":"s3token: fix conf option name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f577ff191ee01c3cb4626805338028815c2b4"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-04 08:03:45.000000000","tz":330},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-04 08:03:45.000000000","tz":330},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b613c75164d7ba11c2fb0240e64a1f07926057e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b613c75164d7ba11c2fb0240e64a1f07926057e3"}]},"branch":"refs/heads/master"},"10bc63ceef01018cca52443a87ac28fe42e5d10f":{"kind":"REWORK","_number":23,"created":"2019-07-09 06:52:46.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/23"}}},"commit":{"parents":[{"commit":"392c967af53a7dd95c725fa8a765b3882a87f471","subject":"Merge \"container-replicator: Add a timeout for get_shard_ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/392c967af53a7dd95c725fa8a765b3882a87f471"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-09 06:50:40.000000000","tz":330},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-09 06:50:40.000000000","tz":330},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10bc63ceef01018cca52443a87ac28fe42e5d10f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10bc63ceef01018cca52443a87ac28fe42e5d10f"}]},"branch":"refs/heads/master"},"ac74dc38668b965bbab7d777ad67c525fb358bd1":{"kind":"REWORK","_number":24,"created":"2019-07-10 06:56:58.000000000","uploader":{"_account_id":26690,"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","username":"hcltech-ssw"},"ref":"refs/changes/08/507808/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/507808/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/507808/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/507808/24"}}},"commit":{"parents":[{"commit":"4ae412e13b1b16e7a043235427229ead75186ce7","subject":"Merge \"Increase some middleware test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ae412e13b1b16e7a043235427229ead75186ce7"}]}],"author":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-10 06:51:18.000000000","tz":330},"committer":{"name":"HCLTech-SSW","email":"hcl_ss_oss@hcl.com","date":"2019-07-10 06:51:18.000000000","tz":330},"subject":"Add ability to undelete an account.","message":"Add ability to undelete an account.\n\nImplemented the fix to support the feature of undeleting the\naccount which was marked for deletion for the reaper.\n\nChange-Id: Ib537065a88ba3b7d97235d74cf8bcae465ccf055\nCloses-Bug: #1521351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac74dc38668b965bbab7d777ad67c525fb358bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac74dc38668b965bbab7d777ad67c525fb358bd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
