)]}'
{"id":"openstack%2Fswift~517389","triplet_id":"openstack%2Fswift~master~I4bb584dadc8d36e2533c5f7b555ce359fde79276","project":"openstack/swift","branch":"master","topic":"new-expirer","hashtags":[],"change_id":"I4bb584dadc8d36e2533c5f7b555ce359fde79276","subject":"Add object-expirer new mode to execute tasks from general task queue","status":"NEW","created":"2017-11-02 18:11:59.000000000","updated":"2021-06-17 05:51:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":117,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c4dcee6e13b8ca8240fca409e02864b8dd43d42c","_number":517389,"virtual_id_number":517389,"owner":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-17 05:51:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-22 08:57:59.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-15 23:24:57.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2018-09-12 19:48:22.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2018-09-13 16:38:36.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2018-09-13 23:08:55.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2018-09-14 04:48:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-03 22:58:05.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-05-04 15:36:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6befe2e75b2909d0829e979b84e14f47d2b1ac00","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-11-02 18:11:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08c02b1086e6c44e9722f91cbdd56f58fe628334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-02 20:03:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/1/check/openstack-tox-pep8/373c4b1/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/89/517389/1/check/openstack-tox-py27/28c0a3c/ : FAILURE in 7m 01s\n- openstack-tox-py35 http://logs.openstack.org/89/517389/1/check/openstack-tox-py35/29e4ba2/ : SUCCESS in 2m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/1/check/build-openstack-sphinx-docs/502c703/html/ : SUCCESS in 4m 50s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/89/517389/1/check/legacy-swift-tox-xfs-tmp-py27/577e1a6/ : FAILURE in 7m 10s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/89/517389/1/check/legacy-swift-tox-xfs-tmp-py35/a60328f/ : SUCCESS in 2m 52s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/89/517389/1/check/legacy-swift-tox-xfs-tmp-func/269a3d0/ : FAILURE in 11m 38s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/89/517389/1/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/df41d8f/ : FAILURE in 11m 11s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/89/517389/1/check/legacy-swift-tox-xfs-tmp-func-encryption/3d17f01/ : FAILURE in 9m 12s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/89/517389/1/check/legacy-swift-tox-xfs-tmp-func-ec/56f43eb/ : FAILURE in 10m 13s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/1/check/legacy-swift-dsvm-functional/455569e/ : FAILURE in 29m 30s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/1/check/legacy-swift-dsvm-functional-identity-v3-only/8cd679c/ : FAILURE in 29m 04s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/1/check/legacy-tempest-dsvm-neutron-full/9e93a49/ : FAILURE in 1h 13m 27s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/89/517389/1/check/legacy-grenade-dsvm-neutron/8b390a1/ : SUCCESS in 1h 04m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0cca645285dae21ab239a6959c3af3c65cd073fe","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2017-11-06 03:01:53.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nDid a quick/light review. I like the way this is going. I\u0027d love to see a package swift.common.tasks that contains the generic methods (eg: get_task_container, ...). Few thoughts below.","accounts_in_message":[],"_revision_number":1},{"id":"1cac2725145b5919c50667e09c98f424716231e1","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-12-20 12:50:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85e16cdfe888844a8df183bcbb45a889c3e8aa41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-20 14:29:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/2/check/openstack-tox-pep8/f2cfa96/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/89/517389/2/check/openstack-tox-py27/b98cde5/ : FAILURE in 4m 39s\n- openstack-tox-py35 http://logs.openstack.org/89/517389/2/check/openstack-tox-py35/c56212c/ : SUCCESS in 2m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/2/check/build-openstack-sphinx-docs/7d0fc56/html/ : SUCCESS in 4m 22s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/89/517389/2/check/legacy-swift-tox-xfs-tmp-py27/132e047/ : FAILURE in 6m 28s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/89/517389/2/check/legacy-swift-tox-xfs-tmp-py35/68ceb4f/ : SUCCESS in 2m 22s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/89/517389/2/check/legacy-swift-tox-xfs-tmp-func/2431e5f/ : SUCCESS in 5m 47s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/89/517389/2/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/03a9018/ : SUCCESS in 9m 35s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/89/517389/2/check/legacy-swift-tox-xfs-tmp-func-encryption/0204611/ : SUCCESS in 6m 21s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/89/517389/2/check/legacy-swift-tox-xfs-tmp-func-ec/dce0818/ : SUCCESS in 8m 05s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/2/check/legacy-swift-dsvm-functional/19e5388/ : SUCCESS in 20m 41s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/2/check/legacy-swift-dsvm-functional-identity-v3-only/8a0d5b3/ : SUCCESS in 23m 31s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/2/check/legacy-tempest-dsvm-neutron-full/cbd29e0/ : SUCCESS in 1h 24m 23s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/89/517389/2/check/legacy-grenade-dsvm-neutron/889dec5/ : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":2},{"id":"462b0ab5af2dacb0b8d9c1b77bb1cd8110998e38","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-12-20 20:46:18.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f0dbf537cb1e9b64d04f9c8f6a9aa048f1e18485","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-12-20 20:47:04.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"972d7ce134329e0c71aec5507464538410f32088","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-12-21 11:29:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b65784b39a06bfb10193bc4b6329cef576e481a6","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-12-21 11:30:18.000000000","message":"Patch Set 3: Workflow-1\n\nUpdated docs.\nI will add tests next.","accounts_in_message":[],"_revision_number":3},{"id":"2e0893672ace34f91c44138bcf08ebcf857ad07b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-21 12:48:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/3/check/openstack-tox-pep8/d39a936/ : FAILURE in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/89/517389/3/check/openstack-tox-py27/e3d26f3/ : FAILURE in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/89/517389/3/check/openstack-tox-py35/c7e0dfb/ : SUCCESS in 3m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/3/check/build-openstack-sphinx-docs/58468e8/html/ : SUCCESS in 4m 09s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/89/517389/3/check/legacy-swift-tox-xfs-tmp-py27/16e66b6/ : FAILURE in 10m 18s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/89/517389/3/check/legacy-swift-tox-xfs-tmp-py35/25acf32/ : SUCCESS in 2m 29s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/89/517389/3/check/legacy-swift-tox-xfs-tmp-func/bcc355f/ : SUCCESS in 8m 48s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/89/517389/3/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/465894f/ : SUCCESS in 6m 29s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/89/517389/3/check/legacy-swift-tox-xfs-tmp-func-encryption/71cd098/ : SUCCESS in 6m 11s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/89/517389/3/check/legacy-swift-tox-xfs-tmp-func-ec/5416d40/ : SUCCESS in 7m 04s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/3/check/legacy-swift-dsvm-functional/595ad85/ : SUCCESS in 24m 31s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/3/check/legacy-swift-dsvm-functional-identity-v3-only/18cc006/ : SUCCESS in 30m 04s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/3/check/legacy-tempest-dsvm-neutron-full/848ca7b/ : SUCCESS in 1h 16m 22s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/89/517389/3/check/legacy-grenade-dsvm-neutron/7542c95/ : SUCCESS in 51m 50s\n- swift-tox-py27 http://logs.openstack.org/89/517389/3/check/swift-tox-py27/47e2402/ : FAILURE in 6m 22s\n- swift-tox-py35 http://logs.openstack.org/89/517389/3/check/swift-tox-py35/a669561/ : SUCCESS in 3m 17s\n- swift-tox-func http://logs.openstack.org/89/517389/3/check/swift-tox-func/fe4fc66/ : SUCCESS in 5m 36s\n- swift-tox-func-post-as-copy http://logs.openstack.org/89/517389/3/check/swift-tox-func-post-as-copy/14a9ab6/ : SUCCESS in 6m 15s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/3/check/swift-tox-func-encryption/31de8bf/ : SUCCESS in 6m 34s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/3/check/swift-tox-func-ec/5b02c11/ : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":3},{"id":"ae4a1b7eb42fff29d3f0780158b2472b14abab4b","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-21 13:15:06.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/517389/7729/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/517389/9708/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/517389/8123/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"abb219d3221714bfd8acb58b7abb1f28bdd5c90d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2017-12-22 03:33:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nGreat work on this Kazuhiro! It looks really cool. I have left some comments inline.\n\nSo to solve the legacy problem, you can turn on the boolean value. Is it expected then that only 1 of you object servers will then deal with legacy expiration? What if its a big cluster that used to use the processes and process (running multiple expirers) with heaps of objects still yet to be expired in the future?\n\nIn that case do we need to have the legecy boolean and in that case still also support the process and processes so they can still work as it used too.. only all new expiring objects will use the new task queue?","accounts_in_message":[],"_revision_number":3},{"id":"b10933cab3b1ec069d46149db3513f91d800a9ab","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-12-22 07:26:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"661a602c9ccc3bbae8ff69f1ca35f08321b4c21b","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2017-12-22 07:28:52.000000000","message":"Patch Set 4: Workflow-1\n\n(3 comments)\n\nMatthew, thank you for your comments!\n\nAbout legacy problem:\nQ: Is it expected then that only 1 of you object servers will then deal with legacy expiration?\nA: Yes.\n\nQ: What if its a big cluster that used to use the processes and process (running multiple expirers) with heaps of objects still yet to be expired in the future?\nA: I think nothing bad will happens.\n   In my understanding, processes and process are required to gain throughput of task execution. Because if task execution throughput is less than task registration throughput, expired-but-not-yet-deleted objects will continue increasing.\n   But in general task queue feature, task registration throughput for legacy task queue is ZERO. So I think only 1 object-expirer for legacy task queue is enough.\n   Of course, legacy task execution will delay as compared to multi-process execution. But it will completed eventually.\n\nHow do you think?","accounts_in_message":[],"_revision_number":4},{"id":"b5ff3521b585f3d173d77abf32342138b92f6b14","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-22 08:57:59.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/517389/9714/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/517389/8129/probetests/ : FAILURE\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/517389/7735/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"039d7e111f0fc3bf1f8750b9b5142a3250e98e02","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-10 11:15:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f70cb61b88f3f982ae4819f0f8549b114137c743","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-10 11:46:15.000000000","message":"Patch Set 5:\n\nAdded unit tests.\nI think no update of functional tests are needed, because this patch doesn\u0027t change expirer\u0027s frontend behaviors.\n\nNow, all unit tests and all functional tests succeed in my environment.\n\nI will update probe tests in this week.","accounts_in_message":[],"_revision_number":5},{"id":"381b546da355e25c234cc4e3a94364a0adb8954e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-10 12:35:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/5/check/build-openstack-sphinx-docs/da17b00/html/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/5/check/openstack-tox-pep8/9fd8f0c/ : SUCCESS in 3m 33s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/5/check/legacy-swift-dsvm-functional/c50e8ee/ : SUCCESS in 23m 06s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/5/check/legacy-swift-dsvm-functional-identity-v3-only/1eca47b/ : SUCCESS in 23m 06s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/5/check/legacy-tempest-dsvm-neutron-full/5b9b52a/ : SUCCESS in 1h 17m 04s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/89/517389/5/check/legacy-grenade-dsvm-neutron/6e8562a/ : SUCCESS in 46m 08s\n- swift-tox-py27 http://logs.openstack.org/89/517389/5/check/swift-tox-py27/7bf33aa/ : FAILURE in 8m 19s\n- swift-tox-py35 http://logs.openstack.org/89/517389/5/check/swift-tox-py35/539c883/ : SUCCESS in 2m 34s\n- swift-tox-func http://logs.openstack.org/89/517389/5/check/swift-tox-func/959b0cc/ : SUCCESS in 7m 25s\n- swift-tox-func-post-as-copy http://logs.openstack.org/89/517389/5/check/swift-tox-func-post-as-copy/271b3d3/ : SUCCESS in 7m 23s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/5/check/swift-tox-func-encryption/3037dd5/ : SUCCESS in 7m 16s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/5/check/swift-tox-func-ec/dfc0fdd/ : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":5},{"id":"4ec093b03b76abdb41bb7e9191e39bb6d3f83904","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-10 20:48:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0008aa3174f05428436422ce7458bc76a8af925a","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-10 20:53:47.000000000","message":"Patch Set 5: Workflow-1\n\nSorry, I forgot to check some of unit tests.\nI need to fix the unit tests too.","accounts_in_message":[],"_revision_number":5},{"id":"9e0cf207ef27975d079b800654da396e9fd67704","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-11 02:09:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/5/check/build-openstack-sphinx-docs/bb02208/html/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/5/check/openstack-tox-pep8/651f0e6/ : SUCCESS in 4m 21s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/5/check/legacy-swift-dsvm-functional/6ef4a97/ : SUCCESS in 22m 52s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/5/check/legacy-swift-dsvm-functional-identity-v3-only/6c73acc/ : SUCCESS in 21m 04s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/5/check/legacy-tempest-dsvm-neutron-full/dd862f7/ : SUCCESS in 1h 22m 33s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/89/517389/5/check/legacy-grenade-dsvm-neutron/8811eca/ : SUCCESS in 44m 54s\n- swift-tox-py27 http://logs.openstack.org/89/517389/5/check/swift-tox-py27/5300694/ : FAILURE in 8m 34s\n- swift-tox-py35 http://logs.openstack.org/89/517389/5/check/swift-tox-py35/7cc442f/ : SUCCESS in 2m 55s\n- swift-tox-func http://logs.openstack.org/89/517389/5/check/swift-tox-func/5585dc7/ : SUCCESS in 6m 27s\n- swift-tox-func-post-as-copy http://logs.openstack.org/89/517389/5/check/swift-tox-func-post-as-copy/9eb44db/ : SUCCESS in 7m 12s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/5/check/swift-tox-func-encryption/84c02cc/ : SUCCESS in 7m 36s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/5/check/swift-tox-func-ec/a2fbb9b/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":5},{"id":"1b015f5302fd44fd70c665f6cc5f1e52c40205b0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-01-11 04:30:08.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)\n\nThis is looking really cool Kazuhiro. Just some inline comments and a few fixes. I still need to look at the tests.","accounts_in_message":[],"_revision_number":5},{"id":"f48005060768183ec5d8a07670858fcaa83fdfae","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-15 09:14:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b47f91f556b7865913557388353bf03315239935","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-15 09:16:39.000000000","message":"Patch Set 6:\n\n(5 comments)\n\nUpdated unit tests and reflect Matthew\u0027s comments.\n\nI need to update probe tests next...","accounts_in_message":[],"_revision_number":6},{"id":"c1981451f82f0009489fdeebc1d5f384b75f36f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-15 10:49:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/6/check/build-openstack-sphinx-docs/24b4129/html/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/6/check/openstack-tox-pep8/d0a9f7f/ : SUCCESS in 4m 01s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/6/check/legacy-swift-dsvm-functional/5af033f/ : SUCCESS in 23m 50s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/6/check/legacy-swift-dsvm-functional-identity-v3-only/bd29406/ : SUCCESS in 22m 21s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/6/check/legacy-tempest-dsvm-neutron-full/2d5fd87/ : SUCCESS in 1h 17m 18s\n- neutron-grenade http://logs.openstack.org/89/517389/6/check/neutron-grenade/1584744/ : SUCCESS in 52m 06s\n- swift-tox-py27 http://logs.openstack.org/89/517389/6/check/swift-tox-py27/2d558eb/ : SUCCESS in 6m 33s\n- swift-tox-py35 http://logs.openstack.org/89/517389/6/check/swift-tox-py35/d3ec03a/ : SUCCESS in 3m 30s\n- swift-tox-func http://logs.openstack.org/89/517389/6/check/swift-tox-func/30dca35/ : SUCCESS in 6m 57s\n- swift-tox-func-post-as-copy http://logs.openstack.org/89/517389/6/check/swift-tox-func-post-as-copy/9d125a0/ : SUCCESS in 7m 03s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/6/check/swift-tox-func-encryption/1fb3a1a/ : SUCCESS in 7m 07s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/6/check/swift-tox-func-ec/c186c44/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":6},{"id":"f9f0b639edcaaacb3ea9b87d4c9c24de51ce3dfe","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-01-15 23:24:57.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)\n\nAwesome nice work Kazuhiro.\n\nWe need to add some documentation back about the legecy process and processes. If you want I can push a follow up patch you can merge in if it makes your life easer.\n\nNext I want to run in in my SAIO and test it as much as I can.","accounts_in_message":[],"_revision_number":6},{"id":"753cb008ec02ef997fc10a9b15b0c15c784a9b1e","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-16 09:35:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"74e86650564379fdd93f3db81c4c932082ad3e7f","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-16 09:41:20.000000000","message":"Patch Set 7:\n\nAdded probe test.\nDidn\u0027t reflect Matthew\u0027s review comments yet.","accounts_in_message":[],"_revision_number":7},{"id":"c6b7e6412fecac72fce458f7e04e5fc9ded99c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-16 11:37:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/7/check/build-openstack-sphinx-docs/416301b/html/ : SUCCESS in 7m 50s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/7/check/openstack-tox-pep8/db1e995/ : SUCCESS in 4m 25s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/7/check/legacy-swift-dsvm-functional/cfe0f58/ : SUCCESS in 22m 53s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/7/check/legacy-swift-dsvm-functional-identity-v3-only/09e8a40/ : SUCCESS in 26m 02s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/7/check/legacy-tempest-dsvm-neutron-full/e71ddd2/ : SUCCESS in 1h 38m 37s\n- neutron-grenade http://logs.openstack.org/89/517389/7/check/neutron-grenade/2a9ca9f/ : SUCCESS in 45m 25s\n- swift-tox-py27 http://logs.openstack.org/89/517389/7/check/swift-tox-py27/02bd498/ : SUCCESS in 7m 11s\n- swift-tox-py35 http://logs.openstack.org/89/517389/7/check/swift-tox-py35/50c8f38/ : SUCCESS in 3m 51s\n- swift-tox-func http://logs.openstack.org/89/517389/7/check/swift-tox-func/ad22ce4/ : SUCCESS in 7m 42s\n- swift-tox-func-post-as-copy http://logs.openstack.org/89/517389/7/check/swift-tox-func-post-as-copy/70ba819/ : SUCCESS in 6m 58s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/7/check/swift-tox-func-encryption/92414f7/ : SUCCESS in 8m 48s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/7/check/swift-tox-func-ec/68cad59/ : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":7},{"id":"b5e46e7b890af8ebc292ccda1a6683dfdf30bea1","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-17 10:00:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"02f19635fb4d2a7d96c2aa25ef8c28e1db3297df","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-17 10:04:27.000000000","message":"Patch Set 8:\n\n(4 comments)\n\nSquashed Matthew\u0027s docs following patch and fixed merge conflicts.\nNow, all review comments are reflected into this patch.\n\nSo I removed \u0027WIP\u0027 from the commit message.\nI think this patch is ready for review.","accounts_in_message":[],"_revision_number":8},{"id":"7858c1c7c5c97baee1dfe70013a38b2880a3c34d","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-17 10:11:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ca0b6ea722a8da3fd26782229059452a364c9bc0","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-17 10:14:30.000000000","message":"Patch Set 9:\n\nI restored some deleted files by mistake when fixing merge conflicts.\nI deleted them again.","accounts_in_message":[],"_revision_number":9},{"id":"b8c4674f90f9b8f74f24f37677f28f6098f8193b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 12:02:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/9/check/build-openstack-sphinx-docs/dbb7f5d/html/ : SUCCESS in 6m 01s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/9/check/openstack-tox-pep8/adbe5dd/ : SUCCESS in 4m 26s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/9/check/legacy-swift-dsvm-functional/5c406f2/ : SUCCESS in 26m 25s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/9/check/legacy-swift-dsvm-functional-identity-v3-only/c58d4fd/ : SUCCESS in 27m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/9/check/legacy-tempest-dsvm-neutron-full/39061e2/ : SUCCESS in 1h 37m 27s\n- neutron-grenade http://logs.openstack.org/89/517389/9/check/neutron-grenade/0c8cc61/ : SUCCESS in 46m 29s\n- swift-tox-py27 http://logs.openstack.org/89/517389/9/check/swift-tox-py27/25f5ead/ : SUCCESS in 12m 30s\n- swift-tox-py35 http://logs.openstack.org/89/517389/9/check/swift-tox-py35/fc4f5a6/ : SUCCESS in 3m 23s\n- swift-tox-func http://logs.openstack.org/89/517389/9/check/swift-tox-func/55c553b/ : SUCCESS in 9m 58s\n- swift-tox-func-post-as-copy http://logs.openstack.org/89/517389/9/check/swift-tox-func-post-as-copy/4015035/ : SUCCESS in 8m 12s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/9/check/swift-tox-func-encryption/114dd96/ : SUCCESS in 8m 08s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/9/check/swift-tox-func-ec/d766ccc/ : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":9},{"id":"cd5b09ddfba0417203101799ba94714992958bec","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-18 06:14:38.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"82e9a3f667c4ee280e99401376b6d8bec844497b","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-18 06:15:10.000000000","message":"Patch Set 10:\n\nFix merge conflict","accounts_in_message":[],"_revision_number":10},{"id":"485dd720fa31823f386427de00295efae94a07a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-18 07:56:25.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/10/check/build-openstack-sphinx-docs/ce61eb4/html/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/10/check/openstack-tox-pep8/bf33a9d/ : SUCCESS in 4m 42s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/10/check/legacy-swift-dsvm-functional/b02853f/ : SUCCESS in 22m 52s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/10/check/legacy-swift-dsvm-functional-identity-v3-only/ec6b4c6/ : SUCCESS in 29m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/10/check/legacy-tempest-dsvm-neutron-full/f8067af/ : SUCCESS in 1h 25m 54s\n- neutron-grenade http://logs.openstack.org/89/517389/10/check/neutron-grenade/4a0f6da/ : SUCCESS in 55m 16s\n- swift-tox-py27 http://logs.openstack.org/89/517389/10/check/swift-tox-py27/c03b855/ : SUCCESS in 8m 18s\n- swift-tox-py35 http://logs.openstack.org/89/517389/10/check/swift-tox-py35/4df5373/ : SUCCESS in 3m 09s\n- swift-tox-func http://logs.openstack.org/89/517389/10/check/swift-tox-func/de62a51/ : SUCCESS in 8m 50s\n- swift-tox-func-post-as-copy http://logs.openstack.org/89/517389/10/check/swift-tox-func-post-as-copy/fbd72a3/ : SUCCESS in 7m 37s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/10/check/swift-tox-func-encryption/f43d2d5/ : SUCCESS in 7m 47s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/10/check/swift-tox-func-ec/dfd5f46/ : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":10},{"id":"d3d9c5d806debd890850a1ef2b3f82c669f484da","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-01-18 09:11:11.000000000","message":"Patch Set 10: Code-Review-1\n\n(10 comments)\n\njust a quick look through the patch set.\n\nI think we should think about the upgrade impact from legacy config to the new config even we added legacy mode here. what i\u0027m worried about is the object-expier.conf is being a section in the object-sever.conf. For newbees, it would be no problem and it\u0027s reasonable because the general task queue needs object-srever ip ports. However, for existing clusters, it would be careful problem on the order of steps to upgrade. Exactly, adding the docs would be just a solution but I\u0027d like to hear opinions with considering the impact if the operator is missing the upgrading steps.","accounts_in_message":[],"_revision_number":10},{"id":"1199d630e34f2e33ef47a1866d8f05c305dfb98a","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-24 12:01:58.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"be9ec447c35b166cfd5e1d0df3c7f7e046b1b588","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-24 12:03:41.000000000","message":"Patch Set 11:\n\n(9 comments)\n\nReflected kota\u0027s comments and fixed merge conflict.\nI will discuss upgrade impact problem on IRC...","accounts_in_message":[],"_revision_number":11},{"id":"12272a7207c3521dbda2dbd4351e327c113de11e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-24 15:56:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/11/check/build-openstack-sphinx-docs/73a2d53/html/ : SUCCESS in 11m 18s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/11/check/openstack-tox-pep8/9a0f9f1/ : SUCCESS in 5m 12s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/11/check/legacy-swift-dsvm-functional/4a94236/ : SUCCESS in 43m 08s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/11/check/legacy-swift-dsvm-functional-identity-v3-only/7011ece/ : SUCCESS in 32m 28s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/11/check/legacy-tempest-dsvm-neutron-full/6332b40/ : FAILURE in 2h 12m 35s\n- neutron-grenade http://logs.openstack.org/89/517389/11/check/neutron-grenade/5feb6d1/ : SUCCESS in 55m 04s\n- swift-tox-py27 http://logs.openstack.org/89/517389/11/check/swift-tox-py27/fb32903/ : SUCCESS in 7m 27s\n- swift-tox-py35 http://logs.openstack.org/89/517389/11/check/swift-tox-py35/35d7040/ : SUCCESS in 3m 31s\n- swift-tox-func http://logs.openstack.org/89/517389/11/check/swift-tox-func/91542c0/ : SUCCESS in 17m 10s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/11/check/swift-tox-func-encryption/c8c5be2/ : SUCCESS in 8m 36s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/11/check/swift-tox-func-ec/653b695/ : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":11},{"id":"4b973daad10aae1b14dfb3469c33628d05bb2365","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-01-24 21:24:34.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"8e79dcae3c17bf2349142c13718ea0dee8a596d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-25 00:29:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/11/check/build-openstack-sphinx-docs/8fce593/html/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/11/check/openstack-tox-pep8/b455323/ : SUCCESS in 4m 09s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/11/check/legacy-swift-dsvm-functional/d4f261b/ : SUCCESS in 22m 01s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/11/check/legacy-swift-dsvm-functional-identity-v3-only/9c2a90b/ : SUCCESS in 26m 58s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/11/check/legacy-tempest-dsvm-neutron-full/7b72d40/ : SUCCESS in 1h 30m 03s\n- neutron-grenade http://logs.openstack.org/89/517389/11/check/neutron-grenade/ff3827a/ : SUCCESS in 43m 50s\n- swift-tox-py27 http://logs.openstack.org/89/517389/11/check/swift-tox-py27/097b5e6/ : SUCCESS in 7m 49s\n- swift-tox-py35 http://logs.openstack.org/89/517389/11/check/swift-tox-py35/54883c8/ : SUCCESS in 3m 26s\n- swift-tox-func http://logs.openstack.org/89/517389/11/check/swift-tox-func/715c2de/ : SUCCESS in 8m 15s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/11/check/swift-tox-func-encryption/a06357d/ : SUCCESS in 7m 44s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/11/check/swift-tox-func-ec/e1ff8c5/ : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":11},{"id":"dea2cd3fc6a9cc2d6ef13a72aee414e19ec01923","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-07 11:08:23.000000000","message":"Patch Set 11: Workflow-1\n\nI\u0027m refactoring this patch and splitting this into small patches.","accounts_in_message":[],"_revision_number":11},{"id":"ace06e66029a132050f6ae07a8e4aca88ff66439","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-09 08:51:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b6723cffda76e1e8ee0c407aad3561415aae5a74","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-09 08:52:17.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"911cdda4af27125213694c9181e09855d21823c2","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-09 10:06:21.000000000","message":"Topic changed from 517389 to expirer-general-task-queue","accounts_in_message":[],"_revision_number":12},{"id":"f7014cf427e11690efe19cca1df208fa1c779d4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-09 10:25:06.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/12/check/build-openstack-sphinx-docs/3ac76be/html/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/12/check/openstack-tox-pep8/1c64a27/ : FAILURE in 3m 55s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/12/check/legacy-swift-dsvm-functional/096aa5e/ : SUCCESS in 28m 15s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/12/check/legacy-swift-dsvm-functional-identity-v3-only/efc7ee7/ : SUCCESS in 26m 33s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/12/check/legacy-tempest-dsvm-neutron-full/0ba1415/ : SUCCESS in 1h 30m 48s\n- neutron-grenade http://logs.openstack.org/89/517389/12/check/neutron-grenade/f6b549f/ : SUCCESS in 50m 12s\n- swift-tox-py27 http://logs.openstack.org/89/517389/12/check/swift-tox-py27/5230995/ : SUCCESS in 6m 49s\n- swift-tox-py35 http://logs.openstack.org/89/517389/12/check/swift-tox-py35/7dbeb41/ : SUCCESS in 3m 01s\n- swift-tox-func http://logs.openstack.org/89/517389/12/check/swift-tox-func/64e977a/ : SUCCESS in 8m 36s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/12/check/swift-tox-func-encryption/4e8eae2/ : SUCCESS in 7m 10s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/12/check/swift-tox-func-ec/77b890c/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":12},{"id":"f3ff3c5f9d87ff0db615a6bf2b195a4840be149c","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-13 12:09:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"afbb6d136a0394177485bf8114dade0350be18ba","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-13 12:12:36.000000000","message":"Patch Set 13: Workflow-1\n\nI keep refactoring this patch now...","accounts_in_message":[],"_revision_number":13},{"id":"be5d50a1443d7617025c5014f1cd7b7d71ddaad6","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-13 12:57:08.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"74a2adad3ffe8ecdacf8c8f03a7a438e85884780","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 14:31:05.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/14/check/build-openstack-sphinx-docs/0438090/html/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/14/check/openstack-tox-pep8/e0050eb/ : SUCCESS in 3m 46s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/14/check/legacy-swift-dsvm-functional/cc4101f/ : SUCCESS in 22m 32s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/14/check/legacy-swift-dsvm-functional-identity-v3-only/0b45b2e/ : SUCCESS in 25m 59s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/14/check/legacy-tempest-dsvm-neutron-full/37b8245/ : SUCCESS in 1h 32m 25s\n- neutron-grenade http://logs.openstack.org/89/517389/14/check/neutron-grenade/74d1f8c/ : SUCCESS in 51m 54s\n- swift-tox-py27 http://logs.openstack.org/89/517389/14/check/swift-tox-py27/f4655f3/ : SUCCESS in 7m 49s\n- swift-tox-py35 http://logs.openstack.org/89/517389/14/check/swift-tox-py35/ceb3d84/ : SUCCESS in 3m 09s\n- swift-tox-func http://logs.openstack.org/89/517389/14/check/swift-tox-func/d3a9fbe/ : SUCCESS in 7m 12s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/14/check/swift-tox-func-encryption/d74aa7b/ : SUCCESS in 6m 55s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/14/check/swift-tox-func-ec/250c1e7/ : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":14},{"id":"055647ae8f08cb1b0d8d531ef866a253fd139ce7","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-19 11:30:54.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b021edeee733e23e23a5cfe35167f51bc5b364fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-19 13:02:58.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/15/check/build-openstack-sphinx-docs/3636429/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/15/check/openstack-tox-pep8/3d56a02/ : SUCCESS in 4m 10s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/15/check/legacy-swift-dsvm-functional/5a3a4c4/ : SUCCESS in 22m 48s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/15/check/legacy-swift-dsvm-functional-identity-v3-only/c2c8746/ : SUCCESS in 29m 11s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/15/check/legacy-tempest-dsvm-neutron-full/58d7be1/ : SUCCESS in 1h 25m 17s\n- neutron-grenade http://logs.openstack.org/89/517389/15/check/neutron-grenade/202de3f/ : SUCCESS in 54m 21s\n- swift-tox-py27 http://logs.openstack.org/89/517389/15/check/swift-tox-py27/83b5ebc/ : SUCCESS in 7m 29s\n- swift-tox-py35 http://logs.openstack.org/89/517389/15/check/swift-tox-py35/51e5dfb/ : SUCCESS in 3m 25s\n- swift-tox-func http://logs.openstack.org/89/517389/15/check/swift-tox-func/2b4673c/ : SUCCESS in 6m 42s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/15/check/swift-tox-func-encryption/009fd3f/ : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/15/check/swift-tox-func-ec/599efba/ : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":15},{"id":"1d6702be3927cbe3e6741d3bcda8e624f4915d9c","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-20 01:44:56.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"cea5457f0bc874e0bfad12287d149673cfba78f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-20 03:30:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/15/check/build-openstack-sphinx-docs/880fba0/html/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/15/check/openstack-tox-pep8/458b4bb/ : SUCCESS in 3m 37s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/15/check/legacy-swift-dsvm-functional/6d4c7d0/ : SUCCESS in 24m 31s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/15/check/legacy-swift-dsvm-functional-identity-v3-only/9638ceb/ : SUCCESS in 24m 01s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/15/check/legacy-tempest-dsvm-neutron-full/3dfb6a9/ : SUCCESS in 1h 39m 51s\n- neutron-grenade http://logs.openstack.org/89/517389/15/check/neutron-grenade/139e679/ : SUCCESS in 48m 01s\n- swift-tox-py27 http://logs.openstack.org/89/517389/15/check/swift-tox-py27/977fc94/ : SUCCESS in 6m 36s\n- swift-tox-py35 http://logs.openstack.org/89/517389/15/check/swift-tox-py35/598dd5a/ : SUCCESS in 3m 42s\n- swift-tox-func http://logs.openstack.org/89/517389/15/check/swift-tox-func/f4d2049/ : SUCCESS in 8m 26s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/15/check/swift-tox-func-encryption/b5fd754/ : SUCCESS in 7m 52s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/15/check/swift-tox-func-ec/ed9a5ac/ : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":15},{"id":"1135b41c996074c86322933fceb6582a3ac24f18","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-27 09:36:45.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f31f3e4608f4fd8b33a5a689226ae276d5258844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-27 10:58:59.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/16/check/build-openstack-sphinx-docs/008bd5f/html/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/16/check/openstack-tox-pep8/9ed289e/ : SUCCESS in 4m 29s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/16/check/legacy-swift-dsvm-functional/b031c13/ : SUCCESS in 27m 55s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/16/check/legacy-swift-dsvm-functional-identity-v3-only/15af32e/ : SUCCESS in 25m 39s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/16/check/legacy-tempest-dsvm-neutron-full/c78f33e/ : SUCCESS in 1h 20m 44s\n- neutron-grenade http://logs.openstack.org/89/517389/16/check/neutron-grenade/4d5f7db/ : SUCCESS in 40m 36s\n- swift-tox-py27 http://logs.openstack.org/89/517389/16/check/swift-tox-py27/e80b397/ : SUCCESS in 6m 32s\n- swift-tox-py35 http://logs.openstack.org/89/517389/16/check/swift-tox-py35/28c2a1b/ : SUCCESS in 4m 00s\n- swift-tox-func http://logs.openstack.org/89/517389/16/check/swift-tox-func/eb40ef6/ : SUCCESS in 7m 10s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/16/check/swift-tox-func-encryption/4aedb14/ : SUCCESS in 7m 17s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/16/check/swift-tox-func-ec/2bea7f7/ : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":16},{"id":"aa705c01d661408efdcc083852ca04184509ab91","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-28 12:15:27.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d69f71bc3dddf65094b1c4a2dfbc744c1df0d600","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-28 13:55:41.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/17/check/build-openstack-sphinx-docs/d95f033/html/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/17/check/openstack-tox-pep8/e26b93c/ : SUCCESS in 4m 03s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/17/check/legacy-swift-dsvm-functional/aeffea6/ : SUCCESS in 26m 14s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/17/check/legacy-swift-dsvm-functional-identity-v3-only/23f64e1/ : SUCCESS in 21m 58s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/17/check/legacy-tempest-dsvm-neutron-full/f1e4e66/ : SUCCESS in 1h 34m 56s\n- neutron-grenade http://logs.openstack.org/89/517389/17/check/neutron-grenade/145f426/ : SUCCESS in 44m 22s\n- swift-tox-py27 http://logs.openstack.org/89/517389/17/check/swift-tox-py27/f2ebc6f/ : SUCCESS in 7m 02s\n- swift-tox-py35 http://logs.openstack.org/89/517389/17/check/swift-tox-py35/037642f/ : SUCCESS in 3m 40s\n- swift-tox-func http://logs.openstack.org/89/517389/17/check/swift-tox-func/1454742/ : SUCCESS in 6m 41s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/17/check/swift-tox-func-encryption/6276ddc/ : SUCCESS in 7m 44s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/17/check/swift-tox-func-ec/ad17927/ : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":17},{"id":"a0b59e68d1acb151cfed57e7e11e8a7e21c0faaa","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-28 15:04:39.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"35dda0be1709221077f8da0dda6132bf89d4f94d","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-28 15:05:36.000000000","message":"Patch Set 18:\n\nrebased and fixed small mistakes","accounts_in_message":[],"_revision_number":18},{"id":"435644f93258b07875ef3b4dee52a88295d001cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-28 16:35:05.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/18/check/build-openstack-sphinx-docs/6124e79/html/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/18/check/openstack-tox-pep8/44fb655/ : SUCCESS in 3m 49s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/18/check/legacy-swift-dsvm-functional/831c390/ : SUCCESS in 22m 39s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/18/check/legacy-swift-dsvm-functional-identity-v3-only/474fec1/ : SUCCESS in 22m 19s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/18/check/legacy-tempest-dsvm-neutron-full/c6ba507/ : SUCCESS in 1h 24m 50s\n- neutron-grenade http://logs.openstack.org/89/517389/18/check/neutron-grenade/911b8ef/ : SUCCESS in 44m 38s\n- swift-tox-py27 http://logs.openstack.org/89/517389/18/check/swift-tox-py27/2dcffa3/ : SUCCESS in 6m 31s\n- swift-tox-py35 http://logs.openstack.org/89/517389/18/check/swift-tox-py35/92cb8fb/ : FAILURE in 4m 06s\n- swift-tox-func http://logs.openstack.org/89/517389/18/check/swift-tox-func/1b52420/ : SUCCESS in 7m 09s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/18/check/swift-tox-func-encryption/22991dd/ : SUCCESS in 6m 23s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/18/check/swift-tox-func-ec/b76982d/ : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":18},{"id":"d9e83358083301bb7f1eac546d55ef8884f6ad22","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-28 17:10:54.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"1159701bd8b72f768611895d5566c2460753efd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-28 18:27:51.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/18/check/build-openstack-sphinx-docs/2e114f2/html/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/18/check/openstack-tox-pep8/89df178/ : SUCCESS in 4m 10s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/18/check/legacy-swift-dsvm-functional/2547990/ : SUCCESS in 28m 21s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/18/check/legacy-swift-dsvm-functional-identity-v3-only/b2f8718/ : SUCCESS in 21m 09s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/18/check/legacy-tempest-dsvm-neutron-full/147e70a/ : SUCCESS in 1h 14m 34s\n- neutron-grenade http://logs.openstack.org/89/517389/18/check/neutron-grenade/5c4b48f/ : SUCCESS in 51m 24s\n- swift-tox-py27 http://logs.openstack.org/89/517389/18/check/swift-tox-py27/65ffceb/ : SUCCESS in 8m 27s\n- swift-tox-py35 http://logs.openstack.org/89/517389/18/check/swift-tox-py35/36ec584/ : SUCCESS in 3m 40s\n- swift-tox-func http://logs.openstack.org/89/517389/18/check/swift-tox-func/0356d5f/ : SUCCESS in 6m 09s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/18/check/swift-tox-func-encryption/3b1bf1d/ : SUCCESS in 7m 37s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/18/check/swift-tox-func-ec/3bcb6c1/ : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":18},{"id":"1b6f4fff6eb33ed783069d9642166f83ee4777d3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-02-28 23:41:29.000000000","message":"Patch Set 18:\n\nThe reason why endless rechecks for py35 is unrelated to Kazuhiro\u0027s patch, it\u0027s this:\n\n\u003e\u003e\u003e now \u003d 1519830570.6949349\n\u003e\u003e\u003e now\n1519830570.6949348\n\u003e\u003e\u003e timestamp \u003d Timestamp(now)\n\u003e\u003e\u003e timestamp\n1519830570.69493_0000000000000000\n\u003e\u003e\u003e timestamp \u003d Timestamp(now, offset\u003d1)\n\u003e\u003e\u003e timestamp\n1519830570.69493_0000000000000001\n\u003e\u003e\u003e value \u003d \u0027%f\u0027 % now\n\u003e\u003e\u003e value\n\u00271519830570.694935\u0027\n\u003e\u003e\u003e timestamp \u003e value\nFalse\n\u003e\u003e\u003e","accounts_in_message":[],"_revision_number":18},{"id":"e5080362edf5f92b2ecc6e9a955f18cb3a5a671e","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-01 09:35:36.000000000","message":"Patch Set 18: Workflow-1\n\nPete, thank you for comments!\n\nI got advice that the tests difference between lagacy and general should be processed by overriding parameters or methods, not parameterizing test functions from kota.\nSo I will change test implementation.","accounts_in_message":[],"_revision_number":18},{"id":"c74467eaf71e3c32c037348d14a1b6a78b0c4458","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-01 18:34:14.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"65e1a5abbfd0ade4c64ff1a22ad2455518676527","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-01 20:00:48.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/19/check/build-openstack-sphinx-docs/ce0fdcc/html/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/19/check/openstack-tox-pep8/e26a9b5/ : SUCCESS in 4m 04s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/19/check/legacy-swift-dsvm-functional/8eb4515/ : SUCCESS in 23m 59s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/19/check/legacy-swift-dsvm-functional-identity-v3-only/f264c99/ : SUCCESS in 27m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/19/check/legacy-tempest-dsvm-neutron-full/1f5d6f7/ : SUCCESS in 1h 23m 07s\n- neutron-grenade http://logs.openstack.org/89/517389/19/check/neutron-grenade/df5e88d/ : SUCCESS in 46m 43s\n- swift-tox-py27 http://logs.openstack.org/89/517389/19/check/swift-tox-py27/e11a9fa/ : SUCCESS in 6m 54s\n- swift-tox-py35 http://logs.openstack.org/89/517389/19/check/swift-tox-py35/fd889cf/ : SUCCESS in 3m 10s\n- swift-tox-func http://logs.openstack.org/89/517389/19/check/swift-tox-func/599a5bd/ : SUCCESS in 6m 58s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/19/check/swift-tox-func-encryption/7d2650b/ : SUCCESS in 8m 45s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/19/check/swift-tox-func-ec/7f9ce0c/ : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":19},{"id":"1350d2c0dab70031a30d240c5cbf8fa456074606","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-20 08:32:47.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"6a5ba71a2a9fe83320be3d0f43074db8b8aea400","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-20 08:34:13.000000000","message":"Patch Set 19:\n\nReflected kota\u0027s advice.","accounts_in_message":[],"_revision_number":19},{"id":"02dd50e0146a9d66582846a711929104d6149894","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-20 10:23:20.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/20/check/build-openstack-sphinx-docs/3fd2885/html/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/20/check/openstack-tox-pep8/f0373ff/ : SUCCESS in 3m 44s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/20/check/legacy-swift-dsvm-functional/45a4c10/ : SUCCESS in 24m 46s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/20/check/legacy-swift-dsvm-functional-identity-v3-only/5a632b4/ : SUCCESS in 21m 53s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/20/check/legacy-tempest-dsvm-neutron-full/28e1ac7/ : POST_FAILURE in 1h 47m 19s\n- neutron-grenade http://logs.openstack.org/89/517389/20/check/neutron-grenade/8acf11e/ : POST_FAILURE in 1h 12m 57s\n- swift-tox-py27 http://logs.openstack.org/89/517389/20/check/swift-tox-py27/2e3b4f0/ : SUCCESS in 7m 00s\n- swift-tox-py35 http://logs.openstack.org/89/517389/20/check/swift-tox-py35/39a8a11/ : SUCCESS in 3m 39s\n- swift-tox-func http://logs.openstack.org/89/517389/20/check/swift-tox-func/aabce4b/ : SUCCESS in 7m 31s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/20/check/swift-tox-func-encryption/eb041c2/ : SUCCESS in 7m 30s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/20/check/swift-tox-func-ec/b7ffa34/ : SUCCESS in 7m 50s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/20/check/swift-tox-func-domain-remap-staticweb/fa85b6b/ : SUCCESS in 7m 02s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/20/check/swift-probetests-centos-7/5c8d780/ : FAILURE in 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e09e89fa6d64313431edb836d567cb71a9b10592","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-20 11:29:37.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"7ff99349e9889d9f0b286f561e031f6df50724d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-20 13:17:33.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/20/check/build-openstack-sphinx-docs/88ec02e/html/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/20/check/openstack-tox-pep8/4a92769/ : SUCCESS in 3m 54s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/20/check/legacy-swift-dsvm-functional/fe98d78/ : SUCCESS in 25m 25s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/20/check/legacy-swift-dsvm-functional-identity-v3-only/1deef3d/ : SUCCESS in 27m 50s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/20/check/legacy-tempest-dsvm-neutron-full/3461454/ : SUCCESS in 1h 40m 07s\n- neutron-grenade http://logs.openstack.org/89/517389/20/check/neutron-grenade/cf88ebb/ : SUCCESS in 51m 38s\n- swift-tox-py27 http://logs.openstack.org/89/517389/20/check/swift-tox-py27/6fc6f00/ : SUCCESS in 7m 07s\n- swift-tox-py35 http://logs.openstack.org/89/517389/20/check/swift-tox-py35/5bf55d6/ : SUCCESS in 3m 47s\n- swift-tox-func http://logs.openstack.org/89/517389/20/check/swift-tox-func/a47b88f/ : SUCCESS in 8m 17s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/20/check/swift-tox-func-encryption/98784eb/ : SUCCESS in 6m 47s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/20/check/swift-tox-func-domain-remap-staticweb/8c733bd/ : SUCCESS in 7m 00s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/20/check/swift-tox-func-ec/2efe3fb/ : SUCCESS in 7m 19s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/20/check/swift-probetests-centos-7/da6b2e7/ : FAILURE in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5e989f27711580dfa1b3eb0ceb3195c15605519a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-03-20 15:35:33.000000000","message":"Patch Set 20:\n\nI worry a bit about the upgrade impact revealed by the failing probe test ... are we going to need to inline an internal-client.conf similar to what we did for container-sync? See https://github.com/openstack/swift/blob/2.17.0/swift/container/sync.py#L47-L97\n\nI suppose having a busted expirer during the upgrade isn\u0027t the worst thing... we still 404 just fine, we just won\u0027t clean up until we can get a fresh config out...","accounts_in_message":[],"_revision_number":20},{"id":"7219c36a46258ff85a832b56361e3b8c4f770461","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-26 10:22:42.000000000","message":"Patch Set 20:\n\n\u003e are we going to need to inline an internal-client.conf similar to what we did for container-sync?\n\nI don\u0027t think so. The reason why container-sync needs inline internal-client.conf is that container-sync have used direct-client before. So inline conf is required to get backward compatibility.\n\nIn this patch, object-expirer can work with \"object-expirer.conf\" in legacy-queue mode without any change of the conf file. So I think this change has backward compatibility.\n\nThe probe test failures come from difference between configure file style of legacy-queue mode and general-queue mode. In my opinion, we should think general-queue mode object-expirer as a kind of new daemon. So there are no backward compatibility for general-queue mode and this is intentional.\n\nAnyway, I need to fix probe tests...","accounts_in_message":[],"_revision_number":20},{"id":"b48bf38ba97c76bb133e7ce2efe83fdf0bdccd00","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-27 08:45:39.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"3c585b9c6f7a66a585bc03db3da7eb15cd19f8d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 10:25:49.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/21/check/build-openstack-sphinx-docs/aec6395/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/21/check/openstack-tox-pep8/eeebc26/ : SUCCESS in 3m 47s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/21/check/legacy-swift-dsvm-functional/dc43ce5/ : SUCCESS in 26m 58s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/21/check/legacy-swift-dsvm-functional-identity-v3-only/8dbe51b/ : SUCCESS in 21m 50s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/21/check/legacy-tempest-dsvm-neutron-full/6274e04/ : SUCCESS in 1h 35m 28s\n- neutron-grenade http://logs.openstack.org/89/517389/21/check/neutron-grenade/4dbb488/ : SUCCESS in 43m 41s\n- swift-tox-py27 http://logs.openstack.org/89/517389/21/check/swift-tox-py27/449cca5/ : SUCCESS in 6m 52s\n- swift-tox-py35 http://logs.openstack.org/89/517389/21/check/swift-tox-py35/2b9c669/ : SUCCESS in 3m 43s\n- swift-tox-func http://logs.openstack.org/89/517389/21/check/swift-tox-func/d73f864/ : SUCCESS in 7m 29s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/21/check/swift-tox-func-encryption/44a6ddd/ : SUCCESS in 7m 12s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/21/check/swift-tox-func-ec/c58031f/ : SUCCESS in 8m 13s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/21/check/swift-tox-func-domain-remap-staticweb/ce2ebce/ : SUCCESS in 8m 58s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/21/check/swift-probetests-centos-7/368015d/ : SUCCESS in 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f00ad28d311f9adf31c53215fbb6853eae229dab","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-03-28 06:27:47.000000000","message":"Patch Set 21:\n\nProbe tests are fixed.","accounts_in_message":[],"_revision_number":21},{"id":"347d9ee23ef8d1ec7bef63fc35105abcd33f2ddc","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-03-30 09:05:51.000000000","message":"Patch Set 21: Code-Review-1\n\n(4 comments)\n\nStill looking yet, but push my current small comments before i finished this week work.","accounts_in_message":[],"_revision_number":21},{"id":"4793cd2f34a56c9bd839aab3dfaf8f998e07d22b","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-05 12:53:45.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"897b8921431f2fc8ee08921cdc95b723defee4f7","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-05 12:56:25.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"d833c231557ad480a7168d411f66ed7add4e99b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-05 14:34:59.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/22/check/build-openstack-sphinx-docs/89c2b22/html/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/22/check/openstack-tox-pep8/3a620e2/ : SUCCESS in 4m 53s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/22/check/legacy-swift-dsvm-functional/73fb2a4/ : SUCCESS in 33m 06s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/22/check/legacy-swift-dsvm-functional-identity-v3-only/9070f7a/ : SUCCESS in 27m 04s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/22/check/legacy-tempest-dsvm-neutron-full/c44df66/ : SUCCESS in 1h 35m 09s\n- neutron-grenade http://logs.openstack.org/89/517389/22/check/neutron-grenade/8fe6997/ : SUCCESS in 55m 34s\n- swift-tox-py27 http://logs.openstack.org/89/517389/22/check/swift-tox-py27/aff77f7/ : SUCCESS in 13m 04s\n- swift-tox-py35 http://logs.openstack.org/89/517389/22/check/swift-tox-py35/14da57a/ : SUCCESS in 3m 32s\n- swift-tox-func http://logs.openstack.org/89/517389/22/check/swift-tox-func/3fc31d8/ : SUCCESS in 8m 05s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/22/check/swift-tox-func-encryption/0e3fdb5/ : SUCCESS in 13m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/22/check/swift-tox-func-domain-remap-staticweb/73d8963/ : SUCCESS in 7m 36s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/22/check/swift-tox-func-ec/ba57af8/ : SUCCESS in 9m 22s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/22/check/swift-probetests-centos-7/561db95/ : SUCCESS in 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"15294de9ab24458a4d70875e59841746bc26905b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-04-11 05:26:37.000000000","message":"Patch Set 22: Code-Review-1\n\n(2 comments)\n\nThe design documents raised significant concerns about the migration, but the actual code is almost perfect (if I read it right): if one just updates the code, the old configurations continue to work as before, I think. There is only one problem: the changed PID file name. IMHO that is unnecessary. (-1 for discussion)","accounts_in_message":[],"_revision_number":22},{"id":"369676fdf93671aaeede8c751afbaa6ed703cbc3","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-04-12 06:37:46.000000000","message":"Patch Set 22:\n\nPete:\n\nIs that about \u0027legacy-expirer\u0027 in the manager? It is what I just was talking about with Kazuhiro a few days ago in-person. Then, I think he will fix it to be consistent with the older behavior (i.e. object-expirer). Thanks for good comments.","accounts_in_message":[],"_revision_number":22},{"id":"0b8a2b2f20dc4e1efe2fd329572f74cb52603cd9","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-17 09:36:12.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"88477240217e2e3e8c123f3f5676198413c3043f","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-17 09:44:26.000000000","message":"Patch Set 23:\n\n(2 comments)\n\nPete, Kota, thank you for review.\n\nI simplified PID file name implementation. I think this implementation is better than previous patch set :)","accounts_in_message":[],"_revision_number":23},{"id":"79ea818197cb00ae66b1ed4d4567cbea7379d93d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-17 11:38:38.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/23/check/build-openstack-sphinx-docs/05176eb/html/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/23/check/openstack-tox-pep8/7cd37b7/ : SUCCESS in 3m 25s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/23/check/legacy-swift-dsvm-functional/614fd08/ : SUCCESS in 21m 38s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/23/check/legacy-swift-dsvm-functional-identity-v3-only/945c24f/ : SUCCESS in 26m 11s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/23/check/legacy-tempest-dsvm-neutron-full/871fac2/ : SUCCESS in 1h 59m 15s\n- neutron-grenade http://logs.openstack.org/89/517389/23/check/neutron-grenade/8d259c4/ : SUCCESS in 40m 12s\n- swift-tox-py27 http://logs.openstack.org/89/517389/23/check/swift-tox-py27/f64ccd2/ : SUCCESS in 6m 41s\n- swift-tox-py35 http://logs.openstack.org/89/517389/23/check/swift-tox-py35/af286ff/ : SUCCESS in 3m 59s\n- swift-tox-func http://logs.openstack.org/89/517389/23/check/swift-tox-func/25af95f/ : SUCCESS in 6m 38s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/23/check/swift-tox-func-encryption/96d547a/ : SUCCESS in 7m 26s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/23/check/swift-tox-func-domain-remap-staticweb/559d3d1/ : SUCCESS in 6m 25s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/23/check/swift-tox-func-ec/5e392f0/ : SUCCESS in 7m 25s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/23/check/swift-probetests-centos-7/89fa4d6/ : FAILURE in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"c5db46ceb2c0987c677e66c7d83b9995c8be8dc0","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-18 00:38:22.000000000","message":"Patch Set 23:\n\nSome probe tests failed... I will fix them.","accounts_in_message":[],"_revision_number":23},{"id":"11b3d2a52f134370ca9c5ad4e2ed31de8ab60027","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-18 00:54:03.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"a04b9d999c38a67d59d266485b31dae13cb345f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-18 02:25:41.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/23/check/build-openstack-sphinx-docs/8e4cc7a/html/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/23/check/openstack-tox-pep8/ebb0f59/ : SUCCESS in 3m 41s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/23/check/legacy-swift-dsvm-functional/1c62102/ : SUCCESS in 22m 20s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/23/check/legacy-swift-dsvm-functional-identity-v3-only/5d1772d/ : SUCCESS in 24m 21s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/23/check/legacy-tempest-dsvm-neutron-full/0414cbb/ : SUCCESS in 1h 26m 10s\n- neutron-grenade http://logs.openstack.org/89/517389/23/check/neutron-grenade/3113924/ : SUCCESS in 53m 47s\n- swift-tox-py27 http://logs.openstack.org/89/517389/23/check/swift-tox-py27/335cafd/ : SUCCESS in 6m 45s\n- swift-tox-py35 http://logs.openstack.org/89/517389/23/check/swift-tox-py35/ce00641/ : SUCCESS in 3m 53s\n- swift-tox-func http://logs.openstack.org/89/517389/23/check/swift-tox-func/c7366bd/ : SUCCESS in 7m 03s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/23/check/swift-tox-func-encryption/d4bc9d9/ : SUCCESS in 8m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/23/check/swift-tox-func-domain-remap-staticweb/4666344/ : SUCCESS in 7m 21s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/23/check/swift-tox-func-ec/7b1b0cf/ : SUCCESS in 8m 21s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/23/check/swift-probetests-centos-7/71de127/ : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"2344b1e75de06ea72d27a4143aaab223e483f664","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-18 04:56:56.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"a315d248771a15d70f9929479c829b838f4e2cfc","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-04-18 05:10:06.000000000","message":"Patch Set 24:\n\nFixed probe tests.\nThe failures are caused by object-expirer.conf in saio.\nObject-expirers ran as legacy-mode with the conf.","accounts_in_message":[],"_revision_number":24},{"id":"3056d2da8dbb48aa42c6f2a63eeb1d29be25d56e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-18 06:23:36.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/24/check/build-openstack-sphinx-docs/2cdd62e/html/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/24/check/openstack-tox-pep8/2a6de5d/ : SUCCESS in 4m 30s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/24/check/legacy-swift-dsvm-functional/cc0bce7/ : SUCCESS in 24m 11s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/24/check/legacy-swift-dsvm-functional-identity-v3-only/8743913/ : SUCCESS in 23m 55s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/24/check/legacy-tempest-dsvm-neutron-full/13967c1/ : SUCCESS in 1h 24m 19s\n- neutron-grenade http://logs.openstack.org/89/517389/24/check/neutron-grenade/953b15b/ : SUCCESS in 46m 06s\n- swift-tox-py27 http://logs.openstack.org/89/517389/24/check/swift-tox-py27/9f958b6/ : SUCCESS in 7m 48s\n- swift-tox-py35 http://logs.openstack.org/89/517389/24/check/swift-tox-py35/ffcf4e0/ : SUCCESS in 3m 44s\n- swift-tox-func http://logs.openstack.org/89/517389/24/check/swift-tox-func/e0465ab/ : SUCCESS in 7m 13s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/24/check/swift-tox-func-encryption/2bbe629/ : SUCCESS in 7m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/24/check/swift-tox-func-domain-remap-staticweb/6319bdc/ : SUCCESS in 8m 05s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/24/check/swift-tox-func-ec/05f3fce/ : SUCCESS in 9m 05s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/24/check/swift-probetests-centos-7/43daae8/ : SUCCESS in 22m 16s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"84708e501677bfd4d6fd51e2a034503274a800f2","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-05-24 22:34:33.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"e4a4b95e0d811829115aa6b103624c44a7c3c75b","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-05-24 22:37:08.000000000","message":"Patch Set 25:\n\nRebased and Fixed merge conflict.","accounts_in_message":[],"_revision_number":25},{"id":"e9bb8efb7d1c1e59517f8401ec763880a843e1b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-25 00:11:51.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/25/check/build-openstack-sphinx-docs/966d4c2/html/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/25/check/openstack-tox-pep8/b475bf0/ : SUCCESS in 5m 03s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/25/check/legacy-swift-dsvm-functional/b61f583/ : SUCCESS in 26m 57s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/25/check/legacy-swift-dsvm-functional-identity-v3-only/7ebe8a9/ : SUCCESS in 22m 59s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/25/check/legacy-tempest-dsvm-neutron-full/101d93e/ : SUCCESS in 1h 26m 36s\n- neutron-grenade http://logs.openstack.org/89/517389/25/check/neutron-grenade/79399c4/ : SUCCESS in 48m 47s\n- swift-tox-py27 http://logs.openstack.org/89/517389/25/check/swift-tox-py27/0863e6f/ : SUCCESS in 9m 44s\n- swift-tox-py35 http://logs.openstack.org/89/517389/25/check/swift-tox-py35/ce7becf/ : SUCCESS in 5m 11s\n- swift-tox-func http://logs.openstack.org/89/517389/25/check/swift-tox-func/becf3d0/ : SUCCESS in 7m 53s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/25/check/swift-tox-func-encryption/6234bbc/ : SUCCESS in 8m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/25/check/swift-tox-func-domain-remap-staticweb/507d9a8/ : SUCCESS in 10m 46s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/25/check/swift-tox-func-ec/751f58c/ : SUCCESS in 11m 00s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/25/check/swift-tox-func-s3api/268c8c3/ : SUCCESS in 11m 13s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/25/check/swift-probetests-centos-7/33371b1/ : SUCCESS in 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"dbaa2102b566e7398a2f7add47c3cea975827a73","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-06-08 01:28:45.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"239cca823bc62b90efa4c918e5d72b8ae231ce77","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-06-08 01:30:36.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"26bae8222ab5217b703d085455cbb7f6dd49fcb0","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-06-08 01:31:27.000000000","message":"Patch Set 27:\n\nMerge conflict is fixed.","accounts_in_message":[],"_revision_number":27},{"id":"e578878790d5a8fc7ab696b328e9b75381eab0ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-08 03:13:22.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/517389/27/check/build-openstack-sphinx-docs/14a7aac/ : FAILURE in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/27/check/openstack-tox-pep8/3f6d217/ : SUCCESS in 4m 50s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/27/check/legacy-swift-dsvm-functional/f4aaafa/ : SUCCESS in 26m 29s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/27/check/legacy-swift-dsvm-functional-identity-v3-only/6eb4bba/ : SUCCESS in 28m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/27/check/legacy-tempest-dsvm-neutron-full/c1f282e/ : SUCCESS in 1h 36m 30s\n- neutron-grenade http://logs.openstack.org/89/517389/27/check/neutron-grenade/15d6b0d/ : SUCCESS in 53m 47s\n- swift-tox-py27 http://logs.openstack.org/89/517389/27/check/swift-tox-py27/2d4e6b3/ : SUCCESS in 9m 34s\n- swift-tox-py35 http://logs.openstack.org/89/517389/27/check/swift-tox-py35/76addfc/ : SUCCESS in 3m 41s\n- swift-tox-func http://logs.openstack.org/89/517389/27/check/swift-tox-func/97bee02/ : SUCCESS in 8m 45s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/27/check/swift-tox-func-encryption/a58df87/ : SUCCESS in 8m 21s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/27/check/swift-tox-func-domain-remap-staticweb/03a9664/ : SUCCESS in 8m 30s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/27/check/swift-tox-func-ec/440cbab/ : SUCCESS in 8m 27s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/27/check/swift-tox-func-s3api/3d97129/ : SUCCESS in 10m 16s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/27/check/swift-probetests-centos-7/40a491e/ : SUCCESS in 31m 21s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"02e5e862b01225c760484b72866519a27656b234","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-09-10 20:05:27.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"a58c2f06444273b2c80a8539a83a89d3bf539cdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-10 21:41:15.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/28/check/openstack-tox-docs/7a6bce3/html/ : SUCCESS in 4m 43s\n- swift-tox-py27 http://logs.openstack.org/89/517389/28/check/swift-tox-py27/a1b1008/ : SUCCESS in 10m 59s\n- swift-tox-py35 http://logs.openstack.org/89/517389/28/check/swift-tox-py35/106455d/ : SUCCESS in 4m 39s\n- swift-tox-func http://logs.openstack.org/89/517389/28/check/swift-tox-func/e21ca17/ : SUCCESS in 8m 42s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/28/check/swift-tox-func-encryption/e8e1fe1/ : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/28/check/swift-tox-func-domain-remap-staticweb/a77a0cd/ : SUCCESS in 9m 23s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/28/check/swift-tox-func-ec/f934321/ : SUCCESS in 9m 55s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/28/check/swift-tox-func-s3api/6d79fb4/ : SUCCESS in 10m 38s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/28/check/swift-probetests-centos-7/ee31450/ : SUCCESS in 31m 42s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/28/check/swift-dsvm-functional/10ddfbd/ : SUCCESS in 22m 35s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/517389/28/check/legacy-tempest-dsvm-neutron-full/c9e9bc4/ : SUCCESS in 1h 32m 48s\n- neutron-grenade http://logs.openstack.org/89/517389/28/check/neutron-grenade/c4aaabb/ : SUCCESS in 47m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/517389/28/check/openstack-tox-lower-constraints/13ccb0c/ : SUCCESS in 7m 45s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/28/check/openstack-tox-pep8/a2e2519/ : SUCCESS in 5m 15s\n- legacy-swift-dsvm-functional http://logs.openstack.org/89/517389/28/check/legacy-swift-dsvm-functional/65708e7/ : SUCCESS in 30m 39s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/89/517389/28/check/legacy-swift-dsvm-functional-identity-v3-only/8d83092/ : SUCCESS in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"fd2ba1cd8ababcff88e580f2719e088b061324fa","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-10 22:35:06.000000000","message":"Patch Set 28:\n\n\u003e kota\nThank you for your patch update!","accounts_in_message":[],"_revision_number":28},{"id":"9a48c99816e1fef740d3dc4696ee412790ef18b6","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 05:21:14.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"d79111014e1c6301732163dd63531529008f0226","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 05:37:11.000000000","message":"Patch Set 29:\n\nChanges in patch set 29 are following:\n1. Added `enqueue_expirirng_objects_to_legacy_queue` option to proxy-server and object-server. So, expirer tasks\u0027 destination queue is configurable now.\n2. Restored the removed legacy task probe tests. So, there are both of legacy task probe tests and general task probe tests.\n3. Renamed `execute_legacuy_task` option to `dequeue_from_legacy_queue`.\n4. Now, default value of `dequeue_from_legacy_queue` is true. This default value is not depends on file name (\u0027object-expirer.conf\u0027 or \u0027object-server.conf\u0027). Because it is more simple.\n5. Now, object-expirer general task queue mode works with `servers_per_port` option.\n\nThese changes are depends on discussion with kota.\n\nNow, I\u0027m splitting this patch. So this patch is NOT ready for review.","accounts_in_message":[],"_revision_number":29},{"id":"daae680910b8322baf6bdb97e566c637d8bfe6ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 05:50:17.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/29/check/openstack-tox-docs/d248ab0/html/ : SUCCESS in 3m 31s\n- swift-tox-py27 http://logs.openstack.org/89/517389/29/check/swift-tox-py27/c782a1a/ : FAILURE in 10m 49s\n- swift-tox-py35 http://logs.openstack.org/89/517389/29/check/swift-tox-py35/b7288b9/ : SUCCESS in 4m 12s\n- swift-tox-py36 http://logs.openstack.org/89/517389/29/check/swift-tox-py36/ba9afec/ : SUCCESS in 4m 10s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/29/check/swift-tox-func/a094c16/ : SUCCESS in 7m 08s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/29/check/swift-tox-func-encryption/8f94250/ : SUCCESS in 9m 44s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/29/check/swift-tox-func-domain-remap-staticweb/92b179a/ : SUCCESS in 9m 12s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/29/check/swift-tox-func-ec/c3206ef/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/29/check/swift-tox-func-s3api/07a7547/ : SUCCESS in 11m 04s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/29/check/swift-probetests-centos-7/4e6d6b0/ : FAILURE in 28m 03s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/29/check/swift-dsvm-functional/5d916a3/ : SUCCESS in 21m 05s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/29/check/swift-dsvm-functional-ipv6/af6bf30/ : SUCCESS in 20m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/517389/29/check/openstack-tox-lower-constraints/b351324/ : SUCCESS in 9m 21s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/29/check/openstack-tox-pep8/03d9e7e/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":29},{"id":"12d96c04bd903c7708739d3c66786b34305b60a7","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 06:27:46.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"2286e81d8fbb27293eaedc17c008442cfec567e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 07:36:30.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/30/check/openstack-tox-docs/c42378e/html/ : SUCCESS in 4m 10s\n- swift-tox-py27 http://logs.openstack.org/89/517389/30/check/swift-tox-py27/d9286eb/ : SUCCESS in 11m 21s\n- swift-tox-py35 http://logs.openstack.org/89/517389/30/check/swift-tox-py35/7740c7a/ : SUCCESS in 4m 58s\n- swift-tox-py36 http://logs.openstack.org/89/517389/30/check/swift-tox-py36/63c94a8/ : SUCCESS in 5m 50s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/30/check/swift-tox-func/f6a49ce/ : SUCCESS in 9m 35s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/30/check/swift-tox-func-encryption/0b71729/ : SUCCESS in 9m 52s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/30/check/swift-tox-func-domain-remap-staticweb/4d39e37/ : SUCCESS in 9m 37s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/30/check/swift-tox-func-ec/eb5b59c/ : SUCCESS in 9m 56s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/30/check/swift-tox-func-s3api/fa10803/ : SUCCESS in 10m 33s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/30/check/swift-probetests-centos-7/2791f3a/ : SUCCESS in 31m 23s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/30/check/swift-dsvm-functional/5052110/ : SUCCESS in 28m 13s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/30/check/swift-dsvm-functional-ipv6/4be9dda/ : SUCCESS in 21m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/517389/30/check/openstack-tox-lower-constraints/6e375f8/ : SUCCESS in 8m 22s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/30/check/openstack-tox-pep8/2b3c452/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":30},{"id":"f3434fdd29d429c5ed651be1e8f63f8b5e7d4d1c","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 08:36:47.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"c6fe528893f567b977fa4d7da67cd7f2f22b444b","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 08:38:42.000000000","message":"Patch Set 31:\n\nFollowing patch is split from this patch.\nhttps://review.openstack.org/#/c/601950/","accounts_in_message":[],"_revision_number":31},{"id":"6459e06391d4c2376983db253b2e725f8a349f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 09:19:21.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/31/check/openstack-tox-docs/a00f4c7/html/ : SUCCESS in 3m 52s\n- swift-tox-py27 http://logs.openstack.org/89/517389/31/check/swift-tox-py27/b6e3ad2/ : SUCCESS in 9m 56s\n- swift-tox-py35 http://logs.openstack.org/89/517389/31/check/swift-tox-py35/7d852b4/ : SUCCESS in 4m 38s\n- swift-tox-py36 http://logs.openstack.org/89/517389/31/check/swift-tox-py36/4c677e2/ : SUCCESS in 4m 49s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/31/check/swift-tox-func/74337ec/ : SUCCESS in 8m 21s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/31/check/swift-tox-func-encryption/cac9e6c/ : SUCCESS in 9m 37s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/31/check/swift-tox-func-domain-remap-staticweb/88678fe/ : SUCCESS in 8m 37s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/31/check/swift-tox-func-ec/6a7b003/ : SUCCESS in 9m 15s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/31/check/swift-tox-func-s3api/84d0a10/ : SUCCESS in 12m 12s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/31/check/swift-probetests-centos-7/e140844/ : SUCCESS in 41m 12s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/31/check/swift-dsvm-functional/0a10c1c/ : SUCCESS in 26m 44s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/31/check/swift-dsvm-functional-ipv6/4f0d92d/ : SUCCESS in 26m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/517389/31/check/openstack-tox-lower-constraints/2702b07/ : SUCCESS in 8m 11s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/31/check/openstack-tox-pep8/b7ee10a/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":31},{"id":"0f57b97aca101432509fc0ba2aac9123fa77086c","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 16:15:44.000000000","message":"Patch Set 32: Patch Set 31 was rebased","accounts_in_message":[],"_revision_number":32},{"id":"e792de70987196f00b71d4bd9397c24bff9b979e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 17:57:20.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/32/check/openstack-tox-docs/aad5916/html/ : SUCCESS in 4m 17s\n- swift-tox-py27 http://logs.openstack.org/89/517389/32/check/swift-tox-py27/fa9b5c9/ : SUCCESS in 10m 16s\n- swift-tox-py35 http://logs.openstack.org/89/517389/32/check/swift-tox-py35/3027a09/ : SUCCESS in 5m 11s\n- swift-tox-py36 http://logs.openstack.org/89/517389/32/check/swift-tox-py36/a7f467c/ : SUCCESS in 4m 46s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/32/check/swift-tox-func/e87f803/ : SUCCESS in 8m 12s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/32/check/swift-tox-func-encryption/60883c3/ : SUCCESS in 10m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/32/check/swift-tox-func-domain-remap-staticweb/b537159/ : SUCCESS in 9m 11s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/32/check/swift-tox-func-ec/6740964/ : SUCCESS in 10m 20s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/32/check/swift-tox-func-s3api/e357bbf/ : SUCCESS in 10m 27s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/32/check/swift-probetests-centos-7/2ee1a4d/ : SUCCESS in 26m 20s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/32/check/swift-dsvm-functional/5c979b8/ : SUCCESS in 22m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/32/check/swift-dsvm-functional-ipv6/1321ab3/ : SUCCESS in 25m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/517389/32/check/openstack-tox-lower-constraints/11cab04/ : SUCCESS in 7m 58s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/32/check/openstack-tox-pep8/a361ecb/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":32},{"id":"b7157d8855b7f98289261db4c398541fb9891f05","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-09-12 19:48:22.000000000","message":"Patch Set 32:\n\n(1 comment)\n\nsmall comment.","accounts_in_message":[],"_revision_number":32},{"id":"041027f764edd15e707cb1a771abee972d393a5e","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-09-13 16:38:36.000000000","message":"Patch Set 32: Code-Review-1\n\n(2 comments)\n\nIssue:\nThe object-expirer can run in only one mode: legacy or generic. On a cluster where object-expirer are deployed on object-servers, an object-expirer would not be consuming generic queue if it is designed to be a legacy queue consumer.\n\nSolution (as discussed at the PTG Denver 2018):\nThe object-expirer must be able to run in one of the 3 following mode:\n- legacy\n- generic\n- legacy and generic\n\nThe condition to choose the mode is:\n- on non object-servers, run in legacy mode, get configuration from object-expirer.conf\n- on object-server:\n  - if there is only object-expirer.conf, run in legacy mode\n  - if there is a section [object-expirer] in object-server.conf, run in legacy+generic mode\n\nThe upgrade process for operators will be:\n1/ Upgrade cluster\n   \u003d\u003e expirer tasks are created in legacy queue\n   \u003d\u003e object-expirer consume tasks from the legacy queue\n\n2/ Deploy object-expirer on all object-servers (with config):\n   \u003d\u003e expirer tasks are created in legacy queue\n   \u003d\u003e object-expirer consume tasks from legacy and generic queues\n\n3/ Update proxy/object config to create tasks in generic queue\n   \u003d\u003e expirer tasks are created in generic queue\n   \u003d\u003e object-expirer consume tasks from legacy and generic queues\n\nTo determine if the server is an object-server:\n - look for an object ring\n - check that myips are in one of the object rings","accounts_in_message":[],"_revision_number":32},{"id":"746d8f68c420ac4164702e6692cec1b4558a2775","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-13 18:10:32.000000000","message":"Patch Set 32:\n\n(3 comments)","accounts_in_message":[],"_revision_number":32},{"id":"2b74f0a798fd0a9888657d724cdcec90005b85fb","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-13 21:26:43.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"0559bfe5ed65a6afad698782c0ed5ece780c9e25","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-13 21:28:42.000000000","message":"Patch Set 33:\n\n(5 comments)\n\nFixed for some comments and fixed probe tests because general mode probe test didn\u0027t work.\n\nRomain\u0027s upgrade process is not reflected yet. But I will update for the process.","accounts_in_message":[],"_revision_number":33},{"id":"973f4cd9a8eb8865d93cbd4adda51133ba327077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 22:32:02.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/33/check/openstack-tox-docs/68ba7f7/html/ : SUCCESS in 3m 57s\n- swift-tox-py27 http://logs.openstack.org/89/517389/33/check/swift-tox-py27/32e3511/ : SUCCESS in 11m 25s\n- swift-tox-py35 http://logs.openstack.org/89/517389/33/check/swift-tox-py35/1c4cea0/ : SUCCESS in 4m 33s\n- swift-tox-py36 http://logs.openstack.org/89/517389/33/check/swift-tox-py36/07bdda4/ : SUCCESS in 4m 46s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/33/check/swift-tox-func/a552a2b/ : SUCCESS in 9m 27s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/33/check/swift-tox-func-encryption/689c3b2/ : SUCCESS in 9m 47s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/33/check/swift-tox-func-domain-remap-staticweb/955b579/ : SUCCESS in 9m 23s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/33/check/swift-tox-func-ec/9c580ef/ : SUCCESS in 10m 15s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/33/check/swift-tox-func-s3api/b5971e0/ : SUCCESS in 11m 16s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/33/check/swift-probetests-centos-7/423badf/ : FAILURE in 32m 52s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/33/check/swift-dsvm-functional/29042fc/ : SUCCESS in 22m 51s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/33/check/swift-dsvm-functional-ipv6/12e13ca/ : SUCCESS in 26m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/517389/33/check/openstack-tox-lower-constraints/4431f44/ : SUCCESS in 7m 42s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/33/check/openstack-tox-pep8/8a70a9f/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":33},{"id":"14ce75ce9001196fbad4e6b5a08a64f944dba602","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-13 23:08:55.000000000","message":"Patch Set 33: Code-Review+1\n\nThis looks fine but the probe tests failure is worrying. I had some legitimate failures there that neither functests nor unit tests caught.","accounts_in_message":[],"_revision_number":33},{"id":"2757928e49b6a7717166ca1133866090c7bd3c07","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-14 03:57:18.000000000","message":"Patch Set 33:\n\nI wonder why probe tests fail. In my SAIO, all tests succeed.\nSo I try to recheck to confirm whether the failure is reproductive on gerrit or not.\n\nNOTE: This patch will be updated to reflect discussion with romain.","accounts_in_message":[],"_revision_number":33},{"id":"a21b3cff86a3637e9ff973fa980ac21037a1299c","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-14 03:57:39.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"70ed3a2c29778ea3ad09a7bd2023b5478955161f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-14 04:35:49.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/33/check/openstack-tox-docs/98a563b/html/ : SUCCESS in 3m 35s\n- swift-tox-py27 http://logs.openstack.org/89/517389/33/check/swift-tox-py27/507cf53/ : SUCCESS in 9m 33s\n- swift-tox-py35 http://logs.openstack.org/89/517389/33/check/swift-tox-py35/8f766df/ : SUCCESS in 4m 35s\n- swift-tox-py36 http://logs.openstack.org/89/517389/33/check/swift-tox-py36/b471198/ : SUCCESS in 4m 43s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/33/check/swift-tox-func/599b61d/ : SUCCESS in 8m 14s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/33/check/swift-tox-func-encryption/c0aca59/ : SUCCESS in 8m 27s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/33/check/swift-tox-func-domain-remap-staticweb/1eacaff/ : SUCCESS in 7m 49s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/33/check/swift-tox-func-ec/c993d31/ : SUCCESS in 9m 37s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/33/check/swift-tox-func-s3api/ac030e9/ : SUCCESS in 10m 39s\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/33/check/swift-probetests-centos-7/4d08d35/ : FAILURE in 37m 07s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/33/check/swift-dsvm-functional/e098f43/ : SUCCESS in 26m 26s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/33/check/swift-dsvm-functional-ipv6/22b344c/ : SUCCESS in 27m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/517389/33/check/openstack-tox-lower-constraints/f262b5d/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/33/check/openstack-tox-pep8/f295207/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":33},{"id":"7a89c21b965a9f4558ff6228930843a90e4746b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-09-14 04:48:20.000000000","message":"Patch Set 33:\n\nYou might need to update the configs under doc/saio/swift -- we start from there when running probe tests in the gate.","accounts_in_message":[],"_revision_number":33},{"id":"9e54c0215bca8c8bac8292f95f437c5c91df4e08","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-11-28 08:36:11.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"48f599673ee7d0a5e85b07334c1598ffa434f100","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-11-28 08:39:08.000000000","message":"Patch Set 34:\n\nRebased to check gate test behavior with the latest code.","accounts_in_message":[],"_revision_number":34},{"id":"d003a00784678b340ed8f4633b385f2a467b639a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 09:54:27.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/34/check/openstack-tox-docs/6bb6df7/html/ : SUCCESS in 5m 17s\n- swift-tox-py27 http://logs.openstack.org/89/517389/34/check/swift-tox-py27/65843f9/ : FAILURE in 11m 39s\n- swift-tox-py35 http://logs.openstack.org/89/517389/34/check/swift-tox-py35/64cb0ac/ : SUCCESS in 9m 06s\n- swift-tox-py36 http://logs.openstack.org/89/517389/34/check/swift-tox-py36/4225cc7/ : SUCCESS in 7m 32s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/34/check/swift-tox-func/0c93d1d/ : FAILURE in 20m 17s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/34/check/swift-tox-func-encryption/3969036/ : FAILURE in 11m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/34/check/swift-tox-func-domain-remap-staticweb/211cc70/ : FAILURE in 11m 09s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/34/check/swift-tox-func-ec/2a7b592/ : FAILURE in 13m 03s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/34/check/swift-tox-func-s3api/52a031e/ : FAILURE in 13m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/34/check/swift-tox-func-s3api-ceph-s3tests-tempauth/68311b8/ : SUCCESS in 13m 51s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/34/check/swift-probetests-centos-7/4dd8b01/ : FAILURE in 29m 41s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/34/check/swift-dsvm-functional/dd012d7/ : FAILURE in 20m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/34/check/swift-dsvm-functional-ipv6/52c7134/ : FAILURE in 22m 37s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/34/check/swift-tox-lower-constraints/04377ad/ : FAILURE in 13m 08s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/34/check/openstack-tox-pep8/674219a/ : FAILURE in 5m 17s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/34/check/swift-multinode-rolling-upgrade/0e43d5b/ : FAILURE in 16m 51s","accounts_in_message":[],"_revision_number":34},{"id":"92f146e50f5342d308f1fd9588542468f6ce33ab","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 05:32:48.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"938472a58a988775ea899aa15bd942fa9645bc67","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 05:34:59.000000000","message":"Patch Set 35:\n\nRebased","accounts_in_message":[],"_revision_number":35},{"id":"d374dc226a4a094abe6e08810617fa1bbeb96f0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 06:13:25.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/35/check/openstack-tox-docs/63fd524/html/ : SUCCESS in 4m 01s\n- swift-tox-py27 http://logs.openstack.org/89/517389/35/check/swift-tox-py27/ed4ed85/ : FAILURE in 11m 25s\n- swift-tox-py35 http://logs.openstack.org/89/517389/35/check/swift-tox-py35/0eb56fb/ : SUCCESS in 5m 37s\n- swift-tox-py36 http://logs.openstack.org/89/517389/35/check/swift-tox-py36/6cd2186/ : SUCCESS in 5m 18s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/35/check/swift-tox-func/aeaa030/ : SUCCESS in 7m 57s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/35/check/swift-tox-func-encryption/3450013/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/35/check/swift-tox-func-domain-remap-staticweb/52eb720/ : SUCCESS in 7m 41s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/35/check/swift-tox-func-ec/7975ee9/ : SUCCESS in 9m 40s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/35/check/swift-tox-func-s3api/206e8c3/ : SUCCESS in 9m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/35/check/swift-tox-func-s3api-ceph-s3tests-tempauth/379d04a/ : SUCCESS in 12m 23s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/35/check/swift-probetests-centos-7/f8fe70d/ : SUCCESS in 35m 21s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/35/check/swift-dsvm-functional/a494dad/ : SUCCESS in 21m 38s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/35/check/swift-dsvm-functional-ipv6/5fdbb13/ : SUCCESS in 25m 36s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/35/check/swift-tox-lower-constraints/4e7f95d/ : FAILURE in 12m 12s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/35/check/openstack-tox-pep8/22baf08/ : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/35/check/swift-multinode-rolling-upgrade/a84c957/ : SUCCESS in 15m 16s","accounts_in_message":[],"_revision_number":35},{"id":"79d4833a65f22ebeea08444f4a4fa6692e3c04b1","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 06:30:18.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"842cc0ddc0751bbb8b0d968bb0f7941c30b86127","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 06:31:07.000000000","message":"Patch Set 36:\n\nFixed variable name.","accounts_in_message":[],"_revision_number":36},{"id":"2eb5bb4e48b41669ed39caaea0e7cf43344fb86e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 06:59:31.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/36/check/openstack-tox-docs/26112bb/html/ : SUCCESS in 3m 48s\n- swift-tox-py27 http://logs.openstack.org/89/517389/36/check/swift-tox-py27/e6d8ae9/ : FAILURE in 9m 53s\n- swift-tox-py35 http://logs.openstack.org/89/517389/36/check/swift-tox-py35/10a81a4/ : SUCCESS in 5m 15s\n- swift-tox-py36 http://logs.openstack.org/89/517389/36/check/swift-tox-py36/602d1f5/ : SUCCESS in 4m 54s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/36/check/swift-tox-func/7b0c2da/ : SUCCESS in 7m 44s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/36/check/swift-tox-func-encryption/d78a394/ : SUCCESS in 7m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/36/check/swift-tox-func-domain-remap-staticweb/e23dccb/ : SUCCESS in 8m 07s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/36/check/swift-tox-func-ec/acc105c/ : SUCCESS in 8m 46s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/36/check/swift-tox-func-s3api/09b04d7/ : SUCCESS in 9m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/36/check/swift-tox-func-s3api-ceph-s3tests-tempauth/054b12c/ : SUCCESS in 12m 28s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/36/check/swift-probetests-centos-7/7d0ba9e/ : FAILURE in 28m 21s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/36/check/swift-dsvm-functional/048cf96/ : SUCCESS in 23m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/36/check/swift-dsvm-functional-ipv6/5f2c41d/ : SUCCESS in 23m 53s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/36/check/swift-tox-lower-constraints/eacadb1/ : FAILURE in 10m 05s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/36/check/openstack-tox-pep8/094f477/ : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/36/check/swift-multinode-rolling-upgrade/cfacda3/ : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":36},{"id":"1b3b7bece1bcaebcc982bf7050f6ab23d75b3460","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 07:44:22.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"c04f5c6d8d3d8d1c266df33cff1e0a849ebabc49","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 07:54:08.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"0029e19068938dbf92d07e4f05358be0de1cea5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 08:34:39.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/38/check/openstack-tox-docs/d172f32/html/ : SUCCESS in 3m 54s\n- swift-tox-py27 http://logs.openstack.org/89/517389/38/check/swift-tox-py27/1115dac/ : SUCCESS in 10m 06s\n- swift-tox-py35 http://logs.openstack.org/89/517389/38/check/swift-tox-py35/1e07d75/ : SUCCESS in 5m 10s\n- swift-tox-py36 http://logs.openstack.org/89/517389/38/check/swift-tox-py36/2591656/ : SUCCESS in 5m 09s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/38/check/swift-tox-func/c3a8cce/ : SUCCESS in 7m 27s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/38/check/swift-tox-func-encryption/5d26e65/ : SUCCESS in 8m 40s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/38/check/swift-tox-func-domain-remap-staticweb/0702dbd/ : SUCCESS in 8m 23s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/38/check/swift-tox-func-ec/2303c2a/ : SUCCESS in 8m 45s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/38/check/swift-tox-func-s3api/cac33a5/ : SUCCESS in 9m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/38/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fd3f40f/ : SUCCESS in 11m 04s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/38/check/swift-probetests-centos-7/3e8c6b8/ : FAILURE in 35m 45s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/38/check/swift-dsvm-functional/22af607/ : SUCCESS in 24m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/38/check/swift-dsvm-functional-ipv6/35b52f7/ : SUCCESS in 23m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/38/check/swift-tox-lower-constraints/35fd7ec/ : SUCCESS in 10m 23s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/38/check/openstack-tox-pep8/16cbc25/ : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/38/check/swift-multinode-rolling-upgrade/ea50825/ : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":38},{"id":"a07e973deb0ee8378c58b2c5721136f6f67ab5f1","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 08:40:36.000000000","message":"Patch Set 38:\n\nrecheck","accounts_in_message":[],"_revision_number":38},{"id":"1a949eca3cbd957b18273a0f74793193afd75265","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 09:21:19.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/38/check/openstack-tox-docs/5a495c7/html/ : SUCCESS in 3m 43s\n- swift-tox-py27 http://logs.openstack.org/89/517389/38/check/swift-tox-py27/10daa07/ : SUCCESS in 13m 58s\n- swift-tox-py35 http://logs.openstack.org/89/517389/38/check/swift-tox-py35/991d63e/ : SUCCESS in 12m 36s\n- swift-tox-py36 http://logs.openstack.org/89/517389/38/check/swift-tox-py36/c831ad4/ : SUCCESS in 5m 31s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/38/check/swift-tox-func/51c34f6/ : SUCCESS in 12m 06s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/38/check/swift-tox-func-encryption/c46bf81/ : SUCCESS in 8m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/38/check/swift-tox-func-domain-remap-staticweb/2a6d654/ : FAILURE in 8m 32s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/38/check/swift-tox-func-ec/923632e/ : FAILURE in 8m 56s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/38/check/swift-tox-func-s3api/d43edc4/ : FAILURE in 10m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/38/check/swift-tox-func-s3api-ceph-s3tests-tempauth/bcb02a8/ : SUCCESS in 12m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/38/check/swift-probetests-centos-7/74d21aa/ : FAILURE in 36m 01s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/38/check/swift-dsvm-functional/e9c8688/ : SUCCESS in 21m 15s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/38/check/swift-dsvm-functional-ipv6/e4d65d6/ : SUCCESS in 25m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/38/check/swift-tox-lower-constraints/3a6b208/ : SUCCESS in 9m 43s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/38/check/openstack-tox-pep8/4306786/ : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/38/check/swift-multinode-rolling-upgrade/04f3a7e/ : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":38},{"id":"e5267948833ca020eea47eea9942b7d6075b9154","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 01:41:14.000000000","message":"Patch Set 38:\n\nrecheck","accounts_in_message":[],"_revision_number":38},{"id":"fb0600e4affdc940356888578ca1a5c16deb0f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 02:20:14.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/38/check/openstack-tox-docs/7c6066e/html/ : SUCCESS in 3m 34s\n- swift-tox-py27 http://logs.openstack.org/89/517389/38/check/swift-tox-py27/1e9f090/ : SUCCESS in 9m 55s\n- swift-tox-py35 http://logs.openstack.org/89/517389/38/check/swift-tox-py35/f9d1be0/ : SUCCESS in 6m 04s\n- swift-tox-py36 http://logs.openstack.org/89/517389/38/check/swift-tox-py36/10279db/ : SUCCESS in 4m 35s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/38/check/swift-tox-func/2d265d1/ : SUCCESS in 8m 07s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/38/check/swift-tox-func-encryption/2ab8469/ : SUCCESS in 7m 41s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/38/check/swift-tox-func-domain-remap-staticweb/d04136e/ : SUCCESS in 8m 54s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/38/check/swift-tox-func-ec/b338edb/ : SUCCESS in 9m 11s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/38/check/swift-tox-func-s3api/293d5b7/ : SUCCESS in 8m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/38/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e88cac9/ : SUCCESS in 10m 33s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/38/check/swift-probetests-centos-7/9a20644/ : FAILURE in 34m 23s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/38/check/swift-dsvm-functional/65b0c1a/ : SUCCESS in 22m 54s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/38/check/swift-dsvm-functional-ipv6/ed07b39/ : SUCCESS in 29m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/38/check/swift-tox-lower-constraints/333ec10/ : SUCCESS in 9m 29s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/38/check/openstack-tox-pep8/be29602/ : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/38/check/swift-multinode-rolling-upgrade/234ec76/ : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":38},{"id":"9c50d8d9b58f73c52d36283e9979609aa09079d9","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 02:30:06.000000000","message":"Patch Set 38: Workflow-1\n\nUmm... some probe tests\u0027 results of TestObjectExpirerGeneralMode are unstable at gate.\n\nFirst test: test_expirer_delete_returns_outdated_404 and test_expirer_object_split_brain failed\nSecond test: test_expirer_delete_returns_outdated_404 failed\nThird test: test_expirer_delete_returns_outdated_412 and test_expirer_object_split_brain failed\n\nThese tests\u0027 results are always \"ok\" in my environment.\n\nI\u0027m investigating behavior of these tests.","accounts_in_message":[],"_revision_number":38},{"id":"a5d761f3d5491ec3837cf2f4205c93aa65eb1d9e","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 11:47:27.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"613022c8809f3ab950437963c189a89a7d7d75e9","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 11:55:47.000000000","message":"Patch Set 39:\n\nThe probe tests\u0027 failures came from lack of object-expirer for general task queue.\nAt patch set 38, only one object-expirer for 127.0.0.1 ran.\nHowever there were 4 object-servers for 127.0.0.1, 127.0.0.2, 127.0.0.3, and 127.0.0.4, so quarter of expiring tasks are executed.\nAt patch set 39, this bug is fixed.","accounts_in_message":[],"_revision_number":39},{"id":"173964bb4486436a98084ac4ab624f951f5c53df","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 12:06:17.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"9143bd34a25798d5e142ad6320f267fb7d66dd69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 12:47:29.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/40/check/openstack-tox-docs/f5ec9b8/html/ : SUCCESS in 3m 52s\n- swift-tox-py27 http://logs.openstack.org/89/517389/40/check/swift-tox-py27/34af31b/ : SUCCESS in 10m 25s\n- swift-tox-py35 http://logs.openstack.org/89/517389/40/check/swift-tox-py35/8e7668d/ : SUCCESS in 5m 11s\n- swift-tox-py36 http://logs.openstack.org/89/517389/40/check/swift-tox-py36/c042a25/ : SUCCESS in 9m 16s (non-voting)\n- swift-tox-func http://logs.openstack.org/89/517389/40/check/swift-tox-func/a7dc072/ : SUCCESS in 8m 51s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/40/check/swift-tox-func-encryption/ba4b846/ : SUCCESS in 8m 04s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/40/check/swift-tox-func-domain-remap-staticweb/c551d9a/ : SUCCESS in 10m 05s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/40/check/swift-tox-func-ec/4337ac3/ : SUCCESS in 8m 44s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/40/check/swift-tox-func-s3api/e727333/ : SUCCESS in 8m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/40/check/swift-tox-func-s3api-ceph-s3tests-tempauth/df9e8da/ : SUCCESS in 14m 54s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/40/check/swift-probetests-centos-7/1f00761/ : SUCCESS in 39m 53s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/40/check/swift-dsvm-functional/7264555/ : SUCCESS in 21m 20s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/40/check/swift-dsvm-functional-ipv6/99b7e63/ : SUCCESS in 25m 36s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/40/check/swift-tox-lower-constraints/7b8929e/ : SUCCESS in 12m 18s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/40/check/openstack-tox-pep8/fa7eb0a/ : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/40/check/swift-multinode-rolling-upgrade/a8d1871/ : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":40},{"id":"499f937b078e0e23200f887e3973c8bfea2596b0","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-02 22:03:51.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"f9bb2ecdcb396695f72a1ff381a6aeaa33f89699","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:49:21.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/41/check/openstack-tox-docs/ac7d1a9/html/ : SUCCESS in 4m 34s\n- swift-tox-py27 http://logs.openstack.org/89/517389/41/check/swift-tox-py27/5bf63e1/ : FAILURE in 17m 24s\n- swift-tox-py35 http://logs.openstack.org/89/517389/41/check/swift-tox-py35/64bc457/ : FAILURE in 8m 22s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/89/517389/41/check/swift-tox-py36/9591ad2/ : FAILURE in 8m 24s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/89/517389/41/check/swift-tox-py37/2cf4fb9/ : FAILURE in 13m 13s\n- swift-tox-func http://logs.openstack.org/89/517389/41/check/swift-tox-func/ce32171/ : SUCCESS in 8m 24s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/41/check/swift-tox-func-encryption/420fd67/ : SUCCESS in 15m 04s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/41/check/swift-tox-func-domain-remap-staticweb/0b38c9b/ : SUCCESS in 8m 45s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/41/check/swift-tox-func-ec/7021700/ : SUCCESS in 10m 16s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/41/check/swift-tox-func-s3api/6731e24/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/41/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9307778/ : TIMED_OUT in 31m 21s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/41/check/swift-probetests-centos-7/ba62d0a/ : SUCCESS in 30m 16s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/41/check/swift-dsvm-functional/f30b7ff/ : SUCCESS in 32m 07s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/41/check/swift-dsvm-functional-ipv6/40e6796/ : SUCCESS in 29m 24s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/41/check/swift-tox-lower-constraints/fb15ac7/ : FAILURE in 11m 05s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/41/check/openstack-tox-pep8/9cdf193/ : FAILURE in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/41/check/swift-multinode-rolling-upgrade/eff1d08/ : SUCCESS in 16m 13s","accounts_in_message":[],"_revision_number":41},{"id":"61abf84d6e787ddb249b42f4eb2be61a9b14f013","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-02 22:55:08.000000000","message":"Patch Set 42: Published edit on patch set 41.","accounts_in_message":[],"_revision_number":42},{"id":"d4e767a0464974c3ecf1b6210ef2feca35694092","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 23:45:10.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/42/check/openstack-tox-docs/c63f44a/html/ : SUCCESS in 4m 09s\n- swift-tox-py27 http://logs.openstack.org/89/517389/42/check/swift-tox-py27/0022c25/ : SUCCESS in 12m 48s\n- swift-tox-py35 http://logs.openstack.org/89/517389/42/check/swift-tox-py35/29bea34/ : FAILURE in 9m 58s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/89/517389/42/check/swift-tox-py36/c4b80b5/ : FAILURE in 9m 30s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/89/517389/42/check/swift-tox-py37/5e08ac7/ : FAILURE in 9m 49s\n- swift-tox-func http://logs.openstack.org/89/517389/42/check/swift-tox-func/0e87c4c/ : SUCCESS in 9m 55s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/42/check/swift-tox-func-encryption/a93cc57/ : SUCCESS in 13m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/42/check/swift-tox-func-domain-remap-staticweb/1ae3ae2/ : SUCCESS in 10m 41s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/42/check/swift-tox-func-ec/30ad794/ : SUCCESS in 11m 39s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/42/check/swift-tox-func-s3api/4e5f5b4/ : SUCCESS in 11m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/42/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a658db0/ : SUCCESS in 11m 27s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/42/check/swift-probetests-centos-7/5d94427/ : SUCCESS in 45m 49s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/42/check/swift-dsvm-functional/f45ea52/ : SUCCESS in 25m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/42/check/swift-dsvm-functional-ipv6/fccb52e/ : SUCCESS in 26m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/42/check/swift-tox-lower-constraints/be0f522/ : SUCCESS in 11m 59s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/42/check/openstack-tox-pep8/0f67bf1/ : SUCCESS in 7m 06s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/42/check/swift-multinode-rolling-upgrade/403abce/ : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":42},{"id":"012ad4c6317a3ecc9eb3b4e18c94771d7165e5cb","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-03 15:19:41.000000000","message":"Patch Set 43: Published edit on patch set 42.","accounts_in_message":[],"_revision_number":43},{"id":"77d6623c5a96d3b2b74b73b97eeb3fa34b0e5ff5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-03 15:40:30.000000000","message":"Patch Set 42:\n\n(32 comments)\n\nOk, I think i\u0027ve got this all loaded into my head...\n\nThis is my biggest problem:\n\nTHIS CHANGE:\n\n\tvagrant@saio:~$ time nosetests swift/test/probe/test_object_expirer.py \n\t..........\n\t----------------------------------------------------------------------\n\tRan 10 tests in 1829.204s\n\n\tOK\n\n\treal\t30m30.007s\n\tuser\t9m55.756s\n\tsys\t5m34.880s\n\nMASTER:\n\n\tvagrant@saio:~$ time nosetests swift/test/probe/test_object_expirer.py \n\t.....\n\t----------------------------------------------------------------------\n\tRan 5 tests in 161.084s\n\n\tOK\n\n\treal\t2m41.843s\n\tuser\t0m50.648s\n\tsys\t1m2.408s\n\nand I think it relates to my trepidation about the task queue naming, which I still need to think about.","accounts_in_message":[],"_revision_number":42},{"id":"6429153ac540bb7de46db842974ea5a3ec8b7b1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:04:44.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/43/check/openstack-tox-docs/14dbade/html/ : SUCCESS in 4m 25s\n- swift-tox-py27 http://logs.openstack.org/89/517389/43/check/swift-tox-py27/f66f7dc/ : SUCCESS in 9m 58s\n- swift-tox-py35 http://logs.openstack.org/89/517389/43/check/swift-tox-py35/eb114d6/ : SUCCESS in 9m 19s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/89/517389/43/check/swift-tox-py36/18c6e24/ : SUCCESS in 7m 38s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/89/517389/43/check/swift-tox-py37/647b15c/ : SUCCESS in 8m 04s\n- swift-tox-func http://logs.openstack.org/89/517389/43/check/swift-tox-func/ab34d50/ : SUCCESS in 9m 25s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/43/check/swift-tox-func-encryption/ee77ca8/ : SUCCESS in 8m 25s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/43/check/swift-tox-func-domain-remap-staticweb/9311299/ : SUCCESS in 8m 25s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/43/check/swift-tox-func-ec/0a10a03/ : SUCCESS in 10m 20s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/43/check/swift-tox-func-s3api/0fae300/ : SUCCESS in 10m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/43/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0029412/ : SUCCESS in 11m 26s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/43/check/swift-probetests-centos-7/eeb4a9f/ : SUCCESS in 39m 29s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/43/check/swift-dsvm-functional/8509e3e/ : SUCCESS in 25m 25s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/43/check/swift-dsvm-functional-ipv6/6d7f2d4/ : SUCCESS in 25m 35s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/43/check/swift-tox-lower-constraints/82f3e8f/ : SUCCESS in 13m 10s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/43/check/openstack-tox-pep8/ddb662d/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/43/check/swift-multinode-rolling-upgrade/700ca1a/ : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":43},{"id":"159a1128325f25c0724b6d6f0485e161102ef890","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-03 22:58:05.000000000","message":"Patch Set 43: Code-Review-1\n\n(1 comment)\n\nRecap of discussion at PTG:\n\nThere dequeue_from_legacy_task_queue option should be *additive* we  need three modes\n\n1) you can run ONLY from legacy using object-expirer.conf\n2) you will BY DEFAULT run ONLY from general-task-q using [object-expirer] in object-server.conf\n3) you MAY enable running from BOTH legacy and general-task-q using dequeue_from_legacy OR processes/process in the [object-expirer] in object-server.conf\n\nWith a modest part power and a couple of rings there\u0027s a lot of requests to list the general task queue, many of them 404 on a sparse development environment.  This can be mitigated by reducing part_power to 6, or changing the module index so that ONLY the replica_index \u003d\u003d 0 node lists the container, and does ALL the rows... however a different queue layout would be more efficient for some operations:\n\nthe general task queue should look like:\n\n.task-\u003cring\u003e-\u003ctype\u003e/\u003cpartition\u003e\n\nAn instance of the expirer might just list it\u0027s container partition, but if it gets a few 404s, it could with a small random sleep list it\u0027s account and \"skip ahead\"\n\nAlso this layout would have more reasonable cardinality in the containers, with probably between 10-100K objects per container instead of only 1\n\nThere\u0027s a few other things/comments that would be nice to cleanup, but these are the larger changes that should be addressed first.","accounts_in_message":[],"_revision_number":43},{"id":"4db95580b0ffba9f689fb619cf47cb8d73f88a5c","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-04 10:38:45.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"bb8275232b3e44b3515f8111f3317d6d72b1038a","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-04 10:44:37.000000000","message":"Patch Set 44:\n\n(10 comments)\n\nUpdate for some clay\u0027s comments.\nThis update doesn\u0027t include \"task queue three modes\" and \".task-\u003cring\u003e-\u003ctype\u003e/\u003cpartition\u003e style queue\".","accounts_in_message":[],"_revision_number":44},{"id":"c50453dbd011f77ac9645d09d6d15db481c29afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 11:23:45.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/44/check/openstack-tox-docs/b59db10/ : FAILURE in 3m 39s\n- swift-tox-py27 http://logs.openstack.org/89/517389/44/check/swift-tox-py27/3852f36/ : SUCCESS in 10m 39s\n- swift-tox-py35 http://logs.openstack.org/89/517389/44/check/swift-tox-py35/bd7a892/ : SUCCESS in 10m 03s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/89/517389/44/check/swift-tox-py36/da06568/ : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/89/517389/44/check/swift-tox-py37/909cd69/ : SUCCESS in 9m 45s\n- swift-tox-func http://logs.openstack.org/89/517389/44/check/swift-tox-func/341cf87/ : FAILURE in 3m 16s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/44/check/swift-tox-func-encryption/ee4d461/ : FAILURE in 3m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/44/check/swift-tox-func-domain-remap-staticweb/8e644f1/ : FAILURE in 3m 26s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/44/check/swift-tox-func-ec/61490cb/ : FAILURE in 3m 16s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/44/check/swift-tox-func-s3api/f4e7797/ : FAILURE in 3m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/44/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ae17ae2/ : SUCCESS in 13m 55s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/44/check/swift-probetests-centos-7/54fa643/ : SUCCESS in 40m 43s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/44/check/swift-dsvm-functional/b10c36d/ : FAILURE in 13m 32s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/44/check/swift-dsvm-functional-ipv6/544a35c/ : FAILURE in 13m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/44/check/swift-tox-lower-constraints/652616c/ : SUCCESS in 13m 55s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/44/check/openstack-tox-pep8/c406e1b/ : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/44/check/swift-multinode-rolling-upgrade/8be01be/ : FAILURE in 9m 49s","accounts_in_message":[],"_revision_number":44},{"id":"9e9910ef3a328e4dcc0b7c42118ee4d0c915c856","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-04 14:00:19.000000000","message":"Patch Set 45: Published edit on patch set 44.","accounts_in_message":[],"_revision_number":45},{"id":"749449d085b6885cb6f67617c254d8d9fa3bfabb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 14:49:19.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/45/check/openstack-tox-docs/81e93fd/ : FAILURE in 3m 13s\n- swift-tox-py27 http://logs.openstack.org/89/517389/45/check/swift-tox-py27/6592aeb/ : SUCCESS in 9m 45s\n- swift-tox-py35 http://logs.openstack.org/89/517389/45/check/swift-tox-py35/11114c9/ : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/89/517389/45/check/swift-tox-py36/416b722/ : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/89/517389/45/check/swift-tox-py37/d9429bc/ : SUCCESS in 8m 07s\n- swift-tox-func http://logs.openstack.org/89/517389/45/check/swift-tox-func/2ab9152/ : FAILURE in 3m 03s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/45/check/swift-tox-func-encryption/7fb55fc/ : FAILURE in 3m 53s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/45/check/swift-tox-func-domain-remap-staticweb/1860458/ : FAILURE in 4m 28s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/45/check/swift-tox-func-ec/c3d5790/ : FAILURE in 3m 00s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/45/check/swift-tox-func-s3api/19834a9/ : FAILURE in 3m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/45/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4aa66bc/ : SUCCESS in 12m 21s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/45/check/swift-probetests-centos-7/f886294/ : SUCCESS in 43m 20s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/45/check/swift-dsvm-functional/3663fb9/ : FAILURE in 16m 04s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/45/check/swift-dsvm-functional-ipv6/8cf7a77/ : FAILURE in 16m 02s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/45/check/swift-tox-lower-constraints/03d6856/ : SUCCESS in 11m 47s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/45/check/openstack-tox-pep8/bba538a/ : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/45/check/swift-multinode-rolling-upgrade/9081d06/ : FAILURE in 7m 52s","accounts_in_message":[],"_revision_number":45},{"id":"4d062b2db1d0d3260456c74385a7e8ddff9940c1","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-04 14:51:14.000000000","message":"Patch Set 46: Published edit on patch set 45.","accounts_in_message":[],"_revision_number":46},{"id":"ad9d0c2e3ee1b42a69b8368d46090ecc5c8793b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 15:36:28.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/517389/46/check/openstack-tox-docs/04ca0dc/html/ : SUCCESS in 3m 56s\n- swift-tox-py27 http://logs.openstack.org/89/517389/46/check/swift-tox-py27/799c312/ : SUCCESS in 12m 27s\n- swift-tox-py35 http://logs.openstack.org/89/517389/46/check/swift-tox-py35/0cae078/ : SUCCESS in 10m 00s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/89/517389/46/check/swift-tox-py36/79f60d1/ : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/89/517389/46/check/swift-tox-py37/e1a6ba1/ : SUCCESS in 10m 09s\n- swift-tox-func http://logs.openstack.org/89/517389/46/check/swift-tox-func/839fb5b/ : FAILURE in 3m 05s\n- swift-tox-func-encryption http://logs.openstack.org/89/517389/46/check/swift-tox-func-encryption/c355875/ : FAILURE in 3m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/89/517389/46/check/swift-tox-func-domain-remap-staticweb/f3b8750/ : FAILURE in 3m 00s\n- swift-tox-func-ec http://logs.openstack.org/89/517389/46/check/swift-tox-func-ec/918c02f/ : FAILURE in 3m 08s\n- swift-tox-func-s3api http://logs.openstack.org/89/517389/46/check/swift-tox-func-s3api/4a1fe0b/ : FAILURE in 3m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/89/517389/46/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1f433bc/ : SUCCESS in 12m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/89/517389/46/check/swift-probetests-centos-7/40a2c3d/ : SUCCESS in 39m 32s\n- swift-dsvm-functional http://logs.openstack.org/89/517389/46/check/swift-dsvm-functional/dc6fc86/ : FAILURE in 14m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/89/517389/46/check/swift-dsvm-functional-ipv6/f8fcdea/ : FAILURE in 12m 34s\n- swift-tox-lower-constraints http://logs.openstack.org/89/517389/46/check/swift-tox-lower-constraints/06de6c2/ : SUCCESS in 11m 31s\n- openstack-tox-pep8 http://logs.openstack.org/89/517389/46/check/openstack-tox-pep8/ef74d44/ : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/89/517389/46/check/swift-multinode-rolling-upgrade/5940fa1/ : FAILURE in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"f7d7fb410fdf1a8b19d21ca3a834e9892b47872c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-17 04:18:27.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"8540d3ba93ede71d3fc305212ac0388067ab5635","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 04:56:13.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b254ba3b75b64ba490be227317d9c487 : FAILURE in 34m 16s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59c52be808344c02b3391a66032f0178 : FAILURE in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"c4dcee6e13b8ca8240fca409e02864b8dd43d42c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 05:51:51.000000000","message":"Patch Set 47: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ec94b6d05b4362bbfca40c8e4cd615 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/610a42b9574741668f83b208705f7efc : SUCCESS in 57m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0eabe15ac80a42aaab184e1f70e11614 : FAILURE in 22m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/89ff13d33921435ba5735a817b208e71 : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2beb85dd0c8e41c7a8c8ac1a3ce7125a : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ca8a0de092f4975a02926f204b789a6 : FAILURE in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5037371cda0c4bfb8b78a53ae7cfb047 : FAILURE in 14m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d00d5321722a432cbbe96783f34d7934 : FAILURE in 15m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aca9225ef1364f4781fa7bc3e1ba8ade : FAILURE in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0080f29a23f34d75b9587483647370f1 : FAILURE in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3bf6c67eeb3643f892c8bcfacf43a9f6 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b1abbdcb127a47c0b0b3066d1b200e20 : SUCCESS in 22m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7d588a7da6a4ae1998ec473b20d733f : SUCCESS in 15m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e335f6576444104b43ee2b7d46e0614 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/584281c3aa5145f68acf3dff937f55bb : SUCCESS in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ddb045d7a0714754ad476cf0637e1506 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5cc8bedc6a9545b2be305ac2f71e5a70 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e5331dedcb8047b6bdb5bf00a76c5d1e : SUCCESS in 23m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/812623d8121c4b6586f6df5a27bcf47d : FAILURE in 54m 40s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5e576f8c7f3346a38da70a226393ecee : FAILURE in 59m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8ce07f1cb084d9db9355e62c60d6468 : TIMED_OUT in 1h 32m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5a20eb824bb5460b8aa3fbf5ad283536 : TIMED_OUT in 1h 32m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b344678c21f94e3aa8c0ff258d6092a5 : FAILURE in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/471d504611bf4734b954599606edbea8 : FAILURE in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59ef24f59fc64d4c9c60286053a91f7f : SUCCESS in 26m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e45b1e3c75924bbd8414cb9df88fe112 : FAILURE in 20m 11s","accounts_in_message":[],"_revision_number":47}],"current_revision_number":47,"current_revision":"88c7674f7a342aa04b680afb9c6c4d0f186d4ccf","revisions":{"e5e6b8b3c8f15a8be23840335f0c10c35bf1ca4c":{"kind":"REWORK","_number":1,"created":"2017-11-02 18:11:59.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/1"}}},"commit":{"parents":[{"commit":"449d83fb0c8668f444baf3c74cf15c26a70bb02e","subject":"Doc uses alias instead of aliases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/449d83fb0c8668f444baf3c74cf15c26a70bb02e"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"subject":"WIP: Update object expirer to use general task queue system","message":"WIP: Update object expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch update\nobject expirer\u0027s task system to use the general task queue system.\n\nThe general task queue system will be used in automated tiering\ntask system.\n\n[1]: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5e6b8b3c8f15a8be23840335f0c10c35bf1ca4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5e6b8b3c8f15a8be23840335f0c10c35bf1ca4c"}]},"branch":"refs/heads/master"},"7dfa014b851b435058fbdd6a46b155a44f3d538c":{"kind":"REWORK","_number":2,"created":"2017-12-20 12:50:10.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/2"}}},"commit":{"parents":[{"commit":"449d83fb0c8668f444baf3c74cf15c26a70bb02e","subject":"Doc uses alias instead of aliases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/449d83fb0c8668f444baf3c74cf15c26a70bb02e"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-12-20 13:01:26.000000000","tz":0},"subject":"WIP: Update object expirer to use general task queue system","message":"WIP: Update object expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch update\nobject expirer\u0027s task system to use the general task queue system.\n\nThe general task queue system will be used in automated tiering\ntask system.\n\n[1]: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dfa014b851b435058fbdd6a46b155a44f3d538c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dfa014b851b435058fbdd6a46b155a44f3d538c"}]},"branch":"refs/heads/master"},"638242bdbf0bde0dcce517785d465a6efc84106c":{"kind":"REWORK","_number":3,"created":"2017-12-21 11:29:01.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/3"}}},"commit":{"parents":[{"commit":"449d83fb0c8668f444baf3c74cf15c26a70bb02e","subject":"Doc uses alias instead of aliases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/449d83fb0c8668f444baf3c74cf15c26a70bb02e"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-12-21 11:41:46.000000000","tz":0},"subject":"WIP: Update object expirer to use general task queue system","message":"WIP: Update object expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch update\nobject expirer\u0027s task system to use the general task queue system.\n\nThe general task queue system will be used in automated tiering\ntask system.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/638242bdbf0bde0dcce517785d465a6efc84106c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/638242bdbf0bde0dcce517785d465a6efc84106c"}]},"branch":"refs/heads/master"},"d43ee63a3e7525aab720c615fb454e68e0465167":{"kind":"REWORK","_number":4,"created":"2017-12-22 07:26:20.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/4"}}},"commit":{"parents":[{"commit":"449d83fb0c8668f444baf3c74cf15c26a70bb02e","subject":"Doc uses alias instead of aliases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/449d83fb0c8668f444baf3c74cf15c26a70bb02e"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-12-22 07:06:45.000000000","tz":0},"subject":"WIP: Update object expirer to use general task queue system","message":"WIP: Update object expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch update\nobject expirer\u0027s task system to use the general task queue system.\n\nThe general task queue system will be used in automated tiering\ntask system.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d43ee63a3e7525aab720c615fb454e68e0465167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d43ee63a3e7525aab720c615fb454e68e0465167"}]},"branch":"refs/heads/master"},"e2d59d78b61ee1322fc8e62357936dd1d6f4a749":{"kind":"REWORK","_number":5,"created":"2018-01-10 11:15:53.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/5"}}},"commit":{"parents":[{"commit":"b0242f4fdc7d77b6c328ead519310d9ebb85f193","subject":"Merge \"Fix intermittent check_delete_headers failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0242f4fdc7d77b6c328ead519310d9ebb85f193"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-01-10 11:28:50.000000000","tz":0},"subject":"WIP: Update object expirer to use general task queue system","message":"WIP: Update object expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch update\nobject expirer\u0027s task system to use the general task queue system.\n\nThe general task queue system will be used in automated tiering\ntask system.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2d59d78b61ee1322fc8e62357936dd1d6f4a749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2d59d78b61ee1322fc8e62357936dd1d6f4a749"}]},"branch":"refs/heads/master"},"67295b2f469dffc205d32b64f772bdeb624fb466":{"kind":"REWORK","_number":6,"created":"2018-01-15 09:14:28.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/6"}}},"commit":{"parents":[{"commit":"b0242f4fdc7d77b6c328ead519310d9ebb85f193","subject":"Merge \"Fix intermittent check_delete_headers failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0242f4fdc7d77b6c328ead519310d9ebb85f193"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-01-15 09:27:42.000000000","tz":0},"subject":"WIP: Update object expirer to use general task queue system","message":"WIP: Update object expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch update\nobject expirer\u0027s task system to use the general task queue system.\n\nThe general task queue system will be used in automated tiering\ntask system.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67295b2f469dffc205d32b64f772bdeb624fb466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67295b2f469dffc205d32b64f772bdeb624fb466"}]},"branch":"refs/heads/master"},"2e8fcbbc2dbe77366aa4ce5be9472672a82d013d":{"kind":"REWORK","_number":7,"created":"2018-01-16 09:35:32.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/7"}}},"commit":{"parents":[{"commit":"b0242f4fdc7d77b6c328ead519310d9ebb85f193","subject":"Merge \"Fix intermittent check_delete_headers failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0242f4fdc7d77b6c328ead519310d9ebb85f193"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-01-16 09:48:52.000000000","tz":0},"subject":"WIP: Update object expirer to use general task queue system","message":"WIP: Update object expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch update\nobject expirer\u0027s task system to use the general task queue system.\n\nThe general task queue system will be used in automated tiering\ntask system.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e8fcbbc2dbe77366aa4ce5be9472672a82d013d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e8fcbbc2dbe77366aa4ce5be9472672a82d013d"}]},"branch":"refs/heads/master"},"4f7239ff041518fed3bae9a45e5f945c38c627df":{"kind":"REWORK","_number":8,"created":"2018-01-17 10:00:07.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/8"}}},"commit":{"parents":[{"commit":"7227d9967b052e0f6f29ee3b09b0ca3748f45b1c","subject":"Merge \"Add tests for X-Backend-Clean-Expiring-Object-Queue true\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7227d9967b052e0f6f29ee3b09b0ca3748f45b1c"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-01-17 07:31:56.000000000","tz":0},"subject":"Update object-expirer to use general task queue system","message":"Update object-expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch updates\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f7239ff041518fed3bae9a45e5f945c38c627df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f7239ff041518fed3bae9a45e5f945c38c627df"}]},"branch":"refs/heads/master"},"42c198f349bfe6dc5149ed411af73074f1d9c6e6":{"kind":"REWORK","_number":9,"created":"2018-01-17 10:11:34.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/9"}}},"commit":{"parents":[{"commit":"7227d9967b052e0f6f29ee3b09b0ca3748f45b1c","subject":"Merge \"Add tests for X-Backend-Clean-Expiring-Object-Queue true\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7227d9967b052e0f6f29ee3b09b0ca3748f45b1c"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-01-17 10:23:32.000000000","tz":0},"subject":"Update object-expirer to use general task queue system","message":"Update object-expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch updates\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42c198f349bfe6dc5149ed411af73074f1d9c6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42c198f349bfe6dc5149ed411af73074f1d9c6e6"}]},"branch":"refs/heads/master"},"9b13cfe5e27eb0bccba54a2d9f156aa6958867dc":{"kind":"REWORK","_number":10,"created":"2018-01-18 06:14:38.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/10"}}},"commit":{"parents":[{"commit":"d707fc7b6d0ceb4556dddfc258c5de8c4baff05c","subject":"DRY out tests until the stone bleeds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d707fc7b6d0ceb4556dddfc258c5de8c4baff05c"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-01-18 05:13:50.000000000","tz":0},"subject":"Update object-expirer to use general task queue system","message":"Update object-expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch updates\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b13cfe5e27eb0bccba54a2d9f156aa6958867dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b13cfe5e27eb0bccba54a2d9f156aa6958867dc"}]},"branch":"refs/heads/master"},"a18284a285b826a38827f463e0edcbd62fb5c5f3":{"kind":"REWORK","_number":11,"created":"2018-01-24 12:01:58.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/11"}}},"commit":{"parents":[{"commit":"813e6ad265d00088931c0637f439a65bcb496336","subject":"Merge \"object-server can 409 in response to x-if-delete-at\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/813e6ad265d00088931c0637f439a65bcb496336"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-01-24 12:15:30.000000000","tz":0},"subject":"Update object-expirer to use general task queue system","message":"Update object-expirer to use general task queue system\n\nAt first step for general task queue system [1], this patch updates\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a18284a285b826a38827f463e0edcbd62fb5c5f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a18284a285b826a38827f463e0edcbd62fb5c5f3"}]},"branch":"refs/heads/master"},"7e3950e03876bb81fd2abbefbc5d08968acd1ca8":{"kind":"REWORK","_number":12,"created":"2018-02-09 08:51:50.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/12"}}},"commit":{"parents":[{"commit":"f9fe7114a0228e2940b37e5e1b0cf42c50f82ad0","subject":"Refactor expirer\u0027s task round robin implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9fe7114a0228e2940b37e5e1b0cf42c50f82ad0"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-09 08:50:28.000000000","tz":0},"subject":"WIP: Add object-expirer new mode to execute tasks from general task queue","message":"WIP: Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3950e03876bb81fd2abbefbc5d08968acd1ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3950e03876bb81fd2abbefbc5d08968acd1ca8"}]},"branch":"refs/heads/master"},"3413a3211d69613a7e1b2da8c4778a7306eeadbb":{"kind":"REWORK","_number":13,"created":"2018-02-13 12:09:27.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/13"}}},"commit":{"parents":[{"commit":"a50049f42fde86d7e7ff6f32ca8d03f6c56ffc1b","subject":"Split expirer methods into smaller ones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a50049f42fde86d7e7ff6f32ca8d03f6c56ffc1b"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:07:46.000000000","tz":0},"subject":"WIP: Add object-expirer new mode to execute tasks from general task queue","message":"WIP: Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3413a3211d69613a7e1b2da8c4778a7306eeadbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3413a3211d69613a7e1b2da8c4778a7306eeadbb"}]},"branch":"refs/heads/master"},"2a261b60e73f855a7910e295a5c084a95ebb41c4":{"kind":"REWORK","_number":14,"created":"2018-02-13 12:57:08.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/14"}}},"commit":{"parents":[{"commit":"aca37769b9f301f3640819acbe8ff1198acc5d5c","subject":"Parameterize expirer unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aca37769b9f301f3640819acbe8ff1198acc5d5c"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:56:53.000000000","tz":0},"subject":"WIP: Add object-expirer new mode to execute tasks from general task queue","message":"WIP: Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a261b60e73f855a7910e295a5c084a95ebb41c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a261b60e73f855a7910e295a5c084a95ebb41c4"}]},"branch":"refs/heads/master"},"eb4201bf5b5bc8df36340aaf5fcabd0be95671d6":{"kind":"REWORK","_number":15,"created":"2018-02-19 11:30:54.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/15"}}},"commit":{"parents":[{"commit":"aca37769b9f301f3640819acbe8ff1198acc5d5c","subject":"Parameterize expirer unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aca37769b9f301f3640819acbe8ff1198acc5d5c"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-19 11:26:26.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nThe general task queue system will be used in auto tiering task\nsystem too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb4201bf5b5bc8df36340aaf5fcabd0be95671d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb4201bf5b5bc8df36340aaf5fcabd0be95671d6"}]},"branch":"refs/heads/master"},"b287a5b583485a888512a8ca9c3b793ca5d8a66d":{"kind":"REWORK","_number":16,"created":"2018-02-27 09:36:45.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/16"}}},"commit":{"parents":[{"commit":"ab23e1b51395d7ff209e7a967924504c4cb4561c","subject":"Parameterize expirer unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab23e1b51395d7ff209e7a967924504c4cb4561c"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-27 09:35:13.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tssks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b287a5b583485a888512a8ca9c3b793ca5d8a66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b287a5b583485a888512a8ca9c3b793ca5d8a66d"}]},"branch":"refs/heads/master"},"827c93a7b0fdf219379dac4a26bb89baee9bd5c8":{"kind":"REWORK","_number":17,"created":"2018-02-28 12:15:27.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/17"}}},"commit":{"parents":[{"commit":"fc4c40d1db0a310ce37354f87ff6e7d3c26872e3","subject":"Parameterize expirer unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc4c40d1db0a310ce37354f87ff6e7d3c26872e3"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-28 12:13:40.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tssks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/827c93a7b0fdf219379dac4a26bb89baee9bd5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/827c93a7b0fdf219379dac4a26bb89baee9bd5c8"}]},"branch":"refs/heads/master"},"25c61bcbe0534cd944baa0cf89cee7e0431bbea6":{"kind":"REWORK","_number":18,"created":"2018-02-28 15:04:39.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/18"}}},"commit":{"parents":[{"commit":"a77ddf4e36eec3d881bc5a56fee7f1cc539cc044","subject":"Parameterize expirer unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a77ddf4e36eec3d881bc5a56fee7f1cc539cc044"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2017-11-02 18:15:31.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-28 14:56:19.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tssks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25c61bcbe0534cd944baa0cf89cee7e0431bbea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25c61bcbe0534cd944baa0cf89cee7e0431bbea6"}]},"branch":"refs/heads/master"},"8466824a1e49aa7831e0b60332d75f3ba9bb7130":{"kind":"REWORK","_number":19,"created":"2018-03-01 18:34:14.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/19"}}},"commit":{"parents":[{"commit":"1fadffeae0defb5c8956247e9118e74c8405f8ba","subject":"Split expirer methods and parametrize task account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fadffeae0defb5c8956247e9118e74c8405f8ba"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-03-01 18:29:08.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tssks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8466824a1e49aa7831e0b60332d75f3ba9bb7130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8466824a1e49aa7831e0b60332d75f3ba9bb7130"}]},"branch":"refs/heads/master"},"53a4f71b30d210adeacc056886ee0590f77e1bd5":{"kind":"REWORK","_number":20,"created":"2018-03-20 08:32:47.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/20"}}},"commit":{"parents":[{"commit":"bac79f2daeb1a4f103bb81daf0b1a52fbce632c3","subject":"Unify expirer unit test task queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bac79f2daeb1a4f103bb81daf0b1a52fbce632c3"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-03-20 08:30:30.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tssks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53a4f71b30d210adeacc056886ee0590f77e1bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53a4f71b30d210adeacc056886ee0590f77e1bd5"}]},"branch":"refs/heads/master"},"b76221d99cebf4f48aa138488fed0a6c66df8765":{"kind":"REWORK","_number":21,"created":"2018-03-27 08:45:39.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/21"}}},"commit":{"parents":[{"commit":"bac79f2daeb1a4f103bb81daf0b1a52fbce632c3","subject":"Unify expirer unit test task queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bac79f2daeb1a4f103bb81daf0b1a52fbce632c3"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-03-27 08:44:11.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tssks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b76221d99cebf4f48aa138488fed0a6c66df8765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b76221d99cebf4f48aa138488fed0a6c66df8765"}]},"branch":"refs/heads/master"},"3aae6e384b6eeee0f53a8c78bd783d05c54221a7":{"kind":"REWORK","_number":22,"created":"2018-04-05 12:53:45.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/22"}}},"commit":{"parents":[{"commit":"bac79f2daeb1a4f103bb81daf0b1a52fbce632c3","subject":"Unify expirer unit test task queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bac79f2daeb1a4f103bb81daf0b1a52fbce632c3"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-04-05 11:39:40.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aae6e384b6eeee0f53a8c78bd783d05c54221a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aae6e384b6eeee0f53a8c78bd783d05c54221a7"}]},"branch":"refs/heads/master"},"26eb4283daf628f53517c4f697ec1f4982fd7861":{"kind":"REWORK","_number":23,"created":"2018-04-17 09:36:12.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/23"}}},"commit":{"parents":[{"commit":"bac79f2daeb1a4f103bb81daf0b1a52fbce632c3","subject":"Unify expirer unit test task queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bac79f2daeb1a4f103bb81daf0b1a52fbce632c3"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-04-13 01:50:51.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26eb4283daf628f53517c4f697ec1f4982fd7861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26eb4283daf628f53517c4f697ec1f4982fd7861"}]},"branch":"refs/heads/master"},"6aa55d279d7a1f8f40bfe2a11223824a2703ee53":{"kind":"REWORK","_number":24,"created":"2018-04-18 04:56:56.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/24"}}},"commit":{"parents":[{"commit":"bac79f2daeb1a4f103bb81daf0b1a52fbce632c3","subject":"Unify expirer unit test task queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bac79f2daeb1a4f103bb81daf0b1a52fbce632c3"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-04-18 04:56:50.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6aa55d279d7a1f8f40bfe2a11223824a2703ee53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6aa55d279d7a1f8f40bfe2a11223824a2703ee53"}]},"branch":"refs/heads/master"},"1b88046587f46167ae1accf44fc74daa77e1d258":{"kind":"REWORK","_number":25,"created":"2018-05-24 22:34:33.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/25"}}},"commit":{"parents":[{"commit":"8403ca391544bbe4c477ad1d20fea048ae5d2d03","subject":"Merge \"Let make_db_file_path accept epoch\u003dNone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8403ca391544bbe4c477ad1d20fea048ae5d2d03"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-05-24 22:33:46.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b88046587f46167ae1accf44fc74daa77e1d258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b88046587f46167ae1accf44fc74daa77e1d258"}]},"branch":"refs/heads/master"},"4f176c55984a7e2b235a0ce6ae3a832c7e1106f9":{"kind":"REWORK","_number":26,"created":"2018-06-08 01:28:45.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/26"}}},"commit":{"parents":[{"commit":"c01c43d982e811d2468a59f4fcc3b440e1200066","subject":"Merge \"Adds read_only middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c01c43d982e811d2468a59f4fcc3b440e1200066"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-06-08 01:18:03.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f176c55984a7e2b235a0ce6ae3a832c7e1106f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f176c55984a7e2b235a0ce6ae3a832c7e1106f9"}]},"branch":"refs/heads/master"},"accc12efe7ae4da213fda566bd628b4a3dd7403c":{"kind":"REWORK","_number":27,"created":"2018-06-08 01:30:36.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/27"}}},"commit":{"parents":[{"commit":"c01c43d982e811d2468a59f4fcc3b440e1200066","subject":"Merge \"Adds read_only middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c01c43d982e811d2468a59f4fcc3b440e1200066"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-06-08 01:32:01.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/accc12efe7ae4da213fda566bd628b4a3dd7403c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/accc12efe7ae4da213fda566bd628b4a3dd7403c"}]},"branch":"refs/heads/master"},"e2a4afb1f6a718c94d3ece3fcb4f71b9f85f8c0f":{"kind":"REWORK","_number":28,"created":"2018-09-10 20:05:27.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/89/517389/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/28"}}},"commit":{"parents":[{"commit":"1f29508327c0f5c1b06b1b06b06400055681583f","subject":"Allow for backwards compatibility with swift3.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f29508327c0f5c1b06b1b06b06400055681583f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-09-10 20:05:12.000000000","tz":540},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2a4afb1f6a718c94d3ece3fcb4f71b9f85f8c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2a4afb1f6a718c94d3ece3fcb4f71b9f85f8c0f"}]},"branch":"refs/heads/master"},"d9e3b4a22b7221eb0d768b751f6ca60800346aad":{"kind":"REWORK","_number":29,"created":"2018-09-12 05:21:14.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/29"}}},"commit":{"parents":[{"commit":"1f29508327c0f5c1b06b1b06b06400055681583f","subject":"Allow for backwards compatibility with swift3.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f29508327c0f5c1b06b1b06b06400055681583f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 05:00:42.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e3b4a22b7221eb0d768b751f6ca60800346aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e3b4a22b7221eb0d768b751f6ca60800346aad"}]},"branch":"refs/heads/master"},"0ed5948c8062350e1a9f6bb7d2b30d32e7bbeb8b":{"kind":"REWORK","_number":30,"created":"2018-09-12 06:27:46.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/30"}}},"commit":{"parents":[{"commit":"1f29508327c0f5c1b06b1b06b06400055681583f","subject":"Allow for backwards compatibility with swift3.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f29508327c0f5c1b06b1b06b06400055681583f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 06:10:40.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ed5948c8062350e1a9f6bb7d2b30d32e7bbeb8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ed5948c8062350e1a9f6bb7d2b30d32e7bbeb8b"}]},"branch":"refs/heads/master"},"9e0a069272d8f9094346770f368431b836863f4a":{"kind":"REWORK","_number":31,"created":"2018-09-12 08:36:47.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/31"}}},"commit":{"parents":[{"commit":"6f581b544324f21a34929fcd0ca166e21e8a0487","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f581b544324f21a34929fcd0ca166e21e8a0487"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 08:19:15.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e0a069272d8f9094346770f368431b836863f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e0a069272d8f9094346770f368431b836863f4a"}]},"branch":"refs/heads/master"},"bde878d37311e2f74581adf0847e93173f3bb42c":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2018-09-12 16:15:44.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/32"}}},"commit":{"parents":[{"commit":"7e7613ff9b337f1c6bb31c9be98da4a24ceb302e","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e7613ff9b337f1c6bb31c9be98da4a24ceb302e"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 16:15:44.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bde878d37311e2f74581adf0847e93173f3bb42c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bde878d37311e2f74581adf0847e93173f3bb42c"}]},"branch":"refs/heads/master"},"609dc6b7b3b6c95b073d6beba7a98844c60557a8":{"kind":"REWORK","_number":33,"created":"2018-09-13 21:26:43.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/33"}}},"commit":{"parents":[{"commit":"7e7613ff9b337f1c6bb31c9be98da4a24ceb302e","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e7613ff9b337f1c6bb31c9be98da4a24ceb302e"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-13 21:24:11.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/609dc6b7b3b6c95b073d6beba7a98844c60557a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/609dc6b7b3b6c95b073d6beba7a98844c60557a8"}]},"branch":"refs/heads/master"},"be05644f936de8d839cb3dd07a438cb01636623c":{"kind":"REWORK","_number":34,"created":"2018-11-28 08:36:11.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/34"}}},"commit":{"parents":[{"commit":"ff5efc0255132902aa88706df2d17046e0e8838f","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5efc0255132902aa88706df2d17046e0e8838f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-11-28 08:33:33.000000000","tz":540},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be05644f936de8d839cb3dd07a438cb01636623c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be05644f936de8d839cb3dd07a438cb01636623c"}]},"branch":"refs/heads/master"},"7b8d2840373d3c195a71abb6d660c3122f3aeaa2":{"kind":"REWORK","_number":35,"created":"2018-12-26 05:32:48.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/35"}}},"commit":{"parents":[{"commit":"f0d94b2dadf131ed1eb64efe07b59d801009fd55","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d94b2dadf131ed1eb64efe07b59d801009fd55"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-26 05:30:53.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b8d2840373d3c195a71abb6d660c3122f3aeaa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b8d2840373d3c195a71abb6d660c3122f3aeaa2"}]},"branch":"refs/heads/master"},"5f47f792e4175cf9228b4e0dbdbfb2684b17cdbb":{"kind":"REWORK","_number":36,"created":"2018-12-26 06:30:18.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/36"}}},"commit":{"parents":[{"commit":"f0d94b2dadf131ed1eb64efe07b59d801009fd55","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d94b2dadf131ed1eb64efe07b59d801009fd55"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-26 06:28:57.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f47f792e4175cf9228b4e0dbdbfb2684b17cdbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f47f792e4175cf9228b4e0dbdbfb2684b17cdbb"}]},"branch":"refs/heads/master"},"8bcdb915bd54a46bfd59d3555a93c33d7b489f21":{"kind":"REWORK","_number":37,"created":"2018-12-26 07:44:22.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/37"}}},"commit":{"parents":[{"commit":"f0d94b2dadf131ed1eb64efe07b59d801009fd55","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d94b2dadf131ed1eb64efe07b59d801009fd55"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-26 07:43:49.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bcdb915bd54a46bfd59d3555a93c33d7b489f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bcdb915bd54a46bfd59d3555a93c33d7b489f21"}]},"branch":"refs/heads/master"},"4dfa25f681f5adc8bea0a1fa9fe1f9f9677f7bf0":{"kind":"REWORK","_number":38,"created":"2018-12-26 07:54:08.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/38"}}},"commit":{"parents":[{"commit":"f0d94b2dadf131ed1eb64efe07b59d801009fd55","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d94b2dadf131ed1eb64efe07b59d801009fd55"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-26 07:53:31.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dfa25f681f5adc8bea0a1fa9fe1f9f9677f7bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dfa25f681f5adc8bea0a1fa9fe1f9f9677f7bf0"}]},"branch":"refs/heads/master"},"14d113ff14127c0cce9668fa91daa6f12149c6f6":{"kind":"REWORK","_number":39,"created":"2018-12-27 11:47:27.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/39"}}},"commit":{"parents":[{"commit":"99095807a919b8f8cb1e26a8f30a1f8f457894e1","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99095807a919b8f8cb1e26a8f30a1f8f457894e1"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-27 11:46:39.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14d113ff14127c0cce9668fa91daa6f12149c6f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14d113ff14127c0cce9668fa91daa6f12149c6f6"}]},"branch":"refs/heads/master"},"9e90cec88105bc4745389ebcafcd2b20ec76c5ba":{"kind":"REWORK","_number":40,"created":"2018-12-27 12:06:17.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/40"}}},"commit":{"parents":[{"commit":"99095807a919b8f8cb1e26a8f30a1f8f457894e1","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99095807a919b8f8cb1e26a8f30a1f8f457894e1"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-27 12:04:23.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e90cec88105bc4745389ebcafcd2b20ec76c5ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e90cec88105bc4745389ebcafcd2b20ec76c5ba"}]},"branch":"refs/heads/master"},"39077a22731aed5ba686fdc3a8ccd2b8dd269846":{"kind":"REWORK","_number":41,"created":"2019-05-02 22:03:51.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/89/517389/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/41"}}},"commit":{"parents":[{"commit":"3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-05-02 22:03:35.000000000","tz":540},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39077a22731aed5ba686fdc3a8ccd2b8dd269846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39077a22731aed5ba686fdc3a8ccd2b8dd269846"}]},"branch":"refs/heads/master"},"d36ff26e2620a50e43aa194df48dff7748ae2a13":{"kind":"REWORK","_number":42,"created":"2019-05-02 22:55:08.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/42"}}},"commit":{"parents":[{"commit":"3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2019-05-02 22:54:53.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d36ff26e2620a50e43aa194df48dff7748ae2a13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d36ff26e2620a50e43aa194df48dff7748ae2a13"}]},"branch":"refs/heads/master"},"59e541d41d3772393bf50d6631973c29df0ac59c":{"kind":"REWORK","_number":43,"created":"2019-05-03 15:19:41.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/43"}}},"commit":{"parents":[{"commit":"3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2019-05-03 15:19:18.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59e541d41d3772393bf50d6631973c29df0ac59c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59e541d41d3772393bf50d6631973c29df0ac59c"}]},"branch":"refs/heads/master"},"2bfcddf165a95f1c0bff405ee9fdb2b65eb22721":{"kind":"REWORK","_number":44,"created":"2019-05-04 10:38:45.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/44"}}},"commit":{"parents":[{"commit":"3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2019-05-04 10:29:28.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bfcddf165a95f1c0bff405ee9fdb2b65eb22721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bfcddf165a95f1c0bff405ee9fdb2b65eb22721"}]},"branch":"refs/heads/master"},"72286c948707bce82d6471bbafe0b5e95c0296b9":{"kind":"REWORK","_number":45,"created":"2019-05-04 14:00:19.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/45"}}},"commit":{"parents":[{"commit":"3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2019-05-04 14:00:04.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72286c948707bce82d6471bbafe0b5e95c0296b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72286c948707bce82d6471bbafe0b5e95c0296b9"}]},"branch":"refs/heads/master"},"d2a20f815e81b44f469d307bbd1d51bc4d6fee85":{"kind":"REWORK","_number":46,"created":"2019-05-04 14:51:14.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/89/517389/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/46"}}},"commit":{"parents":[{"commit":"3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401","subject":"Enable to configure object-expirer in object-server.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2019-05-04 14:50:56.000000000","tz":0},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2a20f815e81b44f469d307bbd1d51bc4d6fee85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2a20f815e81b44f469d307bbd1d51bc4d6fee85"}]},"branch":"refs/heads/master"},"88c7674f7a342aa04b680afb9c6c4d0f186d4ccf":{"kind":"REWORK","_number":47,"created":"2021-06-17 04:18:27.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/89/517389/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/517389/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/517389/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/517389/47"}}},"commit":{"parents":[{"commit":"977f1bcc61077833c8473f7dee7a53e5605def2f","subject":"Merge \"Add unit test for missing whole EC fragments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/977f1bcc61077833c8473f7dee7a53e5605def2f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-02-13 12:42:18.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-17 04:18:22.000000000","tz":600},"subject":"Add object-expirer new mode to execute tasks from general task queue","message":"Add object-expirer new mode to execute tasks from general task queue\n\nAt first step for general task queue system [1], this patch adds\nobject-expirer\u0027s task system to use the general task queue. With\nthis change, object-expirer\u0027s task queue will gain scalabilily. a\nlarge number of object-expirers will be able to run on a large swift\ncluster at the same time.\n\nAfter this patch, enqueue target of expiring tasks are automatically changed to\ngeneral task queue. Dequeue target of expiring tasks can be choosen by operator\nfrom legacy (\u003dthe task queue used before this patch) or general task queue.\n\nThe general task queue system will be used in auto tiering task system too.\n\n[1] Basic idea: https://wiki.openstack.org/wiki/Swift/ideas/task-execution\n[2] Details: https://docs.google.com/document/d/11sBbB6pBvLYNeM9wjTdvvsJIu8Dl8i13UH2NfRVNOqg/edit#\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: I4bb584dadc8d36e2533c5f7b555ce359fde79276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88c7674f7a342aa04b680afb9c6c4d0f186d4ccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88c7674f7a342aa04b680afb9c6c4d0f186d4ccf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
