)]}'
{"id":"openstack%2Fswift~525770","triplet_id":"openstack%2Fswift~master~Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046","project":"openstack/swift","branch":"master","topic":"bug/1733634","hashtags":[],"change_id":"Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046","subject":"tempurl: Make the digest algorithm configurable","status":"MERGED","created":"2017-12-05 21:55:13.000000000","updated":"2018-02-01 03:51:06.000000000","submitted":"2018-02-01 03:51:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"525770-1517457066725-b5c247cf","meta_rev_id":"e069c233df16a434f9dbad4e3cabfe6068bbd90f","_number":525770,"virtual_id_number":525770,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":2,"date":"2018-02-01 03:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2018-01-31 02:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2018-01-31 09:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2018-01-31 09:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-29 17:21:26.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-05 19:18:48.000000000","updated_by":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"reviewer":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"state":"REVIEWER"},{"updated":"2018-01-31 02:36:15.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2018-01-31 09:38:55.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2018-02-01 03:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1911ebeeb74e1afad602ecb7de4c85072d2c7266","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-05 21:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f5f09a9181d4443f3e76b08ab20e32bf7e9ffaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-06 01:47:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/525770/1/check/openstack-tox-pep8/b4a5512/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/70/525770/1/check/openstack-tox-py27/c403967/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/70/525770/1/check/openstack-tox-py35/b662ade/ : SUCCESS in 3m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/70/525770/1/check/build-openstack-sphinx-docs/9fc5398/html/ : SUCCESS in 4m 52s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/70/525770/1/check/legacy-swift-tox-xfs-tmp-py27/af1a03a/ : SUCCESS in 7m 01s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/70/525770/1/check/legacy-swift-tox-xfs-tmp-py35/d6ae31b/ : SUCCESS in 2m 44s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/70/525770/1/check/legacy-swift-tox-xfs-tmp-func/135a7ba/ : SUCCESS in 7m 06s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/70/525770/1/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/61cd7ac/ : SUCCESS in 8m 37s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/70/525770/1/check/legacy-swift-tox-xfs-tmp-func-encryption/975fd2f/ : SUCCESS in 6m 38s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/70/525770/1/check/legacy-swift-tox-xfs-tmp-func-ec/c1551b2/ : SUCCESS in 7m 27s\n- legacy-swift-dsvm-functional http://logs.openstack.org/70/525770/1/check/legacy-swift-dsvm-functional/1d2e11f/ : SUCCESS in 24m 30s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/70/525770/1/check/legacy-swift-dsvm-functional-identity-v3-only/8f73234/ : SUCCESS in 25m 11s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/70/525770/1/check/legacy-tempest-dsvm-neutron-full/4817f05/ : SUCCESS in 1h 21m 54s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/70/525770/1/check/legacy-grenade-dsvm-neutron/7f63e62/ : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0b74b96401a7f3da995bb982c884364ce630a12b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-28 04:20:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fdb2c4c2d7736f7695d3ac62a0a61f5d57759a27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 05:43:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/70/525770/2/check/build-openstack-sphinx-docs/a5e486a/html/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/70/525770/2/check/openstack-tox-pep8/f46c49f/ : SUCCESS in 3m 29s\n- legacy-swift-dsvm-functional http://logs.openstack.org/70/525770/2/check/legacy-swift-dsvm-functional/3ba97ff/ : SUCCESS in 26m 37s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/70/525770/2/check/legacy-swift-dsvm-functional-identity-v3-only/f6e4f10/ : SUCCESS in 25m 08s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/70/525770/2/check/legacy-tempest-dsvm-neutron-full/c715d44/ : SUCCESS in 1h 21m 35s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/70/525770/2/check/legacy-grenade-dsvm-neutron/28245e2/ : SUCCESS in 49m 03s\n- swift-tox-py27 http://logs.openstack.org/70/525770/2/check/swift-tox-py27/2201db8/ : SUCCESS in 7m 11s\n- swift-tox-py35 http://logs.openstack.org/70/525770/2/check/swift-tox-py35/7adefc0/ : SUCCESS in 3m 05s\n- swift-tox-func http://logs.openstack.org/70/525770/2/check/swift-tox-func/f08f3ac/ : SUCCESS in 9m 01s\n- swift-tox-func-post-as-copy http://logs.openstack.org/70/525770/2/check/swift-tox-func-post-as-copy/15c19b0/ : SUCCESS in 6m 51s\n- swift-tox-func-encryption http://logs.openstack.org/70/525770/2/check/swift-tox-func-encryption/1831c46/ : SUCCESS in 7m 21s\n- swift-tox-func-ec http://logs.openstack.org/70/525770/2/check/swift-tox-func-ec/4d2ca6c/ : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":2},{"id":"229d70d33780cc20b148f64c47cf6f87a98e751a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-28 19:12:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ff027cd162df45d27bd9a29d544017fe67cc676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 20:40:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/70/525770/3/check/build-openstack-sphinx-docs/f448ace/html/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/70/525770/3/check/openstack-tox-pep8/f5ef73d/ : SUCCESS in 5m 47s\n- legacy-swift-dsvm-functional http://logs.openstack.org/70/525770/3/check/legacy-swift-dsvm-functional/6b35820/ : SUCCESS in 22m 23s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/70/525770/3/check/legacy-swift-dsvm-functional-identity-v3-only/958a8f0/ : SUCCESS in 21m 43s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/70/525770/3/check/legacy-tempest-dsvm-neutron-full/0864b49/ : SUCCESS in 1h 25m 36s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/70/525770/3/check/legacy-grenade-dsvm-neutron/6ce5804/ : SUCCESS in 49m 16s\n- swift-tox-py27 http://logs.openstack.org/70/525770/3/check/swift-tox-py27/294d82f/ : SUCCESS in 7m 22s\n- swift-tox-py35 http://logs.openstack.org/70/525770/3/check/swift-tox-py35/97df137/ : SUCCESS in 2m 48s\n- swift-tox-func http://logs.openstack.org/70/525770/3/check/swift-tox-func/46554d2/ : SUCCESS in 8m 19s\n- swift-tox-func-post-as-copy http://logs.openstack.org/70/525770/3/check/swift-tox-func-post-as-copy/9a7dda1/ : SUCCESS in 6m 39s\n- swift-tox-func-encryption http://logs.openstack.org/70/525770/3/check/swift-tox-func-encryption/9b66b57/ : SUCCESS in 8m 34s\n- swift-tox-func-ec http://logs.openstack.org/70/525770/3/check/swift-tox-func-ec/29de60d/ : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":3},{"id":"5061cee629a6e499d3271cf248e841d377b5a811","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-29 02:44:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525770/7756/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525770/9735/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525770/8150/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cda6f66601afe0a511eeb7660643d397cf899f57","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-29 17:21:26.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525770/8152/probetests/ : ABORTED\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525770/7758/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525770/9737/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"3d13974001f6666ab9641231f81498ff26a5c8fc","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2018-01-05 19:18:48.000000000","message":"Patch Set 3: Code-Review-1\n\nI think it\u0027d be nice to have the algorithm as part of the query parameter, so instead of abc123def456, make it more like sha256:abc123def456. That\u0027d cut down on the number of digests we have to compute in the tempurl middleware.\n\nWith this exact set of available hash functions, we can determine it by length, but if we add another hash function whose output size is the same as one of our existing ones, then we\u0027re stuck computing multiple HMACs no matter what we do.","accounts_in_message":[],"_revision_number":3},{"id":"636c3bf4bd23c7797d79721023d238bf40000d70","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-10 19:12:33.000000000","message":"Patch Set 3:\n\n\u003e I think it\u0027d be nice to have the algorithm as part of the query parameter, so instead of abc123def456, make it more like sha256:abc123def456\n\nYeah, I thought about that (actually, almost *exactly* that -- I was also thinking of switching to base64 since a hex SHA-512 gets pretty long), but didn\u0027t want to be needlessly different from what cloud files seems to be doing -- see https://github.com/inflatador/ptemper/blob/master/ptemper.py#L186-L187 which got linked to on the bug report.\n\nI\u0027ll play around with it, see what I come up with... I might also rethink how to approach func testing -- it seems unnecessary to have full copies of all tests that just switch between\n\n- \u003chex-encoded sha1\u003e\n- \u003chex-encoded sha256\u003e\n- sha1:\u003cbase64-encoded sha1\u003e\n- sha256:\u003cbase64-encoded sha256\u003e\n- sha512:\u003cbase64-encoded sha512\u003e\n- whirlpool:\u003cbase64-encoded whirlpool\u003e\n- blake2b:\u003cbase64-encoded blake2b\u003e\n- ...whatever else we may end up adding in the future...","accounts_in_message":[],"_revision_number":3},{"id":"ea3776679b7eb3f1e67525d125a1978ab71c5fc6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-23 21:38:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2cc650bb0e013429e37cbfdc0eb78950e461ad7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 23:51:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/70/525770/4/check/build-openstack-sphinx-docs/8985732/html/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/70/525770/4/check/openstack-tox-pep8/6568500/ : SUCCESS in 4m 26s\n- legacy-swift-dsvm-functional http://logs.openstack.org/70/525770/4/check/legacy-swift-dsvm-functional/f0477e7/ : SUCCESS in 33m 20s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/70/525770/4/check/legacy-swift-dsvm-functional-identity-v3-only/e469e36/ : SUCCESS in 38m 33s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/70/525770/4/check/legacy-tempest-dsvm-neutron-full/81d1800/ : SUCCESS in 1h 28m 14s\n- neutron-grenade http://logs.openstack.org/70/525770/4/check/neutron-grenade/29409db/ : SUCCESS in 49m 18s\n- swift-tox-py27 http://logs.openstack.org/70/525770/4/check/swift-tox-py27/1fc9473/ : FAILURE in 7m 07s\n- swift-tox-py35 http://logs.openstack.org/70/525770/4/check/swift-tox-py35/ac70bc3/ : SUCCESS in 4m 58s\n- swift-tox-func http://logs.openstack.org/70/525770/4/check/swift-tox-func/5bb7959/ : SUCCESS in 15m 28s\n- swift-tox-func-encryption http://logs.openstack.org/70/525770/4/check/swift-tox-func-encryption/eb419ae/ : SUCCESS in 9m 18s\n- swift-tox-func-ec http://logs.openstack.org/70/525770/4/check/swift-tox-func-ec/e1b3031/ : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":4},{"id":"2d02dbc40cc72d7dfd20e0727ac8aea9c4553022","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-29 21:25:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6176584a2a71ba22893431efc07a2112f0cbc07c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 00:40:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/70/525770/5/check/build-openstack-sphinx-docs/95976f5/html/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/70/525770/5/check/openstack-tox-pep8/2136e5c/ : SUCCESS in 3m 52s\n- legacy-swift-dsvm-functional http://logs.openstack.org/70/525770/5/check/legacy-swift-dsvm-functional/921373c/ : SUCCESS in 24m 34s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/70/525770/5/check/legacy-swift-dsvm-functional-identity-v3-only/960d033/ : SUCCESS in 36m 20s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/70/525770/5/check/legacy-tempest-dsvm-neutron-full/e54e699/ : SUCCESS in 1h 52m 02s\n- neutron-grenade http://logs.openstack.org/70/525770/5/check/neutron-grenade/d67f10b/ : SUCCESS in 50m 32s\n- swift-tox-py27 http://logs.openstack.org/70/525770/5/check/swift-tox-py27/99e2c15/ : SUCCESS in 12m 04s\n- swift-tox-py35 http://logs.openstack.org/70/525770/5/check/swift-tox-py35/d300fa2/ : SUCCESS in 3m 03s\n- swift-tox-func http://logs.openstack.org/70/525770/5/check/swift-tox-func/7c5e5ed/ : SUCCESS in 11m 24s\n- swift-tox-func-encryption http://logs.openstack.org/70/525770/5/check/swift-tox-func-encryption/67c2971/ : SUCCESS in 11m 40s\n- swift-tox-func-ec http://logs.openstack.org/70/525770/5/check/swift-tox-func-ec/ae94175/ : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":5},{"id":"8fb083b510cffa1399cef4369a0e5158d02d6edc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-01-30 03:38:06.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nAwesome work Tim!\n\nMissing param in doc string and maybe add a test cap. Though might be bordering on a NIT.","accounts_in_message":[],"_revision_number":5},{"id":"11ebde64613f07bf8079eade8e8496b49c5b1fca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-31 02:19:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e6b60298f94e102041fd6b66c16abf40ad29facb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-31 02:21:57.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8a9caeeb9b76e4060b4feef6a431d1b1e665122c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-01-31 02:36:15.000000000","message":"Patch Set 6: Code-Review+2\n\nyeah, I don\u0027t want to engineer a flaky test.. although I also don\u0027t want to have a test that could potentially never end. Is a fail better then an infinite loop.\n\nAnyway, you\u0027ve gone through and done the maths.. I also ran the thing all night and the highest number of iterations I got was 10. And honestly 10 iterations is nothing and that\u0027s after thousands of runs. So I\u0027m not too worried about it looping too long now. It\u0027ll only be a problem if urlsafe_b64 decided to not use either \u0027_\u0027 or \u0027-\u0027 :P\n\nSorry for my pedantic-iness :)","accounts_in_message":[],"_revision_number":6},{"id":"09d6366d3fc57a7fa8993debb8f8b7442b3e404c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 05:51:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/70/525770/6/check/build-openstack-sphinx-docs/8a16111/html/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/70/525770/6/check/openstack-tox-pep8/d869b53/ : SUCCESS in 4m 33s\n- legacy-swift-dsvm-functional http://logs.openstack.org/70/525770/6/check/legacy-swift-dsvm-functional/b60c400/ : SUCCESS in 22m 35s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/70/525770/6/check/legacy-swift-dsvm-functional-identity-v3-only/5da77fd/ : SUCCESS in 24m 14s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/70/525770/6/check/legacy-tempest-dsvm-neutron-full/0f53287/ : SUCCESS in 1h 18m 42s\n- neutron-grenade http://logs.openstack.org/70/525770/6/check/neutron-grenade/cdca84c/ : SUCCESS in 45m 27s\n- swift-tox-py27 http://logs.openstack.org/70/525770/6/check/swift-tox-py27/ee1b191/ : SUCCESS in 7m 31s\n- swift-tox-py35 http://logs.openstack.org/70/525770/6/check/swift-tox-py35/71ccac7/ : SUCCESS in 3m 20s\n- swift-tox-func http://logs.openstack.org/70/525770/6/check/swift-tox-func/b3f2fe3/ : SUCCESS in 9m 54s\n- swift-tox-func-encryption http://logs.openstack.org/70/525770/6/check/swift-tox-func-encryption/1666c22/ : SUCCESS in 8m 31s\n- swift-tox-func-ec http://logs.openstack.org/70/525770/6/check/swift-tox-func-ec/a985b9d/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":6},{"id":"b227f1d40a50b7ed2177c73678ad345c7cab8d09","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2018-01-31 09:14:10.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)\n\nGreat work Tim! Can\u0027t wait to see this merged. Didn\u0027t had a chance to run tests yet, so only +1 for now. Will try to run some tests later.","accounts_in_message":[],"_revision_number":6},{"id":"a8756a5668171727c68cfbdab74d7027576ebfcf","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2018-01-31 09:38:55.000000000","message":"Patch Set 6: Workflow+1 Code-Review+2\n\nWell, did a few tests on my machine and it works as expected. Let\u0027s merge this!","accounts_in_message":[],"_revision_number":6},{"id":"637e2ee8308246a81d9177a92231c1e1e3e8d70a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 13:47:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"91a19b27aab88665aed16ff74c8ccf170389bda8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 03:51:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/70/525770/6/gate/build-openstack-sphinx-docs/116480b/html/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/70/525770/6/gate/openstack-tox-pep8/cbe7148/ : SUCCESS in 6m 13s\n- legacy-swift-dsvm-functional http://logs.openstack.org/70/525770/6/gate/legacy-swift-dsvm-functional/0d515b9/ : SUCCESS in 25m 47s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/70/525770/6/gate/legacy-tempest-dsvm-neutron-full/0e388a1/ : SUCCESS in 1h 31m 08s\n- neutron-grenade http://logs.openstack.org/70/525770/6/gate/neutron-grenade/07b178b/ : SUCCESS in 1h 12m 30s\n- swift-tox-py27 http://logs.openstack.org/70/525770/6/gate/swift-tox-py27/21875d8/ : SUCCESS in 10m 16s\n- swift-tox-py35 http://logs.openstack.org/70/525770/6/gate/swift-tox-py35/407a50d/ : SUCCESS in 6m 39s\n- swift-tox-func http://logs.openstack.org/70/525770/6/gate/swift-tox-func/0949003/ : SUCCESS in 11m 16s\n- swift-tox-func-encryption http://logs.openstack.org/70/525770/6/gate/swift-tox-func-encryption/18a8dd1/ : SUCCESS in 10m 37s\n- swift-tox-func-ec http://logs.openstack.org/70/525770/6/gate/swift-tox-func-ec/abd1d4e/ : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":6},{"id":"24ac13ae66b31e3e6c9a7cf07174256e6c966d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 03:51:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5a4d3bdfc49d61ee6e333c8eb1dad4383ae9308b","revisions":{"dc9ba71d3e45a600ec4fc38e417279456b047862":{"kind":"REWORK","_number":1,"created":"2017-12-05 21:55:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/525770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/525770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/525770/1"}}},"commit":{"parents":[{"commit":"b808c806b4e7000aed45a0026dd7407cd1616891","subject":"Remove testTooLongName tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b808c806b4e7000aed45a0026dd7407cd1616891"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 20:19:37.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:34.000000000","tz":0},"subject":"tempurl: Make the digest algorithm configurable","message":"tempurl: Make the digest algorithm configurable\n\n... and add support for SHA-256 and SHA-512 by default. This allows us\nto start moving toward replacing SHA-1-based signatures. We\u0027ve known\nthis would eventually be necessary for a while [1], and earlier this\nyear we\u0027ve seen SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nRelated-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc9ba71d3e45a600ec4fc38e417279456b047862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc9ba71d3e45a600ec4fc38e417279456b047862"}]},"branch":"refs/heads/master"},"4e725e3b516c1a86b7104fd5693eced62c2a2b8a":{"kind":"REWORK","_number":2,"created":"2017-12-28 04:20:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/525770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/525770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/525770/2"}}},"commit":{"parents":[{"commit":"26822232b3872a184ccad9f367c262130579be00","subject":"Merge \"Fix sometimes-flaky container name functional test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26822232b3872a184ccad9f367c262130579be00"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 20:19:37.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-28 04:12:59.000000000","tz":-480},"subject":"tempurl: Make the digest algorithm configurable","message":"tempurl: Make the digest algorithm configurable\n\n... and add support for SHA-256 and SHA-512 by default. This allows us\nto start moving toward replacing SHA-1-based signatures. We\u0027ve known\nthis would eventually be necessary for a while [1], and earlier this\nyear we\u0027ve seen SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nRelated-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e725e3b516c1a86b7104fd5693eced62c2a2b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e725e3b516c1a86b7104fd5693eced62c2a2b8a"}]},"branch":"refs/heads/master"},"b5512d007bc48067fc062158b00e9df2e8a66aad":{"kind":"REWORK","_number":3,"created":"2017-12-28 19:12:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/525770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/525770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/525770/3"}}},"commit":{"parents":[{"commit":"61fe6aae81d00597c777a64ac337a8dfb990f0c2","subject":"Better mock out OSErrors in test_replicator before raising them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61fe6aae81d00597c777a64ac337a8dfb990f0c2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 20:19:37.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-28 19:11:40.000000000","tz":0},"subject":"tempurl: Make the digest algorithm configurable","message":"tempurl: Make the digest algorithm configurable\n\n... and add support for SHA-256 and SHA-512 by default. This allows us\nto start moving toward replacing SHA-1-based signatures. We\u0027ve known\nthis would eventually be necessary for a while [1], and earlier this\nyear we\u0027ve seen SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nRelated-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5512d007bc48067fc062158b00e9df2e8a66aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5512d007bc48067fc062158b00e9df2e8a66aad"}]},"branch":"refs/heads/master"},"3445d13b8340247db688c7b75f773d587ea9bb02":{"kind":"REWORK","_number":4,"created":"2018-01-23 21:38:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/525770/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/525770/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/525770/4"}}},"commit":{"parents":[{"commit":"61fe6aae81d00597c777a64ac337a8dfb990f0c2","subject":"Better mock out OSErrors in test_replicator before raising them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61fe6aae81d00597c777a64ac337a8dfb990f0c2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 20:19:37.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-23 21:17:13.000000000","tz":0},"subject":"tempurl: Make the digest algorithm configurable","message":"tempurl: Make the digest algorithm configurable\n\n... and add support for SHA-256 and SHA-512 by default. This allows us\nto start moving toward replacing SHA-1-based signatures. We\u0027ve known\nthis would eventually be necessary for a while [1], and earlier this\nyear we\u0027ve seen SHA-1 collisions [2].\n\nAdditionally, allow signatures to be base64-encoded, provided they start\nwith a digest name followed by a colon. Trailing padding is optional for\nbase64-encoded signatures, and both normal and \"url-safe\" modes are\nsupported. For example, all of the following SHA-1 signatures are\nequivalent:\n\n   da39a3ee5e6b4b0d3255bfef95601890afd80709\n   sha1:2jmj7l5rSw0yVb/vlWAYkK/YBwk\u003d\n   sha1:2jmj7l5rSw0yVb/vlWAYkK/YBwk\n   sha1:2jmj7l5rSw0yVb_vlWAYkK_YBwk\u003d\n   sha1:2jmj7l5rSw0yVb_vlWAYkK_YBwk\n\nThis was done for two reasons:\n\n   1. A hex-encoded SHA-512 is rather lengthy at 128 characters.\n   2. This will allow us to more-easily add support for different\n      digests with the same bit length in the future.\n\nBase64-encoding is required for SHA-512 signatures; hex-encoding is\nsupported for SHA-256 signatures so we aren\u0027t needlessly breaking from\nwhat Rackspace is doing.\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nRelated-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3445d13b8340247db688c7b75f773d587ea9bb02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3445d13b8340247db688c7b75f773d587ea9bb02"}]},"branch":"refs/heads/master"},"bf910ad675dbec00129e481171ff554a191f3cd8":{"kind":"REWORK","_number":5,"created":"2018-01-29 21:25:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/525770/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/525770/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/525770/5"}}},"commit":{"parents":[{"commit":"61fe6aae81d00597c777a64ac337a8dfb990f0c2","subject":"Better mock out OSErrors in test_replicator before raising them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61fe6aae81d00597c777a64ac337a8dfb990f0c2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 20:19:37.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-29 21:18:10.000000000","tz":0},"subject":"tempurl: Make the digest algorithm configurable","message":"tempurl: Make the digest algorithm configurable\n\n... and add support for SHA-256 and SHA-512 by default. This allows us\nto start moving toward replacing SHA-1-based signatures. We\u0027ve known\nthis would eventually be necessary for a while [1], and earlier this\nyear we\u0027ve seen SHA-1 collisions [2].\n\nAdditionally, allow signatures to be base64-encoded, provided they start\nwith a digest name followed by a colon. Trailing padding is optional for\nbase64-encoded signatures, and both normal and \"url-safe\" modes are\nsupported. For example, all of the following SHA-1 signatures are\nequivalent:\n\n   da39a3ee5e6b4b0d3255bfef95601890afd80709\n   sha1:2jmj7l5rSw0yVb/vlWAYkK/YBwk\u003d\n   sha1:2jmj7l5rSw0yVb/vlWAYkK/YBwk\n   sha1:2jmj7l5rSw0yVb_vlWAYkK_YBwk\u003d\n   sha1:2jmj7l5rSw0yVb_vlWAYkK_YBwk\n\n(Note that \"normal\" base64 encodings will require that you url encode\nall \"+\" characters as \"%2B\" so they aren\u0027t misinterpretted as spaces.)\n\nThis was done for two reasons:\n\n   1. A hex-encoded SHA-512 is rather lengthy at 128 characters -- 88\n      isn\u0027t *that* much better, but it\u0027s something.\n   2. This will allow us to more-easily add support for different\n      digests with the same bit length in the future.\n\nBase64-encoding is required for SHA-512 signatures; hex-encoding is\nsupported for SHA-256 signatures so we aren\u0027t needlessly breaking from\nwhat Rackspace is doing.\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nRelated-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf910ad675dbec00129e481171ff554a191f3cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf910ad675dbec00129e481171ff554a191f3cd8"}]},"branch":"refs/heads/master"},"5a4d3bdfc49d61ee6e333c8eb1dad4383ae9308b":{"kind":"REWORK","_number":6,"created":"2018-01-31 02:19:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/525770/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/525770/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/525770/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/525770/6"}}},"commit":{"parents":[{"commit":"61fe6aae81d00597c777a64ac337a8dfb990f0c2","subject":"Better mock out OSErrors in test_replicator before raising them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61fe6aae81d00597c777a64ac337a8dfb990f0c2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 20:19:37.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-31 02:19:18.000000000","tz":0},"subject":"tempurl: Make the digest algorithm configurable","message":"tempurl: Make the digest algorithm configurable\n\n... and add support for SHA-256 and SHA-512 by default. This allows us\nto start moving toward replacing SHA-1-based signatures. We\u0027ve known\nthis would eventually be necessary for a while [1], and earlier this\nyear we\u0027ve seen SHA-1 collisions [2].\n\nAdditionally, allow signatures to be base64-encoded, provided they start\nwith a digest name followed by a colon. Trailing padding is optional for\nbase64-encoded signatures, and both normal and \"url-safe\" modes are\nsupported. For example, all of the following SHA-1 signatures are\nequivalent:\n\n   da39a3ee5e6b4b0d3255bfef95601890afd80709\n   sha1:2jmj7l5rSw0yVb/vlWAYkK/YBwk\u003d\n   sha1:2jmj7l5rSw0yVb/vlWAYkK/YBwk\n   sha1:2jmj7l5rSw0yVb_vlWAYkK_YBwk\u003d\n   sha1:2jmj7l5rSw0yVb_vlWAYkK_YBwk\n\n(Note that \"normal\" base64 encodings will require that you url encode\nall \"+\" characters as \"%2B\" so they aren\u0027t misinterpretted as spaces.)\n\nThis was done for two reasons:\n\n   1. A hex-encoded SHA-512 is rather lengthy at 128 characters -- 88\n      isn\u0027t *that* much better, but it\u0027s something.\n   2. This will allow us to more-easily add support for different\n      digests with the same bit length in the future.\n\nBase64-encoding is required for SHA-512 signatures; hex-encoding is\nsupported for SHA-256 signatures so we aren\u0027t needlessly breaking from\nwhat Rackspace is doing.\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nRelated-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a4d3bdfc49d61ee6e333c8eb1dad4383ae9308b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a4d3bdfc49d61ee6e333c8eb1dad4383ae9308b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
