)]}'
{"id":"openstack%2Fswift~525771","triplet_id":"openstack%2Fswift~master~I6e6fa76671c860191a2ce921cb6caddc859b1066","project":"openstack/swift","branch":"master","topic":"bug/1733634","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-06-01 15:54:25.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-03-31 11:00:35.000000000","reason":"Pete Zaitcev replied on the change"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-06-01 15:54:25.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2022-06-01 15:54:25.000000000","reason":"Change was submitted"},"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-06-01 15:54:25.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-03-10 22:51:04.000000000","reason":"removed on reply"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-31 19:14:09.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I6e6fa76671c860191a2ce921cb6caddc859b1066","subject":"tempurl: Deprecate sha1 signatures","status":"MERGED","created":"2017-12-05 21:55:13.000000000","updated":"2022-06-07 15:57:02.000000000","submitted":"2022-06-01 15:54:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"525771-bug/1733634","meta_rev_id":"fcd90d428e711c871b54d6d9956b2438ba4e703c","_number":525771,"virtual_id_number":525771,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-01 15:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2022-05-24 07:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-31 19:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-31 19:14:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-28 20:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-30 06:18:48.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-08 15:43:29.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"state":"CC"},{"updated":"2022-03-10 12:16:45.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-03-10 22:50:48.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2022-03-10 22:51:04.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"CC"},{"updated":"2022-03-11 04:13:46.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2022-03-11 15:31:05.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2022-03-23 03:51:47.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2022-05-24 07:32:00.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2022-06-07 15:57:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"d466571e5591e41bc5e956d627daca4543b491e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-05 21:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33e6d3033ddb69a2d505eca21f6cb8f7df1fe26c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 23:18:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/71/525771/1/check/openstack-tox-pep8/d6f9db5/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/71/525771/1/check/openstack-tox-py27/f822ad2/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/71/525771/1/check/openstack-tox-py35/4918759/ : SUCCESS in 4m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/71/525771/1/check/build-openstack-sphinx-docs/31114e1/html/ : SUCCESS in 4m 54s\n- legacy-swift-tox-xfs-tmp-py27 http://logs.openstack.org/71/525771/1/check/legacy-swift-tox-xfs-tmp-py27/87cff9a/ : SUCCESS in 6m 59s\n- legacy-swift-tox-xfs-tmp-py35 http://logs.openstack.org/71/525771/1/check/legacy-swift-tox-xfs-tmp-py35/3c16554/ : SUCCESS in 3m 53s\n- legacy-swift-tox-xfs-tmp-func http://logs.openstack.org/71/525771/1/check/legacy-swift-tox-xfs-tmp-func/7224a8f/ : SUCCESS in 6m 57s\n- legacy-swift-tox-xfs-tmp-func-post-as-copy http://logs.openstack.org/71/525771/1/check/legacy-swift-tox-xfs-tmp-func-post-as-copy/13a16af/ : SUCCESS in 7m 10s\n- legacy-swift-tox-xfs-tmp-func-encryption http://logs.openstack.org/71/525771/1/check/legacy-swift-tox-xfs-tmp-func-encryption/e0187da/ : SUCCESS in 8m 07s\n- legacy-swift-tox-xfs-tmp-func-ec http://logs.openstack.org/71/525771/1/check/legacy-swift-tox-xfs-tmp-func-ec/80e1515/ : SUCCESS in 9m 42s\n- legacy-swift-dsvm-functional http://logs.openstack.org/71/525771/1/check/legacy-swift-dsvm-functional/564d148/ : SUCCESS in 23m 09s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/71/525771/1/check/legacy-swift-dsvm-functional-identity-v3-only/30bf7c8/ : SUCCESS in 20m 11s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/71/525771/1/check/legacy-tempest-dsvm-neutron-full/1f7081b/ : FAILURE in 1h 13m 18s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/71/525771/1/check/legacy-grenade-dsvm-neutron/8a7c018/ : SUCCESS in 48m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6e6d121011711289f3e68a6af2f7e518685e7bc8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-06 00:04:49.000000000","message":"Patch Set 1: Workflow-1\n\nWill probably need at least a cycle between the related change and this one -- but may as well have a starting point.\n\nAlso, needs an UpgradeImpact tag \u0026 description.","accounts_in_message":[],"_revision_number":1},{"id":"d22a6beefddf3da1730ce600a2bdf6a2bdf9f99d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-28 04:20:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50e246ee07bcb3578ab170aa1c419fd796dc5498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 05:34:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/71/525771/2/check/build-openstack-sphinx-docs/8269801/html/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/71/525771/2/check/openstack-tox-pep8/6ce047a/ : SUCCESS in 3m 46s\n- legacy-swift-dsvm-functional http://logs.openstack.org/71/525771/2/check/legacy-swift-dsvm-functional/90ae5ad/ : FAILURE in 21m 54s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/71/525771/2/check/legacy-swift-dsvm-functional-identity-v3-only/742befc/ : FAILURE in 17m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/71/525771/2/check/legacy-tempest-dsvm-neutron-full/583164d/ : FAILURE in 1h 11m 27s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/71/525771/2/check/legacy-grenade-dsvm-neutron/c702edf/ : SUCCESS in 45m 35s\n- swift-tox-py27 http://logs.openstack.org/71/525771/2/check/swift-tox-py27/d7673b7/ : SUCCESS in 7m 34s\n- swift-tox-py35 http://logs.openstack.org/71/525771/2/check/swift-tox-py35/eb76d62/ : SUCCESS in 2m 42s\n- swift-tox-func http://logs.openstack.org/71/525771/2/check/swift-tox-func/c6282d6/ : FAILURE in 7m 58s\n- swift-tox-func-post-as-copy http://logs.openstack.org/71/525771/2/check/swift-tox-func-post-as-copy/35518d3/ : FAILURE in 6m 23s\n- swift-tox-func-encryption http://logs.openstack.org/71/525771/2/check/swift-tox-func-encryption/f37f909/ : FAILURE in 7m 21s\n- swift-tox-func-ec http://logs.openstack.org/71/525771/2/check/swift-tox-func-ec/32091f4/ : FAILURE in 8m 41s","accounts_in_message":[],"_revision_number":2},{"id":"10c7cb63a3d062a35843df8ca83e4cea7eeba2b2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2017-12-28 19:12:22.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e5d680835b349d6f4aa3fa81d0951b7aba3ed848","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 20:25:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/71/525771/3/check/build-openstack-sphinx-docs/b25c691/html/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/71/525771/3/check/openstack-tox-pep8/dbb1e4d/ : SUCCESS in 3m 29s\n- legacy-swift-dsvm-functional http://logs.openstack.org/71/525771/3/check/legacy-swift-dsvm-functional/1e2f39d/ : SUCCESS in 22m 01s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/71/525771/3/check/legacy-swift-dsvm-functional-identity-v3-only/27b9c98/ : SUCCESS in 30m 09s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/71/525771/3/check/legacy-tempest-dsvm-neutron-full/b366cb0/ : FAILURE in 1h 11m 31s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/71/525771/3/check/legacy-grenade-dsvm-neutron/27e7e56/ : SUCCESS in 59m 02s\n- swift-tox-py27 http://logs.openstack.org/71/525771/3/check/swift-tox-py27/e3599d6/ : SUCCESS in 5m 40s\n- swift-tox-py35 http://logs.openstack.org/71/525771/3/check/swift-tox-py35/e6eac8f/ : SUCCESS in 5m 00s\n- swift-tox-func http://logs.openstack.org/71/525771/3/check/swift-tox-func/ec1f616/ : SUCCESS in 6m 29s\n- swift-tox-func-post-as-copy http://logs.openstack.org/71/525771/3/check/swift-tox-func-post-as-copy/21cdbac/ : SUCCESS in 7m 05s\n- swift-tox-func-encryption http://logs.openstack.org/71/525771/3/check/swift-tox-func-encryption/59b144b/ : SUCCESS in 6m 23s\n- swift-tox-func-ec http://logs.openstack.org/71/525771/3/check/swift-tox-func-ec/5e1d490/ : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":3},{"id":"eaf119bc034072b0d477f4aca143917872d03825","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-29 04:23:51.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525771/7757/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525771/9736/ : FAILURE\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525771/8151/probetests/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ba4dca8ab191cc956d0f19cb3acbadc2b25fb6f4","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2017-12-30 06:18:48.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525771/8153/probetests/ : ABORTED\n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525771/7759/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/525771/9738/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"304027d1c3bdff6c57b0accebbca9b2dd76c8265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-30 06:11:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2d587f727dc6d594e75e7ee16c9b59edc055178e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 07:23:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce8d7c4396f47e6be0c6e86cb3eff97 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/0f7de87dceac4346a87628d3740926db : SUCCESS in 44m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94655aebfe8241d3a8482abb4a418582 : FAILURE in 33m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea2e874022cd45bba63f2a5614f50243 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba17d96361a74ec7ae0ab943a0d9fe90 : SUCCESS in 9m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1956e437721248218f78c7a62c313aaf : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23cbbb99cef14a89972f94b402862292 : SUCCESS in 17m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/636874d997c0451f913f5e70d59bda22 : SUCCESS in 16m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e08792446a3472791c502a4afe9d2d2 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/69ed02ee443c444d93bd771ea654831a : SUCCESS in 21m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/85b7b9c16c424fe2abae5825d26f97b7 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45d26b41b15d42f3a99a3c28601d775e : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3739a09905fd48879c7c1ea17e6d16a8 : SUCCESS in 18m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9c711440fb014bb690caea73c25f3282 : SUCCESS in 22m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7842bfe1d5634dc4bb2786275da3fcf7 : SUCCESS in 12m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0dd86b4617494b34babb594f7f9db4d7 : SUCCESS in 21m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c2f0e71dd1e41959e71bdf30363b914 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e6e26b904d845719bec9c97cac0bc1b : SUCCESS in 27m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c04a03f01d524e7a9610b9c1833c0294 : SUCCESS in 43m 29s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/64718e01846841e7bfa07fb2fbf05988 : SUCCESS in 58m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3db5d43cc134262a29d28d8e71d29e6 : FAILURE in 25m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ea76a05f2384d53bcbd0b2544e92922 : FAILURE in 25m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b114c51dcb1a4aa2a61e77e7867a6b07 : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7e9f11831d4637975b6b14a996c1f3 : SUCCESS in 7m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2fb2af6589d24ef3b03b5efbec5a50f3 : SUCCESS in 19m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f681660fcaab473db1a54ec339a9eb3a : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":4},{"id":"2ec6d2346263bf442e26731bc502a68a4db7ccff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 07:56:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12548d500c4b4a53bf5695afe590f80c : SUCCESS in 44m 33s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c03415700ea4449aa47a5bb12606cfa4 : SUCCESS in 50m 22s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/ad1c7b7f73b14739821fe8880b76200f : SUCCESS in 1h 37m 00s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/78a739f69cf542afaba6e40760a1cf6a : SUCCESS in 51m 38s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae7c9dd206d747669be72f91d4b6cdf5 : SUCCESS in 50m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8f3ead4965e899ba98f56320aa6cb768a01bace","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-08 15:43:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b25e9a52f00be930d41cced52f11763b13db663e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-03-08 16:25:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"0a42cdef02c1130f7678386c311d6fba522f2c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 17:24:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fd8bf1ea644b9f926c27a066407105 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/dbefa8bb53a7421ab58655ee10607a0d : SUCCESS in 48m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/700b46672ce64eccb85f6b035a6c7fce : FAILURE in 31m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8431c1a3cf294323b908b244a92081b0 : SUCCESS in 41m 47s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/46f5e80e4ef14784bb34a59d228e1913 : FAILURE in 19m 57s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9ff90b70967b4f50ae0eb0cdbc0ad14a : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/250accd4661440979fd5602de2599186 : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fda93e0b66ea43b988dab64153dde9f8 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f3fb0906583f48ad9ebf6a9ea438aa62 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7c7f87334814267a282974b87a977d8 : FAILURE in 16m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61a1d227b00947409a0b1e400263516b : FAILURE in 15m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7b814d2b9b5c4b1cbe21fd8b49beeae1 : FAILURE in 19m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14ccce7bfaff45dbbcc453d65b54584c : SUCCESS in 21m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8010e7bfc6b64cc2b679e333bc99290c : SUCCESS in 24m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/020a7771796d4e5394c931a5907a5eac : SUCCESS in 22m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/920e68ce599243b08ad34c5f9a369e7b : SUCCESS in 22m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/823d2f7bbe9a44a5a2ea9db4ce453a76 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/abd8d13d91804b95a626f2fff59f3642 : SUCCESS in 20m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d8c282229614ddda42ba006c57f2fa3 : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46a8d298d5164ed2b6d96f868cd829e9 : SUCCESS in 28m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d8a82818d324bf6b5e733540d460ffe : SUCCESS in 42m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f3da8a1563141c19657124e624017f2 : SUCCESS in 55m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd9c2979d6db4b8bbe4e786671b8d7c9 : FAILURE in 24m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1baf47dd5a344721b2486495c090b221 : FAILURE in 25m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2e7d3f558d74ef5a57323ac764d1052 : FAILURE in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9a87889ee741b394f7086b446c62d7 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7949cb381d7b48a2a8c0de80485cfe59 : SUCCESS in 28m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31aa65db633445e1b837cacbc66e9131 : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":5},{"id":"6a02a97b94ee389f35fef8460566c197fc73687f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 20:03:57.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1290fdff530145068d8afa4528e85751 : FAILURE in 1h 00m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe873e19698f4ed6828014b3ce622603 : FAILURE in 52m 45s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/6147e5d6f5d14a8bb0761fa3d9c802ff : TIMED_OUT in 3h 02m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dce53723babe47f8af7757f012566bea : SUCCESS in 55m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/129c111693064b02a54789968e60adab : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":5},{"id":"fc104bdf06a40a1ffe445e6e6506a6e913458682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-09 09:52:30.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"1a5309cb6a0588a2ccad98aaa6613d8c000ada28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 11:02:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adab653eabd443368df615a1902f16dd : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/810c11a192374b84bfedd7d8ed908443 : SUCCESS in 40m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3ca8f8f1b1d416da76a14cbc3949201 : SUCCESS in 48m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19f14b799c0b42e3a5c99f3c4048a857 : SUCCESS in 41m 17s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bded9e18bbf44fa3bb66f0b540697126 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/517ed6e59a424d6899021c88123a9814 : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cf075013ada14e1086330e5a511cace9 : SUCCESS in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30b6322a64a24482bdd945ef8b66aa35 : FAILURE in 7m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b23294eb3c8e4125928b17a0ef8c11ed : SUCCESS in 7m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b2020da027eb4c1e97255ca45ef3783f : FAILURE in 18m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/359b7a75164b4343b8ce0463c86f072b : FAILURE in 14m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9f54c26dbc2489795af2fab058133d5 : FAILURE in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e005b12f3a694330ad99bfdcd9980db5 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1dc047e961934795947af011ac88d79e : SUCCESS in 20m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/50cfcb6deaae44b89676f3dee226a4b8 : SUCCESS in 17m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5e76cb49af5d4b19925d0e5e6ea7b39a : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/59e1da6a937740858b13f9411213df18 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4562186987cb4385a603d24780032314 : SUCCESS in 19m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/472f4ef8fd8544879c09c9075ec04fb2 : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2e341e2db254e788e59249d262241d9 : SUCCESS in 26m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7fd3c683775f4cd88006cbeb4d35322b : SUCCESS in 53m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed6762f9bde34bc49dd282a22bb801e1 : SUCCESS in 1h 08m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e386b87bf34848ccb88777e301f88c29 : FAILURE in 29m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/168b910c577f4a9ca265574ebb96aed9 : FAILURE in 29m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca3a913b3ed04f4eaaa6058821088c62 : FAILURE in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fa1e9f648c0483eb81d2ec44f85a584 : SUCCESS in 9m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/40f755e6188c485eb0f8c8daf4772f4d : SUCCESS in 31m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0df15b4b5a7546eba2ed793d44b7345f : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":6},{"id":"2f722a09632c97d31844bdec168d2c131ace0543","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 12:58:22.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1d4f46bc0e84e5c81fbb29f9b12cdf3 : FAILURE in 41m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ff700209bb140a09c5dbde170eabe83 : FAILURE in 40m 18s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b28596259b6646768d584f9eddde7436 : TIMED_OUT in 3h 02m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d527dc809e3a471283e2d8a1be608501 : SUCCESS in 46m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/608ac85485dd4d84bf209a523433337d : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":6},{"id":"707daeb0bc7e7eb6557c1efc6617ffb48780e989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-09 16:07:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"69f59b4aabd1a69c8f94cedf9ee9a1a8ffab442a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 17:09:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf2c26c227244e7b063e455731c0fed : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/3210adc56d6e45b8a060fea4d2055afb : SUCCESS in 52m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c87c3e3f9964678887f0838082af6b3 : SUCCESS in 46m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0eff2d51a636454aa5fc750479e706b4 : SUCCESS in 49m 38s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/51d4c2d8bc164df8959d3a18ec56ef45 : SUCCESS in 22m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/604b1377d3b642249a0c9bf842a3951e : SUCCESS in 24m 41s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9041655e37ea40489136068bfc55772c : SUCCESS in 25m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8965d0f00d844340a2c8b6dc38ee7526 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/313e5cb082434956a9298ca008271ea2 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a68a4322e5aa4c38bb3c2c20abd29226 : SUCCESS in 22m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8c3685279ad4eaca02d4d71bf1a4bb9 : SUCCESS in 21m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6032e3ea72834de299a90d952465750c : SUCCESS in 21m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a501b668e0142f8985bf2a12697a9a6 : SUCCESS in 20m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9bca548011e54ab7a85fcd18d8ea9672 : SUCCESS in 28m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/388d92cf21ac48f2bc4ab5db2ee12136 : SUCCESS in 23m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0925fe71190e45db91dd25ba346d1831 : SUCCESS in 19m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dbb60f9c76814191ab224db64ecc43df : SUCCESS in 18m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/224d87a3987a4ac0bbf141853558e8a1 : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5786237a31084599958e073c0bc470cd : SUCCESS in 7m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/352581393c9644c7bb09aa20633659c8 : SUCCESS in 21m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bd60b47b146b41a0a5e67bf05f74ed7d : SUCCESS in 45m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6a88a789ebf4d7a849eb073df134546 : SUCCESS in 51m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac7ceba1e54e481b982c116ce0902915 : FAILURE in 24m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/144d2ab407c74900989615ded7822fad : FAILURE in 23m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d70c6a0a292442f86307467828fcb14 : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e5126c8b2d1451b96c4b617e9b8b906 : SUCCESS in 7m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f309b36ab6bd401ca26d8802b9b3c8b4 : SUCCESS in 23m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfe0a44d502b4b9aad1ee37fcf17e851 : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":7},{"id":"2af56f4f9373b58ebecd3a549d06b7dca2ec3d18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 19:13:05.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa7dc1706be7446799a9c8c65a02c71a : SUCCESS in 41m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b20f358e4063422bbfdc7d27a7fe170e : SUCCESS in 41m 24s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/a31b6a6242a74fceb3be334677c0b764 : TIMED_OUT in 3h 03m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8922f025b8f4097987683a1658947bd : SUCCESS in 47m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f73fb85818b419bb31fe4cb1a847a79 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":7},{"id":"f9467159fb3e16e8421040bc4380b9f653d66cda","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-09 21:52:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ff8495ae12aa5888d58839a16ee3c94931124574","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-10 11:36:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e9731615a26be093989a3e8127bbdefb1c605a48","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-10 12:16:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"06797cb7eaaf15445ebafe62b705c6028ec21246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 12:48:13.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887520427ea9498697d5ac285b8276ae : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/6008a703e40b4be2aa2612f23f7b8e76 : SUCCESS in 52m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e941564c66840a6aeb6fe49e646eb91 : SUCCESS in 48m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53652057f9cb4598b498e99672e80600 : SUCCESS in 41m 04s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d5cda2815f1847079866a03317134ed3 : SUCCESS in 23m 04s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/afe3ccfeb22c4ca08c9a4666e4d152b5 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/60cc155a99634481a5bc028b2af55e03 : SUCCESS in 22m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78bbbf5fbcd7458a9c3ac8a2f905594b : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a5970038573c43bfb5794eac9a58e6d4 : SUCCESS in 8m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7bee324981064615a7a3f4a70fc287c2 : SUCCESS in 20m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51046c516a6a4160bf57a3aac9d1121c : SUCCESS in 20m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9268b1d54e654f13a7f6c48dd3b22668 : SUCCESS in 14m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e6ced53dc46486d8e8b7be9b6127142 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5554028e03e4e29a1448e685e6bcd1f : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37c9024649064475a93d41ce6b9f16d6 : SUCCESS in 20m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ed84db77dcdd484d905645a62aa242c4 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8748b62a66e848de88f61d780267f814 : SUCCESS in 18m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/580766d90bbb4367bcd51e15cad359ea : SUCCESS in 17m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fdff1a23cde241a2be69eb4628f74d2f : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1f9e413540f4610b19031481afacca9 : SUCCESS in 21m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d106b97234f441a299292d312e1029fc : SUCCESS in 45m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/af55cc88511a4b538220f640f4120f25 : SUCCESS in 58m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3781af6f8dc047b2bbda1e515a6da922 : FAILURE in 33m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad6cd1314995493f95f4beefaf64b231 : FAILURE in 33m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/276c5f742b7346be9807737a2c74bedc : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b903b4a78c4e65bbf7d78ac3033069 : SUCCESS in 7m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6596d44743cc4de18b840247bc3d57cd : SUCCESS in 22m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1444f017dad34e849c03c72c8577c6e0 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":7},{"id":"81e999d8095449cf28f4555c9279f6c749937363","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:53:53.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be52e804926447468b6c77f4193003ae : SUCCESS in 1h 01m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4088225f9a0c42bd91da57f3553bb25d : SUCCESS in 58m 13s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b2edf06d9d0f4fe988105a6c768ed7c6 : TIMED_OUT in 3h 03m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/71c833d6d17f41d0bff10f964cdc269a : SUCCESS in 1h 12m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9943e2d1f2d1471392015413ef3236e7 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":7},{"id":"05bbf1c4114098f8234b8f6dfc1eef2fbc8ec506","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-11 04:13:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4ff395e564a5f9fa61d0247b79e33b5ea3a69f40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-16 05:22:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"23e0d090dfbb4fedaf5c80bd8229bd8af5ec278a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 06:35:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48d7551c23648baa99cc498e0a853ce : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/aeb9ee5e1d7d4539a35728135dee346e : SUCCESS in 1h 03m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d86f44436e134cfda203c93c1b151c28 : SUCCESS in 51m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8e00e19015c4dafaa57cb1c8dc0a6c3 : SUCCESS in 56m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/70873ff3cd5746129c73c17932941f2a : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f8b24e89a60c40ce95c7e99c7264855e : SUCCESS in 27m 44s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/412bbbcdc3a3453e96d479b334269808 : SUCCESS in 26m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/198d313ecfae4701b3ffe933d69da605 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1a034709538d494684521af90ab74611 : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d80701050f99498ea28123442e7e67fd : SUCCESS in 23m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80f55b3e4aec42a08b69960b8e39a9bd : SUCCESS in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99ce86f302d247c39d19eda56aa71cf4 : SUCCESS in 21m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db070ad199f44ac6980aaa75b5e24d7e : SUCCESS in 30m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e26111bf61db44a8a0f7a424ba50e357 : SUCCESS in 21m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2992930a647640a59bc4eaa672abe3d0 : SUCCESS in 20m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/372954f470d347438bee39ad2e66a35d : SUCCESS in 23m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2fa20428d10b433cbd96aceb42cd4bfa : SUCCESS in 26m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/14c4c5b7940543eda32b764b951c151f : SUCCESS in 18m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/569dc411e16a4801b25ac11ac1eb7a41 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54856378f74a4abea6ca0231702f5c33 : SUCCESS in 25m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/586622a35af54feb93e708e2e5daff2c : SUCCESS in 47m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0812dc7018d489e8e9d9a30a76a1772 : SUCCESS in 1h 09m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f6030f5f92843bc98d46c226eb07fe9 : FAILURE in 34m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/672b35d0261e4c6d84183c65ad26dfde : FAILURE in 25m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2491cfd825c946f4aaa4580c161f31b2 : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7384b2b8f8c45bab40b31de3b54b39c : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/361c621e617643f0816909e84504810c : SUCCESS in 19m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db3bc2581dfe433a9d4014d9c58f372f : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":8},{"id":"62d5cf8e93795ff8a422948a5cc35534bb55720a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 08:27:47.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b543ee3068d943c48db1a3faacacb2d6 : TIMED_OUT in 1h 32m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fac6c7479cbe4a22a5d9449b87a9f84d : SUCCESS in 31m 13s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3e49de7c29934e0891e4c700eae3b874 : TIMED_OUT in 3h 02m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/baecfccdf5f64612bc24427cf4c6a76a : SUCCESS in 59m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f51df868882d4ac580bbbe11008b78af : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":8},{"id":"ecba02b0226a81e44a5cac9935d450e4ac2f11e2","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-23 03:51:47.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"37dbd6b7fee9cc35bc360650b80c42b41b0858c5","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-03-23 13:50:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"56bdd7ea2ce394e6fbf41e57933099c775926bd3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-24 04:39:37.000000000","message":"Patch Set 8:\n\n(9 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8789a8aa74935fc2f7c2a07807fa5d089b468878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-24 04:40:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2cdbd0c8782411e25117db038b1b3aa5995a2d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 05:44:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7a651a8d1b4bd49cf165a8309163ef : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/43c1186fd5d94e9d9795f4ef722fa7da : SUCCESS in 52m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d52303cc206542e89ce8e13178685225 : SUCCESS in 48m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9789ff85f5e40d285ddb071eaf1503c : SUCCESS in 47m 49s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d74911f783f94dd2bbe713c62eca2a47 : SUCCESS in 25m 18s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/37bdc9db3ef047108e9c848611df4458 : SUCCESS in 25m 33s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0a06d79c31ac42509a05f250487c7d9b : SUCCESS in 26m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73480f4ecfe3437db99ed8512167324f : SUCCESS in 11m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8e358e4b9ca34b48b77b63c791957ed7 : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc70915fce0a4a6996293860ad2eee30 : FAILURE in 22m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d139a34c09564bf1aaf593cd06de4981 : FAILURE in 21m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/29c56d861b2e4e2396d417a36b054800 : FAILURE in 21m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a663bdf10414552bfc871b620e5b8dd : SUCCESS in 22m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08a455a7a84a4886a09d430f21cd69b2 : SUCCESS in 22m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0790947e796427290b60f1eb10c0b20 : SUCCESS in 24m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/21123217193d4c8bb7781006841126e5 : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/849e060ade7f4b3bae18ac39f9550038 : SUCCESS in 21m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/da4d130f957e4ceea4045c174af4b5b4 : SUCCESS in 22m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ebb39039201f488591594b963c5919c2 : SUCCESS in 17m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f509184a95a24758a82cb5c80faefabc : SUCCESS in 26m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b6e1c6cb9a642879798b35cd30e569e : SUCCESS in 59m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fedaf28ea334bca85ab0474e87e6674 : SUCCESS in 1h 02m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9b5e27c1acd54360b37ab8ba74274394 : FAILURE in 27m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e951893f75c457c8d27eb44082b536b : FAILURE in 28m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2662d2b0a454ddfb4819231be3b2239 : FAILURE in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09b3fbc0aee4a139b6363d7ef985ed4 : SUCCESS in 7m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca64260ba82e4fc9bac67c3034c5b1aa : SUCCESS in 28m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f95a8d69650435ea65f35b6042b6304 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":9},{"id":"0131ce7f0d74a307ec89dc0c09ac779376ba8a40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 07:45:20.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ee220ac986146b3a156bb84b4f0e4b0 : FAILURE in 51m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff8a4643addd4beba5539b148d72ad5d : FAILURE in 50m 24s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/86ea31f491c64bebb78102d41aa70b0b : TIMED_OUT in 3h 02m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2dd562b78e17407195ff6fb1aae6b463 : TIMED_OUT in 1h 33m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed745d338d1840a9a79dd476757fb669 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":9},{"id":"2594f121b44251f9da2d8536d705e773747c5960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-24 20:48:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e827be036f309115e5d0ce9e7dccfe0bc7f856e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 21:59:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f4a0a6ac7d42b4a45c402a59ff9dbd : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8adebdbabb1a4bf19f6b9f709d13baed : SUCCESS in 54m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a629ed784e864459995cfa82c27d5d94 : SUCCESS in 45m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4511eba08ee84dab9afa926640a06136 : SUCCESS in 51m 18s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e6431bfcc18745589e0ac19f42e523df : SUCCESS in 21m 30s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/97938e8e80e548059e3e358cc3345ea9 : SUCCESS in 23m 57s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c2115e5fb78d418c9f008a1fe367f7d9 : SUCCESS in 26m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/461672d9078d475c910e72df2d98cdea : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8d8e174f840044719414738c289ae01a : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ebffda3c57d4fad80be35ca53794ab7 : SUCCESS in 21m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8e4ae1e34d74c12abcd98af36d721f2 : SUCCESS in 31m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/27e5efc2da2b4ca7bd76c3fdbe412084 : SUCCESS in 19m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28521459c5414566b885cfcbff8e5f8d : SUCCESS in 25m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0d471a24bc94879a70e34c7a8a68fb1 : SUCCESS in 20m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8bf79bc2ca2048368e7aa09368ec9302 : SUCCESS in 24m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52031722757b40d4bc7e70134ba32a31 : SUCCESS in 24m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3f1d8473ad894d52ba69526904c20588 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/461f5338d3da4af5a1dce059d2b1ef77 : SUCCESS in 19m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9c3e1d582694aad97f93e853459769d : SUCCESS in 8m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55ada4593ff741b691e0bc5370eea54f : SUCCESS in 27m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7c9e4b647be042da937b6b16ecab5b07 : SUCCESS in 56m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5291721daca442a08b937b10e54961f0 : SUCCESS in 1h 05m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46dbca8474ca4e1388f4ea9014e5482d : FAILURE in 30m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9d43c1544c34794887e2fc46ae15b6f : FAILURE in 24m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27a2c255977b49a18ee0d40b21c51a85 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f633f5ac3774790833e1c6077dcf851 : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0681fd75244e4f889368c6ef23052b27 : SUCCESS in 29m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e5321e0874a4afc86ad70a3f0974521 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":10},{"id":"70af5eba2d3adff0ac64581c7984c86bb5c70c7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 23:29:06.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/578f0e9b67de43d18859b1b79ba7b626 : SUCCESS in 38m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c3bf4b426754a848b67d958b17364a8 : SUCCESS in 48m 25s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/0686a7c051cc480f95b8bb9662c5f5b2 : FAILURE in 2h 38m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d1269e23f4b40409764df8402d70c32 : SUCCESS in 41m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9f0eabad8874983bd46604e1fe9c9c9 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":10},{"id":"d3acab593ee6a70d6b05727ff228b41c0c29a734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-28 21:39:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"afb354cae80d5f4785be5e6bfdead01b3bc26459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 23:25:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e84d9770c24a9abc23f2080eb143e4 : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/dc4d01fbdc42482c95184e338e3e193a : SUCCESS in 46m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1322146853a246b3b61ae536c241824f : SUCCESS in 44m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a96487fdaf0341c5a2f8903b1c5df964 : SUCCESS in 42m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d98f8d4b967147baa6dab053bc748b32 : SUCCESS in 23m 59s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eb037a03097c458fbd0365338724536a : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/798b5ee1786c44aba4373b8edbbc21b2 : SUCCESS in 22m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/006733752106422fb0fccf6c1b4ff282 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3a0623668c3745cab54c4e39d95dbc24 : SUCCESS in 12m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/281591e04b384502a452310971c5225a : SUCCESS in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a39abd78a91c4856a490c479248913d9 : SUCCESS in 18m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3d9e010a97f4474823fc87d337e02f6 : SUCCESS in 18m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c131fc49246f4d1690344c30da95eb4b : SUCCESS in 22m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3090e8cfff974a2f985b18b408756849 : SUCCESS in 22m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e5727f99bef45928fd2b29aa8bf319e : SUCCESS in 21m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/51ed3856a4f144d094067ceb58d2d7cf : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f7cdcaa5b9424a1b99e402fd0f015582 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/73c00857af0a4e74b0f400abe651212b : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3df11b4835a146fda18154655e791952 : SUCCESS in 9m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4240c79281e74699834a0fe005bbf16b : SUCCESS in 26m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e427bd0261f24708b735e97c2e4cc59d : SUCCESS in 53m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0081df18d03f453aaaa967bbcc82809e : SUCCESS in 57m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d83874387baf410688a6689a6f5f3f00 : FAILURE in 32m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d34ac98990c4fd99d9a5d3f203143a0 : FAILURE in 33m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/263dae8d231640169b67539c8561e7f7 : SUCCESS in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a3ec87f59d7456097886db89b42fb4a : SUCCESS in 7m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06d0653ac11a49c1ac3559868119022c : SUCCESS in 21m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce45288110154743b9276425ea71f7e7 : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":11},{"id":"851c452bf1067c498a78bda8ac4ec8dce8a12cc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 01:05:50.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60e7299607544c07ad4ade02c15495e7 : SUCCESS in 49m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/040feeaea0c34396b28b8f4d53ad2650 : SUCCESS in 55m 31s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/1036ad2c4ee64ed0bbd4dd6abd86c015 : FAILURE in 2h 39m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c93dacba40374e88a6ea3aaf14fbd2e4 : SUCCESS in 1h 04m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbf015d92adc4e4fa35a6103a0a3376c : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":11},{"id":"1023b1b6b3a78d8fe1e388deabdd75d83e26ec63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-30 23:07:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bead447a21e739cbc9974cb1d310d533f1c02482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 00:06:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45933d160d134eebb4e510a05fb613eb : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2ad0e9d351ab40b2b9f1db8749dbcb42 : POST_FAILURE in 51m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d01345e76d3a4caeaff8a50b74b297a2 : SUCCESS in 42m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e64bea86982a45829a974f70b547d8a6 : SUCCESS in 48m 20s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0bbaf312872e4e6da5f4ef36dd7e2465 : SUCCESS in 28m 10s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7e20e22d5d204f54ab8fb7a0cef205ea : SUCCESS in 29m 03s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8c00cd4a8b1743ed941fb5c1f693712a : SUCCESS in 31m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/269f948522874524b49fd471533e4053 : SUCCESS in 7m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/33546dbcf98b40aeb78ed9aabc788242 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1568719a71b245ccad25531419a5de46 : SUCCESS in 23m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e5d1e29d0f649c48d29ac8eb1ebd1fe : SUCCESS in 23m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/be2b52e2f39b4518b1e436ab56d9d0aa : SUCCESS in 23m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/69decd33b77e4f7c951e254d44442c37 : SUCCESS in 26m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4cfb604bb624ffebcf21d74119b65b0 : SUCCESS in 23m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d88a65b36b7640d6a25934481030ab9e : SUCCESS in 27m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d36d14dcb424d0aa9bcbb2e79f3ae8e : SUCCESS in 21m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f29dd20ab99b431ea8821ce726fd81b7 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8276ad9ce51a4d5f810d84ff8d4062a6 : SUCCESS in 18m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d10f391eefce4dc695e868e8bf8b07cc : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3343ce0fc72d497ca34a6e24c0c567b5 : SUCCESS in 17m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7b98996ab15489cbb92238423455e23 : SUCCESS in 48m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/428b49089d8842bd89ef28de4e3a7228 : SUCCESS in 57m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84f92e7f818b4f46bb0c525b5cc28494 : SUCCESS in 44m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98730229bf3d4b72abcf7952b2fbf47e : SUCCESS in 43m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72c196890d36471bab9e6860256db716 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61df07ee833e4168b30a56c8cb135f8f : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a653625962b4b508dbc2f05cb0469eb : SUCCESS in 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eadb73b6238c4acf825477dd172187a3 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":12},{"id":"e7342075bf220b50d29ba0529e7ecb549ed0114f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-31 01:28:30.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"aa413ec947b2fc817a59ce65cde701063a33f2bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 01:59:06.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a54133303da4ca093fa709853172aa7 : SUCCESS in 56m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3c17ba75cbe4a299bb0503cf810742c : SUCCESS in 53m 04s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d866541ac95d470f843ca4d39531c419 : FAILURE in 2h 49m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3c2904c8fa94ef785c7dd86357658ef : SUCCESS in 56m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/520db87fda5f451eb6829cb798fca12b : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":12},{"id":"a4f40764eae5fc7598d6c8ff14fe9f3fc31990c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 03:00:48.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0670a7309f5648b59d8d4f76c62ec7b0 : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b3f797d5235d40eda3e8848c4c7182f1 : SUCCESS in 59m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/34a6fdbeceb94e3b92b90ce4f61f99c7 : SUCCESS in 49m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/376764b0a60b4cdca13dd33b345dbeb4 : SUCCESS in 55m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7395f3c90a5a49198cb107344a73e2d0 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c9cd9431861541c18a33efb522fe4b62 : SUCCESS in 22m 27s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/92c0633b53104e7c99fb64b1b9291efb : SUCCESS in 18m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b94c91fd06ca4ade9af16a556e7fca7e : SUCCESS in 7m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89167673ead7457ea5a95ed318bbf6f9 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c118b9ed33ad44afb9dab0d33a93969e : SUCCESS in 13m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44975b7938724796b6225400737b1eb1 : TIMED_OUT in 1h 31m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a6a85bee94444fb9b41d698b65ab476 : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ba05f699b0544efb9c0f7f2bcb97406 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a43b80c96514d1785d03f4e646c68d9 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/403f586b481c43f7a6c05d6f9a4a82fa : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7af0c6bce1d84e1e93ba142307057670 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e5cd3a3597de42f4b735de3dc471f72e : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2700cf41a9bb4603854a0e407cc97737 : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5aacfa56d3f49adb2135a7598349451 : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df12a764f2df42c8a23be9550ca8595f : SUCCESS in 19m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c13ff4beae34b9191cc60404f31ecc1 : SUCCESS in 54m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b337ff035479492dafdc54234f5e91b1 : SUCCESS in 1h 00m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae3041017e0a44a58631e6c3f6901bef : SUCCESS in 29m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43847efbe5244756a9269075864b754e : SUCCESS in 44m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2459215f3d07437bbc9bbfed10345e8d : SUCCESS in 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ac5cef909349ef9597d5f8251884a2 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3330c5bf69c44a7bc5634e214e8e3a7 : SUCCESS in 17m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/750bd9a1b1314e7faf7a6fc4cc767701 : SUCCESS in 49m 45s","accounts_in_message":[],"_revision_number":12},{"id":"68007f889debce46a39677f46522980d7b8e2ad2","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-31 03:20:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"887622e2a6a0c91cbf04b98f04c390cc72de5555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 04:55:08.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b1f5b3b01a540c7bb076273ddd4a568 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/26854629418b47bcbf0724baa2b4574f : SUCCESS in 56m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7365de0196934c4d917467429e4690af : SUCCESS in 51m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b9350666c6c4b5c9ac5202453435b6e : SUCCESS in 40m 09s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9dd6b52a8d55425688829e887fe9c3f4 : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3c056b0f24fb419590598706470bbbe1 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5a382de009fb41cdb224f79ba850a58e : SUCCESS in 23m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c78450f428da4c6e8dc716cdb4c57e31 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ebea11708ff7482291d78d6212231238 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32900f9219ae4423b381b8fe65259e5b : SUCCESS in 20m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6210e601b56043b9b9366d506214fdcc : TIMED_OUT in 1h 31m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e77373b650ec4517aad34ce4da1ac494 : SUCCESS in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f9fb25bbfa440baaac3479ba035b341 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73cc3b1b4d9a4a3ca5a18e223ece30bc : SUCCESS in 15m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e876921a500243e9a8c9e859bc8cb9de : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1009355e480d45a296a98f8844e47bc9 : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/087309372cd8426b87dd14c3dd9d299d : SUCCESS in 38m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0091e20db0094e5f9d07b6f063f3f183 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/739a0e1a603a4ce0ba077a19347753eb : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88fe2ef6930c467ab449634eae82c709 : SUCCESS in 21m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7fd54541ea054ba0a6519561e70d6b87 : SUCCESS in 38m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f536a5361c2745b88722c02aaf97afc2 : SUCCESS in 52m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed79c3403e2b4d348eb20f6fa3801c3b : SUCCESS in 56m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5885bd3745174fe38c4251431465525b : SUCCESS in 56m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/098a9a30bcdf4dcda557e76ee88a5e8a : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3485378374245cfb01465663454bdc1 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1cf16cb572e452d87c29eccdc567aa6 : SUCCESS in 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f66ef45b57b148aa804b2d2ee21c4c27 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":12},{"id":"be8fe370ea770104a86ebf48e3df420c130b6645","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 06:18:57.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2cf4861bac3488bb32f791664afeccd : SUCCESS in 40m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00dd2ad3b9074026858e3272d8717e94 : SUCCESS in 37m 39s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/00ef2bfb84ec492fb5ee6c374604ec33 : FAILURE in 2h 56m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/00634b392fcd4e3f88eded900ea2cc01 : SUCCESS in 51m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/22abd7f7958f4b05af5635cd91101fdd : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":12},{"id":"14140b4b69b35ec20a0f4b6a9041829dda489846","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-31 11:00:35.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c5912f47770c1e53de29e15c8c481714e84ee481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 13:41:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247b6f39a16e4338aa91bc9997300b67 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/43431582955c4d82a2e2ff7b27c25d59 : SUCCESS in 1h 05m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92a1ce7da4554dc4ac1d547161ce422e : SUCCESS in 45m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60a69f5f5cac4e4caaabc45b4f9e0b56 : SUCCESS in 52m 47s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/64faeaf5e01e408189d6859e510b934c : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/46ebd8c5d43f497c97034b9675dc8d2c : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/43ab32aa4a0c448db6598de641d3825f : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91d0c1f7ea334350a315866e5b3e3a5f : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/093bb183018845febea0de8472604acf : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ac156c63bcb4d5c919c54ad2f037aac : SUCCESS in 16m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/145365eff21d4c0a9d6353f9c65848cc : SUCCESS in 16m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dcb4c6454c4d49479eac7c8a6f9445a4 : SUCCESS in 15m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/037fb44280014503ae8deb5f6978d055 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe032ab689fc47cbb5d56691b3fbd848 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76d27a52d09347e5a3e2a08d3e8322c7 : SUCCESS in 18m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/55c9a0796fe14a4d932763bfd677b067 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/96f12353b1b24cdb9e79ddedc105a60f : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/50d164ff338b44c592d261628a834a8e : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03b92c8133d3482c922c83210f788330 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4634beb0512e4c989541d3972d431de3 : SUCCESS in 20m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d1637d97b1043b98e36eb0e4d0ce7a4 : SUCCESS in 57m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/67d2fa076bc741c6b26712e5bc8c15fb : SUCCESS in 52m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db8519dd98a1422ea04fae6cff5b668d : SUCCESS in 42m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d64ae166955748549ae61059dfec92ad : SUCCESS in 41m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14642752d7aa4367998f08af7d96a645 : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e14292df86b463f9b76296a8e8f630e : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ee648e4d5684f3ca1abcec206baf7e6 : SUCCESS in 20m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c1b823cc0cb437bb0123a4cfa5106ee : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":12},{"id":"15a7be848a451df8096c1be025039865c0ae5799","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 14:23:38.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dda9572da8314528a5f68245486b398a : SUCCESS in 55m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8531e35e765e428b8745a45f9b5f1a04 : SUCCESS in 51m 19s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/bed2cbf2b6f24f34a08ff0cda7b30049 : TIMED_OUT in 3h 01m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b7bd6474a864f739c4854dd6b1e7525 : SUCCESS in 55m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f97371e62cc6491a9f2389951a3aba91 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":12},{"id":"6c8eceece8842b1603e017c80a2dc3a6864935c2","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-31 20:40:30.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"e45471300e0e1e995b61c26deb6da3506db628d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-22 10:43:32.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0182f9fb1168f8a9d7dee4491a5db48062a8466b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:58:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a44aab78c242bfa178df322aa8ac7c : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/e3c66252642347a2980bb73f9b55f6e1 : SUCCESS in 1h 00m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66b99bd2be6848c6a7e6fa13f84c9cfa : SUCCESS in 51m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4920a95dd9e54622929eca10bc459d80 : SUCCESS in 53m 33s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c22750495aa845448655e7e2b4fb7029 : SUCCESS in 34m 16s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d57963b894b84b84b80b7a4e5c0c1896 : SUCCESS in 35m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/35f01d4fd65c4eea8548080616c5d36b : SUCCESS in 38m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad1811f4c67c422fa3488d21043b7953 : SUCCESS in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a8c74bb745a344cbbefe2d6432cbcc10 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1134824715cb40c485d8de4c6941be8f : SUCCESS in 31m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/881a53a32de74b05a6e64b50c6695ef1 : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7718f52cce9a4b3481cb3c265e294719 : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f747d1dba4014a8ab3d18fdf6074ead0 : SUCCESS in 19m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5a68571caa2a4f329e40ff98d8922bfc : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b444a3c53454b2aa7930ede6cc7b35c : SUCCESS in 18m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ed2463475ea4194b8f95ef51d3e74bd : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f8eaae4f3f78408db36710c8d1921eb8 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/52af72fd1aa84076a49594deabd943b8 : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89ecb0d587984d3aa5ade61969095d43 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/19c1702a49654bd8b4e399b5288be0b0 : SUCCESS in 16m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0713f923e5254da69f72946c7e5752e1 : SUCCESS in 55m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7eff3eb2aabf410895cc81266a2b4944 : SUCCESS in 1h 04m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7609b7d2b0a54ab0b8f974131b74abf9 : SUCCESS in 42m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5019cb45f8ba41b9b246705d891162a2 : SUCCESS in 49m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eed5468ca33341b194ae4fac6c426d97 : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c8481f83144f3ab75e884d15d38140 : SUCCESS in 7m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0141ed25f2ac4a60853cfbbdecaa9649 : SUCCESS in 23m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9745601fef6f40a99c6313fa70293e6f : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":13},{"id":"175123bd8680ceee137e04196c856e8c0627eba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 15:46:19.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa583c638c9044e0aa8f5fa63c3b3fc7 : TIMED_OUT in 1h 31m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec3cc75030fc48deb4041420ce4d00e9 : SUCCESS in 47m 23s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/0809b89fa56d4ea8bc313ce9a6e4963d : TIMED_OUT in 3h 01m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/403fa05b90b6497b953de824dc94c8ab : SUCCESS in 56m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/60c1563e3dd94c6887c9505871d5b983 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":13},{"id":"b2097bdd3604adc38e753163bf4b431edbccf2b8","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-05-13 12:39:00.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b60873d40f7fc671468f6d93604e4c36857598b0","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-05-13 12:56:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3ef9e8d8ac7291d043eb3cf1a4aa3b77a50e6503","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-17 07:39:06.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"e3be49a2fc20feb454cef27af3b0e803bfe9d60b","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-05-24 07:32:00.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6dae31771136c1101705b16286e0367bbea78990","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-26 07:15:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d03b59be2f4a772b2509e596c65d2ed604edb1ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-27 05:27:13.000000000","message":"Patch Set 13: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"5e03bac3fd14bf81ca6407cb901e68cd600ad4e2","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-05-27 09:59:12.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"dbb2f0afbd735f1fda04ad674f194bc321cd7e4d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-31 19:14:09.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4f965083f9721148f0dba03e7b95fa38de5af2ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 13:01:00.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"df89359aa17e4340845e0dd8c49a1b0d0e053ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 15:54:25.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f68b806554d403aa606aa5d8a4e6e76 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/493c74b247ae4004b9a1d3638d2bc68e : SUCCESS in 51m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8271290f040a4ff3bcd727b5576f7c9c : SUCCESS in 48m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06f2ed93bba0447eaa56caed29a1d80e : SUCCESS in 47m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3b4f1b86699f4c64a82ff94ebd4d1d92 : SUCCESS in 8m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3c1108495cb641fb987036edd0429b98 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b20f16ac8504ac7b535a3558c21f153 : SUCCESS in 19m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6ddad4ae2c949f08862a4bca717539d : SUCCESS in 16m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d381c7183b35411abe9bea04f1cf02bb : SUCCESS in 13m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c5716eb20b843d78c99685fd6eb223f : SUCCESS in 20m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2767f0ba41a94d91a7b0ac3c7d17753d : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9442ac0909ca423c8c11b06430277431 : SUCCESS in 15m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/08f39321729049d4989964ecd0bff547 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dfd771d361f04ae7a643e15a8a429991 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef2115acf16147838c8cdc6203e6fc80 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45ba3931d4cd4715ae8a10146a904cf4 : SUCCESS in 6m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/57d5117811f6428ebe4bd2d0b9e790bd : SUCCESS in 50m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/296d80d48a454865845a8afc86ce2af0 : SUCCESS in 55m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5075909c796d4752bfdacf8aea7dc43e : SUCCESS in 40m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40b7df4dfa9f4145bbf43c590ed787f9 : SUCCESS in 32m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62bc5df7dbc0464fa5bcd9b14b2577c8 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92001ce214284657955093776b72401a : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41afc40275254cb499202e8ea2729246 : SUCCESS in 26m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f237d20d660d429680518a6b1d52b1d4 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":13},{"id":"51a80ded54c1dfa82a0e20c1bcc22a9d55326943","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 15:54:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"11f1c6d443dc1a6746407784a9e57a2b92022b9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 15:56:32.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4e77354a0f47a9b6a738776de4358a : SUCCESS in 1m 26s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2c0d6d60a6cf4be3be414bc5b29d14ea : SUCCESS in 54s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/02b9de96878144f7b0f956e617034661 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fcd90d428e711c871b54d6d9956b2438ba4e703c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-06-07 15:57:02.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"118cf2ba8af97dbbd78271126e22cb80f18f9adc","revisions":{"2a4ce6bd5031ccf3f06216c2866e9783d843b996":{"kind":"REWORK","_number":1,"created":"2017-12-05 21:55:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/525771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/1"}}},"commit":{"parents":[{"commit":"dc9ba71d3e45a600ec4fc38e417279456b047862","subject":"tempurl: Make the digest algorithm configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc9ba71d3e45a600ec4fc38e417279456b047862"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nearlier this year we\u0027ve seen SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a4ce6bd5031ccf3f06216c2866e9783d843b996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a4ce6bd5031ccf3f06216c2866e9783d843b996"}]},"branch":"refs/heads/master"},"927ada3711974091c9a8aa4dd1270a5653f28b34":{"kind":"REWORK","_number":2,"created":"2017-12-28 04:20:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/525771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/2"}}},"commit":{"parents":[{"commit":"4e725e3b516c1a86b7104fd5693eced62c2a2b8a","subject":"tempurl: Make the digest algorithm configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e725e3b516c1a86b7104fd5693eced62c2a2b8a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-28 04:19:54.000000000","tz":-480},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nearlier this year we\u0027ve seen SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/927ada3711974091c9a8aa4dd1270a5653f28b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/927ada3711974091c9a8aa4dd1270a5653f28b34"}]},"branch":"refs/heads/master"},"47c713ba1be2835f4ca1d3027cbf40de6fb54e14":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-12-28 19:12:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/525771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/3"}}},"commit":{"parents":[{"commit":"b5512d007bc48067fc062158b00e9df2e8a66aad","subject":"tempurl: Make the digest algorithm configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5512d007bc48067fc062158b00e9df2e8a66aad"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-28 19:11:41.000000000","tz":0},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nearlier this year we\u0027ve seen SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c713ba1be2835f4ca1d3027cbf40de6fb54e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c713ba1be2835f4ca1d3027cbf40de6fb54e14"}]},"branch":"refs/heads/master"},"25af5096506186c02516393d9c81735422e0e91e":{"kind":"REWORK","_number":4,"created":"2021-10-30 06:11:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/525771/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/4"}}},"commit":{"parents":[{"commit":"27478f07c3f45067a2644140b7da2b356ca9ac60","subject":"Merge \"proxy: Remove a bunch of logging translations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27478f07c3f45067a2644140b7da2b356ca9ac60"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-30 06:06:56.000000000","tz":-420},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25af5096506186c02516393d9c81735422e0e91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25af5096506186c02516393d9c81735422e0e91e"}]},"branch":"refs/heads/master"},"5ec5ac084cc6983981b2b4ec70c64141f6546794":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-03-08 16:25:13.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/525771/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/5"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-08 16:25:13.000000000","tz":0},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ec5ac084cc6983981b2b4ec70c64141f6546794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ec5ac084cc6983981b2b4ec70c64141f6546794"}]},"branch":"refs/heads/master","description":"Rebase"},"2bed4cda37061432bf64260e7da2c29f5e7846d2":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-03-09 09:52:30.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/71/525771/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/6"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-03-09 09:52:30.000000000","tz":0},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\n\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bed4cda37061432bf64260e7da2c29f5e7846d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bed4cda37061432bf64260e7da2c29f5e7846d2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"246aeb3f49fec6fd1cefdb2a3e64eae344e51a1d":{"kind":"REWORK","_number":7,"created":"2022-03-09 16:07:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/525771/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/7"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-09 16:07:00.000000000","tz":-480},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/246aeb3f49fec6fd1cefdb2a3e64eae344e51a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/246aeb3f49fec6fd1cefdb2a3e64eae344e51a1d"}]},"branch":"refs/heads/master"},"b1fc91f4f4fa2a7518f72e713f8da07afa14dca8":{"kind":"REWORK","_number":8,"created":"2022-03-16 05:22:10.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/525771/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/8"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-16 00:40:54.000000000","tz":660},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1fc91f4f4fa2a7518f72e713f8da07afa14dca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1fc91f4f4fa2a7518f72e713f8da07afa14dca8"}]},"branch":"refs/heads/master"},"b76bd41b4ee8cbf17823fcdbed5e114e903987fc":{"kind":"REWORK","_number":9,"created":"2022-03-24 04:40:01.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/71/525771/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/9"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-03-24 04:36:33.000000000","tz":-300},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b76bd41b4ee8cbf17823fcdbed5e114e903987fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b76bd41b4ee8cbf17823fcdbed5e114e903987fc"}]},"branch":"refs/heads/master"},"a8fd79aa14318f9f2a859a478debb08bec0a5660":{"kind":"REWORK","_number":10,"created":"2022-03-24 20:48:24.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/71/525771/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/10"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-03-24 20:48:17.000000000","tz":-300},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8fd79aa14318f9f2a859a478debb08bec0a5660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8fd79aa14318f9f2a859a478debb08bec0a5660"}]},"branch":"refs/heads/master"},"2d5981e4046a41914681b272f873e0f4c5daf0e3":{"kind":"REWORK","_number":11,"created":"2022-03-28 21:39:57.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/71/525771/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/11"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-03-28 21:39:42.000000000","tz":-300},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d5981e4046a41914681b272f873e0f4c5daf0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d5981e4046a41914681b272f873e0f4c5daf0e3"}]},"branch":"refs/heads/master"},"c8aad9b87a60dfff701f31cdee74748172a06669":{"kind":"REWORK","_number":12,"created":"2022-03-30 23:07:35.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/71/525771/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/12"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-03-30 23:07:29.000000000","tz":-300},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8aad9b87a60dfff701f31cdee74748172a06669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8aad9b87a60dfff701f31cdee74748172a06669"}]},"branch":"refs/heads/master"},"118cf2ba8af97dbbd78271126e22cb80f18f9adc":{"kind":"REWORK","_number":13,"created":"2022-04-22 10:43:32.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/525771/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/525771/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/525771/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/525771/13"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2017-12-05 21:52:51.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-22 10:43:01.000000000","tz":600},"subject":"tempurl: Deprecate sha1 signatures","message":"tempurl: Deprecate sha1 signatures\n\nWe\u0027ve known this would eventually be necessary for a while [1], and\nway back in 2017 we started seeing SHA-1 collisions [2].\n\n[1] https://www.schneier.com/blog/archives/2012/10/when_will_we_se.html\n[2] https://security.googleblog.com/2017/02/announcing-first-sha1-collision.html\n\nUpgradeImpact:\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\"sha1\" has been removed from the default set of `allowed_digests` in the\ntempurl middleware config. If your cluster still has clients requiring\nthe use of SHA-1,\n\n- explicitly configure `allowed_digests` to include \"sha1\" and\n- encourage your clients to move to more-secure algorithms.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/832771\nChange-Id: I6e6fa76671c860191a2ce921cb6caddc859b1066\nRelated-Change: Ia9dd1a91cc3c9c946f5f029cdefc9e66bcf01046\nCloses-Bug: #1733634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/118cf2ba8af97dbbd78271126e22cb80f18f9adc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/118cf2ba8af97dbbd78271126e22cb80f18f9adc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
