)]}'
{"id":"openstack%2Fswift~533027","triplet_id":"openstack%2Fswift~master~Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee","project":"openstack/swift","branch":"master","topic":"533030","hashtags":[],"change_id":"Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee","subject":"Fix up some Content-Type handling in account/container listings","status":"MERGED","created":"2018-01-12 01:47:45.000000000","updated":"2020-03-04 07:46:50.000000000","submitted":"2020-03-04 07:44:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"533027-1583307899995-cb471cbf","meta_rev_id":"11e1781712f113f658e4978918d16a0b234d252f","_number":533027,"virtual_id_number":533027,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 07:44:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":0,"date":"2020-03-04 04:39:23.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-03-02 16:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-03-02 16:40:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-29 07:26:16.000000000","updated_by":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"reviewer":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"state":"REVIEWER"},{"updated":"2020-03-02 16:40:39.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-03-04 07:44:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51505dca5f91c75772d8ad8ee545eca7ad1346ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-12 01:47:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e1b675c823a9c314827db8c7cb532d77c5fbc64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-12 03:34:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/27/533027/1/check/build-openstack-sphinx-docs/5aa33cd/html/ : SUCCESS in 6m 50s\n- openstack-tox-pep8 http://logs.openstack.org/27/533027/1/check/openstack-tox-pep8/56703a3/ : SUCCESS in 5m 17s\n- legacy-swift-dsvm-functional http://logs.openstack.org/27/533027/1/check/legacy-swift-dsvm-functional/4d75b26/ : SUCCESS in 26m 02s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/27/533027/1/check/legacy-swift-dsvm-functional-identity-v3-only/6f43c95/ : SUCCESS in 24m 38s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/27/533027/1/check/legacy-tempest-dsvm-neutron-full/dab6ce3/ : SUCCESS in 1h 15m 58s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/27/533027/1/check/legacy-grenade-dsvm-neutron/56024e4/ : SUCCESS in 47m 03s\n- swift-tox-py27 http://logs.openstack.org/27/533027/1/check/swift-tox-py27/1d00839/ : SUCCESS in 6m 27s\n- swift-tox-py35 http://logs.openstack.org/27/533027/1/check/swift-tox-py35/a81e500/ : SUCCESS in 3m 40s\n- swift-tox-func http://logs.openstack.org/27/533027/1/check/swift-tox-func/e52272b/ : SUCCESS in 7m 39s\n- swift-tox-func-post-as-copy http://logs.openstack.org/27/533027/1/check/swift-tox-func-post-as-copy/bed0a1f/ : SUCCESS in 6m 41s\n- swift-tox-func-encryption http://logs.openstack.org/27/533027/1/check/swift-tox-func-encryption/92dc678/ : SUCCESS in 9m 04s\n- swift-tox-func-ec http://logs.openstack.org/27/533027/1/check/swift-tox-func-ec/018315a/ : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":1},{"id":"198be36b7918ee0e1e5eb290039e4397783d057c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-01-17 02:56:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nneeds, tests - maybe\n\n    diff --git a/test/unit/common/middleware/test_listing_formats.py b/test/unit/common/middleware/test_listing_formats.py\n    index 8577d86..b2c1cfe 100644\n    --- a/test/unit/common/middleware/test_listing_formats.py\n    +++ b/test/unit/common/middleware/test_listing_formats.py\n    @@ -16,7 +16,7 @@\n     import json\n     import unittest\n     \n    -from swift.common.swob import Request, HTTPOk\n    +from swift.common.swob import Request, HTTPOk, HTTPNoContent\n     from swift.common.middleware import listing_formats\n     from test.unit.common.middleware.helpers import FakeSwift\n     \n    @@ -82,6 +82,18 @@ class TestListingFormats(unittest.TestCase):\n             self.assertEqual(self.fake_swift.calls[-1], (\n                 \u0027GET\u0027, \u0027/v1/a?format\u003djson\u0027))\n     \n    +    def test_valid_content_type_on_xml_head(self):\n    +        self.fake_swift.register(\u0027HEAD\u0027, \u0027/v1/a\u0027, HTTPNoContent, {\n    +            \u0027Content-Length\u0027: str(len(self.fake_account_listing)),\n    +            \u0027Content-Type\u0027: \u0027application/json\u0027}, self.fake_account_listing)\n    +        req \u003d Request.blank(\u0027/v1/a?format\u003dxml\u0027, method\u003d\u0027HEAD\u0027)\n    +        resp \u003d req.get_response(self.app)\n    +        self.assertEqual(resp.body, \u0027\u0027)\n    +        self.assertEqual(resp.headers[\u0027Content-Type\u0027],\n    +                         \u0027application/xml; charset\u003dutf-8\u0027)\n    +        self.assertEqual(self.fake_swift.calls[-1], (\n    +            \u0027HEAD\u0027, \u0027/v1/a?format\u003djson\u0027))\n    +\n         def test_valid_container(self):\n             self.fake_swift.register(\u0027GET\u0027, \u0027/v1/a/c\u0027, HTTPOk, {\n                 \u0027Content-Length\u0027: str(len(self.fake_container_listing)),","accounts_in_message":[],"_revision_number":1},{"id":"9f181a8ec14a798c2ab7858e38ef5f28e3adee5f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-29 22:04:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b4f9f27b46f835a9a3ee850859e226609466813f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-01-29 22:04:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29e3cd9aee5a11ed5d1682b5bc08d335a54416f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 00:42:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/27/533027/2/check/build-openstack-sphinx-docs/4993199/html/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/27/533027/2/check/openstack-tox-pep8/72498bc/ : SUCCESS in 5m 37s\n- legacy-swift-dsvm-functional http://logs.openstack.org/27/533027/2/check/legacy-swift-dsvm-functional/c326a71/ : SUCCESS in 28m 20s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/27/533027/2/check/legacy-swift-dsvm-functional-identity-v3-only/e2b7e8a/ : SUCCESS in 22m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/27/533027/2/check/legacy-tempest-dsvm-neutron-full/81d90cb/ : SUCCESS in 1h 26m 28s\n- neutron-grenade http://logs.openstack.org/27/533027/2/check/neutron-grenade/d566634/ : SUCCESS in 56m 06s\n- swift-tox-py27 http://logs.openstack.org/27/533027/2/check/swift-tox-py27/34d9fc8/ : SUCCESS in 7m 41s\n- swift-tox-py35 http://logs.openstack.org/27/533027/2/check/swift-tox-py35/1264bd2/ : SUCCESS in 4m 20s\n- swift-tox-func http://logs.openstack.org/27/533027/2/check/swift-tox-func/eebe5bc/ : SUCCESS in 13m 47s\n- swift-tox-func-encryption http://logs.openstack.org/27/533027/2/check/swift-tox-func-encryption/71a4c9c/ : SUCCESS in 8m 58s\n- swift-tox-func-ec http://logs.openstack.org/27/533027/2/check/swift-tox-func-ec/d5254cd/ : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":2},{"id":"cb12e212fa2df148a7ccec5e8ce79d0e90e4272f","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-07 03:50:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nI like this change :)\nBut I added -1 because I have a question about this change.\nPlease check my comment.","accounts_in_message":[],"_revision_number":2},{"id":"b9bef847a1ffd54e00ee2abfa67e557e52c6884e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-02-07 05:36:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81fc57c276127162e5996c2f8455a1f7c1bfe1a9","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-02-07 06:45:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02a30f92421b63ba9a39d64eb09175e05f822c16","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-03-01 21:11:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf58688f71d1cf77597aaf88870d532240535f62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-03-01 21:42:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"186619128e563f066cb459743e2f15f84cc22107","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-01 23:04:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/27/533027/4/check/build-openstack-sphinx-docs/d9b8763/html/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/27/533027/4/check/openstack-tox-pep8/d3ba7ea/ : SUCCESS in 3m 46s\n- legacy-swift-dsvm-functional http://logs.openstack.org/27/533027/4/check/legacy-swift-dsvm-functional/aeed4b6/ : SUCCESS in 23m 58s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/27/533027/4/check/legacy-swift-dsvm-functional-identity-v3-only/da1e070/ : SUCCESS in 26m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/27/533027/4/check/legacy-tempest-dsvm-neutron-full/c39a22c/ : SUCCESS in 1h 20m 26s\n- neutron-grenade http://logs.openstack.org/27/533027/4/check/neutron-grenade/1ed2c76/ : SUCCESS in 51m 24s\n- swift-tox-py27 http://logs.openstack.org/27/533027/4/check/swift-tox-py27/80c25e3/ : SUCCESS in 6m 17s\n- swift-tox-py35 http://logs.openstack.org/27/533027/4/check/swift-tox-py35/a5b0e57/ : SUCCESS in 3m 37s\n- swift-tox-func http://logs.openstack.org/27/533027/4/check/swift-tox-func/ca15e56/ : SUCCESS in 7m 12s\n- swift-tox-func-encryption http://logs.openstack.org/27/533027/4/check/swift-tox-func-encryption/234806a/ : SUCCESS in 7m 16s\n- swift-tox-func-ec http://logs.openstack.org/27/533027/4/check/swift-tox-func-ec/9a5d992/ : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":4},{"id":"56fb88f28bdc554b4d8a41b71430863b0e1744be","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-05-29 07:26:16.000000000","message":"Patch Set 4: Code-Review+1\n\nUnit tests are enough now.\nLGTM!","accounts_in_message":[],"_revision_number":4},{"id":"f5e0cdf72be9f399b443b406c4249ecd073b1e86","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-28 00:22:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"33c01b48b4bc4c6a7f5e428e9e9ce0f5b0e84bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 03:11:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bacd6e86d3c54b3390238a3eeb31c3b4 : SUCCESS in 19m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e4bb727a4034929b934b1f7e1c21c16 : SUCCESS in 1h 08m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90c9009f7dc446a984a7df7d133eca9b : SUCCESS in 51m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26e8bfdca38145bc9a537986e022f295 : SUCCESS in 11m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/79b839b24dc342efb2d7ab2b893f6517 : SUCCESS in 14m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ebd0cb6dd2e340d2a5083da7f8db7169 : SUCCESS in 16m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20cf6216e6bc423085bf149ec5cd7a5b : FAILURE in 19m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/432ae69e5b004274bcba7ab6d5ba3709 : FAILURE in 16m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4106451e9944654ab5f5900bb23cad0 : FAILURE in 16m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62f2930e435e4da9bab2d2d874294c11 : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/294353aa1db24ef9a333f67d79ab7fe2 : SUCCESS in 15m 40s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/82c63574dc4e44819f36ab9312a20c89 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5e7ec2015ce3433e934d8654fa5ddfbb : FAILURE in 13m 44s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/35347db3a0a44d3a8fd4b64c9b32fdd3 : SUCCESS in 21m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b2d019767e4f41f5ad1d9373f53a5d59 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8944b205f17940029bf3e4baafe8632a : SUCCESS in 22m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5028eab2612f4e458abd78b463be86e4 : SUCCESS in 11m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6589be0308fb417f8cf1957c0d056c7d : SUCCESS in 13m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6eaf99f5810c4cbdb714da7d21aad4f7 : SUCCESS in 16m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/037b1c8e7069468bbb0ea7b19eb4c8b7 : SUCCESS in 19m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b6da637a56924f03b8d167d243b39105 : SUCCESS in 44m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81fff5adacfb4896b96146246fd43018 : SUCCESS in 31m 16s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a5ea736167a340c0b79ca1ad57b241b7 : SUCCESS in 30m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36a37e5861924c659ecf40f3007f5074 : SUCCESS in 39m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2195213ce82e457e93d27dc85a706001 : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50cd89c4391348e4b600884f1b4739b1 : SUCCESS in 8m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11b889d48b8a40f6986c5af66958575f : SUCCESS in 40m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59334026ff7341caa3a0ff64740a1ba5 : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":5},{"id":"23b5415f021897a8348ca4fa1e7f4c4a45d5c835","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-29 02:32:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4f62dfe2a5fe713aba34abdb34efe8d814685d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 03:54:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08267b1fa4b49cfb23f473a3279d17f : SUCCESS in 12m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6818f73b700947cc895a825120affc21 : SUCCESS in 1h 00m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59630b1e237342a387b1cfecd57f8b75 : SUCCESS in 56m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7b861c8affc454e9135d9e1fd242826 : SUCCESS in 12m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89a3c35cddf24a81947108871c5b7c2e : SUCCESS in 12m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58cd8930267041708ed0aab635eb7135 : FAILURE in 13m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2fea260235404474a505b6b9cee37f38 : FAILURE in 20m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d0054e73c2234522896c00c1739b453e : FAILURE in 20m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a99523254545348d82a0f90c4bb36e : FAILURE in 21m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/754bcecca9d44ac691b2cf0a7ed6fc51 : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce4407b918a643a99d8db1b1c35c9e31 : SUCCESS in 12m 23s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8a2bb4d59d864b6a9e023e8c076daef4 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ab757b3c98c459091fd5944fd2bd781 : SUCCESS in 18m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/22a71e4e131a41cb8cc5740a127f3e41 : SUCCESS in 30m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a18dff1949cc467a9a7f64876bbf3c8c : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/14af0a19509e43168c3240abf7a5d576 : SUCCESS in 23m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7e1ebd81902345929fcd031e9a803d40 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/57f8fe312df547e699146471ee21576f : SUCCESS in 14m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/adb56462523042c9a69c46a899c9dd50 : SUCCESS in 17m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6afda9a3bc254ad9b0f1808b6ba19ac1 : SUCCESS in 23m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cbfccd6f07cf4e2796109c39c22b6f7a : SUCCESS in 45m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c7becb330f74fc48103f551aaab114b : SUCCESS in 41m 16s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e00b5c76710a444b9aa6c0bd91711572 : SUCCESS in 39m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/488e60710ae9457184612c4df81272d5 : SUCCESS in 41m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bb4d316b9e342129bdfad14b1b062c0 : FAILURE in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2407f051bd3044e5a5a835706d1b556d : SUCCESS in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12168d47c3c2467994ae5f4f48850e8b : SUCCESS in 24m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04e094ac002d4da4b2ead5eeaf60335a : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":6},{"id":"ca6e17f93143a01a8c89db705e8b4de0b5262f92","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-29 05:37:29.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8ed2f00b48be67f60d3f5b5cd1823c75b02b6728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 06:48:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d37fbccb7fe4afbae6014922775d75f : SUCCESS in 9m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04a49fba982a4858be78a05e9d9aa31f : SUCCESS in 1h 05m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f7a8af4d55c402f9151fd3bb41016ab : SUCCESS in 52m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/984f592966be42a185eced955b5994c7 : SUCCESS in 9m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/276209ef02854ae1ac22c4d384f77078 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47c52f892b2d42368a8d922a497ad644 : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d2d0b5cfd28b45f6bcd07fb0b31d5037 : SUCCESS in 12m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63e7363485634e5d97ca33e5de3158c6 : SUCCESS in 13m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c732d6b0b499488782f1998df155bb85 : SUCCESS in 13m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc33f644606c43d4bd99889cba9c1d7b : SUCCESS in 10m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/687093882e9841f8bc65117b4b12745b : SUCCESS in 13m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/87621d64b35249f9b3c296e4d43c2b06 : SUCCESS in 10m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff351adb0068463591bfd46ffa3e69b6 : SUCCESS in 11m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e40d16356b4c4617938fff16737148a3 : SUCCESS in 15m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a429dfe980ff4e7e80c02b1bd7e6b9e5 : SUCCESS in 10m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b8759a9e91d04e47bf4c559b892a8fc1 : SUCCESS in 13m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/75e28f39a73e435888f0829cb4abf15d : SUCCESS in 10m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8e1a1db420ea40fcb5ecbad2e719c9d6 : SUCCESS in 11m 21s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/61d11ba5fcbe4ff6ab442bde4c57d335 : SUCCESS in 16m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/838b78deeffd4624b3192996f4cc9a43 : SUCCESS in 18m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/56bfc7ab0d234ca48f6fd9c845eea0d6 : SUCCESS in 44m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b211f99657f463383e763362ee48c72 : SUCCESS in 31m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9d3946cd1da143dd84db68d7f50bfeeb : SUCCESS in 32m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e393a7ca0c24d70b3faf638900a589e : SUCCESS in 33m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7240224db1ef439384ee45e271de957f : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc7dc7168134b28bd54d8b7bb0cf044 : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/beb0ebccfd0c403a8d686abb408fb8a2 : SUCCESS in 25m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a970b0573704bd591074409e92ef227 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":6},{"id":"ed08bcbb74d5ebdb094180e63b17619a58cec123","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-03-02 16:40:39.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nyeah i guess this is fine, except, I mean our HEAD responses don\u0027t vary that much based on the accept headers.  We lie about content length on purpose.\n\nhttps://gist.github.com/clayg/dfe1733776716ecc3a2d1c9d5512d1fd\n\nI\u0027m going to trust this bad browser caching behavior caused a problem somewhere and isn\u0027t just an attempt to be more pedantically correctly","accounts_in_message":[],"_revision_number":6},{"id":"d10b85326c032814784a1a027fceaefc877ad0f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 16:41:46.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"409501e4af527fea3990677f84ee7c37a095ddda","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-02 16:48:32.000000000","message":"Patch Set 6:\n\nNoticed it while poking at CORS tests, though Safari (at least) still has some trouble: https://review.opendev.org/#/c/533028/6/test/functional/cors/test-container.js@69","accounts_in_message":[],"_revision_number":6},{"id":"197bca722c0aba498ea3ad2a09613b86b3e3a44c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 00:23:25.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d192febfe647bb8576e586a0f01866 : SUCCESS in 12m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c36f6ab33c68458fbfb6dc0d4a0f927b : SUCCESS in 1h 03m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/596e855e76774fca8223c415d3344a1a : SUCCESS in 57m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6b84f1c9c74348c79a59c2ca60a1891e : SUCCESS in 12m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/17b98f187572443eaaedf4f943588f0b : SUCCESS in 15m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/493b53bbce5b4edd9cd08eac4f018930 : SUCCESS in 25m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b90d1de4c234931ac1a15a5db1355bc : SUCCESS in 14m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7ebb726a6ef949f7a165239504aa3cfe : SUCCESS in 22m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e4aaef5c5c884831864d64419923a359 : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/09c3ef8643c248e1bc18227869f1dd76 : SUCCESS in 22m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/659c8f0908c14c47a44f0c585faa72f3 : SUCCESS in 14m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/008f67d15a5b4e2284a8fccaf59b050d : SUCCESS in 13m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a24f93e46a44da89049d3b6e17a9b8a : SUCCESS in 13m 57s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4bffa5f771894527a4779265e014c899 : SUCCESS in 18m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7df5ca7f3d1f4ad7829ee948561a7707 : SUCCESS in 13m 34s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/626eadc726234dfab08f9e240232cd6b : SUCCESS in 26m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/092a28fd94d640e3960b752b3ec14aa7 : SUCCESS in 14m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d43932d876a74b609f56dc05aa99d679 : SUCCESS in 20m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f7a041e746804087b4e9a4338bdbbe5a : SUCCESS in 49m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4707889bdafe48cc8c377988b77295f9 : FAILURE in 29m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3fd8b83d040a4353b2d29eae299eaed4 : SUCCESS in 37m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cd856d568904d2cb2794fbe274b24b9 : SUCCESS in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0f56f35ddf740e9ba1f678dd6f4fff9 : SUCCESS in 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4c0bab35634dbdb77dd5264ca2a89b : SUCCESS in 11m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f56b8507c9e49e989bf6f448d995c9b : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":6},{"id":"374c4863accc54ffabe830214166f60f8b3ddac9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-03 00:40:06.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e10fed22a3fd95b7897921e74f5d3a306d32a9c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:02:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d8ba2eaf5b49fea3d9fb0c82f398f6 : SUCCESS in 11m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/db3bc096e74d4c58b6abd361309eb646 : SUCCESS in 56m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f484ab660a734cd68ba6215417832c07 : SUCCESS in 54m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b2ef1f15d254098ace27bc83c4184cc : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7ac4691b42b7487bb37e5679a9c8a435 : SUCCESS in 11m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b900407f9cb64b4281d83c0ed5374d32 : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/45381c44390548dd99e22002de2d3301 : SUCCESS in 18m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f66f6ba01035461e907d3acf75ab6fdc : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/720d4e832b0449ffa8d67456761c9b78 : SUCCESS in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2e4bccc633e44ef8a06758e3fcfd0a8 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96d47c58034f4840a6594ac2665b9215 : SUCCESS in 17m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/944bca7d55cd4d94b15f579fdf848df5 : SUCCESS in 13m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ab3357b411b40e18000f55f50fbd30c : SUCCESS in 14m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5ad9337c490841858efdfa9368dd092a : SUCCESS in 19m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c574db3e473a4392b0cc7bede5929533 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d39dcc6f685d4038b344fac7672d3086 : SUCCESS in 14m 57s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/eecab104ae0045f6988f88e55b0a5407 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/292d7e857e9b40a38651f0916c9ea451 : SUCCESS in 17m 21s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/32012ad003e646f19e05403ac0a5bbfe : SUCCESS in 20m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66c1c974beb54d87afc4bc313d74e7da : SUCCESS in 18m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b14143e0073b4175abd49d8b29c8e15e : SUCCESS in 44m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/021b39f8eaff4d0696e20c9ce113628d : SUCCESS in 34m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ed6647210cd540b2b7360ec170d04953 : SUCCESS in 31m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef7234848d3c46caaf12b82aa709422a : SUCCESS in 30m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d45738b10de4b868ee7360de1b30ade : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808485bb3e694f5793656db6c137ed18 : SUCCESS in 8m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b3c4b0e7a704857ae2252fc3f6d1caa : SUCCESS in 31m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f853b30bdf444ec9a327b9eae9e2057 : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":6},{"id":"aa822d1b15918b73a895d33e32714f998bc295ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:20:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"305561f08c0b93bd7e303ffcd93da03f1d72d250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 15:34:28.000000000","message":"Patch Set 6:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"199e2a1aab016db02e150bde0d5af3d14c49d749","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 18:29:48.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38fabd995ed34abdbf23dfd489059ac1 : SUCCESS in 12m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/74c5bf73a4be4d21898bac7968d06ec2 : FAILURE in 40m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee7c98fcef7e4ebe924092ed69751581 : SUCCESS in 51m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ca7c5090f4954d418ed47b4a8ee056f8 : SUCCESS in 11m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e9e28eaa3e2547039fbea93676011ac8 : SUCCESS in 13m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/761ab22c85c349f09884b5fc75dd372f : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/04db3ff242f644908afcc6cd1e8c7b43 : SUCCESS in 21m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec99996df9e6428faea0243ced2ea33e : SUCCESS in 18m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2bbe36287246441cbae6741b6b62a3e1 : SUCCESS in 14m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9fc4d5a9702d4497a94ced8ccae337fe : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/367aa8f42ecf4b679bedbdc320fd3abb : SUCCESS in 14m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/299509b39cb2428e932c8701335965cb : SUCCESS in 14m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3214b7308ee4b5eb34777932d15b2c8 : SUCCESS in 16m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fdb8820e2b0347f896998f0c52b2e0d5 : SUCCESS in 24m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2092b8cc88784c18994eb5617933f59f : SUCCESS in 13m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f04837efc37c4dfbb564921664077f2f : SUCCESS in 18m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/245105b68b6043d8b81ba35d07327b60 : SUCCESS in 18m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6b6c09a82c274142bcbdd623a58493c6 : SUCCESS in 22m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3cb2aede7172458b914b64c3f96c8d26 : FAILURE in 55m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee4f9b463a82426fa9727eeb799c69cf : SUCCESS in 35m 11s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1805f205488e486981aabc3a7f5ed2fd : SUCCESS in 30m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea1328aebede42ab946674cd9aa005d2 : SUCCESS in 42m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4ad48b236014915bc05bdb68b7a3ee6 : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e3f4f57d54448e182770cb8f2af46ce : SUCCESS in 11m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a492b2b0a98f4ecc937949e45cec122b : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":6},{"id":"c8bd69fdffbefd9f3fd9e9fc491a9ba527eefd82","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-03 19:24:18.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ff2769b3e8f8f956d9d88607118298159716087f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 01:10:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f034e88f48f4196ba141ad58ce7c732 : SUCCESS in 10m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/570951dc98414e19a143f55b9e2cb8ee : SUCCESS in 59m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0355db57de0c40228ac1e5b5ddfcb05c : FAILURE in 42m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a5b4797da8e4d81b96269552ca68440 : SUCCESS in 11m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b5ebbf7fc1844bd68529eea555dcdc11 : SUCCESS in 11m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/edbde65ad5fb417497b5fc87bf30b38f : SUCCESS in 25m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/464002e9a2ef427183fe34d04cfde6b4 : SUCCESS in 13m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/852fd8e52dc44a3a81596f9bbee4a39e : SUCCESS in 17m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cefbd2cbe5244acdbb452e6544f27125 : SUCCESS in 20m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35f317985d1e4b288c05521d728a8e95 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/47bfe893d53c4a70a092379e4787ecc8 : SUCCESS in 23m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1e42eeb772fb48d7bdf10b219d0816f5 : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d56cfb3816024ab198d1990b11c2eacf : SUCCESS in 11m 54s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d7105168090a4c778895a6be9397f4f0 : SUCCESS in 21m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c67cd41790c24477b7f47b6e5c295941 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ca3104b1140c43cd8a0179eaad6a5036 : SUCCESS in 13m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9942ae3455e947f2a5cb7895194170d4 : SUCCESS in 11m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6581ac5f5af84b2e94a8a8156aa367d8 : SUCCESS in 11m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1b0e0e1824d74efeba43f525c0c2f7c9 : SUCCESS in 17m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1af8c3d497d74e5cbd71174226f758e5 : SUCCESS in 24m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f56b92a1f4104ee3be82c6f9d1f7e67b : SUCCESS in 54m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2d7864b1f9c4532a8250e8b1909fc32 : SUCCESS in 32m 09s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f3e53f07203747819db9420823123b25 : SUCCESS in 33m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88fbdd342c59400a97351e9d9cb73e4c : SUCCESS in 33m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3c0a0d6b0eb4e2dac5284685da31fb8 : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f27f49d41c945cfaa8135e9ef98468c : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b95a6fd9e5e04f8590cf1882a14f2c77 : FAILURE in 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d56c7160bdd4deabd94446e68229474 : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":6},{"id":"baf42975e2335773d04afdcd6cbdadb8f6565bd1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-04 04:39:23.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"012c45eb059dafa0f0e28ab8bb68f01f4880a80c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 05:53:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17307b82765243fdb282287261e9e8e8 : SUCCESS in 9m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1b51e409bde54a419638ef1386aacc39 : SUCCESS in 58m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7575f4f6be924be2aa5b440889af2b06 : SUCCESS in 56m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f5df1cef52c0436ea005af238149ac3d : SUCCESS in 9m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c9f44790666d4d49b78a73ecb8e0e8b5 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f898189ccf343ed922f31c318d6d2b1 : SUCCESS in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79c46138ce3a4092b54379afda9d282d : SUCCESS in 16m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/83c311b7591647669838572bf41ffaa6 : SUCCESS in 16m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b69ead75b02c4e218afbad9816aa16d9 : SUCCESS in 12m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1974661eb26e4aa6a61632b82576f5bb : SUCCESS in 10m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4babac796bfc4e518cfc2cb694c6e497 : SUCCESS in 12m 47s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8e98d8bbb24d4f2099ce2f5ab3f51cdf : SUCCESS in 11m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80dcb81c60de45e7b44d24d7943e3608 : SUCCESS in 10m 47s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e862ecd34d7b44b699448eca24d1d1ff : SUCCESS in 18m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ccac4ea2795e4318843716799343eaee : SUCCESS in 9m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e221e2cd1ea544c3a0a0b9463eb31226 : SUCCESS in 13m 28s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b6de41a1e69e43889ed636acc750f3ed : SUCCESS in 9m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f24d6b82bc1445c19c539d42ed15141f : SUCCESS in 14m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/24d796aca6fa435a96126c08bbdec0e3 : SUCCESS in 15m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e603ddaeb4c544e8abbe58996db41309 : SUCCESS in 22m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d97bb4b1f1e243de9f6915029ffee22e : SUCCESS in 46m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cbf9dadda844436fbbd127d9f1673ac1 : SUCCESS in 28m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/724d991ca9d04539b39ebe2e9f4b9cf1 : SUCCESS in 33m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a2c0aa92769340ccad7b545c7f2c0b7f : SUCCESS in 28m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba26e1b0f8c346148be446d63eebd30e : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d98c14a43649b495254cf864078765 : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/743bd1e4f55a46b3aad464bfb2d591e0 : SUCCESS in 20m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4190efe0e024459392543babe8cac6fa : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":6},{"id":"52904b03363ec9019ee6eac682be8abd41780844","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 05:53:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a53223be2830dd2f51342ab9ffa43276a9e8f1b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:44:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"49ad5028182cf41bc6401dcd8e1ddfe55a0a6303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:44:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ec4b90176045dca80bc844ba2f35a9 : SUCCESS in 10m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e36d9ef2d37347cc83a13c92216fdcdb : SUCCESS in 59m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/895d53e26dfc4454b4f32c00b4ac8db6 : SUCCESS in 55m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/344a635ed4544bf1a071431e3d86a836 : SUCCESS in 13m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fe675062582c47fd84d14ef68bd0b58f : SUCCESS in 16m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd0e6cede5d842bba8135cbc3f9eccc5 : SUCCESS in 16m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b839be536bbc443ba1e47d02f92cb1d4 : SUCCESS in 18m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3833e9a0c96f4eae8e6d777252f2a905 : SUCCESS in 20m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d50f281beeef4fecaad7e814dc678b7b : SUCCESS in 19m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b826b8ba15424b4da624eb47c542f294 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db482a133aeb466dbb184ae582f9ceb7 : SUCCESS in 15m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dd29f555d030490a896e15b7805ff1e9 : SUCCESS in 12m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/acb4c110116c4bf5b13afa272378cce2 : SUCCESS in 12m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1c25087cc52d49089f344133201156ff : SUCCESS in 22m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/18876a19259544fcb37ea405f4a4d876 : SUCCESS in 20m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/26763687780a4823b1d62ea8d553982f : SUCCESS in 16m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f04f4843be0d40ae9eb084dc01aa9cdd : SUCCESS in 23m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f22cada5eaa442fe948b9c39ed405b62 : SUCCESS in 17m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e5a2c0ce1e340f29e48978f210124c1 : SUCCESS in 43m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5820b369e5194245869abf7fceecc633 : SUCCESS in 34m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/515f1e478f2b48a4ac268a30e54d293c : SUCCESS in 37m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/857ce61240ad4296ae29f1ab3a6aa719 : SUCCESS in 41m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26a2645816e04334975e1493d0285b90 : SUCCESS in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef38130c12784ec98e8875917f1bb2cf : SUCCESS in 8m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c02a86c12dc405293606b4483ba2f80 : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":6},{"id":"1c536cd40850ba4f66a7ca76c9912e7a836c5a8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:46:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446ba6b76ec74362b8a43318b9106349 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/22e3d649ce85436c99111828b42a395e : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/aba718c351c54c568b8bbd79a00c0956 : SUCCESS in 1m 25s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3b65a5998cc921d2763cf1a9ec1e40b88491262d","revisions":{"b4cff2b3b5a5b9e7eadd5113a62379912486d2e6":{"kind":"REWORK","_number":1,"created":"2018-01-12 01:47:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/533027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/533027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/533027/1"}}},"commit":{"parents":[{"commit":"5917ea0ea157acb0a54f9d8b180a0bbc03577121","subject":"Merge \"Change exit code when displaying empty rings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5917ea0ea157acb0a54f9d8b180a0bbc03577121"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-10 06:16:41.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-12 01:44:50.000000000","tz":-480},"subject":"Fix up some Content-Type handling in account/container listings","message":"Fix up some Content-Type handling in account/container listings\n\nUpdate content type on 204 (not just 200) to properly handle HEAD\nrequests from xml/txt listings.\n\nAdd \"Vary: Accept\" header to listings, since otherwise, browsers may\nserve the wrong content type from cache (even though we *would have*\nsent the *right* type if it actually sent the request).\n\nChange-Id: Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4cff2b3b5a5b9e7eadd5113a62379912486d2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4cff2b3b5a5b9e7eadd5113a62379912486d2e6"}]},"branch":"refs/heads/master"},"904df7fefca4415c5f4e00baa579b3ef3fe1816e":{"kind":"REWORK","_number":2,"created":"2018-01-29 22:04:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/533027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/533027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/533027/2"}}},"commit":{"parents":[{"commit":"5917ea0ea157acb0a54f9d8b180a0bbc03577121","subject":"Merge \"Change exit code when displaying empty rings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5917ea0ea157acb0a54f9d8b180a0bbc03577121"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-10 06:16:41.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-29 22:03:36.000000000","tz":-480},"subject":"Fix up some Content-Type handling in account/container listings","message":"Fix up some Content-Type handling in account/container listings\n\nUpdate content type on 204 (not just 200) to properly handle HEAD\nrequests from xml/txt listings.\n\nAdd \"Vary: Accept\" header to listings, since otherwise, browsers may\nserve the wrong content type from cache (even though we *would have*\nsent the *right* type if it actually sent the request).\n\nChange-Id: Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/904df7fefca4415c5f4e00baa579b3ef3fe1816e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/904df7fefca4415c5f4e00baa579b3ef3fe1816e"}]},"branch":"refs/heads/master"},"f9ce49a8fd986910d49905a11722aa2d84ae7fd5":{"kind":"REWORK","_number":3,"created":"2018-03-01 21:11:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/533027/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/533027/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/533027/3"}}},"commit":{"parents":[{"commit":"10943b89b97ba927e2d651eca454870bf4699f59","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10943b89b97ba927e2d651eca454870bf4699f59"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-10 06:16:41.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-03-01 21:09:44.000000000","tz":0},"subject":"Fix up some Content-Type handling in account/container listings","message":"Fix up some Content-Type handling in account/container listings\n\nUpdate content type on 204 (not just 200) to properly handle HEAD\nrequests from xml/txt listings.\n\nAdd \"Vary: Accept\" header to listings, since otherwise, browsers may\nserve the wrong content type from cache (even though we *would have*\nsent the *right* type if it actually sent the request).\n\nChange-Id: Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9ce49a8fd986910d49905a11722aa2d84ae7fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9ce49a8fd986910d49905a11722aa2d84ae7fd5"}]},"branch":"refs/heads/master"},"4dfcf2150795b387083967faf1ced4e79a5c78b4":{"kind":"REWORK","_number":4,"created":"2018-03-01 21:42:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/533027/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/533027/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/533027/4"}}},"commit":{"parents":[{"commit":"10943b89b97ba927e2d651eca454870bf4699f59","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10943b89b97ba927e2d651eca454870bf4699f59"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-10 06:16:41.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-03-01 21:42:11.000000000","tz":0},"subject":"Fix up some Content-Type handling in account/container listings","message":"Fix up some Content-Type handling in account/container listings\n\nUpdate content type on 204 (not just 200) to properly handle HEAD\nrequests from xml/txt listings.\n\nAdd \"Vary: Accept\" header to listings, since otherwise, browsers may\nserve the wrong content type from cache (even though we *would have*\nsent the *right* type if it actually sent the request).\n\nChange-Id: Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dfcf2150795b387083967faf1ced4e79a5c78b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dfcf2150795b387083967faf1ced4e79a5c78b4"}]},"branch":"refs/heads/master"},"8e7aec5842298bf97a793803b3a1023a449e95c5":{"kind":"REWORK","_number":5,"created":"2020-02-28 00:22:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/533027/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/533027/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/533027/5"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-10 06:16:41.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-24 18:46:40.000000000","tz":-480},"subject":"Fix up some Content-Type handling in account/container listings","message":"Fix up some Content-Type handling in account/container listings\n\nUpdate content type on 204 (not just 200) to properly handle HEAD\nrequests from xml/txt listings.\n\nAdd \"Vary: Accept\" header to listings, since otherwise, browsers may\nserve the wrong content type from cache (even though we *would have*\nsent the *right* type if it actually sent the request).\n\nChange-Id: Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7aec5842298bf97a793803b3a1023a449e95c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7aec5842298bf97a793803b3a1023a449e95c5"}]},"branch":"refs/heads/master"},"3b65a5998cc921d2763cf1a9ec1e40b88491262d":{"kind":"REWORK","_number":6,"created":"2020-02-29 02:32:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/533027/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/533027/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/533027/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/533027/6"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-10 06:16:41.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-29 02:32:38.000000000","tz":-480},"subject":"Fix up some Content-Type handling in account/container listings","message":"Fix up some Content-Type handling in account/container listings\n\nUpdate content type on 204 (not just 200) to properly handle HEAD\nrequests from xml/txt listings.\n\nAdd \"Vary: Accept\" header to listings, since otherwise, browsers may\nserve the wrong content type from cache (even though we *would have*\nsent the *right* type if it actually sent the request).\n\nChange-Id: Iaa333aaca36a8dc2df65d38ef2173e3a6e2000ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b65a5998cc921d2763cf1a9ec1e40b88491262d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b65a5998cc921d2763cf1a9ec1e40b88491262d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
