)]}'
{"id":"openstack%2Fswift~541141","triplet_id":"openstack%2Fswift~master~I98582732ca5ccb2e7d2369b53abf9aa8c0ede00c","project":"openstack/swift","branch":"master","topic":"reconstructor-recon","hashtags":[],"change_id":"I98582732ca5ccb2e7d2369b53abf9aa8c0ede00c","subject":"Add and pipe reconstructor stats through recon","status":"MERGED","created":"2018-02-06 06:27:24.000000000","updated":"2021-08-23 23:35:14.000000000","submitted":"2021-08-23 23:33:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"541141-reconstructor-recon","meta_rev_id":"b21963affc197ebae53963937c20d0e40351f6b5","_number":541141,"virtual_id_number":541141,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 23:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2021-08-21 04:25:00.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-23 18:02:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-23 18:02:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-06 07:21:37.000000000","updated_by":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-06 08:04:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-21 04:25:00.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2021-08-23 18:02:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"0b90cc4714fb80c5c3490635b049a3a844bcd82b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-02-06 06:27:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4daf148ef04a468bbd2929e39b3117dd253e4bb","author":{"_account_id":13052,"name":"SwiftStack Cluster CI","email":"openstack-ci@swiftstack.com","username":"swiftstack-cluster-ci","tags":["SERVICE_USER"]},"date":"2018-02-06 07:21:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* cluster-swift-tox-func-ec https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/541141/7902/ : FAILURE\n\n* cluster-swift-tox-func https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/541141/9882/ : SUCCESS\n\n* vm-saio-probe https://8b86aea46fb38e6450f2-0e5f4c086da474abc1df58826577db2f.ssl.cf1.rackcdn.com/541141/8280/probetests/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"fc47db2c8d2bb348e4397cf6d9f8ae35adfca27d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-06 08:04:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/41/541141/1/check/build-openstack-sphinx-docs/83ba83c/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/41/541141/1/check/openstack-tox-pep8/a8f6841/ : SUCCESS in 6m 08s\n- legacy-swift-dsvm-functional http://logs.openstack.org/41/541141/1/check/legacy-swift-dsvm-functional/2bd5bbb/ : SUCCESS in 23m 52s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/41/541141/1/check/legacy-swift-dsvm-functional-identity-v3-only/413db29/ : SUCCESS in 22m 09s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/41/541141/1/check/legacy-tempest-dsvm-neutron-full/75c152f/ : SUCCESS in 1h 31m 55s\n- neutron-grenade http://logs.openstack.org/41/541141/1/check/neutron-grenade/b795ec0/ : SUCCESS in 56m 13s\n- swift-tox-py27 http://logs.openstack.org/41/541141/1/check/swift-tox-py27/3889725/ : SUCCESS in 6m 42s\n- swift-tox-py35 http://logs.openstack.org/41/541141/1/check/swift-tox-py35/62cf722/ : SUCCESS in 4m 26s\n- swift-tox-func http://logs.openstack.org/41/541141/1/check/swift-tox-func/7ff1f50/ : SUCCESS in 8m 02s\n- swift-tox-func-encryption http://logs.openstack.org/41/541141/1/check/swift-tox-func-encryption/4812416/ : SUCCESS in 6m 29s\n- swift-tox-func-ec http://logs.openstack.org/41/541141/1/check/swift-tox-func-ec/7b2b4f3/ : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1},{"id":"52f6bff5c16d5e8682c043e84aef698a6a3028b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-27 07:08:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"259ca2cf05a9602cfc87847341e0c2c53fe1d71b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-27 07:16:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84f36a997ee09da343e3e0e8fb149eb6c8deaf85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:24:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e018fbdc634a64b0184fb3ca85ef2c : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/a8ba984bfb9e412597bf007c6eb976a9 : SUCCESS in 59m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/469cdae4eb0f421aafbf099485017b07 : SUCCESS in 42m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecf97334ecc248019e55538b3c263359 : SUCCESS in 9m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bf375536e27748f4bc0d1d08babb0b61 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3742bedcbd140258f1f3479982ec970 : SUCCESS in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd2b76d3043043f3a3ab16a95941836a : SUCCESS in 17m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24649792da244f21aa23c9fe90fd67d4 : SUCCESS in 17m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e89dd8d269334b3bbfa24be5db56cb8a : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e0d0f00d7824a4d85efe70573e4539e : SUCCESS in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2897e9070725420bae115f6399d645a8 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62b10301f574468284bf1c265b50a2ee : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dbfb2e0390dd46c48dbe2ab3ff58536f : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/38d13ebe0fce4bb09a15187cb66d4c2c : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0be3354683bd4d6286a99242de7ada58 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a99818bf6c47499b955fb7e3db0069fe : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a337304180942309e65861f3744cbfc : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce9bc2fc35064300bb3271e08bac7783 : SUCCESS in 20m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/beb7dd64e1d24b8cb076e8a49296df90 : SUCCESS in 50m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c183e47d855b4d37ac7d3d9826b40453 : SUCCESS in 47m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d6eed81cd2c41b7aa7d38bcc2216aa5 : SUCCESS in 40m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17577ea1465643d88bde35a4719f5432 : SUCCESS in 38m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44c712eeb99646739e049ccc8caeeb71 : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c96a5cb69074df89d559407cf39b57a : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63d510b5f7a64fa0ad49276b02f95e30 : SUCCESS in 23m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7d21ac1e1834c668141f8c272c52498 : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":3},{"id":"56926361cbf074ba4e6376e3c36b6f8c8432ea5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:33:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f20f54d7ee846d3b3f2782682d5b69f : TIMED_OUT in 1h 31m 10s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4e224a8756540cf8783b994a08d1a10 : TIMED_OUT in 1h 31m 22s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/7e88e96a1b734f0c937d639654ea51c0 : TIMED_OUT in 2h 01m 11s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/58535a6c8c544772b310ca79bca83a80 : TIMED_OUT in 1h 31m 21s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cfa50c7173544549b48bba8867e61b8 : TIMED_OUT in 1h 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd0d0cb17031bcbf38f02a3ced1a938d3c89f0ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-08-20 00:04:07.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2ed525bd34030b1f5450d7d16ef8db2fa61d24f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 01:15:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0096d46e5d4494dbfc94dd35c851cd4 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/41424cd5898948fd8c0b7657e3ea9a6d : SUCCESS in 1h 06m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d70a82301904832bb092c2eaab74b46 : SUCCESS in 37m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b2f52f3c2734f2fb978910238e2a868 : FAILURE in 7m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1b6472fdbdd54f5fb959db58be969f7e : FAILURE in 7m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/037792b36a614e9c88cdbb9b5f15cfe9 : SUCCESS in 16m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6b4a180bdaa45259eb8d9f2dcd73433 : SUCCESS in 27m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c159789a44a94ee5a6092f710eb9b8b2 : SUCCESS in 20m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d15a109e68bf48b0b8d7d542dc2987b9 : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c863f77b4a34fc4a241720cd33fe198 : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/051c555700264e7481fbf344faa96157 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/01fe9a69c2e443d2a221417df3f03747 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7d0bd7aff634a31a18e54b76a29089e : SUCCESS in 18m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b02a293a0194b2b86110c80a5810441 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2f7f8b9e85084c07a609c2fa7b822824 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/412b697bb9ad4d1eb18cad3598b2c11b : SUCCESS in 20m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/666583e77ca548429e1c3fdb7589f21e : SUCCESS in 15m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c7d31f67cbb74624b7e6d279ec3373b5 : SUCCESS in 23m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7fa516cc887b4888aaff84c61dcafd9a : SUCCESS in 52m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bf32388a34994eb898f0ebc2c7f15b9e : SUCCESS in 53m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfdd816fe56947a799c5362548bac65b : SUCCESS in 43m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bda53c0431514491bdd037b23f0f27e2 : SUCCESS in 47m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/971b437b3d124956828d85c37b43ba72 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20232eb90d794c699f279af96bf8a61f : SUCCESS in 7m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e8184edf3274f6da15c228899cab630 : SUCCESS in 34m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aee73372f1d648a5833ee82d6e836f58 : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":4},{"id":"995c5f03fe6b939ae647a29ef8098d2f0420ce31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 01:29:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be535088f08c43bebe9edfc0adb3ca06 : SUCCESS in 29m 19s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eaa4818fdf9f4cafab6b77ddc18b1b0f : SUCCESS in 43m 00s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/0bfef7488130434cae21efb35dd24665 : SUCCESS in 1h 18m 07s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/713d64e2d5a34b16995258bbe6a619c8 : SUCCESS in 35m 40s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a1dc38855194bdeb3624bc986b805f7 : SUCCESS in 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"882121f20b9564c972e73c325fd0a02e4563ab6e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-08-21 04:25:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6c9c23162292386e6826a31f97d51dd30322e2cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-23 18:02:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"acd72faf57f6efded2f638e067df62f2381f4fea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 18:03:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7b7145e9c92b944947b14ed8e419108f9351e271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 23:33:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2027965ecdf848399a0465aa5ee7c131 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b41d47892f76434e85d5e046cd3c860d : SUCCESS in 59m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45f43048c1b74b559ee71f5a9dcc9561 : SUCCESS in 42m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/df0d24510bf84593b62c95abac359ed4 : POST_FAILURE in 9m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/242b67cba6b04d1dbfe56cc6c9fd5ada : POST_FAILURE in 9m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b72667ec9a842cba8146485f4e76267 : SUCCESS in 21m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2184512d19dd49ee88d45c8c8f65f81d : SUCCESS in 16m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/261ca1df9aa84a75be998c9e2fd68b42 : SUCCESS in 32m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/52b4439abbfb45a6be03af60a61a4c54 : SUCCESS in 17m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48bd67ee6fd04569bca66c2591245dca : SUCCESS in 34m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb40a39f3d4f42489893a899a687abf1 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e81e75d367514c83a715965fa0414d4f : SUCCESS in 28m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76f302f15f3f4ac69e1ab55425e215ab : SUCCESS in 19m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a033d6664bad43c89898b38322bc2ff6 : SUCCESS in 27m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2b074ba2845c4bfaa5f2003355948239 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5846692d9344997a1b2bcc6a8ddc27f : SUCCESS in 19m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a21665c6a9e41b4acfe6ade01f6499f : SUCCESS in 9m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/69a7e5f9f2164e4180740aeb6b82574a : SUCCESS in 47m 57s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b6079716ecb04b109c77b1b4c3aa41fa : SUCCESS in 56m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc9b271f44604e0c920acd39f566d7e3 : SUCCESS in 40m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d183cc9f3ec2402e9ed5401b8d5059d0 : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb51f03a44f545668b2d11e59b7a6b3e : SUCCESS in 20m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e89840d448574fd5a4a186bf5f7d3b5d : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c74ec5d194044afe8bed3c87484b4be4 : SUCCESS in 24m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/315ac4697d71492699739b69131f97c0 : SUCCESS in 44m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a8bed3107d00058ec1515ca4a59e5c37f32522fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 23:33:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b21963affc197ebae53963937c20d0e40351f6b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 23:35:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d38ef9856434aeeb6d98a70296cf8ba : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0c097e8cc67948d4bc1e9bc9c6671210 : FAILURE in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/dc229e9f4d0b4d638c6fc89fe5c79244 : FAILURE in 51s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7a105b5ef01e990d52ef2111b23707904cd41867","revisions":{"ca1458b8882ef2849c6924885907c4199e5230d9":{"kind":"REWORK","_number":1,"created":"2018-02-06 06:27:24.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/41/541141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/541141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/541141/1"}}},"commit":{"parents":[{"commit":"32d1b3246f6e5fd6176d2af1c22a3ac6d0fe6a17","subject":"Merge \"authors/changelog updates for 2.17.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32d1b3246f6e5fd6176d2af1c22a3ac6d0fe6a17"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2018-02-06 06:19:25.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2018-02-06 06:19:25.000000000","tz":0},"subject":"Add and pipe reconstructor stats through recon","message":"Add and pipe reconstructor stats through recon\n\nThis patch plumbs the object-reconstructor stats that are dropped\ninto recon cache out through the middleware and swift-recon tool.\n\nThis adds a \u0027/recon/reconstuction/object\u0027 to the middleware. As such\nthe swift-recon tool has grown a \u0027-R\u0027 or \u0027--reconstruction\u0027 option\naccess this data from each node.\n\nPlus some tests and documentation updates.\n\nChange-Id: I98582732ca5ccb2e7d2369b53abf9aa8c0ede00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca1458b8882ef2849c6924885907c4199e5230d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca1458b8882ef2849c6924885907c4199e5230d9"}]},"branch":"refs/heads/master"},"8302ba6f7803514ddeb0d4a01f80c3f41a22a17d":{"kind":"REWORK","_number":2,"created":"2021-07-27 07:08:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/41/541141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/541141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/541141/2"}}},"commit":{"parents":[{"commit":"a8f15128639f7e62a3e8d51c607b45c496206191","subject":"AUTHORS/CHANGELOG for 2.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8f15128639f7e62a3e8d51c607b45c496206191"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2018-02-06 06:19:25.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-27 07:07:41.000000000","tz":600},"subject":"Add and pipe reconstructor stats through recon","message":"Add and pipe reconstructor stats through recon\n\nThis patch plumbs the object-reconstructor stats that are dropped\ninto recon cache out through the middleware and swift-recon tool.\n\nThis adds a \u0027/recon/reconstuction/object\u0027 to the middleware. As such\nthe swift-recon tool has grown a \u0027-R\u0027 or \u0027--reconstruction\u0027 option\naccess this data from each node.\n\nPlus some tests and documentation updates.\n\nChange-Id: I98582732ca5ccb2e7d2369b53abf9aa8c0ede00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8302ba6f7803514ddeb0d4a01f80c3f41a22a17d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8302ba6f7803514ddeb0d4a01f80c3f41a22a17d"}]},"branch":"refs/heads/master"},"436c0ca4a1440fc214f57c320e290397be646b48":{"kind":"REWORK","_number":3,"created":"2021-07-27 07:16:23.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/41/541141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/541141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/541141/3"}}},"commit":{"parents":[{"commit":"a8f15128639f7e62a3e8d51c607b45c496206191","subject":"AUTHORS/CHANGELOG for 2.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8f15128639f7e62a3e8d51c607b45c496206191"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2018-02-06 06:19:25.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-27 07:15:40.000000000","tz":600},"subject":"Add and pipe reconstructor stats through recon","message":"Add and pipe reconstructor stats through recon\n\nThis patch plumbs the object-reconstructor stats that are dropped\ninto recon cache out through the middleware and swift-recon tool.\n\nThis adds a \u0027/recon/reconstuction/object\u0027 to the middleware. As such\nthe swift-recon tool has grown a \u0027-R\u0027 or \u0027--reconstruction\u0027 option\naccess this data from each node.\n\nPlus some tests and documentation updates.\n\nChange-Id: I98582732ca5ccb2e7d2369b53abf9aa8c0ede00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/436c0ca4a1440fc214f57c320e290397be646b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/436c0ca4a1440fc214f57c320e290397be646b48"}]},"branch":"refs/heads/master"},"7a105b5ef01e990d52ef2111b23707904cd41867":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-08-20 00:04:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/41/541141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/541141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/541141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/541141/4"}}},"commit":{"parents":[{"commit":"a8f15128639f7e62a3e8d51c607b45c496206191","subject":"AUTHORS/CHANGELOG for 2.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8f15128639f7e62a3e8d51c607b45c496206191"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2018-02-06 06:19:25.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-08-20 00:03:40.000000000","tz":0},"subject":"Add and pipe reconstructor stats through recon","message":"Add and pipe reconstructor stats through recon\n\nThis patch plumbs the object-reconstructor stats that are dropped\ninto recon cache out through the middleware and swift-recon tool.\n\nThis adds a \u0027/recon/reconstruction/object\u0027 to the middleware. As such\nthe swift-recon tool has grown a \u0027-R\u0027 or \u0027--reconstruction\u0027 option\naccess this data from each node.\n\nPlus some tests and documentation updates.\n\nChange-Id: I98582732ca5ccb2e7d2369b53abf9aa8c0ede00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a105b5ef01e990d52ef2111b23707904cd41867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a105b5ef01e990d52ef2111b23707904cd41867"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
