)]}'
{"id":"openstack%2Fswift~548948","triplet_id":"openstack%2Fswift~master~I29e30ef45fe3f8a026e7897127ffae08a6a80cd9","project":"openstack/swift","branch":"master","topic":"logFormat","hashtags":[],"change_id":"I29e30ef45fe3f8a026e7897127ffae08a6a80cd9","subject":"Make log format for requests configurable","status":"MERGED","created":"2018-03-01 13:23:33.000000000","updated":"2019-05-03 21:08:43.000000000","submitted":"2019-05-03 21:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"548948-1556917723330-76a6aaaf","meta_rev_id":"c01e3ac1a322175cc69a25ba7a204e74410969ee","_number":548948,"virtual_id_number":548948,"owner":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":2,"date":"2019-05-03 21:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},{"value":0,"date":"2019-05-03 17:07:44.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-05-03 15:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2019-05-03 15:59:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-01 13:23:52.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2018-09-13 03:44:22.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2018-10-26 22:15:24.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-02 22:21:41.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2019-05-03 15:59:50.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2019-05-03 17:07:44.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-03 21:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f34ab88f9631e2bdf7f25f2c766921745f4d1234","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-03-01 13:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe53bc16321ee4f36c4451c2495e1d58027962e4","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2018-03-01 13:23:52.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\n    Your patch will now be tested automatically by OpenStack testing frameworks\n    and once the automatic tests pass, it will be reviewed by other friendly\n    developers. They will give you feedback and may require you to refine it.\n\n    People seldom get their patch approved on the first try, so don\u0027t be\n    concerned if requested to make corrections. Feel free to modify your patch\n    and resubmit a new change-set.\n\n    Patches usually take 3 to 7 days to be reviewed so be patient and be\n    available on IRC to ask and answer questions about your work. Also it\n    takes generally at least a couple of weeks for cores to get around to\n    reviewing code. The more you participate in the community the more\n    rewarding it is for you. You may also notice that the more you get to know\n    people and get to be known, the faster your patches will be reviewed and\n    eventually approved. Get to know others and become known by doing code\n    reviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\n    Thanks again for supporting OpenStack, we look forward to working with you.\n\n    IRC: https://wiki.openstack.org/wiki/IRC\n    Workflow: http://docs.openstack.org/infra/manual/developers.html\n    Commit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"f0716c58a128399bf0c8e0f46b81f3f68fe69e02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-01 14:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/1/check/build-openstack-sphinx-docs/59f0902/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/1/check/openstack-tox-pep8/872abdf/ : SUCCESS in 4m 05s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/1/check/legacy-swift-dsvm-functional/6faca09/ : SUCCESS in 22m 52s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/1/check/legacy-swift-dsvm-functional-identity-v3-only/4ceb014/ : SUCCESS in 22m 55s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/1/check/legacy-tempest-dsvm-neutron-full/7960b83/ : SUCCESS in 1h 23m 11s\n- neutron-grenade http://logs.openstack.org/48/548948/1/check/neutron-grenade/b8395f8/ : SUCCESS in 56m 00s\n- swift-tox-py27 http://logs.openstack.org/48/548948/1/check/swift-tox-py27/4b44fe6/ : SUCCESS in 7m 01s\n- swift-tox-py35 http://logs.openstack.org/48/548948/1/check/swift-tox-py35/f560834/ : SUCCESS in 3m 02s\n- swift-tox-func http://logs.openstack.org/48/548948/1/check/swift-tox-func/ab8b08b/ : SUCCESS in 7m 11s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/1/check/swift-tox-func-encryption/d96541b/ : SUCCESS in 8m 20s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/1/check/swift-tox-func-ec/747440b/ : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"78f8c96d08afd2ede6d8586ca67f56836a83d8d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-03-02 01:06:10.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI generally like it; seems useful for GDPR, yeah? Might be nice to document it in etc/proxy-server.conf-sample too.\n\nAre we going to need similar changes to the various storage servers? Say around https://github.com/openstack/swift/blob/2.17.0/swift/obj/server.py#L1162-L1168\n\nOr is the plan to just turn off log_requests on the storage servers?","accounts_in_message":[],"_revision_number":1},{"id":"e161816cb3ddaad400f2e6244f07fa0dc8dd098e","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-03-05 10:11:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a23091198e46839e9191aa750c7d9e58709e9296","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-03-05 10:21:33.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nHi,\n\nYes, it could be ;)\n\nFor now, I only change the access-logs part which is easy because the log building is already well split. For the others parts (rest of proxy, object servers, ...) it will be more complex, but I think it can be useful for operators to be able to customize all logs for a better integration with log management tools.\n\nAdding the log-name in the log message didn\u0027t change the final log. In fact, the log-name is added here by default (\u0027server\u0027 variable): https://github.com/openstack/swift/blob/master/swift/common/utils.py#L1873. In this change, we override the fmt parameter with \"%(message)s\" only, and the \u0027server\u0027 variable (the log-name), is added in the log building.\n\nI forgot to edit etc/proxy-server.conf-sample and the doc/source/logs.rst, it\u0027s done.","accounts_in_message":[],"_revision_number":2},{"id":"8ab3b617f0fb7cc2449a15aee6a10218d921438e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-05 11:34:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/2/check/build-openstack-sphinx-docs/f01f910/html/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/2/check/openstack-tox-pep8/07384a1/ : SUCCESS in 3m 37s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/2/check/legacy-swift-dsvm-functional/bed77c9/ : SUCCESS in 23m 21s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/2/check/legacy-swift-dsvm-functional-identity-v3-only/c4ae270/ : SUCCESS in 23m 37s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/2/check/legacy-tempest-dsvm-neutron-full/3efc945/ : SUCCESS in 1h 21m 07s\n- neutron-grenade http://logs.openstack.org/48/548948/2/check/neutron-grenade/266f08c/ : SUCCESS in 42m 06s\n- swift-tox-py27 http://logs.openstack.org/48/548948/2/check/swift-tox-py27/6a03719/ : SUCCESS in 7m 14s\n- swift-tox-py35 http://logs.openstack.org/48/548948/2/check/swift-tox-py35/0e1f996/ : SUCCESS in 3m 24s\n- swift-tox-func http://logs.openstack.org/48/548948/2/check/swift-tox-func/0b29a03/ : SUCCESS in 6m 44s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/2/check/swift-tox-func-encryption/cd12cd2/ : SUCCESS in 6m 36s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/2/check/swift-tox-func-ec/c447408/ : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4c0a3814374dfd1483c2be129ce3832a2c17d746","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-03-07 17:50:08.000000000","message":"Patch Set 2:\n\nI love the work here.\n\nI\u0027m concerned about adding a new field to the beginning of the line. Doing so will break existing log parsers, right? Can we add the field to the end of the line instead? That will allow well-written log parsers to still work after the cluster is upgraded.","accounts_in_message":[],"_revision_number":2},{"id":"5cc62c00235514deec491c2d7e625cf3505cfbf4","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-03-08 11:13:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de5ce56ea91744bdde2e777df8526ee4c32dbfe7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-08 12:47:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/3/check/build-openstack-sphinx-docs/8a3a5f5/html/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/3/check/openstack-tox-pep8/f1fdd9a/ : SUCCESS in 3m 53s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/3/check/legacy-swift-dsvm-functional/a1b8be9/ : SUCCESS in 22m 49s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/3/check/legacy-swift-dsvm-functional-identity-v3-only/dfe892d/ : SUCCESS in 21m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/3/check/legacy-tempest-dsvm-neutron-full/4636405/ : SUCCESS in 1h 32m 32s\n- neutron-grenade http://logs.openstack.org/48/548948/3/check/neutron-grenade/d1cf02d/ : SUCCESS in 42m 21s\n- swift-tox-py27 http://logs.openstack.org/48/548948/3/check/swift-tox-py27/4890936/ : SUCCESS in 6m 47s\n- swift-tox-py35 http://logs.openstack.org/48/548948/3/check/swift-tox-py35/f1f16ee/ : SUCCESS in 3m 33s\n- swift-tox-func http://logs.openstack.org/48/548948/3/check/swift-tox-func/dad7439/ : SUCCESS in 7m 57s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/3/check/swift-tox-func-encryption/21b2382/ : SUCCESS in 7m 35s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/3/check/swift-tox-func-ec/4a40a37/ : SUCCESS in 9m 59s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/3/check/swift-probetests-centos-7/f38edde/ : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"078c0e5bc5e1ab2c91cd153295643f9beaf03804","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-03-08 14:11:21.000000000","message":"Patch Set 3:\n\nHi Tim and John,\n\nI understand your concerns about the \"{log_name}: \" that I added in the message part of logs.\nI removed this part in the Patch Set 3 I just pushed. Now, it doesn\u0027t change the default log format.\n\nThere are 2 things I want to do:\nI want to be able to customize logs (I want to write a template in config to have my logs in a LTSV format).\nI want to have anonymized logs.\n\nThe reason why I added the \"{log_name}: \" is because there is a point that annoys me for the LTSV formatting:\nThe default format (fmt) in the get_logger() in common/utils.py is: \"%(server)s: %(message)s\". There is a space character after the \u0027:\u0027. In Syslog, the message part of the log is everything after this \u0027:\u0027 including this space. So, in my actual log management process, I have to remove it to get exactly the same part of the message (equivalent to the %(message)).\n\nMaybe is it worth to work on the templating of the message in a first change, and then we will think about the format of logs in a second patch. It could be the opportunity to review the format for all servers and not only the access-logs part of the proxy server.","accounts_in_message":[],"_revision_number":3},{"id":"8bcfc84cde4fb70d459f34e4bae7d7d96f1f598d","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-05-30 16:21:13.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"444423fbe943d7b0bb1aabed8a0dca5299373ddd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-30 18:49:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/4/check/build-openstack-sphinx-docs/f92e98c/html/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/4/check/openstack-tox-pep8/2aff83f/ : SUCCESS in 5m 29s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/4/check/legacy-swift-dsvm-functional/bed2654/ : SUCCESS in 23m 14s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/4/check/legacy-swift-dsvm-functional-identity-v3-only/59ad2f7/ : SUCCESS in 21m 50s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/4/check/legacy-tempest-dsvm-neutron-full/0c1dddf/ : SUCCESS in 2h 20m 19s\n- neutron-grenade http://logs.openstack.org/48/548948/4/check/neutron-grenade/a8aef7a/ : SUCCESS in 1h 27m 37s\n- swift-tox-py27 http://logs.openstack.org/48/548948/4/check/swift-tox-py27/5021104/ : SUCCESS in 11m 40s\n- swift-tox-py35 http://logs.openstack.org/48/548948/4/check/swift-tox-py35/2f86dff/ : SUCCESS in 4m 41s\n- swift-tox-func http://logs.openstack.org/48/548948/4/check/swift-tox-func/6dfa900/ : SUCCESS in 8m 51s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/4/check/swift-tox-func-encryption/e7843c9/ : SUCCESS in 10m 18s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/4/check/swift-tox-func-domain-remap-staticweb/3550872/ : SUCCESS in 7m 51s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/4/check/swift-tox-func-ec/b86972a/ : SUCCESS in 9m 20s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/4/check/swift-tox-func-s3api/3771621/ : SUCCESS in 9m 54s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/4/check/swift-probetests-centos-7/943ce00/ : SUCCESS in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be74264c32aacbdc99645121e79982c37c3a5edc","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-05-30 21:54:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nLooks good, some small things to fix.","accounts_in_message":[],"_revision_number":4},{"id":"a6b375949f0607a54caf4562c49c5ec1545b4280","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-06-12 13:02:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"59d00b969d0c59d398ecc3750108bde1c4ea9d96","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-06-12 13:12:43.000000000","message":"Patch Set 5:\n\n\u003e Uploaded patch set 5.\n\nHi Romain.\n\nIn this change, I fixed what you said. I changed the config variable name for the templating of logs to \"log_msg_template\".\n\nI also added the variable \"log_anonymization_salt\" which is a salt for md5 hashs of anonymized information.\n\nThanks for your advices.","accounts_in_message":[],"_revision_number":5},{"id":"a54607e00e8dde64887ce428b561c5b42f3b56bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-12 14:37:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/5/check/build-openstack-sphinx-docs/531dce6/html/ : SUCCESS in 6m 51s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/5/check/openstack-tox-pep8/5e9e34d/ : SUCCESS in 6m 17s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/5/check/legacy-swift-dsvm-functional/e5157dd/ : SUCCESS in 31m 30s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/5/check/legacy-swift-dsvm-functional-identity-v3-only/0672fd8/ : SUCCESS in 24m 55s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/5/check/legacy-tempest-dsvm-neutron-full/e419d2f/ : SUCCESS in 1h 24m 04s\n- neutron-grenade http://logs.openstack.org/48/548948/5/check/neutron-grenade/80e1c91/ : SUCCESS in 52m 25s\n- swift-tox-py27 http://logs.openstack.org/48/548948/5/check/swift-tox-py27/863599d/ : SUCCESS in 11m 11s\n- swift-tox-py35 http://logs.openstack.org/48/548948/5/check/swift-tox-py35/d551f36/ : SUCCESS in 5m 31s\n- swift-tox-func http://logs.openstack.org/48/548948/5/check/swift-tox-func/5f329a7/ : SUCCESS in 8m 26s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/5/check/swift-tox-func-encryption/5c142f6/ : SUCCESS in 9m 41s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/5/check/swift-tox-func-domain-remap-staticweb/8c77b4e/ : SUCCESS in 10m 46s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/5/check/swift-tox-func-ec/d9e8200/ : SUCCESS in 11m 01s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/5/check/swift-tox-func-s3api/d2bb887/ : SUCCESS in 10m 46s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/5/check/swift-probetests-centos-7/dd69131/ : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"767352ccf2f1174b2f3fbbbf520097769b9bc325","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-06-14 22:40:06.000000000","message":"Patch Set 5: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"275a62dd0786b82739903572d27a2cd8a0d4c5f6","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-06-18 13:26:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"00ae3fc71a64bedbb8852d4fa1e9ac802a84e6b5","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-06-18 13:45:03.000000000","message":"Patch Set 6:\n\n\u003e Uploaded patch set 6.\n\nHi Tim,\n\nIt is true that the management of \u0027keywords\u0027 in replacement was weird.\nI changed many things in this patch set :)\n\nNow, I use the possibility of the formater to call properties during formatting. In the log_msg_template it is possible to write {object} or {object.anonymized} (second one is the hash). For timestamps it is possible to get a date with {end_time.datetime} or to build date with {end_time.H} (using strftime directives).\n\nAnd, if the configuration is wrong, an error will be raised during the initialization and not later :)\n\nThanks for your comments,","accounts_in_message":[],"_revision_number":6},{"id":"eec1b279878436ac1ff271a4b63911e9bf422123","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-18 14:58:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/6/check/build-openstack-sphinx-docs/b307f9c/html/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/6/check/openstack-tox-pep8/5a1cb11/ : FAILURE in 5m 35s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/6/check/legacy-swift-dsvm-functional/47c799b/ : SUCCESS in 28m 15s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/6/check/legacy-swift-dsvm-functional-identity-v3-only/a0d97f7/ : SUCCESS in 34m 11s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/6/check/legacy-tempest-dsvm-neutron-full/89110a2/ : SUCCESS in 1h 28m 04s\n- neutron-grenade http://logs.openstack.org/48/548948/6/check/neutron-grenade/df09fd4/ : SUCCESS in 48m 00s\n- swift-tox-py27 http://logs.openstack.org/48/548948/6/check/swift-tox-py27/f7ce641/ : SUCCESS in 10m 00s\n- swift-tox-py35 http://logs.openstack.org/48/548948/6/check/swift-tox-py35/19cabc9/ : SUCCESS in 5m 44s\n- swift-tox-func http://logs.openstack.org/48/548948/6/check/swift-tox-func/ccddcb1/ : SUCCESS in 10m 38s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/6/check/swift-tox-func-encryption/0f47733/ : SUCCESS in 11m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/6/check/swift-tox-func-domain-remap-staticweb/07d35f5/ : SUCCESS in 9m 32s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/6/check/swift-tox-func-ec/53d7d83/ : SUCCESS in 9m 28s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/6/check/swift-tox-func-s3api/9e97de7/ : SUCCESS in 10m 03s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/6/check/swift-probetests-centos-7/fd210cc/ : SUCCESS in 32m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0b0c6ad58c75c64714765ff5bae838744f5181a","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-06-18 15:25:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e0abb044643c9b116bdf02ba9718ef998fb20ba8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-18 17:49:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/7/check/build-openstack-sphinx-docs/071b5b8/html/ : SUCCESS in 5m 50s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/7/check/openstack-tox-pep8/f617d50/ : SUCCESS in 6m 16s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/7/check/legacy-swift-dsvm-functional/6a38f7d/ : SUCCESS in 26m 43s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/7/check/legacy-swift-dsvm-functional-identity-v3-only/444bfbe/ : SUCCESS in 25m 03s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/7/check/legacy-tempest-dsvm-neutron-full/598d196/ : SUCCESS in 1h 56m 56s\n- neutron-grenade http://logs.openstack.org/48/548948/7/check/neutron-grenade/c23c37f/ : SUCCESS in 47m 14s\n- swift-tox-py27 http://logs.openstack.org/48/548948/7/check/swift-tox-py27/8980bed/ : SUCCESS in 10m 42s\n- swift-tox-py35 http://logs.openstack.org/48/548948/7/check/swift-tox-py35/bb309ba/ : SUCCESS in 5m 49s\n- swift-tox-func http://logs.openstack.org/48/548948/7/check/swift-tox-func/c05efde/ : SUCCESS in 9m 24s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/7/check/swift-tox-func-encryption/4b5fb76/ : SUCCESS in 8m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/7/check/swift-tox-func-domain-remap-staticweb/a889077/ : SUCCESS in 9m 14s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/7/check/swift-tox-func-ec/92dfc83/ : SUCCESS in 9m 37s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/7/check/swift-tox-func-s3api/ca4f72d/ : SUCCESS in 10m 24s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/7/check/swift-probetests-centos-7/37d6b2c/ : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"06432bd3ee6fb19d0dc1eef47932efa73a9b2b2c","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-08-20 16:21:14.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"a0d74421b28405caa9105cffe1aea0e25ecb23cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-20 17:56:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/48/548948/8/check/build-openstack-sphinx-docs/b7b8ce6/html/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/8/check/openstack-tox-pep8/d200acf/ : SUCCESS in 5m 38s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/8/check/legacy-swift-dsvm-functional/b6990ec/ : SUCCESS in 28m 45s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/8/check/legacy-swift-dsvm-functional-identity-v3-only/521b906/ : SUCCESS in 24m 27s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/8/check/legacy-tempest-dsvm-neutron-full/2284842/ : SUCCESS in 1h 30m 49s\n- neutron-grenade http://logs.openstack.org/48/548948/8/check/neutron-grenade/e97f4d5/ : SUCCESS in 47m 48s\n- swift-tox-py27 http://logs.openstack.org/48/548948/8/check/swift-tox-py27/8764e96/ : SUCCESS in 11m 04s\n- swift-tox-py35 http://logs.openstack.org/48/548948/8/check/swift-tox-py35/6d436ab/ : FAILURE in 4m 51s\n- swift-tox-func http://logs.openstack.org/48/548948/8/check/swift-tox-func/e7c97c4/ : SUCCESS in 8m 10s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/8/check/swift-tox-func-encryption/9e2df45/ : SUCCESS in 12m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/8/check/swift-tox-func-domain-remap-staticweb/6bf4963/ : SUCCESS in 8m 39s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/8/check/swift-tox-func-ec/be2e151/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/8/check/swift-tox-func-s3api/49bd6ed/ : SUCCESS in 18m 45s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/8/check/swift-probetests-centos-7/cb61663/ : SUCCESS in 25m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/548948/8/check/openstack-tox-lower-constraints/e38fd28/ : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":8},{"id":"350f2b8272b3cb2bcb1f20e7f769b077eef952e7","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-09-07 07:15:22.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"bbea5a0fe46bb538041b1a2a1a5ce6664a0f4820","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 08:56:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/9/check/openstack-tox-docs/89814d7/html/ : SUCCESS in 4m 20s\n- swift-tox-py27 http://logs.openstack.org/48/548948/9/check/swift-tox-py27/8979feb/ : SUCCESS in 11m 36s\n- swift-tox-py35 http://logs.openstack.org/48/548948/9/check/swift-tox-py35/c0a14f1/ : FAILURE in 5m 05s\n- swift-tox-func http://logs.openstack.org/48/548948/9/check/swift-tox-func/d3d4200/ : SUCCESS in 12m 16s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/9/check/swift-tox-func-encryption/d510d07/ : SUCCESS in 15m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/9/check/swift-tox-func-domain-remap-staticweb/90df951/ : SUCCESS in 9m 32s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/9/check/swift-tox-func-ec/a50285a/ : SUCCESS in 10m 17s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/9/check/swift-tox-func-s3api/332500e/ : SUCCESS in 11m 24s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/9/check/swift-probetests-centos-7/db90e5b/ : SUCCESS in 30m 14s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/9/check/swift-dsvm-functional/aff34ff/ : SUCCESS in 27m 19s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/9/check/legacy-tempest-dsvm-neutron-full/9ee2519/ : SUCCESS in 1h 30m 43s\n- neutron-grenade http://logs.openstack.org/48/548948/9/check/neutron-grenade/3360179/ : SUCCESS in 51m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/548948/9/check/openstack-tox-lower-constraints/3d5fda0/ : SUCCESS in 7m 40s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/9/check/openstack-tox-pep8/b07395e/ : SUCCESS in 5m 22s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/9/check/legacy-swift-dsvm-functional/6a4989a/ : SUCCESS in 28m 43s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/9/check/legacy-swift-dsvm-functional-identity-v3-only/1efd5eb/ : SUCCESS in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"402700ecbaf73df4c2ce8348405735e363f85a24","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-09-07 10:01:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"967010355ecc2f269b4918fa94476502cec4bdb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 11:37:59.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/10/check/openstack-tox-docs/1cc8646/html/ : SUCCESS in 4m 38s\n- swift-tox-py27 http://logs.openstack.org/48/548948/10/check/swift-tox-py27/31d127a/ : SUCCESS in 10m 14s\n- swift-tox-py35 http://logs.openstack.org/48/548948/10/check/swift-tox-py35/3541606/ : FAILURE in 4m 43s\n- swift-tox-func http://logs.openstack.org/48/548948/10/check/swift-tox-func/24fde10/ : SUCCESS in 8m 37s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/10/check/swift-tox-func-encryption/8ae9162/ : SUCCESS in 8m 56s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/10/check/swift-tox-func-domain-remap-staticweb/6247572/ : SUCCESS in 8m 59s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/10/check/swift-tox-func-ec/f8373f4/ : SUCCESS in 8m 48s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/10/check/swift-tox-func-s3api/91aede3/ : SUCCESS in 10m 08s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/10/check/swift-probetests-centos-7/f8857f7/ : SUCCESS in 23m 58s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/10/check/swift-dsvm-functional/2027770/ : SUCCESS in 27m 48s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/10/check/legacy-tempest-dsvm-neutron-full/c841a1c/ : SUCCESS in 1h 33m 47s\n- neutron-grenade http://logs.openstack.org/48/548948/10/check/neutron-grenade/79183ec/ : SUCCESS in 55m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/548948/10/check/openstack-tox-lower-constraints/1f2e596/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/10/check/openstack-tox-pep8/58b3407/ : SUCCESS in 4m 42s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/10/check/legacy-swift-dsvm-functional/4b44eed/ : SUCCESS in 24m 42s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/10/check/legacy-swift-dsvm-functional-identity-v3-only/215976b/ : SUCCESS in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"66992fff6c5cae963a20af59dae3937f1e338610","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-09-07 12:24:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a1e4a58e183c5d24c56b4c734707a943743976be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 14:07:22.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/11/check/openstack-tox-docs/bb40df6/html/ : SUCCESS in 4m 27s\n- swift-tox-py27 http://logs.openstack.org/48/548948/11/check/swift-tox-py27/a883379/ : SUCCESS in 9m 30s\n- swift-tox-py35 http://logs.openstack.org/48/548948/11/check/swift-tox-py35/cf3b441/ : FAILURE in 4m 41s\n- swift-tox-func http://logs.openstack.org/48/548948/11/check/swift-tox-func/686df72/ : SUCCESS in 10m 11s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/11/check/swift-tox-func-encryption/9702da5/ : SUCCESS in 9m 42s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/11/check/swift-tox-func-domain-remap-staticweb/7626f74/ : SUCCESS in 8m 09s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/11/check/swift-tox-func-ec/04ad43a/ : SUCCESS in 9m 03s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/11/check/swift-tox-func-s3api/4e3d26b/ : SUCCESS in 9m 08s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/11/check/swift-probetests-centos-7/ca8b944/ : SUCCESS in 25m 04s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/11/check/swift-dsvm-functional/37b1aa5/ : SUCCESS in 22m 58s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/11/check/legacy-tempest-dsvm-neutron-full/d30c1fa/ : SUCCESS in 1h 37m 07s\n- neutron-grenade http://logs.openstack.org/48/548948/11/check/neutron-grenade/7670605/ : SUCCESS in 49m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/548948/11/check/openstack-tox-lower-constraints/3cdf8a1/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/11/check/openstack-tox-pep8/51a7968/ : SUCCESS in 5m 14s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/11/check/legacy-swift-dsvm-functional/449ca81/ : SUCCESS in 26m 15s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/11/check/legacy-swift-dsvm-functional-identity-v3-only/6f4e641/ : SUCCESS in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"388380cb94a4a32f81f53d85500fda0994c6a4dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-09-07 17:07:21.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"4c0ecc13bf4ba2ded4da935939ae4e99df5b495f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 19:44:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/12/check/openstack-tox-docs/964e81d/html/ : SUCCESS in 3m 39s\n- swift-tox-py27 http://logs.openstack.org/48/548948/12/check/swift-tox-py27/a2155b1/ : SUCCESS in 10m 32s\n- swift-tox-py35 http://logs.openstack.org/48/548948/12/check/swift-tox-py35/416cfe0/ : SUCCESS in 4m 40s\n- swift-tox-func http://logs.openstack.org/48/548948/12/check/swift-tox-func/ef4077f/ : SUCCESS in 8m 08s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/12/check/swift-tox-func-encryption/3bed6fc/ : SUCCESS in 9m 13s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/12/check/swift-tox-func-domain-remap-staticweb/a190292/ : SUCCESS in 8m 27s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/12/check/swift-tox-func-ec/53899ba/ : FAILURE in 10m 20s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/12/check/swift-tox-func-s3api/d66c313/ : SUCCESS in 15m 45s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/12/check/swift-probetests-centos-7/95be030/ : SUCCESS in 25m 25s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/12/check/swift-dsvm-functional/d3aeb9b/ : SUCCESS in 28m 42s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/12/check/legacy-tempest-dsvm-neutron-full/53687cb/ : SUCCESS in 1h 31m 57s\n- neutron-grenade http://logs.openstack.org/48/548948/12/check/neutron-grenade/987535a/ : SUCCESS in 1h 02m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/548948/12/check/openstack-tox-lower-constraints/8c03245/ : SUCCESS in 8m 42s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/12/check/openstack-tox-pep8/653390d/ : SUCCESS in 4m 39s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/12/check/legacy-swift-dsvm-functional/81308d0/ : SUCCESS in 24m 46s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/12/check/legacy-swift-dsvm-functional-identity-v3-only/c704fb8/ : SUCCESS in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6ae4b4b89c5b624225d00edb87e2c8ffc7a0439e","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2018-09-10 16:25:26.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"2eaa790958f1d8f518523f1e0d2d28ec925591f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-10 18:13:05.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/12/check/openstack-tox-docs/386a19a/html/ : SUCCESS in 4m 04s\n- swift-tox-py27 http://logs.openstack.org/48/548948/12/check/swift-tox-py27/6b6b0d3/ : SUCCESS in 10m 59s\n- swift-tox-py35 http://logs.openstack.org/48/548948/12/check/swift-tox-py35/40d039b/ : SUCCESS in 4m 31s\n- swift-tox-func http://logs.openstack.org/48/548948/12/check/swift-tox-func/8506063/ : SUCCESS in 8m 03s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/12/check/swift-tox-func-encryption/e064e83/ : SUCCESS in 9m 04s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/12/check/swift-tox-func-domain-remap-staticweb/66a08e8/ : SUCCESS in 8m 27s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/12/check/swift-tox-func-ec/b0c4018/ : FAILURE in 10m 35s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/12/check/swift-tox-func-s3api/ac1b460/ : SUCCESS in 9m 44s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/12/check/swift-probetests-centos-7/23fceac/ : SUCCESS in 29m 49s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/12/check/swift-dsvm-functional/84f501d/ : SUCCESS in 25m 03s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/548948/12/check/legacy-tempest-dsvm-neutron-full/24154ab/ : SUCCESS in 1h 40m 38s\n- neutron-grenade http://logs.openstack.org/48/548948/12/check/neutron-grenade/727d0cd/ : SUCCESS in 49m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/548948/12/check/openstack-tox-lower-constraints/f096756/ : SUCCESS in 7m 43s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/12/check/openstack-tox-pep8/6f2b911/ : SUCCESS in 4m 52s\n- legacy-swift-dsvm-functional http://logs.openstack.org/48/548948/12/check/legacy-swift-dsvm-functional/98cd9d3/ : SUCCESS in 26m 46s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/48/548948/12/check/legacy-swift-dsvm-functional-identity-v3-only/3062ad5/ : SUCCESS in 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d59dfc2d54f3d7e63a09429a1db72f777cb025d0","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-09-13 03:12:11.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"6eb7647681771d150d20d6dc3ef9c8eae7515c30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 03:43:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/12/check/openstack-tox-docs/f12d25e/html/ : SUCCESS in 4m 15s\n- swift-tox-py27 http://logs.openstack.org/48/548948/12/check/swift-tox-py27/aa1fa9e/ : SUCCESS in 12m 45s\n- swift-tox-py35 http://logs.openstack.org/48/548948/12/check/swift-tox-py35/6802407/ : SUCCESS in 5m 12s\n- swift-tox-py36 http://logs.openstack.org/48/548948/12/check/swift-tox-py36/3a52627/ : SUCCESS in 6m 07s (non-voting)\n- swift-tox-func http://logs.openstack.org/48/548948/12/check/swift-tox-func/869756d/ : SUCCESS in 9m 20s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/12/check/swift-tox-func-encryption/e87f662/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/12/check/swift-tox-func-domain-remap-staticweb/2dcc803/ : SUCCESS in 10m 02s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/12/check/swift-tox-func-ec/9d43a9e/ : SUCCESS in 9m 57s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/12/check/swift-tox-func-s3api/81f5b8b/ : SUCCESS in 10m 34s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/12/check/swift-probetests-centos-7/06b6ca2/ : SUCCESS in 26m 27s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/12/check/swift-dsvm-functional/137023c/ : SUCCESS in 25m 22s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/12/check/swift-dsvm-functional-ipv6/33f65ee/ : SUCCESS in 21m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/548948/12/check/openstack-tox-lower-constraints/ce8991f/ : SUCCESS in 8m 41s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/12/check/openstack-tox-pep8/afe002b/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":12},{"id":"800cd8de3bfea9cd238003f2cd9797ebd528401a","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-09-13 03:44:22.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"4908c39f1bdf65762e90f3b0ac46aae06d1a5bc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 04:07:17.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 http://logs.openstack.org/48/548948/12/experimental/swift-tox-py27-centos-7/09e1942/ : SUCCESS in 13m 13s\n- swift-tox-func-centos-7 http://logs.openstack.org/48/548948/12/experimental/swift-tox-func-centos-7/3458736/ : SUCCESS in 10m 58s\n- swift-tox-func-encryption-centos-7 http://logs.openstack.org/48/548948/12/experimental/swift-tox-func-encryption-centos-7/7790ee6/ : SUCCESS in 11m 55s\n- swift-tox-func-ec-centos-7 http://logs.openstack.org/48/548948/12/experimental/swift-tox-func-ec-centos-7/71ce48c/ : SUCCESS in 21m 18s","accounts_in_message":[],"_revision_number":12},{"id":"4bf0026ecf7f1c7a47a9d24f4384f5188db5bd9d","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-26 09:03:50.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"552d490f726648384b9248333794299db16ae4be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 11:55:02.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/13/check/openstack-tox-docs/ce595a4/html/ : SUCCESS in 4m 13s\n- swift-tox-py27 http://logs.openstack.org/48/548948/13/check/swift-tox-py27/87a617c/ : SUCCESS in 10m 25s\n- swift-tox-py35 http://logs.openstack.org/48/548948/13/check/swift-tox-py35/d0b422a/ : FAILURE in 5m 24s\n- swift-tox-py36 http://logs.openstack.org/48/548948/13/check/swift-tox-py36/05c0ffd/ : FAILURE in 4m 57s (non-voting)\n- swift-tox-func http://logs.openstack.org/48/548948/13/check/swift-tox-func/1cef777/ : SUCCESS in 8m 33s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/13/check/swift-tox-func-encryption/52758d1/ : SUCCESS in 9m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/13/check/swift-tox-func-domain-remap-staticweb/b3cfd38/ : SUCCESS in 9m 06s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/13/check/swift-tox-func-ec/5cb6176/ : SUCCESS in 19m 11s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/13/check/swift-tox-func-s3api/0a28898/ : SUCCESS in 11m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/13/check/swift-tox-func-s3api-ceph-s3tests-tempauth/3a1023a/ : SUCCESS in 11m 13s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/13/check/swift-probetests-centos-7/691e46b/ : SUCCESS in 28m 09s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/13/check/swift-dsvm-functional/30a1e01/ : SUCCESS in 27m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/13/check/swift-dsvm-functional-ipv6/e06d765/ : SUCCESS in 29m 01s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/13/check/swift-tox-lower-constraints/77d472e/ : SUCCESS in 12m 04s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/13/check/openstack-tox-pep8/84ba6fa/ : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/13/check/swift-multinode-rolling-upgrade/5046ba9/ : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":13},{"id":"a569996a4cbfa3a70e7efd7e3bfd4a243f293009","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-26 17:47:53.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6c29e5129fc977f4d652fd53e5333448ef4f20a0","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-10-26 22:15:24.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0630a862fd150d4c110c87c93849735866fd14d3","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-29 09:51:14.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"26e5b567d750ce46131f07750dbdec071dc24abf","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-29 10:03:36.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"70269e31e87716fe59da15cdd8335a905935d1b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-29 10:39:30.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/14/check/openstack-tox-docs/64a1b3c/html/ : SUCCESS in 4m 33s\n- swift-tox-py27 http://logs.openstack.org/48/548948/14/check/swift-tox-py27/3689f93/ : SUCCESS in 10m 48s\n- swift-tox-py35 http://logs.openstack.org/48/548948/14/check/swift-tox-py35/b1f6a1d/ : FAILURE in 5m 54s\n- swift-tox-py36 http://logs.openstack.org/48/548948/14/check/swift-tox-py36/aa56535/ : FAILURE in 4m 32s (non-voting)\n- swift-tox-func http://logs.openstack.org/48/548948/14/check/swift-tox-func/55cdbcd/ : FAILURE in 20m 45s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/14/check/swift-tox-func-encryption/1f41dcf/ : SUCCESS in 20m 56s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/14/check/swift-tox-func-domain-remap-staticweb/466e74b/ : SUCCESS in 8m 10s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/14/check/swift-tox-func-ec/3e44ac4/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/14/check/swift-tox-func-s3api/966a6f4/ : SUCCESS in 10m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/14/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f72bd45/ : SUCCESS in 9m 09s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/14/check/swift-probetests-centos-7/58a9888/ : SUCCESS in 27m 10s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/14/check/swift-dsvm-functional/4b19ead/ : SUCCESS in 27m 06s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/14/check/swift-dsvm-functional-ipv6/e1db44d/ : SUCCESS in 26m 41s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/14/check/swift-tox-lower-constraints/327b407/ : SUCCESS in 10m 08s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/14/check/openstack-tox-pep8/a07c22f/ : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/14/check/swift-multinode-rolling-upgrade/f4e6545/ : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":14},{"id":"b080dacac76fcb47e3a21bb09febfe86cf28ac2d","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-29 16:22:01.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6fb15c768479dda9a483f46bb96c0bea25fc4005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-29 21:59:45.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/15/check/openstack-tox-docs/ccc3c73/html/ : SUCCESS in 4m 09s\n- swift-tox-py27 http://logs.openstack.org/48/548948/15/check/swift-tox-py27/899e87c/ : SUCCESS in 10m 36s\n- swift-tox-py35 http://logs.openstack.org/48/548948/15/check/swift-tox-py35/31ac4e6/ : SUCCESS in 4m 57s\n- swift-tox-py36 http://logs.openstack.org/48/548948/15/check/swift-tox-py36/596819f/ : SUCCESS in 5m 19s (non-voting)\n- swift-tox-func http://logs.openstack.org/48/548948/15/check/swift-tox-func/fd1a9c8/ : SUCCESS in 10m 25s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/15/check/swift-tox-func-encryption/75f00e2/ : SUCCESS in 13m 39s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/15/check/swift-tox-func-domain-remap-staticweb/da78b79/ : SUCCESS in 14m 32s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/15/check/swift-tox-func-ec/81c1eda/ : SUCCESS in 14m 33s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/15/check/swift-tox-func-s3api/449fb3f/ : SUCCESS in 15m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/15/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0c1e25f/ : SUCCESS in 10m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/15/check/swift-probetests-centos-7/c97f4d7/ : SUCCESS in 34m 25s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/15/check/swift-dsvm-functional/efa3873/ : SUCCESS in 26m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/15/check/swift-dsvm-functional-ipv6/066ec43/ : SUCCESS in 26m 56s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/15/check/swift-tox-lower-constraints/8fac530/ : SUCCESS in 10m 24s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/15/check/openstack-tox-pep8/6ca5e5f/ : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/15/check/swift-multinode-rolling-upgrade/6c81989/ : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":15},{"id":"f6e9a63ea9dc99bb5224d0806929c3aab4bcaeb1","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2019-04-19 14:58:17.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"607844bfc714a2067b46ebfc186df7f20496f412","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2019-04-24 13:31:37.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"94eeee915a19e1f748426ec3c10d2c96d31112c8","author":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"date":"2019-04-24 13:32:39.000000000","message":"Patch Set 17:\n\n\u003e Patch Set 16: Patch Set 15 was rebased\n\nI think Zuul failed to start tests. Rebased again","accounts_in_message":[],"_revision_number":17},{"id":"48c6a235eff8d0490521029b6679400b21fd18ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 17:58:55.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/17/check/openstack-tox-docs/e7b4957/html/ : SUCCESS in 3m 42s\n- swift-tox-py27 http://logs.openstack.org/48/548948/17/check/swift-tox-py27/e47abfa/ : SUCCESS in 9m 54s\n- swift-tox-py35 http://logs.openstack.org/48/548948/17/check/swift-tox-py35/c8604cf/ : SUCCESS in 8m 12s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/48/548948/17/check/swift-tox-py36/b6e62bc/ : SUCCESS in 12m 13s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/48/548948/17/check/swift-tox-py37/b2e59fe/ : SUCCESS in 8m 44s\n- swift-tox-func http://logs.openstack.org/48/548948/17/check/swift-tox-func/3c8a055/ : SUCCESS in 9m 26s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/17/check/swift-tox-func-encryption/788f780/ : SUCCESS in 8m 42s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/17/check/swift-tox-func-domain-remap-staticweb/fbc71eb/ : SUCCESS in 12m 01s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/17/check/swift-tox-func-ec/26f7096/ : SUCCESS in 26m 06s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/17/check/swift-tox-func-s3api/f4aa452/ : SUCCESS in 16m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/17/check/swift-tox-func-s3api-ceph-s3tests-tempauth/755a6f1/ : SUCCESS in 11m 53s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/17/check/swift-probetests-centos-7/98dad3a/ : FAILURE in 26m 02s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/17/check/swift-dsvm-functional/119fce3/ : SUCCESS in 25m 40s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/17/check/swift-dsvm-functional-ipv6/f7408c3/ : SUCCESS in 27m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/17/check/swift-tox-lower-constraints/27b18ab/ : SUCCESS in 11m 03s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/17/check/openstack-tox-pep8/8054e52/ : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/17/check/swift-multinode-rolling-upgrade/f1e3459/ : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":17},{"id":"aa48492d689e0d597c1f6bd4f6ec1e0cf37ae3e3","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2019-05-02 20:44:35.000000000","message":"Patch Set 17:\n\n(2 comments)\n\nThis looks really good. I\u0027ve got a couple of questions inline.","accounts_in_message":[],"_revision_number":17},{"id":"5b424e35669c16421df52f902e51c13b85fda882","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2019-05-02 22:16:28.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9616367f4735047cac9639fdaadf45dd8bae1efa","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2019-05-02 22:20:06.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"6b8fd314c1803ab4d9a6e3eeee7f7bb863466f52","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2019-05-02 22:21:41.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"8d1b9341069ebf62f58c36f3232c6fc84858c4f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 23:03:53.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/19/check/openstack-tox-docs/f237738/html/ : SUCCESS in 3m 55s\n- swift-tox-py27 http://logs.openstack.org/48/548948/19/check/swift-tox-py27/27ea4f9/ : SUCCESS in 12m 01s\n- swift-tox-py35 http://logs.openstack.org/48/548948/19/check/swift-tox-py35/9a98ee6/ : SUCCESS in 9m 24s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/48/548948/19/check/swift-tox-py36/80cf5d7/ : SUCCESS in 9m 57s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/48/548948/19/check/swift-tox-py37/4648c6a/ : FAILURE in 8m 53s\n- swift-tox-func http://logs.openstack.org/48/548948/19/check/swift-tox-func/e776e7b/ : SUCCESS in 9m 41s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/19/check/swift-tox-func-encryption/a2afbff/ : SUCCESS in 9m 37s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/19/check/swift-tox-func-domain-remap-staticweb/14cce21/ : FAILURE in 12m 15s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/19/check/swift-tox-func-ec/f1f6ddf/ : SUCCESS in 14m 02s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/19/check/swift-tox-func-s3api/2a32661/ : SUCCESS in 11m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/19/check/swift-tox-func-s3api-ceph-s3tests-tempauth/068f4c9/ : SUCCESS in 11m 16s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/19/check/swift-probetests-centos-7/532ebfc/ : SUCCESS in 32m 11s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/19/check/swift-dsvm-functional/f2d8a32/ : SUCCESS in 34m 55s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/19/check/swift-dsvm-functional-ipv6/c3cebe1/ : SUCCESS in 29m 02s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/19/check/swift-tox-lower-constraints/6558fc0/ : SUCCESS in 13m 59s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/19/check/openstack-tox-pep8/fa4d415/ : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/19/check/swift-multinode-rolling-upgrade/9658d40/ : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":19},{"id":"48c4a3bd5a8b08f1b5a497a8fb213f15d916c60c","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2019-05-02 23:18:25.000000000","message":"Patch Set 19:\n\n(1 comment)\n\nPlease add the anonymized docs to the logs.rst file in addition to the manpage. Otherwise, looks great!","accounts_in_message":[],"_revision_number":19},{"id":"80b70e2813c78d3d4ff722f80ea210d5d043e566","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2019-05-02 23:43:59.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"07baf3d062bd700750f64d381d9c496f8bc0e7ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 00:32:24.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/20/check/openstack-tox-docs/6ccad25/html/ : SUCCESS in 4m 11s\n- swift-tox-py27 http://logs.openstack.org/48/548948/20/check/swift-tox-py27/6d68fd8/ : SUCCESS in 12m 54s\n- swift-tox-py35 http://logs.openstack.org/48/548948/20/check/swift-tox-py35/da95137/ : SUCCESS in 10m 27s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/48/548948/20/check/swift-tox-py36/c62614f/ : SUCCESS in 9m 27s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/48/548948/20/check/swift-tox-py37/8cf3bd2/ : SUCCESS in 9m 54s\n- swift-tox-func http://logs.openstack.org/48/548948/20/check/swift-tox-func/4455df6/ : SUCCESS in 9m 03s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/20/check/swift-tox-func-encryption/f01bbfe/ : SUCCESS in 14m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/20/check/swift-tox-func-domain-remap-staticweb/0955464/ : SUCCESS in 26m 43s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/20/check/swift-tox-func-ec/8f94fe2/ : SUCCESS in 9m 39s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/20/check/swift-tox-func-s3api/34493d5/ : SUCCESS in 11m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/20/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1f286f1/ : SUCCESS in 13m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/20/check/swift-probetests-centos-7/fba8098/ : SUCCESS in 29m 41s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/20/check/swift-dsvm-functional/a1842b1/ : SUCCESS in 22m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/20/check/swift-dsvm-functional-ipv6/9af7a27/ : TIMED_OUT in 47m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/20/check/swift-tox-lower-constraints/ec7a966/ : SUCCESS in 11m 21s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/20/check/openstack-tox-pep8/525ab20/ : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/20/check/swift-multinode-rolling-upgrade/e34b21f/ : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":20},{"id":"27a14a154c7de432301548023734eff5a956f6df","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 13:10:04.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"d6a981bf34369132117dd4554d8191fd39dafb22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 13:58:51.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/20/check/openstack-tox-docs/0f0f47f/html/ : SUCCESS in 4m 36s\n- swift-tox-py27 http://logs.openstack.org/48/548948/20/check/swift-tox-py27/b2a57cc/ : SUCCESS in 16m 15s\n- swift-tox-py35 http://logs.openstack.org/48/548948/20/check/swift-tox-py35/b5110c2/ : SUCCESS in 12m 17s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/48/548948/20/check/swift-tox-py36/cb45771/ : SUCCESS in 13m 07s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/48/548948/20/check/swift-tox-py37/43ea86f/ : SUCCESS in 13m 03s\n- swift-tox-func http://logs.openstack.org/48/548948/20/check/swift-tox-func/1609a2d/ : SUCCESS in 14m 21s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/20/check/swift-tox-func-encryption/da7b4b4/ : SUCCESS in 18m 28s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/20/check/swift-tox-func-domain-remap-staticweb/809d406/ : SUCCESS in 10m 15s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/20/check/swift-tox-func-ec/2a803bb/ : SUCCESS in 13m 37s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/20/check/swift-tox-func-s3api/803e87d/ : SUCCESS in 17m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/20/check/swift-tox-func-s3api-ceph-s3tests-tempauth/22a4513/ : SUCCESS in 19m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/20/check/swift-probetests-centos-7/2f27681/ : SUCCESS in 35m 31s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/20/check/swift-dsvm-functional/3c2f555/ : SUCCESS in 35m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/20/check/swift-dsvm-functional-ipv6/bf8df91/ : SUCCESS in 32m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/20/check/swift-tox-lower-constraints/12c216f/ : SUCCESS in 11m 11s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/20/check/openstack-tox-pep8/fa5e622/ : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/20/check/swift-multinode-rolling-upgrade/67e14e3/ : SUCCESS in 16m 10s","accounts_in_message":[],"_revision_number":20},{"id":"2c197e8f1a87ab0f46d0f6acbbf0339a3b254456","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2019-05-03 15:59:50.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\nthis looks great!","accounts_in_message":[],"_revision_number":20},{"id":"c0db85c413dd7b7dacb8f6982e714cb4b8fe359c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:00:05.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"f8ba4507cf8a9b3910a22ad04e56134532ed9d4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:34:58.000000000","message":"Patch Set 20: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/20/gate/openstack-tox-docs/5a063e6/html/ : SUCCESS in 4m 26s\n- swift-tox-py27 http://logs.openstack.org/48/548948/20/gate/swift-tox-py27/c2e9652/ : SUCCESS in 10m 12s\n- swift-tox-py37 http://logs.openstack.org/48/548948/20/gate/swift-tox-py37/6b50b83/ : SUCCESS in 8m 57s\n- swift-tox-func http://logs.openstack.org/48/548948/20/gate/swift-tox-func/cad568a/ : SUCCESS in 9m 50s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/20/gate/swift-tox-func-encryption/ca468f1/ : SUCCESS in 8m 21s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/20/gate/swift-tox-func-domain-remap-staticweb/82c0aab/ : SUCCESS in 9m 49s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/20/gate/swift-tox-func-ec/33119c1/ : SUCCESS in 9m 41s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/20/gate/swift-tox-func-s3api/b0ef05f/ : FAILURE in 14m 18s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/20/gate/swift-probetests-centos-7/a1a895a/ : SUCCESS in 29m 26s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/20/gate/swift-dsvm-functional/dee88e0/ : SUCCESS in 27m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/20/gate/swift-dsvm-functional-ipv6/5378ac8/ : SUCCESS in 28m 05s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/20/gate/swift-tox-lower-constraints/3496238/ : FAILURE in 14m 39s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/20/gate/openstack-tox-pep8/a5e51dd/ : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":20},{"id":"188629f93b4cde1f6b0f6ba775a75216a2ab1e87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 17:07:44.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"4194eab2627b5fcc5c9c754e6164a5cb2b8d5116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:33:03.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/20/check/openstack-tox-docs/8d68276/html/ : SUCCESS in 5m 05s\n- swift-tox-py27 http://logs.openstack.org/48/548948/20/check/swift-tox-py27/0235699/ : SUCCESS in 12m 25s\n- swift-tox-py35 http://logs.openstack.org/48/548948/20/check/swift-tox-py35/3fa9031/ : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/48/548948/20/check/swift-tox-py36/71e6a45/ : SUCCESS in 10m 57s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/48/548948/20/check/swift-tox-py37/1a10d83/ : SUCCESS in 9m 36s\n- swift-tox-func http://logs.openstack.org/48/548948/20/check/swift-tox-func/159a00e/ : SUCCESS in 8m 08s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/20/check/swift-tox-func-encryption/3618a30/ : SUCCESS in 7m 52s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/20/check/swift-tox-func-domain-remap-staticweb/5149210/ : SUCCESS in 8m 59s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/20/check/swift-tox-func-ec/77aadac/ : SUCCESS in 8m 48s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/20/check/swift-tox-func-s3api/80d5403/ : SUCCESS in 10m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/48/548948/20/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c59f079/ : SUCCESS in 14m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/20/check/swift-probetests-centos-7/4fcf1bc/ : SUCCESS in 32m 18s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/20/check/swift-dsvm-functional/478f4ef/ : SUCCESS in 30m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/20/check/swift-dsvm-functional-ipv6/2cccea3/ : SUCCESS in 26m 20s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/20/check/swift-tox-lower-constraints/b196288/ : SUCCESS in 13m 03s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/20/check/openstack-tox-pep8/acd2c73/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/48/548948/20/check/swift-multinode-rolling-upgrade/7c3fd08/ : SUCCESS in 16m 28s","accounts_in_message":[],"_revision_number":20},{"id":"47bc97ddd70321b6407b1fd2c6f2e300ec56ff73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:33:50.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"86c3261f9d1ad23249e5096b7d191dae8f44c363","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:08:42.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/548948/20/gate/openstack-tox-docs/f9db4c5/html/ : SUCCESS in 4m 59s\n- swift-tox-py27 http://logs.openstack.org/48/548948/20/gate/swift-tox-py27/7ba56bb/ : SUCCESS in 10m 32s\n- swift-tox-py37 http://logs.openstack.org/48/548948/20/gate/swift-tox-py37/5479bd6/ : SUCCESS in 11m 23s\n- swift-tox-func http://logs.openstack.org/48/548948/20/gate/swift-tox-func/b1972ad/ : SUCCESS in 8m 37s\n- swift-tox-func-encryption http://logs.openstack.org/48/548948/20/gate/swift-tox-func-encryption/3ef51d0/ : SUCCESS in 8m 41s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/48/548948/20/gate/swift-tox-func-domain-remap-staticweb/b63e9b9/ : SUCCESS in 13m 04s\n- swift-tox-func-ec http://logs.openstack.org/48/548948/20/gate/swift-tox-func-ec/5ee088a/ : SUCCESS in 11m 00s\n- swift-tox-func-s3api http://logs.openstack.org/48/548948/20/gate/swift-tox-func-s3api/a3b953c/ : SUCCESS in 12m 58s\n- swift-probetests-centos-7 http://logs.openstack.org/48/548948/20/gate/swift-probetests-centos-7/1d39b6d/ : SUCCESS in 30m 55s\n- swift-dsvm-functional http://logs.openstack.org/48/548948/20/gate/swift-dsvm-functional/87abb2b/ : SUCCESS in 27m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/48/548948/20/gate/swift-dsvm-functional-ipv6/9470317/ : SUCCESS in 33m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/48/548948/20/gate/swift-tox-lower-constraints/177027b/ : SUCCESS in 16m 11s\n- openstack-tox-pep8 http://logs.openstack.org/48/548948/20/gate/openstack-tox-pep8/feb56e5/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":20},{"id":"ad588e43a62e9b9f095c1c54c85dba851d245a01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:08:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"a4cc353375264e4bcd8f9aef2f0701ae800e0e73","revisions":{"cc91e7285071069983a3c9d0ba0008981c74b0ad":{"kind":"REWORK","_number":1,"created":"2018-03-01 13:23:33.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/1"}}},"commit":{"parents":[{"commit":"78439d95f46a8dff695896e060cf1834871a1e6f","subject":"Merge \"py3: port common/memcached.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78439d95f46a8dff695896e060cf1834871a1e6f"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 13:20:58.000000000","tz":60},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the access_log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substitutes with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to include anonymized information\nlike a hash of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc91e7285071069983a3c9d0ba0008981c74b0ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc91e7285071069983a3c9d0ba0008981c74b0ad"}]},"branch":"refs/heads/master"},"7b036ae3594d25009004de4a8de8c27c80e45bfb":{"kind":"REWORK","_number":2,"created":"2018-03-05 10:11:29.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/2"}}},"commit":{"parents":[{"commit":"78439d95f46a8dff695896e060cf1834871a1e6f","subject":"Merge \"py3: port common/memcached.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78439d95f46a8dff695896e060cf1834871a1e6f"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-05 09:41:37.000000000","tz":60},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the proxylogging_log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substitutes with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to include anonymized information\nlike a hash of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b036ae3594d25009004de4a8de8c27c80e45bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b036ae3594d25009004de4a8de8c27c80e45bfb"}]},"branch":"refs/heads/master"},"ae3f8828eb736fef1e2e41259ef52ecd1e38c2b8":{"kind":"REWORK","_number":3,"created":"2018-03-08 11:13:19.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/3"}}},"commit":{"parents":[{"commit":"78439d95f46a8dff695896e060cf1834871a1e6f","subject":"Merge \"py3: port common/memcached.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78439d95f46a8dff695896e060cf1834871a1e6f"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-08 11:01:39.000000000","tz":60},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the proxylogging_log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substitutes with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to include anonymized information\nlike a hash of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae3f8828eb736fef1e2e41259ef52ecd1e38c2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae3f8828eb736fef1e2e41259ef52ecd1e38c2b8"}]},"branch":"refs/heads/master"},"f2a2db9660a35b7e0b177e2393605ca5c74a5b2d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-05-30 16:21:13.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/4"}}},"commit":{"parents":[{"commit":"c03701d7b7450a0e48531e7cb4c95e0c0aeaf43e","subject":"Merge \"Remove unhelpful log message in copy middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c03701d7b7450a0e48531e7cb4c95e0c0aeaf43e"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-05-30 16:20:10.000000000","tz":120},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the proxylogging_log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substitutes with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to include anonymized information\nlike a hash of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a2db9660a35b7e0b177e2393605ca5c74a5b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a2db9660a35b7e0b177e2393605ca5c74a5b2d"}]},"branch":"refs/heads/master"},"bbb1c42e968559d4a40d21ac3935c6389c00d4ad":{"kind":"REWORK","_number":5,"created":"2018-06-12 13:02:25.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/5"}}},"commit":{"parents":[{"commit":"8faa26d203355d222e5ccc2413a83809c7203778","subject":"Merge \"Rearrange some error handling to reduce indent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8faa26d203355d222e5ccc2413a83809c7203778"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-06-12 12:59:56.000000000","tz":120},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to include anonymized information\nlike a hash of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbb1c42e968559d4a40d21ac3935c6389c00d4ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbb1c42e968559d4a40d21ac3935c6389c00d4ad"}]},"branch":"refs/heads/master"},"3861d950018059b3ab3e8e18288f4a76c71d9730":{"kind":"REWORK","_number":6,"created":"2018-06-18 13:26:02.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/6"}}},"commit":{"parents":[{"commit":"4aca486fc4821b4b571623f79a2b44e9148367ad","subject":"Merge \"func tests: Rename storage_url to storage_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aca486fc4821b4b571623f79a2b44e9148367ad"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-06-18 13:22:07.000000000","tz":120},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to anonymized (hash) some parts of the\nlog like the path of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3861d950018059b3ab3e8e18288f4a76c71d9730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3861d950018059b3ab3e8e18288f4a76c71d9730"}]},"branch":"refs/heads/master"},"52666522bf90ef88b10fc3b4c77dbff58ad92eb0":{"kind":"REWORK","_number":7,"created":"2018-06-18 15:25:01.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/7"}}},"commit":{"parents":[{"commit":"4aca486fc4821b4b571623f79a2b44e9148367ad","subject":"Merge \"func tests: Rename storage_url to storage_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aca486fc4821b4b571623f79a2b44e9148367ad"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-06-18 15:24:23.000000000","tz":120},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to anonymized (hash) some parts of the\nlog like the path of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52666522bf90ef88b10fc3b4c77dbff58ad92eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52666522bf90ef88b10fc3b4c77dbff58ad92eb0"}]},"branch":"refs/heads/master"},"e55cc0b6b77bad02c76e6307adebb7d129ae8a7d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2018-08-20 16:21:14.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/8"}}},"commit":{"parents":[{"commit":"ed18495faa693edf4768a47eb579cdd3ab22d994","subject":"Merge \"Add debugging info to SignatureDoesNotMatch responses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed18495faa693edf4768a47eb579cdd3ab22d994"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-08-20 16:17:17.000000000","tz":120},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to anonymized (hash) some parts of the\nlog like the path of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e55cc0b6b77bad02c76e6307adebb7d129ae8a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e55cc0b6b77bad02c76e6307adebb7d129ae8a7d"}]},"branch":"refs/heads/master"},"04037e0d190cd56addef6a69632f3c136e49c0b7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-09-07 07:15:22.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/48/548948/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/9"}}},"commit":{"parents":[{"commit":"1f29508327c0f5c1b06b1b06b06400055681583f","subject":"Allow for backwards compatibility with swift3.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f29508327c0f5c1b06b1b06b06400055681583f"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-09-07 07:15:22.000000000","tz":0},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to anonymized (hash) some parts of the\nlog like the path of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04037e0d190cd56addef6a69632f3c136e49c0b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04037e0d190cd56addef6a69632f3c136e49c0b7"}]},"branch":"refs/heads/master"},"3350bb8fadf68a83182797be20a1eff44868d821":{"kind":"REWORK","_number":10,"created":"2018-09-07 10:01:19.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/10"}}},"commit":{"parents":[{"commit":"1f29508327c0f5c1b06b1b06b06400055681583f","subject":"Allow for backwards compatibility with swift3.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f29508327c0f5c1b06b1b06b06400055681583f"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-09-07 09:58:41.000000000","tz":120},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to anonymized (hash) some parts of the\nlog like the path of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3350bb8fadf68a83182797be20a1eff44868d821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3350bb8fadf68a83182797be20a1eff44868d821"}]},"branch":"refs/heads/master"},"cde9a6cbc642f8b0e178842317201dc6ba7f830f":{"kind":"REWORK","_number":11,"created":"2018-09-07 12:24:26.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/11"}}},"commit":{"parents":[{"commit":"1f29508327c0f5c1b06b1b06b06400055681583f","subject":"Allow for backwards compatibility with swift3.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f29508327c0f5c1b06b1b06b06400055681583f"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-09-07 12:24:06.000000000","tz":120},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to anonymized (hash) some parts of the\nlog like the path of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cde9a6cbc642f8b0e178842317201dc6ba7f830f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cde9a6cbc642f8b0e178842317201dc6ba7f830f"}]},"branch":"refs/heads/master"},"5eb2809cc8cc8267c31f6899b91a378e4d9762e9":{"kind":"REWORK","_number":12,"created":"2018-09-07 17:07:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/548948/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/12"}}},"commit":{"parents":[{"commit":"1f29508327c0f5c1b06b1b06b06400055681583f","subject":"Allow for backwards compatibility with swift3.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f29508327c0f5c1b06b1b06b06400055681583f"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2018-09-07 17:07:06.000000000","tz":0},"subject":"Add template in proxy to create custom and anonymous access-logs","message":"Add template in proxy to create custom and anonymous access-logs\n\nAdd the log_msg_template option in proxy-server.conf.\nIt should be a string with some patterns (like that: {pattern}) which will\nbe substituted with the corresponding log information. This allows customized\nproxy-access logs. It is also possible to anonymized (hash) some parts of the\nlog like the path of the request.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5eb2809cc8cc8267c31f6899b91a378e4d9762e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5eb2809cc8cc8267c31f6899b91a378e4d9762e9"}]},"branch":"refs/heads/master"},"cdafc6aadaf26435fd5e32011aefff7ca6df4bb7":{"kind":"REWORK","_number":13,"created":"2018-10-26 09:03:50.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/48/548948/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/13"}}},"commit":{"parents":[{"commit":"6bb8229b75de74c11f46de54bbca2d10bfe3bada","subject":"Merge \"Require gettext for all non-SUSE distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bb8229b75de74c11f46de54bbca2d10bfe3bada"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-10-26 09:01:49.000000000","tz":120},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdafc6aadaf26435fd5e32011aefff7ca6df4bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdafc6aadaf26435fd5e32011aefff7ca6df4bb7"}]},"branch":"refs/heads/master"},"a3f46f4b1d809780271e31ff184716ec81a6970d":{"kind":"REWORK","_number":14,"created":"2018-10-29 09:51:14.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/48/548948/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/14"}}},"commit":{"parents":[{"commit":"24bf5eea8cd5d8d4ce3d3cb166a48bf8781ccad9","subject":"Merge \"Fix up the test for .ismount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24bf5eea8cd5d8d4ce3d3cb166a48bf8781ccad9"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-10-29 09:49:35.000000000","tz":60},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f46f4b1d809780271e31ff184716ec81a6970d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f46f4b1d809780271e31ff184716ec81a6970d"}]},"branch":"refs/heads/master"},"7753c1065872ca350dae63a73784fa207c9402bc":{"kind":"REWORK","_number":15,"created":"2018-10-29 16:22:01.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/48/548948/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/15"}}},"commit":{"parents":[{"commit":"24bf5eea8cd5d8d4ce3d3cb166a48bf8781ccad9","subject":"Merge \"Fix up the test for .ismount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24bf5eea8cd5d8d4ce3d3cb166a48bf8781ccad9"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-10-29 16:21:30.000000000","tz":60},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7753c1065872ca350dae63a73784fa207c9402bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7753c1065872ca350dae63a73784fa207c9402bc"}]},"branch":"refs/heads/master"},"6bfa721f0170426b2dd98bf69eeb04ca05cce531":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-04-19 14:58:17.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/16"}}},"commit":{"parents":[{"commit":"3be4d2f64b7897cf643e90f8697710e9bab472cd","subject":"Merge \"Clean up dlo unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3be4d2f64b7897cf643e90f8697710e9bab472cd"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2019-04-19 14:58:17.000000000","tz":0},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bfa721f0170426b2dd98bf69eeb04ca05cce531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bfa721f0170426b2dd98bf69eeb04ca05cce531"}]},"branch":"refs/heads/master"},"2d2c6dfcac2cee3756f7dc58d352f23e1ebd8863":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-04-24 13:31:37.000000000","uploader":{"_account_id":27887,"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","username":"gillesbiannic"},"ref":"refs/changes/48/548948/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/17"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2019-04-24 13:31:37.000000000","tz":0},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d2c6dfcac2cee3756f7dc58d352f23e1ebd8863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d2c6dfcac2cee3756f7dc58d352f23e1ebd8863"}]},"branch":"refs/heads/master"},"ccea9605af727a15d2a5a987f5a6a018f24c48d0":{"kind":"REWORK","_number":18,"created":"2019-05-02 22:16:28.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/48/548948/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/18"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-05-02 22:15:30.000000000","tz":-360},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccea9605af727a15d2a5a987f5a6a018f24c48d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccea9605af727a15d2a5a987f5a6a018f24c48d0"}]},"branch":"refs/heads/master"},"a68439c1395fd42bc5dec8a1cfee7c9a3cfd2ce9":{"kind":"REWORK","_number":19,"created":"2019-05-02 22:20:06.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/48/548948/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/19"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-05-02 22:19:25.000000000","tz":-360},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a68439c1395fd42bc5dec8a1cfee7c9a3cfd2ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a68439c1395fd42bc5dec8a1cfee7c9a3cfd2ce9"}]},"branch":"refs/heads/master"},"a4cc353375264e4bcd8f9aef2f0701ae800e0e73":{"kind":"REWORK","_number":20,"created":"2019-05-02 23:43:59.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/48/548948/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/548948/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/548948/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/548948/20"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Gilles Biannic","email":"gilles.biannic@corp.ovh.com","date":"2018-03-01 10:31:12.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-05-02 23:43:25.000000000","tz":-360},"subject":"Make log format for requests configurable","message":"Make log format for requests configurable\n\nAdd the log_msg_template option in proxy-server.conf and log_format in\na/c/o-server.conf. It is a string parsable by Python\u0027s format()\nfunction. Some fields containing user data might be anonymized by using\nlog_anonymization_method and log_anonymization_salt.\n\nChange-Id: I29e30ef45fe3f8a026e7897127ffae08a6a80cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cc353375264e4bcd8f9aef2f0701ae800e0e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cc353375264e4bcd8f9aef2f0701ae800e0e73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
