)]}'
{"id":"openstack%2Fswift~569884","triplet_id":"openstack%2Fswift~master~I6a24e704346ef04d1521b837c48f86cd5d98963d","project":"openstack/swift","branch":"master","topic":"569882","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-09 15:40:34.000000000","reason":"Clay Gerrard replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6a24e704346ef04d1521b837c48f86cd5d98963d","subject":"Clean up direct_client\u0027s gen_headers","status":"NEW","created":"2018-05-21 23:40:32.000000000","updated":"2021-12-09 15:40:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"315f83c0e3b0f234bf23b22a305f0ed7fdb56c59","_number":569884,"virtual_id_number":569884,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-06 06:32:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2021-12-09 15:40:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-22 18:46:50.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2018-05-23 17:02:07.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2018-08-24 02:03:52.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2018-10-31 01:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-09 15:40:34.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"ce30837e42d39fbadbe3bd124463fc5a230e0ebf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-05-21 23:40:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49ec5e0d078617b07a8c62bcb01a81fa0a38bd7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-22 01:24:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/84/569884/1/check/build-openstack-sphinx-docs/1a2c2b1/html/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/84/569884/1/check/openstack-tox-pep8/b4f5dab/ : SUCCESS in 3m 58s\n- legacy-swift-dsvm-functional http://logs.openstack.org/84/569884/1/check/legacy-swift-dsvm-functional/9e50d3b/ : SUCCESS in 22m 18s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/84/569884/1/check/legacy-swift-dsvm-functional-identity-v3-only/a5e3276/ : SUCCESS in 27m 01s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/84/569884/1/check/legacy-tempest-dsvm-neutron-full/596453a/ : SUCCESS in 1h 42m 19s\n- neutron-grenade http://logs.openstack.org/84/569884/1/check/neutron-grenade/f2e6e92/ : SUCCESS in 42m 22s\n- swift-tox-py27 http://logs.openstack.org/84/569884/1/check/swift-tox-py27/b46bdb9/ : FAILURE in 9m 12s\n- swift-tox-py35 http://logs.openstack.org/84/569884/1/check/swift-tox-py35/f14a1d4/ : SUCCESS in 3m 56s\n- swift-tox-func http://logs.openstack.org/84/569884/1/check/swift-tox-func/c3f7794/ : SUCCESS in 6m 38s\n- swift-tox-func-encryption http://logs.openstack.org/84/569884/1/check/swift-tox-func-encryption/f24eb97/ : SUCCESS in 7m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/84/569884/1/check/swift-tox-func-domain-remap-staticweb/b9ee0f1/ : SUCCESS in 7m 28s\n- swift-tox-func-ec http://logs.openstack.org/84/569884/1/check/swift-tox-func-ec/bd18eb9/ : SUCCESS in 7m 58s\n- swift-tox-func-s3api http://logs.openstack.org/84/569884/1/check/swift-tox-func-s3api/b083fab/ : SUCCESS in 9m 21s\n- swift-probetests-centos-7 http://logs.openstack.org/84/569884/1/check/swift-probetests-centos-7/320b529/ : TIMED_OUT in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"225d8b660d3bf44ad278fda86d5cad874304072a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-05-22 17:18:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ed41f0a6f0f31c76c6b81bb0d9698b69d983828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-22 18:45:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/84/569884/2/check/build-openstack-sphinx-docs/c5540d6/html/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/84/569884/2/check/openstack-tox-pep8/c833425/ : SUCCESS in 4m 52s\n- legacy-swift-dsvm-functional http://logs.openstack.org/84/569884/2/check/legacy-swift-dsvm-functional/4f4614f/ : SUCCESS in 22m 40s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/84/569884/2/check/legacy-swift-dsvm-functional-identity-v3-only/8b4ac07/ : SUCCESS in 23m 06s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/84/569884/2/check/legacy-tempest-dsvm-neutron-full/bfc8366/ : SUCCESS in 1h 24m 27s\n- neutron-grenade http://logs.openstack.org/84/569884/2/check/neutron-grenade/da9db9a/ : SUCCESS in 56m 18s\n- swift-tox-py27 http://logs.openstack.org/84/569884/2/check/swift-tox-py27/faca10a/ : SUCCESS in 9m 40s\n- swift-tox-py35 http://logs.openstack.org/84/569884/2/check/swift-tox-py35/8a74a4b/ : SUCCESS in 4m 28s\n- swift-tox-func http://logs.openstack.org/84/569884/2/check/swift-tox-func/7f20a18/ : SUCCESS in 8m 52s\n- swift-tox-func-encryption http://logs.openstack.org/84/569884/2/check/swift-tox-func-encryption/f35221f/ : SUCCESS in 8m 29s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/84/569884/2/check/swift-tox-func-domain-remap-staticweb/8160832/ : SUCCESS in 7m 32s\n- swift-tox-func-ec http://logs.openstack.org/84/569884/2/check/swift-tox-func-ec/cc81d7c/ : SUCCESS in 7m 51s\n- swift-tox-func-s3api http://logs.openstack.org/84/569884/2/check/swift-tox-func-s3api/92a1f3f/ : SUCCESS in 10m 07s\n- swift-probetests-centos-7 http://logs.openstack.org/84/569884/2/check/swift-probetests-centos-7/174e690/ : SUCCESS in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c184b87be39f617c5217b58965ecb744bfe0a2f","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-05-22 18:46:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"55e954fcf0d0592b2c6db2893891ea24fb790a7a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2018-05-23 17:02:07.000000000","message":"Patch Set 2:\n\nI tried similar on feature/deep to get the user agent plumbed in, saw the unit tests fail and realised that however reasonable the change seems to us I\u0027d never know if anyone else relies on that current behaviour","accounts_in_message":[],"_revision_number":2},{"id":"6dd610523d8ed6dba51fbc93bf6a0ef72e2c855a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-05-23 21:39:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03419c41018a550822a54145082cc3ebb9496158","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-08-24 02:03:52.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nNice clean up, left a comment inline, not blocking, just a question. Will give it a bit of time then will come back to +A","accounts_in_message":[],"_revision_number":2},{"id":"49c6f1dfb22d9f52152a6a9fcefe85dc718e6cfc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-30 22:15:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c92ad0cb69d33a1cef0e11379ace9ae08d0edfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-31 01:39:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/569884/3/check/openstack-tox-docs/d284bf9/html/ : SUCCESS in 4m 18s\n- swift-tox-py27 http://logs.openstack.org/84/569884/3/check/swift-tox-py27/4d2ac4e/ : SUCCESS in 12m 09s\n- swift-tox-py35 http://logs.openstack.org/84/569884/3/check/swift-tox-py35/dc87b9f/ : SUCCESS in 5m 46s\n- swift-tox-py36 http://logs.openstack.org/84/569884/3/check/swift-tox-py36/aabef0f/ : SUCCESS in 4m 49s (non-voting)\n- swift-tox-func http://logs.openstack.org/84/569884/3/check/swift-tox-func/596908a/ : SUCCESS in 9m 03s\n- swift-tox-func-encryption http://logs.openstack.org/84/569884/3/check/swift-tox-func-encryption/636d9fc/ : SUCCESS in 9m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/84/569884/3/check/swift-tox-func-domain-remap-staticweb/5862aef/ : SUCCESS in 10m 06s\n- swift-tox-func-ec http://logs.openstack.org/84/569884/3/check/swift-tox-func-ec/001aa14/ : SUCCESS in 13m 07s\n- swift-tox-func-s3api http://logs.openstack.org/84/569884/3/check/swift-tox-func-s3api/c6de191/ : SUCCESS in 11m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/84/569884/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/089277d/ : SUCCESS in 10m 47s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/84/569884/3/check/swift-probetests-centos-7/43041fe/ : SUCCESS in 27m 47s\n- swift-dsvm-functional http://logs.openstack.org/84/569884/3/check/swift-dsvm-functional/0839d41/ : SUCCESS in 27m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/84/569884/3/check/swift-dsvm-functional-ipv6/c69d718/ : SUCCESS in 22m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/84/569884/3/check/swift-tox-lower-constraints/0e8e298/ : SUCCESS in 10m 44s\n- openstack-tox-pep8 http://logs.openstack.org/84/569884/3/check/openstack-tox-pep8/7dddb5d/ : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/84/569884/3/check/swift-multinode-rolling-upgrade/366876e/ : SUCCESS in 15m 38s","accounts_in_message":[],"_revision_number":3},{"id":"c24edf452f814deebb95f34cdb5171a85a6d58eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-06 05:33:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"66273fdfb1874f73ca787cc57a03476ea9a4496a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-06 06:32:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f3aa293a1f4e94b63754e6770918ed : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/acaa43e5f441419db949a28e5fc91df7 : SUCCESS in 55m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fad2d2afde94af8a26c1cc668d077fd : SUCCESS in 32m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e3e89759dd084dfaba8d4f579a24a0e8 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b9bf09193c041b891b14068eddd9189 : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dac7093b156f40d1b3890b0869addf2b : SUCCESS in 12m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d833d80c273245bbaeaaffac4f9be0ac : SUCCESS in 21m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8189bb675ad45b9816d4214aaf4a1b5 : SUCCESS in 13m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/954b7a58c23445a89dd88828ba2bb03f : SUCCESS in 13m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/243153af58454b83a015c56109d0323e : SUCCESS in 12m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b62dc91963104588817fc6ea059da586 : SUCCESS in 23m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0175ad89ec594adc96d3c0383e38dfe9 : SUCCESS in 24m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c86c17097b774f87a07e76c87c998a20 : SUCCESS in 15m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ec28ca2f2be64a13abbed3c974079cc7 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5d86097236ef48bcbae21b5eac0d86f5 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d2967850a7745ee97009874197fcc36 : SUCCESS in 14m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d82ed7fe4e7745f99e4e2fa44b622f9d : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c5cfe1bd36947ed8357841f83d7aa54 : SUCCESS in 18m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fff641a4a7cc405fb046d144ce149c5b : SUCCESS in 44m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7d3a91ca83d64340a6dd6f0044bd4a32 : SUCCESS in 47m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a91eb05c73c483ab6f4b0879a538e7e : SUCCESS in 38m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1e977e349fa4fa792d5941af7743746 : SUCCESS in 40m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ae4c8d7684246ebbd076c9bde3f5a5c : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a40e5992515c4d4383ba326a9fd026ba : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18e3af8b96c04180bd87d3dada4445e9 : SUCCESS in 24m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d34a836e4e14f44be1df580175abc67 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":4},{"id":"906b5ce820a026006b34f0ce8fdc4d623e223c2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-06 07:27:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83806ed6efc040698c08b182c18dcbb6 : SUCCESS in 54m 46s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7334569ad1e84efc91d695ff972df726 : SUCCESS in 48m 31s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/093a0b18222d44389e181e9b6e0d779d : FAILURE in 1h 51m 47s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/88e3502e0c0544979e7a5465180e5c34 : SUCCESS in 1h 02m 17s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca34ba55f28a41c4a4f00d82086e69f2 : SUCCESS in 48m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"315f83c0e3b0f234bf23b22a305f0ed7fdb56c59","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-09 15:40:34.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c64515d5fc14616b81f41568d8323a622a930c09","revisions":{"63e81805aa5b1800712e3a648b58525e70d98230":{"kind":"REWORK","_number":1,"created":"2018-05-21 23:40:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/569884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/569884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/569884/1"}}},"commit":{"parents":[{"commit":"37dafa03feea67d15547ad52bcec5994b8243dd4","subject":"Fix probe tests in the gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37dafa03feea67d15547ad52bcec5994b8243dd4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-05-21 19:46:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-05-21 23:39:58.000000000","tz":-420},"subject":"Clean up direct_client\u0027s gen_headers","message":"Clean up direct_client\u0027s gen_headers\n\nWe *always* want X-Timestamp and User-Agent headers; it\u0027s just that\n*some* tests supply their own.\n\nAlso, drop a bunch of\n\n   if headers is None:\n       headers \u003d {}\n\nboilerplate -- not only was gen_headers() content to do that itself, but\nHeaderKeyDict would happily eat None as well!\n\nChange-Id: I6a24e704346ef04d1521b837c48f86cd5d98963d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e81805aa5b1800712e3a648b58525e70d98230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e81805aa5b1800712e3a648b58525e70d98230"}]},"branch":"refs/heads/master"},"713010350d0fc8a8c547bd11d919b05346384a46":{"kind":"REWORK","_number":2,"created":"2018-05-22 17:18:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/569884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/569884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/569884/2"}}},"commit":{"parents":[{"commit":"8e09b6740a3338c3814b705230cad57125c69fed","subject":"Merge \"s3api: simplify BaseAclHandler.request_with\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e09b6740a3338c3814b705230cad57125c69fed"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-05-21 19:46:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-05-22 17:17:55.000000000","tz":-420},"subject":"Clean up direct_client\u0027s gen_headers","message":"Clean up direct_client\u0027s gen_headers\n\nWe *always* want X-Timestamp and User-Agent headers; it\u0027s just that\n*some* things supply their own.\n\nAlso, drop a bunch of\n\n   if headers is None:\n       headers \u003d {}\n\nboilerplate -- not only was gen_headers() content to do that itself, but\nHeaderKeyDict would happily eat None as well!\n\nChange-Id: I6a24e704346ef04d1521b837c48f86cd5d98963d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/713010350d0fc8a8c547bd11d919b05346384a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/713010350d0fc8a8c547bd11d919b05346384a46"}]},"branch":"refs/heads/master"},"8cb40774f1197a38921181b51ed919aa38a4d897":{"kind":"REWORK","_number":3,"created":"2018-10-30 22:15:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/569884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/569884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/569884/3"}}},"commit":{"parents":[{"commit":"2cfe31551db57de2489208c7d0f53aab5549749c","subject":"Merge \"Allow multiple keymasters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cfe31551db57de2489208c7d0f53aab5549749c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-05-21 19:46:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-10-30 22:15:03.000000000","tz":0},"subject":"Clean up direct_client\u0027s gen_headers","message":"Clean up direct_client\u0027s gen_headers\n\nWe *always* want X-Timestamp and User-Agent headers; it\u0027s just that\n*some* things supply their own.\n\nAlso, drop a bunch of\n\n   if headers is None:\n       headers \u003d {}\n\nboilerplate -- not only was gen_headers() content to do that itself, but\nHeaderKeyDict would happily eat None as well!\n\nChange-Id: I6a24e704346ef04d1521b837c48f86cd5d98963d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cb40774f1197a38921181b51ed919aa38a4d897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cb40774f1197a38921181b51ed919aa38a4d897"}]},"branch":"refs/heads/master"},"c64515d5fc14616b81f41568d8323a622a930c09":{"kind":"REWORK","_number":4,"created":"2021-11-06 05:33:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/569884/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/569884/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/569884/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/569884/4"}}},"commit":{"parents":[{"commit":"27478f07c3f45067a2644140b7da2b356ca9ac60","subject":"Merge \"proxy: Remove a bunch of logging translations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27478f07c3f45067a2644140b7da2b356ca9ac60"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-05-21 19:46:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-11-06 05:32:47.000000000","tz":-420},"subject":"Clean up direct_client\u0027s gen_headers","message":"Clean up direct_client\u0027s gen_headers\n\nWe *always* want X-Timestamp and User-Agent headers; it\u0027s just that\n*some* things supply their own.\n\nAlso, drop a bunch of\n\n   if headers is None:\n       headers \u003d {}\n\nboilerplate -- not only was gen_headers() content to do that itself, but\nHeaderKeyDict would happily eat None as well!\n\nChange-Id: I6a24e704346ef04d1521b837c48f86cd5d98963d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c64515d5fc14616b81f41568d8323a622a930c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c64515d5fc14616b81f41568d8323a622a930c09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
