)]}'
{"id":"openstack%2Fswift~580333","triplet_id":"openstack%2Fswift~master~Ic61153948e35f1c09b05bfc97dfac3fb487b0898","project":"openstack/swift","branch":"master","topic":"bug/1780204","hashtags":[],"change_id":"Ic61153948e35f1c09b05bfc97dfac3fb487b0898","subject":"Use get_container_info to check existence before container PUT","status":"MERGED","created":"2018-07-05 10:04:27.000000000","updated":"2019-07-02 02:50:41.000000000","submitted":"2019-07-01 21:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"580333-1562016244349-f800f651","meta_rev_id":"d433915d5ab7736914f24aaf43b9d98ffda4147e","_number":580333,"virtual_id_number":580333,"owner":{"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-07-01 21:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-27 00:03:56.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":2,"date":"2019-06-27 14:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2019-07-01 20:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-12 08:29:01.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2019-06-27 00:03:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-27 14:16:13.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-07-01 20:20:13.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-07-01 21:24:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6f1cbb96accd35f54ae4e9e3e86817a4a7db234","author":{"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},"date":"2018-07-05 10:04:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffb1035bbbfc1c75d427905d72b828247ddeaab9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-05 11:42:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/580333/1/check/build-openstack-sphinx-docs/56e2928/html/ : SUCCESS in 5m 55s\n- openstack-tox-pep8 http://logs.openstack.org/33/580333/1/check/openstack-tox-pep8/cdf079b/ : SUCCESS in 6m 08s\n- legacy-swift-dsvm-functional http://logs.openstack.org/33/580333/1/check/legacy-swift-dsvm-functional/3d0cc52/ : SUCCESS in 26m 05s\n- legacy-swift-dsvm-functional-identity-v3-only http://logs.openstack.org/33/580333/1/check/legacy-swift-dsvm-functional-identity-v3-only/24b690a/ : SUCCESS in 26m 19s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/33/580333/1/check/legacy-tempest-dsvm-neutron-full/43e6952/ : SUCCESS in 1h 35m 27s\n- neutron-grenade http://logs.openstack.org/33/580333/1/check/neutron-grenade/a47bfe1/ : SUCCESS in 50m 52s\n- swift-tox-py27 http://logs.openstack.org/33/580333/1/check/swift-tox-py27/ae4738e/ : SUCCESS in 10m 04s\n- swift-tox-py35 http://logs.openstack.org/33/580333/1/check/swift-tox-py35/188f5de/ : SUCCESS in 5m 09s\n- swift-tox-func http://logs.openstack.org/33/580333/1/check/swift-tox-func/a28725c/ : SUCCESS in 8m 02s\n- swift-tox-func-encryption http://logs.openstack.org/33/580333/1/check/swift-tox-func-encryption/36c7b23/ : SUCCESS in 10m 18s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/580333/1/check/swift-tox-func-domain-remap-staticweb/571745a/ : SUCCESS in 9m 20s\n- swift-tox-func-ec http://logs.openstack.org/33/580333/1/check/swift-tox-func-ec/d05f73e/ : SUCCESS in 10m 25s\n- swift-tox-func-s3api http://logs.openstack.org/33/580333/1/check/swift-tox-func-s3api/77ff80e/ : SUCCESS in 10m 33s\n- swift-probetests-centos-7 http://logs.openstack.org/33/580333/1/check/swift-probetests-centos-7/b272d4d/ : FAILURE in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4cca48c19167a8b02880034f700e59116824591","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-07-12 08:29:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nCould you have the summary how the change to HEAD container mitigate the load at the container server?\n\nWhile checking the container server code, it seems like HEAD container also have a task to update container state from .pending_file when getting the db stat so that I\u0027m not sure how the change is effective to us.\n\nPlus, set my -1 for my inline that the new change will be missing the existing error case. Any ideas?","accounts_in_message":[],"_revision_number":1},{"id":"088fd5e244904739f65c351cb5dc8ab28108e5f7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-07-24 06:14:37.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI think what we\u0027re really after is a get_container_info call -- that way once *a* HEAD succeeds, subsequent calls (within a minute or so, depending on tuning) get served from memcached and we don\u0027t actually have to hit the container layer.\n\nAgreed with Kota that we still need to worry about BucketAlreadyExists errors though, because concurrent actors make everything difficult.","accounts_in_message":[],"_revision_number":1},{"id":"ed3a0cd013d66c241aca3f87ade2fa62b842b0a6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 18:45:18.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nRebase will be coming shortly; I\u0027ll also\n\n- see what it looks like to use get_container_info and\n- try to wrap my head around the s3_acl implications.","accounts_in_message":[],"_revision_number":1},{"id":"a8d39c1426b04178cd64ed1fd8676bab2344c653","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 18:45:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f91c3c61660866bab0337c0eaaa4bb3577617591","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-25 21:44:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)\n\nLGTM, gate tests are taking forever, but unit and func tests pass on my machine.","accounts_in_message":[],"_revision_number":2},{"id":"9cac7d9238bdaecc479984b2b64178ee33e8b8ce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 22:04:47.000000000","message":"Patch Set 2: Workflow-1\n\nYeah, probably should address those two points.","accounts_in_message":[],"_revision_number":2},{"id":"c14d193461ce57c9337d89e4e6e5ade678d7747b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:42:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/580333/2/check/openstack-tox-docs/3c7c00d/html/ : SUCCESS in 4m 19s\n- swift-build-image http://logs.openstack.org/33/580333/2/check/swift-build-image/c00795e/ : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/33/580333/2/check/swift-tox-py27/d71457a/ : SUCCESS in 10m 20s\n- swift-tox-py36 http://logs.openstack.org/33/580333/2/check/swift-tox-py36/fa1da0f/ : SUCCESS in 11m 33s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/33/580333/2/check/swift-tox-py37/66dd0a4/ : SUCCESS in 11m 14s\n- swift-tox-func http://logs.openstack.org/33/580333/2/check/swift-tox-func/3823a1b/ : SUCCESS in 9m 40s\n- swift-tox-func-py37 http://logs.openstack.org/33/580333/2/check/swift-tox-func-py37/199c4d7/ : SUCCESS in 8m 56s\n- swift-tox-func-encryption http://logs.openstack.org/33/580333/2/check/swift-tox-func-encryption/d776b13/ : SUCCESS in 9m 13s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/580333/2/check/swift-tox-func-domain-remap-staticweb/46e77e6/ : SUCCESS in 9m 07s\n- swift-tox-func-ec http://logs.openstack.org/33/580333/2/check/swift-tox-func-ec/baf8959/ : SUCCESS in 11m 48s\n- swift-tox-func-s3api http://logs.openstack.org/33/580333/2/check/swift-tox-func-s3api/799b7e5/ : SUCCESS in 11m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/33/580333/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/16ebf9a/ : SUCCESS in 14m 43s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/33/580333/2/check/swift-probetests-centos-7/b1b68d7/ : SUCCESS in 33m 04s\n- swift-dsvm-functional http://logs.openstack.org/33/580333/2/check/swift-dsvm-functional/0633783/ : SUCCESS in 26m 42s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/580333/2/check/swift-dsvm-functional-ipv6/2640e90/ : SUCCESS in 30m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/33/580333/2/check/swift-tox-lower-constraints/6d298b8/ : SUCCESS in 13m 16s\n- openstack-tox-pep8 http://logs.openstack.org/33/580333/2/check/openstack-tox-pep8/c4eb4a4/ : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/33/580333/2/check/swift-multinode-rolling-upgrade/861b97a/ : FAILURE in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abe13d582da9e25117a20b382132320d965cc61b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-27 00:01:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa053969000d63be2764bf91b1e9a9177f9246c1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-27 00:03:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"883f7a53b6697aee50af4afbe3aff2022c5a0188","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 05:20:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/580333/3/check/openstack-tox-docs/dbe185b/html/ : SUCCESS in 4m 14s\n- swift-build-image http://logs.openstack.org/33/580333/3/check/swift-build-image/ab63ec8/ : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/33/580333/3/check/swift-tox-py27/78b72b4/ : SUCCESS in 10m 28s\n- swift-tox-py36 http://logs.openstack.org/33/580333/3/check/swift-tox-py36/f745e8f/ : SUCCESS in 11m 28s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/33/580333/3/check/swift-tox-py37/5bf4897/ : SUCCESS in 10m 36s\n- swift-tox-func http://logs.openstack.org/33/580333/3/check/swift-tox-func/1912c98/ : SUCCESS in 8m 29s\n- swift-tox-func-py37 http://logs.openstack.org/33/580333/3/check/swift-tox-func-py37/fd3c487/ : SUCCESS in 8m 44s\n- swift-tox-func-encryption http://logs.openstack.org/33/580333/3/check/swift-tox-func-encryption/0c9c8cf/ : SUCCESS in 8m 45s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/580333/3/check/swift-tox-func-domain-remap-staticweb/705f4be/ : SUCCESS in 9m 11s\n- swift-tox-func-ec http://logs.openstack.org/33/580333/3/check/swift-tox-func-ec/e565a19/ : SUCCESS in 9m 23s\n- swift-tox-func-s3api http://logs.openstack.org/33/580333/3/check/swift-tox-func-s3api/bf38f52/ : SUCCESS in 11m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/33/580333/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/68d4441/ : SUCCESS in 16m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/33/580333/3/check/swift-probetests-centos-7/5a982c8/ : SUCCESS in 31m 20s\n- swift-dsvm-functional http://logs.openstack.org/33/580333/3/check/swift-dsvm-functional/0c67581/ : SUCCESS in 26m 12s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/580333/3/check/swift-dsvm-functional-ipv6/08a5fea/ : SUCCESS in 31m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/33/580333/3/check/swift-tox-lower-constraints/19c13bf/ : SUCCESS in 11m 31s\n- openstack-tox-pep8 http://logs.openstack.org/33/580333/3/check/openstack-tox-pep8/1f37d8f/ : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/33/580333/3/check/swift-multinode-rolling-upgrade/eb67202/ : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45ccd3f0deba8ef62d9de9033ec8fff25d04fb91","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-27 14:16:13.000000000","message":"Patch Set 3: Code-Review+2\n\nlgtm. noticed during testing that our func tests also provide some code coverage.","accounts_in_message":[],"_revision_number":3},{"id":"83db19ae5a345b187129de71a742b9a1a285dff6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-07-01 20:20:13.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"97d5857210ad2d63ec5dfa6238824cc9a4d4d3fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 20:20:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"474c51a0a9f4339435cb9c886cc3beac1aa46747","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:24:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"48a4276368929e7ceb152efc03e19a60b11ac5ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:24:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/580333/3/gate/openstack-tox-docs/0afaf92/html/ : SUCCESS in 3m 55s\n- swift-upload-image http://logs.openstack.org/33/580333/3/gate/swift-upload-image/0a2e9fb/ : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/33/580333/3/gate/swift-tox-py27/a26263b/ : SUCCESS in 11m 25s\n- swift-tox-py37 http://logs.openstack.org/33/580333/3/gate/swift-tox-py37/9f856a5/ : SUCCESS in 13m 36s\n- swift-tox-func http://logs.openstack.org/33/580333/3/gate/swift-tox-func/74fca01/ : SUCCESS in 8m 47s\n- swift-tox-func-py37 http://logs.openstack.org/33/580333/3/gate/swift-tox-func-py37/3540428/ : SUCCESS in 9m 02s\n- swift-tox-func-encryption http://logs.openstack.org/33/580333/3/gate/swift-tox-func-encryption/7e5255d/ : SUCCESS in 9m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/580333/3/gate/swift-tox-func-domain-remap-staticweb/ab049f4/ : SUCCESS in 10m 14s\n- swift-tox-func-ec http://logs.openstack.org/33/580333/3/gate/swift-tox-func-ec/0bb0297/ : SUCCESS in 9m 16s\n- swift-tox-func-s3api http://logs.openstack.org/33/580333/3/gate/swift-tox-func-s3api/d7639bb/ : SUCCESS in 12m 37s\n- swift-probetests-centos-7 http://logs.openstack.org/33/580333/3/gate/swift-probetests-centos-7/0703f9b/ : SUCCESS in 38m 04s\n- swift-dsvm-functional http://logs.openstack.org/33/580333/3/gate/swift-dsvm-functional/3093dd5/ : SUCCESS in 27m 54s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/580333/3/gate/swift-dsvm-functional-ipv6/cd2654b/ : SUCCESS in 30m 14s\n- swift-tox-lower-constraints http://logs.openstack.org/33/580333/3/gate/swift-tox-lower-constraints/dcc3ef1/ : SUCCESS in 13m 33s\n- openstack-tox-pep8 http://logs.openstack.org/33/580333/3/gate/openstack-tox-pep8/9314bfd/ : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":3},{"id":"040f4340abd4d2256e62f7d3c396a9f1aa1adcff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:25:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/33/580333/3/promote/swift-promote-image/823b2d7/ : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad73305ed6d1c54961bd3727fc02afa05e03a24f","author":{"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},"date":"2019-07-02 02:50:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e02b9b798007b9f8c70e33c22fa6189d81a41ca3","revisions":{"04fe71ee07922c240fc167874d9398fb819c399c":{"kind":"REWORK","_number":1,"created":"2018-07-05 10:04:27.000000000","uploader":{"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},"ref":"refs/changes/33/580333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/580333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/580333/1"}}},"commit":{"parents":[{"commit":"cb3692f8db4b2e2f94567658a2b9cb1ab8c0834a","subject":"Merge \"swob.Match: remove quotes when checking __contains__\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb3692f8db4b2e2f94567658a2b9cb1ab8c0834a"}]}],"author":{"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","date":"2018-07-05 09:46:50.000000000","tz":480},"committer":{"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","date":"2018-07-05 09:47:38.000000000","tz":480},"subject":"HEAD to check existence before container PUT","message":"HEAD to check existence before container PUT\n\nPUT request on an existing container will trigger an update on\ncontainer db. When disks where container db landed are under\nheavy loads, update on the container db may fail due to LockTimout.\n\nHence, we first check existence, if it\u0027s not there, we PUT.\n\nChange-Id: Ic61153948e35f1c09b05bfc97dfac3fb487b0898\nCloses-Bug: 1780204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04fe71ee07922c240fc167874d9398fb819c399c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04fe71ee07922c240fc167874d9398fb819c399c"}]},"branch":"refs/heads/master"},"a39a493017fc1e04e544fa84425d80054837ec7b":{"kind":"REWORK","_number":2,"created":"2019-06-25 18:45:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/580333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/580333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/580333/2"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","date":"2018-07-05 09:46:50.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-25 18:36:17.000000000","tz":-420},"subject":"HEAD to check existence before container PUT","message":"HEAD to check existence before container PUT\n\nPUT request on an existing container will trigger an update on\ncontainer db. When disks where container db landed are under\nheavy loads, update on the container db may fail due to LockTimout.\n\nHence, we first check existence, if it\u0027s not there, we PUT.\n\nChange-Id: Ic61153948e35f1c09b05bfc97dfac3fb487b0898\nCloses-Bug: 1780204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a39a493017fc1e04e544fa84425d80054837ec7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a39a493017fc1e04e544fa84425d80054837ec7b"}]},"branch":"refs/heads/master"},"e02b9b798007b9f8c70e33c22fa6189d81a41ca3":{"kind":"REWORK","_number":3,"created":"2019-06-27 00:01:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/580333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/580333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/580333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/580333/3"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","date":"2018-07-05 09:46:50.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-27 00:01:00.000000000","tz":-420},"subject":"Use get_container_info to check existence before container PUT","message":"Use get_container_info to check existence before container PUT\n\nPUT request on an existing container will trigger an update on\ncontainer db. When disks where container db landed are under\nheavy loads, update on the container db may fail due to LockTimout.\n\nHence, we first check existence, if it\u0027s not there, we PUT.\n\nChange-Id: Ic61153948e35f1c09b05bfc97dfac3fb487b0898\nCloses-Bug: 1780204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e02b9b798007b9f8c70e33c22fa6189d81a41ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e02b9b798007b9f8c70e33c22fa6189d81a41ca3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
