)]}'
{"id":"openstack%2Fswift~601950","triplet_id":"openstack%2Fswift~master~Ib21568f9b9d8547da87a99d65ae73a550e9c3230","project":"openstack/swift","branch":"master","topic":"expirer-section-in-object-server-conf","hashtags":[],"change_id":"Ib21568f9b9d8547da87a99d65ae73a550e9c3230","subject":"Enable to configure object-expirer in object-server.conf","status":"MERGED","created":"2018-09-12 08:25:43.000000000","updated":"2019-05-05 00:25:11.000000000","submitted":"2019-05-05 00:25:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":47,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"601950-1557015911439-fc295863","meta_rev_id":"8685e901b4ad21051b9e9f9fc03feb109c1210cb","_number":601950,"virtual_id_number":601950,"owner":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-05-05 00:25:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2019-05-04 15:45:02.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-05-04 15:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"value":1,"date":"2019-05-04 15:52:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-08 06:29:10.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-02-01 15:49:03.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2019-03-27 21:36:29.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2019-05-02 20:28:15.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-05-03 20:19:19.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-04 15:52:55.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-05 00:25:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c93075d32d51553e6eef1c662592d5f4d4004b91","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 08:25:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4275b3bb1c3e02b82cf90e6a00aa3f926e0aa159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 08:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/1/check/openstack-tox-docs/2acbe39/html/ : SUCCESS in 3m 48s\n- swift-tox-py27 http://logs.openstack.org/50/601950/1/check/swift-tox-py27/17e2c4e/ : SUCCESS in 11m 19s\n- swift-tox-py35 http://logs.openstack.org/50/601950/1/check/swift-tox-py35/f757337/ : SUCCESS in 4m 57s\n- swift-tox-py36 http://logs.openstack.org/50/601950/1/check/swift-tox-py36/a3bd3ed/ : SUCCESS in 4m 53s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/1/check/swift-tox-func/2c9b6ad/ : SUCCESS in 9m 50s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/1/check/swift-tox-func-encryption/ccde02c/ : SUCCESS in 9m 27s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/1/check/swift-tox-func-domain-remap-staticweb/781014a/ : SUCCESS in 8m 28s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/1/check/swift-tox-func-ec/de9d315/ : SUCCESS in 14m 05s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/1/check/swift-tox-func-s3api/ba6db53/ : SUCCESS in 24m 06s\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/1/check/swift-probetests-centos-7/28ecb4c/ : SUCCESS in 27m 41s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/1/check/swift-dsvm-functional/92ca800/ : SUCCESS in 20m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/1/check/swift-dsvm-functional-ipv6/89258c9/ : SUCCESS in 21m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/601950/1/check/openstack-tox-lower-constraints/ae2bf61/ : SUCCESS in 8m 03s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/1/check/openstack-tox-pep8/a14ae9c/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b62c7fbf8e1c5e70b41b3b6c1f1e9bf54ffa7c01","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-12 16:11:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c87dab20de75e819cfe78d04abe284185c613e4","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-09-12 16:36:24.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\nThe direction seems nice to have.\n\nThat is because we can have directory-type configuration so that, we can push the file that has object-expirer section into the object-server conf dir instead of setting extra object-expirer.conf. So I\u0027m sure this is not so big impact for Swift operators but the docs on how to migrate their conf should be included in this patch.\n\nSo just +1 for this direction.","accounts_in_message":[],"_revision_number":2},{"id":"78095902e7f7017d7a7fa580901c144df7aa7798","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 18:00:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/2/check/openstack-tox-docs/1b1b963/html/ : SUCCESS in 4m 05s\n- swift-tox-py27 http://logs.openstack.org/50/601950/2/check/swift-tox-py27/6eef47b/ : SUCCESS in 9m 52s\n- swift-tox-py35 http://logs.openstack.org/50/601950/2/check/swift-tox-py35/945e945/ : SUCCESS in 4m 47s\n- swift-tox-py36 http://logs.openstack.org/50/601950/2/check/swift-tox-py36/ca224ef/ : SUCCESS in 4m 37s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/2/check/swift-tox-func/1d49a91/ : SUCCESS in 9m 13s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/2/check/swift-tox-func-encryption/f2dea3f/ : SUCCESS in 8m 38s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/2/check/swift-tox-func-domain-remap-staticweb/e0b1915/ : SUCCESS in 10m 10s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/2/check/swift-tox-func-ec/7a542e1/ : SUCCESS in 9m 29s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/2/check/swift-tox-func-s3api/4873b31/ : SUCCESS in 10m 53s\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/2/check/swift-probetests-centos-7/0683140/ : SUCCESS in 25m 50s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/2/check/swift-dsvm-functional/cff820f/ : SUCCESS in 26m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/2/check/swift-dsvm-functional-ipv6/a6b4ee3/ : SUCCESS in 22m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/601950/2/check/openstack-tox-lower-constraints/31fc4f1/ : SUCCESS in 7m 27s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/2/check/openstack-tox-pep8/efe6be4/ : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5b3db1975b0fd9e8e0168396e7bf3f6dac566a60","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-09-13 21:08:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eeb03c8b40b661b5881c92429d6b14288697afc9","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-14 23:57:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c1171d437840fb9363f7a4b807b53e1283d9c36","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-14 23:57:52.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6f7dd5b9bc1aa5acc7995d01a3a50397772e8216","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-09-15 00:07:37.000000000","message":"Patch Set 3: Workflow-1\n\nas discuss with kazuhiro quickly, this is still wip so set workflow -1","accounts_in_message":[],"_revision_number":3},{"id":"d9cf6daa0c979fb6697919b00fb20975a19df227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 00:29:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/3/check/openstack-tox-docs/91a1da0/html/ : SUCCESS in 3m 49s\n- swift-tox-py27 http://logs.openstack.org/50/601950/3/check/swift-tox-py27/5367a22/ : SUCCESS in 9m 44s\n- swift-tox-py35 http://logs.openstack.org/50/601950/3/check/swift-tox-py35/b4e9df0/ : SUCCESS in 4m 27s\n- swift-tox-py36 http://logs.openstack.org/50/601950/3/check/swift-tox-py36/a33553e/ : SUCCESS in 4m 46s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/3/check/swift-tox-func/b66a189/ : SUCCESS in 13m 22s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/3/check/swift-tox-func-encryption/a8e0e24/ : SUCCESS in 8m 21s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/3/check/swift-tox-func-domain-remap-staticweb/e3ee815/ : SUCCESS in 7m 44s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/3/check/swift-tox-func-ec/a968ec0/ : SUCCESS in 9m 35s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/3/check/swift-tox-func-s3api/bceb6b6/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b4b5f97/ : SUCCESS in 9m 29s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/3/check/swift-probetests-centos-7/979148f/ : SUCCESS in 22m 31s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/3/check/swift-dsvm-functional/65faf86/ : SUCCESS in 27m 19s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/3/check/swift-dsvm-functional-ipv6/5cb4184/ : SUCCESS in 22m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/601950/3/check/openstack-tox-lower-constraints/416ca8c/ : SUCCESS in 7m 45s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/3/check/openstack-tox-pep8/7c0f82d/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"328dc39d60f6c1f19f883259b9d2210f097364da","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-09-21 12:09:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"84f426c35c8254edb239c8beca26a3c57603adc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 13:06:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/50/601950/4/check/openstack-tox-cover/0e45d4e/cover/ : SUCCESS in 8m 48s\n- openstack-tox-docs http://logs.openstack.org/50/601950/4/check/openstack-tox-docs/ea2a563/html/ : SUCCESS in 4m 36s\n- swift-tox-py27 http://logs.openstack.org/50/601950/4/check/swift-tox-py27/b31fbb3/ : SUCCESS in 12m 05s\n- swift-tox-py35 http://logs.openstack.org/50/601950/4/check/swift-tox-py35/d7aa6b4/ : SUCCESS in 5m 35s\n- swift-tox-py36 http://logs.openstack.org/50/601950/4/check/swift-tox-py36/8e4b96b/ : SUCCESS in 4m 52s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/4/check/swift-tox-func/96dafd3/ : SUCCESS in 9m 39s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/4/check/swift-tox-func-encryption/3a95eb6/ : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/4/check/swift-tox-func-domain-remap-staticweb/74ab14c/ : SUCCESS in 9m 28s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/4/check/swift-tox-func-ec/791a18a/ : SUCCESS in 10m 47s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/4/check/swift-tox-func-s3api/071f72c/ : SUCCESS in 11m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/41f7a13/ : SUCCESS in 9m 20s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/4/check/swift-probetests-centos-7/00f377b/ : SUCCESS in 25m 20s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/4/check/swift-dsvm-functional/99289fb/ : SUCCESS in 31m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/4/check/swift-dsvm-functional-ipv6/7f14a91/ : SUCCESS in 27m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/4/check/swift-tox-lower-constraints/b6cd6f5/ : SUCCESS in 10m 47s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/4/check/openstack-tox-pep8/f58b67d/ : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/4/check/swift-multinode-rolling-upgrade/197497b/ : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":4},{"id":"3e9860f392936575cc2d76a47bdac6eabb3477ec","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-10-08 06:29:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)\n\nOnly part way through the review (still only in the documentation and config side. But I need to head for a meeting and then stopping for dinner, so will push up all my current comments.","accounts_in_message":[],"_revision_number":4},{"id":"026132459906885529c06f9230f620ab2920fcd7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-10-09 05:16:16.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3d276b9e18a76b20a02ffc25671f0fbcb3b3b60e","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-10-10 09:51:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2eb295b976e037243e310ccd9e3dfc82ce9602f9","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-10-10 09:51:33.000000000","message":"Patch Set 5:\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1987fad97d84bc1043b84a556261d540fc546941","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 11:03:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/50/601950/5/check/openstack-tox-cover/99ca721/cover/ : SUCCESS in 10m 35s\n- openstack-tox-docs http://logs.openstack.org/50/601950/5/check/openstack-tox-docs/524d0ba/html/ : SUCCESS in 4m 49s\n- swift-tox-py27 http://logs.openstack.org/50/601950/5/check/swift-tox-py27/6be59f9/ : SUCCESS in 16m 57s\n- swift-tox-py35 http://logs.openstack.org/50/601950/5/check/swift-tox-py35/12a61e9/ : SUCCESS in 4m 48s\n- swift-tox-py36 http://logs.openstack.org/50/601950/5/check/swift-tox-py36/6a35361/ : SUCCESS in 4m 26s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/5/check/swift-tox-func/7df1846/ : SUCCESS in 8m 39s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/5/check/swift-tox-func-encryption/1533a59/ : SUCCESS in 18m 29s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/5/check/swift-tox-func-domain-remap-staticweb/86d7313/ : SUCCESS in 9m 51s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/5/check/swift-tox-func-ec/0d08b64/ : SUCCESS in 9m 13s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/5/check/swift-tox-func-s3api/0e536df/ : SUCCESS in 13m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2335170/ : SUCCESS in 10m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/5/check/swift-probetests-centos-7/b92a3ab/ : SUCCESS in 28m 28s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/5/check/swift-dsvm-functional/5cc6f44/ : SUCCESS in 28m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/5/check/swift-dsvm-functional-ipv6/5e4a5ff/ : SUCCESS in 24m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/5/check/swift-tox-lower-constraints/7eed9ba/ : SUCCESS in 12m 40s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/5/check/openstack-tox-pep8/e7ffec1/ : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/5/check/swift-multinode-rolling-upgrade/4c61ad9/ : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":5},{"id":"aa3bd12f3f3f32d8896a4dc6b93aebbbd58a1633","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-11-28 08:36:11.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"57ce75b6b399273ea3390b7d94453c145e29a9e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 09:49:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/6/check/openstack-tox-docs/625774b/html/ : SUCCESS in 4m 23s\n- swift-tox-py27 http://logs.openstack.org/50/601950/6/check/swift-tox-py27/412e9da/ : SUCCESS in 11m 25s\n- swift-tox-py35 http://logs.openstack.org/50/601950/6/check/swift-tox-py35/8d90ff8/ : SUCCESS in 10m 48s\n- swift-tox-py36 http://logs.openstack.org/50/601950/6/check/swift-tox-py36/c37f473/ : SUCCESS in 6m 13s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/6/check/swift-tox-func/6dea7aa/ : SUCCESS in 8m 54s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/6/check/swift-tox-func-encryption/8002ecd/ : SUCCESS in 14m 20s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/6/check/swift-tox-func-domain-remap-staticweb/7d2c6ac/ : SUCCESS in 8m 41s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/6/check/swift-tox-func-ec/45bb14d/ : SUCCESS in 14m 24s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/6/check/swift-tox-func-s3api/8859405/ : SUCCESS in 11m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/6/check/swift-tox-func-s3api-ceph-s3tests-tempauth/5e8b8d2/ : SUCCESS in 10m 38s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/6/check/swift-probetests-centos-7/3d14910/ : SUCCESS in 25m 19s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/6/check/swift-dsvm-functional/98b5280/ : SUCCESS in 31m 16s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/6/check/swift-dsvm-functional-ipv6/6a5d62d/ : SUCCESS in 28m 06s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/6/check/swift-tox-lower-constraints/b590769/ : SUCCESS in 12m 19s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/6/check/openstack-tox-pep8/2743f75/ : SUCCESS in 9m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/6/check/swift-multinode-rolling-upgrade/374f353/ : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":6},{"id":"035f75cc52304033db0d1c907ee6b505d6fcf324","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 02:41:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4102a79e3ce8da4d856cc20ed704e867ac14a02e","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 02:44:26.000000000","message":"Patch Set 7:\n\nSquashed matthew\u0027s following patch[1].\n\n[1]: https://review.openstack.org/#/c/616076","accounts_in_message":[],"_revision_number":7},{"id":"75583b181e3ae3adc756173c5b0f42c8d9120570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 03:15:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/7/check/openstack-tox-docs/e93cdf1/html/ : SUCCESS in 3m 32s\n- swift-tox-py27 http://logs.openstack.org/50/601950/7/check/swift-tox-py27/53dcc90/ : SUCCESS in 10m 15s\n- swift-tox-py35 http://logs.openstack.org/50/601950/7/check/swift-tox-py35/0f85fe3/ : SUCCESS in 5m 29s\n- swift-tox-py36 http://logs.openstack.org/50/601950/7/check/swift-tox-py36/48bded4/ : SUCCESS in 4m 46s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/7/check/swift-tox-func/edf7263/ : SUCCESS in 9m 13s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/7/check/swift-tox-func-encryption/eb02205/ : SUCCESS in 8m 47s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/7/check/swift-tox-func-domain-remap-staticweb/a046990/ : SUCCESS in 8m 36s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/7/check/swift-tox-func-ec/1529e13/ : SUCCESS in 8m 50s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/7/check/swift-tox-func-s3api/65bf2c0/ : SUCCESS in 9m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/7/check/swift-tox-func-s3api-ceph-s3tests-tempauth/da4d9d1/ : SUCCESS in 12m 17s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/7/check/swift-probetests-centos-7/112d1a6/ : SUCCESS in 29m 01s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/7/check/swift-dsvm-functional/6ee1141/ : SUCCESS in 20m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/7/check/swift-dsvm-functional-ipv6/ec7d380/ : SUCCESS in 26m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/7/check/swift-tox-lower-constraints/d5c5ec5/ : SUCCESS in 11m 12s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/7/check/openstack-tox-pep8/4dfb5cc/ : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/7/check/swift-multinode-rolling-upgrade/16f7455/ : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":7},{"id":"03ecf228cbdf73158102c7b5c12e3028a80b2dc5","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 04:29:53.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1ac318841febb9d9adfb8e487be5ce6c67be67c4","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 04:31:28.000000000","message":"Patch Set 8:\n\nremoved extra space","accounts_in_message":[],"_revision_number":8},{"id":"c6c1c4cf22341f20fa13f1a149078680d614477d","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 04:52:14.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"76eb076409f61c3e706f60960e87600dbc5bd03f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 05:23:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/9/check/openstack-tox-docs/a4e0b62/html/ : SUCCESS in 3m 35s\n- swift-tox-py27 http://logs.openstack.org/50/601950/9/check/swift-tox-py27/d210621/ : SUCCESS in 14m 22s\n- swift-tox-py35 http://logs.openstack.org/50/601950/9/check/swift-tox-py35/2d12aac/ : SUCCESS in 5m 10s\n- swift-tox-py36 http://logs.openstack.org/50/601950/9/check/swift-tox-py36/59b1591/ : SUCCESS in 6m 17s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/9/check/swift-tox-func/a6c2b6b/ : SUCCESS in 10m 03s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/9/check/swift-tox-func-encryption/2190eb7/ : SUCCESS in 9m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/9/check/swift-tox-func-domain-remap-staticweb/a6f227e/ : SUCCESS in 12m 05s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/9/check/swift-tox-func-ec/2cb7583/ : SUCCESS in 8m 03s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/9/check/swift-tox-func-s3api/7ed06a7/ : FAILURE in 9m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/9/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ff77130/ : SUCCESS in 12m 44s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/9/check/swift-probetests-centos-7/9e5c0be/ : SUCCESS in 29m 51s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/9/check/swift-dsvm-functional/e10b103/ : SUCCESS in 23m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/9/check/swift-dsvm-functional-ipv6/15040bf/ : SUCCESS in 24m 30s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/9/check/swift-tox-lower-constraints/f79bcd8/ : SUCCESS in 9m 20s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/9/check/openstack-tox-pep8/41796a3/ : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/9/check/swift-multinode-rolling-upgrade/e82d834/ : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":9},{"id":"94ddaa9cad3909fd697306288886ce0a202b4feb","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 05:30:26.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"b5e282bff297ab0e582e555a5d98bbc60f701a8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 06:02:36.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/9/check/openstack-tox-docs/f5a9b74/html/ : SUCCESS in 3m 56s\n- swift-tox-py27 http://logs.openstack.org/50/601950/9/check/swift-tox-py27/2178583/ : SUCCESS in 10m 00s\n- swift-tox-py35 http://logs.openstack.org/50/601950/9/check/swift-tox-py35/e1e2ff3/ : SUCCESS in 5m 20s\n- swift-tox-py36 http://logs.openstack.org/50/601950/9/check/swift-tox-py36/bc552ec/ : SUCCESS in 9m 24s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/9/check/swift-tox-func/90e9c63/ : SUCCESS in 8m 47s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/9/check/swift-tox-func-encryption/68fbe69/ : FAILURE in 8m 45s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/9/check/swift-tox-func-domain-remap-staticweb/21a500e/ : SUCCESS in 7m 41s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/9/check/swift-tox-func-ec/03f638c/ : FAILURE in 8m 36s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/9/check/swift-tox-func-s3api/0a50936/ : SUCCESS in 11m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/9/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4e0a0b6/ : SUCCESS in 10m 45s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/9/check/swift-probetests-centos-7/da12511/ : SUCCESS in 30m 04s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/9/check/swift-dsvm-functional/e1df8fa/ : SUCCESS in 22m 29s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/9/check/swift-dsvm-functional-ipv6/70d98e8/ : SUCCESS in 25m 20s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/9/check/swift-tox-lower-constraints/74ece94/ : SUCCESS in 9m 33s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/9/check/openstack-tox-pep8/0de511d/ : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/9/check/swift-multinode-rolling-upgrade/ee4023e/ : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":9},{"id":"ec064e213d8be488661f3a99e0f9490dcfe518f3","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 06:21:31.000000000","message":"Patch Set 9:\n\nUmm...at first test for patch set 9, only swift-tox-func-s3api failed.\nAt second test for patch set 9, swift-tox-func-encryption and swift-tox-func-ec failed.\nThe errors came from 503: \u0027Service Unavailable\u0027.\n\nSo, I think the FAILUREs don\u0027t come from the codes.","accounts_in_message":[],"_revision_number":9},{"id":"5c75071e628f50f9cefc84e25e73d1b9eb944129","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 06:21:45.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"d8cbdb37aeb6c1bf6950da8cbc8d4dc3eec248f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 06:56:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/9/check/openstack-tox-docs/71aab5a/html/ : SUCCESS in 4m 07s\n- swift-tox-py27 http://logs.openstack.org/50/601950/9/check/swift-tox-py27/1314489/ : SUCCESS in 12m 06s\n- swift-tox-py35 http://logs.openstack.org/50/601950/9/check/swift-tox-py35/728f394/ : SUCCESS in 5m 55s\n- swift-tox-py36 http://logs.openstack.org/50/601950/9/check/swift-tox-py36/dbc18bc/ : SUCCESS in 5m 40s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/9/check/swift-tox-func/30926bc/ : SUCCESS in 8m 53s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/9/check/swift-tox-func-encryption/591a84c/ : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/9/check/swift-tox-func-domain-remap-staticweb/f7ba056/ : SUCCESS in 8m 20s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/9/check/swift-tox-func-ec/a2a788d/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/9/check/swift-tox-func-s3api/3066930/ : SUCCESS in 9m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/9/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f7f91bd/ : SUCCESS in 12m 22s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/9/check/swift-probetests-centos-7/0742ced/ : SUCCESS in 29m 37s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/9/check/swift-dsvm-functional/8d22ff4/ : SUCCESS in 22m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/9/check/swift-dsvm-functional-ipv6/00e4182/ : SUCCESS in 25m 23s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/9/check/swift-tox-lower-constraints/e4cda6a/ : SUCCESS in 11m 31s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/9/check/openstack-tox-pep8/5609215/ : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/9/check/swift-multinode-rolling-upgrade/f25243b/ : SUCCESS in 17m 00s","accounts_in_message":[],"_revision_number":9},{"id":"424a5fdc69a2c9d79a36760630870402d305a60a","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-26 07:12:40.000000000","message":"Patch Set 9:\n\nok. ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"06994b42852341ec92e65b79db2e4b94550c9d24","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 09:52:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ae460a9e44af997fc72f3e356277260385b65800","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 09:55:32.000000000","message":"Patch Set 10:\n\nI found a bug that the manager cannot find config path of object-server/1.conf.d with [object-expirer] section.\nThe patch set 10 fixed the bug.","accounts_in_message":[],"_revision_number":10},{"id":"99e54cad5a08a11e56d8dbfa877d253cbe035cb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 10:27:05.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/10/check/openstack-tox-docs/f0ce277/html/ : SUCCESS in 4m 01s\n- swift-tox-py27 http://logs.openstack.org/50/601950/10/check/swift-tox-py27/aebd032/ : SUCCESS in 10m 05s\n- swift-tox-py35 http://logs.openstack.org/50/601950/10/check/swift-tox-py35/1ec6839/ : SUCCESS in 5m 07s\n- swift-tox-py36 http://logs.openstack.org/50/601950/10/check/swift-tox-py36/d40269e/ : SUCCESS in 5m 17s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/10/check/swift-tox-func/4af247c/ : SUCCESS in 7m 41s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/10/check/swift-tox-func-encryption/0b4e163/ : SUCCESS in 7m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/10/check/swift-tox-func-domain-remap-staticweb/51e77aa/ : SUCCESS in 7m 58s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/10/check/swift-tox-func-ec/299154a/ : SUCCESS in 8m 15s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/10/check/swift-tox-func-s3api/8041ed6/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/10/check/swift-tox-func-s3api-ceph-s3tests-tempauth/44f7101/ : SUCCESS in 10m 27s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/10/check/swift-probetests-centos-7/b50bc13/ : SUCCESS in 28m 22s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/10/check/swift-dsvm-functional/c9443c5/ : SUCCESS in 23m 15s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/10/check/swift-dsvm-functional-ipv6/0b24dec/ : SUCCESS in 27m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/10/check/swift-tox-lower-constraints/ddb6b5a/ : SUCCESS in 9m 33s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/10/check/openstack-tox-pep8/47a2d13/ : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/10/check/swift-multinode-rolling-upgrade/9fdca9f/ : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":10},{"id":"847b38f70fcaa942f473c4f3f40b7b8d358ac135","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 10:41:11.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5aeebbb5f754d0c1a019b81e15b4559288b327d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 11:15:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/11/check/openstack-tox-docs/970abb1/html/ : SUCCESS in 3m 29s\n- swift-tox-py27 http://logs.openstack.org/50/601950/11/check/swift-tox-py27/cd260ce/ : SUCCESS in 9m 53s\n- swift-tox-py35 http://logs.openstack.org/50/601950/11/check/swift-tox-py35/9782091/ : SUCCESS in 6m 16s\n- swift-tox-py36 http://logs.openstack.org/50/601950/11/check/swift-tox-py36/cbaf18e/ : SUCCESS in 4m 53s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/11/check/swift-tox-func/571affc/ : SUCCESS in 10m 09s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/11/check/swift-tox-func-encryption/497a7cc/ : SUCCESS in 8m 20s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/11/check/swift-tox-func-domain-remap-staticweb/af6d2dd/ : SUCCESS in 8m 24s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/11/check/swift-tox-func-ec/2a44fed/ : SUCCESS in 8m 46s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/11/check/swift-tox-func-s3api/dfb9817/ : SUCCESS in 11m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/11/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ef4dc7b/ : TIMED_OUT in 30m 56s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/11/check/swift-probetests-centos-7/bf198f3/ : SUCCESS in 28m 37s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/11/check/swift-dsvm-functional/433905a/ : SUCCESS in 27m 47s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/11/check/swift-dsvm-functional-ipv6/d751cf2/ : SUCCESS in 25m 21s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/11/check/swift-tox-lower-constraints/e082933/ : SUCCESS in 10m 08s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/11/check/openstack-tox-pep8/bbf3d97/ : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/11/check/swift-multinode-rolling-upgrade/d75b2b7/ : SUCCESS in 16m 05s","accounts_in_message":[],"_revision_number":11},{"id":"5ab4f933472b7aeee864cb8957cfe6f0d559d6ca","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2018-12-27 11:39:05.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"b03a5805a866d7c6264625b1e613da947f25515c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 12:23:22.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/11/check/openstack-tox-docs/9e9619b/html/ : SUCCESS in 3m 55s\n- swift-tox-py27 http://logs.openstack.org/50/601950/11/check/swift-tox-py27/da83c2d/ : SUCCESS in 9m 31s\n- swift-tox-py35 http://logs.openstack.org/50/601950/11/check/swift-tox-py35/c2f6daf/ : SUCCESS in 5m 13s\n- swift-tox-py36 http://logs.openstack.org/50/601950/11/check/swift-tox-py36/7e23b27/ : SUCCESS in 5m 19s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/11/check/swift-tox-func/d11b36b/ : SUCCESS in 8m 22s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/11/check/swift-tox-func-encryption/8b9ab6e/ : SUCCESS in 8m 07s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/11/check/swift-tox-func-domain-remap-staticweb/88a05a2/ : SUCCESS in 8m 29s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/11/check/swift-tox-func-ec/2643ed0/ : SUCCESS in 9m 19s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/11/check/swift-tox-func-s3api/bbf6c69/ : SUCCESS in 10m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/11/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f3a1925/ : SUCCESS in 10m 27s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/11/check/swift-probetests-centos-7/ab4ebda/ : SUCCESS in 43m 16s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/11/check/swift-dsvm-functional/7e55c68/ : SUCCESS in 25m 47s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/11/check/swift-dsvm-functional-ipv6/564d6ba/ : SUCCESS in 21m 18s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/11/check/swift-tox-lower-constraints/803b978/ : SUCCESS in 9m 55s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/11/check/openstack-tox-pep8/3b708bb/ : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/11/check/swift-multinode-rolling-upgrade/c34c634/ : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":11},{"id":"ab5af9304b45d32bd72ef6f0728df8ce0ac0acf9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-01-29 02:56:48.000000000","message":"Patch Set 11: Code-Review+2\n\nThis looks ready to me.","accounts_in_message":[],"_revision_number":11},{"id":"6e40093cd56c46cdce614fe44021bf4c091552a8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-01-29 02:59:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"14917f4245cea3b22c291762ca1e1bb423d8ac20","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2019-02-01 15:49:03.000000000","message":"Patch Set 11:\n\n(2 comments)\n\nI\u0027m wondering if we could do better than check on the file name.\n\n1/ we can assume that if the process IS NOT on an object-server, we need to run legacy mode\n2/ we can assume that if the process IS on an object-server, we need to run new mode\n3/ can we check the content of the config (like pipeline variable) to check if this is an object-server or not","accounts_in_message":[],"_revision_number":11},{"id":"93202bd908ca2e47397c0e246f414908e0d048d1","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-02-18 10:06:12.000000000","message":"Patch Set 11:\n\n(2 comments)\n\nPete, Romain, thank you for your reviews.\n\n\u003e 3/ can we check the content of the config (like pipeline variable) to check if this is an object-server or not\n\nI think checking config with its content will be complicated.\n\nAFAIK, except for pipeline, any variables are optional. So only pipeline variable can be used from config content.\n\nIf we check a config with its pipeline variable, we need to check pipeline members contains object-server or not. Because both of object-server.conf and object-expirer.conf contains pipeline variable.\n\nIf I understand correctly, pipeline variable names are editable.\nIt means that following object-server.conf can exist.\n\n\u003d\u003d\u003d object-server.conf \u003d\u003d\u003d\n[pipeline:main]\npipeline \u003d healthcheck recon obj-sv\n\n[app:obj-sv]\nuse \u003d egg:swift#object\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nSo the pipeline check requires check of egg. But it depends on egg settings. The implementation will be complicated.\n\n\nHow about following behavior?\n\nCase 1: conf file path contains \"object-expirer\"\n  - The expirer run as legacy mode.\n\nCase 2: conf file path contains \"object-server\"\n  - The expirer\u0027s mode depends on \"dequeue_from_legacy\" setting.\n  - \"dequeue_from_legacy\"\u0027s default is false.\n\nCase 3: otherwise (e.g. \"object-expiration.conf\", \"obj-sv.conf\")\n  - The expirer\u0027s mode depends on \"dequeue_from_legacy\" setting.\n  - \"dequeue_from_legacy\"\u0027s default is **true** for backward compatibility.\n  - If the default value is used, the fact is logged as warn-log.\n      - The log message will be\n        \"\"\"\n        Object-expirer\u0027s `dequeue_from_legacy` value is not set in the conf file.\n        It runs with `dequeue_from_legacy \u003d\u003d True` for backward compatibility.\n        It means that the object-expirer\u0027s configuration may be in the style of object-expirer.conf.\n        Object-expirer.conf style is deprecated. Move the configuration into object-server.conf.\n        If you want suppress this warning, set `dequeue_from_legacy` explicitly.\n        \"\"\"\n      - Any other message ideas are welcome :)","accounts_in_message":[],"_revision_number":11},{"id":"26a4b412fa80feabd7c17f053f0713029123c4ab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-03-27 21:32:08.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\ni don\u0027t really understand how this works - if I\u0027m confused please ignore me","accounts_in_message":[],"_revision_number":11},{"id":"3cc64ff26797e360b0879ad30767ae56281a6f90","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-03-27 21:36:29.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6c74a35c3c6e24156b299f720896a03df4936317","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-03-29 04:38:57.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"90ea6b2c60d1794a4997e8843844dea9eebb5d19","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-03-29 04:43:55.000000000","message":"Patch Set 12:\n\n(5 comments)\n\nI fixed some bugs and updated commit message for reviewers.","accounts_in_message":[],"_revision_number":12},{"id":"2e7e3d1164c2adc2f5642ce2c223520e612d8e6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 05:12:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/12/check/openstack-tox-docs/adefe9f/html/ : SUCCESS in 4m 24s\n- swift-tox-py27 http://logs.openstack.org/50/601950/12/check/swift-tox-py27/447067a/ : FAILURE in 11m 38s\n- swift-tox-py35 http://logs.openstack.org/50/601950/12/check/swift-tox-py35/cd75a71/ : SUCCESS in 9m 30s\n- swift-tox-py36 http://logs.openstack.org/50/601950/12/check/swift-tox-py36/5399257/ : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/12/check/swift-tox-py37/501e7dc/ : SUCCESS in 9m 02s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/12/check/swift-tox-func/6cdfa99/ : SUCCESS in 8m 57s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/12/check/swift-tox-func-encryption/dca55c7/ : SUCCESS in 8m 39s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/12/check/swift-tox-func-domain-remap-staticweb/3c8343e/ : SUCCESS in 8m 18s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/12/check/swift-tox-func-ec/6e1c7d8/ : SUCCESS in 8m 20s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/12/check/swift-tox-func-s3api/f12cad6/ : SUCCESS in 9m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/12/check/swift-tox-func-s3api-ceph-s3tests-tempauth/96ab7d8/ : SUCCESS in 13m 14s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/12/check/swift-probetests-centos-7/dbe102c/ : SUCCESS in 32m 58s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/12/check/swift-dsvm-functional/bb0a176/ : SUCCESS in 24m 10s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/12/check/swift-dsvm-functional-ipv6/968122c/ : SUCCESS in 25m 25s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/12/check/swift-tox-lower-constraints/c8a5ad2/ : FAILURE in 13m 45s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/12/check/openstack-tox-pep8/dd29d00/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/12/check/swift-multinode-rolling-upgrade/8a643be/ : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":12},{"id":"3f1d48c4278993000f4e237665e6938770dc308b","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-03-29 05:39:42.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c18d18742a5fab418c47738b04463c5385a6a7bf","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-03-29 05:44:00.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"24117794d0ce4a616ea1a6eb5996163119311a99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 06:13:48.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/13/check/openstack-tox-docs/ec25b71/html/ : SUCCESS in 3m 57s\n- swift-tox-py27 http://logs.openstack.org/50/601950/13/check/swift-tox-py27/e5603f7/ : SUCCESS in 10m 51s\n- swift-tox-py35 http://logs.openstack.org/50/601950/13/check/swift-tox-py35/461ff79/ : SUCCESS in 8m 59s\n- swift-tox-py36 http://logs.openstack.org/50/601950/13/check/swift-tox-py36/23ba1e2/ : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/13/check/swift-tox-py37/3cc6245/ : SUCCESS in 8m 15s (non-voting)\n- swift-tox-func http://logs.openstack.org/50/601950/13/check/swift-tox-func/9e7c5b0/ : SUCCESS in 8m 32s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/13/check/swift-tox-func-encryption/e6ecac2/ : SUCCESS in 9m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/13/check/swift-tox-func-domain-remap-staticweb/3a383a3/ : SUCCESS in 8m 12s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/13/check/swift-tox-func-ec/256ce48/ : SUCCESS in 8m 50s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/13/check/swift-tox-func-s3api/3b251da/ : SUCCESS in 9m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/13/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f0ff644/ : SUCCESS in 11m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/13/check/swift-probetests-centos-7/e0197b8/ : SUCCESS in 28m 47s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/13/check/swift-dsvm-functional/a9d4d22/ : SUCCESS in 26m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/13/check/swift-dsvm-functional-ipv6/fe9aa3e/ : SUCCESS in 27m 02s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/13/check/swift-tox-lower-constraints/6c5e863/ : SUCCESS in 12m 26s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/13/check/openstack-tox-pep8/c242133/ : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/13/check/swift-multinode-rolling-upgrade/a716ca4/ : SUCCESS in 18m 40s","accounts_in_message":[],"_revision_number":13},{"id":"ccf53185426e1aa42f9ebc77be4f0b1a52e148e5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-02 17:16:16.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)\n\nOk, I understand the reasoning for this option being the way it is.\n\nOperators will want to \"migrate\" expirer to object layer but many object nodes will not run in legacy mode.\n\nIf you currently run with n_proccesses \u003d 24 and you have 64 object nodes you probably want to migrate those options to 24 of the 64 object nodes and turn dequeue_from_legacy to false on the rest.","accounts_in_message":[],"_revision_number":13},{"id":"3a49891bdb4e100066aac36f18f2206951c5b461","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-02 20:28:15.000000000","message":"Patch Set 13: Code-Review+2\n\nI found it akward to have to deal with the processes/process option(s) and ALSO the dequeue_from_legacy option...\n\nI want to look at the follow up patch incase I missed something that maybe should get pulled out into a earlier patch, but that could probably that can be follow-up...\n\nI don\u0027t think I have any reason not merge this.  It WOMM!","accounts_in_message":[],"_revision_number":13},{"id":"3b3efe30fb6f10419aa206257d2cbd8f03d0ab48","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-02 21:55:56.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"2d2bdd1808d2c263f0b0d3a9a895bf969679b3b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:31:35.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/14/check/openstack-tox-docs/a74ab00/html/ : SUCCESS in 4m 18s\n- swift-tox-py27 http://logs.openstack.org/50/601950/14/check/swift-tox-py27/a02f719/ : SUCCESS in 10m 01s\n- swift-tox-py35 http://logs.openstack.org/50/601950/14/check/swift-tox-py35/8b21663/ : SUCCESS in 10m 21s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/14/check/swift-tox-py36/37021ca/ : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/14/check/swift-tox-py37/94c2399/ : SUCCESS in 10m 20s\n- swift-tox-func http://logs.openstack.org/50/601950/14/check/swift-tox-func/7369c09/ : SUCCESS in 9m 24s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/14/check/swift-tox-func-encryption/c69c79a/ : SUCCESS in 8m 30s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/14/check/swift-tox-func-domain-remap-staticweb/6621864/ : SUCCESS in 8m 52s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/14/check/swift-tox-func-ec/2cfb65f/ : SUCCESS in 10m 07s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api/4658974/ : SUCCESS in 11m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7d2bc26/ : SUCCESS in 11m 32s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/14/check/swift-probetests-centos-7/34fd8b7/ : SUCCESS in 30m 14s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional/847e76b/ : SUCCESS in 25m 16s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional-ipv6/faa4ba9/ : SUCCESS in 27m 40s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/14/check/swift-tox-lower-constraints/9937a4e/ : SUCCESS in 12m 15s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/14/check/openstack-tox-pep8/143c788/ : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/14/check/swift-multinode-rolling-upgrade/b45a941/ : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":14},{"id":"57f3b202f6f83cda1235bd468bd6f16530ef6051","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-03 20:19:19.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"24931113f6007dd3a69b4da71de8b3cfa2bdf652","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 20:20:11.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"ded85c9a1085470a7d2ff5c90085602752ced3e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 22:12:02.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/14/gate/openstack-tox-docs/ef5a9ec/html/ : SUCCESS in 4m 01s\n- swift-tox-py27 http://logs.openstack.org/50/601950/14/gate/swift-tox-py27/5a37d62/ : TIMED_OUT in 40m 56s\n- swift-tox-py37 http://logs.openstack.org/50/601950/14/gate/swift-tox-py37/9128b51/ : FAILURE in 38m 44s\n- swift-tox-func http://logs.openstack.org/50/601950/14/gate/swift-tox-func/fbd16c8/ : SUCCESS in 27m 48s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/14/gate/swift-tox-func-encryption/8db195e/ : SUCCESS in 10m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/14/gate/swift-tox-func-domain-remap-staticweb/5f82889/ : SUCCESS in 10m 37s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/14/gate/swift-tox-func-ec/bf6a879/ : SUCCESS in 11m 23s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/14/gate/swift-tox-func-s3api/52f7b59/ : SUCCESS in 10m 37s\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/14/gate/swift-probetests-centos-7/3726197/ : SUCCESS in 31m 48s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/14/gate/swift-dsvm-functional/3d0ba98/ : SUCCESS in 27m 15s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/14/gate/swift-dsvm-functional-ipv6/0e96c61/ : SUCCESS in 27m 01s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/14/gate/swift-tox-lower-constraints/ef882b4/ : SUCCESS in 11m 37s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/14/gate/openstack-tox-pep8/4e3abaa/ : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":14},{"id":"8669abcb360d5566524ab31ed37c2e41c21a867f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 04:35:37.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"9949606fbd23e3dd2b6e31525da99417bac7733b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 05:18:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/14/check/openstack-tox-docs/ce0317c/html/ : SUCCESS in 3m 56s\n- swift-tox-py27 http://logs.openstack.org/50/601950/14/check/swift-tox-py27/f50ccce/ : SUCCESS in 9m 55s\n- swift-tox-py35 http://logs.openstack.org/50/601950/14/check/swift-tox-py35/0f6a6e4/ : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/14/check/swift-tox-py36/06984e2/ : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/14/check/swift-tox-py37/e47eec0/ : SUCCESS in 7m 39s\n- swift-tox-func http://logs.openstack.org/50/601950/14/check/swift-tox-func/741846e/ : FAILURE in 3m 11s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/14/check/swift-tox-func-encryption/20c77d5/ : FAILURE in 3m 14s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/14/check/swift-tox-func-domain-remap-staticweb/693694d/ : FAILURE in 3m 04s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/14/check/swift-tox-func-ec/2ffc2a9/ : FAILURE in 3m 16s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api/8e83136/ : FAILURE in 3m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api-ceph-s3tests-tempauth/70504cd/ : SUCCESS in 10m 50s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/14/check/swift-probetests-centos-7/57c9eb0/ : SUCCESS in 27m 27s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional/eaf4f86/ : FAILURE in 16m 52s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional-ipv6/7f3ef01/ : FAILURE in 17m 23s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/14/check/swift-tox-lower-constraints/4f8ac91/ : SUCCESS in 11m 32s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/14/check/openstack-tox-pep8/1f63d57/ : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/14/check/swift-multinode-rolling-upgrade/ffe8079/ : FAILURE in 7m 41s","accounts_in_message":[],"_revision_number":14},{"id":"b8dcb4c3598c353bb73e816350e8caaa3b7b415a","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-04 07:00:52.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"dadf4ccdf8f293009c004e7ed342baf3574bead5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 07:34:49.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/14/check/openstack-tox-docs/b39c458/html/ : SUCCESS in 3m 44s\n- swift-tox-py27 http://logs.openstack.org/50/601950/14/check/swift-tox-py27/e528ca0/ : SUCCESS in 9m 53s\n- swift-tox-py35 http://logs.openstack.org/50/601950/14/check/swift-tox-py35/11bdafa/ : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/14/check/swift-tox-py36/3565e11/ : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/14/check/swift-tox-py37/c002408/ : SUCCESS in 7m 58s\n- swift-tox-func http://logs.openstack.org/50/601950/14/check/swift-tox-func/bddfaaa/ : FAILURE in 3m 19s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/14/check/swift-tox-func-encryption/422742b/ : FAILURE in 3m 14s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/14/check/swift-tox-func-domain-remap-staticweb/6145ce1/ : FAILURE in 3m 09s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/14/check/swift-tox-func-ec/b742d1f/ : FAILURE in 3m 13s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api/f62b4b6/ : FAILURE in 3m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api-ceph-s3tests-tempauth/3e30b35/ : SUCCESS in 11m 34s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/14/check/swift-probetests-centos-7/0a6a74d/ : SUCCESS in 28m 39s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional/3853a58/ : FAILURE in 18m 11s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional-ipv6/d3e683f/ : FAILURE in 17m 05s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/14/check/swift-tox-lower-constraints/c510aed/ : SUCCESS in 12m 07s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/14/check/openstack-tox-pep8/f106d04/ : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/14/check/swift-multinode-rolling-upgrade/092557a/ : FAILURE in 7m 34s","accounts_in_message":[],"_revision_number":14},{"id":"5f92b1a076e65032bfb40c5074e2b324846c0bd3","author":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"date":"2019-05-04 14:01:12.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"35b2335973071ea98d04b582ad0c7f5fb6ad02a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 14:38:24.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/14/check/openstack-tox-docs/c651f15/html/ : SUCCESS in 3m 20s\n- swift-tox-py27 http://logs.openstack.org/50/601950/14/check/swift-tox-py27/e9825fb/ : SUCCESS in 13m 00s\n- swift-tox-py35 http://logs.openstack.org/50/601950/14/check/swift-tox-py35/ab78b2b/ : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/14/check/swift-tox-py36/40d5146/ : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/14/check/swift-tox-py37/660da59/ : SUCCESS in 7m 17s\n- swift-tox-func http://logs.openstack.org/50/601950/14/check/swift-tox-func/dbba1ce/ : FAILURE in 3m 15s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/14/check/swift-tox-func-encryption/cce6a09/ : FAILURE in 3m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/14/check/swift-tox-func-domain-remap-staticweb/65f4a6d/ : FAILURE in 5m 55s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/14/check/swift-tox-func-ec/23b17ec/ : FAILURE in 3m 52s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api/9efdd0a/ : FAILURE in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/14/check/swift-tox-func-s3api-ceph-s3tests-tempauth/da7a391/ : SUCCESS in 10m 53s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/14/check/swift-probetests-centos-7/9aeb532/ : SUCCESS in 32m 05s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional/e8069c9/ : FAILURE in 14m 58s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/14/check/swift-dsvm-functional-ipv6/c024d64/ : FAILURE in 18m 28s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/14/check/swift-tox-lower-constraints/5c3e42c/ : SUCCESS in 11m 14s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/14/check/openstack-tox-pep8/c964df2/ : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/14/check/swift-multinode-rolling-upgrade/1523a3f/ : FAILURE in 9m 27s","accounts_in_message":[],"_revision_number":14},{"id":"9d289289728a48da48d7fd32efd23704e06a491a","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-04 15:45:02.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"18d6e7078204e8c0ba14a48cd9659a50e6e95548","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 15:52:55.000000000","message":"Patch Set 15: Workflow+1\n\nRe-approving based on prior approval.","accounts_in_message":[],"_revision_number":15},{"id":"dc63c483946a2d202fc767360a50e00fe04ac355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:20:12.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/check/openstack-tox-docs/67c24f3/html/ : SUCCESS in 4m 00s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/check/swift-tox-py27/01ec2e3/ : SUCCESS in 9m 53s\n- swift-tox-py35 http://logs.openstack.org/50/601950/15/check/swift-tox-py35/777cf9f/ : SUCCESS in 10m 24s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/15/check/swift-tox-py36/7d97042/ : SUCCESS in 7m 43s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/check/swift-tox-py37/319b44c/ : SUCCESS in 9m 19s\n- swift-tox-func http://logs.openstack.org/50/601950/15/check/swift-tox-func/ca15940/ : SUCCESS in 8m 38s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/check/swift-tox-func-encryption/bcc022f/ : SUCCESS in 8m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/check/swift-tox-func-domain-remap-staticweb/c4c2fbb/ : SUCCESS in 8m 47s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/check/swift-tox-func-ec/0485108/ : SUCCESS in 8m 31s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api/8362cb5/ : SUCCESS in 9m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api-ceph-s3tests-tempauth/5f75bdb/ : SUCCESS in 11m 23s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/check/swift-probetests-centos-7/fe58293/ : FAILURE in 29m 54s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional/92ce735/ : SUCCESS in 26m 26s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional-ipv6/5d27bc3/ : SUCCESS in 26m 00s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/check/swift-tox-lower-constraints/349a1d0/ : SUCCESS in 10m 49s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/check/openstack-tox-pep8/9cf1d79/ : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/15/check/swift-multinode-rolling-upgrade/f678072/ : FAILURE in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b2a28b1dbeaf7157beb497d8c10a74aca2ef061f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 16:44:06.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"10f6151f432d5e94a2cfb70dc273c9c2db22fbed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 17:35:20.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/check/openstack-tox-docs/3cf6ce3/html/ : SUCCESS in 4m 45s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/check/swift-tox-py27/c88240b/ : SUCCESS in 28m 11s\n- swift-tox-py35 http://logs.openstack.org/50/601950/15/check/swift-tox-py35/a83af28/ : SUCCESS in 18m 16s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/15/check/swift-tox-py36/b17bdc9/ : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/check/swift-tox-py37/f43b300/ : SUCCESS in 9m 57s\n- swift-tox-func http://logs.openstack.org/50/601950/15/check/swift-tox-func/cb3feca/ : SUCCESS in 24m 18s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/check/swift-tox-func-encryption/64aae8e/ : SUCCESS in 20m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/check/swift-tox-func-domain-remap-staticweb/d1251a3/ : SUCCESS in 19m 07s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/check/swift-tox-func-ec/83603de/ : FAILURE in 11m 34s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api/ebd2533/ : SUCCESS in 12m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api-ceph-s3tests-tempauth/952b8ed/ : SUCCESS in 15m 07s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/check/swift-probetests-centos-7/1b53475/ : FAILURE in 30m 55s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional/279c0c9/ : SUCCESS in 24m 03s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional-ipv6/f16ada1/ : SUCCESS in 25m 07s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/check/swift-tox-lower-constraints/b62a45f/ : SUCCESS in 13m 38s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/check/openstack-tox-pep8/7adaaed/ : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/15/check/swift-multinode-rolling-upgrade/47586cc/ : FAILURE in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8db4b071517e4ba00345391581fbc03ee8ecf7b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 17:41:18.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"ddd51a79fc66c072598496735cb00d90fe968685","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:16:11.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/check/openstack-tox-docs/e516240/html/ : SUCCESS in 3m 51s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/check/swift-tox-py27/34af435/ : FAILURE in 12m 15s\n- swift-tox-py35 http://logs.openstack.org/50/601950/15/check/swift-tox-py35/8b6b603/ : FAILURE in 15m 29s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/15/check/swift-tox-py36/157faf1/ : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/check/swift-tox-py37/0fa92d4/ : SUCCESS in 15m 27s\n- swift-tox-func http://logs.openstack.org/50/601950/15/check/swift-tox-func/951fb06/ : SUCCESS in 9m 11s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/check/swift-tox-func-encryption/94196c8/ : SUCCESS in 16m 18s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/check/swift-tox-func-domain-remap-staticweb/bd79e57/ : SUCCESS in 8m 36s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/check/swift-tox-func-ec/bdfcf6e/ : SUCCESS in 8m 49s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api/6621b3d/ : SUCCESS in 9m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api-ceph-s3tests-tempauth/aa107c0/ : SUCCESS in 11m 55s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/check/swift-probetests-centos-7/d683905/ : SUCCESS in 29m 22s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional/5fa18c5/ : SUCCESS in 28m 20s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional-ipv6/8b233ab/ : SUCCESS in 26m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/check/swift-tox-lower-constraints/3befdef/ : SUCCESS in 12m 25s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/check/openstack-tox-pep8/9705d40/ : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/15/check/swift-multinode-rolling-upgrade/08b100b/ : FAILURE in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"82ce54bde94aee746fcb3ce1a66a54384d05070c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 18:16:42.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"d08adc56ddf451623ccd05a18c5dd353bc25734d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:56:19.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/check/openstack-tox-docs/a7478dd/html/ : SUCCESS in 4m 06s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/check/swift-tox-py27/d08d70b/ : SUCCESS in 24m 46s\n- swift-tox-py35 http://logs.openstack.org/50/601950/15/check/swift-tox-py35/81957ac/ : SUCCESS in 14m 03s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/15/check/swift-tox-py36/147dd7a/ : SUCCESS in 14m 33s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/check/swift-tox-py37/af03f23/ : SUCCESS in 20m 25s\n- swift-tox-func http://logs.openstack.org/50/601950/15/check/swift-tox-func/c31202e/ : SUCCESS in 11m 41s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/check/swift-tox-func-encryption/c6c79e8/ : SUCCESS in 8m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/check/swift-tox-func-domain-remap-staticweb/c7e193c/ : SUCCESS in 9m 40s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/check/swift-tox-func-ec/dddcae9/ : SUCCESS in 16m 14s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api/4da2916/ : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api-ceph-s3tests-tempauth/6522eeb/ : SUCCESS in 12m 28s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/check/swift-probetests-centos-7/97e7bc8/ : SUCCESS in 38m 56s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional/5fe2cb7/ : SUCCESS in 25m 38s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional-ipv6/77f8aeb/ : SUCCESS in 21m 31s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/check/swift-tox-lower-constraints/4d07fa7/ : SUCCESS in 16m 50s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/check/openstack-tox-pep8/dde65ad/ : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/15/check/swift-multinode-rolling-upgrade/d9983c5/ : FAILURE in 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2d08e6a8797d9c92baeb2bc3947f537f93ab6eab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:56:45.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"a28fa561aa89e297904de4750f1d632e5733be96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 20:32:20.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/gate/openstack-tox-docs/8462ca5/html/ : SUCCESS in 4m 23s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/gate/swift-tox-py27/74f03cb/ : SUCCESS in 12m 27s\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/gate/swift-tox-py37/cd444a1/ : SUCCESS in 14m 30s\n- swift-tox-func http://logs.openstack.org/50/601950/15/gate/swift-tox-func/85fcb9d/ : SUCCESS in 8m 17s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/gate/swift-tox-func-encryption/d9a12e6/ : SUCCESS in 15m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/gate/swift-tox-func-domain-remap-staticweb/e7272e5/ : SUCCESS in 38m 45s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/gate/swift-tox-func-ec/48a7cc5/ : SUCCESS in 17m 38s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/gate/swift-tox-func-s3api/b89d3d0/ : TIMED_OUT in 40m 44s\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/gate/swift-probetests-centos-7/f843051/ : SUCCESS in 31m 14s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/gate/swift-dsvm-functional/adf24a4/ : SUCCESS in 45m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/gate/swift-dsvm-functional-ipv6/a3a1af7/ : SUCCESS in 28m 32s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/gate/swift-tox-lower-constraints/a9e6f3e/ : SUCCESS in 11m 17s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/gate/openstack-tox-pep8/cf6a3db/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":15},{"id":"7d6db7a92a306bd85b261e085b9f13a076e8d7a4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 20:39:22.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"5f92ea6230f53c240f1ae97435a51d8917bbe24b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 21:14:53.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/check/openstack-tox-docs/bd3748e/html/ : SUCCESS in 4m 24s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/check/swift-tox-py27/62b231b/ : SUCCESS in 17m 38s\n- swift-tox-py35 http://logs.openstack.org/50/601950/15/check/swift-tox-py35/09e6847/ : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/50/601950/15/check/swift-tox-py36/73b0215/ : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/check/swift-tox-py37/0008a41/ : SUCCESS in 7m 31s\n- swift-tox-func http://logs.openstack.org/50/601950/15/check/swift-tox-func/cf45b64/ : SUCCESS in 15m 46s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/check/swift-tox-func-encryption/d4eccd2/ : SUCCESS in 8m 40s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/check/swift-tox-func-domain-remap-staticweb/44f6699/ : FAILURE in 15m 52s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/check/swift-tox-func-ec/0147df9/ : SUCCESS in 9m 39s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api/84ef1c3/ : SUCCESS in 9m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api-ceph-s3tests-tempauth/5b3a1f0/ : SUCCESS in 11m 23s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/check/swift-probetests-centos-7/e1b068f/ : SUCCESS in 30m 09s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional/5ea7396/ : SUCCESS in 25m 19s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional-ipv6/2e21d76/ : SUCCESS in 27m 21s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/check/swift-tox-lower-constraints/bbc0ab4/ : FAILURE in 15m 15s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/check/openstack-tox-pep8/60329e9/ : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/15/check/swift-multinode-rolling-upgrade/f2ae24e/ : FAILURE in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"be7cc215f22b540023d4d1049fdb87ee01fe686c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 23:03:57.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"bd7e92db27dacb5cb68206e3c584c55f8f172ed0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 23:36:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/check/openstack-tox-docs/ee2d2aa/html/ : SUCCESS in 4m 00s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/check/swift-tox-py27/ef87261/ : SUCCESS in 10m 10s\n- swift-tox-py36 http://logs.openstack.org/50/601950/15/check/swift-tox-py36/2b85fed/ : SUCCESS in 13m 58s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/check/swift-tox-py37/585acf6/ : SUCCESS in 8m 03s\n- swift-tox-func http://logs.openstack.org/50/601950/15/check/swift-tox-func/105133e/ : SUCCESS in 7m 52s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/check/swift-tox-func-encryption/cfebe74/ : SUCCESS in 8m 27s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/check/swift-tox-func-domain-remap-staticweb/baf1ac3/ : SUCCESS in 8m 45s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/check/swift-tox-func-ec/faf2592/ : SUCCESS in 9m 01s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api/a62de3e/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/50/601950/15/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ecd94fb/ : SUCCESS in 11m 34s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/check/swift-probetests-centos-7/21e697a/ : SUCCESS in 29m 32s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional/70d56e6/ : SUCCESS in 27m 42s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/check/swift-dsvm-functional-ipv6/220796e/ : SUCCESS in 26m 57s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/check/swift-tox-lower-constraints/18bceff/ : SUCCESS in 12m 30s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/check/openstack-tox-pep8/c18ac89/ : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/50/601950/15/check/swift-multinode-rolling-upgrade/e6ba2c9/ : FAILURE in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0a1397232a6f90d5481946c09a1ddebaf2328d5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 23:37:10.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ca717adc545582e870bf2b03ad0a9e4918f4db87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 00:25:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"7e70bac1f1db6d2b90fcbe50a7b1e2b17c216b60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 00:25:11.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/601950/15/gate/openstack-tox-docs/f2802c7/html/ : SUCCESS in 3m 55s\n- swift-tox-py27 http://logs.openstack.org/50/601950/15/gate/swift-tox-py27/4b0719e/ : SUCCESS in 32m 53s\n- swift-tox-py37 http://logs.openstack.org/50/601950/15/gate/swift-tox-py37/7d731e1/ : SUCCESS in 25m 47s\n- swift-tox-func http://logs.openstack.org/50/601950/15/gate/swift-tox-func/c115397/ : SUCCESS in 29m 26s\n- swift-tox-func-encryption http://logs.openstack.org/50/601950/15/gate/swift-tox-func-encryption/3c55886/ : SUCCESS in 26m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/50/601950/15/gate/swift-tox-func-domain-remap-staticweb/afbbb6c/ : SUCCESS in 29m 48s\n- swift-tox-func-ec http://logs.openstack.org/50/601950/15/gate/swift-tox-func-ec/496f3bf/ : SUCCESS in 30m 58s\n- swift-tox-func-s3api http://logs.openstack.org/50/601950/15/gate/swift-tox-func-s3api/01753b0/ : SUCCESS in 10m 11s\n- swift-probetests-centos-7 http://logs.openstack.org/50/601950/15/gate/swift-probetests-centos-7/c4ced43/ : SUCCESS in 28m 36s\n- swift-dsvm-functional http://logs.openstack.org/50/601950/15/gate/swift-dsvm-functional/caaaf54/ : SUCCESS in 25m 39s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/50/601950/15/gate/swift-dsvm-functional-ipv6/aba2f0f/ : SUCCESS in 37m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/50/601950/15/gate/swift-tox-lower-constraints/e392c20/ : SUCCESS in 11m 16s\n- openstack-tox-pep8 http://logs.openstack.org/50/601950/15/gate/openstack-tox-pep8/f0c96c4/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"443f029a584c5bb7dd7b910ddacd8fa562272dfb","revisions":{"6f581b544324f21a34929fcd0ca166e21e8a0487":{"kind":"REWORK","_number":1,"created":"2018-09-12 08:25:43.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/1"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 08:05:30.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nobject-expirer.conf can be used in the same manner as before.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f581b544324f21a34929fcd0ca166e21e8a0487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f581b544324f21a34929fcd0ca166e21e8a0487"}]},"branch":"refs/heads/master"},"7e7613ff9b337f1c6bb31c9be98da4a24ceb302e":{"kind":"REWORK","_number":2,"created":"2018-09-12 16:11:06.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/2"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 16:04:27.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf, only object-server.conf is used.\nObject-expirer.conf is used only when all object-server.conf\ndoesn\u0027t have \"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e7613ff9b337f1c6bb31c9be98da4a24ceb302e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e7613ff9b337f1c6bb31c9be98da4a24ceb302e"}]},"branch":"refs/heads/master"},"5d0e763c62902e25f2c37b7b63890d638f947641":{"kind":"REWORK","_number":3,"created":"2018-09-14 23:57:00.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/3"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-14 23:55:56.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf, only object-server.conf is used.\nObject-expirer.conf is used only when all object-server.conf\ndoesn\u0027t have \"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d0e763c62902e25f2c37b7b63890d638f947641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d0e763c62902e25f2c37b7b63890d638f947641"}]},"branch":"refs/heads/master"},"24db649ea44147a6dbb1f4365cc94309cef4f03c":{"kind":"REWORK","_number":4,"created":"2018-09-21 12:09:20.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/4"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-21 11:51:29.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24db649ea44147a6dbb1f4365cc94309cef4f03c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24db649ea44147a6dbb1f4365cc94309cef4f03c"}]},"branch":"refs/heads/master"},"ca59e1dabd65e9914942e6fa66665cd1b11be93d":{"kind":"REWORK","_number":5,"created":"2018-10-10 09:51:09.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/5"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-10-10 09:50:29.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca59e1dabd65e9914942e6fa66665cd1b11be93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca59e1dabd65e9914942e6fa66665cd1b11be93d"}]},"branch":"refs/heads/master"},"ff5efc0255132902aa88706df2d17046e0e8838f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-11-28 08:36:11.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/6"}}},"commit":{"parents":[{"commit":"e58dce1e4e088dfcf265bd0f2fc92a9046808c8f","subject":"Merge \"Use eventlet.patcher.original to get Python select module in get_hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e58dce1e4e088dfcf265bd0f2fc92a9046808c8f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-11-28 08:31:56.000000000","tz":540},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5efc0255132902aa88706df2d17046e0e8838f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5efc0255132902aa88706df2d17046e0e8838f"}]},"branch":"refs/heads/master"},"01c6820fb010768bfb32e40ee2cefd4a0463ad44":{"kind":"REWORK","_number":7,"created":"2018-12-26 02:41:48.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/7"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-26 02:25:50.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01c6820fb010768bfb32e40ee2cefd4a0463ad44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01c6820fb010768bfb32e40ee2cefd4a0463ad44"}]},"branch":"refs/heads/master"},"a2a8aed681fd2bbbe60ac8f3ab9c0a584ecd09cc":{"kind":"REWORK","_number":8,"created":"2018-12-26 04:29:53.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/8"}}},"commit":{"parents":[{"commit":"4bdc9b003a8ab1989105720ece04c1b3b516c7be","subject":"Merge \"Add IPv6 DSVM tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bdc9b003a8ab1989105720ece04c1b3b516c7be"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-26 04:29:27.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2a8aed681fd2bbbe60ac8f3ab9c0a584ecd09cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2a8aed681fd2bbbe60ac8f3ab9c0a584ecd09cc"}]},"branch":"refs/heads/master"},"f0d94b2dadf131ed1eb64efe07b59d801009fd55":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-12-26 04:52:14.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/9"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-26 04:51:30.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d94b2dadf131ed1eb64efe07b59d801009fd55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d94b2dadf131ed1eb64efe07b59d801009fd55"}]},"branch":"refs/heads/master"},"29b56d6d97b1e74bcf1e33867384d5efc2690f92":{"kind":"REWORK","_number":10,"created":"2018-12-27 09:52:58.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/10"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-27 09:49:13.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b56d6d97b1e74bcf1e33867384d5efc2690f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b56d6d97b1e74bcf1e33867384d5efc2690f92"}]},"branch":"refs/heads/master"},"99095807a919b8f8cb1e26a8f30a1f8f457894e1":{"kind":"REWORK","_number":11,"created":"2018-12-27 10:41:11.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/11"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-12-27 10:40:50.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99095807a919b8f8cb1e26a8f30a1f8f457894e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99095807a919b8f8cb1e26a8f30a1f8f457894e1"}]},"branch":"refs/heads/master"},"e0498217469b8640d5e409899ce75298db0a007a":{"kind":"REWORK","_number":12,"created":"2019-03-29 04:38:57.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/12"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2019-03-29 04:11:17.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\nThere are two differences between \"object-expirer.conf\" style and\n\"object-server.conf\" style.\n\nThe first difference is `dequeue_from_legacy` default value.\n`dequeue_from_legacy` defines task queue mode. In \"object-expirer.conf\"\nstyle, the default mode is legacy queue. In \"object-server.conf\" style,\nthe default mode is general queue. But general mode means no-op mode\nfor now, because general task queue is not implemented yet.\n\nThe second difference is internal client config. In \"object-expirer.conf\"\nstyle, config file of internal client is the object-expirer.conf itself.\nIn \"object-server.conf\" style, config file of internal client is\nanother file.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0498217469b8640d5e409899ce75298db0a007a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0498217469b8640d5e409899ce75298db0a007a"}]},"branch":"refs/heads/master"},"52e3336abf8485d3bfc267dd447b208fba1fb68f":{"kind":"REWORK","_number":13,"created":"2019-03-29 05:39:42.000000000","uploader":{"_account_id":14766,"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","username":"miyahara"},"ref":"refs/changes/50/601950/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/13"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2019-03-29 05:33:39.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\nThere are two differences between \"object-expirer.conf\" style and\n\"object-server.conf\" style.\n\nThe first difference is `dequeue_from_legacy` default value.\n`dequeue_from_legacy` defines task queue mode. In \"object-expirer.conf\"\nstyle, the default mode is legacy queue. In \"object-server.conf\" style,\nthe default mode is general queue. But general mode means no-op mode\nfor now, because general task queue is not implemented yet.\n\nThe second difference is internal client config. In \"object-expirer.conf\"\nstyle, config file of internal client is the object-expirer.conf itself.\nIn \"object-server.conf\" style, config file of internal client is\nanother file.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e3336abf8485d3bfc267dd447b208fba1fb68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e3336abf8485d3bfc267dd447b208fba1fb68f"}]},"branch":"refs/heads/master"},"3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-05-02 21:55:56.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/50/601950/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/14"}}},"commit":{"parents":[{"commit":"f55167a735632cdc89fb6f83f97e4fe3ed09b7cc","subject":"Merge \"Increase node_timeout in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f55167a735632cdc89fb6f83f97e4fe3ed09b7cc"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-05-02 21:55:56.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\nThere are two differences between \"object-expirer.conf\" style and\n\"object-server.conf\" style.\n\nThe first difference is `dequeue_from_legacy` default value.\n`dequeue_from_legacy` defines task queue mode. In \"object-expirer.conf\"\nstyle, the default mode is legacy queue. In \"object-server.conf\" style,\nthe default mode is general queue. But general mode means no-op mode\nfor now, because general task queue is not implemented yet.\n\nThe second difference is internal client config. In \"object-expirer.conf\"\nstyle, config file of internal client is the object-expirer.conf itself.\nIn \"object-server.conf\" style, config file of internal client is\nanother file.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269a3c1e903ec97fbf48e6b2bbe411fcdfe3401"}]},"branch":"refs/heads/master"},"443f029a584c5bb7dd7b910ddacd8fa562272dfb":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-05-04 15:45:02.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/50/601950/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/601950/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/601950/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/601950/15"}}},"commit":{"parents":[{"commit":"b8824f052b80b04b5d56ce339741e2de4af85f3a","subject":"Merge \"Fix ostestr invocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8824f052b80b04b5d56ce339741e2de4af85f3a"}]}],"author":{"name":"Kazuhiro MIYAHARA","email":"miyahara.kazuhiro@lab.ntt.co.jp","date":"2018-09-12 07:35:51.000000000","tz":0},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-05-04 15:45:02.000000000","tz":0},"subject":"Enable to configure object-expirer in object-server.conf","message":"Enable to configure object-expirer in object-server.conf\n\nTo prepare for object-expirer\u0027s general task queue feature [1],\nthis patch enables to configure object-expirer in object-server.conf.\nObject-expirer.conf can be used in the same manner as before, but deprecated.\n\nIf both of object-server.conf with \"object-expirer\" section and\nobject-expirer.conf are in a node, only object-server.conf is used.\nObject-expirer.conf is used only if all object-server.conf doesn\u0027t have\n\"object-expirer\" section.\n\nThere are two differences between \"object-expirer.conf\" style and\n\"object-server.conf\" style.\n\nThe first difference is `dequeue_from_legacy` default value.\n`dequeue_from_legacy` defines task queue mode. In \"object-expirer.conf\"\nstyle, the default mode is legacy queue. In \"object-server.conf\" style,\nthe default mode is general queue. But general mode means no-op mode\nfor now, because general task queue is not implemented yet.\n\nThe second difference is internal client config. In \"object-expirer.conf\"\nstyle, config file of internal client is the object-expirer.conf itself.\nIn \"object-server.conf\" style, config file of internal client is\nanother file.\n\n[1]: https://review.openstack.org/#/c/517389/\n\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: Ib21568f9b9d8547da87a99d65ae73a550e9c3230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/443f029a584c5bb7dd7b910ddacd8fa562272dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/443f029a584c5bb7dd7b910ddacd8fa562272dfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
