)]}'
{"id":"openstack%2Fswift~602205","triplet_id":"openstack%2Fswift~master~Iebc5cd4c22db159db3b26685e02b37a028eb2be6","project":"openstack/swift","branch":"master","topic":"pre-stage-update-commit","hashtags":[],"change_id":"Iebc5cd4c22db159db3b26685e02b37a028eb2be6","subject":"Refactor obj.server.ObjectController.PUT","status":"MERGED","created":"2018-09-12 23:31:53.000000000","updated":"2018-10-11 02:33:06.000000000","submitted":"2018-10-11 02:33:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"602205-1539225186837-a7f5251b","meta_rev_id":"50b0a3023218e6cf5723ba1e28b8a33ae9930c82","_number":602205,"virtual_id_number":602205,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2018-10-11 02:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-09-14 06:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-10-10 16:16:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-10 16:16:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2018-10-11 02:33:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c206e8591244845b26041111153e692eec49315e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-12 23:31:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8cb859b7fabc82f32a3f4a9957971c09e1be09d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 23:56:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/602205/1/check/openstack-tox-docs/82c2f53/html/ : SUCCESS in 4m 03s\n- swift-tox-py27 http://logs.openstack.org/05/602205/1/check/swift-tox-py27/fac8e28/ : FAILURE in 10m 27s\n- swift-tox-py35 http://logs.openstack.org/05/602205/1/check/swift-tox-py35/03804ca/ : SUCCESS in 4m 30s\n- swift-tox-py36 http://logs.openstack.org/05/602205/1/check/swift-tox-py36/536cb0e/ : SUCCESS in 4m 38s (non-voting)\n- swift-tox-func http://logs.openstack.org/05/602205/1/check/swift-tox-func/bf55f67/ : SUCCESS in 9m 14s\n- swift-tox-func-encryption http://logs.openstack.org/05/602205/1/check/swift-tox-func-encryption/6ab8789/ : SUCCESS in 12m 41s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/602205/1/check/swift-tox-func-domain-remap-staticweb/eca2f23/ : SUCCESS in 8m 34s\n- swift-tox-func-ec http://logs.openstack.org/05/602205/1/check/swift-tox-func-ec/915dda1/ : SUCCESS in 9m 15s\n- swift-tox-func-s3api http://logs.openstack.org/05/602205/1/check/swift-tox-func-s3api/69b9bf5/ : SUCCESS in 10m 15s\n- swift-probetests-centos-7 http://logs.openstack.org/05/602205/1/check/swift-probetests-centos-7/cd11bef/ : SUCCESS in 22m 54s\n- swift-dsvm-functional http://logs.openstack.org/05/602205/1/check/swift-dsvm-functional/701044d/ : SUCCESS in 21m 18s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/602205/1/check/swift-dsvm-functional-ipv6/848a09a/ : SUCCESS in 23m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/602205/1/check/openstack-tox-lower-constraints/f40de33/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/05/602205/1/check/openstack-tox-pep8/900ca79/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a148e3c0307f33022410a288e95f2e58b5540be3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-09-13 07:12:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(5 comments)\n\nI like the direction -- seems to make PUT much more readable. Test failure repros locally, though.","accounts_in_message":[],"_revision_number":1},{"id":"8f467c9ddae267f44191bab3b070b42218b71f24","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 13:58:56.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nThanks for the comments!  I don\u0027t think I quite grok the md5 bug/test-failure yet.","accounts_in_message":[],"_revision_number":1},{"id":"c987584d086ad81dacf1d92f9b7ba8329be8e9e9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 14:09:30.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nok, let me see if there\u0027s another way to go about it...","accounts_in_message":[],"_revision_number":1},{"id":"94615c430a09142f90368421346fdd163b159de2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 14:21:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cfa0356640e10d3f7424548b155a55a4399ddca","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 14:34:19.000000000","message":"Patch Set 1:\n\nAnother fix:\n\n\tdiff --git a/test/unit/proxy/test_server.py b/test/unit/proxy/test_server.py\n\tindex f53f87c85..8f48b1211 100644\n\t--- a/test/unit/proxy/test_server.py\n\t+++ b/test/unit/proxy/test_server.py\n\t@@ -5859,7 +5859,7 @@ class BaseTestECObjectController(BaseTestObjectController):\n\t\t     countdown[0] -\u003d 1\n\t\t     return hasher\n\t \n\t-        obj \u003d \u0027uvarovite-esurience-cerated-symphysic\u0027\n\t+        obj \u003d \u0027uvarovite-esurience-cerated-symphysic\u0027 * 10739\n\t\t prolis \u003d _test_sockets[0]\n\t\t prosrv \u003d _test_servers[0]\n\t\t sock \u003d connect_tcp((\u0027localhost\u0027, prolis.getsockname()[1]))","accounts_in_message":[],"_revision_number":1},{"id":"4a0cdef598134e1667f0e643ada86255270a25bb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 14:41:38.000000000","message":"Patch Set 1:\n\nSolve with coupling instead of timing?\n\n\tdiff --git a/test/unit/proxy/test_server.py b/test/unit/proxy/test_server.py\n\tindex f53f87c85..c2ca35bc9 100644\n\t--- a/test/unit/proxy/test_server.py\n\t+++ b/test/unit/proxy/test_server.py\n\t@@ -5854,6 +5854,10 @@ class BaseTestECObjectController(BaseTestObjectController):\n\t \n\t\t def busted_md5_constructor(initial_str\u003d\"\"):\n\t\t     hasher \u003d md5(initial_str)\n\t+            if initial_str:\n\t+                # footers hashers initialize with the entire json document in\n\t+                # the initializer, we don\u0027t need to mess with those.\n\t+                return hasher\n\t\t     if countdown[0] \u003e 0:\n\t\t\t hasher.update(\u0027wrong\u0027)\n\t\t     countdown[0] -\u003d 1","accounts_in_message":[],"_revision_number":1},{"id":"47a989cd01c41bd6d095f8c5b8cb94610e36735f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 16:55:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c9c61d796ede87743395fd429e336cd90cd372d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 17:29:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/602205/2/check/openstack-tox-docs/95d2bd2/html/ : SUCCESS in 4m 32s\n- swift-tox-py27 http://logs.openstack.org/05/602205/2/check/swift-tox-py27/9fc2d63/ : FAILURE in 9m 45s\n- swift-tox-py35 http://logs.openstack.org/05/602205/2/check/swift-tox-py35/b4b260e/ : SUCCESS in 4m 39s\n- swift-tox-py36 http://logs.openstack.org/05/602205/2/check/swift-tox-py36/9909522/ : SUCCESS in 5m 02s (non-voting)\n- swift-tox-func http://logs.openstack.org/05/602205/2/check/swift-tox-func/ce22f88/ : SUCCESS in 9m 32s\n- swift-tox-func-encryption http://logs.openstack.org/05/602205/2/check/swift-tox-func-encryption/a8f6106/ : SUCCESS in 11m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/602205/2/check/swift-tox-func-domain-remap-staticweb/ac27812/ : SUCCESS in 8m 32s\n- swift-tox-func-ec http://logs.openstack.org/05/602205/2/check/swift-tox-func-ec/b0e747b/ : SUCCESS in 10m 13s\n- swift-tox-func-s3api http://logs.openstack.org/05/602205/2/check/swift-tox-func-s3api/aa193fe/ : SUCCESS in 13m 45s\n- swift-probetests-centos-7 http://logs.openstack.org/05/602205/2/check/swift-probetests-centos-7/73464de/ : SUCCESS in 28m 42s\n- swift-dsvm-functional http://logs.openstack.org/05/602205/2/check/swift-dsvm-functional/28e054f/ : SUCCESS in 21m 42s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/602205/2/check/swift-dsvm-functional-ipv6/6145786/ : SUCCESS in 30m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/602205/2/check/openstack-tox-lower-constraints/3cf66b1/ : SUCCESS in 8m 18s\n- openstack-tox-pep8 http://logs.openstack.org/05/602205/2/check/openstack-tox-pep8/3a5b8a2/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f4758c3fc8a926cd706043274fab3ce70fc4cdfc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 18:12:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"51619f73dab00454b3da441e8d74d935acc0b6b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 18:46:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/602205/3/check/openstack-tox-docs/bb1db3e/html/ : SUCCESS in 3m 59s\n- swift-tox-py27 http://logs.openstack.org/05/602205/3/check/swift-tox-py27/56d2d0d/ : SUCCESS in 9m 56s\n- swift-tox-py35 http://logs.openstack.org/05/602205/3/check/swift-tox-py35/4799834/ : SUCCESS in 5m 20s\n- swift-tox-py36 http://logs.openstack.org/05/602205/3/check/swift-tox-py36/a98b7eb/ : SUCCESS in 5m 46s (non-voting)\n- swift-tox-func http://logs.openstack.org/05/602205/3/check/swift-tox-func/cb22549/ : SUCCESS in 8m 37s\n- swift-tox-func-encryption http://logs.openstack.org/05/602205/3/check/swift-tox-func-encryption/2c59706/ : SUCCESS in 9m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/602205/3/check/swift-tox-func-domain-remap-staticweb/894a56d/ : SUCCESS in 9m 26s\n- swift-tox-func-ec http://logs.openstack.org/05/602205/3/check/swift-tox-func-ec/537e425/ : SUCCESS in 17m 47s\n- swift-tox-func-s3api http://logs.openstack.org/05/602205/3/check/swift-tox-func-s3api/be4f617/ : SUCCESS in 10m 23s\n- swift-probetests-centos-7 http://logs.openstack.org/05/602205/3/check/swift-probetests-centos-7/a5c2197/ : SUCCESS in 25m 31s\n- swift-dsvm-functional http://logs.openstack.org/05/602205/3/check/swift-dsvm-functional/7412566/ : SUCCESS in 24m 06s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/602205/3/check/swift-dsvm-functional-ipv6/cfc5233/ : SUCCESS in 25m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/602205/3/check/openstack-tox-lower-constraints/56eb61c/ : SUCCESS in 6m 44s\n- openstack-tox-pep8 http://logs.openstack.org/05/602205/3/check/openstack-tox-pep8/79abbb7/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ed211ea54baae3069512c293b4e7fddfa306d6ad","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2018-09-13 21:45:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0781ee904f9d1505e512eb78c0fc25b6ed9ba76b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 22:18:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/602205/4/check/openstack-tox-docs/c360b8d/html/ : SUCCESS in 4m 38s\n- swift-tox-py27 http://logs.openstack.org/05/602205/4/check/swift-tox-py27/7425f95/ : SUCCESS in 11m 36s\n- swift-tox-py35 http://logs.openstack.org/05/602205/4/check/swift-tox-py35/fd0e05a/ : SUCCESS in 4m 50s\n- swift-tox-py36 http://logs.openstack.org/05/602205/4/check/swift-tox-py36/9f64795/ : SUCCESS in 5m 19s (non-voting)\n- swift-tox-func http://logs.openstack.org/05/602205/4/check/swift-tox-func/d042900/ : SUCCESS in 7m 52s\n- swift-tox-func-encryption http://logs.openstack.org/05/602205/4/check/swift-tox-func-encryption/a56a911/ : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/602205/4/check/swift-tox-func-domain-remap-staticweb/86dd09d/ : SUCCESS in 8m 34s\n- swift-tox-func-ec http://logs.openstack.org/05/602205/4/check/swift-tox-func-ec/39ddae9/ : SUCCESS in 10m 16s\n- swift-tox-func-s3api http://logs.openstack.org/05/602205/4/check/swift-tox-func-s3api/8abc5df/ : SUCCESS in 11m 45s\n- swift-probetests-centos-7 http://logs.openstack.org/05/602205/4/check/swift-probetests-centos-7/64ac429/ : SUCCESS in 28m 16s\n- swift-dsvm-functional http://logs.openstack.org/05/602205/4/check/swift-dsvm-functional/a31e921/ : SUCCESS in 23m 28s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/602205/4/check/swift-dsvm-functional-ipv6/f174461/ : SUCCESS in 29m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/602205/4/check/openstack-tox-lower-constraints/efa8230/ : SUCCESS in 8m 28s\n- openstack-tox-pep8 http://logs.openstack.org/05/602205/4/check/openstack-tox-pep8/92605b4/ : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"bd5534fb297fd968c1b7157cd6a55c58d4df7740","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-09-14 06:58:08.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)\n\nI\u0027ll let Pete take a look tomorrow -- I think it all makes sense, though?","accounts_in_message":[],"_revision_number":4},{"id":"5d01c8de251e5f6af23778e95cb4e849828f7f03","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-10 16:16:28.000000000","message":"Patch Set 4: Workflow+1\n\nYeah, I feel like breaking up PUT is a laudable goal in and of itself.","accounts_in_message":[],"_revision_number":4},{"id":"ace0db67a3dec2fdfc0ca98f4a0450bcf98301d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 16:16:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"24e34c475fde043f3b64b72affda0af40f368d4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 20:04:07.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/602205/4/gate/openstack-tox-docs/9003213/html/ : SUCCESS in 5m 00s\n- swift-tox-py27 http://logs.openstack.org/05/602205/4/gate/swift-tox-py27/40a31d2/ : SUCCESS in 11m 47s\n- swift-tox-py35 http://logs.openstack.org/05/602205/4/gate/swift-tox-py35/74928a1/ : SUCCESS in 4m 52s\n- swift-tox-func http://logs.openstack.org/05/602205/4/gate/swift-tox-func/480a9cd/ : SUCCESS in 9m 52s\n- swift-tox-func-encryption http://logs.openstack.org/05/602205/4/gate/swift-tox-func-encryption/b3d6d32/ : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/602205/4/gate/swift-tox-func-domain-remap-staticweb/2583cba/ : SUCCESS in 8m 32s\n- swift-tox-func-ec http://logs.openstack.org/05/602205/4/gate/swift-tox-func-ec/2901570/ : SUCCESS in 10m 03s\n- swift-tox-func-s3api http://logs.openstack.org/05/602205/4/gate/swift-tox-func-s3api/56d3ade/ : SUCCESS in 9m 30s\n- swift-probetests-centos-7 http://logs.openstack.org/05/602205/4/gate/swift-probetests-centos-7/dd032f1/ : FAILURE in 32m 49s\n- swift-dsvm-functional http://logs.openstack.org/05/602205/4/gate/swift-dsvm-functional/5e1f7e5/ : SUCCESS in 22m 33s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/602205/4/gate/swift-dsvm-functional-ipv6/19e6959/ : SUCCESS in 28m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/05/602205/4/gate/swift-tox-lower-constraints/184f708/ : SUCCESS in 11m 03s\n- openstack-tox-pep8 http://logs.openstack.org/05/602205/4/gate/openstack-tox-pep8/2a23a66/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":4},{"id":"ff8880b3f0f79270313eddd787b79adb8de055fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-10 20:10:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"929807c0b711b7fda0f39b89438a95e7e5a82561","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 21:25:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/05/602205/4/check/openstack-tox-cover/2e813b4/cover/ : SUCCESS in 8m 25s\n- openstack-tox-docs http://logs.openstack.org/05/602205/4/check/openstack-tox-docs/0e8369c/html/ : SUCCESS in 4m 20s\n- swift-tox-py27 http://logs.openstack.org/05/602205/4/check/swift-tox-py27/bd3287a/ : SUCCESS in 11m 42s\n- swift-tox-py35 http://logs.openstack.org/05/602205/4/check/swift-tox-py35/49bcfc5/ : SUCCESS in 4m 43s\n- swift-tox-py36 http://logs.openstack.org/05/602205/4/check/swift-tox-py36/732c03c/ : SUCCESS in 4m 58s (non-voting)\n- swift-tox-func http://logs.openstack.org/05/602205/4/check/swift-tox-func/5cfff1b/ : SUCCESS in 8m 23s\n- swift-tox-func-encryption http://logs.openstack.org/05/602205/4/check/swift-tox-func-encryption/1fa71b2/ : SUCCESS in 10m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/602205/4/check/swift-tox-func-domain-remap-staticweb/3c802b9/ : SUCCESS in 9m 06s\n- swift-tox-func-ec http://logs.openstack.org/05/602205/4/check/swift-tox-func-ec/887933c/ : SUCCESS in 9m 09s\n- swift-tox-func-s3api http://logs.openstack.org/05/602205/4/check/swift-tox-func-s3api/0ed24b0/ : SUCCESS in 9m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/602205/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/98ccf38/ : SUCCESS in 9m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/602205/4/check/swift-probetests-centos-7/dde16da/ : SUCCESS in 27m 05s\n- swift-dsvm-functional http://logs.openstack.org/05/602205/4/check/swift-dsvm-functional/1bdf3c5/ : SUCCESS in 23m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/602205/4/check/swift-dsvm-functional-ipv6/4ac23fa/ : SUCCESS in 29m 12s\n- swift-tox-lower-constraints http://logs.openstack.org/05/602205/4/check/swift-tox-lower-constraints/4ba79f4/ : SUCCESS in 10m 52s\n- openstack-tox-pep8 http://logs.openstack.org/05/602205/4/check/openstack-tox-pep8/253fa7e/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/602205/4/check/swift-multinode-rolling-upgrade/f1dda34/ : SUCCESS in 16m 42s","accounts_in_message":[],"_revision_number":4},{"id":"3aad2e3849a6340d01b6fbf851fc07a75190ec13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 21:25:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b922c31c7328643e85548d98aefd27c8882a957b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-11 02:33:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/602205/4/gate/openstack-tox-docs/7dac30c/html/ : SUCCESS in 4m 32s\n- swift-tox-py27 http://logs.openstack.org/05/602205/4/gate/swift-tox-py27/7eedaf1/ : SUCCESS in 11m 56s\n- swift-tox-py35 http://logs.openstack.org/05/602205/4/gate/swift-tox-py35/fbd741b/ : SUCCESS in 5m 05s\n- swift-tox-func http://logs.openstack.org/05/602205/4/gate/swift-tox-func/ef92536/ : SUCCESS in 9m 39s\n- swift-tox-func-encryption http://logs.openstack.org/05/602205/4/gate/swift-tox-func-encryption/be6d5c1/ : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/602205/4/gate/swift-tox-func-domain-remap-staticweb/0550b1c/ : SUCCESS in 9m 52s\n- swift-tox-func-ec http://logs.openstack.org/05/602205/4/gate/swift-tox-func-ec/884ebeb/ : SUCCESS in 9m 11s\n- swift-tox-func-s3api http://logs.openstack.org/05/602205/4/gate/swift-tox-func-s3api/a1c3a20/ : SUCCESS in 10m 20s\n- swift-probetests-centos-7 http://logs.openstack.org/05/602205/4/gate/swift-probetests-centos-7/66c56d0/ : SUCCESS in 28m 36s\n- swift-dsvm-functional http://logs.openstack.org/05/602205/4/gate/swift-dsvm-functional/f40e13c/ : SUCCESS in 24m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/602205/4/gate/swift-dsvm-functional-ipv6/0755fca/ : SUCCESS in 24m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/05/602205/4/gate/swift-tox-lower-constraints/64bbebb/ : SUCCESS in 11m 50s\n- openstack-tox-pep8 http://logs.openstack.org/05/602205/4/gate/openstack-tox-pep8/2018f77/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":4},{"id":"c964b52776370b64fc37268f12c777ef9ef53ed7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-11 02:33:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cbfa585d3b48d2d8e2aedb9cedc62226dade8798","revisions":{"9865b22c871ff5e79e4b0ec71f471267feb4692b":{"kind":"REWORK","_number":1,"created":"2018-09-12 23:31:53.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/602205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/602205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/602205/1"}}},"commit":{"parents":[{"commit":"33c76507531073e431148a4f65da0a40e5ba88eb","subject":"Add writer method to DiskFile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33c76507531073e431148a4f65da0a40e5ba88eb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-12 20:04:14.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-12 23:30:07.000000000","tz":-300},"subject":"Refactor obj.server.ObjectController.PUT","message":"Refactor obj.server.ObjectController.PUT\n\nChange-Id: Iebc5cd4c22db159db3b26685e02b37a028eb2be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9865b22c871ff5e79e4b0ec71f471267feb4692b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9865b22c871ff5e79e4b0ec71f471267feb4692b"}]},"branch":"refs/heads/master"},"46baa5b601ed091cfe22e4208cdd41c0f5423fb7":{"kind":"REWORK","_number":2,"created":"2018-09-13 16:55:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/602205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/602205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/602205/2"}}},"commit":{"parents":[{"commit":"1c8d8aeb72ad25008ad71576e2785183577b1895","subject":"Add a chunks_finished to BaseDiskFileWriter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c8d8aeb72ad25008ad71576e2785183577b1895"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-12 20:04:14.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-13 16:54:46.000000000","tz":-300},"subject":"Refactor obj.server.ObjectController.PUT","message":"Refactor obj.server.ObjectController.PUT\n\nChange-Id: Iebc5cd4c22db159db3b26685e02b37a028eb2be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46baa5b601ed091cfe22e4208cdd41c0f5423fb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46baa5b601ed091cfe22e4208cdd41c0f5423fb7"}]},"branch":"refs/heads/master"},"0ac03f0b6ef80df9bb8834a48cad8d49f32b215a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-09-13 18:12:39.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/602205/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/602205/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/602205/3"}}},"commit":{"parents":[{"commit":"52ecbf95399a7e4c0d062af8d885ef7ce684eda4","subject":"Add a chunks_finished to BaseDiskFileWriter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52ecbf95399a7e4c0d062af8d885ef7ce684eda4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-12 20:04:14.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-13 18:12:19.000000000","tz":-300},"subject":"Refactor obj.server.ObjectController.PUT","message":"Refactor obj.server.ObjectController.PUT\n\nChange-Id: Iebc5cd4c22db159db3b26685e02b37a028eb2be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ac03f0b6ef80df9bb8834a48cad8d49f32b215a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ac03f0b6ef80df9bb8834a48cad8d49f32b215a"}]},"branch":"refs/heads/master"},"cbfa585d3b48d2d8e2aedb9cedc62226dade8798":{"kind":"REWORK","_number":4,"created":"2018-09-13 21:45:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/602205/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/602205/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/602205/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/602205/4"}}},"commit":{"parents":[{"commit":"52ecbf95399a7e4c0d062af8d885ef7ce684eda4","subject":"Add a chunks_finished to BaseDiskFileWriter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52ecbf95399a7e4c0d062af8d885ef7ce684eda4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-12 20:04:14.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2018-09-13 21:42:20.000000000","tz":-300},"subject":"Refactor obj.server.ObjectController.PUT","message":"Refactor obj.server.ObjectController.PUT\n\nChange-Id: Iebc5cd4c22db159db3b26685e02b37a028eb2be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbfa585d3b48d2d8e2aedb9cedc62226dade8798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbfa585d3b48d2d8e2aedb9cedc62226dade8798"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
