)]}'
{"id":"openstack%2Fswift~609168","triplet_id":"openstack%2Fswift~master~I7c251a758dfc1fedb3fb61e351de305b431afa79","project":"openstack/swift","branch":"master","topic":"591862","hashtags":[],"change_id":"I7c251a758dfc1fedb3fb61e351de305b431afa79","subject":"We don\u0027t have to keep the retrieved token anymore","status":"MERGED","created":"2018-10-09 23:51:44.000000000","updated":"2019-09-20 01:53:11.000000000","submitted":"2019-09-20 01:50:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"609168-1568944224051-ab427080","meta_rev_id":"156fc1bd326706a1e1bc6dc910202f28d9bb4d52","_number":609168,"virtual_id_number":609168,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":2,"date":"2019-09-20 01:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"all":[{"value":2,"date":"2019-05-14 05:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-20 00:21:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 05:52:25.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2019-09-20 01:50:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4abfbcd2fee7c8d14d654d7bc72ea23d432675e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-09 23:51:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eda822b783b6304df0865f20229f83112d51dd66","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-09 23:52:17.000000000","message":"Patch Set 1:\n\nPorted from swift3 patch https://review.openstack.org/#/c/430145/","accounts_in_message":[],"_revision_number":1},{"id":"fd3423e9d46fd108ff2d53301c0d556b4c725e79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 00:30:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/609168/1/check/openstack-tox-cover/1286f3b/cover/ : SUCCESS in 8m 06s\n- openstack-tox-docs http://logs.openstack.org/68/609168/1/check/openstack-tox-docs/5c26beb/html/ : SUCCESS in 4m 26s\n- swift-tox-py27 http://logs.openstack.org/68/609168/1/check/swift-tox-py27/dd088e1/ : SUCCESS in 10m 26s\n- swift-tox-py35 http://logs.openstack.org/68/609168/1/check/swift-tox-py35/f394323/ : SUCCESS in 4m 15s\n- swift-tox-py36 http://logs.openstack.org/68/609168/1/check/swift-tox-py36/b50bfef/ : SUCCESS in 4m 18s (non-voting)\n- swift-tox-func http://logs.openstack.org/68/609168/1/check/swift-tox-func/57add82/ : SUCCESS in 7m 47s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/1/check/swift-tox-func-encryption/c29aeaa/ : SUCCESS in 10m 52s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/1/check/swift-tox-func-domain-remap-staticweb/5d7943a/ : SUCCESS in 10m 04s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/1/check/swift-tox-func-ec/8d82992/ : SUCCESS in 9m 43s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/1/check/swift-tox-func-s3api/f4c8595/ : SUCCESS in 11m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/019abde/ : SUCCESS in 10m 36s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/1/check/swift-probetests-centos-7/f8b7005/ : FAILURE in 26m 58s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/1/check/swift-dsvm-functional/cb4e6b7/ : SUCCESS in 25m 59s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/1/check/swift-dsvm-functional-ipv6/442ec44/ : SUCCESS in 31m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/1/check/swift-tox-lower-constraints/0cb1257/ : SUCCESS in 10m 09s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/1/check/openstack-tox-pep8/e98b145/ : FAILURE in 4m 39s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/1/check/swift-multinode-rolling-upgrade/c1eaf93/ : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"dc23432007bae305c90240dc34ce3a221d7a8690","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-10 00:41:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"10acbcf15546e9c5e8ab308af38c90c89606e511","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 01:13:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/609168/2/check/openstack-tox-cover/63ebaac/cover/ : SUCCESS in 8m 14s\n- openstack-tox-docs http://logs.openstack.org/68/609168/2/check/openstack-tox-docs/94637a5/html/ : SUCCESS in 4m 06s\n- swift-tox-py27 http://logs.openstack.org/68/609168/2/check/swift-tox-py27/382daa6/ : SUCCESS in 12m 20s\n- swift-tox-py35 http://logs.openstack.org/68/609168/2/check/swift-tox-py35/7ee7965/ : SUCCESS in 4m 42s\n- swift-tox-py36 http://logs.openstack.org/68/609168/2/check/swift-tox-py36/44397bc/ : SUCCESS in 4m 35s (non-voting)\n- swift-tox-func http://logs.openstack.org/68/609168/2/check/swift-tox-func/58c425b/ : SUCCESS in 10m 22s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/2/check/swift-tox-func-encryption/8c78602/ : SUCCESS in 9m 53s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/2/check/swift-tox-func-domain-remap-staticweb/8ac7db8/ : SUCCESS in 7m 58s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/2/check/swift-tox-func-ec/bb6dc4c/ : SUCCESS in 10m 33s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/2/check/swift-tox-func-s3api/cb02cc9/ : SUCCESS in 15m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/038291a/ : SUCCESS in 9m 42s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/2/check/swift-probetests-centos-7/5210540/ : SUCCESS in 24m 44s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/2/check/swift-dsvm-functional/98fd738/ : SUCCESS in 26m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/2/check/swift-dsvm-functional-ipv6/61937d9/ : SUCCESS in 24m 16s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/2/check/swift-tox-lower-constraints/0e6fc61/ : SUCCESS in 12m 08s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/2/check/openstack-tox-pep8/b5d68f2/ : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/2/check/swift-multinode-rolling-upgrade/e4382df/ : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8c58f2562aaff1a1a31b4acc1c536fbb8663a02d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-20 00:09:47.000000000","message":"Patch Set 2: Code-Review+2\n\nI dropped test_s3api_with_s3_token_no_pass_token_to_auth_token when I ported it over, since it didn\u0027t provide a useful distinction from test_s3api_with_s3_token_and_auth_token, but this seems good to me. Maybe I ought to let someone else spot-check, since I was the one to pull it over from swift3?","accounts_in_message":[],"_revision_number":2},{"id":"8f4cefb381eff101a105235a1d73f43d3f57231c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-12-12 00:16:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fb39b2a530e4b0a8a8e14110fbb79919ae9f88d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-12-12 00:18:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"302ed09f1c9b2de313d024a376ecf237797e7a31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 00:50:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/609168/3/check/openstack-tox-docs/09546c0/html/ : SUCCESS in 4m 04s\n- swift-tox-py27 http://logs.openstack.org/68/609168/3/check/swift-tox-py27/6b8d44c/ : SUCCESS in 10m 22s\n- swift-tox-py35 http://logs.openstack.org/68/609168/3/check/swift-tox-py35/a85dc31/ : SUCCESS in 5m 54s\n- swift-tox-py36 http://logs.openstack.org/68/609168/3/check/swift-tox-py36/b25f529/ : SUCCESS in 5m 27s (non-voting)\n- swift-tox-func http://logs.openstack.org/68/609168/3/check/swift-tox-func/79ec3cc/ : SUCCESS in 8m 50s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/3/check/swift-tox-func-encryption/de3af2c/ : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/3/check/swift-tox-func-domain-remap-staticweb/14a230e/ : SUCCESS in 9m 17s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/3/check/swift-tox-func-ec/d75f4da/ : SUCCESS in 9m 48s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/3/check/swift-tox-func-s3api/0760a3b/ : SUCCESS in 11m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/334559b/ : SUCCESS in 12m 35s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/3/check/swift-probetests-centos-7/74bfe38/ : SUCCESS in 29m 35s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/3/check/swift-dsvm-functional/84e2d84/ : SUCCESS in 27m 12s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/3/check/swift-dsvm-functional-ipv6/d24ed06/ : SUCCESS in 27m 32s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/3/check/swift-tox-lower-constraints/adf4c43/ : SUCCESS in 10m 20s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/3/check/openstack-tox-pep8/e94dc03/ : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/3/check/swift-multinode-rolling-upgrade/b5bfc50/ : SUCCESS in 16m 38s","accounts_in_message":[],"_revision_number":3},{"id":"20576646fc5a0e95b2c22e085f3ba85e3d42c5f8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-03-26 21:10:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f6159d5180a0948bf870cb0289c58a8d42858ce7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-03-26 21:10:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7c991aba6236e96c4518ece92e257a3d8ada685","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 00:05:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/609168/4/check/openstack-tox-docs/7faf491/html/ : SUCCESS in 3m 56s\n- swift-tox-py27 http://logs.openstack.org/68/609168/4/check/swift-tox-py27/59e8599/ : SUCCESS in 12m 55s\n- swift-tox-py35 http://logs.openstack.org/68/609168/4/check/swift-tox-py35/3e01b0a/ : SUCCESS in 9m 17s\n- swift-tox-py36 http://logs.openstack.org/68/609168/4/check/swift-tox-py36/49f9957/ : SUCCESS in 9m 18s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/68/609168/4/check/swift-tox-py37/0468392/ : SUCCESS in 8m 23s (non-voting)\n- swift-tox-func http://logs.openstack.org/68/609168/4/check/swift-tox-func/e91e7dd/ : SUCCESS in 10m 22s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/4/check/swift-tox-func-encryption/82683c9/ : SUCCESS in 9m 23s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/4/check/swift-tox-func-domain-remap-staticweb/5a8a42e/ : SUCCESS in 8m 53s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/4/check/swift-tox-func-ec/ce87098/ : SUCCESS in 9m 43s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/4/check/swift-tox-func-s3api/0fa4fa2/ : SUCCESS in 11m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/765331c/ : SUCCESS in 12m 02s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/4/check/swift-probetests-centos-7/0e3d9da/ : SUCCESS in 29m 24s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/4/check/swift-dsvm-functional/940f1c7/ : SUCCESS in 30m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/4/check/swift-dsvm-functional-ipv6/b985e62/ : SUCCESS in 27m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/4/check/swift-tox-lower-constraints/bb62831/ : SUCCESS in 10m 52s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/4/check/openstack-tox-pep8/0280ebc/ : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/4/check/swift-multinode-rolling-upgrade/341537d/ : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":4},{"id":"5cb703778d4374ba54b48a8dfe57b05cc8d63142","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-03 23:12:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nChange looks good. Small questions still exist. Please checkout them.","accounts_in_message":[],"_revision_number":4},{"id":"03261a90c8fa3727667ffa1b707147786133c6ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 23:42:48.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"39785dac7b5f5d8879a9c77583bc257b7eb3b0bd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 23:42:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f7a25c8ebac8dca937a1026fd0203c778dd3d715","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 00:35:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/609168/5/check/openstack-tox-docs/799e726/html/ : SUCCESS in 4m 11s\n- swift-tox-py27 http://logs.openstack.org/68/609168/5/check/swift-tox-py27/3ed4b9b/ : SUCCESS in 10m 56s\n- swift-tox-py35 http://logs.openstack.org/68/609168/5/check/swift-tox-py35/229cce5/ : SUCCESS in 8m 19s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/68/609168/5/check/swift-tox-py36/55d4c32/ : SUCCESS in 10m 09s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/68/609168/5/check/swift-tox-py37/9f0f9bc/ : SUCCESS in 8m 52s\n- swift-tox-func http://logs.openstack.org/68/609168/5/check/swift-tox-func/eb053cf/ : SUCCESS in 16m 26s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/5/check/swift-tox-func-encryption/5df1eb9/ : SUCCESS in 9m 16s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/5/check/swift-tox-func-domain-remap-staticweb/c8ef600/ : SUCCESS in 21m 08s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/5/check/swift-tox-func-ec/a1b536f/ : SUCCESS in 10m 02s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/5/check/swift-tox-func-s3api/7aade95/ : SUCCESS in 20m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/aeed4b9/ : SUCCESS in 12m 03s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/5/check/swift-probetests-centos-7/ec60d0a/ : SUCCESS in 31m 41s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/5/check/swift-dsvm-functional/5cd8106/ : SUCCESS in 28m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/5/check/swift-dsvm-functional-ipv6/7bc812c/ : SUCCESS in 23m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/5/check/swift-tox-lower-constraints/193bb08/ : SUCCESS in 12m 51s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/5/check/openstack-tox-pep8/b706edc/ : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/5/check/swift-multinode-rolling-upgrade/cb78a85/ : SUCCESS in 21m 41s","accounts_in_message":[],"_revision_number":5},{"id":"d4cbe72fb68d1213d140c8221c6f96f93ee023ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-07 18:30:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5b21c0ed5863639846138b452438395e09c71255","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 20:09:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/68/609168/6/check/openstack-tox-docs/43f6da0/html/ : SUCCESS in 4m 10s\n- swift-tox-py27 http://logs.openstack.org/68/609168/6/check/swift-tox-py27/87926bc/ : FAILURE in 10m 55s\n- swift-tox-py36 http://logs.openstack.org/68/609168/6/check/swift-tox-py36/26a3828/ : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/68/609168/6/check/swift-tox-py37/1819a52/ : FAILURE in 22m 04s\n- swift-tox-func http://logs.openstack.org/68/609168/6/check/swift-tox-func/a1eb91e/ : SUCCESS in 8m 11s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/6/check/swift-tox-func-encryption/b099cd9/ : SUCCESS in 9m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/6/check/swift-tox-func-domain-remap-staticweb/a17bd6b/ : SUCCESS in 8m 39s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/6/check/swift-tox-func-ec/d5ecfeb/ : SUCCESS in 8m 55s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/6/check/swift-tox-func-s3api/c2b7e44/ : TIMED_OUT in 40m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/6/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0c062c9/ : TIMED_OUT in 31m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/6/check/swift-probetests-centos-7/fb546dc/ : SUCCESS in 31m 31s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/6/check/swift-dsvm-functional/feafa60/ : SUCCESS in 27m 06s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/6/check/swift-dsvm-functional-ipv6/5a0fa22/ : SUCCESS in 35m 07s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/6/check/swift-tox-lower-constraints/eb6e6cf/ : FAILURE in 26m 29s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/6/check/openstack-tox-pep8/dfe23e2/ : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/6/check/swift-multinode-rolling-upgrade/b4d49c0/ : FAILURE in 8m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c009355d440a211b92f6bd20f9f806cfe7a23336","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 18:21:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a2bceb92691970b5d459c864e7d830a220fa199d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 21:17:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/68/609168/7/check/openstack-tox-docs/669ed85/html/ : SUCCESS in 3m 59s\n- swift-build-image http://logs.openstack.org/68/609168/7/check/swift-build-image/8cc5256/ : SUCCESS in 8m 03s\n- swift-tox-py27 http://logs.openstack.org/68/609168/7/check/swift-tox-py27/26dcd1e/ : SUCCESS in 10m 00s\n- swift-tox-py36 http://logs.openstack.org/68/609168/7/check/swift-tox-py36/e251be4/ : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/68/609168/7/check/swift-tox-py37/b5a4da5/ : SUCCESS in 8m 13s\n- swift-tox-func http://logs.openstack.org/68/609168/7/check/swift-tox-func/d913df8/ : SUCCESS in 9m 59s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/7/check/swift-tox-func-encryption/c1c1d5c/ : SUCCESS in 9m 19s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/7/check/swift-tox-func-domain-remap-staticweb/98a6d43/ : SUCCESS in 9m 12s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/7/check/swift-tox-func-ec/d87aae2/ : SUCCESS in 8m 44s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/7/check/swift-tox-func-s3api/9e2451f/ : SUCCESS in 9m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/7/check/swift-tox-func-s3api-ceph-s3tests-tempauth/5e49254/ : SUCCESS in 12m 25s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/7/check/swift-probetests-centos-7/de6b0c5/ : FAILURE in 8m 53s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/7/check/swift-dsvm-functional/a786157/ : SUCCESS in 26m 29s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/7/check/swift-dsvm-functional-ipv6/96d0bbe/ : SUCCESS in 29m 45s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/7/check/swift-tox-lower-constraints/68924c0/ : SUCCESS in 14m 20s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/7/check/openstack-tox-pep8/f980ad6/ : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/7/check/swift-multinode-rolling-upgrade/d954ec5/ : FAILURE in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1403581615f000a6e9d8c3e20669935bfe0fd4c4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 21:21:45.000000000","message":"Patch Set 7:\n\nrecheck\n\nNo, this did not affect container DB locking.","accounts_in_message":[],"_revision_number":7},{"id":"85e979224a26d370c44336879bff64e6c1f689ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 22:37:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/609168/7/check/openstack-tox-docs/22393c0/html/ : SUCCESS in 4m 09s\n- swift-build-image http://logs.openstack.org/68/609168/7/check/swift-build-image/0af29c6/ : SUCCESS in 8m 34s\n- swift-tox-py27 http://logs.openstack.org/68/609168/7/check/swift-tox-py27/9f335ee/ : SUCCESS in 9m 39s\n- swift-tox-py36 http://logs.openstack.org/68/609168/7/check/swift-tox-py36/88f4064/ : SUCCESS in 8m 12s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/68/609168/7/check/swift-tox-py37/06b4b74/ : SUCCESS in 9m 02s\n- swift-tox-func http://logs.openstack.org/68/609168/7/check/swift-tox-func/9f9e6cf/ : SUCCESS in 9m 13s\n- swift-tox-func-encryption http://logs.openstack.org/68/609168/7/check/swift-tox-func-encryption/a61f3a6/ : SUCCESS in 8m 37s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/609168/7/check/swift-tox-func-domain-remap-staticweb/de4b9d8/ : SUCCESS in 8m 32s\n- swift-tox-func-ec http://logs.openstack.org/68/609168/7/check/swift-tox-func-ec/10cf835/ : SUCCESS in 9m 26s\n- swift-tox-func-s3api http://logs.openstack.org/68/609168/7/check/swift-tox-func-s3api/f70c87f/ : SUCCESS in 10m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/609168/7/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e98cab6/ : SUCCESS in 13m 43s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/609168/7/check/swift-probetests-centos-7/a66d0b3/ : SUCCESS in 31m 40s\n- swift-dsvm-functional http://logs.openstack.org/68/609168/7/check/swift-dsvm-functional/5d5675b/ : SUCCESS in 23m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/609168/7/check/swift-dsvm-functional-ipv6/33711ba/ : SUCCESS in 24m 45s\n- swift-tox-lower-constraints http://logs.openstack.org/68/609168/7/check/swift-tox-lower-constraints/894622c/ : SUCCESS in 11m 09s\n- openstack-tox-pep8 http://logs.openstack.org/68/609168/7/check/openstack-tox-pep8/55c3ac1/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/609168/7/check/swift-multinode-rolling-upgrade/d8bf9bb/ : FAILURE in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"44e4fc4ce197ccd61940cd6bbd776c18c6ba1f0d","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-14 05:52:25.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)\n\nLooks fine to me. Thanks.","accounts_in_message":[],"_revision_number":7},{"id":"78f1948dcbf7b0745a2aefb20dc67fd0256e3160","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-14 16:21:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b19151b437a5b9943924efc27386880b1fa99d05","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-20 00:21:00.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3e149b9474392687dd7a00358210f3a3ae050b55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 00:21:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4b827d46c3d7e6d3e4c5b13248de6810a8a950d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 01:50:23.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a00f46f59944af9bf2692c679d11987 : SUCCESS in 7m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1c18bee9b6c442de9f795ccac7c2cd3f : SUCCESS in 1h 01m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6baf70e042424c5eb2907f211e62f6d5 : SUCCESS in 1h 04m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3c4bb2a197b9404081799b6232e96785 : SUCCESS in 10m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b6f655a34e8e431ba38186506fd6ecb8 : SUCCESS in 15m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5864144236841acb623252f5592d102 : SUCCESS in 12m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4b6ea9db2045412a9ffdf70d06ce4abd : SUCCESS in 12m 16s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f54ffe24ed2d4dafad07dec3ba6a85e4 : SUCCESS in 9m 25s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/aa3319283a3741f393f4d7cae8d4d95b : SUCCESS in 10m 32s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1e5b5111531d4e9abedfb2592f3fb3ca : SUCCESS in 11m 36s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d32ff4c0a3354619a611f11ff0b868a1 : SUCCESS in 10m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/838e0e411f1d46b7ab2b63867e78fa94 : SUCCESS in 13m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5117c2c674f14b48a39f2ca1af0497a1 : SUCCESS in 13m 38s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2e83ee21a734434a86bfa7e65a99ccf8 : SUCCESS in 9m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/41c5f411880d42c2a1b96218a2e468a3 : SUCCESS in 10m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/10e87adb6204428cab83e5ac936c4350 : SUCCESS in 12m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0697eb543333427097e986993d4048e6 : SUCCESS in 33m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2eebd99d6eb24be5ae4629c7be4a6928 : SUCCESS in 30m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/45a15fd9eb24466fad09150ebb42c419 : SUCCESS in 31m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9dfde3d77464bfdb7a7ad39c7d0f101 : SUCCESS in 39m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba447a3f1b7b4a398f889cff5afa3bf4 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26cf8020dab740a89ea8c1711bc4c1ec : SUCCESS in 6m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2060b885e74c47b8a5dcee1e1514478e : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":7},{"id":"e62a8af940bfcd6838c6d8b849885c139829046e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 01:50:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"7eb46f6e86a9d253fb4d2e45283301160014d5b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 01:53:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f3b28127a642ac866ab8238835dce2 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0214a0c67d084d60a924b203371f5d84 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4aa71aa25caed34f36fafe2de025425aa1d1e0b2","revisions":{"a6b04cfcafa06da11da89eb3616b2416f2f1ca04":{"kind":"REWORK","_number":1,"created":"2018-10-09 23:51:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/609168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/609168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/609168/1"}}},"commit":{"parents":[{"commit":"b08b6f88f9316705557b7dc32c45fcb82a267ed3","subject":"s3api: Stop mangling Authorization header for v4 signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b08b6f88f9316705557b7dc32c45fcb82a267ed3"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-10-09 23:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-10-09 23:42:18.000000000","tz":-420},"subject":"We don\u0027t have to keep the retrieved token anymore","message":"We don\u0027t have to keep the retrieved token anymore\n\nSince the change in s3_token_middleware to retrieve the auth info\nfrom keystone directly, now, we don\u0027t need to have any tokens provided\nby keystone in the request header as X-Auth-Token.\n\nNote that this makes the pipeline ordering change documented in the\nrelated changes mandatory, even when working with a v2 Keystone server.\n\nChange-Id: I7c251a758dfc1fedb3fb61e351de305b431afa79\nRelated-Change: I21e38884a2aefbb94b76c76deccd815f01db7362\nRelated-Change: Ic9af387b9192f285f0f486e7171eefb23968007e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6b04cfcafa06da11da89eb3616b2416f2f1ca04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6b04cfcafa06da11da89eb3616b2416f2f1ca04"}]},"branch":"refs/heads/master"},"59ae982d507e8d9ffa560c8fe3dc7ab7bb6ee22b":{"kind":"REWORK","_number":2,"created":"2018-10-10 00:41:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/609168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/609168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/609168/2"}}},"commit":{"parents":[{"commit":"b08b6f88f9316705557b7dc32c45fcb82a267ed3","subject":"s3api: Stop mangling Authorization header for v4 signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b08b6f88f9316705557b7dc32c45fcb82a267ed3"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-10-09 23:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2018-10-10 00:41:41.000000000","tz":0},"subject":"We don\u0027t have to keep the retrieved token anymore","message":"We don\u0027t have to keep the retrieved token anymore\n\nSince the change in s3_token_middleware to retrieve the auth info\nfrom keystone directly, now, we don\u0027t need to have any tokens provided\nby keystone in the request header as X-Auth-Token.\n\nNote that this makes the pipeline ordering change documented in the\nrelated changes mandatory, even when working with a v2 Keystone server.\n\nChange-Id: I7c251a758dfc1fedb3fb61e351de305b431afa79\nRelated-Change: I21e38884a2aefbb94b76c76deccd815f01db7362\nRelated-Change: Ic9af387b9192f285f0f486e7171eefb23968007e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59ae982d507e8d9ffa560c8fe3dc7ab7bb6ee22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59ae982d507e8d9ffa560c8fe3dc7ab7bb6ee22b"}]},"branch":"refs/heads/master"},"af2e058476253d10369a76f6063bdddbdb63ae2a":{"kind":"REWORK","_number":3,"created":"2018-12-12 00:16:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/609168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/609168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/609168/3"}}},"commit":{"parents":[{"commit":"014511d31e4d2711913d2e7ee55f58a904125c60","subject":"s3api: Stop mangling Authorization header for v4 signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/014511d31e4d2711913d2e7ee55f58a904125c60"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-10-09 23:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-12-12 00:16:17.000000000","tz":-480},"subject":"We don\u0027t have to keep the retrieved token anymore","message":"We don\u0027t have to keep the retrieved token anymore\n\nSince the change in s3_token_middleware to retrieve the auth info\nfrom keystone directly, now, we don\u0027t need to have any tokens provided\nby keystone in the request header as X-Auth-Token.\n\nNote that this makes the pipeline ordering change documented in the\nrelated changes mandatory, even when working with a v2 Keystone server.\n\nChange-Id: I7c251a758dfc1fedb3fb61e351de305b431afa79\nRelated-Change: I21e38884a2aefbb94b76c76deccd815f01db7362\nRelated-Change: Ic9af387b9192f285f0f486e7171eefb23968007e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af2e058476253d10369a76f6063bdddbdb63ae2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af2e058476253d10369a76f6063bdddbdb63ae2a"}]},"branch":"refs/heads/master"},"92b3d93118199e82b64b2a9b7700c21eb6c89ad6":{"kind":"REWORK","_number":4,"created":"2019-03-26 21:10:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/609168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/609168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/609168/4"}}},"commit":{"parents":[{"commit":"89e5927f7dd94fc28b3847944eb7dd227d516fa8","subject":"Fix mocking time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89e5927f7dd94fc28b3847944eb7dd227d516fa8"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-10-09 23:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-03-26 21:10:07.000000000","tz":-420},"subject":"We don\u0027t have to keep the retrieved token anymore","message":"We don\u0027t have to keep the retrieved token anymore\n\nSince the change in s3_token_middleware to retrieve the auth info\nfrom keystone directly, now, we don\u0027t need to have any tokens provided\nby keystone in the request header as X-Auth-Token.\n\nNote that this makes the pipeline ordering change documented in the\nrelated changes mandatory, even when working with a v2 Keystone server.\n\nChange-Id: I7c251a758dfc1fedb3fb61e351de305b431afa79\nRelated-Change: I21e38884a2aefbb94b76c76deccd815f01db7362\nRelated-Change: Ic9af387b9192f285f0f486e7171eefb23968007e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92b3d93118199e82b64b2a9b7700c21eb6c89ad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92b3d93118199e82b64b2a9b7700c21eb6c89ad6"}]},"branch":"refs/heads/master"},"13ba43e8f8532b9cbc26d8b68aeee1cf4dda39d4":{"kind":"REWORK","_number":5,"created":"2019-05-03 23:42:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/609168/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/609168/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/609168/5"}}},"commit":{"parents":[{"commit":"89e5927f7dd94fc28b3847944eb7dd227d516fa8","subject":"Fix mocking time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89e5927f7dd94fc28b3847944eb7dd227d516fa8"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-10-09 23:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-03 23:42:45.000000000","tz":-420},"subject":"We don\u0027t have to keep the retrieved token anymore","message":"We don\u0027t have to keep the retrieved token anymore\n\nSince the change in s3_token_middleware to retrieve the auth info\nfrom keystone directly, now, we don\u0027t need to have any tokens provided\nby keystone in the request header as X-Auth-Token.\n\nNote that this makes the pipeline ordering change documented in the\nrelated changes mandatory, even when working with a v2 Keystone server.\n\nChange-Id: I7c251a758dfc1fedb3fb61e351de305b431afa79\nRelated-Change: I21e38884a2aefbb94b76c76deccd815f01db7362\nRelated-Change: Ic9af387b9192f285f0f486e7171eefb23968007e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13ba43e8f8532b9cbc26d8b68aeee1cf4dda39d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13ba43e8f8532b9cbc26d8b68aeee1cf4dda39d4"}]},"branch":"refs/heads/master"},"0d0b9e873cffe9a9e0f54abb8f6dbcc1a8b0a0af":{"kind":"REWORK","_number":6,"created":"2019-05-07 18:30:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/609168/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/609168/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/609168/6"}}},"commit":{"parents":[{"commit":"f9898ec8f3b27954a6544db91ca36f20b5b74feb","subject":"Merge \"[Trivial fix]Remove unnecessary slash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9898ec8f3b27954a6544db91ca36f20b5b74feb"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-10-09 23:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 18:29:50.000000000","tz":-420},"subject":"We don\u0027t have to keep the retrieved token anymore","message":"We don\u0027t have to keep the retrieved token anymore\n\nSince the change in s3_token_middleware to retrieve the auth info\nfrom keystone directly, now, we don\u0027t need to have any tokens provided\nby keystone in the request header as X-Auth-Token.\n\nNote that this makes the pipeline ordering change documented in the\nrelated changes mandatory, even when working with a v2 Keystone server.\n\nChange-Id: I7c251a758dfc1fedb3fb61e351de305b431afa79\nRelated-Change: I21e38884a2aefbb94b76c76deccd815f01db7362\nRelated-Change: Ic9af387b9192f285f0f486e7171eefb23968007e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d0b9e873cffe9a9e0f54abb8f6dbcc1a8b0a0af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d0b9e873cffe9a9e0f54abb8f6dbcc1a8b0a0af"}]},"branch":"refs/heads/master"},"4aa71aa25caed34f36fafe2de025425aa1d1e0b2":{"kind":"REWORK","_number":7,"created":"2019-05-09 18:21:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/609168/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/609168/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/609168/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/609168/7"}}},"commit":{"parents":[{"commit":"032cf3b3b4599bd720131c8f02cc6ed8a155683d","subject":"Update docker secret.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/032cf3b3b4599bd720131c8f02cc6ed8a155683d"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2018-10-09 23:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-09 18:21:00.000000000","tz":-420},"subject":"We don\u0027t have to keep the retrieved token anymore","message":"We don\u0027t have to keep the retrieved token anymore\n\nSince the change in s3_token_middleware to retrieve the auth info\nfrom keystone directly, now, we don\u0027t need to have any tokens provided\nby keystone in the request header as X-Auth-Token.\n\nNote that this makes the pipeline ordering change documented in the\nrelated changes mandatory, even when working with a v2 Keystone server.\n\nChange-Id: I7c251a758dfc1fedb3fb61e351de305b431afa79\nRelated-Change: I21e38884a2aefbb94b76c76deccd815f01db7362\nRelated-Change: Ic9af387b9192f285f0f486e7171eefb23968007e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aa71aa25caed34f36fafe2de025425aa1d1e0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aa71aa25caed34f36fafe2de025425aa1d1e0b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
