)]}'
{"id":"openstack%2Fswift~609991","triplet_id":"openstack%2Fswift~master~Ie94d0de9c8b57a1e1926b9915be4b3c8913b0310","project":"openstack/swift","branch":"master","topic":"ssync-requests","hashtags":[],"change_id":"Ie94d0de9c8b57a1e1926b9915be4b3c8913b0310","subject":"Allow parallel requests during a SSYNC replication","status":"ABANDONED","created":"2018-10-12 09:58:31.000000000","updated":"2018-11-06 13:13:41.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"da122a69e9aba3ec703a2ea79b3986b61328fd46","_number":609991,"virtual_id_number":609991,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2018-10-12 12:37:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"date":"2018-10-12 22:26:29.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2018-10-12 11:56:44.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-12 12:37:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-12 22:26:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"39bc0a6bff9fb78ab6e9a53db31b1f0ca89e50d7","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-12 09:58:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3facdcca1e647c5d67709dcf3e7f12a90f4a2ebd","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-12 10:03:28.000000000","message":"Patch Set 1: Workflow-1\n\nThis gives us a (almost) 2x perf improvement on updates() function during revert","accounts_in_message":[],"_revision_number":1},{"id":"46a9d104aaedd123f04747ac5bc663b94c2f1938","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-12 10:50:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/91/609991/1/check/openstack-tox-cover/8a81b72/cover/ : SUCCESS in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/91/609991/1/check/openstack-tox-docs/83154c4/html/ : SUCCESS in 4m 11s\n- swift-tox-py27 http://logs.openstack.org/91/609991/1/check/swift-tox-py27/8e162c5/ : FAILURE in 11m 40s\n- swift-tox-py35 http://logs.openstack.org/91/609991/1/check/swift-tox-py35/3379831/ : SUCCESS in 4m 55s\n- swift-tox-py36 http://logs.openstack.org/91/609991/1/check/swift-tox-py36/ceeb7d4/ : SUCCESS in 5m 06s (non-voting)\n- swift-tox-func http://logs.openstack.org/91/609991/1/check/swift-tox-func/e2cbef9/ : SUCCESS in 8m 43s\n- swift-tox-func-encryption http://logs.openstack.org/91/609991/1/check/swift-tox-func-encryption/3311508/ : SUCCESS in 9m 28s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/91/609991/1/check/swift-tox-func-domain-remap-staticweb/849af1f/ : SUCCESS in 9m 34s\n- swift-tox-func-ec http://logs.openstack.org/91/609991/1/check/swift-tox-func-ec/8f0792e/ : SUCCESS in 9m 56s\n- swift-tox-func-s3api http://logs.openstack.org/91/609991/1/check/swift-tox-func-s3api/92aaf15/ : FAILURE in 13m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/91/609991/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f28e1bb/ : SUCCESS in 9m 36s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/91/609991/1/check/swift-probetests-centos-7/73b48de/ : SUCCESS in 29m 36s\n- swift-dsvm-functional http://logs.openstack.org/91/609991/1/check/swift-dsvm-functional/a1953e1/ : SUCCESS in 28m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/91/609991/1/check/swift-dsvm-functional-ipv6/8c6bcbb/ : SUCCESS in 22m 22s\n- swift-tox-lower-constraints http://logs.openstack.org/91/609991/1/check/swift-tox-lower-constraints/51e077b/ : FAILURE in 11m 38s\n- openstack-tox-pep8 http://logs.openstack.org/91/609991/1/check/openstack-tox-pep8/711af1c/ : FAILURE in 4m 52s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/91/609991/1/check/swift-multinode-rolling-upgrade/4d4a1d5/ : SUCCESS in 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f750d1046cdfba1b4865a2d2e96aba8ab5899928","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-12 11:55:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e867530221028333abe947c2682f81a89ecdabc","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-10-12 11:56:44.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"0d4c95cd22389f19fe608addc1f36dd379b4aee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-12 12:37:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/609991/2/check/openstack-tox-cover/16c6ed2/cover/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/91/609991/2/check/openstack-tox-docs/5a16552/html/ : SUCCESS in 4m 30s\n- swift-tox-py27 http://logs.openstack.org/91/609991/2/check/swift-tox-py27/cca58b4/ : SUCCESS in 12m 28s\n- swift-tox-py35 http://logs.openstack.org/91/609991/2/check/swift-tox-py35/3768629/ : SUCCESS in 4m 30s\n- swift-tox-py36 http://logs.openstack.org/91/609991/2/check/swift-tox-py36/4103f75/ : SUCCESS in 4m 15s (non-voting)\n- swift-tox-func http://logs.openstack.org/91/609991/2/check/swift-tox-func/d6c1fa3/ : SUCCESS in 23m 39s\n- swift-tox-func-encryption http://logs.openstack.org/91/609991/2/check/swift-tox-func-encryption/a1b1f8f/ : SUCCESS in 8m 38s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/91/609991/2/check/swift-tox-func-domain-remap-staticweb/74f8db4/ : SUCCESS in 8m 14s\n- swift-tox-func-ec http://logs.openstack.org/91/609991/2/check/swift-tox-func-ec/21db839/ : SUCCESS in 8m 37s\n- swift-tox-func-s3api http://logs.openstack.org/91/609991/2/check/swift-tox-func-s3api/3a22e0f/ : SUCCESS in 12m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/91/609991/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/73a9541/ : SUCCESS in 9m 40s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/91/609991/2/check/swift-probetests-centos-7/32fe898/ : SUCCESS in 26m 42s\n- swift-dsvm-functional http://logs.openstack.org/91/609991/2/check/swift-dsvm-functional/6fffd65/ : SUCCESS in 25m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/91/609991/2/check/swift-dsvm-functional-ipv6/4913445/ : SUCCESS in 24m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/91/609991/2/check/swift-tox-lower-constraints/e6bb3f5/ : SUCCESS in 11m 02s\n- openstack-tox-pep8 http://logs.openstack.org/91/609991/2/check/openstack-tox-pep8/28d1ca3/ : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/91/609991/2/check/swift-multinode-rolling-upgrade/5184699/ : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":2},{"id":"901c34b9e0b5be7c1d3a1dbbe9d0015e0ea7404c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-10-12 22:26:29.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3fa0857496d962aa61491609776b2ae2e17c4870","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2018-11-06 13:13:41.000000000","message":"Abandoned\n\nAbandoning, in favor of https://review.openstack.org/#/c/613987/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14fc4244d3919bd753e5e515f846ae3b3f16fd86","revisions":{"de8466cbf218054ca063d713974e3de46749e4b3":{"kind":"REWORK","_number":1,"created":"2018-10-12 09:58:31.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/91/609991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/609991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/609991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/609991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/609991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/609991/1"}}},"commit":{"parents":[{"commit":"d7955048b3e6b17be8ae6f4915e9e6700016efd1","subject":"Reduce code duplication in SSYNC sender","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7955048b3e6b17be8ae6f4915e9e6700016efd1"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-10-12 06:54:06.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-10-12 09:44:08.000000000","tz":120},"subject":"Allow parallel requests during a SSYNC replication","message":"Allow parallel requests during a SSYNC replication\n\nChange-Id: Ie94d0de9c8b57a1e1926b9915be4b3c8913b0310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de8466cbf218054ca063d713974e3de46749e4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de8466cbf218054ca063d713974e3de46749e4b3"}]},"branch":"refs/heads/master"},"14fc4244d3919bd753e5e515f846ae3b3f16fd86":{"kind":"REWORK","_number":2,"created":"2018-10-12 11:55:26.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/91/609991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/609991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/609991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/609991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/609991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/609991/2"}}},"commit":{"parents":[{"commit":"2913319e08658701134f72c0d779b210302816a5","subject":"Reduce code duplication in SSYNC sender","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2913319e08658701134f72c0d779b210302816a5"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-10-12 06:54:06.000000000","tz":120},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2018-10-12 11:50:56.000000000","tz":120},"subject":"Allow parallel requests during a SSYNC replication","message":"Allow parallel requests during a SSYNC replication\n\nChange-Id: Ie94d0de9c8b57a1e1926b9915be4b3c8913b0310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14fc4244d3919bd753e5e515f846ae3b3f16fd86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14fc4244d3919bd753e5e515f846ae3b3f16fd86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
