)]}'
{"id":"openstack%2Fswift~619303","triplet_id":"openstack%2Fswift~master~I3603247e0c36299de2107aa8e494a3f87647696f","project":"openstack/swift","branch":"master","topic":"py3/mw-bulk","hashtags":[],"change_id":"I3603247e0c36299de2107aa8e494a3f87647696f","subject":"py3: port bulk middleware","status":"ABANDONED","created":"2018-11-21 16:27:18.000000000","updated":"2019-05-17 20:55:14.000000000","total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef418c3d01d9317e5be11d539ddeeb0530d3d554","_number":619303,"virtual_id_number":619303,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-04-24 04:27:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-07 00:04:26.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-23 08:41:27.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-04-24 04:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-07 00:04:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"971c5f794eda6d7f33fc788002ecd8232c342474","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2018-11-21 16:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5844ecb424f8fbdf6925b6a3518b09ebb6187eff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-21 17:34:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/1/check/openstack-tox-docs/7c1bd87/html/ : SUCCESS in 5m 18s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/1/check/build-openstack-releasenotes/030828c/html/ : SUCCESS in 6m 26s\n- swift-tox-py27 http://logs.openstack.org/03/619303/1/check/swift-tox-py27/0be9c34/ : SUCCESS in 11m 34s\n- swift-tox-py35 http://logs.openstack.org/03/619303/1/check/swift-tox-py35/fefadcd/ : SUCCESS in 7m 03s\n- swift-tox-py36 http://logs.openstack.org/03/619303/1/check/swift-tox-py36/15993e9/ : SUCCESS in 5m 56s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/1/check/swift-tox-func/471a593/ : SUCCESS in 14m 03s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/1/check/swift-tox-func-encryption/ea680ef/ : SUCCESS in 13m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/1/check/swift-tox-func-domain-remap-staticweb/e481c65/ : SUCCESS in 12m 40s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/1/check/swift-tox-func-ec/a06a387/ : SUCCESS in 15m 58s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/1/check/swift-tox-func-s3api/74fb2fc/ : SUCCESS in 15m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/efd9940/ : SUCCESS in 15m 56s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/1/check/swift-probetests-centos-7/92df9c4/ : SUCCESS in 32m 26s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/1/check/swift-dsvm-functional/450dffc/ : SUCCESS in 26m 04s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/1/check/swift-dsvm-functional-ipv6/61a4cf6/ : SUCCESS in 25m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/1/check/swift-tox-lower-constraints/0f8603b/ : SUCCESS in 12m 13s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/1/check/openstack-tox-pep8/99a9590/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/1/check/swift-multinode-rolling-upgrade/559ceae/ : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a38aeb43e775b5d09286002bbcbaed880e8da185","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-11-23 08:41:27.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nOnly just started the review on this. I then got side tracked and realised I need to rebuild my py3 saio env.. and now it\u0027s Friday night, so will finish reviewing and testing the py3 patches next week.\n\n\\o/ Beer o\u0027clock!","accounts_in_message":[],"_revision_number":1},{"id":"1dc19b297763c08eb6ab112a7dc7252ea54ca78a","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-02-28 02:31:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbbce3d6dbc846c0fcb43a2d1d8164f897e9f110","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 03:12:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/2/check/openstack-tox-docs/5b3066a/html/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/2/check/build-openstack-releasenotes/33eae41/html/ : SUCCESS in 6m 15s\n- swift-tox-py27 http://logs.openstack.org/03/619303/2/check/swift-tox-py27/84cdc45/ : SUCCESS in 10m 54s\n- swift-tox-py35 http://logs.openstack.org/03/619303/2/check/swift-tox-py35/fae99b5/ : SUCCESS in 12m 57s\n- swift-tox-py36 http://logs.openstack.org/03/619303/2/check/swift-tox-py36/24d3ffc/ : SUCCESS in 8m 43s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/2/check/swift-tox-func/bc3c407/ : SUCCESS in 14m 32s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/2/check/swift-tox-func-encryption/2a18bcc/ : SUCCESS in 9m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/2/check/swift-tox-func-domain-remap-staticweb/4878f20/ : SUCCESS in 14m 23s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/2/check/swift-tox-func-ec/3aa23ef/ : SUCCESS in 10m 43s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/2/check/swift-tox-func-s3api/e015ac2/ : SUCCESS in 16m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/32059d7/ : SUCCESS in 12m 53s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/2/check/swift-probetests-centos-7/1fc6191/ : FAILURE in 33m 21s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/2/check/swift-dsvm-functional/81d78c6/ : SUCCESS in 23m 07s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/2/check/swift-dsvm-functional-ipv6/5cf5f83/ : SUCCESS in 32m 05s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/2/check/swift-tox-lower-constraints/94f8dd3/ : SUCCESS in 11m 35s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/2/check/openstack-tox-pep8/8fe5123/ : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/2/check/swift-multinode-rolling-upgrade/7f6d472/ : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d1d0909ba749fa5590de917b227955945a0b7844","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-03-01 23:35:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69e6935c9f82bb38a2a34a91d31614e83f9e1c3c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-03-02 03:49:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3f9a8484d536fc5e5f115440c388e9d9b67c613","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-03-02 03:49:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e4568746158f61bcbbfa9e9e6e0dc0f5699bcde1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-02 04:25:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/3/check/openstack-tox-docs/915bb97/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/3/check/build-openstack-releasenotes/7e82186/html/ : SUCCESS in 5m 34s\n- swift-tox-py27 http://logs.openstack.org/03/619303/3/check/swift-tox-py27/13c9538/ : SUCCESS in 12m 16s\n- swift-tox-py35 http://logs.openstack.org/03/619303/3/check/swift-tox-py35/ce5d7f8/ : SUCCESS in 10m 47s\n- swift-tox-py36 http://logs.openstack.org/03/619303/3/check/swift-tox-py36/21db7e5/ : SUCCESS in 9m 44s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/3/check/swift-tox-func/40cb8d4/ : SUCCESS in 8m 04s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/3/check/swift-tox-func-encryption/551e219/ : SUCCESS in 8m 36s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/3/check/swift-tox-func-domain-remap-staticweb/4b6e9a1/ : SUCCESS in 10m 23s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/3/check/swift-tox-func-ec/7cec108/ : SUCCESS in 8m 55s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/3/check/swift-tox-func-s3api/1aca568/ : SUCCESS in 11m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/564581d/ : SUCCESS in 10m 58s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/3/check/swift-probetests-centos-7/123910d/ : FAILURE in 28m 06s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/3/check/swift-dsvm-functional/13d5f18/ : SUCCESS in 27m 55s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/3/check/swift-dsvm-functional-ipv6/3b27984/ : SUCCESS in 28m 53s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/3/check/swift-tox-lower-constraints/1355458/ : SUCCESS in 9m 24s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/3/check/openstack-tox-pep8/e2ea17d/ : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/3/check/swift-multinode-rolling-upgrade/f6dab66/ : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2750507d358ec54d015743f4e17d15f056ae0490","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-03-06 00:12:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"12c8d5264cb9f699fa76abc0dc750bba7a230ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 02:24:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/4/check/openstack-tox-docs/f9e1fe4/html/ : SUCCESS in 5m 16s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/4/check/build-openstack-releasenotes/23ef623/html/ : SUCCESS in 6m 56s\n- swift-tox-py27 http://logs.openstack.org/03/619303/4/check/swift-tox-py27/d6067f3/ : SUCCESS in 11m 43s\n- swift-tox-py35 http://logs.openstack.org/03/619303/4/check/swift-tox-py35/3d20aab/ : SUCCESS in 8m 43s\n- swift-tox-py36 http://logs.openstack.org/03/619303/4/check/swift-tox-py36/f4a3c80/ : SUCCESS in 8m 10s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/4/check/swift-tox-func/3b98571/ : SUCCESS in 8m 46s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/4/check/swift-tox-func-encryption/6218d17/ : SUCCESS in 9m 16s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/4/check/swift-tox-func-domain-remap-staticweb/ef6b5a2/ : SUCCESS in 8m 30s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/4/check/swift-tox-func-ec/578b0fa/ : SUCCESS in 12m 02s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/4/check/swift-tox-func-s3api/820f3d1/ : SUCCESS in 9m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b243419/ : SUCCESS in 13m 39s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/4/check/swift-probetests-centos-7/8565b07/ : SUCCESS in 29m 46s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/4/check/swift-dsvm-functional/3d06dca/ : SUCCESS in 22m 58s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/4/check/swift-dsvm-functional-ipv6/99a96cc/ : SUCCESS in 23m 17s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/4/check/swift-tox-lower-constraints/6bfec42/ : SUCCESS in 16m 23s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/4/check/openstack-tox-pep8/124f743/ : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/4/check/swift-multinode-rolling-upgrade/ef93658/ : FAILURE in 16m 52s","accounts_in_message":[],"_revision_number":4},{"id":"a68018d4b3238912592959f2d5b66c7c3d1dc2dd","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-03-06 02:36:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5daa88afe2d3feea96387f391e4b0648741aa9a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 03:15:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/4/check/openstack-tox-docs/c5d54c4/html/ : SUCCESS in 4m 08s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/4/check/build-openstack-releasenotes/ee7ec32/html/ : SUCCESS in 6m 55s\n- swift-tox-py27 http://logs.openstack.org/03/619303/4/check/swift-tox-py27/25aa0b8/ : SUCCESS in 11m 35s\n- swift-tox-py35 http://logs.openstack.org/03/619303/4/check/swift-tox-py35/778b8df/ : SUCCESS in 8m 06s\n- swift-tox-py36 http://logs.openstack.org/03/619303/4/check/swift-tox-py36/f1bc943/ : SUCCESS in 9m 06s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/4/check/swift-tox-func/8eefea8/ : SUCCESS in 10m 44s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/4/check/swift-tox-func-encryption/c41e84e/ : SUCCESS in 9m 32s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/4/check/swift-tox-func-domain-remap-staticweb/803c0ec/ : SUCCESS in 15m 48s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/4/check/swift-tox-func-ec/a484610/ : SUCCESS in 9m 53s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/4/check/swift-tox-func-s3api/38bcccd/ : SUCCESS in 9m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fdcc5dc/ : SUCCESS in 13m 09s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/4/check/swift-probetests-centos-7/d1bf4db/ : SUCCESS in 37m 45s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/4/check/swift-dsvm-functional/d96a983/ : SUCCESS in 23m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/4/check/swift-dsvm-functional-ipv6/6c1fe84/ : SUCCESS in 25m 16s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/4/check/swift-tox-lower-constraints/3abfda9/ : SUCCESS in 11m 05s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/4/check/openstack-tox-pep8/9f98cbf/ : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/4/check/swift-multinode-rolling-upgrade/f2dffc0/ : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":4},{"id":"c186958d4600ec23de226ec0d6da76445036d2a6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-03-11 21:46:49.000000000","message":"Patch Set 4:\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9dd77ecf53943b395b4473a409da0f95bee0e924","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-03-28 02:37:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a268fcf8a0a28aa54167e41df0daad711e328de4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 03:12:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/5/check/openstack-tox-docs/78d46af/html/ : SUCCESS in 5m 06s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/5/check/build-openstack-releasenotes/63338ac/html/ : SUCCESS in 6m 42s\n- swift-tox-py27 http://logs.openstack.org/03/619303/5/check/swift-tox-py27/acf48a3/ : SUCCESS in 13m 33s\n- swift-tox-py35 http://logs.openstack.org/03/619303/5/check/swift-tox-py35/ad7e88f/ : SUCCESS in 10m 15s\n- swift-tox-py36 http://logs.openstack.org/03/619303/5/check/swift-tox-py36/1cf1ed1/ : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/619303/5/check/swift-tox-py37/e55903b/ : SUCCESS in 11m 19s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/5/check/swift-tox-func/f3d0098/ : SUCCESS in 8m 32s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/5/check/swift-tox-func-encryption/957392d/ : SUCCESS in 8m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/5/check/swift-tox-func-domain-remap-staticweb/a9286ef/ : SUCCESS in 8m 24s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/5/check/swift-tox-func-ec/493c5bb/ : SUCCESS in 9m 50s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/5/check/swift-tox-func-s3api/01923d3/ : SUCCESS in 12m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/87c1a30/ : SUCCESS in 13m 39s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/5/check/swift-probetests-centos-7/6acb93f/ : SUCCESS in 30m 45s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/5/check/swift-dsvm-functional/0807dcc/ : SUCCESS in 28m 04s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/5/check/swift-dsvm-functional-ipv6/0e26bcd/ : SUCCESS in 27m 26s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/5/check/swift-tox-lower-constraints/a152301/ : SUCCESS in 13m 06s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/5/check/openstack-tox-pep8/3c32de2/ : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/5/check/swift-multinode-rolling-upgrade/6a6d0af/ : SUCCESS in 17m 00s","accounts_in_message":[],"_revision_number":5},{"id":"ccdabb570b1e12b8905e14b927aed8cff694edd4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-04-06 06:02:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"99adcaeaeb97b8d91a3a2612f3bf76819ca5fb05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-06 07:04:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/6/check/openstack-tox-docs/a7818d9/html/ : SUCCESS in 5m 27s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/6/check/build-openstack-releasenotes/f010f6d/html/ : SUCCESS in 8m 47s\n- swift-tox-py27 http://logs.openstack.org/03/619303/6/check/swift-tox-py27/0dba6cb/ : FAILURE in 18m 37s\n- swift-tox-py35 http://logs.openstack.org/03/619303/6/check/swift-tox-py35/ea88cbc/ : SUCCESS in 9m 47s\n- swift-tox-py36 http://logs.openstack.org/03/619303/6/check/swift-tox-py36/f319e1b/ : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/619303/6/check/swift-tox-py37/c05eb46/ : SUCCESS in 9m 21s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/6/check/swift-tox-func/3852f6d/ : SUCCESS in 9m 46s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/6/check/swift-tox-func-encryption/9d95d8c/ : SUCCESS in 11m 21s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/6/check/swift-tox-func-domain-remap-staticweb/a5163ac/ : SUCCESS in 10m 37s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/6/check/swift-tox-func-ec/aeb096d/ : SUCCESS in 10m 40s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/6/check/swift-tox-func-s3api/c1fdcda/ : SUCCESS in 11m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/6/check/swift-tox-func-s3api-ceph-s3tests-tempauth/23235fe/ : SUCCESS in 13m 56s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/6/check/swift-probetests-centos-7/b96ec33/ : SUCCESS in 28m 56s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/6/check/swift-dsvm-functional/ae3e2ff/ : SUCCESS in 27m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/6/check/swift-dsvm-functional-ipv6/2432699/ : SUCCESS in 26m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/6/check/swift-tox-lower-constraints/b882618/ : SUCCESS in 14m 32s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/6/check/openstack-tox-pep8/a39e408/ : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/6/check/swift-multinode-rolling-upgrade/59baa94/ : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":6},{"id":"a70a1c4dd6e55017442568877baccf8e2e9b1679","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-08 16:48:12.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6d3a5604c81942bfb88689fc8b47dc02d6530621","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-08 16:48:18.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4b71c9cff120231fdec51d6831485ae791902ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 18:33:36.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/6/check/openstack-tox-docs/89a5b9a/html/ : SUCCESS in 5m 28s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/6/check/build-openstack-releasenotes/5acd1bc/html/ : SUCCESS in 7m 49s\n- swift-tox-py27 http://logs.openstack.org/03/619303/6/check/swift-tox-py27/b583692/ : SUCCESS in 13m 23s\n- swift-tox-py35 http://logs.openstack.org/03/619303/6/check/swift-tox-py35/d55eedf/ : SUCCESS in 10m 46s\n- swift-tox-py36 http://logs.openstack.org/03/619303/6/check/swift-tox-py36/5d0f129/ : SUCCESS in 12m 07s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/619303/6/check/swift-tox-py37/95da325/ : SUCCESS in 10m 26s (non-voting)\n- swift-tox-func http://logs.openstack.org/03/619303/6/check/swift-tox-func/9de3330/ : SUCCESS in 10m 48s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/6/check/swift-tox-func-encryption/0196268/ : SUCCESS in 11m 06s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/6/check/swift-tox-func-domain-remap-staticweb/718cf3f/ : SUCCESS in 14m 24s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/6/check/swift-tox-func-ec/60b3f65/ : SUCCESS in 13m 13s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/6/check/swift-tox-func-s3api/26c15c0/ : SUCCESS in 10m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/6/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9ce3993/ : SUCCESS in 13m 51s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/6/check/swift-probetests-centos-7/5ce101c/ : FAILURE in 27m 56s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/6/check/swift-dsvm-functional/033a2d0/ : SUCCESS in 29m 43s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/6/check/swift-dsvm-functional-ipv6/fccd57b/ : SUCCESS in 37m 30s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/6/check/swift-tox-lower-constraints/cda355f/ : SUCCESS in 13m 27s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/6/check/openstack-tox-pep8/e1d87bf/ : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/6/check/swift-multinode-rolling-upgrade/df3a23c/ : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":6},{"id":"76e02db917c26ac70c659c2418e48ad897316ccc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-09 15:08:41.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"57d6b74f0b04b7f53774544daf701344aa957b67","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-04-24 00:50:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a5222b65d0c6122b1ae9e0636e696e739ce38048","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-04-24 01:23:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a6cb0b0446d2c740a336094e96c5c7e5da28ce31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:27:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/03/619303/7/check/openstack-tox-docs/279da77/html/ : SUCCESS in 4m 00s\n- build-openstack-releasenotes http://logs.openstack.org/03/619303/7/check/build-openstack-releasenotes/5c1ae5b/html/ : SUCCESS in 7m 07s\n- swift-tox-py27 http://logs.openstack.org/03/619303/7/check/swift-tox-py27/9eb7121/ : SUCCESS in 10m 22s\n- swift-tox-py35 http://logs.openstack.org/03/619303/7/check/swift-tox-py35/f0fc13e/ : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/03/619303/7/check/swift-tox-py36/8dd3747/ : SUCCESS in 8m 20s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/619303/7/check/swift-tox-py37/7f04cc6/ : SUCCESS in 8m 37s\n- swift-tox-func http://logs.openstack.org/03/619303/7/check/swift-tox-func/c752ad8/ : SUCCESS in 8m 03s\n- swift-tox-func-encryption http://logs.openstack.org/03/619303/7/check/swift-tox-func-encryption/00ab2f9/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/619303/7/check/swift-tox-func-domain-remap-staticweb/6baf231/ : SUCCESS in 10m 35s\n- swift-tox-func-ec http://logs.openstack.org/03/619303/7/check/swift-tox-func-ec/b696e84/ : SUCCESS in 10m 54s\n- swift-tox-func-s3api http://logs.openstack.org/03/619303/7/check/swift-tox-func-s3api/9749ce1/ : SUCCESS in 10m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/619303/7/check/swift-tox-func-s3api-ceph-s3tests-tempauth/87367b6/ : SUCCESS in 12m 14s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/619303/7/check/swift-probetests-centos-7/7cf7c47/ : SUCCESS in 30m 47s\n- swift-dsvm-functional http://logs.openstack.org/03/619303/7/check/swift-dsvm-functional/c485c5e/ : SUCCESS in 31m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/619303/7/check/swift-dsvm-functional-ipv6/1e05c0a/ : SUCCESS in 28m 15s\n- swift-tox-lower-constraints http://logs.openstack.org/03/619303/7/check/swift-tox-lower-constraints/c4b7f7b/ : SUCCESS in 11m 23s\n- openstack-tox-pep8 http://logs.openstack.org/03/619303/7/check/openstack-tox-pep8/449b3a5/ : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/619303/7/check/swift-multinode-rolling-upgrade/04b8eb3/ : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":7},{"id":"9be147159c4b7d2789a4102e2f92ec398b0aee1e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-07 00:04:26.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nSo I\u0027m still trying to follow the thread... I\u0027ve got a couple concerns...\n\nFWIW, I pushed up https://review.opendev.org/#/c/657505/ while trying to wrap my head around this; now I\u0027ve just gotta reconcile the two approaches. There\u0027s certainly something of an advantage to ensuring that our list of objects to delete is in native strings when we get to SLO...","accounts_in_message":[],"_revision_number":7},{"id":"3ec3d34fb8b7f434d2124ff10494a81298111940","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-17 20:55:14.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/657505/ is better","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0a54a53a11899df735b4576d99ca98059ab43c0d","revisions":{"5182dc06f3832837c571adc6df161abb97615d1b":{"kind":"REWORK","_number":1,"created":"2018-11-21 16:27:18.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/03/619303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/619303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/619303/1"}}},"commit":{"parents":[{"commit":"582f0585e8bc450e0aa04ef054dd56e80bacf285","subject":"py3: encryption follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/582f0585e8bc450e0aa04ef054dd56e80bacf285"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"subject":"py3: port bulk middleware","message":"py3: port bulk middleware\n\nThis includes a reimplementation of unquote().\n\nChange-Id: I3603247e0c36299de2107aa8e494a3f87647696f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5182dc06f3832837c571adc6df161abb97615d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5182dc06f3832837c571adc6df161abb97615d1b"}]},"branch":"refs/heads/master"},"e21aa7364e39479722cc1b3f37f4a2e8ba77d922":{"kind":"REWORK","_number":2,"created":"2019-02-28 02:31:33.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/03/619303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/619303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/619303/2"}}},"commit":{"parents":[{"commit":"3d4ed23c087315c4103a72999c9b2495931caef6","subject":"Merge \"Address some review comments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d4ed23c087315c4103a72999c9b2495931caef6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-02-28 02:22:32.000000000","tz":-360},"subject":"py3: port bulk middleware","message":"py3: port bulk middleware\n\nThis has a coupe of dark spots.\n\nFirst is a re-implementation of unquote(). We tried everything to\nmake the standard library work, but to no avail. Generally we\u0027re\ntrying to shed workarounds like bufferedhttp.py, not add new ones.\n\nSecond, and worse, is a user-visible API incompatibility. The code\nworks as long as things are fine, but when an error happens, we\nmust return the name of the failed object. This surprisibly turned\nout to be impossible to do in a compatible way. See the\ntest_bulk_delete_works_unicode(). We\u0027re currently pushing this\npatch out with this known deficiency just to unlock the overall\nprogress of the py3 port.\n\nChange-Id: I3603247e0c36299de2107aa8e494a3f87647696f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e21aa7364e39479722cc1b3f37f4a2e8ba77d922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e21aa7364e39479722cc1b3f37f4a2e8ba77d922"}]},"branch":"refs/heads/master"},"9be3cba4e4edfb6915b5759c5a9c433e7e2abfcd":{"kind":"REWORK","_number":3,"created":"2019-03-02 03:49:57.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/03/619303/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/619303/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/619303/3"}}},"commit":{"parents":[{"commit":"349c54998461accd25b84e32781a10938d4167a4","subject":"py3: port object auditor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/349c54998461accd25b84e32781a10938d4167a4"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-03-02 03:30:46.000000000","tz":-360},"subject":"py3: port bulk middleware","message":"py3: port bulk middleware\n\nChange-Id: I3603247e0c36299de2107aa8e494a3f87647696f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9be3cba4e4edfb6915b5759c5a9c433e7e2abfcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9be3cba4e4edfb6915b5759c5a9c433e7e2abfcd"}]},"branch":"refs/heads/master"},"ba7f345385c131028451862e42b7cbeb542c5f66":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-06 00:12:47.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/03/619303/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/619303/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/619303/4"}}},"commit":{"parents":[{"commit":"f9329ca5ff5cab807ead0a9dcf1a4e0fba7c27ae","subject":"Merge \"docs: clean up SAIO formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9329ca5ff5cab807ead0a9dcf1a4e0fba7c27ae"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-03-06 00:00:19.000000000","tz":-360},"subject":"py3: port bulk middleware","message":"py3: port bulk middleware\n\nChange-Id: I3603247e0c36299de2107aa8e494a3f87647696f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba7f345385c131028451862e42b7cbeb542c5f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba7f345385c131028451862e42b7cbeb542c5f66"}]},"branch":"refs/heads/master"},"765326847a4cff02793736afce9e1562711d1e83":{"kind":"REWORK","_number":5,"created":"2019-03-28 02:37:43.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/03/619303/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/619303/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/619303/5"}}},"commit":{"parents":[{"commit":"89e5927f7dd94fc28b3847944eb7dd227d516fa8","subject":"Fix mocking time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89e5927f7dd94fc28b3847944eb7dd227d516fa8"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-03-28 02:37:22.000000000","tz":-300},"subject":"py3: port bulk middleware","message":"py3: port bulk middleware\n\nChange-Id: I3603247e0c36299de2107aa8e494a3f87647696f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/765326847a4cff02793736afce9e1562711d1e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/765326847a4cff02793736afce9e1562711d1e83"}]},"branch":"refs/heads/master"},"6301bb854b94f8f980639c39c90123117f008009":{"kind":"REWORK","_number":6,"created":"2019-04-06 06:02:18.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/03/619303/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/619303/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/619303/6"}}},"commit":{"parents":[{"commit":"89e5927f7dd94fc28b3847944eb7dd227d516fa8","subject":"Fix mocking time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89e5927f7dd94fc28b3847944eb7dd227d516fa8"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-04-06 06:02:11.000000000","tz":-300},"subject":"py3: port bulk middleware","message":"py3: port bulk middleware\n\nChange-Id: I3603247e0c36299de2107aa8e494a3f87647696f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6301bb854b94f8f980639c39c90123117f008009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6301bb854b94f8f980639c39c90123117f008009"}]},"branch":"refs/heads/master"},"0a54a53a11899df735b4576d99ca98059ab43c0d":{"kind":"REWORK","_number":7,"created":"2019-04-24 00:50:27.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/03/619303/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/619303/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/619303/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/619303/7"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2018-11-21 16:26:37.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-04-24 00:49:22.000000000","tz":-300},"subject":"py3: port bulk middleware","message":"py3: port bulk middleware\n\nThis makes the get_response_body() return bytes. It makes little\ndifference in bulk itself, but helps slo a lot.\n\nChange-Id: I3603247e0c36299de2107aa8e494a3f87647696f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a54a53a11899df735b4576d99ca98059ab43c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a54a53a11899df735b4576d99ca98059ab43c0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
