)]}'
{"id":"openstack%2Fswift~620440","triplet_id":"openstack%2Fswift~master~Ic2e29474505426dea77e178bf94d891f150d851b","project":"openstack/swift","branch":"master","topic":"bug/1805539","hashtags":[],"change_id":"Ic2e29474505426dea77e178bf94d891f150d851b","subject":"Fix a potential bug","status":"MERGED","created":"2018-11-28 02:59:51.000000000","updated":"2019-09-05 05:43:29.000000000","submitted":"2019-09-05 05:41:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"620440-1567662104925-07ca9f8f","meta_rev_id":"f7a9fa494ba2dcb11d4e9ca045ce6fb6528828fe","_number":620440,"virtual_id_number":620440,"owner":{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-12-19 07:18:49.000000000","_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"date":"2018-11-28 03:00:07.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"date":"2019-09-04 11:42:59.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"date":"2018-11-28 03:58:47.000000000","_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":2,"date":"2019-09-05 05:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-31 04:40:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-31 04:40:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-28 03:00:07.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2018-11-28 03:58:47.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2018-12-19 07:18:49.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-08-31 04:40:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-09-04 11:42:59.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-09-05 05:41:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30469d84801f4c65b7e43a5d7bcdc78c207314ee","author":{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},"date":"2018-11-28 02:59:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eeb1296eaad0c11fdfa5898974f1d305193b8e97","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2018-11-28 03:00:07.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"e24a58cada61b68186e6c07355ec19f9ad2fc0e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 03:47:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/40/620440/1/check/openstack-tox-docs/9b6e580/html/ : SUCCESS in 5m 44s\n- swift-tox-py27 http://logs.openstack.org/40/620440/1/check/swift-tox-py27/54614d6/ : SUCCESS in 14m 53s\n- swift-tox-py35 http://logs.openstack.org/40/620440/1/check/swift-tox-py35/78f8fbd/ : SUCCESS in 9m 28s\n- swift-tox-py36 http://logs.openstack.org/40/620440/1/check/swift-tox-py36/5dfc518/ : SUCCESS in 7m 54s (non-voting)\n- swift-tox-func http://logs.openstack.org/40/620440/1/check/swift-tox-func/bb779a9/ : SUCCESS in 15m 10s\n- swift-tox-func-encryption http://logs.openstack.org/40/620440/1/check/swift-tox-func-encryption/516cc41/ : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/40/620440/1/check/swift-tox-func-domain-remap-staticweb/0013228/ : SUCCESS in 9m 54s\n- swift-tox-func-ec http://logs.openstack.org/40/620440/1/check/swift-tox-func-ec/f3f3c93/ : SUCCESS in 13m 03s\n- swift-tox-func-s3api http://logs.openstack.org/40/620440/1/check/swift-tox-func-s3api/6071297/ : SUCCESS in 12m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/40/620440/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/6033699/ : SUCCESS in 10m 59s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/40/620440/1/check/swift-probetests-centos-7/07ac664/ : SUCCESS in 29m 00s\n- swift-dsvm-functional http://logs.openstack.org/40/620440/1/check/swift-dsvm-functional/26137f8/ : FAILURE in 42m 05s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/40/620440/1/check/swift-dsvm-functional-ipv6/8a206f7/ : SUCCESS in 22m 45s\n- swift-tox-lower-constraints http://logs.openstack.org/40/620440/1/check/swift-tox-lower-constraints/63437af/ : SUCCESS in 14m 00s\n- openstack-tox-pep8 http://logs.openstack.org/40/620440/1/check/openstack-tox-pep8/b9973f8/ : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/40/620440/1/check/swift-multinode-rolling-upgrade/03407fc/ : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a8bab94ececca12796dd97b06a4aaf4a46aa3e81","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-11-28 03:57:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b211ed20943024d5185c6ded7b8057d021bfb4ae","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-11-28 03:58:22.000000000","message":"Patch Set 1:\n\nArzhana, I think the test failure was something transient related to the environment. I\u0027ve rechecked it, so we\u0027ll see what another pass says. Thanks for your patch","accounts_in_message":[],"_revision_number":1},{"id":"c712696e8229e0d2dce288c9e5694312313cd783","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-11-28 03:58:47.000000000","message":"Patch Set 1:\n\n*Arzhna, sorry","accounts_in_message":[],"_revision_number":1},{"id":"e0b926a7ab8afebdfacb2a17105d5aaddb871e15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 04:41:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/40/620440/1/check/openstack-tox-docs/ac51b8a/html/ : SUCCESS in 9m 06s\n- swift-tox-py27 http://logs.openstack.org/40/620440/1/check/swift-tox-py27/3d3d6fe/ : SUCCESS in 10m 22s\n- swift-tox-py35 http://logs.openstack.org/40/620440/1/check/swift-tox-py35/0ac58b2/ : SUCCESS in 5m 32s\n- swift-tox-py36 http://logs.openstack.org/40/620440/1/check/swift-tox-py36/7457c57/ : SUCCESS in 5m 19s (non-voting)\n- swift-tox-func http://logs.openstack.org/40/620440/1/check/swift-tox-func/056c3f0/ : SUCCESS in 8m 01s\n- swift-tox-func-encryption http://logs.openstack.org/40/620440/1/check/swift-tox-func-encryption/2bbe1ce/ : SUCCESS in 7m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/40/620440/1/check/swift-tox-func-domain-remap-staticweb/9e65787/ : SUCCESS in 9m 20s\n- swift-tox-func-ec http://logs.openstack.org/40/620440/1/check/swift-tox-func-ec/a960834/ : SUCCESS in 9m 24s\n- swift-tox-func-s3api http://logs.openstack.org/40/620440/1/check/swift-tox-func-s3api/ccc4070/ : SUCCESS in 19m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/40/620440/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2b5ebee/ : SUCCESS in 11m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/40/620440/1/check/swift-probetests-centos-7/d097ba4/ : SUCCESS in 25m 53s\n- swift-dsvm-functional http://logs.openstack.org/40/620440/1/check/swift-dsvm-functional/69c7f9a/ : SUCCESS in 23m 08s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/40/620440/1/check/swift-dsvm-functional-ipv6/61b675f/ : SUCCESS in 28m 29s\n- swift-tox-lower-constraints http://logs.openstack.org/40/620440/1/check/swift-tox-lower-constraints/7d6db7d/ : SUCCESS in 38m 40s\n- openstack-tox-pep8 http://logs.openstack.org/40/620440/1/check/openstack-tox-pep8/0f5f527/ : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/40/620440/1/check/swift-multinode-rolling-upgrade/cc6def0/ : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b068d9e296cdac7e0d7698d559e8623aa809531e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-12-19 07:18:49.000000000","message":"Patch Set 1:\n\nThanks Arzhna, good spot. Currently it doesn\u0027t effect anything but yes it\u0027s clearly wrong and could cause an issue in the future. Great stop.\n\nI\u0027ll have a closer look tomorrow and give this patch some love.\n\nThanks so much for getting involved and helping make Swift better!","accounts_in_message":[],"_revision_number":1},{"id":"2e09b137508ba3e9fc2a2e15745c2e308ff4d68c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-31 04:40:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI take it this was just found while reading code? Nice spot!\n\nI really wanted to add a test to verify that we got it right... but found that since we pass the audit_location.path/hash_dir_path as a _datadir kwarg to __init__, we actually *ignore* the partition arg anyway. Sigh.\n\nThis is clearly closer to what was intended, though, and should have merged a while back. Sorry for the delay.","accounts_in_message":[],"_revision_number":1},{"id":"83e7b452551fd46ce0abf162b43592a62a4d7c5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 04:41:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89105a2c5a1fa088a1f290f36f06d2ba97281d05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 05:41:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716b2881e7fa478fa527a5ad48524909 : SUCCESS in 5m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/505c6acabf774e94b9ddac9f1e9f64ed : SUCCESS in 57m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c86343a69acd455891b18b6cad9e8f95 : SUCCESS in 54m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4a1d9a409de84451a63ae3ce285515bb : SUCCESS in 10m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9fa72c23f87a4031aa05aec5495bd7d1 : SUCCESS in 11m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5857524b52a14fda8e0038f6ec990fb1 : SUCCESS in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/40b8acf229f645daaadd2fab028d927a : SUCCESS in 12m 41s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/e2823bc0595441adbb21b53090ee70ef : SUCCESS in 9m 49s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/eb17e694624748e1b28b9fa63d37c4fd : SUCCESS in 10m 06s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/45a1ab827ebd48b192e4021dd3ed83c4 : SUCCESS in 11m 23s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/354ebd2ffc1646eab3423b29eff7c724 : SUCCESS in 10m 04s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/10c613426fca4cdd848d82b267439170 : SUCCESS in 13m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/663fac8fb6424e46b4a2f8c3c213e304 : SUCCESS in 9m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2ca7cbb4ba944077bdedc8f9d6fd2fe4 : SUCCESS in 10m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5c1f865faf324bbd81b2cbf4769af0c4 : SUCCESS in 11m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/34ec5e9fe36a436f90e3b6184c99e841 : SUCCESS in 12m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cb0af1d47b0247c28a6569f4021056d7 : FAILURE in 32m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/747a2b249979459fb7d9fac2ac95601b : SUCCESS in 27m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4d96c90a922c48d49128cbb460c944d0 : SUCCESS in 31m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d67606990284f8c9c47fae32cd7dbe7 : SUCCESS in 29m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecafebe95210440b9cc6008d60aa8f70 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cdb122489eb416b84bd399d695de01f : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a74d09e9d55a024508209249c6798937f02fc4fa","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-09-04 11:42:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7388cf7949d989f5c9a7f02a6d9a5d49f82bb4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddda8f322b0446babef3a97ab88897c5 : SUCCESS in 15m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/950b203cc30443b38913180eb6f1b44f : SUCCESS in 1h 04m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/544d7f89383749318cd6eccd78289422 : SUCCESS in 1h 01m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/57caca33a0744ad2873c4cbd353beff1 : SUCCESS in 10m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3fbefe64ec54706b3a5727485f39b0f : SUCCESS in 23m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e3a39183074da8af24b6746f644c1b : SUCCESS in 18m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7791ad77aa1f4a05879e0a5a99fc1ea3 : SUCCESS in 25m 06s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5e48ca2325124afe8680d9f9def17644 : SUCCESS in 14m 13s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/aa72d2765cdd4a71a50e1f303e68419b : SUCCESS in 14m 12s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/4c4a0013ba294066ac380a4f55f8abdc : SUCCESS in 11m 32s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9f50dad9cdbf49bba48f4b1f87715364 : SUCCESS in 16m 54s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/5da238edb21a4e02867d3dae198974fa : SUCCESS in 12m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/698856fd63c043e7a39e38266453d41c : SUCCESS in 20m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4cb50598ad6641faa32f5936849d63dc : SUCCESS in 11m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0c8f493f73654c03a4d7001dc8021881 : SUCCESS in 11m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9aec5df05e2546f9b7d3db1c39d3d98e : SUCCESS in 13m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cdb01e870e0c436fbb246bc7fe37bbf4 : SUCCESS in 13m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd467a96b7e544fdaf7bbb4ad402cab4 : SUCCESS in 15m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1b1317c34af847dca95976661bc83525 : SUCCESS in 33m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03ac4e9164904ff6bc8faa2012e0de89 : SUCCESS in 28m 33s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/96e5075e89f940dfb45791ea0331e895 : SUCCESS in 30m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b4810629d464f1ba4d985ef7e438a3b : SUCCESS in 29m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8f724dd4bd40a2941cde3dbbb0556d : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc909a2203e4cd9b972f929dc13a92f : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30d49dbaf0c94a888e9695872fdb788a : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":1},{"id":"05fdf00c3868eaec5d2248a5b0a9116acd4d4cbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:42:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f661ab5298be890aa27d359e6504d1b3ee4a3493","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 15:58:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3445dad977746c9bab0893cb19d8e14 : SUCCESS in 5m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/efb140e57e204eb6b50a590d14dd57bd : SUCCESS in 1h 11m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39916220ea6a44a09384438a189d3ada : SUCCESS in 54m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6992e3f40e79488d92ef0f7db050102e : SUCCESS in 10m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8503cbdc75b34ec2bcb920cada42b116 : SUCCESS in 14m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e884b6be12e14f16a76fd51a3b34fcc3 : SUCCESS in 13m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/73a6dfc594674f52823023784be900d0 : TIMED_OUT in 42m 01s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/59f153c809194413b3f962db0be07145 : SUCCESS in 11m 29s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a7cc5df7f3f0454e8bd022465ccf5c5d : SUCCESS in 11m 50s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/fdcb44c378a845298a7fdf8dfc917f84 : SUCCESS in 13m 26s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/10481dd46d2743009da8340788dadbe9 : SUCCESS in 17m 21s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/55a2f8443e3e4e0e97bd7d8e392267c7 : SUCCESS in 14m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b04efb17ddd14895a40c5e3e89a44bb9 : SUCCESS in 12m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/600443069f174f50a5d10557dcc36210 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5e176f9d90ed4009a207eb5e1a518153 : SUCCESS in 10m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8a0dcf98619b459ba2e221c74d99cd82 : SUCCESS in 16m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/22d1989641354a6baea57004f0d0ef0d : SUCCESS in 30m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5623d5db82524bb989f1f77c66c73cb4 : SUCCESS in 26m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3a7ce07ed20b4aa4835ac6fe7e75e871 : SUCCESS in 28m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4f086c6afbc4ece82a7991a5c793cbb : SUCCESS in 25m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c6555009b4e41d19f5cb7addbf4997c : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c5dc2acaaa4dda8efe5ebae05f96a4 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a021e6e94187a676a7f45605ba1196b378b689c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-04 17:09:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c6273e9d31ded6588ca1ac3928c33ccb1d1ad1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:56:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc3b73e1fbc64f77a11bbc49ae395260 : SUCCESS in 6m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1fd6b6a297ba441aae3de378b5147a5c : SUCCESS in 54m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43d39347ccf74ebda7b91aa31ac36b03 : SUCCESS in 50m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a49ae3026035449dae00403e2f7b8099 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bcc0a6ffbfc4b94ba25004bdcedcea3 : SUCCESS in 11m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56a9023944824db88b5bae887baf4337 : SUCCESS in 15m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/782ac46e5fe44f50897fd782b860e2a8 : SUCCESS in 13m 12s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3cea45d233b241d496508f190ea66db5 : SUCCESS in 11m 29s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/0c3fd585ff1740929bbfa173058e4d56 : SUCCESS in 12m 23s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/6a7d196d52f04a229cb1ca63eb070577 : SUCCESS in 9m 16s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/6baeca0d224a4631b90445f72e5950e0 : SUCCESS in 11m 06s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/cce763bbb135427cb83e485d30294eef : SUCCESS in 13m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a2cc698fbf3640a8bb3b6a4e618ac483 : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a1d46539d1134ff59deb5b93dba270b9 : SUCCESS in 12m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/121450554dd841cf904ba6cef628c1e2 : SUCCESS in 11m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cd1804151599445099b58c743838a518 : SUCCESS in 11m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/308aa59915ef407f953c57e9b12566c5 : SUCCESS in 14m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad94d8fa030e4bd29cf1569ae272f2e6 : SUCCESS in 14m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85b0ada9d94d4ffcb861e7383e13b690 : SUCCESS in 36m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76fb51e3744d4f21ae0aab850f5c299d : SUCCESS in 32m 17s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f22a7c3509bf41ba9c5bf1d688077338 : SUCCESS in 28m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e36592ba650f409e828e1032ad8308cd : SUCCESS in 28m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9ade80eb26c49768b2b3177c1643b66 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6646755acccc456fabd55afa124abe58 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ee232027669407bb05e95263d1543c9 : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c4b508a048d4e5293ef05b1cd497eb1ff929f649","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:57:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1838f577531651b76899819b115567094f153cf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:41:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b26548f345f400e913f6d92499bb751 : SUCCESS in 5m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4134fafa911c4978956e72f642461bc6 : SUCCESS in 1h 03m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75771498e72a42508ba20c05b6575682 : SUCCESS in 54m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/600ec6eb28c14411a89981e1003aae88 : SUCCESS in 11m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9beee64e7674078a40fd470fe66bc19 : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73604d7d9c2d469697435cf57229699f : SUCCESS in 13m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/81caf15c01c5492ea55cce60759da899 : SUCCESS in 15m 06s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/591c7cd260834bad9aea692aa0295747 : SUCCESS in 10m 33s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/5240c088ce9f46aab358f78d85245b22 : SUCCESS in 11m 43s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/f65872cf2e274401ba18bc0b83571a5e : SUCCESS in 14m 00s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d1bf7770441a40f7a2ee689f255b47fb : SUCCESS in 11m 12s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/7fcd24b97b3548198f0d7f1d2f890b4b : SUCCESS in 13m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0c5b492d5a9a47ccbf5b40de37bd385a : SUCCESS in 10m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/925d476bc43d492e88f32bb962454f89 : SUCCESS in 10m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/471a53583fd64a859b41526114f3407f : SUCCESS in 11m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/28a7aaace0884aa48fe78a80d90325a7 : SUCCESS in 12m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/643aae486a16488a8dd79331639bfff5 : SUCCESS in 36m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f49491d9e910498bbd19d92d9871f83d : SUCCESS in 33m 43s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b6b800fc5aeb4bcd89eff01d27c64404 : SUCCESS in 30m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad4a45d7c2d54aee9320f8bbd09e2302 : SUCCESS in 27m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55f80f601312422bbaa6faaf2ebea156 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfca39350454f078e1ba3f685c2cf15 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"84814161faaebb104666bcd8273a33916a5a06e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:41:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bedec65b06c3bf34bf4552ed3cc5f1dadc024d29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:43:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef058d3eed24331aa8be15801f69527 : SUCCESS in 1m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1aeb8dc55ce24424a5e6f77d05ee8a42 : SUCCESS in 1m 15s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eed76d8bed446518bff2ca4af18259f7637c430e","revisions":{"eed76d8bed446518bff2ca4af18259f7637c430e":{"kind":"REWORK","_number":1,"created":"2018-11-28 02:59:51.000000000","uploader":{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},"ref":"refs/changes/40/620440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/620440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/620440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/620440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/620440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/620440/1"}}},"commit":{"parents":[{"commit":"e58dce1e4e088dfcf265bd0f2fc92a9046808c8f","subject":"Merge \"Use eventlet.patcher.original to get Python select module in get_hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e58dce1e4e088dfcf265bd0f2fc92a9046808c8f"}]}],"author":{"name":"arzhna","email":"arzhna@gmail.com","date":"2018-11-28 02:15:05.000000000","tz":540},"committer":{"name":"arzhna","email":"arzhna@gmail.com","date":"2018-11-28 02:59:34.000000000","tz":540},"subject":"Fix a potential bug","message":"Fix a potential bug\n\nIn the class method from_hash_dir(), the arguments to input when creating an instance of the BaseDiskFile class are incorrect.\nThe __init__() method of BaseDiskFile class receive the arguments in order of mgr, device_path, partition and etc.\nHowever, in from_hash_dir() method, the order of arguments are mgr, device_path, None and partition\nThe class method from_hash_dir() is used by the Object Auditor.\nIf the partition argument is used in the new DiskFile implementations, exception may occur.\nIt will be cause object auditing to failed and the object will be quarantine by the Object Auditor.\n\nCloses-Bug: #1805539\nChange-Id: Ic2e29474505426dea77e178bf94d891f150d851b\n\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eed76d8bed446518bff2ca4af18259f7637c430e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eed76d8bed446518bff2ca4af18259f7637c430e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
