)]}'
{"id":"openstack%2Fswift~624462","triplet_id":"openstack%2Fswift~master~Ifddbb9df60207dfaaf0434877f4fe6af5c26bdf8","project":"openstack/swift","branch":"master","topic":"603529","hashtags":[],"change_id":"Ifddbb9df60207dfaaf0434877f4fe6af5c26bdf8","subject":"s3token: When caching is enabled, default auth_type to password","status":"MERGED","created":"2018-12-11 18:44:21.000000000","updated":"2020-01-31 19:05:41.000000000","submitted":"2020-01-31 19:01:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"624462-1580497270990-641ca554","meta_rev_id":"853141bae99a75b1b7e40a1f0214cd1cefa5f45f","_number":624462,"virtual_id_number":624462,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 19:01:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-31 17:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-31 17:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 19:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"727fd2e9da9be4e63903e456bc4dcfbddf49937b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-12-11 18:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb1ead50ed965f724e5ad4c06915085c1e2be382","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 19:25:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/62/624462/1/check/openstack-tox-docs/0bf27c8/html/ : SUCCESS in 4m 23s\n- swift-tox-py27 http://logs.openstack.org/62/624462/1/check/swift-tox-py27/ffed7e6/ : SUCCESS in 11m 37s\n- swift-tox-py35 http://logs.openstack.org/62/624462/1/check/swift-tox-py35/b3c09d8/ : SUCCESS in 5m 20s\n- swift-tox-py36 http://logs.openstack.org/62/624462/1/check/swift-tox-py36/da000f9/ : SUCCESS in 5m 33s (non-voting)\n- swift-tox-func http://logs.openstack.org/62/624462/1/check/swift-tox-func/90ba4f6/ : SUCCESS in 8m 43s\n- swift-tox-func-encryption http://logs.openstack.org/62/624462/1/check/swift-tox-func-encryption/802df24/ : SUCCESS in 8m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/62/624462/1/check/swift-tox-func-domain-remap-staticweb/a3446fa/ : SUCCESS in 10m 00s\n- swift-tox-func-ec http://logs.openstack.org/62/624462/1/check/swift-tox-func-ec/2cf5d82/ : SUCCESS in 8m 51s\n- swift-tox-func-s3api http://logs.openstack.org/62/624462/1/check/swift-tox-func-s3api/b27e986/ : SUCCESS in 11m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/62/624462/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c31974d/ : SUCCESS in 10m 42s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/62/624462/1/check/swift-probetests-centos-7/b85b2d1/ : SUCCESS in 29m 34s\n- swift-dsvm-functional http://logs.openstack.org/62/624462/1/check/swift-dsvm-functional/9b3201b/ : SUCCESS in 22m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/62/624462/1/check/swift-dsvm-functional-ipv6/e067905/ : SUCCESS in 30m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/62/624462/1/check/swift-tox-lower-constraints/2d01fa5/ : SUCCESS in 10m 21s\n- openstack-tox-pep8 http://logs.openstack.org/62/624462/1/check/openstack-tox-pep8/4c64e07/ : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/62/624462/1/check/swift-multinode-rolling-upgrade/2f18812/ : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8e30698637e400a5f814bba773adc88c6aed5324","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 17:44:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0375986a7718d3687c09b6af44fac8cad4a9531e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:45:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52e4510d7d185a2c5ca63bfc313b4fddc6aee05d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:01:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b52a116b0b4116bf1f044206aa73d1 : SUCCESS in 9m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/904b6cdea6d047eab3a59d88d886c9fe : SUCCESS in 1h 02m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d1a453549c8421d8771ce12d19b3b17 : SUCCESS in 53m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a35558c28ab640caa3e084b8205c4b47 : SUCCESS in 10m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e042d863448e4df19c394fbfde2b992d : SUCCESS in 10m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/67b3103f26754b889305b23789732f48 : SUCCESS in 17m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00a60007e69349ccaa7fedc5b704fc13 : SUCCESS in 13m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3f4a63db313f4a38a3d147c309f8a22d : SUCCESS in 14m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/af62e81dbeb74be986f20a3c5d6e96e7 : SUCCESS in 20m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f464a068e6554d6ea8a518c1b20346e6 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b02c89889424ce38735be7435ff3965 : SUCCESS in 13m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/286415be522e4e1aa63bae36b3c0c623 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c40000473e004124a11c7a7deeb27d40 : SUCCESS in 12m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/85cbcb1b988f45a0966c5292e79a933f : SUCCESS in 16m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/aa33f474029b4fa2a8eb734f8142caf9 : SUCCESS in 19m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a2fded04681c41059c8b733ec676bf22 : SUCCESS in 11m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/499b7b37ef624654930fa3e5d1dab772 : SUCCESS in 12m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/072422e56189400ebd5b963609c0e551 : SUCCESS in 22m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8495b0b66ce44593a62fd535dd5764ab : SUCCESS in 46m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce37cd29b64f4290b8bb9afb4fa6fc1d : SUCCESS in 31m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2e0f650e62d1469f90fdc716cfc5853d : SUCCESS in 33m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30020c9192e74f79b18dc99abc39d69e : SUCCESS in 36m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbc50015856a430fb4a3e028091d8c13 : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/582b0f2465e44470be17aa5a8137604f : SUCCESS in 11m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8cc5f4b0b0c4cc5bddfd65ec4a6c895 : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7b7941629cfb4e24e25af98e24411556366b74a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:01:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b9f1b9737e3e4b5626b28c00f2a50332d1f238f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7eb8bffeddd4846becec6b91c462eb7 : SUCCESS in 1m 28s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/99520699f27b4652990a26e03fb0e262 : SUCCESS in 59s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8ce4bd599a714060ba73944281716bc2 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec70cf9605752e3b8a5c33ee7ef145f97df7567f","revisions":{"ec70cf9605752e3b8a5c33ee7ef145f97df7567f":{"kind":"REWORK","_number":1,"created":"2018-12-11 18:44:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/624462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/624462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/624462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/624462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/624462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/624462/1"}}},"commit":{"parents":[{"commit":"b0aea93603abd98ce5c7d963e3b8b9b418d9c261","subject":"s3 secret caching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0aea93603abd98ce5c7d963e3b8b9b418d9c261"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-12-10 19:58:18.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-12-11 18:41:36.000000000","tz":0},"subject":"s3token: When caching is enabled, default auth_type to password","message":"s3token: When caching is enabled, default auth_type to password\n\nOtherwise, if you just set secret_cache_duration and skip configuring\nauth_type, you get a not-very-obvious\n\n    TypeError: sequence item 0: expected string, NoneType found\n\nbubbling up from stevedore. Now, you\u0027d get something like\n\n    MissingRequiredOptions: Auth plugin requires parameters which were\n    not given: auth_url\n\nChange-Id: Ifddbb9df60207dfaaf0434877f4fe6af5c26bdf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec70cf9605752e3b8a5c33ee7ef145f97df7567f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec70cf9605752e3b8a5c33ee7ef145f97df7567f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
