)]}'
{"id":"openstack%2Fswift~625082","triplet_id":"openstack%2Fswift~master~I18accffb2cf6ba6a3fff6fd5d95f06a424d1d919","project":"openstack/swift","branch":"master","topic":"sample_config_update","hashtags":[],"change_id":"I18accffb2cf6ba6a3fff6fd5d95f06a424d1d919","subject":"added value and notes to a sample config file for s3token","status":"MERGED","created":"2018-12-13 18:46:01.000000000","updated":"2020-02-04 10:20:02.000000000","submitted":"2020-02-04 10:18:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"625082-1580811512782-ed07d331","meta_rev_id":"e9c75b0fcb6869f613e09d00c29457bf4923300f","_number":625082,"virtual_id_number":625082,"owner":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-04 10:18:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-31 18:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-31 18:48:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-14 05:32:28.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-01-31 18:48:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-02-04 10:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5805ba50f3a219943ac242e0414af215928d99a","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-13 18:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c41995e749c3be2751e0e8d48e3199fe23ebe24b","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-13 18:48:01.000000000","message":"Patch Set 1:\n\nI\u0027d really like to add the config values needed for https://github.com/openstack/swift/blob/master/swift/common/middleware/s3api/s3token.py#L167-L178\n\n...but it\u0027s really hard to find what those should be by digging in the keystoneclient code. Anyone know what they are?\n\nRegardless of adding those, this patch should help anyway.","accounts_in_message":[],"_revision_number":1},{"id":"dad03107845a9e5e8da48a9b6fec23f4ca44f7a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 19:25:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/82/625082/1/check/openstack-tox-docs/29a878d/html/ : SUCCESS in 4m 04s\n- swift-tox-py27 http://logs.openstack.org/82/625082/1/check/swift-tox-py27/0e984f2/ : SUCCESS in 13m 51s\n- swift-tox-py35 http://logs.openstack.org/82/625082/1/check/swift-tox-py35/35e75bc/ : SUCCESS in 7m 20s\n- swift-tox-py36 http://logs.openstack.org/82/625082/1/check/swift-tox-py36/1e4e1ca/ : SUCCESS in 5m 36s (non-voting)\n- swift-tox-func http://logs.openstack.org/82/625082/1/check/swift-tox-func/a7fa8f3/ : SUCCESS in 8m 41s\n- swift-tox-func-encryption http://logs.openstack.org/82/625082/1/check/swift-tox-func-encryption/38b3f07/ : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/625082/1/check/swift-tox-func-domain-remap-staticweb/50d1387/ : SUCCESS in 8m 08s\n- swift-tox-func-ec http://logs.openstack.org/82/625082/1/check/swift-tox-func-ec/2b35cb5/ : SUCCESS in 9m 48s\n- swift-tox-func-s3api http://logs.openstack.org/82/625082/1/check/swift-tox-func-s3api/e27a4ea/ : SUCCESS in 11m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/82/625082/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2d5cfdc/ : SUCCESS in 12m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/82/625082/1/check/swift-probetests-centos-7/97d11f4/ : SUCCESS in 31m 27s\n- swift-dsvm-functional http://logs.openstack.org/82/625082/1/check/swift-dsvm-functional/3ddee68/ : SUCCESS in 23m 01s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/625082/1/check/swift-dsvm-functional-ipv6/614a387/ : SUCCESS in 25m 37s\n- swift-tox-lower-constraints http://logs.openstack.org/82/625082/1/check/swift-tox-lower-constraints/9e8f691/ : SUCCESS in 11m 12s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/82/625082/1/check/swift-multinode-rolling-upgrade/0c746eb/ : SUCCESS in 16m 57s","accounts_in_message":[],"_revision_number":1},{"id":"029915f9fc72b574e6f12467ea2717aa2b42e899","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-12-13 19:42:07.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nGo merge https://review.openstack.org/#/c/624462/ so if users *only* configure this, our error won\u0027t suck.\n\nOn the config values, it should more or less match your creds in auth_token, though permissions may need to be a bit different.","accounts_in_message":[],"_revision_number":1},{"id":"b208006182146ff2617f9b730760badcfba8558b","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2018-12-14 05:27:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9f67694eed75aef240dd0ad76615ca61367259d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2018-12-14 05:32:28.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"3cdc7b4ddeb242e42317d3af1fad53131884ba3d","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-14 18:03:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9865d9ad1eed18b88199fcb1911d20f1b83e9d29","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-14 18:04:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"272fb9f3771314dc292593273c37bd28a453924b","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-14 18:13:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"88927405edd9afba95bc296ffa58d03f7571af65","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-14 18:13:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ebfce04c0c4a585ebb0a1d70db356acb57ea448","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 18:54:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/82/625082/4/check/openstack-tox-docs/d6d082b/html/ : SUCCESS in 3m 59s\n- swift-tox-py27 http://logs.openstack.org/82/625082/4/check/swift-tox-py27/a3e1a9e/ : SUCCESS in 10m 08s\n- swift-tox-py35 http://logs.openstack.org/82/625082/4/check/swift-tox-py35/3ab8be1/ : SUCCESS in 5m 31s\n- swift-tox-py36 http://logs.openstack.org/82/625082/4/check/swift-tox-py36/1a36013/ : SUCCESS in 5m 01s (non-voting)\n- swift-tox-func http://logs.openstack.org/82/625082/4/check/swift-tox-func/2b738d9/ : SUCCESS in 9m 18s\n- swift-tox-func-encryption http://logs.openstack.org/82/625082/4/check/swift-tox-func-encryption/3716fcb/ : SUCCESS in 18m 05s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/625082/4/check/swift-tox-func-domain-remap-staticweb/3040e79/ : SUCCESS in 16m 14s\n- swift-tox-func-ec http://logs.openstack.org/82/625082/4/check/swift-tox-func-ec/04724c7/ : SUCCESS in 16m 37s\n- swift-tox-func-s3api http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api/65d52a6/ : SUCCESS in 10m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/31ad2dc/ : SUCCESS in 12m 34s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/82/625082/4/check/swift-probetests-centos-7/94f55d3/ : FAILURE in 31m 45s\n- swift-dsvm-functional http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional/3d79b6e/ : SUCCESS in 28m 12s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional-ipv6/11260d3/ : SUCCESS in 29m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/82/625082/4/check/swift-tox-lower-constraints/f4ac74b/ : SUCCESS in 10m 31s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/82/625082/4/check/swift-multinode-rolling-upgrade/47d886c/ : FAILURE in 8m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a7b327e0c0bb7b69755d0c159d5299ee0a821dc6","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-15 00:49:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c16db62a2d7e89e4ee2bf75f5f2919469de26d04","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-15 00:49:55.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"16b954eaa7c99edc94d96462f76f40ed84444afd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-15 01:37:48.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py27-centos-7 http://logs.openstack.org/82/625082/4/experimental/swift-tox-py27-centos-7/90bf034/ : SUCCESS in 12m 17s\n- swift-tox-func-centos-7 http://logs.openstack.org/82/625082/4/experimental/swift-tox-func-centos-7/06ac9bd/ : SUCCESS in 9m 28s\n- swift-tox-func-encryption-centos-7 http://logs.openstack.org/82/625082/4/experimental/swift-tox-func-encryption-centos-7/666ba04/ : SUCCESS in 11m 38s\n- swift-tox-func-ec-centos-7 http://logs.openstack.org/82/625082/4/experimental/swift-tox-func-ec-centos-7/53d4973/ : SUCCESS in 12m 38s\n- swift-multinode-rolling-upgrade-queens http://logs.openstack.org/82/625082/4/experimental/swift-multinode-rolling-upgrade-queens/ea4f70d/ : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":4},{"id":"86a820abc8604d9ba370ab8350edd444ce727110","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-15 01:50:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/82/625082/4/check/openstack-tox-docs/39a228b/html/ : SUCCESS in 3m 35s\n- swift-tox-py27 http://logs.openstack.org/82/625082/4/check/swift-tox-py27/ba15005/ : SUCCESS in 15m 40s\n- swift-tox-py35 http://logs.openstack.org/82/625082/4/check/swift-tox-py35/5c52d18/ : SUCCESS in 7m 12s\n- swift-tox-py36 http://logs.openstack.org/82/625082/4/check/swift-tox-py36/502621d/ : SUCCESS in 5m 05s (non-voting)\n- swift-tox-func http://logs.openstack.org/82/625082/4/check/swift-tox-func/74f2ceb/ : SUCCESS in 13m 29s\n- swift-tox-func-encryption http://logs.openstack.org/82/625082/4/check/swift-tox-func-encryption/d484523/ : SUCCESS in 13m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/625082/4/check/swift-tox-func-domain-remap-staticweb/7ad12c6/ : SUCCESS in 14m 02s\n- swift-tox-func-ec http://logs.openstack.org/82/625082/4/check/swift-tox-func-ec/1b797e2/ : SUCCESS in 9m 08s\n- swift-tox-func-s3api http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api/0496e28/ : SUCCESS in 15m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/190f911/ : SUCCESS in 12m 55s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/82/625082/4/check/swift-probetests-centos-7/46366de/ : SUCCESS in 32m 50s\n- swift-dsvm-functional http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional/ffc6cc1/ : SUCCESS in 24m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional-ipv6/8f07cf1/ : SUCCESS in 21m 36s\n- swift-tox-lower-constraints http://logs.openstack.org/82/625082/4/check/swift-tox-lower-constraints/ac1b275/ : SUCCESS in 10m 25s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/82/625082/4/check/swift-multinode-rolling-upgrade/527f003/ : FAILURE in 7m 34s","accounts_in_message":[],"_revision_number":4},{"id":"651890ad8bbb292eed1fe32b3b702667b1dfd376","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-15 18:33:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1a5418751a646e1c8f49e3b15fedbe8f36f56087","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-15 19:43:16.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/82/625082/4/check/openstack-tox-docs/361e05d/html/ : SUCCESS in 5m 00s\n- swift-tox-py27 http://logs.openstack.org/82/625082/4/check/swift-tox-py27/e4988c0/ : SUCCESS in 11m 01s\n- swift-tox-py35 http://logs.openstack.org/82/625082/4/check/swift-tox-py35/9d31795/ : SUCCESS in 4m 58s\n- swift-tox-py36 http://logs.openstack.org/82/625082/4/check/swift-tox-py36/6dffe0d/ : SUCCESS in 5m 14s (non-voting)\n- swift-tox-func http://logs.openstack.org/82/625082/4/check/swift-tox-func/1b21b09/ : SUCCESS in 8m 04s\n- swift-tox-func-encryption http://logs.openstack.org/82/625082/4/check/swift-tox-func-encryption/5be40c4/ : SUCCESS in 8m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/625082/4/check/swift-tox-func-domain-remap-staticweb/beda824/ : SUCCESS in 7m 39s\n- swift-tox-func-ec http://logs.openstack.org/82/625082/4/check/swift-tox-func-ec/a874776/ : FAILURE in 13m 27s\n- swift-tox-func-s3api http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api/bef87d4/ : SUCCESS in 10m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/de9a63a/ : SUCCESS in 14m 21s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/82/625082/4/check/swift-probetests-centos-7/fef0e96/ : FAILURE in 39m 49s\n- swift-dsvm-functional http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional/fae3646/ : SUCCESS in 24m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional-ipv6/4d6cf83/ : SUCCESS in 24m 25s\n- swift-tox-lower-constraints http://logs.openstack.org/82/625082/4/check/swift-tox-lower-constraints/6427aea/ : SUCCESS in 10m 21s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/82/625082/4/check/swift-multinode-rolling-upgrade/5b6be79/ : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":4},{"id":"9d56cd8cd21837b8ef295eeb8b0a4888c2ad27d9","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2018-12-16 19:54:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d106ee1f156bad1826b32c1645f206d338aeaa30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-16 20:43:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/82/625082/4/check/openstack-tox-docs/99cb75d/html/ : SUCCESS in 3m 50s\n- swift-tox-py27 http://logs.openstack.org/82/625082/4/check/swift-tox-py27/3be575c/ : SUCCESS in 22m 20s\n- swift-tox-py35 http://logs.openstack.org/82/625082/4/check/swift-tox-py35/b2b7ae6/ : SUCCESS in 5m 02s\n- swift-tox-py36 http://logs.openstack.org/82/625082/4/check/swift-tox-py36/0ead32b/ : SUCCESS in 4m 54s (non-voting)\n- swift-tox-func http://logs.openstack.org/82/625082/4/check/swift-tox-func/a63eb27/ : SUCCESS in 19m 46s\n- swift-tox-func-encryption http://logs.openstack.org/82/625082/4/check/swift-tox-func-encryption/8131662/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/625082/4/check/swift-tox-func-domain-remap-staticweb/93a5f0d/ : SUCCESS in 23m 28s\n- swift-tox-func-ec http://logs.openstack.org/82/625082/4/check/swift-tox-func-ec/24b6965/ : SUCCESS in 8m 30s\n- swift-tox-func-s3api http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api/1e124be/ : SUCCESS in 9m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/82/625082/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a11fda8/ : SUCCESS in 14m 39s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/82/625082/4/check/swift-probetests-centos-7/26bd5a4/ : SUCCESS in 31m 04s\n- swift-dsvm-functional http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional/d8b8035/ : SUCCESS in 23m 30s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/625082/4/check/swift-dsvm-functional-ipv6/5ca1533/ : SUCCESS in 40m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/82/625082/4/check/swift-tox-lower-constraints/9ae6b0e/ : SUCCESS in 9m 56s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/82/625082/4/check/swift-multinode-rolling-upgrade/c7613bd/ : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":4},{"id":"a5cc2be9c3419852c86ab846799651002e88108d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-12-20 16:18:27.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"788c419999ff193682ca232fa2435d714bdd8d25","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 18:48:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"30feb6e0e14c80b644b6c0b6ec83a095f442b651","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 18:48:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7290aa66ef67af70bdff9f0e5367c4617b742bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 20:10:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e569d3d2157a4962a68208d1223f3766 : SUCCESS in 9m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/213144bbceda44c88aca2172d1afaa65 : SUCCESS in 59m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd0e50f5f78d4343840af91a0b208b3c : SUCCESS in 50m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40d8e1225b3240bd86664c10452c58cb : SUCCESS in 11m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/239d152d6faf4a4b88f61e5e59b1573c : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7355b63586c54539a2431a798c627162 : SUCCESS in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6a8f88862bfa493caf83d4feccffd4cf : SUCCESS in 12m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15a44cbca1104b0da6428743205a8094 : SUCCESS in 18m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9ba64e3e54e349cb8f4b10e446998131 : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5009b3227aa46d2a03571847ffb551c : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e99eb9d37d5b4547a9f379b8bc3be123 : SUCCESS in 13m 13s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ebe774b1a89f4df98b4b2e98fc57869d : SUCCESS in 29m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed8b15075eb74912807c1bb6a474f9a0 : SUCCESS in 11m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3ac3d6a200d24d0982fb0b065b61b80c : SUCCESS in 17m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1e00f8c73167429285731e881bdc84be : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e6a169b131314b3dbe8da4b7b6410c0b : SUCCESS in 13m 20s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e8783419e93e4af3bbd2d0c8021d7015 : SUCCESS in 12m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2152e181a14347e9abf325c70a8e3b55 : SUCCESS in 13m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6830f6edb194412f8d36120677f5b623 : SUCCESS in 18m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c35c053a563b4a3bbd1820703f7d55ce : SUCCESS in 19m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d8ae94c414a9430686eed00dfdd2d41e : SUCCESS in 49m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8cb64944d87b40918cce5efa506bae77 : SUCCESS in 41m 15s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7dd426564b614dfe83f36da255b44d78 : SUCCESS in 28m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb14c5398b854fe6850e0bab1e67995b : SUCCESS in 31m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73707fa4e8484ed2ba7f2d42bc59a4fe : SUCCESS in 14m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba92d81ab7a243b4bb81ef9d6d5309df : SUCCESS in 21m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed70e8b7945740a8bd7a2e8b60202365 : SUCCESS in 1h 15m 44s","accounts_in_message":[],"_revision_number":5},{"id":"75e470b7b6f5cabb8c17a4a354ab8525070c7ed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 20:10:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ccf039e6a6d6527e39c171fda681701b49191682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 21:26:39.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53640f8e60f54715a649be4b798e55de : SUCCESS in 9m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d370bc57b1554735801bee1396be8c82 : SUCCESS in 1h 00m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c226c9691b94472eae7f6c59df4d973f : SUCCESS in 53m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/45a7f851cd7445d38aaa8654ff84b200 : SUCCESS in 11m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8cf081b9fd1a4bfc8d4dddcff4f2df8d : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94acfa2bbf80488ca2ed1347e694bdbf : SUCCESS in 15m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/147b62b9b15b4e30942b035f8ea8a6e3 : SUCCESS in 13m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/31921985391f46ce9278ebb44580a7ed : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bc64b333d1474b66b807158f4685c528 : SUCCESS in 13m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39ac5ebdedfa4d45a0787e396d4e0112 : SUCCESS in 12m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5a4db001f9ab4f3a888e13d86ac8f85f : SUCCESS in 17m 55s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f3ebfd399e694d878271fdc106e26641 : SUCCESS in 13m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8eebfed93f8a49e1afb7c5443cc00fd8 : SUCCESS in 11m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f477cd061472485b891b7e9d5b231b16 : SUCCESS in 20m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b26c243834e14432bb785be9cc9a90b6 : SUCCESS in 12m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8112c83c1946463994479c3d5ca10f7f : SUCCESS in 13m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1fbe6552a4cd4336854b3e78abf3dc82 : SUCCESS in 14m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/10eb58f2a0644acfb97cbc8e76c049a5 : SUCCESS in 17m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d30b670463d04260b802fae17384645e : TIMED_OUT in 1h 01m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9e209f908be4a8eae2a8d417f6c3241 : SUCCESS in 30m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f40d8c013fac43f78078bd117aef84ef : SUCCESS in 34m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f31007768ede482393fcba9b2438b289 : SUCCESS in 45m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec7ec2c76b9c428caafd2a117b9747ba : SUCCESS in 15m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/578a5a0bed2446eb88b7a657e5e044a8 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":5},{"id":"c399a806a87dbd2a696c0773078bfc5655d2fd25","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 21:40:20.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b0db34e76db68fe3deb7796513a12768c8acb5c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 22:49:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7b22cec69341719fd8356c31dfb098 : SUCCESS in 7m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/97a39fd0e406493a8933cd2195145c51 : SUCCESS in 1h 01m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2bd384e7bb234cc1a0ec4add7fa01daf : SUCCESS in 54m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8a50a06d2f94f95b56c8e5c8b8dac1f : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad5822daa98b445894c5c2fec0053249 : SUCCESS in 8m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d31a61f66854a468c98fcd217f17f00 : SUCCESS in 15m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/782a84fc1b8f4b248a428dbbc77e3ff6 : SUCCESS in 16m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b822d9412f1943b2a2a655b6c4c9efaa : SUCCESS in 15m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1fcfe111c4814b339bba288b20c8c9b6 : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49f5cad6e305487e969468a088533a99 : SUCCESS in 12m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/190dc3d6a7404f8d9de6efc703a97607 : SUCCESS in 13m 08s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b2281bc926ae401a9b94ee0836cca0c0 : SUCCESS in 9m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/754478c984c64f87bf4cf24d53538f0d : SUCCESS in 10m 09s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/b660e150d75842278387fb5cde7c7aa9 : SUCCESS in 19m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0ecc74009d1048bd9fccab8e0dae6454 : SUCCESS in 11m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8490d0e19fce4807a2e44218b726465f : SUCCESS in 16m 00s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1145ceb0e34b4626acca410da9cd8be7 : SUCCESS in 9m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8d66d1f9e06c49739fa92f34df897f23 : TIMED_OUT in 41m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/119d789421d647ba8bf9ccea493fb640 : TIMED_OUT in 40m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8555f4d3eeca471284626e1f8b7d9cad : TIMED_OUT in 41m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9a79ad8a5b794feca7ccf65e5f301326 : SUCCESS in 53m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2162a47fa16a4cdc992fc16cb9aa49da : SUCCESS in 29m 00s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/62a6d0ffccf14749b056dcd7013e7ada : TIMED_OUT in 47m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e656600c89a240bfa12952c379a2005d : SUCCESS in 32m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/075dcc37070d45e8aa78c5e3c29c9aa7 : SUCCESS in 13m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3bc43e4f8fd74f878a3c663a87202c3f : SUCCESS in 21m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db1e09cb8c7241e4b74c85b5d753f8ca : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":5},{"id":"129b3827e87f390610bdd63b182cfb5f596c59f5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 23:00:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d1aadb021adce583396829add2f34a750f1b83d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 00:12:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e614f29469ad4b16a8361c940c4834d5 : SUCCESS in 8m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9c9668f38fe645a5a083ac01206fbaac : SUCCESS in 56m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd54ab26952d4f239ec18eb7f084f59c : SUCCESS in 49m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a0cb90d8bd404fa58d1286e618731d88 : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/198a6e9969a24ec5a237db2f29e95003 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc3932449a544f8e9eb19b517f7af73b : SUCCESS in 15m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eef0930e233846b5b7f84c5e735f22ea : FAILURE in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8e35070b59242e1b94829ca1158bab8 : SUCCESS in 15m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/085ce493367a419d8ba01f2580daafbb : SUCCESS in 15m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af7d7630915d420c92dabb139693290d : SUCCESS in 12m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d5fc79b4c9249be804c5795d2f5325c : SUCCESS in 14m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/057f94c943f544de962ea4e1eba39c70 : SUCCESS in 11m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7da136dd51144beb9f36a4c31ffb2728 : SUCCESS in 13m 02s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f52031b65cec49caa34a5e6f62373818 : SUCCESS in 19m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c05309c460a04f46b08be22cfea792c3 : SUCCESS in 12m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/93ecd0be0df045c49ce0c1af908f17eb : SUCCESS in 16m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3340044d43ef48c0b134aecb4e66603d : SUCCESS in 11m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cb4862e70f474eb1b26bb4a1b8c0b01c : SUCCESS in 13m 02s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4ab9546b96f54172a3f056a5907ad891 : SUCCESS in 19m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/72725c455c4e4a11b4cfa47ce78434dd : SUCCESS in 25m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/13d08d3139bb4b1c947fa52c38af541a : SUCCESS in 44m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a85344efe7f4eb0a15cabf7ab3206c9 : SUCCESS in 29m 15s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/59797b7d4eb943968a5bd4f7061e44c9 : SUCCESS in 36m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10aa6608dea645ba9988d031de835660 : SUCCESS in 30m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9440aba707b47ef90792ad50d674e15 : SUCCESS in 14m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30d0f4cdc6964030ab70c4beab9b91b7 : SUCCESS in 17m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b8c02d8d54b4aacbe7b7ee62e526634 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":5},{"id":"6f591b0b366562ab0d34a5aab81923d463e2074f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-01 00:15:01.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ba62e123ed5858b4ff8dd284dd61a717e578af7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 01:28:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5449df8f540a454d9d3d4174114cd8ea : SUCCESS in 8m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4e1684a8aefd41fab394feacf78392d3 : SUCCESS in 1h 12m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68e6d9557b204c34bfc32c1b9a06d6f7 : SUCCESS in 1h 01m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35451ae3754d4c0dafc989abbf6f5d7e : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d6916339a2eb4587afea23dae87ec1aa : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0cea400415c43c29fca768642538422 : SUCCESS in 19m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f33b3c768d474e0bbcb24176a706c9cc : SUCCESS in 11m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4189ae49eed44631a04139efadc71df3 : SUCCESS in 18m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6fb1f7d101754f2ea2b4c1f33a71928a : SUCCESS in 14m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/89b3635f5004487da5965beb4f84f7ae : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b1ff6e01639b42ff99544e89adc07bd4 : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/587c500fa81e47308dbdd6996fc07845 : SUCCESS in 12m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dfd709e0e05a48b19cda389b7ed3b3e2 : SUCCESS in 19m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ee6ed06511304a5baffc1931a4a318c8 : FAILURE in 23m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dcc977651c5044b3a16656791cf01abc : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7feca4ee113f4264bdb49053b040e9fd : SUCCESS in 22m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e8f4f766d40c47c1b0413a349ebcd081 : SUCCESS in 9m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/85ec2a36ac6a49dbb1950560e35765ed : SUCCESS in 21m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/bad1fb14210f4479baa98131f63373aa : SUCCESS in 14m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18aba6a4fe494a51b6952beb3d8adb95 : SUCCESS in 17m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/29926774eadd4eef88a46bc80898efcb : SUCCESS in 55m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ba6344fbd4504164ac65225372e51b4d : SUCCESS in 33m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fe9717d761cc446585773bf5003f6999 : SUCCESS in 30m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/844bf86b315a4405abed024b9ab12578 : SUCCESS in 38m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aed086df8394f31b94807964a38b03a : SUCCESS in 12m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a55cf905034043cfa25fc07126fbee13 : SUCCESS in 29m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b48ea50da984817b6b548d27736ec83 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":5},{"id":"bf038d5899bcca6e455a5df1fcef017238d466c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-03 21:57:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"869bdc9a9ec9d22d5fa69fb473fb0409e7648bf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 00:01:04.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c36472beb34c23bd7582cde7c68fca : SUCCESS in 11m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7942997095da439b8b4cba51602fc35a : FAILURE in 37m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c479273188364685b5d60413ad7d8957 : SUCCESS in 52m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47b25559881743a7b22120db45b132df : SUCCESS in 10m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f4d4f65e962a45659c086a9d332c3699 : SUCCESS in 11m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8052d7a374dd4697802a0a65eb965a58 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a47c9ebd778e45f394c4c4bb0841deef : SUCCESS in 14m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/116942df6b924ea5aa7d54ebe6c543de : SUCCESS in 13m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/98577647b26e4e5b8dd83732030ee7eb : SUCCESS in 19m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5386e5fdac50430a9a97dbca99fa8680 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad05f15405b5435c873adeaba7977dd0 : SUCCESS in 16m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/296a28d9a16a4c6581214de57028d056 : SUCCESS in 10m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b43cb38e5284f988d13b168bb176251 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c8fda7a2c9a44fb7a9859b9ddad87df5 : SUCCESS in 27m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4ae57f12436a40a9b15f49139f5af978 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/898a4171d842446282a52881fdb9e744 : SUCCESS in 15m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d4480fb080224386adfbe0dbf073a872 : SUCCESS in 12m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/54fe547f91f64849b965ab1d2bb2df79 : SUCCESS in 12m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a481dcfb512344f1800e03a1339df88b : SUCCESS in 15m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8bd8f4c231a24b8b9ab95707166077f0 : SUCCESS in 25m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1d2c1852278468d8da2305f58b24ff6 : SUCCESS in 46m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e57b94814b84e8e902301d43dad1ba5 : SUCCESS in 30m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c56b8ca670f14b21ab75b00d9bd63bcd : SUCCESS in 39m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3515d7025b8a40ff8f1b5198fb5852e2 : SUCCESS in 35m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/485b0e8ee56540f2a4f5d3561fa082a4 : SUCCESS in 17m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/282f1c9502e24f3ca53bd66f6b4a2afb : SUCCESS in 31m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18aa8370fa2a4c3ca4dc66cf76887037 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":5},{"id":"0952a408912dc7654145d7af5b7e4f08c22415b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-04 00:22:13.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"def5fd2b2f1eea77bffec30ce28bcc36eff87d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 01:33:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93826a47a60e42be893055099853e527 : SUCCESS in 8m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9036baea58fe49b78b5aa66642ed155c : SUCCESS in 54m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b460a22970fc4e35a6e95c6a033be402 : SUCCESS in 51m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/883adb4292e5413bb30d7e009c62c7f5 : SUCCESS in 10m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/84c8ed99fb7a4eb998edc30291a88eea : SUCCESS in 10m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bada33c8a25a435f973b7305349e420b : SUCCESS in 14m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6fc1e84679be45b993f035f090c77aa4 : SUCCESS in 15m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ab577f6c37a4684a3f0b4886791312a : SUCCESS in 15m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2a0d85aecbc34d72aaf8cc77a89865f6 : SUCCESS in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d89d79cd622c41579c226d7d6ac81d1d : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/afd276f66ddf4cfc886c1e68659ab396 : SUCCESS in 12m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4a1a3eccc379458196d27bc168eb3689 : SUCCESS in 10m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f31651275501482a81d8d53f74883c57 : SUCCESS in 11m 00s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2e823a3e81af444a8a7674ca819503a0 : SUCCESS in 18m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f07a5da29fce4d0fbfbcc5acc81059c7 : SUCCESS in 10m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bce2f79763d945f2b9dc6841be7e1d4c : SUCCESS in 13m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/69511e8944ac4e46b9f0dc8be9f32def : SUCCESS in 10m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/312b1f30657445e5bd8f4da46b066f39 : SUCCESS in 10m 40s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a1f0f0cd0f3f4f87a897400a5b262d29 : SUCCESS in 16m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75e0a0ce744543cb9323c96d1fc86a7d : SUCCESS in 18m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/14b5fdbf9a324d7bbf1f735be5d31ff6 : SUCCESS in 44m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1cdada958cf4453b9875718fb4910a0d : SUCCESS in 34m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b59519d870f742d7b4782591203cd05d : SUCCESS in 27m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b9b23899a0145459a4bb0b385a67abd : SUCCESS in 29m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/208db0b482c44674a49ca246e8202555 : SUCCESS in 11m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff99e372dd0d42fc8a690d6fff5fe07f : SUCCESS in 21m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c32ad6dd16cc4bd893f38c05e5761269 : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":5},{"id":"dfd940618f40fe1ff64751e9fbf4e7c06f07f2f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 01:33:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8536466d1a096b5baf5a37de0f51cfd272de59b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 10:18:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5406c2051c8177996fb08df42d4d24b9588f6d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 10:18:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954b2880db364f5b907e63eb1bca2b5d : SUCCESS in 10m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/45b75040ddd64eceb637b9502b72f593 : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c32111c58ad4358b0a23ab5f7e0bb19 : SUCCESS in 59m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/85f538928a27492e8b6d48e97305cb1e : POST_FAILURE in 13m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1659dba6f75b407c888a3a39a52b94d7 : SUCCESS in 11m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2509785224874e7cb39811cd61ea4df9 : SUCCESS in 14m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5edd918c40f0482784fd428ebd685649 : SUCCESS in 14m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7462883781a743d798b49a2026ae464f : SUCCESS in 16m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df7baf765bbc4b05bff39e973f2868c2 : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/142dfe8b09cd4778b19045c930734f95 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ffdcf7b946f74492bebf2933377b2787 : SUCCESS in 16m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ee4802a41d304f40a4451714344fa7c6 : SUCCESS in 12m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/025c4f18d67f42f4b9226f0cded69f34 : SUCCESS in 13m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c0e3139ed4f547b0bec238869ad7516e : SUCCESS in 18m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4c0a2fab77dc46cdbefb9761d98b6973 : SUCCESS in 12m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f425aa57c2a949118b4f83e7611e5be4 : SUCCESS in 13m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8124d26cf0124edaa3deaca19a1d2dfd : SUCCESS in 14m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/66acb6b196ff45608f138f199131ffcf : SUCCESS in 19m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dabc5981333b424baca08c9f6d4c3867 : SUCCESS in 52m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82842905b5ca42ff8f655d483f74def6 : SUCCESS in 34m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ee0fef45b1e04dc99d797fdc35460891 : SUCCESS in 26m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7e81b62a600475091a5a54437416755 : SUCCESS in 29m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9afcca1d56594d75930a3d471c6cbf53 : SUCCESS in 15m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e332a34979a4f8f8fe22a504b504a24 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":5},{"id":"b9227be306fc38b5f920f8c8da0baf9f418e00a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 10:20:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054d459b4aa14b0f8969099add1cc868 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/00bba8c19d744fabb77eb307f0f7f46a : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d5a4d3159fc74a048fe52c6e8eaf9d97 : SUCCESS in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d358b9130d80f5fff492de03abf74676a06e61c3","revisions":{"4eca1bb7f9b6df2f49293284b9d95488219b9445":{"kind":"REWORK","_number":1,"created":"2018-12-13 18:46:01.000000000","uploader":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"ref":"refs/changes/82/625082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/625082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/625082/1"}}},"commit":{"parents":[{"commit":"b9d2c08e8d37fd1507ef27c1d641aec6b23ca312","subject":"Merge \"Fix SSYNC concurrency on partition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9d2c08e8d37fd1507ef27c1d641aec6b23ca312"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-13 18:45:51.000000000","tz":-480},"committer":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-13 18:45:51.000000000","tz":-480},"subject":"added value and notes to a sample config file for s3token","message":"added value and notes to a sample config file for s3token\n\nChange-Id: I18accffb2cf6ba6a3fff6fd5d95f06a424d1d919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eca1bb7f9b6df2f49293284b9d95488219b9445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eca1bb7f9b6df2f49293284b9d95488219b9445"}]},"branch":"refs/heads/master"},"2c26c5fc0b143782a3bc479bc70adf624bf9aed5":{"kind":"REWORK","_number":2,"created":"2018-12-14 18:03:26.000000000","uploader":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"ref":"refs/changes/82/625082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/625082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/625082/2"}}},"commit":{"parents":[{"commit":"b9d2c08e8d37fd1507ef27c1d641aec6b23ca312","subject":"Merge \"Fix SSYNC concurrency on partition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9d2c08e8d37fd1507ef27c1d641aec6b23ca312"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-13 18:45:51.000000000","tz":-480},"committer":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-14 18:03:14.000000000","tz":-480},"subject":"added value and notes to a sample config file for s3token","message":"added value and notes to a sample config file for s3token\n\nChange-Id: I18accffb2cf6ba6a3fff6fd5d95f06a424d1d919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c26c5fc0b143782a3bc479bc70adf624bf9aed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c26c5fc0b143782a3bc479bc70adf624bf9aed5"}]},"branch":"refs/heads/master"},"6584d7e564fc7dc494790e5a8e0d1bf92642bfba":{"kind":"REWORK","_number":3,"created":"2018-12-14 18:13:15.000000000","uploader":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"ref":"refs/changes/82/625082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/625082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/625082/3"}}},"commit":{"parents":[{"commit":"b9d2c08e8d37fd1507ef27c1d641aec6b23ca312","subject":"Merge \"Fix SSYNC concurrency on partition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9d2c08e8d37fd1507ef27c1d641aec6b23ca312"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-13 18:45:51.000000000","tz":-480},"committer":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-14 18:13:06.000000000","tz":-480},"subject":"added value and notes to a sample config file for s3token","message":"added value and notes to a sample config file for s3token\n\nChange-Id: I18accffb2cf6ba6a3fff6fd5d95f06a424d1d919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6584d7e564fc7dc494790e5a8e0d1bf92642bfba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6584d7e564fc7dc494790e5a8e0d1bf92642bfba"}]},"branch":"refs/heads/master"},"c6ac13b0cc4afac13b3a47b4dd920bd6da8d7a68":{"kind":"REWORK","_number":4,"created":"2018-12-14 18:13:44.000000000","uploader":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"ref":"refs/changes/82/625082/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/625082/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/625082/4"}}},"commit":{"parents":[{"commit":"b9d2c08e8d37fd1507ef27c1d641aec6b23ca312","subject":"Merge \"Fix SSYNC concurrency on partition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9d2c08e8d37fd1507ef27c1d641aec6b23ca312"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-13 18:45:51.000000000","tz":-480},"committer":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-14 18:13:37.000000000","tz":-480},"subject":"added value and notes to a sample config file for s3token","message":"added value and notes to a sample config file for s3token\n\nChange-Id: I18accffb2cf6ba6a3fff6fd5d95f06a424d1d919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6ac13b0cc4afac13b3a47b4dd920bd6da8d7a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6ac13b0cc4afac13b3a47b4dd920bd6da8d7a68"}]},"branch":"refs/heads/master"},"d358b9130d80f5fff492de03abf74676a06e61c3":{"kind":"REWORK","_number":5,"created":"2020-01-31 18:48:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/625082/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/625082/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/625082/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/625082/5"}}},"commit":{"parents":[{"commit":"5cd854b783ce80356ec62e709f0817cbe699c8cd","subject":"Authors/changelog for swift 2.24.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cd854b783ce80356ec62e709f0817cbe699c8cd"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2018-12-13 18:45:51.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-31 18:47:58.000000000","tz":-480},"subject":"added value and notes to a sample config file for s3token","message":"added value and notes to a sample config file for s3token\n\nChange-Id: I18accffb2cf6ba6a3fff6fd5d95f06a424d1d919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d358b9130d80f5fff492de03abf74676a06e61c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d358b9130d80f5fff492de03abf74676a06e61c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
