)]}'
{"id":"openstack%2Fswift~626663","triplet_id":"openstack%2Fswift~master~Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c","project":"openstack/swift","branch":"master","topic":"add_rolling_test","hashtags":[],"change_id":"Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c","subject":"Add rolling upgrade tests","status":"NEW","created":"2018-12-20 19:44:18.000000000","updated":"2019-07-18 19:19:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66158e5662c16fcd6a9c32902d757f09b9ba2478","_number":626663,"virtual_id_number":626663,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2019-07-18 19:19:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-20 22:53:58.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-01-14 23:26:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-07-18 19:19:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c768ad25017965d68b8c0e28a4be0fb4a3a0463b","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-20 19:44:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6e0a56a53bc067fe595f1b39981ede17942b437","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 20:40:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/63/626663/1/check/openstack-tox-docs/972925d/html/ : SUCCESS in 3m 50s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/1/check/build-openstack-releasenotes/dd200ce/html/ : SUCCESS in 6m 34s\n- swift-tox-py27 http://logs.openstack.org/63/626663/1/check/swift-tox-py27/4c7c0ea/ : SUCCESS in 15m 11s\n- swift-tox-py35 http://logs.openstack.org/63/626663/1/check/swift-tox-py35/5fbb4d0/ : SUCCESS in 7m 24s\n- swift-tox-py36 http://logs.openstack.org/63/626663/1/check/swift-tox-py36/423863c/ : SUCCESS in 5m 45s (non-voting)\n- swift-tox-func http://logs.openstack.org/63/626663/1/check/swift-tox-func/5de417a/ : SUCCESS in 12m 00s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/1/check/swift-tox-func-encryption/a91edad/ : SUCCESS in 8m 43s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/1/check/swift-tox-func-domain-remap-staticweb/e771a67/ : SUCCESS in 12m 22s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/1/check/swift-tox-func-ec/be259db/ : FAILURE in 37m 59s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/1/check/swift-tox-func-s3api/0c62fdc/ : SUCCESS in 11m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/daa3b8d/ : TIMED_OUT in 30m 49s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/1/check/swift-probetests-centos-7/fa4e0f9/ : SUCCESS in 31m 49s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/1/check/swift-dsvm-functional/b48b9da/ : SUCCESS in 21m 51s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/1/check/swift-dsvm-functional-ipv6/8f25604/ : SUCCESS in 23m 40s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/1/check/swift-tox-lower-constraints/223c2ba/ : SUCCESS in 10m 59s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/1/check/openstack-tox-pep8/b5d3a03/ : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/1/check/swift-multinode-rolling-upgrade/bd56f58/ : RETRY_LIMIT in 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c7ac22790aa62bbeb85ec5803a3d01f9b50e4317","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-20 20:46:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8743862f0f5a743994fab7d5600f0302198a2268","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-20 21:08:17.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"e0909e9bee1cfb74b9586a73e0cf6c068e44da06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 21:36:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/63/626663/2/check/openstack-tox-docs/3a4fd99/html/ : SUCCESS in 4m 09s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/2/check/build-openstack-releasenotes/255ba9d/html/ : SUCCESS in 7m 27s\n- swift-tox-py27 http://logs.openstack.org/63/626663/2/check/swift-tox-py27/40fc5c5/ : SUCCESS in 12m 34s\n- swift-tox-py35 http://logs.openstack.org/63/626663/2/check/swift-tox-py35/4a428a2/ : SUCCESS in 5m 55s\n- swift-tox-py36 http://logs.openstack.org/63/626663/2/check/swift-tox-py36/032eff9/ : SUCCESS in 6m 19s (non-voting)\n- swift-tox-func http://logs.openstack.org/63/626663/2/check/swift-tox-func/ffae204/ : SUCCESS in 8m 56s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/2/check/swift-tox-func-encryption/fccbb81/ : SUCCESS in 8m 36s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/2/check/swift-tox-func-domain-remap-staticweb/72f70e2/ : SUCCESS in 7m 54s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/2/check/swift-tox-func-ec/1da043d/ : FAILURE in 9m 28s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/2/check/swift-tox-func-s3api/bbc0861/ : SUCCESS in 10m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b7c73e3/ : FAILURE in 13m 03s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/2/check/swift-probetests-centos-7/a7e0845/ : SUCCESS in 42m 51s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/2/check/swift-dsvm-functional/3f4c851/ : SUCCESS in 23m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/2/check/swift-dsvm-functional-ipv6/0e941a5/ : SUCCESS in 24m 38s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/2/check/swift-tox-lower-constraints/a8e4919/ : SUCCESS in 10m 27s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/2/check/openstack-tox-pep8/269d70e/ : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/2/check/swift-multinode-rolling-upgrade/6896779/ : SUCCESS in 24m 06s","accounts_in_message":[],"_revision_number":2},{"id":"edfb9e547efeee605cf9262875ae4b851b11863a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 21:36:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 http://logs.openstack.org/63/626663/2/experimental/swift-tox-py27-centos-7/5c58215/ : SUCCESS in 15m 43s\n- swift-tox-func-centos-7 http://logs.openstack.org/63/626663/2/experimental/swift-tox-func-centos-7/7226b02/ : SUCCESS in 11m 45s\n- swift-tox-func-encryption-centos-7 http://logs.openstack.org/63/626663/2/experimental/swift-tox-func-encryption-centos-7/30fd795/ : SUCCESS in 12m 33s\n- swift-tox-func-ec-centos-7 http://logs.openstack.org/63/626663/2/experimental/swift-tox-func-ec-centos-7/4c420e9/ : SUCCESS in 12m 36s\n- swift-multinode-rolling-upgrade-queens http://logs.openstack.org/63/626663/2/experimental/swift-multinode-rolling-upgrade-queens/b876673/ : SUCCESS in 23m 04s\n- swift-multinode-rolling-upgrade-rocky http://logs.openstack.org/63/626663/2/experimental/swift-multinode-rolling-upgrade-rocky/a0f080c/ : SUCCESS in 22m 55s","accounts_in_message":[],"_revision_number":2},{"id":"dd2d925875717a7ef71b3a202198b05b1fd82cff","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-20 22:19:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"53d821b57df37ff226c82b4a88088308dbef0081","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2018-12-20 22:53:58.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks like some awesome ground work for rolling crud tests. Nice one Thiago. From what I can see, looks great.\n\nI like your idea of an extra object server (you mentioned in IRC). if extra nodes aren\u0027t really an issue. Ideally, 2x proxy, 2x AC and 2x O would been nice. Then we can \"test\" zero downtime time upgrade as well. But that would mean 2 more VMs so something we could decide later. \n\nTentitive +2 for now.. but really should wait for zuul to report back, because the ansible looks right but only running it will really tell :)","accounts_in_message":[],"_revision_number":3},{"id":"9b4b2eed8936b730a2be3f258e508219087d0e12","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-20 22:54:24.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"e16a0648f4293eee774cf79250a9405447b1163f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 22:59:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/63/626663/3/check/openstack-tox-docs/4c79294/html/ : SUCCESS in 3m 53s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/3/check/build-openstack-releasenotes/35b0685/html/ : SUCCESS in 6m 20s\n- swift-tox-py27 http://logs.openstack.org/63/626663/3/check/swift-tox-py27/7992966/ : SUCCESS in 12m 06s\n- swift-tox-py35 http://logs.openstack.org/63/626663/3/check/swift-tox-py35/75d2ad2/ : SUCCESS in 5m 58s\n- swift-tox-py36 http://logs.openstack.org/63/626663/3/check/swift-tox-py36/3eb4a2a/ : SUCCESS in 4m 42s (non-voting)\n- swift-tox-func http://logs.openstack.org/63/626663/3/check/swift-tox-func/a64ef1a/ : SUCCESS in 8m 55s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/3/check/swift-tox-func-encryption/ce7d4b9/ : SUCCESS in 10m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/3/check/swift-tox-func-domain-remap-staticweb/b16f758/ : SUCCESS in 7m 37s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/3/check/swift-tox-func-ec/fa85515/ : SUCCESS in 8m 11s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/3/check/swift-tox-func-s3api/f5a0a4a/ : SUCCESS in 9m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/eca9361/ : SUCCESS in 11m 24s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/3/check/swift-probetests-centos-7/861be47/ : SUCCESS in 29m 26s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/3/check/swift-dsvm-functional/fc9a54a/ : SUCCESS in 22m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/3/check/swift-dsvm-functional-ipv6/a75b23f/ : SUCCESS in 23m 10s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/3/check/swift-tox-lower-constraints/bc76e79/ : FAILURE in 9m 44s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/3/check/openstack-tox-pep8/dd5b7ba/ : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/3/check/swift-multinode-rolling-upgrade/93bd2e7/ : RETRY_LIMIT in 9m 54s","accounts_in_message":[],"_revision_number":3},{"id":"8d0be9c49d1268c037b7c27e4310e1b9f319653e","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-20 23:24:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c948fd30485c52b8b1ea004760b5d8f7cf8dbb32","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-20 23:26:13.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"ebf012010d1344dea9e1fb65c57e5834ff2bfa46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 23:52:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 http://logs.openstack.org/63/626663/4/experimental/swift-tox-py27-centos-7/602a6e7/ : SUCCESS in 12m 31s\n- swift-tox-func-centos-7 http://logs.openstack.org/63/626663/4/experimental/swift-tox-func-centos-7/841124a/ : SUCCESS in 11m 30s\n- swift-tox-func-encryption-centos-7 http://logs.openstack.org/63/626663/4/experimental/swift-tox-func-encryption-centos-7/5ed436a/ : SUCCESS in 9m 29s\n- swift-tox-func-ec-centos-7 http://logs.openstack.org/63/626663/4/experimental/swift-tox-func-ec-centos-7/d414439/ : SUCCESS in 11m 26s\n- swift-multinode-rolling-upgrade-queens http://logs.openstack.org/63/626663/4/experimental/swift-multinode-rolling-upgrade-queens/1cfe8ef/ : SUCCESS in 20m 42s\n- swift-multinode-rolling-upgrade-rocky http://logs.openstack.org/63/626663/4/experimental/swift-multinode-rolling-upgrade-rocky/0960d53/ : SUCCESS in 20m 50s","accounts_in_message":[],"_revision_number":4},{"id":"617392ff71b48549f815007a13f00e813bfd4f2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 00:01:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/63/626663/4/check/openstack-tox-docs/d0a6bbb/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/4/check/build-openstack-releasenotes/d088099/html/ : SUCCESS in 6m 12s\n- swift-tox-py27 http://logs.openstack.org/63/626663/4/check/swift-tox-py27/5c07549/ : SUCCESS in 10m 01s\n- swift-tox-py35 http://logs.openstack.org/63/626663/4/check/swift-tox-py35/3f40f0b/ : SUCCESS in 5m 50s\n- swift-tox-py36 http://logs.openstack.org/63/626663/4/check/swift-tox-py36/1e978f5/ : SUCCESS in 4m 56s (non-voting)\n- swift-tox-func http://logs.openstack.org/63/626663/4/check/swift-tox-func/a9dd7c3/ : SUCCESS in 8m 15s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/4/check/swift-tox-func-encryption/e4bf6eb/ : SUCCESS in 13m 20s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/4/check/swift-tox-func-domain-remap-staticweb/74fd25f/ : SUCCESS in 7m 56s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/4/check/swift-tox-func-ec/c210b7c/ : SUCCESS in 12m 49s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api/13e896d/ : SUCCESS in 9m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e774215/ : SUCCESS in 11m 41s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/4/check/swift-probetests-centos-7/9a6f3ed/ : FAILURE in 27m 16s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional/c1d45d8/ : SUCCESS in 30m 02s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional-ipv6/cf7523a/ : SUCCESS in 26m 29s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/4/check/swift-tox-lower-constraints/a4f7208/ : SUCCESS in 9m 54s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/4/check/openstack-tox-pep8/c57ae75/ : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/4/check/swift-multinode-rolling-upgrade/79c9efa/ : SUCCESS in 20m 46s","accounts_in_message":[],"_revision_number":4},{"id":"65cb21d8f6b98270bfc782ff64f6d69eae4587c5","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-21 01:42:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"695e6c19c04325db3d9a2b30ad6f9ce5de2cf3f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 02:16:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/63/626663/4/check/openstack-tox-docs/dea0936/html/ : SUCCESS in 4m 21s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/4/check/build-openstack-releasenotes/fd3f4ad/html/ : SUCCESS in 6m 35s\n- swift-tox-py27 http://logs.openstack.org/63/626663/4/check/swift-tox-py27/0adcf83/ : SUCCESS in 11m 02s\n- swift-tox-py35 http://logs.openstack.org/63/626663/4/check/swift-tox-py35/17e40c2/ : SUCCESS in 5m 51s\n- swift-tox-py36 http://logs.openstack.org/63/626663/4/check/swift-tox-py36/d90cafd/ : SUCCESS in 5m 58s (non-voting)\n- swift-tox-func http://logs.openstack.org/63/626663/4/check/swift-tox-func/b3e33a8/ : SUCCESS in 9m 45s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/4/check/swift-tox-func-encryption/0cf1c86/ : SUCCESS in 8m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/4/check/swift-tox-func-domain-remap-staticweb/d47686c/ : SUCCESS in 9m 03s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/4/check/swift-tox-func-ec/94f857b/ : SUCCESS in 8m 23s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api/ef804c3/ : SUCCESS in 9m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/047bcc6/ : SUCCESS in 12m 46s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/4/check/swift-probetests-centos-7/9751d7e/ : FAILURE in 30m 09s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional/ec5d356/ : SUCCESS in 22m 07s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional-ipv6/8916cab/ : SUCCESS in 24m 42s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/4/check/swift-tox-lower-constraints/029591b/ : SUCCESS in 10m 40s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/4/check/openstack-tox-pep8/6ff9940/ : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/4/check/swift-multinode-rolling-upgrade/7a35585/ : SUCCESS in 23m 52s","accounts_in_message":[],"_revision_number":4},{"id":"4c93397180886f626206ffaeb44b7a4697e53e73","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-21 15:06:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"51dcaee7bde422279765b8689b9ba70603c2d783","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 15:44:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/63/626663/4/check/openstack-tox-docs/73b3dfe/html/ : SUCCESS in 4m 03s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/4/check/build-openstack-releasenotes/77ec90e/html/ : SUCCESS in 6m 17s\n- swift-tox-py27 http://logs.openstack.org/63/626663/4/check/swift-tox-py27/aa57607/ : SUCCESS in 13m 14s\n- swift-tox-py35 http://logs.openstack.org/63/626663/4/check/swift-tox-py35/705afd6/ : SUCCESS in 8m 29s\n- swift-tox-py36 http://logs.openstack.org/63/626663/4/check/swift-tox-py36/2d52a89/ : SUCCESS in 6m 57s (non-voting)\n- swift-tox-func http://logs.openstack.org/63/626663/4/check/swift-tox-func/45febb1/ : SUCCESS in 23m 29s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/4/check/swift-tox-func-encryption/15d7212/ : SUCCESS in 12m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/4/check/swift-tox-func-domain-remap-staticweb/202f8ec/ : SUCCESS in 11m 58s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/4/check/swift-tox-func-ec/1bb6a38/ : SUCCESS in 9m 05s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api/117aa11/ : SUCCESS in 9m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f58a17e/ : SUCCESS in 12m 58s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/4/check/swift-probetests-centos-7/95bcd95/ : SUCCESS in 29m 46s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional/d15c86e/ : SUCCESS in 26m 23s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional-ipv6/4346ef1/ : SUCCESS in 29m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/4/check/swift-tox-lower-constraints/7a14341/ : FAILURE in 10m 24s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/4/check/openstack-tox-pep8/375075e/ : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/4/check/swift-multinode-rolling-upgrade/2f33309/ : SUCCESS in 22m 55s","accounts_in_message":[],"_revision_number":4},{"id":"74132c25d555cc8e809b0c87812cc289b1676593","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-21 19:06:52.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9b8b11acb332e6a768cf4db7f4e579cfcb43dbc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 19:38:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/626663/4/check/openstack-tox-docs/b6776e2/html/ : SUCCESS in 3m 51s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/4/check/build-openstack-releasenotes/1e39af6/html/ : SUCCESS in 6m 13s\n- swift-tox-py27 http://logs.openstack.org/63/626663/4/check/swift-tox-py27/f9ba5ef/ : SUCCESS in 18m 56s\n- swift-tox-py35 http://logs.openstack.org/63/626663/4/check/swift-tox-py35/d75f173/ : SUCCESS in 5m 48s\n- swift-tox-py36 http://logs.openstack.org/63/626663/4/check/swift-tox-py36/cbca328/ : SUCCESS in 5m 05s (non-voting)\n- swift-tox-func http://logs.openstack.org/63/626663/4/check/swift-tox-func/2a2a84c/ : SUCCESS in 9m 00s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/4/check/swift-tox-func-encryption/3bd9ba4/ : SUCCESS in 7m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/4/check/swift-tox-func-domain-remap-staticweb/5864610/ : SUCCESS in 8m 46s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/4/check/swift-tox-func-ec/731a72c/ : SUCCESS in 9m 46s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api/289e726/ : SUCCESS in 11m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8500fbf/ : SUCCESS in 13m 17s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/4/check/swift-probetests-centos-7/6b8aaff/ : SUCCESS in 30m 01s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional/e8c3137/ : SUCCESS in 20m 52s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/4/check/swift-dsvm-functional-ipv6/024e691/ : SUCCESS in 24m 22s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/4/check/swift-tox-lower-constraints/b600dce/ : SUCCESS in 18m 31s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/4/check/openstack-tox-pep8/5417048/ : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/4/check/swift-multinode-rolling-upgrade/5270585/ : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f68e62d999dde26af7e4343a609a4103d61432f3","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2018-12-21 20:09:25.000000000","message":"Patch Set 4:\n\n@matt, in the past i\u0027ve been reluctant to add more VMs due to resource consumption, also it delays test results for us, since the multinode test sits on queued for a long time. \n\nStill I really like the idea of being able to test \"zero\" downtime upgrade. I guess we should give it a shot at some point and figure out how much delays it adds. I guess we would also have to add a load-balancer in front of the proxies, so that just adds to the config complexity.","accounts_in_message":[],"_revision_number":4},{"id":"b6a08b554e0bdf8852da1796277b3d4118cd6358","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-01-14 23:26:29.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nSeems to make sense... and I can see them running on swift-multinode-rolling-upgrade and swift-multinode-rolling-upgrade-queens which is nice!","accounts_in_message":[],"_revision_number":4},{"id":"407169c42948374d02e5c63867ccff0f53dc2200","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-16 16:31:42.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5493d62cfc9b94c02383f9916555807f5210058b","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-16 16:44:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5b3e09c5f5870b3362a9dbf2dec8b0e4dceb3a7c","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-16 17:36:02.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"e410638f7932b6b94b3ee501f67c29be619f58f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:49:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/63/626663/5/check/build-openstack-api-ref/cf67b69/html/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/63/626663/5/check/openstack-tox-docs/e371abc/html/ : SUCCESS in 6m 06s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/5/check/build-openstack-releasenotes/07e380a/html/ : SUCCESS in 8m 22s\n- swift-build-image http://logs.openstack.org/63/626663/5/check/swift-build-image/28bc8c0/ : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/63/626663/5/check/swift-tox-py27/b402e7d/ : SUCCESS in 10m 40s\n- swift-tox-py36 http://logs.openstack.org/63/626663/5/check/swift-tox-py36/bdf53d7/ : SUCCESS in 10m 51s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/63/626663/5/check/swift-tox-py37/3efa103/ : SUCCESS in 14m 30s\n- swift-tox-func http://logs.openstack.org/63/626663/5/check/swift-tox-func/d5cf539/ : SUCCESS in 11m 44s\n- swift-tox-func-py37 http://logs.openstack.org/63/626663/5/check/swift-tox-func-py37/bd195fd/ : SUCCESS in 10m 58s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/5/check/swift-tox-func-encryption/ac7c95b/ : SUCCESS in 9m 44s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/5/check/swift-tox-func-domain-remap-staticweb/ad78aeb/ : SUCCESS in 9m 32s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/5/check/swift-tox-func-ec/9a9066b/ : SUCCESS in 10m 06s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/5/check/swift-tox-func-s3api/4640b74/ : SUCCESS in 11m 46s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/63/626663/5/check/swift-tox-func-domain-remap-staticweb-py37/e3210d8/ : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7f48dfd/ : SUCCESS in 13m 16s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/5/check/swift-probetests-centos-7/06a4413/ : SUCCESS in 33m 04s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/5/check/swift-dsvm-functional/05a6ebc/ : SUCCESS in 35m 08s\n- swift-dsvm-functional-py3 http://logs.openstack.org/63/626663/5/check/swift-dsvm-functional-py3/e997a8d/ : SUCCESS in 30m 38s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/5/check/swift-dsvm-functional-ipv6/9b0a33c/ : SUCCESS in 34m 38s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/5/check/swift-tox-lower-constraints/3c4d26f/ : SUCCESS in 14m 41s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/5/check/openstack-tox-pep8/54d76b7/ : SUCCESS in 7m 48s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/5/check/swift-multinode-rolling-upgrade/941adcd/ : RETRY_LIMIT in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e7f9e4c268f2292eea8530fd0005bd87d8fa8e6f","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-17 18:05:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bb2374b0a3fd2a640b61e2b2e1a4030657b543c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 19:22:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/63/626663/6/check/build-openstack-api-ref/b920bed/html/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/63/626663/6/check/openstack-tox-docs/4adc184/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/6/check/build-openstack-releasenotes/fc5a95b/html/ : SUCCESS in 7m 49s\n- swift-build-image http://logs.openstack.org/63/626663/6/check/swift-build-image/69e3cbc/ : SUCCESS in 7m 48s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/63/626663/6/check/swift-tox-py27/8b0ebb6/ : SUCCESS in 11m 14s\n- swift-tox-py36 http://logs.openstack.org/63/626663/6/check/swift-tox-py36/a4d6604/ : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/63/626663/6/check/swift-tox-py37/554d66e/ : SUCCESS in 11m 49s\n- swift-tox-func http://logs.openstack.org/63/626663/6/check/swift-tox-func/4d98a63/ : SUCCESS in 9m 05s\n- swift-tox-func-py37 http://logs.openstack.org/63/626663/6/check/swift-tox-func-py37/69c3c72/ : SUCCESS in 10m 30s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/6/check/swift-tox-func-encryption/1347096/ : SUCCESS in 9m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/6/check/swift-tox-func-domain-remap-staticweb/5696f0d/ : SUCCESS in 10m 50s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/6/check/swift-tox-func-ec/67bdf81/ : SUCCESS in 9m 36s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/6/check/swift-tox-func-s3api/761e675/ : SUCCESS in 13m 38s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/63/626663/6/check/swift-tox-func-domain-remap-staticweb-py37/7562a6c/ : SUCCESS in 7m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/6/check/swift-tox-func-s3api-ceph-s3tests-tempauth/516b21c/ : SUCCESS in 12m 47s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/6/check/swift-probetests-centos-7/2f1ae78/ : SUCCESS in 31m 59s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/6/check/swift-dsvm-functional/3b3f3b0/ : SUCCESS in 29m 26s\n- swift-dsvm-functional-py3 http://logs.openstack.org/63/626663/6/check/swift-dsvm-functional-py3/0941f51/ : SUCCESS in 33m 05s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/6/check/swift-dsvm-functional-ipv6/090a4ae/ : SUCCESS in 31m 57s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/6/check/swift-tox-lower-constraints/f66631d/ : SUCCESS in 12m 18s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/6/check/openstack-tox-pep8/0d90d41/ : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/6/check/swift-multinode-rolling-upgrade/641c202/ : RETRY_LIMIT in 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1ed7a8f8e6bfbea89ed2a0496c890dd21d29e3a5","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-18 18:35:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e37826707812ef9ea6109ba96bf72f7b417cc8a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 19:19:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/63/626663/7/check/build-openstack-api-ref/34fb4c9/html/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/63/626663/7/check/openstack-tox-docs/789df3f/html/ : SUCCESS in 5m 23s\n- build-openstack-releasenotes http://logs.openstack.org/63/626663/7/check/build-openstack-releasenotes/fabd374/html/ : SUCCESS in 8m 05s\n- swift-build-image http://logs.openstack.org/63/626663/7/check/swift-build-image/0e80ac1/ : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/63/626663/7/check/swift-tox-py27/bf26e73/ : SUCCESS in 9m 40s\n- swift-tox-py36 http://logs.openstack.org/63/626663/7/check/swift-tox-py36/4f7f2d6/ : SUCCESS in 10m 30s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/63/626663/7/check/swift-tox-py37/0f1f67e/ : SUCCESS in 10m 20s\n- swift-tox-func http://logs.openstack.org/63/626663/7/check/swift-tox-func/0fd55ca/ : SUCCESS in 11m 08s\n- swift-tox-func-py37 http://logs.openstack.org/63/626663/7/check/swift-tox-func-py37/01403c7/ : SUCCESS in 13m 16s\n- swift-tox-func-encryption http://logs.openstack.org/63/626663/7/check/swift-tox-func-encryption/4f2e688/ : SUCCESS in 9m 14s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/63/626663/7/check/swift-tox-func-domain-remap-staticweb/c666162/ : SUCCESS in 13m 34s\n- swift-tox-func-ec http://logs.openstack.org/63/626663/7/check/swift-tox-func-ec/5e48189/ : SUCCESS in 9m 36s\n- swift-tox-func-s3api http://logs.openstack.org/63/626663/7/check/swift-tox-func-s3api/c734836/ : SUCCESS in 16m 42s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/63/626663/7/check/swift-tox-func-domain-remap-staticweb-py37/424d217/ : SUCCESS in 10m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/63/626663/7/check/swift-tox-func-s3api-ceph-s3tests-tempauth/befc5be/ : SUCCESS in 16m 20s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/63/626663/7/check/swift-probetests-centos-7/a1fa6f8/ : SUCCESS in 31m 10s\n- swift-dsvm-functional http://logs.openstack.org/63/626663/7/check/swift-dsvm-functional/d79891b/ : SUCCESS in 26m 08s\n- swift-dsvm-functional-py3 http://logs.openstack.org/63/626663/7/check/swift-dsvm-functional-py3/29de39d/ : SUCCESS in 33m 16s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/63/626663/7/check/swift-dsvm-functional-ipv6/a4e3100/ : SUCCESS in 28m 01s\n- swift-tox-lower-constraints http://logs.openstack.org/63/626663/7/check/swift-tox-lower-constraints/c5d23c4/ : SUCCESS in 13m 48s\n- openstack-tox-pep8 http://logs.openstack.org/63/626663/7/check/openstack-tox-pep8/92926e5/ : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/63/626663/7/check/swift-multinode-rolling-upgrade/9d7996a/ : SUCCESS in 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"de5ade511b0bcd3d946d6a7792d7b19928046e14","revisions":{"139a76e8e2303de956e7b1fbf06112cb2cb6c0f1":{"kind":"REWORK","_number":1,"created":"2018-12-20 19:44:18.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/63/626663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/626663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/626663/1"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-09-14 14:50:48.000000000","tz":-240},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-12-20 19:39:16.000000000","tz":-420},"subject":"WIP: add rolling upgrade tests","message":"WIP: add rolling upgrade tests\n\nAdd a new set of \"rolling\" tests. Their intent is to test basic\nCRUD functionality during a rolling upgrade. The sequence would\nbe something like this:\n 1. Install \"old\" version of Swift\n 2. Run \"rolling_before\" tests. These tests will leave data on cluster\n 3. Upgrade Swift\n 4. Run \"rolling_after\" tests. These tests reads the \"before\" data\n\nIt was necessary to add a new test runner VM. The functional test runner\nis running the functional tests from the old version of swift.\nThe new test runner, needs to have the latest code (master).\n\nChange-Id: Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/139a76e8e2303de956e7b1fbf06112cb2cb6c0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/139a76e8e2303de956e7b1fbf06112cb2cb6c0f1"}]},"branch":"refs/heads/master"},"185c501258a79a093e51e3733447830d0dd19f5f":{"kind":"REWORK","_number":2,"created":"2018-12-20 20:46:30.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/63/626663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/626663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/626663/2"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-09-14 14:50:48.000000000","tz":-240},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-12-20 20:46:14.000000000","tz":-420},"subject":"WIP: add rolling upgrade tests","message":"WIP: add rolling upgrade tests\n\nAdd a new set of \"rolling\" tests. Their intent is to test basic\nCRUD functionality during a rolling upgrade. The sequence would\nbe something like this:\n 1. Install \"old\" version of Swift\n 2. Run \"rolling_before\" tests. These tests will leave data on cluster\n 3. Upgrade Swift\n 4. Run \"rolling_after\" tests. These tests reads the \"before\" data\n\nIt was necessary to add a new test runner VM. The functional test runner\nis running the functional tests from the old version of swift.\nThe new test runner, needs to have the latest code (master).\n\nChange-Id: Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/185c501258a79a093e51e3733447830d0dd19f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/185c501258a79a093e51e3733447830d0dd19f5f"}]},"branch":"refs/heads/master"},"111a208f07c432132ea4544d9f2a65eea49bc4d3":{"kind":"REWORK","_number":3,"created":"2018-12-20 22:19:27.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/63/626663/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/626663/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/626663/3"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-09-14 14:50:48.000000000","tz":-240},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-12-20 22:19:13.000000000","tz":-420},"subject":"Add rolling upgrade tests","message":"Add rolling upgrade tests\n\nAdd a new set of \"rolling\" tests. Their intent is to test basic\nCRUD functionality during a rolling upgrade. The sequence would\nbe something like this:\n 1. Install \"old\" version of Swift\n 2. Run \"rolling_before\" tests. These tests will leave data on cluster\n 3. Upgrade Swift\n 4. Run \"rolling_after\" tests. These tests reads the \"before\" data\n\nIt was necessary to add a new test runner VM. The functional test runner\nis running the functional tests from the old version of swift.\nThe new test runner, needs to have the latest code (master).\n\nChange-Id: Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/111a208f07c432132ea4544d9f2a65eea49bc4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/111a208f07c432132ea4544d9f2a65eea49bc4d3"}]},"branch":"refs/heads/master"},"f39244c92dee7ab86b505da47e76225fc6f90698":{"kind":"REWORK","_number":4,"created":"2018-12-20 23:24:44.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/63/626663/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/626663/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/626663/4"}}},"commit":{"parents":[{"commit":"a7aa2329584f1d02f7d1fa205d56aeadffdc980f","subject":"Merge \"Add a comment about the module list for py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7aa2329584f1d02f7d1fa205d56aeadffdc980f"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-09-14 14:50:48.000000000","tz":-240},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-12-20 23:24:39.000000000","tz":-420},"subject":"Add rolling upgrade tests","message":"Add rolling upgrade tests\n\nAdd a new set of \"rolling\" tests. Their intent is to test basic\nCRUD functionality during a rolling upgrade. The sequence would\nbe something like this:\n 1. Install \"old\" version of Swift\n 2. Run \"rolling_before\" tests. These tests will leave data on cluster\n 3. Upgrade Swift\n 4. Run \"rolling_after\" tests. These tests reads the \"before\" data\n\nIt was necessary to add a new test runner VM. The functional test runner\nis running the functional tests from the old version of swift.\nThe new test runner, needs to have the latest code (master).\n\nChange-Id: Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f39244c92dee7ab86b505da47e76225fc6f90698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f39244c92dee7ab86b505da47e76225fc6f90698"}]},"branch":"refs/heads/master"},"2a71fe345189dea541197eae744db5e7ec9d12eb":{"kind":"REWORK","_number":5,"created":"2019-07-16 16:44:56.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/63/626663/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/626663/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/626663/5"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-09-14 14:50:48.000000000","tz":-240},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-07-16 16:44:41.000000000","tz":60},"subject":"Add rolling upgrade tests","message":"Add rolling upgrade tests\n\nAdd a new set of \"rolling\" tests. Their intent is to test basic\nCRUD functionality during a rolling upgrade. The sequence would\nbe something like this:\n 1. Install \"old\" version of Swift\n 2. Run \"rolling/stage\" tests. These tests will leave data on cluster\n 3. Upgrade Swift\n 4. Run \"rolling/verify\" tests. These tests reads the \"stage\" data\n\nIt was necessary to add a new test runner VM. The functional test runner\nis running the functional tests from the old version of swift.\nThe new test runner, needs to have the latest code (master).\n\nChange-Id: Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a71fe345189dea541197eae744db5e7ec9d12eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a71fe345189dea541197eae744db5e7ec9d12eb"}]},"branch":"refs/heads/master"},"059bc7765c982786e90a3a02cc846eee8732e847":{"kind":"REWORK","_number":6,"created":"2019-07-17 18:05:40.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/63/626663/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/626663/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/626663/6"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-09-14 14:50:48.000000000","tz":-240},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-07-17 18:05:25.000000000","tz":60},"subject":"Add rolling upgrade tests","message":"Add rolling upgrade tests\n\nAdd a new set of \"rolling\" tests. Their intent is to test basic\nCRUD functionality during a rolling upgrade. The sequence would\nbe something like this:\n 1. Install \"old\" version of Swift\n 2. Run \"rolling/stage\" tests. These tests will leave data on cluster\n 3. Upgrade Swift\n 4. Run \"rolling/verify\" tests. These tests reads the \"stage\" data\n\nIt was necessary to add a new test runner VM. The functional test runner\nis running the functional tests from the old version of swift.\nThe new test runner, needs to have the latest code (master).\n\nChange-Id: Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/059bc7765c982786e90a3a02cc846eee8732e847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/059bc7765c982786e90a3a02cc846eee8732e847"}]},"branch":"refs/heads/master"},"de5ade511b0bcd3d946d6a7792d7b19928046e14":{"kind":"REWORK","_number":7,"created":"2019-07-18 18:35:19.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/63/626663/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/626663/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/626663/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/626663/7"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2018-09-14 14:50:48.000000000","tz":-240},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-07-18 18:34:54.000000000","tz":60},"subject":"Add rolling upgrade tests","message":"Add rolling upgrade tests\n\nAdd a new set of \"rolling\" tests. Their intent is to test basic\nCRUD functionality during a rolling upgrade. The sequence would\nbe something like this:\n 1. Install \"old\" version of Swift\n 2. Run \"rolling/stage\" tests. These tests will leave data on cluster\n 3. Upgrade Swift\n 4. Run \"rolling/verify\" tests. These tests reads the \"stage\" data\n\nIt was necessary to add a new test runner VM. The functional test runner\nis running the functional tests from the old version of swift.\nThe new test runner, needs to have the latest code (master).\n\nChange-Id: Ie0e0b30ba7c51f7881126456ac6b6e90a4b4010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de5ade511b0bcd3d946d6a7792d7b19928046e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de5ade511b0bcd3d946d6a7792d7b19928046e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
