)]}'
{"id":"openstack%2Fswift~629402","triplet_id":"openstack%2Fswift~stable%2Fpike~I0a94482ed16cb30ce79074e053e6177fe97bcaa9","project":"openstack/swift","branch":"stable/pike","topic":"pike-slow-expirer","hashtags":[],"change_id":"I0a94482ed16cb30ce79074e053e6177fe97bcaa9","subject":"object-server can 409 in response to x-if-delete-at","status":"MERGED","created":"2019-01-09 08:01:58.000000000","updated":"2019-05-02 21:55:54.000000000","submitted":"2019-05-02 21:55:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"629402-1556834154548-5260f087","meta_rev_id":"a05b6dc06ddc4b62fbecb8320825013d7a2894a0","_number":629402,"virtual_id_number":629402,"owner":{"_account_id":23449,"name":"Pavel Kvasnička","email":"pavel.kvasnicka@firma.seznam.cz","username":"PavelK"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-05-02 21:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-02 18:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-02 18:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-02 18:52:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-02 21:55:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47b78642aeccb5df33e98bed4add36ab65049b41","author":{"_account_id":23449,"name":"Pavel Kvasnička","email":"pavel.kvasnicka@firma.seznam.cz","username":"PavelK"},"date":"2019-01-09 08:01:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cacc34840326560f6d070d42cedc39c50a5ea9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 09:44:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/02/629402/1/check/build-openstack-sphinx-docs/69c4483/html/ : SUCCESS in 5m 06s\n- swift-tox-py27 http://logs.openstack.org/02/629402/1/check/swift-tox-py27/d7fc5cf/ : SUCCESS in 7m 42s\n- swift-tox-py35 http://logs.openstack.org/02/629402/1/check/swift-tox-py35/f4ec2b9/ : SUCCESS in 3m 44s\n- swift-tox-func http://logs.openstack.org/02/629402/1/check/swift-tox-func/620bcc0/ : SUCCESS in 7m 43s\n- swift-tox-func-post-as-copy http://logs.openstack.org/02/629402/1/check/swift-tox-func-post-as-copy/17d3032/ : SUCCESS in 7m 11s\n- swift-tox-func-encryption http://logs.openstack.org/02/629402/1/check/swift-tox-func-encryption/9c8ef64/ : SUCCESS in 6m 09s\n- swift-tox-func-ec http://logs.openstack.org/02/629402/1/check/swift-tox-func-ec/3bf853e/ : SUCCESS in 7m 39s\n- openstack-tox-pep8 http://logs.openstack.org/02/629402/1/check/openstack-tox-pep8/3904eb1/ : SUCCESS in 4m 14s\n- swift-dsvm-functional http://logs.openstack.org/02/629402/1/check/swift-dsvm-functional/ae518ad/ : SUCCESS in 25m 05s\n- swift-dsvm-functional-identity-v2 http://logs.openstack.org/02/629402/1/check/swift-dsvm-functional-identity-v2/5677de8/ : SUCCESS in 19m 33s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/02/629402/1/check/legacy-tempest-dsvm-neutron-full/086d581/ : SUCCESS in 1h 35m 33s\n- neutron-grenade http://logs.openstack.org/02/629402/1/check/neutron-grenade/c5c3d34/ : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":1},{"id":"76e29583a38e20d2b02697469e746033a5c98b21","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-03-30 00:40:13.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks like a reasonably clean backport -- the tests are a little different, but pop just fine without the changes under swift/\n\nI *think* this would be appropriate for a backport? Cleaning up the expirer queue appropriately seems good...","accounts_in_message":[],"_revision_number":1},{"id":"8bde5711c7c2f7da8c395bbf2273816c02d89cb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-02 18:52:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa608fc106c50a1600cbbd933161757c3b7f9ffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 18:53:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a446ad282841cf42d1633218d6104cf6e918ee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 21:55:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4752c5a76d90db46ca21f2faff11fec1e02fb548","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 21:55:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/02/629402/1/gate/build-openstack-sphinx-docs/9ba8e40/html/ : SUCCESS in 5m 06s\n- swift-tox-py27 http://logs.openstack.org/02/629402/1/gate/swift-tox-py27/571e120/ : SUCCESS in 9m 04s\n- swift-tox-py35 http://logs.openstack.org/02/629402/1/gate/swift-tox-py35/5e4fe08/ : SUCCESS in 3m 55s\n- swift-tox-func http://logs.openstack.org/02/629402/1/gate/swift-tox-func/2dd8fc3/ : SUCCESS in 8m 22s\n- swift-tox-func-post-as-copy http://logs.openstack.org/02/629402/1/gate/swift-tox-func-post-as-copy/a92604b/ : SUCCESS in 8m 36s\n- swift-tox-func-encryption http://logs.openstack.org/02/629402/1/gate/swift-tox-func-encryption/1f00471/ : SUCCESS in 8m 34s\n- swift-tox-func-ec http://logs.openstack.org/02/629402/1/gate/swift-tox-func-ec/81eb11c/ : SUCCESS in 8m 05s\n- openstack-tox-pep8 http://logs.openstack.org/02/629402/1/gate/openstack-tox-pep8/419ab1d/ : SUCCESS in 4m 32s\n- swift-dsvm-functional http://logs.openstack.org/02/629402/1/gate/swift-dsvm-functional/e91c7bc/ : SUCCESS in 20m 06s\n- swift-dsvm-functional-identity-v2 http://logs.openstack.org/02/629402/1/gate/swift-dsvm-functional-identity-v2/d2051d3/ : SUCCESS in 20m 29s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/02/629402/1/gate/legacy-tempest-dsvm-neutron-full/aa2434b/ : SUCCESS in 1h 17m 26s\n- neutron-grenade http://logs.openstack.org/02/629402/1/gate/neutron-grenade/2185476/ : SUCCESS in 48m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3627b0c0af417d26b7117fe6c116025706ddf694","revisions":{"3627b0c0af417d26b7117fe6c116025706ddf694":{"kind":"REWORK","_number":1,"created":"2019-01-09 08:01:58.000000000","uploader":{"_account_id":23449,"name":"Pavel Kvasnička","email":"pavel.kvasnicka@firma.seznam.cz","username":"PavelK"},"ref":"refs/changes/02/629402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/629402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/629402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/629402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/629402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/629402/1"}}},"commit":{"parents":[{"commit":"68fc87ad21d63c713ed13aeb6a214357d2d61dd6","subject":"copy: Stop hanging per-request data on middleware instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68fc87ad21d63c713ed13aeb6a214357d2d61dd6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-01-17 22:07:26.000000000","tz":0},"committer":{"name":"Pavel Kvasnička","email":"pavel.kvasnicka@firma.seznam.cz","date":"2019-01-09 08:01:30.000000000","tz":60},"subject":"object-server can 409 in response to x-if-delete-at","message":"object-server can 409 in response to x-if-delete-at\n\nPreviously, if the expirer had a stale work item (because the object\nwas overwritten or deleted, or some other process handled the delete),\nthen it would keep retrying for reclaim_age, but every time it\u0027d get\nback a 412.\n\nNow, have the object-server be smart enough to say, \"I have more recent\ninformation than you\" and let the expirer accept that as success.\n\nChange-Id: I0a94482ed16cb30ce79074e053e6177fe97bcaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3627b0c0af417d26b7117fe6c116025706ddf694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3627b0c0af417d26b7117fe6c116025706ddf694"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
