)]}'
{"id":"openstack%2Fswift~629659","triplet_id":"openstack%2Fswift~master~I71222801023ce63b07b94374f2e340db5a982acd","project":"openstack/swift","branch":"master","topic":"tlo","hashtags":[],"change_id":"I71222801023ce63b07b94374f2e340db5a982acd","subject":"slo_manifest_hook follow-up","status":"NEW","created":"2019-01-09 20:40:57.000000000","updated":"2026-09-11 22:56:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d677c456f6b90164808d7eb67c35452c529f63b8","_number":629659,"virtual_id_number":629659,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 22:56:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 08:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2efdc05827dcf24ddb00bc0f6ea24d4d083687ce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-01-09 20:40:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebf78c674a9eb37264dce3a1a97618d70ddafada","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 21:32:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/59/629659/1/check/openstack-tox-docs/be0e735/html/ : SUCCESS in 4m 10s\n- swift-tox-py27 http://logs.openstack.org/59/629659/1/check/swift-tox-py27/3134975/ : SUCCESS in 10m 12s\n- swift-tox-py35 http://logs.openstack.org/59/629659/1/check/swift-tox-py35/6b5211a/ : SUCCESS in 5m 45s\n- swift-tox-py36 http://logs.openstack.org/59/629659/1/check/swift-tox-py36/6d7edd2/ : SUCCESS in 4m 40s (non-voting)\n- swift-tox-func http://logs.openstack.org/59/629659/1/check/swift-tox-func/6fb83aa/ : SUCCESS in 8m 40s\n- swift-tox-func-encryption http://logs.openstack.org/59/629659/1/check/swift-tox-func-encryption/e2cdeed/ : SUCCESS in 8m 41s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/59/629659/1/check/swift-tox-func-domain-remap-staticweb/faca009/ : SUCCESS in 8m 09s\n- swift-tox-func-ec http://logs.openstack.org/59/629659/1/check/swift-tox-func-ec/9c07e7d/ : SUCCESS in 9m 06s\n- swift-tox-func-s3api http://logs.openstack.org/59/629659/1/check/swift-tox-func-s3api/d445591/ : SUCCESS in 10m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/59/629659/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/69b4078/ : SUCCESS in 11m 16s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/59/629659/1/check/swift-probetests-centos-7/ce3dab7/ : SUCCESS in 29m 30s\n- swift-dsvm-functional http://logs.openstack.org/59/629659/1/check/swift-dsvm-functional/d821708/ : SUCCESS in 25m 26s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/59/629659/1/check/swift-dsvm-functional-ipv6/900e3d1/ : SUCCESS in 26m 07s\n- swift-tox-lower-constraints http://logs.openstack.org/59/629659/1/check/swift-tox-lower-constraints/57d4738/ : FAILURE in 24m 44s\n- openstack-tox-pep8 http://logs.openstack.org/59/629659/1/check/openstack-tox-pep8/ea7196b/ : FAILURE in 5m 10s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/59/629659/1/check/swift-multinode-rolling-upgrade/4fd73ee/ : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6907da1455100b65a78bbd05c471c76e70ab374b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-01-09 21:32:23.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"67db6a4e5c93310a8105e1e9a4aa1a9e29d7b864","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-01-09 23:03:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46c7fd8c3ddc9c6a53450e5ea9327e823cced58e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 23:43:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/59/629659/2/check/openstack-tox-docs/d65a8d0/html/ : SUCCESS in 4m 30s\n- swift-tox-py27 http://logs.openstack.org/59/629659/2/check/swift-tox-py27/a37103d/ : SUCCESS in 10m 59s\n- swift-tox-py35 http://logs.openstack.org/59/629659/2/check/swift-tox-py35/603230e/ : SUCCESS in 6m 19s\n- swift-tox-py36 http://logs.openstack.org/59/629659/2/check/swift-tox-py36/5ad0ff7/ : SUCCESS in 8m 02s (non-voting)\n- swift-tox-func http://logs.openstack.org/59/629659/2/check/swift-tox-func/4a94842/ : SUCCESS in 10m 33s\n- swift-tox-func-encryption http://logs.openstack.org/59/629659/2/check/swift-tox-func-encryption/035dc41/ : SUCCESS in 9m 04s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/59/629659/2/check/swift-tox-func-domain-remap-staticweb/c6106d5/ : SUCCESS in 10m 15s\n- swift-tox-func-ec http://logs.openstack.org/59/629659/2/check/swift-tox-func-ec/b96cfe3/ : SUCCESS in 9m 37s\n- swift-tox-func-s3api http://logs.openstack.org/59/629659/2/check/swift-tox-func-s3api/71a74b3/ : SUCCESS in 11m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/59/629659/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8b0641d/ : SUCCESS in 14m 29s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/59/629659/2/check/swift-probetests-centos-7/fc46da8/ : SUCCESS in 34m 10s\n- swift-dsvm-functional http://logs.openstack.org/59/629659/2/check/swift-dsvm-functional/d7944ba/ : SUCCESS in 28m 37s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/59/629659/2/check/swift-dsvm-functional-ipv6/232d551/ : SUCCESS in 23m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/59/629659/2/check/swift-tox-lower-constraints/3f196cf/ : SUCCESS in 12m 40s\n- openstack-tox-pep8 http://logs.openstack.org/59/629659/2/check/openstack-tox-pep8/6701b0c/ : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/59/629659/2/check/swift-multinode-rolling-upgrade/4caee7b/ : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d8ca9d9fe12ac34a630cd5a338ccd92e3a2122d2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-02 04:04:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c876e8f2dc6bf11058b842009563cecaa3c6ec52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 05:08:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/59/629659/3/check/openstack-tox-docs/35c3f7b/html/ : SUCCESS in 4m 00s\n- grenade-py3 https://logs.opendev.org/59/629659/3/check/grenade-py3/1767160/ : SUCCESS in 58m 27s\n- tempest-integrated-object-storage https://logs.opendev.org/59/629659/3/check/tempest-integrated-object-storage/c7e95ef/ : SUCCESS in 51m 56s\n- swift-build-image https://logs.opendev.org/59/629659/3/check/swift-build-image/23b3e54/ : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/59/629659/3/check/swift-tox-py27/880d73a/ : SUCCESS in 9m 57s\n- swift-tox-py36 https://logs.opendev.org/59/629659/3/check/swift-tox-py36/e39ff2d/ : SUCCESS in 10m 10s\n- swift-tox-py37 https://logs.opendev.org/59/629659/3/check/swift-tox-py37/0a9767e/ : SUCCESS in 10m 26s\n- swift-tox-func https://logs.opendev.org/59/629659/3/check/swift-tox-func/814514c/ : SUCCESS in 8m 45s\n- swift-tox-func-encryption https://logs.opendev.org/59/629659/3/check/swift-tox-func-encryption/c31a425/ : SUCCESS in 8m 54s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/59/629659/3/check/swift-tox-func-domain-remap-staticweb/f287c5e/ : SUCCESS in 10m 47s\n- swift-tox-func-ec https://logs.opendev.org/59/629659/3/check/swift-tox-func-ec/6134bb9/ : SUCCESS in 10m 01s\n- swift-tox-func-s3api https://logs.opendev.org/59/629659/3/check/swift-tox-func-s3api/85f4fe5/ : SUCCESS in 11m 25s\n- swift-tox-func-py37 https://logs.opendev.org/59/629659/3/check/swift-tox-func-py37/29c1c84/ : SUCCESS in 9m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/59/629659/3/check/swift-tox-func-domain-remap-staticweb-py37/7f75d82/ : SUCCESS in 9m 35s\n- swift-tox-func-ec-py37 https://logs.opendev.org/59/629659/3/check/swift-tox-func-ec-py37/fb39027/ : SUCCESS in 9m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/59/629659/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/315e4d1/ : SUCCESS in 13m 15s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/59/629659/3/check/swift-probetests-centos-7/877aadf/ : SUCCESS in 30m 51s\n- swift-dsvm-functional https://logs.opendev.org/59/629659/3/check/swift-dsvm-functional/281247a/ : SUCCESS in 27m 32s\n- swift-dsvm-functional-py3 https://logs.opendev.org/59/629659/3/check/swift-dsvm-functional-py3/e6ef536/ : SUCCESS in 28m 10s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/59/629659/3/check/swift-dsvm-functional-ipv6/7c429ac/ : SUCCESS in 25m 40s\n- swift-tox-lower-constraints https://logs.opendev.org/59/629659/3/check/swift-tox-lower-constraints/041843b/ : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://logs.opendev.org/59/629659/3/check/openstack-tox-pep8/3b5d1a4/ : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/59/629659/3/check/swift-multinode-rolling-upgrade/bc60370/ : SUCCESS in 15m 11s","accounts_in_message":[],"_revision_number":3},{"id":"493f75782ae9562e8fdd715d46a1001f124a4136","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-10 05:30:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1aa2789f8c762a0279c7fee3e4135b791a075702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 06:47:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ae9100d7a44a79a86fd133f869ed3a : SUCCESS in 7m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cbc1456830b54c78a06824563dbb3ecc : SUCCESS in 59m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f746177380d48e7b025e9e5554fb122 : SUCCESS in 54m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0db3e59a21a04e0fa43a839e4490e307 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f4b19d71983e43f8a69d499d9207cd6e : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7c6c4f8ce6844ecb7100f52d5baee27 : SUCCESS in 10m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e69ca4fd4d7047ab98fe432c21d036bc : SUCCESS in 11m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/95753c86a87f4943bd71b88168ea0e03 : SUCCESS in 11m 13s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/a14194505cf34b908b61a936bc00cbd1 : SUCCESS in 9m 18s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/dc8b057e7fdd4755a042b857782f53d8 : SUCCESS in 9m 32s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/8ccb2163a6f64090b3155fa5390e937d : SUCCESS in 9m 17s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9e22a1cf8d0d4c50acd36831084426ef : SUCCESS in 10m 17s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/84a6b695897f47549d3d1373445f6202 : FAILURE in 12m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7d45e699d6c84ceda8419a6d8a05e6ed : SUCCESS in 9m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5af94a578f9845c7a3b27500488a5746 : SUCCESS in 9m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/59d923da43594eeaac9359aa9e8c45ab : SUCCESS in 9m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5d3f458a7fe44b749ad79d7f054e2604 : SUCCESS in 10m 12s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d71a5f3356b84ed181f0c9b3abac1305 : FAILURE in 12m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04e503755a484a2e9ea20b33f05919e8 : SUCCESS in 12m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8610ae4063684453a576a183d940dedb : SUCCESS in 28m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc2fae6f5eb14eff8401fb7cac799db3 : SUCCESS in 27m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/86b42be73bdd49b5ab444c5857f71916 : SUCCESS in 30m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5dd06dc8f0224d39ab8dd23d92c77b00 : SUCCESS in 29m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81ef34bd3a36412faaf9007dfd562a57 : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f1d71383174fce856d27cd5abcb0d7 : SUCCESS in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60fe4f0dbef5472b9677b23d8bd86473 : SUCCESS in 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd5bd84301824eb19feaa8999cd46a24 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":4},{"id":"11f756f6be8f95caf37ef5b290a5c8b2a12ac531","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 07:17:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7d7b369ef08d13cb2452254c001083029daae030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 08:22:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2cd0582a643428891b6890e3e60ff04 : SUCCESS in 6m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cde77901a7b14d61b717ea9312715519 : SUCCESS in 48m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/528375dbd93f4262b178d56f11caf03e : SUCCESS in 50m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b493153e12c846d2a03d589fda6e6c09 : SUCCESS in 11m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b934a6db5e2141a3b2f3976f788d2dfb : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88b934f17a5243e48f6e8fe8e3677408 : SUCCESS in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/acc89e9a3c354c38b917d9982baa6d3a : SUCCESS in 16m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7933ae5e788e47889aba55ceb5a2ae5e : SUCCESS in 13m 40s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/665634781a894533a04797edf305db54 : SUCCESS in 15m 46s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/0f4aaebc7ecb4d2e9fa82adf5a0ba23c : SUCCESS in 15m 43s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/40ca4505338a4da6b443a62683984f44 : SUCCESS in 16m 44s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/644a02c92f5e43c189fb10f269554189 : SUCCESS in 12m 02s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e36e35bb726e413c8c90161352b39c8f : SUCCESS in 20m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d3f08515f223457a861626c1fedec85c : SUCCESS in 9m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/40f6558759a14361af33be214f63128a : SUCCESS in 16m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3a15400f447544f9ae92531f4da1e01e : SUCCESS in 11m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/87b9e67e8ef54b9e9c953e49b0946468 : SUCCESS in 17m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c1e26eb87a004ecfa84480406e8e61f6 : SUCCESS in 13m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81dc300738cb4a73975e6e40ba0062bd : SUCCESS in 15m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/344e6c659a0c49ea83f98ed75402d227 : SUCCESS in 32m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/238134c5bd2943e8a8543d9e3e45a4f9 : SUCCESS in 31m 23s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c2dd1d031b784638941efa8be6f79a4e : SUCCESS in 32m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e6e0f65cb0ec4b13b0bce7228c316536 : SUCCESS in 29m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e8f6f190b514f6b8b47d006ef4476bb : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78246ad52d834078b656c3839d01f16d : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/117f1959b2a14986a7874547b83abf85 : SUCCESS in 18m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83e144d6c72b4429b3f60e630ceb8e2b : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":5},{"id":"57355a8d7cb61f45882b874e0d2a8ca9bf11ace1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-09 22:46:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bf4f91c4ed9bb348bc32233fae35a74b1f9c273f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 00:08:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b1ac74fbf8449a812d44cc33f27883\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a872d538965402a8c1b7189d8cfd8e5 : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/0b1f417643684ca3a2d5510d20a55498 : SUCCESS in 1h 14m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59dd84b1cf2140b2ad439e4fe21e4e4d : SUCCESS in 45m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50fa2caa02044e8ba055f5581c7ecf47 : SUCCESS in 1h 03m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/283a04a709224fa6a3987a8e56c52499 : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/642f43ca80584d6bbd5ffe4c062bebea : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69ba3917ebc341f1933a4bb1332e3a99 : SUCCESS in 18m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c335253f5ce74a5e994ffe3483c5f3ac : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4ac01daa545145288af2dbda6ce93d6b : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c26eb7e8806e4f4eb12a03a5b1a54d8a : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f030623c5a124fcc94be6a0dab9b4c62 : SUCCESS in 22m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3322d0d2a7c44388925ecb1085da6f71 : SUCCESS in 24m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ca42de112ae74bf5849ba2cd19418637 : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2ffffaaf8384857bd3f0a28763cdedd : SUCCESS in 24m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3f0b62e8eee4924ae3a79f45d8cd4f0 : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0ed326025714e85bb2bcc589025cd77 : SUCCESS in 24m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/441cfd16bd5c4a658263e7f6aa9b0934 : SUCCESS in 24m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f11f159ec75b48f6a6c0fe1d79a7ad09 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/35d7aee51786445b89143b06d313e963 : SUCCESS in 22m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/98a599f485a64792b6f28d5a418997f4 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7a7ff8217f14482a698a15f7f8da26c : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c6ee148d9e5b405f8f2ec268c04a2cf6 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c7795aaff83246d8a8842fdc5811e410 : SUCCESS in 1h 17m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/80ac61801a4244388d3db9b6f48004bd : SUCCESS in 1h 19m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/484bdd9b5a95479c8ac59fa1c055d32b : SUCCESS in 1h 19m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/551a764466f647f8835306b737035d8a : SUCCESS in 47m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c608b0d759d9454096cf967dbdde4beb : SUCCESS in 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e63594689cb406f813329f81c60280d : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e98d20ba907941649f0767e675f6faca : SUCCESS in 20m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4f9fb41dcc94b969d4c6453e6d98a48 : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":6},{"id":"dae9a4ade2c43a784d3f32732eee99b9e0d13458","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 00:32:01.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e37f2e6d0043b58b76ec6321b90687\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9d5df459a4d41f1a8f0a616de3a3f47 : SUCCESS in 1h 04m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9d3551a91284678b5bd4177e5f48f15 : SUCCESS in 1h 19m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/737fc32c378d4838ab80b67e62a56115 : SUCCESS in 1h 06m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7485b7e023884ff79588b77fb8647646 : TIMED_OUT in 1h 31m 02s","accounts_in_message":[],"_revision_number":6},{"id":"56ff5115432e584088c653f6f6d4dfeb33b2b75b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 18:24:25.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b179f4bc619f46acc84c138a170fb6e9cfb5db44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 19:39:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a391cd97cd144fc8bf0a04743e8f2a0e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60302745e34c4aba8b5af86ca25fad25 : SUCCESS in 17m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c68fedee31d34816b3a4b74ebdc9de75 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/867114da06c84171ac4be2fcc8436e4b : SUCCESS in 1h 13m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5335976db2a46498a372c2a03192866 : SUCCESS in 19m 31s","accounts_in_message":[],"_revision_number":7},{"id":"ccf1275455ad1883ccc93987fc2a726220e539d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 19:40:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/670408ef9acb47f9a6b5b7a9698f65a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c8747d3f544f57bcb41ba60bac381e : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/0d14f7cdabbf433baa9d19ff96a02446 : SUCCESS in 50m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9577874c765c44c8983f6b9e498249e8 : SUCCESS in 42m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e2b901567c14d68ad453847dd829ff4 : SUCCESS in 39m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f97519e9ccac40f08056996c7a6a5657 : SUCCESS in 47m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7f2621c441b4ac3935c079d7b2931ef : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8162c91cba704e46916258a103382fb4 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa8c0696b4a6422aa31d8feaf10342bf : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38d73be3606d4c0faed056425eed2167 : SUCCESS in 10m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91625068ab094d9d8cb4efe42815c683 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e589aa16a7044ebaede7e1de44ccbef : SUCCESS in 20m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/60d801f9d503477cb13f649235de9b9e : SUCCESS in 21m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7ab2e356bf1464b809094a87fad8857 : SUCCESS in 19m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e46aa07e5f0140eeab78b2659041601f : SUCCESS in 17m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/05c91db6b9194c24876b2a7dda1c07dc : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28ccd139780043f29e8f4d9f7c6d0cc8 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c00063f550cd4062ad614a62f17ea6cf : SUCCESS in 19m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0545cb0da41e4cc8bc48bf8c08a1723d : SUCCESS in 12m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1203af35111b4922b04974ead094506c : SUCCESS in 15m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99dad4583f4140e4b1a8231fdb5675fd : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5930159ebccc4c6a9c0696aefbdb862b : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b1da1da062864193a45782639ecfc864 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/761a0fa755d9446bb4e12d97f1b1b3d1 : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c56daa74fef4738ad67a3ee1c7ab854 : SUCCESS in 56m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b109e08fb564b169927221979051def : SUCCESS in 36m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df78635462004d20ac7d993a643faa1e : SUCCESS in 42m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4b64d9705c34ba2808ee3b85b11aca2 : SUCCESS in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0dac43111d4abeb7bf98357ad60a00 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfa3c7937225473a8a6eb0f7fe95d531 : SUCCESS in 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65ddaa5b2e484e8a80ed452309397ece : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":7},{"id":"efbf98e4fdfa3ae9c90399ce84c7324ceec6d771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-11 20:34:52.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"fab29d451e1597217cc53bfa51bc32a66e0cfb60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 21:37:01.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d55a3e2ea0eb4932ba5f7bee216993e7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0672902bbadc45a68f9b10af866844db : FAILURE in 59m 01s","accounts_in_message":[],"_revision_number":8},{"id":"c31ae9e7cb59e6d6fd0211bde899cff9ee198357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-11 21:42:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7e1446f91cd54648ec010deb3e971a93bf897c5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 22:44:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db9295981c3c411593623e4bfe67f85f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71e78a6978dc4267883bc842baaf74fe : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":9},{"id":"d677c456f6b90164808d7eb67c35452c529f63b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 22:56:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23650dc302db44b9a2c9f4ce16b789e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7374682edf446f59706ec9f2d078bd3 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3b2494cfb184d57832a804145c6f09c : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/b5d723af2cb84c2c9cb0ff092b62ba57 : SUCCESS in 1h 04m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b83aad7b7e44e409f65ff6d3a46b441 : SUCCESS in 35m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd87124c4a664ebcaafe6d928d427ed8 : SUCCESS in 56m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e42ce8afd01a4ddb87fdc0c14d96243d : SUCCESS in 1h 01m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a822358328647aa9c11e6f9db6ca2f4 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64a831595b80481682a57e6e7e6f2f9f : SUCCESS in 12m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd6f94668b654e92a76d7d568c8096ed : SUCCESS in 12m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e117f46cdf5842ffa8acbd4c7039fdc7 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/986a56993c8643b6be8c9a29302f6492 : SUCCESS in 20m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0b1c760d0ac34524a821f059966f34a0 : SUCCESS in 22m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/35e7b718c6144b12a85c8f076305aad9 : SUCCESS in 19m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9c4c0f5ff14741438931471ca01bed89 : SUCCESS in 19m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f835f4cb9c78428bb3782dbfce19961c : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b451e727f16945dd9502cc96e3331272 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/23058a3ab5984224a98ddcc3724cc1a7 : SUCCESS in 14m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/38f2d1603d6845c7a41f10f05e2677cc : SUCCESS in 10m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5c922319c344baaac68ab3f79efff8c : SUCCESS in 5m 58s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c9993203ebad4c368e46bc4022bbfbf6 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c945f6bb2a3b48c18729cf5872775285 : SUCCESS in 1h 06m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c275ea35cb14d0893c43f0e52f97857 : SUCCESS in 55m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd99b4c94eda42b1bae30848d3d2cd88 : SUCCESS in 55m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abeac220d16e4c718ee6ef917f51a8d3 : SUCCESS in 28m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1903eb7190419cab73e908e13d8ff7 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/200753ea58f4467587901a68f70eeb79 : SUCCESS in 19m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69515de684574ed996d8e11561670e72 : SUCCESS in 47m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b4a6a037405c75dc15632c9a913d74d80c1f556f","revisions":{"56cbc89ad3aa6d0774b181d11dd7427526fe16a9":{"kind":"REWORK","_number":1,"created":"2019-01-09 20:40:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/1"}}},"commit":{"parents":[{"commit":"4781b80b1ca26d977e8a731ffe63e689b2e4338c","subject":"Merge \"Add slo_manifest_hook callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4781b80b1ca26d977e8a731ffe63e689b2e4338c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56cbc89ad3aa6d0774b181d11dd7427526fe16a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56cbc89ad3aa6d0774b181d11dd7427526fe16a9"}]},"branch":"refs/heads/master"},"dc4bc54b1d410a7d341f1b18a1550f5711c45d43":{"kind":"REWORK","_number":2,"created":"2019-01-09 23:03:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/2"}}},"commit":{"parents":[{"commit":"4781b80b1ca26d977e8a731ffe63e689b2e4338c","subject":"Merge \"Add slo_manifest_hook callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4781b80b1ca26d977e8a731ffe63e689b2e4338c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 23:03:02.000000000","tz":0},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc4bc54b1d410a7d341f1b18a1550f5711c45d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc4bc54b1d410a7d341f1b18a1550f5711c45d43"}]},"branch":"refs/heads/master"},"cd5b1dbfd1581053fe46cd60e2ab158a88249bae":{"kind":"REWORK","_number":3,"created":"2019-08-02 04:04:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/3"}}},"commit":{"parents":[{"commit":"d33e569bee1bd563d83b79b7aebef50ab2dc9726","subject":"Merge \"py3: Fix title-casing in HeaderKeyDict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d33e569bee1bd563d83b79b7aebef50ab2dc9726"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-02 04:03:15.000000000","tz":-420},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd5b1dbfd1581053fe46cd60e2ab158a88249bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd5b1dbfd1581053fe46cd60e2ab158a88249bae"}]},"branch":"refs/heads/master"},"6374a5f92e1f50e7d321312dedcafaebeef2bb13":{"kind":"REWORK","_number":4,"created":"2019-10-10 05:30:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/4"}}},"commit":{"parents":[{"commit":"1c0661e9203373a92b4eec19571f40309263376a","subject":"Merge \"Start building py3 docker images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c0661e9203373a92b4eec19571f40309263376a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-10 05:30:20.000000000","tz":-420},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nTwo big things:\n\n* Since the data_for_storage we pass to the slo_manifest_hook is\n  intentionally mutable, make sure the hook doesn\u0027t do anything *too*\n  crazy with it.\n\n* If the hook changed the large object\u0027s size (by adding or removing\n  segments, for example), make sure that is reflected in the final\n  manifest metadata.\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6374a5f92e1f50e7d321312dedcafaebeef2bb13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6374a5f92e1f50e7d321312dedcafaebeef2bb13"}]},"branch":"refs/heads/master"},"93d1842dcd043bb804d5c9f175ebac4c3bb88a38":{"kind":"REWORK","_number":5,"created":"2019-11-05 07:17:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/5"}}},"commit":{"parents":[{"commit":"36e2d74f8df7af860f26c7edb842e06396c916f5","subject":"Merge \"Tweak Swift AIO docs, update CentOS and Fedora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36e2d74f8df7af860f26c7edb842e06396c916f5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-05 07:15:52.000000000","tz":0},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nTwo big things:\n\n* Since the data_for_storage we pass to the slo_manifest_hook is\n  intentionally mutable, make sure the hook doesn\u0027t do anything *too*\n  crazy with it.\n\n* If the hook changed the large object\u0027s size (by adding or removing\n  segments, for example), make sure that is reflected in the final\n  manifest metadata.\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93d1842dcd043bb804d5c9f175ebac4c3bb88a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93d1842dcd043bb804d5c9f175ebac4c3bb88a38"}]},"branch":"refs/heads/master"},"5d2a2c2207fc92a0283eeb9e25374b8e4a31aa9f":{"kind":"REWORK","_number":6,"created":"2023-03-09 22:46:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/6"}}},"commit":{"parents":[{"commit":"049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06","subject":"Authors/ChangeLog for 2.31.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-09 22:46:04.000000000","tz":-480},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nTwo big things:\n\n* Since the data_for_storage we pass to the slo_manifest_hook is\n  intentionally mutable, make sure the hook doesn\u0027t do anything *too*\n  crazy with it.\n\n* If the hook changed the large object\u0027s size (by adding or removing\n  segments, for example), make sure that is reflected in the final\n  manifest metadata.\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d2a2c2207fc92a0283eeb9e25374b8e4a31aa9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d2a2c2207fc92a0283eeb9e25374b8e4a31aa9f"}]},"branch":"refs/heads/master"},"f8d4e2075a42a6163be206f863b875149ebcdf9a":{"kind":"REWORK","_number":7,"created":"2024-03-22 18:24:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/7"}}},"commit":{"parents":[{"commit":"537b6885e37cb1830992f8586400fe6f8d7f2345","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/537b6885e37cb1830992f8586400fe6f8d7f2345"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 18:23:45.000000000","tz":-420},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nTwo big things:\n\n* Since the data_for_storage we pass to the slo_manifest_hook is\n  intentionally mutable, make sure the hook doesn\u0027t do anything *too*\n  crazy with it.\n\n* If the hook changed the large object\u0027s size (by adding or removing\n  segments, for example), make sure that is reflected in the final\n  manifest metadata.\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8d4e2075a42a6163be206f863b875149ebcdf9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8d4e2075a42a6163be206f863b875149ebcdf9a"}]},"branch":"refs/heads/master"},"f826ceb6fb63418a59342109ed8b6fe29c9bd255":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2026-09-11 20:34:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/8"}}},"commit":{"parents":[{"commit":"9318950da879bd33b3b5216d0d0cfed4dea45d9c","subject":"Merge \"replicator: Quarantine suffixes we cannot delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9318950da879bd33b3b5216d0d0cfed4dea45d9c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-11 20:34:44.000000000","tz":-420},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nTwo big things:\n\n* Since the data_for_storage we pass to the slo_manifest_hook is\n  intentionally mutable, make sure the hook doesn\u0027t do anything *too*\n  crazy with it.\n\n* If the hook changed the large object\u0027s size (by adding or removing\n  segments, for example), make sure that is reflected in the final\n  manifest metadata.\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f826ceb6fb63418a59342109ed8b6fe29c9bd255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f826ceb6fb63418a59342109ed8b6fe29c9bd255"}]},"branch":"refs/heads/master"},"b4a6a037405c75dc15632c9a913d74d80c1f556f":{"kind":"REWORK","_number":9,"created":"2026-09-11 21:42:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/629659/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/629659/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/629659/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/629659/9"}}},"commit":{"parents":[{"commit":"9318950da879bd33b3b5216d0d0cfed4dea45d9c","subject":"Merge \"replicator: Quarantine suffixes we cannot delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9318950da879bd33b3b5216d0d0cfed4dea45d9c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-09 20:40:32.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-11 21:42:50.000000000","tz":-420},"subject":"slo_manifest_hook follow-up","message":"slo_manifest_hook follow-up\n\nTwo big things:\n\n* Since the data_for_storage we pass to the slo_manifest_hook is\n  intentionally mutable, make sure the hook doesn\u0027t do anything *too*\n  crazy with it.\n\n* If the hook changed the large object\u0027s size (by adding or removing\n  segments, for example), make sure that is reflected in the final\n  manifest metadata.\n\nChange-Id: I71222801023ce63b07b94374f2e340db5a982acd\nRelated-Change: I198c5196e0221a72b14597a06e5ce3c4b2bbf436\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4a6a037405c75dc15632c9a913d74d80c1f556f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4a6a037405c75dc15632c9a913d74d80c1f556f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
