)]}'
{"id":"openstack%2Fswift~630415","triplet_id":"openstack%2Fswift~master~I9f9c4360b54675bf5358a69ffad37897bce150e2","project":"openstack/swift","branch":"master","topic":"630415","hashtags":[],"change_id":"I9f9c4360b54675bf5358a69ffad37897bce150e2","subject":"Fix the handling of keystone groups in tempauth.","status":"NEW","created":"2019-01-12 02:59:30.000000000","updated":"2020-12-09 05:55:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b451916463e23975517d6aebdf2ea548e19b33a5","_number":630415,"virtual_id_number":630415,"owner":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-09 05:55:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-28 23:34:00.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-08-29 00:42:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22b76e0cfebb7ce1ecf4d32e4a8a08385047c6c0","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-01-12 02:59:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d4c8ad03dedcd5792f5e2d23eca84967c3c4662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-12 02:59:43.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"d76760f3f64c5e9d96c34f3f95f4918bf652642d","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-01-12 03:06:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6116544900c5d7d924703a9100a5f416e63145ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-12 03:42:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/630415/2/check/openstack-tox-docs/3ad682a/html/ : SUCCESS in 4m 09s\n- swift-tox-py27 http://logs.openstack.org/15/630415/2/check/swift-tox-py27/fbcd520/ : SUCCESS in 10m 45s\n- swift-tox-py35 http://logs.openstack.org/15/630415/2/check/swift-tox-py35/5e0f63b/ : SUCCESS in 5m 40s\n- swift-tox-py36 http://logs.openstack.org/15/630415/2/check/swift-tox-py36/b034d75/ : SUCCESS in 5m 29s (non-voting)\n- swift-tox-func http://logs.openstack.org/15/630415/2/check/swift-tox-func/5dbad4f/ : SUCCESS in 8m 18s\n- swift-tox-func-encryption http://logs.openstack.org/15/630415/2/check/swift-tox-func-encryption/58d704f/ : SUCCESS in 8m 10s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/630415/2/check/swift-tox-func-domain-remap-staticweb/cf23049/ : SUCCESS in 21m 44s\n- swift-tox-func-ec http://logs.openstack.org/15/630415/2/check/swift-tox-func-ec/8818a86/ : FAILURE in 22m 05s\n- swift-tox-func-s3api http://logs.openstack.org/15/630415/2/check/swift-tox-func-s3api/70399c0/ : SUCCESS in 10m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/630415/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4ea2f03/ : SUCCESS in 12m 03s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/630415/2/check/swift-probetests-centos-7/376f7ab/ : SUCCESS in 31m 02s\n- swift-dsvm-functional http://logs.openstack.org/15/630415/2/check/swift-dsvm-functional/ff924f8/ : SUCCESS in 29m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/630415/2/check/swift-dsvm-functional-ipv6/4eaca59/ : SUCCESS in 25m 18s\n- swift-tox-lower-constraints http://logs.openstack.org/15/630415/2/check/swift-tox-lower-constraints/6f181fe/ : SUCCESS in 9m 01s\n- openstack-tox-pep8 http://logs.openstack.org/15/630415/2/check/openstack-tox-pep8/d055e2b/ : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/630415/2/check/swift-multinode-rolling-upgrade/f629c5c/ : FAILURE in 21m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3f43ff32a2af9cc9200064ee1c443a9d73274767","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-01-14 18:04:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)\n\nI\u0027m a little worried about upgrades and py3 support...","accounts_in_message":[],"_revision_number":2},{"id":"c763be98f040660c2bebac7f7c03f4c5846c065c","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-05-04 22:16:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0bcfdc9cf79df654a6069fb4b934baa61db021c9","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-05-04 22:17:12.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nI suspect I\u0027ll need to iterate more on this, but addressed the comments from the last review.","accounts_in_message":[],"_revision_number":2},{"id":"5f6224291a5dc1de214b11daf267880a899b0557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 23:11:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/630415/3/check/openstack-tox-docs/939976d/html/ : SUCCESS in 3m 42s\n- swift-tox-py27 http://logs.openstack.org/15/630415/3/check/swift-tox-py27/c5a2971/ : FAILURE in 17m 33s\n- swift-tox-py36 http://logs.openstack.org/15/630415/3/check/swift-tox-py36/8e017ea/ : FAILURE in 13m 48s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/630415/3/check/swift-tox-py37/3cb84d3/ : FAILURE in 13m 20s\n- swift-tox-func http://logs.openstack.org/15/630415/3/check/swift-tox-func/62c9bce/ : SUCCESS in 12m 52s\n- swift-tox-func-encryption http://logs.openstack.org/15/630415/3/check/swift-tox-func-encryption/c2d361e/ : SUCCESS in 15m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/630415/3/check/swift-tox-func-domain-remap-staticweb/e50d7ae/ : SUCCESS in 13m 09s\n- swift-tox-func-ec http://logs.openstack.org/15/630415/3/check/swift-tox-func-ec/047ba70/ : SUCCESS in 15m 19s\n- swift-tox-func-s3api http://logs.openstack.org/15/630415/3/check/swift-tox-func-s3api/995a846/ : SUCCESS in 17m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/630415/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a5c1a1d/ : SUCCESS in 17m 17s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/630415/3/check/swift-probetests-centos-7/229f1e8/ : SUCCESS in 29m 41s\n- swift-dsvm-functional http://logs.openstack.org/15/630415/3/check/swift-dsvm-functional/bc2493a/ : SUCCESS in 37m 06s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/630415/3/check/swift-dsvm-functional-ipv6/7a105ac/ : SUCCESS in 34m 19s\n- swift-tox-lower-constraints http://logs.openstack.org/15/630415/3/check/swift-tox-lower-constraints/e6535e8/ : FAILURE in 11m 15s\n- openstack-tox-pep8 http://logs.openstack.org/15/630415/3/check/openstack-tox-pep8/8899424/ : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/630415/3/check/swift-multinode-rolling-upgrade/37f6cde/ : FAILURE in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc3546b9192ffb549d2a3ca6f254ce7658daa605","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-05-06 22:33:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"98e8dd2e50c032af9323fb39d6bff8e030b9c354","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 23:07:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/630415/4/check/openstack-tox-docs/21a7fcf/html/ : SUCCESS in 3m 41s\n- swift-tox-py27 http://logs.openstack.org/15/630415/4/check/swift-tox-py27/ca73872/ : SUCCESS in 12m 42s\n- swift-tox-py36 http://logs.openstack.org/15/630415/4/check/swift-tox-py36/7128cb4/ : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/630415/4/check/swift-tox-py37/440903a/ : SUCCESS in 9m 16s\n- swift-tox-func http://logs.openstack.org/15/630415/4/check/swift-tox-func/32e927f/ : SUCCESS in 10m 13s\n- swift-tox-func-encryption http://logs.openstack.org/15/630415/4/check/swift-tox-func-encryption/a9563f0/ : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/630415/4/check/swift-tox-func-domain-remap-staticweb/619a79d/ : SUCCESS in 10m 25s\n- swift-tox-func-ec http://logs.openstack.org/15/630415/4/check/swift-tox-func-ec/5de1989/ : SUCCESS in 9m 06s\n- swift-tox-func-s3api http://logs.openstack.org/15/630415/4/check/swift-tox-func-s3api/af2efe3/ : SUCCESS in 12m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/630415/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/388e6bd/ : SUCCESS in 10m 59s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/630415/4/check/swift-probetests-centos-7/0abe059/ : SUCCESS in 29m 18s\n- swift-dsvm-functional http://logs.openstack.org/15/630415/4/check/swift-dsvm-functional/9562962/ : SUCCESS in 26m 18s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/630415/4/check/swift-dsvm-functional-ipv6/8f82391/ : SUCCESS in 25m 20s\n- swift-tox-lower-constraints http://logs.openstack.org/15/630415/4/check/swift-tox-lower-constraints/6a0e334/ : SUCCESS in 11m 15s\n- openstack-tox-pep8 http://logs.openstack.org/15/630415/4/check/openstack-tox-pep8/b7633ff/ : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/630415/4/check/swift-multinode-rolling-upgrade/47fc118/ : FAILURE in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eca167a0a1735e74b38bf170cc68e3d7fd62f8bf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-28 23:28:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"69eee6af6bef941e5f817352b6adf4d1c251789f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-28 23:34:00.000000000","message":"Patch Set 5:\n\n(4 comments)\n\nRebased, so we can have some py3 func tests run against it.\n\nI\u0027m gonna be real interested in the py3 dsvm job...","accounts_in_message":[],"_revision_number":5},{"id":"c6fcb0907f6b8aa65d2fa439c2cf8e4578838564","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 00:42:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b092609d9a7841e4961ecd917d95dc9f : SUCCESS in 5m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eb1095f0533642a194da8beecdabbdd2 : SUCCESS in 1h 08m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e00b3f1668b4202b185e5363b3e3561 : SUCCESS in 54m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23b2ab0bd5d84a1193287079011eb479 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a23c349fee86400eb6badf2b5480b9b3 : SUCCESS in 10m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/faeced1891ef474583eaefe8dfaa652a : SUCCESS in 12m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/426ea6a3bd8b4c8b80a959c27d23c3c5 : SUCCESS in 12m 26s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8fbe7cc276d64c4e8aede9e2a4d5601a : SUCCESS in 11m 07s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/27ceda4015704d55aaeb78753ab1d1bd : SUCCESS in 9m 26s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/4cd2cbd01ed148bf92b895a3a5886aa6 : SUCCESS in 10m 34s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/5b0d379e0b9f432389186fbb460ecd45 : SUCCESS in 9m 26s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/524f4899dd7b421fb5f87981f48d530a : SUCCESS in 11m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1c38172b5904e3f9f9675ba4b8118ee : SUCCESS in 10m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/04c174f5fa044f83a1f444dcc6ef5602 : SUCCESS in 11m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ecb41dc963614b199c0790d724feb8a1 : SUCCESS in 10m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cc04888a1dde4284abfd8d5e97a78533 : SUCCESS in 10m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7dd6891df7bd4a058527d987503b7f29 : SUCCESS in 11m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dba376b41bef48419fc644b136525309 : SUCCESS in 16m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f199724604b0446c9ff2de8a2b568f09 : SUCCESS in 31m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12e9f49e918649a1b9bc1134934b3ef0 : SUCCESS in 30m 34s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c874d553097641588ef873d08f62a4ff : SUCCESS in 32m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66c749579a5c43529fd539fa8b070fa4 : SUCCESS in 31m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3144775841a486f83f0c87821909555 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ef480a9af442bc8bd8e12d8e6865e1 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9229a275b03d43bf848f2024be0427e4 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":5},{"id":"cec4a3ee0411486dacffaa8c501542dffc2edd32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 18:50:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d43a8c72f7a6190441137ec4f15fc4c70bba8f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 19:51:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90783f5ef21e40f18cf310afab0f7544 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/92c5d272ed934857b195e10fdc065614 : SUCCESS in 54m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96304eab2cc4409e98085e4b25e914a4 : SUCCESS in 43m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a9123ff55454d2d9e7f676a1f9727e7 : SUCCESS in 9m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e613a57db8be410e9e4dbb520f3fce98 : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/579f8cec1fff422188ec1d3a88f2c044 : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d30065eb2c04a32ae6750b0ada5488e : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04482dff11ec487f987601c7047e8997 : SUCCESS in 15m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/47b6d1784960483daca077e05785dc8d : SUCCESS in 12m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32072a56225d484984b1f9f78a0d918d : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70a22a58d32f49059a3dc9cc0cbe7574 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d55a269837414defbf03b39be0997c71 : SUCCESS in 16m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1a2049bab3c74e60bbbfafd3f7d2f28c : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1905601788cd4bf9b6a023dd02cb84d9 : SUCCESS in 20m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e112770512d34e7ca3e2189ab58ab778 : SUCCESS in 21m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbdabf6a72c249fe8c41863a07033b41 : SUCCESS in 18m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/993d98b69f9f4b00a771077f0a64fddf : FAILURE in 47m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e41ac88bfda144c88bbfb38f6746f1f2 : SUCCESS in 45m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/705b2ab003014a03ba07028d6afc8360 : SUCCESS in 42m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef5a4a64e3cf48b489dfe4d03c084c16 : SUCCESS in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/184bc308dc134d5ea5904fa6a12fe15a : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6fee81363ac14a2ca115913f27443b32 : FAILURE in 8m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d72dec8c1d474518b3543399579b5dad : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":6},{"id":"b4375ac2aee92ab7dab2a667fc484cd5087cab16","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 20:27:57.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"56a3e0e654048446c0cc745b3e80dd94f6344875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 21:29:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5bbb814f6f48ca9eec08365b716052 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/597a6518616f4ab8b9483458945622fe : SUCCESS in 55m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9543d75254a4c448e90e0aa0eaf710c : SUCCESS in 43m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb70842d9dfa412386ea9acf5fad997b : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a791e9fb392844b8bc782bd4b6ccbba5 : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60430688ebec4f5da3a2dc03072016f6 : SUCCESS in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dffc4762a7cf44cfad3e19e0210f862b : SUCCESS in 13m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b4113a1a4a84167aa88bee52caf1dc2 : SUCCESS in 52m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/806a4d15ad2d4551951b3fc26f2b15b1 : SUCCESS in 14m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/354e56bd81264780b48eba436ac59053 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2427d1e6a1a344e19b84a25993609749 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4616d87527649bfacea63c2e529c877 : SUCCESS in 24m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cc3cf7bdfbb54d1daa6dab484ffb96bc : SUCCESS in 23m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/51f3d58b9da6422cbfc4031bd6706438 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/74ccf9f070864e7ca3e18c95caaa8ba8 : SUCCESS in 18m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3b9958648b1a41cea766511dc02a94ec : SUCCESS in 20m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/75a24c36eb2a4531bbab99f9aa8149a2 : FAILURE in 46m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/382394a2c1f44fe88219a2844916f75c : SUCCESS in 52m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea99b6b906574913b3b27de19b13546c : SUCCESS in 44m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61e5823312ed47eaa8abd5e069aef08c : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1e23a749c045d18b2db7697cf8bcd3 : SUCCESS in 8m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df90acefad3f45ac8d504f44843ea6b0 : FAILURE in 9m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c41f33262e84091bf30870e164fce11 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":6},{"id":"42c078f549065583a834e0eef96e46b031997d41","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 22:17:49.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"421703456fe869d1ac93410da77187c541f45376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 23:23:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0021aa24ecc044e9a861c8b98283d051 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/d60e8d3cf39f4a99aadb2d2ba9b8bf9a : SUCCESS in 1h 04m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc6da8c33a7a47eca191abf68f67e117 : SUCCESS in 40m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b24f215a917b49eeb67070ed002da8ca : FAILURE in 3m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/649ba012de674fc699deea79d5e0fee6 : FAILURE in 3m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4c856fcf781b44279987d731d863ad69 : SUCCESS in 24m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4fe9bd198c8e4ebb8ad355a37cbfbe11 : SUCCESS in 12m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1dd1a9b7cb914932ad6879a7e6bba545 : SUCCESS in 23m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8abf28d192994121813e98a8452aae49 : SUCCESS in 12m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc87509022c3459ea273dec068ebe3e2 : SUCCESS in 28m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b9940fdaaf444ffbe9de1b23103473f : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c002a6020aae41d3b5b5b3afb535d3c7 : SUCCESS in 19m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c2124f14b99a436284ff52feff7dcc5d : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/067c4071927a47a4a37e0e2b65d46cdc : SUCCESS in 18m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8b7a87b5d4ac45399efe7f696570309d : SUCCESS in 30m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/424ffa52e72d4580af6c58302cf23ae2 : POST_FAILURE in 5m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e11aef21a0764e49909a0bec65446ae2 : FAILURE in 50m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d298b1383724690a43bf643b811bd6b : SUCCESS in 33m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db3f7db99fb34cdc8ef028a3c0ce1278 : SUCCESS in 42m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31820907ffe14a4da54de45e780e4b1c : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca8349bc85846cabbe47d1493dc2466 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee30ab61db1f4578a04dadd11fef404f : FAILURE in 9m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb09b297c0574d53b36be558f4e19ff1 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":6},{"id":"af66c8bb2c2c0083d6b261d70f13c99f23f92551","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 05:01:08.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b451916463e23975517d6aebdf2ea548e19b33a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 05:55:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2cf5562f0740ca8b10ab7332db721d : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/06f47c882e4b4a50825ff4ef3037327c : SUCCESS in 52m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fcbaba0f7d034d24b44e6e277082efa1 : SUCCESS in 43m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de369ccfb9f14684a41dc0e7a1aec534 : SUCCESS in 18m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/901ba541fec84fb8b0cd908d2b937efe : SUCCESS in 14m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6dacc7b6fc3d4318bcf5f7b02aa993ae : SUCCESS in 12m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd988cc74f964d0a90f9a0b5bfd44dcb : SUCCESS in 15m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/762ef85d1c914ed7bf4b2defad86880a : SUCCESS in 13m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/41871c5a5c36442aa633d6216a62a7f6 : SUCCESS in 12m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b5dd39b46fc14f39a33e7bbb5f24a036 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/197971ccd3ab436eb7b3d0cd66a0ced4 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d68d7462c890404b90772e24364f0aaa : SUCCESS in 16m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ab4b926765b14a10804e4b2e3e32609f : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/436475ad05244c4a8c22398c02ae58ad : SUCCESS in 17m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9293b56678824c93a6c4a97d085f731c : SUCCESS in 19m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7c855e90041447590ea46f06de8aac7 : POST_FAILURE in 5m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d3164dbb3d624085af1ab7b8eba460c1 : SUCCESS in 46m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afcbdd5109894a079d47e3c7b9430ecf : SUCCESS in 42m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2ce9cbd2ab6d49c3986917c934bfed9a : SUCCESS in 46m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba526e90d7e445c286a0b77ab0d6da77 : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf28002314e44e1fb9fad8da725ff990 : SUCCESS in 7m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/afbcd61018be45cfad0b2d3eebf982f6 : FAILURE in 8m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59f82341913e4c479d87f830b03a7d15 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f0db5fcd890c11eddd133a71ff54e9e77fd696a3","revisions":{"41c849d6f1c624a860e8d8810eba7c8a24278338":{"kind":"REWORK","_number":1,"created":"2019-01-12 02:59:30.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/15/630415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/630415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/630415/1"}}},"commit":{"parents":[{"commit":"7a425e606522db5c012f0e981cd069a878caecb1","subject":"Handle non-ASCII characters in S3 v2 listings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a425e606522db5c012f0e981cd069a878caecb1"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 02:57:32.000000000","tz":-480},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 02:57:32.000000000","tz":-480},"subject":"Fix the handling of keystone groups in tempauth.","message":"Fix the handling of keystone groups in tempauth.\n\nStandardizes on using NULL to separate groups and fixes the behavior\nwhere keystoneauth would insert a tuple for remote_user.\n\nChange-Id: I9f9c4360b54675bf5358a69ffad37897bce150e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41c849d6f1c624a860e8d8810eba7c8a24278338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41c849d6f1c624a860e8d8810eba7c8a24278338"}]},"branch":"refs/heads/master"},"9aec3b2d0f84b286295dc9def470bfdc18408fa3":{"kind":"REWORK","_number":2,"created":"2019-01-12 03:06:45.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/15/630415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/630415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/630415/2"}}},"commit":{"parents":[{"commit":"d1e9a8e0cbe0ca562311256a861ced1159e3e158","subject":"Merge \"s3api: Stop relying on container listings during multipart uploads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1e9a8e0cbe0ca562311256a861ced1159e3e158"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 02:57:32.000000000","tz":-480},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 03:05:42.000000000","tz":-480},"subject":"Fix the handling of keystone groups in tempauth.","message":"Fix the handling of keystone groups in tempauth.\n\nStandardizes on using NULL to separate groups and fixes the behavior\nwhere keystoneauth would insert a tuple for remote_user.\n\nChange-Id: I9f9c4360b54675bf5358a69ffad37897bce150e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9aec3b2d0f84b286295dc9def470bfdc18408fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9aec3b2d0f84b286295dc9def470bfdc18408fa3"}]},"branch":"refs/heads/master"},"9f3b6279fcea71a9fa1b9c4d3c02008491e60352":{"kind":"REWORK","_number":3,"created":"2019-05-04 22:16:18.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/15/630415/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/630415/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/630415/3"}}},"commit":{"parents":[{"commit":"f55167a735632cdc89fb6f83f97e4fe3ed09b7cc","subject":"Merge \"Increase node_timeout in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f55167a735632cdc89fb6f83f97e4fe3ed09b7cc"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 02:57:32.000000000","tz":-480},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-05-04 22:16:01.000000000","tz":-420},"subject":"Fix the handling of keystone groups in tempauth.","message":"Fix the handling of keystone groups in tempauth.\n\nStandardizes on using NULL to separate groups and fixes the behavior\nwhere keystoneauth would insert a tuple for remote_user.\n\nChange-Id: I9f9c4360b54675bf5358a69ffad37897bce150e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f3b6279fcea71a9fa1b9c4d3c02008491e60352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f3b6279fcea71a9fa1b9c4d3c02008491e60352"}]},"branch":"refs/heads/master"},"7948246f9fc74751926d02841d6e51720e6b4541":{"kind":"REWORK","_number":4,"created":"2019-05-06 22:33:00.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/15/630415/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/630415/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/630415/4"}}},"commit":{"parents":[{"commit":"f55167a735632cdc89fb6f83f97e4fe3ed09b7cc","subject":"Merge \"Increase node_timeout in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f55167a735632cdc89fb6f83f97e4fe3ed09b7cc"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 02:57:32.000000000","tz":-480},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-05-06 22:32:40.000000000","tz":-420},"subject":"Fix the handling of keystone groups in tempauth.","message":"Fix the handling of keystone groups in tempauth.\n\nStandardizes on using NULL to separate groups and fixes the behavior\nwhere keystoneauth would insert a tuple for remote_user.\n\nChange-Id: I9f9c4360b54675bf5358a69ffad37897bce150e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7948246f9fc74751926d02841d6e51720e6b4541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7948246f9fc74751926d02841d6e51720e6b4541"}]},"branch":"refs/heads/master"},"4694077a7da937c34cf415aa980e42e342893268":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-28 23:28:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/630415/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/630415/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/630415/5"}}},"commit":{"parents":[{"commit":"a32fb30c16062ea64488e918077d635645e33e47","subject":"Use SOURCE_DATE_EPOCH in docs to make build reproducible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a32fb30c16062ea64488e918077d635645e33e47"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 02:57:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-28 17:24:06.000000000","tz":-420},"subject":"Fix the handling of keystone groups in tempauth.","message":"Fix the handling of keystone groups in tempauth.\n\nStandardizes on using NULL to separate groups and fixes the behavior\nwhere keystoneauth would insert a tuple for remote_user.\n\nChange-Id: I9f9c4360b54675bf5358a69ffad37897bce150e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4694077a7da937c34cf415aa980e42e342893268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4694077a7da937c34cf415aa980e42e342893268"}]},"branch":"refs/heads/master"},"f0db5fcd890c11eddd133a71ff54e9e77fd696a3":{"kind":"REWORK","_number":6,"created":"2020-12-08 18:50:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/630415/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/630415/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/630415/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/630415/6"}}},"commit":{"parents":[{"commit":"fe1f1e8271d3c6cdc773b447a55c6a5cdb3efd1d","subject":"Merge \"Give unit tests a second chance to pass\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe1f1e8271d3c6cdc773b447a55c6a5cdb3efd1d"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-01-12 02:57:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-08 18:49:04.000000000","tz":-480},"subject":"Fix the handling of keystone groups in tempauth.","message":"Fix the handling of keystone groups in tempauth.\n\nStandardizes on using NULL to separate groups and fixes the behavior\nwhere keystoneauth would insert a tuple for remote_user.\n\nChange-Id: I9f9c4360b54675bf5358a69ffad37897bce150e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0db5fcd890c11eddd133a71ff54e9e77fd696a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0db5fcd890c11eddd133a71ff54e9e77fd696a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
