)]}'
{"id":"openstack%2Fswift~636217","triplet_id":"openstack%2Fswift~master~I334ef517d833916a3b7be3151a812d4f9c66a6e1","project":"openstack/swift","branch":"master","topic":"slow-probe-fix","hashtags":[],"change_id":"I334ef517d833916a3b7be3151a812d4f9c66a6e1","subject":"Increase node_timeout in gate","status":"MERGED","created":"2019-02-11 20:43:28.000000000","updated":"2019-04-30 22:12:01.000000000","submitted":"2019-04-30 22:12:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636217-1556662321907-e317d9ab","meta_rev_id":"0ae12228849f61b398a0e8c791652630256242e9","_number":636217,"virtual_id_number":636217,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-04-30 22:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-02-27 03:10:21.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-04-30 19:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-02-12 18:36:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-30 20:13:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-27 03:10:21.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-04-30 20:13:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-04-30 22:12:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d065744ba89f530f6844ed6d3a9b0c32e7705613","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-02-11 20:43:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"accb2180de4fb808c84564aca0c8e5f68182c89f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-02-11 21:18:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a8dca7fd211384c377b4973c2c4bd5eef5eb377","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 21:38:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/636217/1/check/openstack-tox-docs/62dc525/html/ : SUCCESS in 5m 32s\n- swift-tox-py27 http://logs.openstack.org/17/636217/1/check/swift-tox-py27/f81108a/ : SUCCESS in 11m 15s\n- swift-tox-py35 http://logs.openstack.org/17/636217/1/check/swift-tox-py35/405228b/ : SUCCESS in 7m 26s\n- swift-tox-py36 http://logs.openstack.org/17/636217/1/check/swift-tox-py36/bdcd5a3/ : SUCCESS in 6m 53s (non-voting)\n- swift-tox-func http://logs.openstack.org/17/636217/1/check/swift-tox-func/536a97a/ : SUCCESS in 9m 29s\n- swift-tox-func-encryption http://logs.openstack.org/17/636217/1/check/swift-tox-func-encryption/b82d4d2/ : SUCCESS in 10m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/636217/1/check/swift-tox-func-domain-remap-staticweb/975aef1/ : SUCCESS in 10m 04s\n- swift-tox-func-ec http://logs.openstack.org/17/636217/1/check/swift-tox-func-ec/35599a3/ : SUCCESS in 9m 30s\n- swift-tox-func-s3api http://logs.openstack.org/17/636217/1/check/swift-tox-func-s3api/1390261/ : SUCCESS in 11m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/17/636217/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/83f2753/ : SUCCESS in 12m 39s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/17/636217/1/check/swift-probetests-centos-7/af4600d/ : SUCCESS in 30m 19s\n- swift-dsvm-functional http://logs.openstack.org/17/636217/1/check/swift-dsvm-functional/297eefa/ : SUCCESS in 27m 43s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/636217/1/check/swift-dsvm-functional-ipv6/a982b4a/ : SUCCESS in 24m 31s\n- swift-tox-lower-constraints http://logs.openstack.org/17/636217/1/check/swift-tox-lower-constraints/2cf6b6f/ : SUCCESS in 11m 07s\n- openstack-tox-pep8 http://logs.openstack.org/17/636217/1/check/openstack-tox-pep8/8b57461/ : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/636217/1/check/swift-multinode-rolling-upgrade/ccc890d/ : SUCCESS in 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a96bc4e747a7d19e66ccc54c8ba01efd4e7bf5e0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-02-11 21:48:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b0bf20c0e2d276230d91912f64078531464d192","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-11 21:50:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good: http://logs.openstack.org/17/636217/1/check/swift-multinode-rolling-upgrade/ccc890d/proxy-server.conf now says\n\n [app:proxy-server]\n use \u003d egg:swift#proxy\n allow_account_management \u003d true\n account_autocreate \u003d true\n node_timeout \u003d 20\n\nWe should maybe do something similar for container updates, too...","accounts_in_message":[],"_revision_number":2},{"id":"aedc70cc14d11a2c1b2135618215555857710bad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 22:22:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/17/636217/2/check/openstack-tox-docs/5de3df9/html/ : SUCCESS in 4m 32s\n- swift-tox-py27 http://logs.openstack.org/17/636217/2/check/swift-tox-py27/b3e8aa3/ : SUCCESS in 10m 47s\n- swift-tox-py35 http://logs.openstack.org/17/636217/2/check/swift-tox-py35/1ca4818/ : SUCCESS in 7m 29s\n- swift-tox-py36 http://logs.openstack.org/17/636217/2/check/swift-tox-py36/a257d64/ : SUCCESS in 8m 00s (non-voting)\n- swift-tox-func http://logs.openstack.org/17/636217/2/check/swift-tox-func/92ecf01/ : SUCCESS in 9m 03s\n- swift-tox-func-encryption http://logs.openstack.org/17/636217/2/check/swift-tox-func-encryption/b903e1f/ : SUCCESS in 8m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/636217/2/check/swift-tox-func-domain-remap-staticweb/8273f0b/ : SUCCESS in 9m 03s\n- swift-tox-func-ec http://logs.openstack.org/17/636217/2/check/swift-tox-func-ec/acc4cc4/ : SUCCESS in 10m 52s\n- swift-tox-func-s3api http://logs.openstack.org/17/636217/2/check/swift-tox-func-s3api/14c9ee7/ : SUCCESS in 10m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/17/636217/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e05b096/ : SUCCESS in 13m 17s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/17/636217/2/check/swift-probetests-centos-7/009a2ba/ : FAILURE in 8m 39s\n- swift-dsvm-functional http://logs.openstack.org/17/636217/2/check/swift-dsvm-functional/3cdaebe/ : SUCCESS in 28m 04s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/636217/2/check/swift-dsvm-functional-ipv6/4d68956/ : SUCCESS in 21m 56s\n- swift-tox-lower-constraints http://logs.openstack.org/17/636217/2/check/swift-tox-lower-constraints/91eb3c9/ : SUCCESS in 10m 42s\n- openstack-tox-pep8 http://logs.openstack.org/17/636217/2/check/openstack-tox-pep8/49a66e2/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/636217/2/check/swift-multinode-rolling-upgrade/1672998/ : SUCCESS in 18m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8fa44badce42c60aaf065797ee35cb1bc9a1534b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-11 23:45:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b7bcba6158c3d7475d3ead335e5f8af6a3e0fae4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-11 23:46:42.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"da98ed6f530e41d7d1fd1d0a6cc542366bd06c48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 00:23:45.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py27-centos-7 http://logs.openstack.org/17/636217/2/experimental/swift-tox-py27-centos-7/4fd5053/ : FAILURE in 15m 18s\n- swift-tox-func-centos-7 http://logs.openstack.org/17/636217/2/experimental/swift-tox-func-centos-7/bed4c94/ : SUCCESS in 11m 50s\n- swift-tox-func-encryption-centos-7 http://logs.openstack.org/17/636217/2/experimental/swift-tox-func-encryption-centos-7/5f51f0f/ : SUCCESS in 10m 29s\n- swift-tox-func-ec-centos-7 http://logs.openstack.org/17/636217/2/experimental/swift-tox-func-ec-centos-7/c450584/ : SUCCESS in 12m 28s\n- swift-multinode-rolling-upgrade-queens http://logs.openstack.org/17/636217/2/experimental/swift-multinode-rolling-upgrade-queens/291d483/ : SUCCESS in 18m 06s\n- swift-multinode-rolling-upgrade-rocky http://logs.openstack.org/17/636217/2/experimental/swift-multinode-rolling-upgrade-rocky/8b7cbda/ : SUCCESS in 20m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6f7712503ed33e297984f930bca527f1d2a25f9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 01:44:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/17/636217/2/check/openstack-tox-docs/5f6f2a7/html/ : SUCCESS in 4m 35s\n- swift-tox-py27 http://logs.openstack.org/17/636217/2/check/swift-tox-py27/2404c3b/ : SUCCESS in 10m 44s\n- swift-tox-py35 http://logs.openstack.org/17/636217/2/check/swift-tox-py35/425d140/ : SUCCESS in 6m 45s\n- swift-tox-py36 http://logs.openstack.org/17/636217/2/check/swift-tox-py36/a7ffb28/ : SUCCESS in 7m 18s (non-voting)\n- swift-tox-func http://logs.openstack.org/17/636217/2/check/swift-tox-func/acb53b4/ : SUCCESS in 9m 34s\n- swift-tox-func-encryption http://logs.openstack.org/17/636217/2/check/swift-tox-func-encryption/11df9e9/ : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/636217/2/check/swift-tox-func-domain-remap-staticweb/afc8682/ : SUCCESS in 9m 01s\n- swift-tox-func-ec http://logs.openstack.org/17/636217/2/check/swift-tox-func-ec/f2fb6e3/ : SUCCESS in 8m 48s\n- swift-tox-func-s3api http://logs.openstack.org/17/636217/2/check/swift-tox-func-s3api/db41c17/ : SUCCESS in 11m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/17/636217/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e5a2da1/ : SUCCESS in 17m 29s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/17/636217/2/check/swift-probetests-centos-7/226d27e/ : FAILURE in 7m 03s\n- swift-dsvm-functional http://logs.openstack.org/17/636217/2/check/swift-dsvm-functional/ccc27ef/ : SUCCESS in 28m 26s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/636217/2/check/swift-dsvm-functional-ipv6/9e8e1aa/ : SUCCESS in 26m 30s\n- swift-tox-lower-constraints http://logs.openstack.org/17/636217/2/check/swift-tox-lower-constraints/66ee339/ : SUCCESS in 11m 51s\n- openstack-tox-pep8 http://logs.openstack.org/17/636217/2/check/openstack-tox-pep8/da7ff68/ : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/636217/2/check/swift-multinode-rolling-upgrade/b8188ed/ : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":2},{"id":"0de6712befd5ef91f729b8b31d0bf465056c3dbb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-12 02:11:26.000000000","message":"Patch Set 2: Code-Review-1 -Workflow\n\nSo I\u0027m realizing that https://github.com/openstack/swift/blob/2.20.0/test/probe/test_container_failures.py#L145-L176 depends on us having a node_timeout \u003c 15s...","accounts_in_message":[],"_revision_number":2},{"id":"5685c4fb8e12d1e0ea2a52b6d245369243221370","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-02-12 15:27:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6d3da0e336421dd378d027cd9a17c550bba8f1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 16:08:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/17/636217/3/check/openstack-tox-docs/7493539/html/ : SUCCESS in 4m 39s\n- swift-tox-py27 http://logs.openstack.org/17/636217/3/check/swift-tox-py27/5f6a2f4/ : SUCCESS in 16m 07s\n- swift-tox-py35 http://logs.openstack.org/17/636217/3/check/swift-tox-py35/6a28898/ : SUCCESS in 7m 00s\n- swift-tox-py36 http://logs.openstack.org/17/636217/3/check/swift-tox-py36/2bcaf5c/ : SUCCESS in 6m 31s (non-voting)\n- swift-tox-func http://logs.openstack.org/17/636217/3/check/swift-tox-func/f641ed0/ : SUCCESS in 8m 56s\n- swift-tox-func-encryption http://logs.openstack.org/17/636217/3/check/swift-tox-func-encryption/036cfe0/ : SUCCESS in 10m 43s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/636217/3/check/swift-tox-func-domain-remap-staticweb/abbbc55/ : FAILURE in 10m 49s\n- swift-tox-func-ec http://logs.openstack.org/17/636217/3/check/swift-tox-func-ec/765dee3/ : SUCCESS in 9m 33s\n- swift-tox-func-s3api http://logs.openstack.org/17/636217/3/check/swift-tox-func-s3api/ea9e8a3/ : SUCCESS in 10m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/17/636217/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7afa7a8/ : SUCCESS in 21m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/17/636217/3/check/swift-probetests-centos-7/36d0433/ : FAILURE in 10m 40s\n- swift-dsvm-functional http://logs.openstack.org/17/636217/3/check/swift-dsvm-functional/ea20385/ : SUCCESS in 25m 50s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/636217/3/check/swift-dsvm-functional-ipv6/099a371/ : SUCCESS in 25m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/17/636217/3/check/swift-tox-lower-constraints/168cef5/ : SUCCESS in 10m 13s\n- openstack-tox-pep8 http://logs.openstack.org/17/636217/3/check/openstack-tox-pep8/ebac3d8/ : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/636217/3/check/swift-multinode-rolling-upgrade/0e616ef/ : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":3},{"id":"6f86c4c05e1b222956de5c95061b9e2cffd0e0fd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-02-12 16:41:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"777948c8dd471d7470739c91e0a49a5b90f658b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 18:21:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/17/636217/4/check/openstack-tox-docs/d34c728/html/ : SUCCESS in 4m 03s\n- swift-tox-py27 http://logs.openstack.org/17/636217/4/check/swift-tox-py27/9cc30b6/ : FAILURE in 12m 46s\n- swift-tox-py35 http://logs.openstack.org/17/636217/4/check/swift-tox-py35/3472ebe/ : SUCCESS in 6m 06s\n- swift-tox-py36 http://logs.openstack.org/17/636217/4/check/swift-tox-py36/236e945/ : SUCCESS in 7m 20s (non-voting)\n- swift-tox-func http://logs.openstack.org/17/636217/4/check/swift-tox-func/109ab59/ : SUCCESS in 9m 34s\n- swift-tox-func-encryption http://logs.openstack.org/17/636217/4/check/swift-tox-func-encryption/361d0d1/ : SUCCESS in 8m 56s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/636217/4/check/swift-tox-func-domain-remap-staticweb/feaf9de/ : SUCCESS in 9m 15s\n- swift-tox-func-ec http://logs.openstack.org/17/636217/4/check/swift-tox-func-ec/20b1d69/ : SUCCESS in 9m 54s\n- swift-tox-func-s3api http://logs.openstack.org/17/636217/4/check/swift-tox-func-s3api/d77573f/ : SUCCESS in 10m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/17/636217/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fe51676/ : SUCCESS in 15m 41s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/17/636217/4/check/swift-probetests-centos-7/c3d2a15/ : SUCCESS in 31m 25s\n- swift-dsvm-functional http://logs.openstack.org/17/636217/4/check/swift-dsvm-functional/9558c31/ : SUCCESS in 25m 04s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/636217/4/check/swift-dsvm-functional-ipv6/87ece5b/ : SUCCESS in 25m 17s\n- swift-tox-lower-constraints http://logs.openstack.org/17/636217/4/check/swift-tox-lower-constraints/c7d77e8/ : SUCCESS in 16m 37s\n- openstack-tox-pep8 http://logs.openstack.org/17/636217/4/check/openstack-tox-pep8/9ffde1a/ : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/636217/4/check/swift-multinode-rolling-upgrade/9a0ede4/ : SUCCESS in 19m 04s","accounts_in_message":[],"_revision_number":4},{"id":"fa75332242d8582908bcb9f1bc74b01a1b28ab74","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-12 18:36:55.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)\n\nrecheck\n\nIDK -- I could get behind this; it\u0027s definitely better... but I worry a bit that we\u0027re adding some more complication without actually guaranteeing that we\u0027ve fixed it...","accounts_in_message":[],"_revision_number":4},{"id":"a3534b8906b64faf4f4f888cd612e39de692f4a1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-02-13 16:34:52.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nI don\u0027t know how to make the gate have less slow vms...\n\nI\u0027m pretty sure upping node_timeout will make the gate tests more reliable... \n\nThis is just the best I\u0027ve come up with so far...\n\nWe can probably do better...","accounts_in_message":[],"_revision_number":4},{"id":"f7f3265292dbe32e806c91536a332750d288ff46","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-16 00:02:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a5f10d799bd6c56f177e8566add7a466658ddaa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-16 01:14:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/636217/4/check/openstack-tox-docs/577e341/html/ : SUCCESS in 5m 06s\n- swift-tox-py27 http://logs.openstack.org/17/636217/4/check/swift-tox-py27/1344024/ : SUCCESS in 11m 02s\n- swift-tox-py35 http://logs.openstack.org/17/636217/4/check/swift-tox-py35/13771e0/ : SUCCESS in 6m 10s\n- swift-tox-py36 http://logs.openstack.org/17/636217/4/check/swift-tox-py36/54b5280/ : SUCCESS in 6m 24s (non-voting)\n- swift-tox-func http://logs.openstack.org/17/636217/4/check/swift-tox-func/c4c48b3/ : SUCCESS in 9m 41s\n- swift-tox-func-encryption http://logs.openstack.org/17/636217/4/check/swift-tox-func-encryption/be9e494/ : SUCCESS in 14m 16s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/636217/4/check/swift-tox-func-domain-remap-staticweb/fbaa317/ : SUCCESS in 9m 10s\n- swift-tox-func-ec http://logs.openstack.org/17/636217/4/check/swift-tox-func-ec/3666b46/ : SUCCESS in 10m 10s\n- swift-tox-func-s3api http://logs.openstack.org/17/636217/4/check/swift-tox-func-s3api/6bfc95c/ : SUCCESS in 15m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/17/636217/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/cc777f5/ : SUCCESS in 14m 37s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/17/636217/4/check/swift-probetests-centos-7/dce860f/ : SUCCESS in 29m 40s\n- swift-dsvm-functional http://logs.openstack.org/17/636217/4/check/swift-dsvm-functional/3559916/ : SUCCESS in 25m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/636217/4/check/swift-dsvm-functional-ipv6/c1cfe34/ : SUCCESS in 25m 45s\n- swift-tox-lower-constraints http://logs.openstack.org/17/636217/4/check/swift-tox-lower-constraints/11b5b33/ : SUCCESS in 10m 54s\n- openstack-tox-pep8 http://logs.openstack.org/17/636217/4/check/openstack-tox-pep8/861b6d3/ : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/636217/4/check/swift-multinode-rolling-upgrade/4649c60/ : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":4},{"id":"1e64e0980a08a68f865f1aa8c96e84b1194d8bf4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-02-27 03:10:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(to be honest, I do not understand this completely, but it looks reasonably harmless.)","accounts_in_message":[],"_revision_number":4},{"id":"3955db19e3acadd2875e47287c15aaf4b87cf6c8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-30 16:55:31.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nJust a couple more questions. I\u0027m coming around to this more and more.","accounts_in_message":[],"_revision_number":4},{"id":"e2d30fb4ebfd2293fbc337829799df9a1ea94a73","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-04-30 19:15:25.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nI\u0027m pretty solidity convinced of the merit of the goal here, but I\u0027m neither happy with nor proud of the kludgy implementation for finding the config.  It\u0027s opaque and brittle - do you have ANY better ideas?\n\nThe only thing I can say for it is that it seems to work in my dev environment and also the gate.","accounts_in_message":[],"_revision_number":4},{"id":"b807149801dc480259b1995d2caa5e03a421c006","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-30 20:13:33.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"260dbf980d138b667ded3064021096d846d1cdee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 20:14:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a5f8bb86433e307f5883d283f49d096e6a007cba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 22:12:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/636217/4/gate/openstack-tox-docs/4749138/html/ : SUCCESS in 5m 03s\n- swift-tox-py27 http://logs.openstack.org/17/636217/4/gate/swift-tox-py27/d55e31c/ : SUCCESS in 12m 01s\n- swift-tox-py37 http://logs.openstack.org/17/636217/4/gate/swift-tox-py37/2d957c5/ : SUCCESS in 9m 43s\n- swift-tox-func http://logs.openstack.org/17/636217/4/gate/swift-tox-func/43b885d/ : SUCCESS in 10m 28s\n- swift-tox-func-encryption http://logs.openstack.org/17/636217/4/gate/swift-tox-func-encryption/15d08ef/ : SUCCESS in 10m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/636217/4/gate/swift-tox-func-domain-remap-staticweb/3078eaf/ : SUCCESS in 9m 31s\n- swift-tox-func-ec http://logs.openstack.org/17/636217/4/gate/swift-tox-func-ec/47c0ec7/ : SUCCESS in 10m 26s\n- swift-tox-func-s3api http://logs.openstack.org/17/636217/4/gate/swift-tox-func-s3api/d24c52d/ : SUCCESS in 11m 06s\n- swift-probetests-centos-7 http://logs.openstack.org/17/636217/4/gate/swift-probetests-centos-7/c1f3eec/ : SUCCESS in 33m 31s\n- swift-dsvm-functional http://logs.openstack.org/17/636217/4/gate/swift-dsvm-functional/07b183e/ : SUCCESS in 27m 23s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/636217/4/gate/swift-dsvm-functional-ipv6/e193108/ : SUCCESS in 27m 02s\n- swift-tox-lower-constraints http://logs.openstack.org/17/636217/4/gate/swift-tox-lower-constraints/8ac21b1/ : SUCCESS in 13m 13s\n- openstack-tox-pep8 http://logs.openstack.org/17/636217/4/gate/openstack-tox-pep8/f6e5b19/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":4},{"id":"9f5dbca9ff8c01875e17d7e56ffca8976881e8fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 22:12:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"771963c92654338011d3dc907044b2ee3f405dd3","revisions":{"6007d128ad61447d496524b2ccab404a0022edfb":{"kind":"REWORK","_number":1,"created":"2019-02-11 20:43:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/17/636217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/636217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/636217/1"}}},"commit":{"parents":[{"commit":"64e5fd364ae044ebfda541fd90bb3551bab36dcb","subject":"Merge \"Stop using duplicate dev IDs in write_fake_ring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e5fd364ae044ebfda541fd90bb3551bab36dcb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-11 20:43:01.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-11 20:43:05.000000000","tz":-360},"subject":"Increase probetest multi-node timeout","message":"Increase probetest multi-node timeout\n\nChange-Id: I334ef517d833916a3b7be3151a812d4f9c66a6e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6007d128ad61447d496524b2ccab404a0022edfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6007d128ad61447d496524b2ccab404a0022edfb"}]},"branch":"refs/heads/master"},"faf3ce4a071d449b3e9ca2dd481fc10c8308620a":{"kind":"REWORK","_number":2,"created":"2019-02-11 21:48:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/17/636217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/636217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/636217/2"}}},"commit":{"parents":[{"commit":"64e5fd364ae044ebfda541fd90bb3551bab36dcb","subject":"Merge \"Stop using duplicate dev IDs in write_fake_ring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e5fd364ae044ebfda541fd90bb3551bab36dcb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-11 20:43:01.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-11 21:48:26.000000000","tz":-360},"subject":"Increase node_timeout in gate","message":"Increase node_timeout in gate\n\nGive storage nodes more time to complete requests for multi-node upgrade\nand probetests.\n\nChange-Id: I334ef517d833916a3b7be3151a812d4f9c66a6e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/faf3ce4a071d449b3e9ca2dd481fc10c8308620a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/faf3ce4a071d449b3e9ca2dd481fc10c8308620a"}]},"branch":"refs/heads/master"},"22cc6a0c5dd2ecb7742b45541f3a98af3a64681d":{"kind":"REWORK","_number":3,"created":"2019-02-12 15:27:59.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/17/636217/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/636217/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/636217/3"}}},"commit":{"parents":[{"commit":"64e5fd364ae044ebfda541fd90bb3551bab36dcb","subject":"Merge \"Stop using duplicate dev IDs in write_fake_ring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e5fd364ae044ebfda541fd90bb3551bab36dcb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-11 20:43:01.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-12 15:27:32.000000000","tz":-360},"subject":"Increase node_timeout in gate","message":"Increase node_timeout in gate\n\nGive storage nodes more time to complete requests for multi-node upgrade\nand probetests.\n\nAlso slightly decouple probetests from default configs.\n\nChange-Id: I334ef517d833916a3b7be3151a812d4f9c66a6e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22cc6a0c5dd2ecb7742b45541f3a98af3a64681d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22cc6a0c5dd2ecb7742b45541f3a98af3a64681d"}]},"branch":"refs/heads/master"},"771963c92654338011d3dc907044b2ee3f405dd3":{"kind":"REWORK","_number":4,"created":"2019-02-12 16:41:41.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/17/636217/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/636217/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/636217/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/636217/4"}}},"commit":{"parents":[{"commit":"64e5fd364ae044ebfda541fd90bb3551bab36dcb","subject":"Merge \"Stop using duplicate dev IDs in write_fake_ring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e5fd364ae044ebfda541fd90bb3551bab36dcb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-11 20:43:01.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-02-12 16:39:17.000000000","tz":-360},"subject":"Increase node_timeout in gate","message":"Increase node_timeout in gate\n\nGive storage nodes more time to complete requests for multi-node upgrade\nand probetests.\n\nAlso slightly decouple probetests from default configs.\n\nChange-Id: I334ef517d833916a3b7be3151a812d4f9c66a6e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771963c92654338011d3dc907044b2ee3f405dd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771963c92654338011d3dc907044b2ee3f405dd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
