)]}'
{"id":"openstack%2Fswift~638015","triplet_id":"openstack%2Fswift~master~Ia44e832e4198e75f3337884c4612f4e9fe3b9e0b","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ia44e832e4198e75f3337884c4612f4e9fe3b9e0b","subject":"versioned_writes: stop trying to catch HTTPPreconditionFailed","status":"MERGED","created":"2019-02-19 21:29:12.000000000","updated":"2019-05-04 18:35:50.000000000","submitted":"2019-05-04 18:35:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"638015-1556994950436-f36ff123","meta_rev_id":"221315ae4b393e8cde4407bda2ca539d8302546d","_number":638015,"virtual_id_number":638015,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-05-04 18:35:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-02-27 03:14:02.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-03 15:09:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-27 03:14:02.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-04 18:35:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ce3e76c89929f673b803ef5909c26fe65205ec1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-19 21:29:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4c8f2e35cd5546e5de111c9e2bc394ea76818b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 22:45:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/check/openstack-tox-docs/ccfb462/html/ : SUCCESS in 5m 00s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/check/swift-tox-py27/b6fdea4/ : SUCCESS in 11m 14s\n- swift-tox-py35 http://logs.openstack.org/15/638015/1/check/swift-tox-py35/421cc14/ : SUCCESS in 6m 43s\n- swift-tox-py36 http://logs.openstack.org/15/638015/1/check/swift-tox-py36/f5092dd/ : SUCCESS in 8m 12s (non-voting)\n- swift-tox-func http://logs.openstack.org/15/638015/1/check/swift-tox-func/46e1c74/ : SUCCESS in 9m 10s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/check/swift-tox-func-encryption/d7a7be6/ : FAILURE in 11m 16s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/check/swift-tox-func-domain-remap-staticweb/9276859/ : SUCCESS in 10m 01s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/check/swift-tox-func-ec/76145f0/ : SUCCESS in 9m 44s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api/7cfb81e/ : SUCCESS in 9m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/bb8b943/ : SUCCESS in 12m 16s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/check/swift-probetests-centos-7/3995f96/ : SUCCESS in 32m 50s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional/1158d66/ : SUCCESS in 27m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional-ipv6/89d8e4d/ : SUCCESS in 30m 53s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/check/swift-tox-lower-constraints/df322fe/ : FAILURE in 15m 38s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/check/openstack-tox-pep8/92306ac/ : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/638015/1/check/swift-multinode-rolling-upgrade/4f35d27/ : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"41c6e585ec9e76bf9a52bb43e6d6e721625254ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-02-19 22:58:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"967d8a8bb1a8008e0cd771444baddcb674ef2657","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 23:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/check/openstack-tox-docs/a7ff027/html/ : SUCCESS in 3m 57s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/check/swift-tox-py27/b9bce37/ : SUCCESS in 10m 45s\n- swift-tox-py35 http://logs.openstack.org/15/638015/1/check/swift-tox-py35/5350c2b/ : SUCCESS in 6m 36s\n- swift-tox-py36 http://logs.openstack.org/15/638015/1/check/swift-tox-py36/178b44f/ : SUCCESS in 5m 48s (non-voting)\n- swift-tox-func http://logs.openstack.org/15/638015/1/check/swift-tox-func/76b4546/ : SUCCESS in 9m 03s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/check/swift-tox-func-encryption/8b694e5/ : SUCCESS in 9m 22s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/check/swift-tox-func-domain-remap-staticweb/0ddb4ee/ : SUCCESS in 9m 05s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/check/swift-tox-func-ec/53c75c7/ : SUCCESS in 14m 28s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api/4eacc8c/ : SUCCESS in 10m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2b44db0/ : SUCCESS in 11m 26s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/check/swift-probetests-centos-7/b94db6d/ : SUCCESS in 31m 51s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional/3b2ff64/ : SUCCESS in 25m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional-ipv6/43f0a9c/ : SUCCESS in 23m 30s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/check/swift-tox-lower-constraints/0385220/ : SUCCESS in 11m 20s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/check/openstack-tox-pep8/8307d8a/ : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/638015/1/check/swift-multinode-rolling-upgrade/b0a6421/ : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9cbd8323c1c9dfc79e77de803d0740720345c216","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-02-27 03:14:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nnice","accounts_in_message":[],"_revision_number":1},{"id":"96f1b248326a825a246de0787658e3c3d59e5a13","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 15:09:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"099daa3bfda6b63bdcb488fa871b0df06888d224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 15:09:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbe147e233c3f78971c3cb6eba35f674c637d9c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 15:46:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/gate/openstack-tox-docs/852bcf0/html/ : SUCCESS in 4m 49s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/gate/swift-tox-py27/c3b4a05/ : FAILURE in 11m 55s\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/gate/swift-tox-py37/b7ba230/ : SUCCESS in 8m 59s\n- swift-tox-func http://logs.openstack.org/15/638015/1/gate/swift-tox-func/6f1fdff/ : SUCCESS in 9m 54s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/gate/swift-tox-func-encryption/aeca31f/ : SUCCESS in 9m 22s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/gate/swift-tox-func-domain-remap-staticweb/c7b3f1e/ : SUCCESS in 8m 28s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/gate/swift-tox-func-ec/cbd4c4f/ : SUCCESS in 8m 45s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/gate/swift-tox-func-s3api/79ddee4/ : SUCCESS in 9m 17s\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/gate/swift-probetests-centos-7/4c079b8/ : SUCCESS in 31m 53s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional/d8838df/ : SUCCESS in 27m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional-ipv6/bfb4add/ : SUCCESS in 25m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/gate/swift-tox-lower-constraints/cc29c60/ : SUCCESS in 11m 51s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/gate/openstack-tox-pep8/504a8f8/ : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"dde1aa413bdce0b0eb959c8a3b463e98f6d8c980","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 17:05:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b2e709835f70abf717ebc5ab8ad415e0ec8f37d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 17:42:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/check/openstack-tox-docs/d8dae6b/html/ : SUCCESS in 4m 04s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/check/swift-tox-py27/261c55a/ : SUCCESS in 10m 04s\n- swift-tox-py35 http://logs.openstack.org/15/638015/1/check/swift-tox-py35/21e8716/ : SUCCESS in 10m 25s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/15/638015/1/check/swift-tox-py36/3daaf49/ : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/check/swift-tox-py37/5e94b71/ : SUCCESS in 10m 58s\n- swift-tox-func http://logs.openstack.org/15/638015/1/check/swift-tox-func/e29bc10/ : SUCCESS in 8m 23s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/check/swift-tox-func-encryption/ada07c6/ : SUCCESS in 10m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/check/swift-tox-func-domain-remap-staticweb/ac450e3/ : SUCCESS in 8m 51s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/check/swift-tox-func-ec/a67a994/ : SUCCESS in 9m 38s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api/a28830c/ : SUCCESS in 12m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0754e82/ : SUCCESS in 11m 48s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/check/swift-probetests-centos-7/4e69465/ : SUCCESS in 30m 43s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional/c4fcc0f/ : SUCCESS in 27m 00s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional-ipv6/175720e/ : SUCCESS in 28m 50s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/check/swift-tox-lower-constraints/efc314b/ : SUCCESS in 23m 42s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/check/openstack-tox-pep8/6b595c1/ : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/638015/1/check/swift-multinode-rolling-upgrade/8466d64/ : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e7265fb4638f20be409ca8255ffc9ba28b6e8094","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 17:42:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3631b436ab9fd6e32f7400a230d9bbab74e66589","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 20:30:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/gate/openstack-tox-docs/4352a17/html/ : SUCCESS in 4m 53s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/gate/swift-tox-py27/ee02806/ : FAILURE in 15m 21s\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/gate/swift-tox-py37/d49bb6c/ : FAILURE in 14m 24s\n- swift-tox-func http://logs.openstack.org/15/638015/1/gate/swift-tox-func/f91e560/ : SUCCESS in 9m 18s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/gate/swift-tox-func-encryption/49ec931/ : SUCCESS in 12m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/gate/swift-tox-func-domain-remap-staticweb/2d831e0/ : SUCCESS in 9m 06s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/gate/swift-tox-func-ec/17f5c8a/ : SUCCESS in 9m 36s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/gate/swift-tox-func-s3api/1ffcd8b/ : SUCCESS in 9m 48s\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/gate/swift-probetests-centos-7/27e791c/ : SUCCESS in 32m 51s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional/0fe9b43/ : SUCCESS in 22m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional-ipv6/373cd30/ : SUCCESS in 29m 14s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/gate/swift-tox-lower-constraints/bc29d22/ : SUCCESS in 12m 33s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/gate/openstack-tox-pep8/0899f3c/ : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"bd2631ce5331df4c1f2c6072730cdce2f86a6b01","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 22:53:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a06c51b38e509f0a73fb9a9eb7a08b5091c16cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:35:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/check/openstack-tox-docs/15e56cb/html/ : SUCCESS in 3m 39s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/check/swift-tox-py27/62ed4ab/ : SUCCESS in 12m 48s\n- swift-tox-py35 http://logs.openstack.org/15/638015/1/check/swift-tox-py35/33c2b7b/ : SUCCESS in 8m 19s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/15/638015/1/check/swift-tox-py36/88ab7b3/ : SUCCESS in 8m 06s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/check/swift-tox-py37/e1abfcb/ : SUCCESS in 8m 37s\n- swift-tox-func http://logs.openstack.org/15/638015/1/check/swift-tox-func/f2b2c19/ : SUCCESS in 34m 24s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/check/swift-tox-func-encryption/580fefa/ : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/check/swift-tox-func-domain-remap-staticweb/ae6dbb0/ : SUCCESS in 8m 26s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/check/swift-tox-func-ec/d44fc45/ : SUCCESS in 8m 28s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api/23b0084/ : SUCCESS in 12m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ce2bd59/ : SUCCESS in 11m 05s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/check/swift-probetests-centos-7/d62651a/ : SUCCESS in 34m 26s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional/41b8320/ : SUCCESS in 30m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional-ipv6/edf2d9a/ : SUCCESS in 32m 52s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/check/swift-tox-lower-constraints/555e435/ : SUCCESS in 20m 47s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/check/openstack-tox-pep8/1fba494/ : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/638015/1/check/swift-multinode-rolling-upgrade/9965b09/ : SUCCESS in 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"242cb349804fb505e8dc30755df2437facc8be03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:35:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e25282d65f929866e4b2a89c005d125449ae91c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 02:40:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/gate/openstack-tox-docs/f7543d5/html/ : SUCCESS in 4m 22s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/gate/swift-tox-py27/874d0b7/ : SUCCESS in 10m 47s\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/gate/swift-tox-py37/0e5a1bb/ : SUCCESS in 11m 25s\n- swift-tox-func http://logs.openstack.org/15/638015/1/gate/swift-tox-func/405875b/ : FAILURE in 3m 46s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/gate/swift-tox-func-encryption/5ab4b47/ : FAILURE in 3m 52s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/gate/swift-tox-func-domain-remap-staticweb/de66a9a/ : FAILURE in 3m 54s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/gate/swift-tox-func-ec/cc4891c/ : FAILURE in 5m 33s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/gate/swift-tox-func-s3api/8991d62/ : FAILURE in 3m 41s\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/gate/swift-probetests-centos-7/30d83bf/ : SUCCESS in 35m 20s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional/739852f/ : FAILURE in 24m 32s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional-ipv6/401b503/ : FAILURE in 18m 10s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/gate/swift-tox-lower-constraints/1ab4671/ : SUCCESS in 19m 27s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/gate/openstack-tox-pep8/d98a7ba/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6ddc64c054442e687ce1422f7001b32c8862abd3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 04:36:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95e2bd7087ce04482d9c393233cce3545ce60ccc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 05:28:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/check/openstack-tox-docs/c7a1565/html/ : SUCCESS in 3m 56s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/check/swift-tox-py27/3b7093b/ : FAILURE in 12m 30s\n- swift-tox-py35 http://logs.openstack.org/15/638015/1/check/swift-tox-py35/92be12e/ : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/15/638015/1/check/swift-tox-py36/743330e/ : SUCCESS in 8m 41s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/check/swift-tox-py37/6984e6d/ : SUCCESS in 9m 30s\n- swift-tox-func http://logs.openstack.org/15/638015/1/check/swift-tox-func/bd32925/ : FAILURE in 3m 24s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/check/swift-tox-func-encryption/a8b76d7/ : FAILURE in 3m 24s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/check/swift-tox-func-domain-remap-staticweb/b38126e/ : FAILURE in 3m 33s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/check/swift-tox-func-ec/40e3c96/ : FAILURE in 3m 34s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api/9ded62e/ : FAILURE in 3m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/eef5fbe/ : SUCCESS in 13m 53s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/check/swift-probetests-centos-7/823b557/ : FAILURE in 29m 17s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional/2f9636d/ : FAILURE in 17m 51s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional-ipv6/1a74438/ : FAILURE in 14m 00s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/check/swift-tox-lower-constraints/7da6a1f/ : SUCCESS in 13m 56s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/check/openstack-tox-pep8/de2e1db/ : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/638015/1/check/swift-multinode-rolling-upgrade/513f478/ : FAILURE in 10m 07s","accounts_in_message":[],"_revision_number":1},{"id":"57f6a6154c160a925abd48e1b712028285d1b5b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 15:53:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3c95e78a984073c308c9af04d6629d76af9b5c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:28:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/check/openstack-tox-docs/751edaa/html/ : SUCCESS in 4m 55s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/check/swift-tox-py27/06d51e4/ : SUCCESS in 13m 06s\n- swift-tox-py35 http://logs.openstack.org/15/638015/1/check/swift-tox-py35/c3a0ee4/ : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/15/638015/1/check/swift-tox-py36/32e9820/ : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/check/swift-tox-py37/a7e19b4/ : SUCCESS in 8m 17s\n- swift-tox-func http://logs.openstack.org/15/638015/1/check/swift-tox-func/5b4bdf0/ : SUCCESS in 10m 02s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/check/swift-tox-func-encryption/ca76ecf/ : SUCCESS in 10m 47s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/check/swift-tox-func-domain-remap-staticweb/3564b9c/ : SUCCESS in 8m 49s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/check/swift-tox-func-ec/16fa9a1/ : SUCCESS in 9m 24s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api/2ef79dd/ : SUCCESS in 14m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/638015/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/22d6d15/ : SUCCESS in 11m 41s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/check/swift-probetests-centos-7/e1084bd/ : SUCCESS in 29m 34s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional/82ffe8b/ : SUCCESS in 28m 29s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/check/swift-dsvm-functional-ipv6/80fdbf5/ : SUCCESS in 29m 34s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/check/swift-tox-lower-constraints/77241c1/ : SUCCESS in 11m 45s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/check/openstack-tox-pep8/cbbd9ba/ : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/638015/1/check/swift-multinode-rolling-upgrade/283cb82/ : FAILURE in 7m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72215e4aa209e2dccc29b7ff43ebe796a6f74c71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:29:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"824eb15fa9d9d109d38038680492e50b5c729294","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:35:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"221315ae4b393e8cde4407bda2ca539d8302546d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:35:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/638015/1/gate/openstack-tox-docs/c1c1baf/html/ : SUCCESS in 5m 10s\n- swift-tox-py27 http://logs.openstack.org/15/638015/1/gate/swift-tox-py27/224b450/ : SUCCESS in 11m 40s\n- swift-tox-py37 http://logs.openstack.org/15/638015/1/gate/swift-tox-py37/4591718/ : SUCCESS in 9m 00s\n- swift-tox-func http://logs.openstack.org/15/638015/1/gate/swift-tox-func/987ecc2/ : SUCCESS in 10m 16s\n- swift-tox-func-encryption http://logs.openstack.org/15/638015/1/gate/swift-tox-func-encryption/e472871/ : SUCCESS in 10m 29s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/638015/1/gate/swift-tox-func-domain-remap-staticweb/5a9e4e3/ : SUCCESS in 10m 39s\n- swift-tox-func-ec http://logs.openstack.org/15/638015/1/gate/swift-tox-func-ec/c3d8d5d/ : SUCCESS in 10m 13s\n- swift-tox-func-s3api http://logs.openstack.org/15/638015/1/gate/swift-tox-func-s3api/d1f7000/ : SUCCESS in 11m 01s\n- swift-probetests-centos-7 http://logs.openstack.org/15/638015/1/gate/swift-probetests-centos-7/f14e65f/ : SUCCESS in 31m 18s\n- swift-dsvm-functional http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional/7b4bf1f/ : SUCCESS in 26m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/638015/1/gate/swift-dsvm-functional-ipv6/57305a4/ : SUCCESS in 26m 40s\n- swift-tox-lower-constraints http://logs.openstack.org/15/638015/1/gate/swift-tox-lower-constraints/1d68a50/ : SUCCESS in 12m 30s\n- openstack-tox-pep8 http://logs.openstack.org/15/638015/1/gate/openstack-tox-pep8/30b43a9/ : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e045d0de213457e97a943d0e6f75fe456c377e1","revisions":{"8e045d0de213457e97a943d0e6f75fe456c377e1":{"kind":"REWORK","_number":1,"created":"2019-02-19 21:29:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/638015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/638015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/638015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/638015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/638015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/638015/1"}}},"commit":{"parents":[{"commit":"89b3adc9fbfc29fc1cac58fef39632aadc30eba4","subject":"Merge \"probetests: make negative assertion more meaningful\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89b3adc9fbfc29fc1cac58fef39632aadc30eba4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-02-19 21:14:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-02-19 21:28:41.000000000","tz":-480},"subject":"versioned_writes: stop trying to catch HTTPPreconditionFailed","message":"versioned_writes: stop trying to catch HTTPPreconditionFailed\n\n* We just caught it to raise another one\n* On newer versions of python, we wouldn\u0027t actually catch\n  anything, because HTTPPreconditionFailed is not an instance\n  of BaseException -- it\u0027s a partial function application\n  which when called will return an exception.\n\nChange-Id: Ia44e832e4198e75f3337884c4612f4e9fe3b9e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e045d0de213457e97a943d0e6f75fe456c377e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e045d0de213457e97a943d0e6f75fe456c377e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
