)]}'
{"id":"openstack%2Fswift~651520","triplet_id":"openstack%2Fswift~feature%2Flosf~Ie6e4c6ae82a485a0ca74505b6cc87a39ba0c8603","project":"openstack/swift","branch":"feature/losf","topic":"651102","hashtags":[],"change_id":"Ie6e4c6ae82a485a0ca74505b6cc87a39ba0c8603","subject":"Change the Makefile to work on vagrant swift all in one","status":"ABANDONED","created":"2019-04-10 12:45:15.000000000","updated":"2019-05-17 13:46:30.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7827c5e3393b5a944182c544f3dd312b4356b586","_number":651520,"virtual_id_number":651520,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-02 14:48:45.000000000","_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":-1,"date":"2019-05-02 01:41:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-02 01:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-02 14:48:45.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"}],"messages":[{"id":"768ba94574f7a0416887c78bff31874d503f994d","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-04-10 12:45:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0055a2ab542b05a8dca71c5980c6bbc474701822","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-04-10 12:46:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ba493a0b097c394c48b8d5034715d095dc0f51b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 13:42:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/20/651520/1/check/openstack-tox-docs/1b3c8ae/ : POST_FAILURE in 9m 20s\n- swift-tox-py27 http://logs.openstack.org/20/651520/1/check/swift-tox-py27/990fcf9/ : FAILURE in 7m 06s\n- swift-tox-py35 http://logs.openstack.org/20/651520/1/check/swift-tox-py35/6caa8da/ : FAILURE in 9m 23s\n- swift-tox-py36 http://logs.openstack.org/20/651520/1/check/swift-tox-py36/46813da/ : FAILURE in 4m 40s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/20/651520/1/check/swift-tox-py37/50a31cb/ : FAILURE in 4m 02s (non-voting)\n- swift-tox-func http://logs.openstack.org/20/651520/1/check/swift-tox-func/9c4f1e5/ : FAILURE in 3m 26s\n- swift-tox-func-encryption http://logs.openstack.org/20/651520/1/check/swift-tox-func-encryption/2201aca/ : FAILURE in 9m 38s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/20/651520/1/check/swift-tox-func-domain-remap-staticweb/1cbe92f/ : FAILURE in 10m 16s\n- swift-tox-func-ec http://logs.openstack.org/20/651520/1/check/swift-tox-func-ec/d6b3f84/ : FAILURE in 4m 20s\n- swift-tox-func-s3api http://logs.openstack.org/20/651520/1/check/swift-tox-func-s3api/1c04495/ : FAILURE in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/20/651520/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b7eb488/ : RETRY_LIMIT in 6m 43s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/20/651520/1/check/swift-probetests-centos-7/6747ad2/ : RETRY_LIMIT in 7m 13s\n- swift-dsvm-functional http://logs.openstack.org/20/651520/1/check/swift-dsvm-functional/cded7f7/ : RETRY_LIMIT in 13m 54s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/20/651520/1/check/swift-dsvm-functional-ipv6/7d30d87/ : RETRY_LIMIT in 10m 55s\n- swift-tox-lower-constraints http://logs.openstack.org/20/651520/1/check/swift-tox-lower-constraints/03a4816/ : FAILURE in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/20/651520/1/check/openstack-tox-pep8/9f748e8/ : FAILURE in 4m 42s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/20/651520/1/check/swift-multinode-rolling-upgrade/d2c6b15/ : SUCCESS in 18m 30s","accounts_in_message":[],"_revision_number":1},{"id":"eeca3559382755079c7258dd441e67796160bb4e","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-04-11 08:05:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a04bac89f006bbd74246cf113bb2715096ab34a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 08:45:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/20/651520/2/check/openstack-tox-docs/12c8e88/html/ : SUCCESS in 5m 34s\n- swift-tox-py27 http://logs.openstack.org/20/651520/2/check/swift-tox-py27/7401f78/ : SUCCESS in 15m 21s\n- swift-tox-py35 http://logs.openstack.org/20/651520/2/check/swift-tox-py35/51cbe17/ : SUCCESS in 12m 12s\n- swift-tox-py36 http://logs.openstack.org/20/651520/2/check/swift-tox-py36/1d83a68/ : SUCCESS in 9m 33s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/20/651520/2/check/swift-tox-py37/94551ad/ : SUCCESS in 11m 24s (non-voting)\n- swift-tox-func http://logs.openstack.org/20/651520/2/check/swift-tox-func/b5879c6/ : SUCCESS in 10m 50s\n- swift-tox-func-encryption http://logs.openstack.org/20/651520/2/check/swift-tox-func-encryption/7186c7f/ : SUCCESS in 14m 35s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/20/651520/2/check/swift-tox-func-domain-remap-staticweb/2c02f10/ : SUCCESS in 10m 21s\n- swift-tox-func-ec http://logs.openstack.org/20/651520/2/check/swift-tox-func-ec/1fdfc9d/ : SUCCESS in 11m 59s\n- swift-tox-func-s3api http://logs.openstack.org/20/651520/2/check/swift-tox-func-s3api/ee112ff/ : SUCCESS in 14m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/20/651520/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1b561a8/ : SUCCESS in 14m 00s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/20/651520/2/check/swift-probetests-centos-7/c1c53c2/ : FAILURE in 36m 06s\n- swift-dsvm-functional http://logs.openstack.org/20/651520/2/check/swift-dsvm-functional/6abab4f/ : SUCCESS in 24m 13s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/20/651520/2/check/swift-dsvm-functional-ipv6/9f2b49a/ : SUCCESS in 29m 59s\n- swift-tox-lower-constraints http://logs.openstack.org/20/651520/2/check/swift-tox-lower-constraints/38ff9c2/ : FAILURE in 18m 12s\n- openstack-tox-pep8 http://logs.openstack.org/20/651520/2/check/openstack-tox-pep8/87777ad/ : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/20/651520/2/check/swift-multinode-rolling-upgrade/8f420be/ : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":2},{"id":"689f378c2dbde79b1f9ce14851837260390e7ce5","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-02 00:25:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bf648e017c64d0774671c6ade0ffdda434fdcca","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-02 00:27:29.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nUpdate the parent base because this might be what I really want to run the losf in the gate. Let me check how the installation work with this. Then possibly I squash this patch to the parent adding you to co-author in the patch. Now the parent is https://review.opendev.org/#/c/656576/6","accounts_in_message":[],"_revision_number":3},{"id":"01cc3464a049ee501bacae08fd5935a8bbc95f41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 01:41:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/20/651520/3/check/openstack-tox-docs/96a01cb/html/ : SUCCESS in 3m 59s\n- swift-tox-py27 http://logs.openstack.org/20/651520/3/check/swift-tox-py27/204bbe2/ : SUCCESS in 9m 49s\n- swift-tox-py35 http://logs.openstack.org/20/651520/3/check/swift-tox-py35/dc4c507/ : SUCCESS in 8m 22s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/20/651520/3/check/swift-tox-py36/76d4f44/ : SUCCESS in 8m 45s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/20/651520/3/check/swift-tox-py37/1467b12/ : SUCCESS in 7m 39s\n- swift-tox-func http://logs.openstack.org/20/651520/3/check/swift-tox-func/a5b31a3/ : SUCCESS in 12m 42s\n- swift-tox-func-encryption http://logs.openstack.org/20/651520/3/check/swift-tox-func-encryption/2826fe9/ : SUCCESS in 9m 12s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/20/651520/3/check/swift-tox-func-domain-remap-staticweb/24d4e8c/ : SUCCESS in 7m 58s\n- swift-tox-func-ec http://logs.openstack.org/20/651520/3/check/swift-tox-func-ec/a655785/ : SUCCESS in 28m 45s\n- swift-tox-func-s3api http://logs.openstack.org/20/651520/3/check/swift-tox-func-s3api/82540ad/ : SUCCESS in 34m 29s\n- swift-tox-func-losf http://logs.openstack.org/20/651520/3/check/swift-tox-func-losf/13055ef/ : RETRY_LIMIT in 22m 46s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/20/651520/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fff6193/ : SUCCESS in 11m 59s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/20/651520/3/check/swift-probetests-centos-7/4f3f20d/ : SUCCESS in 38m 01s\n- swift-dsvm-functional http://logs.openstack.org/20/651520/3/check/swift-dsvm-functional/383cc1a/ : SUCCESS in 25m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/20/651520/3/check/swift-dsvm-functional-ipv6/2afe7ed/ : SUCCESS in 26m 10s\n- swift-tox-lower-constraints http://logs.openstack.org/20/651520/3/check/swift-tox-lower-constraints/11f29c7/ : FAILURE in 14m 31s\n- openstack-tox-pep8 http://logs.openstack.org/20/651520/3/check/openstack-tox-pep8/0f1e362/ : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/20/651520/3/check/swift-multinode-rolling-upgrade/f28b9f5/ : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1aa1ec767849172bf42fd2c2e5034e4d91f86b2c","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-02 14:48:45.000000000","message":"Patch Set 3:\n\nFYI, i squashed this patch to https://review.opendev.org/#/c/656576/. Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"dd874bab71935cfef277920fea5b590e1576cc3f","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-17 13:46:30.000000000","message":"Abandoned\n\nSquashed in 656576","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b9af404cd514e7e18a030f8ae125f7a3cf65045","revisions":{"59665483a52d9caac8f9eaedfd1c26ed885d0d65":{"kind":"REWORK","_number":1,"created":"2019-04-10 12:45:15.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/20/651520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/651520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/651520/1"}}},"commit":{"parents":[{"commit":"c7a68899cbc756d5311ab5828db6f12c5a03c7de","subject":"WIP: add swift-rpc-losf into setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7a68899cbc756d5311ab5828db6f12c5a03c7de"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-04-10 12:39:57.000000000","tz":120},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-04-10 12:45:00.000000000","tz":120},"subject":"Change the Makefile to work on vagrant swift all in one","message":"Change the Makefile to work on vagrant swift all in one\n\nWith the $PATH properly set (needs $HOME/go/bin, and the path to cmake,\nif needed), it works and compiles swift-rpc-losf.\nBut I removed the build/install distinction which we may want to keep?\n\nAlso, -Wl,-rpath does not appear to work yet, it would be nice for the\nbinary not to require setting LD_LIBRARY_PATH to find leveldb.so\nIt seems that the \"go get levigo..\" line compiles it but does not\ninstall the binary, and the \"go get\" line eventually installs the\nbinary, but without the proper LDFLAGS. I don\u0027t know how to specify them\nwith the new go module system, I\u0027ll try again tomorrow\n\nChange-Id: Ie6e4c6ae82a485a0ca74505b6cc87a39ba0c8603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59665483a52d9caac8f9eaedfd1c26ed885d0d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59665483a52d9caac8f9eaedfd1c26ed885d0d65"}]},"branch":"refs/heads/feature/losf"},"45ebe72943b056b9c022dafae97fe1f81c45c1ce":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-11 08:05:40.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/20/651520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/651520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/651520/2"}}},"commit":{"parents":[{"commit":"6f17c84923c38602214613a0852769a1d4baea87","subject":"Fix Makefile to be able to re-run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f17c84923c38602214613a0852769a1d4baea87"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-04-10 12:39:57.000000000","tz":120},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-04-11 08:05:11.000000000","tz":540},"subject":"Change the Makefile to work on vagrant swift all in one","message":"Change the Makefile to work on vagrant swift all in one\n\nWith the $PATH properly set (needs $HOME/go/bin, and the path to cmake,\nif needed), it works and compiles swift-rpc-losf.\nBut I removed the build/install distinction which we may want to keep?\n\nAlso, -Wl,-rpath does not appear to work yet, it would be nice for the\nbinary not to require setting LD_LIBRARY_PATH to find leveldb.so\nIt seems that the \"go get levigo..\" line compiles it but does not\ninstall the binary, and the \"go get\" line eventually installs the\nbinary, but without the proper LDFLAGS. I don\u0027t know how to specify them\nwith the new go module system, I\u0027ll try again tomorrow\n\nChange-Id: Ie6e4c6ae82a485a0ca74505b6cc87a39ba0c8603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45ebe72943b056b9c022dafae97fe1f81c45c1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45ebe72943b056b9c022dafae97fe1f81c45c1ce"}]},"branch":"refs/heads/feature/losf"},"2b9af404cd514e7e18a030f8ae125f7a3cf65045":{"kind":"REWORK","_number":3,"created":"2019-05-02 00:25:36.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/20/651520/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/651520/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/651520/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/651520/3"}}},"commit":{"parents":[{"commit":"c17979745f0a2823f30b0831aae158dacd88d60b","subject":"WIP: try install losf packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c17979745f0a2823f30b0831aae158dacd88d60b"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-04-10 12:39:57.000000000","tz":120},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-05-02 00:24:51.000000000","tz":540},"subject":"Change the Makefile to work on vagrant swift all in one","message":"Change the Makefile to work on vagrant swift all in one\n\nWith the $PATH properly set (needs $HOME/go/bin, and the path to cmake,\nif needed), it works and compiles swift-rpc-losf.\nBut I removed the build/install distinction which we may want to keep?\n\nAlso, -Wl,-rpath does not appear to work yet, it would be nice for the\nbinary not to require setting LD_LIBRARY_PATH to find leveldb.so\nIt seems that the \"go get levigo..\" line compiles it but does not\ninstall the binary, and the \"go get\" line eventually installs the\nbinary, but without the proper LDFLAGS. I don\u0027t know how to specify them\nwith the new go module system, I\u0027ll try again tomorrow\n\nChange-Id: Ie6e4c6ae82a485a0ca74505b6cc87a39ba0c8603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b9af404cd514e7e18a030f8ae125f7a3cf65045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b9af404cd514e7e18a030f8ae125f7a3cf65045"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
