)]}'
{"id":"openstack%2Fswift~651893","triplet_id":"openstack%2Fswift~master~Ifc35aacb2e45004f74c871f08ff3c52bc57c1463","project":"openstack/swift","branch":"master","topic":"651893","hashtags":[],"change_id":"Ifc35aacb2e45004f74c871f08ff3c52bc57c1463","subject":"py3: port s3api","status":"MERGED","created":"2019-04-11 17:49:58.000000000","updated":"2019-05-04 22:49:38.000000000","submitted":"2019-05-04 22:49:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651893-1557010178764-7175b152","meta_rev_id":"566f973d737f8cd8430196c37335b5b1bbcfcba4","_number":651893,"virtual_id_number":651893,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-05-04 22:49:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-03 06:20:14.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-05-04 19:59:27.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-05-04 19:59:27.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 19:59:27.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-04 22:49:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f233ad205af68e102c49260f30dc3011978d02d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-11 17:49:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4c3ee8db598cbc35b8140292f83ccd2101e387b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 18:35:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/1/check/openstack-tox-docs/3edbf37/html/ : SUCCESS in 5m 46s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/1/check/build-openstack-releasenotes/6d97d42/html/ : SUCCESS in 8m 24s\n- swift-tox-py27 http://logs.openstack.org/93/651893/1/check/swift-tox-py27/8f5ba4e/ : SUCCESS in 12m 21s\n- swift-tox-py35 http://logs.openstack.org/93/651893/1/check/swift-tox-py35/f577049/ : FAILURE in 9m 56s\n- swift-tox-py36 http://logs.openstack.org/93/651893/1/check/swift-tox-py36/8a2840d/ : FAILURE in 9m 34s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/1/check/swift-tox-py37/0e5a1c5/ : FAILURE in 13m 04s (non-voting)\n- swift-tox-func http://logs.openstack.org/93/651893/1/check/swift-tox-func/2e136dc/ : SUCCESS in 11m 12s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/1/check/swift-tox-func-encryption/b673036/ : SUCCESS in 10m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/1/check/swift-tox-func-domain-remap-staticweb/2b06c11/ : SUCCESS in 10m 40s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/1/check/swift-tox-func-ec/82149d6/ : SUCCESS in 8m 49s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/1/check/swift-tox-func-s3api/68d5e73/ : SUCCESS in 10m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b3c2779/ : SUCCESS in 13m 50s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/1/check/swift-probetests-centos-7/6e412cb/ : SUCCESS in 32m 51s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/1/check/swift-dsvm-functional/eee765f/ : SUCCESS in 27m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/1/check/swift-dsvm-functional-ipv6/3ed3ed0/ : SUCCESS in 28m 24s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/1/check/swift-tox-lower-constraints/bd40db7/ : SUCCESS in 16m 36s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/1/check/openstack-tox-pep8/99613f5/ : FAILURE in 6m 42s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/1/check/swift-multinode-rolling-upgrade/6969640/ : SUCCESS in 18m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5a3b839f8c99a6e7cfdd0db4055b22bde4047d62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-11 18:56:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8251cb0aa71b204273d3bed79da00df0381d08d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 19:30:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/2/check/openstack-tox-docs/3da3cb0/html/ : SUCCESS in 7m 44s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/2/check/build-openstack-releasenotes/12b619e/html/ : SUCCESS in 9m 13s\n- swift-tox-py27 http://logs.openstack.org/93/651893/2/check/swift-tox-py27/6065feb/ : SUCCESS in 11m 09s\n- swift-tox-py35 http://logs.openstack.org/93/651893/2/check/swift-tox-py35/af0ce9a/ : SUCCESS in 11m 54s\n- swift-tox-py36 http://logs.openstack.org/93/651893/2/check/swift-tox-py36/0e8f08c/ : SUCCESS in 10m 55s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/2/check/swift-tox-py37/afbab0d/ : SUCCESS in 10m 37s (non-voting)\n- swift-tox-func http://logs.openstack.org/93/651893/2/check/swift-tox-func/bcb9e5f/ : SUCCESS in 7m 59s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/2/check/swift-tox-func-encryption/c74602a/ : SUCCESS in 15m 09s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/2/check/swift-tox-func-domain-remap-staticweb/a05640b/ : SUCCESS in 12m 49s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/2/check/swift-tox-func-ec/abb975c/ : SUCCESS in 14m 34s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/2/check/swift-tox-func-s3api/25aa00c/ : SUCCESS in 14m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0099923/ : SUCCESS in 16m 04s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/2/check/swift-probetests-centos-7/9dae4fe/ : SUCCESS in 29m 59s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/2/check/swift-dsvm-functional/dd88990/ : SUCCESS in 25m 30s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/2/check/swift-dsvm-functional-ipv6/604db99/ : SUCCESS in 29m 19s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/2/check/swift-tox-lower-constraints/4f90a8b/ : SUCCESS in 17m 51s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/2/check/openstack-tox-pep8/70e60fd/ : FAILURE in 5m 34s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/2/check/swift-multinode-rolling-upgrade/294714f/ : SUCCESS in 16m 24s","accounts_in_message":[],"_revision_number":2},{"id":"09081ed113c158093562dff33b75171b1f0a8063","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-11 19:32:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da05e729941cdd97e910a867880e2194b052e1e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 23:09:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/3/check/openstack-tox-docs/15a89ff/html/ : SUCCESS in 6m 34s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/3/check/build-openstack-releasenotes/1e2d096/html/ : SUCCESS in 8m 15s\n- swift-tox-py27 http://logs.openstack.org/93/651893/3/check/swift-tox-py27/3c3ff4b/ : SUCCESS in 11m 01s\n- swift-tox-py35 http://logs.openstack.org/93/651893/3/check/swift-tox-py35/12b58e8/ : SUCCESS in 10m 55s\n- swift-tox-py36 http://logs.openstack.org/93/651893/3/check/swift-tox-py36/531de62/ : SUCCESS in 10m 33s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/3/check/swift-tox-py37/89f000e/ : SUCCESS in 10m 23s (non-voting)\n- swift-tox-func http://logs.openstack.org/93/651893/3/check/swift-tox-func/b07a8ae/ : SUCCESS in 8m 42s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/3/check/swift-tox-func-encryption/4834448/ : SUCCESS in 9m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/3/check/swift-tox-func-domain-remap-staticweb/998dc85/ : SUCCESS in 9m 54s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/3/check/swift-tox-func-ec/648cb9b/ : SUCCESS in 10m 51s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/3/check/swift-tox-func-s3api/fe995b7/ : SUCCESS in 16m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b342366/ : SUCCESS in 11m 57s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/3/check/swift-probetests-centos-7/c556f27/ : SUCCESS in 30m 55s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/3/check/swift-dsvm-functional/63ca8b6/ : SUCCESS in 28m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/3/check/swift-dsvm-functional-ipv6/98aa51d/ : SUCCESS in 32m 46s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/3/check/swift-tox-lower-constraints/83bf438/ : SUCCESS in 13m 27s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/3/check/openstack-tox-pep8/5ee6fdb/ : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/3/check/swift-multinode-rolling-upgrade/3502fa4/ : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":3},{"id":"a19bdedcb622c8aaf38222118a334e425dbe6a37","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-04-23 04:25:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\n(see inline)","accounts_in_message":[],"_revision_number":3},{"id":"fef8d8da39a45ae01a4466948e4a2e5fbfc140ed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-24 17:12:38.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"385c3aeb355e491578b53d93eb409608658001cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-24 17:12:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3e8393a1449741ff571d073484856118faa7054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 03:44:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/4/check/openstack-tox-docs/38a24b7/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/4/check/build-openstack-releasenotes/c8dd5fc/html/ : SUCCESS in 7m 49s\n- swift-tox-py27 http://logs.openstack.org/93/651893/4/check/swift-tox-py27/8d6decb/ : SUCCESS in 12m 34s\n- swift-tox-py35 http://logs.openstack.org/93/651893/4/check/swift-tox-py35/80d5edc/ : SUCCESS in 10m 41s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/4/check/swift-tox-py36/f9620d2/ : SUCCESS in 8m 18s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/4/check/swift-tox-py37/d40605f/ : SUCCESS in 10m 14s\n- swift-tox-func http://logs.openstack.org/93/651893/4/check/swift-tox-func/77347bc/ : SUCCESS in 8m 46s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/4/check/swift-tox-func-encryption/f860b06/ : SUCCESS in 8m 20s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/4/check/swift-tox-func-domain-remap-staticweb/87ed51f/ : SUCCESS in 9m 24s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/4/check/swift-tox-func-ec/68c241f/ : SUCCESS in 10m 48s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api/452a175/ : SUCCESS in 11m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e67a6d9/ : SUCCESS in 13m 32s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/4/check/swift-probetests-centos-7/29a6ff3/ : FAILURE in 36m 52s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional/5272440/ : SUCCESS in 30m 33s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional-ipv6/67eb09d/ : SUCCESS in 29m 26s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/4/check/swift-tox-lower-constraints/c9e9bb5/ : SUCCESS in 11m 01s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/4/check/openstack-tox-pep8/9e8717a/ : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/4/check/swift-multinode-rolling-upgrade/b396f07/ : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":4},{"id":"bb7292b44844b99277fb09d4bb7fa33b581f1b7c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-25 19:09:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"043176864cb655466003d31a074a9ce99c3f21ba","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-04-25 20:51:12.000000000","message":"Patch Set 4: Code-Review+2\n\nStill not a big fan of the copy-pasting those absurd expressions, but if double-ports are still a problem, then okay.","accounts_in_message":[],"_revision_number":4},{"id":"6d9b645cc1b39a28c827982fee32ac35fcc3b2c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:15:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/4/check/openstack-tox-docs/a7336c4/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/4/check/build-openstack-releasenotes/5d5b4c9/html/ : SUCCESS in 6m 19s\n- swift-tox-py27 http://logs.openstack.org/93/651893/4/check/swift-tox-py27/5d7573d/ : SUCCESS in 12m 05s\n- swift-tox-py35 http://logs.openstack.org/93/651893/4/check/swift-tox-py35/2cb55bf/ : SUCCESS in 8m 11s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/4/check/swift-tox-py36/a4419c8/ : SUCCESS in 12m 03s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/4/check/swift-tox-py37/dd90a19/ : SUCCESS in 9m 02s\n- swift-tox-func http://logs.openstack.org/93/651893/4/check/swift-tox-func/6d0e561/ : SUCCESS in 9m 03s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/4/check/swift-tox-func-encryption/bb89b33/ : SUCCESS in 8m 16s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/4/check/swift-tox-func-domain-remap-staticweb/f39251e/ : SUCCESS in 7m 48s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/4/check/swift-tox-func-ec/7d574ee/ : SUCCESS in 8m 38s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api/b77ab3f/ : SUCCESS in 9m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/042479d/ : SUCCESS in 14m 36s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/4/check/swift-probetests-centos-7/d0c4357/ : SUCCESS in 32m 39s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional/94fceb7/ : SUCCESS in 24m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional-ipv6/9b71300/ : SUCCESS in 27m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/4/check/swift-tox-lower-constraints/3032c22/ : FAILURE in 34m 40s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/4/check/openstack-tox-pep8/929ce0e/ : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/4/check/swift-multinode-rolling-upgrade/e9a9daf/ : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":4},{"id":"e4ca00acd9a8d01942989e51ee2feef43b69bca1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-29 04:08:24.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"36403946320b83e194c5d468a5be3a8f2db074c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 04:46:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/4/check/openstack-tox-docs/7bfd2dc/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/4/check/build-openstack-releasenotes/9419e31/html/ : SUCCESS in 7m 28s\n- swift-tox-py27 http://logs.openstack.org/93/651893/4/check/swift-tox-py27/0805c1a/ : SUCCESS in 9m 37s\n- swift-tox-py35 http://logs.openstack.org/93/651893/4/check/swift-tox-py35/a0159ce/ : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/4/check/swift-tox-py36/13546c5/ : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/4/check/swift-tox-py37/af54082/ : SUCCESS in 8m 18s\n- swift-tox-func http://logs.openstack.org/93/651893/4/check/swift-tox-func/019fb0e/ : SUCCESS in 7m 56s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/4/check/swift-tox-func-encryption/18b2069/ : SUCCESS in 7m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/4/check/swift-tox-func-domain-remap-staticweb/53bbc1d/ : SUCCESS in 7m 31s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/4/check/swift-tox-func-ec/0350a00/ : SUCCESS in 13m 26s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api/6fc6e57/ : SUCCESS in 9m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f0bc19f/ : SUCCESS in 10m 24s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/4/check/swift-probetests-centos-7/22d030c/ : FAILURE in 29m 46s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional/dfdf142/ : SUCCESS in 26m 47s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional-ipv6/10d98b9/ : SUCCESS in 27m 31s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/4/check/swift-tox-lower-constraints/c762d7e/ : SUCCESS in 10m 57s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/4/check/openstack-tox-pep8/23211cf/ : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/4/check/swift-multinode-rolling-upgrade/b037fe8/ : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":4},{"id":"5c73a586eedc43580eb85b7d371e7c2edc69d2e2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-29 16:31:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f7247977b83d940ed73e11181b40e9321ab1ec49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 17:11:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/4/check/openstack-tox-docs/30c6eeb/html/ : SUCCESS in 3m 39s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/4/check/build-openstack-releasenotes/f869198/html/ : SUCCESS in 6m 12s\n- swift-tox-py27 http://logs.openstack.org/93/651893/4/check/swift-tox-py27/cfe48f8/ : FAILURE in 36m 15s\n- swift-tox-py35 http://logs.openstack.org/93/651893/4/check/swift-tox-py35/917193e/ : SUCCESS in 15m 35s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/4/check/swift-tox-py36/1e9f6a1/ : FAILURE in 26m 13s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/4/check/swift-tox-py37/9003051/ : SUCCESS in 15m 10s\n- swift-tox-func http://logs.openstack.org/93/651893/4/check/swift-tox-func/a1f4245/ : SUCCESS in 9m 28s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/4/check/swift-tox-func-encryption/1fea323/ : SUCCESS in 11m 48s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/4/check/swift-tox-func-domain-remap-staticweb/ad37aad/ : SUCCESS in 10m 44s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/4/check/swift-tox-func-ec/924cafe/ : SUCCESS in 11m 08s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api/7e8b701/ : SUCCESS in 14m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7b577b4/ : SUCCESS in 15m 55s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/4/check/swift-probetests-centos-7/8017fc5/ : FAILURE in 35m 42s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional/7063712/ : SUCCESS in 25m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional-ipv6/c37bea7/ : SUCCESS in 32m 22s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/4/check/swift-tox-lower-constraints/6aec0c7/ : SUCCESS in 11m 35s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/4/check/openstack-tox-pep8/d41b2ea/ : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/4/check/swift-multinode-rolling-upgrade/72b617e/ : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":4},{"id":"3a594fe44a509da1f555549be10f4b941bbf110c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-29 18:09:29.000000000","message":"Patch Set 4:\n\nrecheck\n\nI\u0027m getting soooo tired of that log line assertion.... and not even https://review.opendev.org/#/c/652791/ helped :-(","accounts_in_message":[],"_revision_number":4},{"id":"e5e2d018c637ec43875e430e3d0b2a7f296ab4c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 20:01:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/4/check/openstack-tox-docs/7993562/html/ : SUCCESS in 4m 09s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/4/check/build-openstack-releasenotes/2717c3b/html/ : SUCCESS in 6m 52s\n- swift-tox-py27 http://logs.openstack.org/93/651893/4/check/swift-tox-py27/6967487/ : SUCCESS in 11m 40s\n- swift-tox-py35 http://logs.openstack.org/93/651893/4/check/swift-tox-py35/c851c22/ : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/4/check/swift-tox-py36/5832599/ : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/4/check/swift-tox-py37/f7f39e6/ : SUCCESS in 11m 40s\n- swift-tox-func http://logs.openstack.org/93/651893/4/check/swift-tox-func/177cb60/ : SUCCESS in 8m 46s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/4/check/swift-tox-func-encryption/9052246/ : SUCCESS in 12m 02s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/4/check/swift-tox-func-domain-remap-staticweb/d3cb008/ : SUCCESS in 12m 16s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/4/check/swift-tox-func-ec/27dc2e5/ : SUCCESS in 8m 34s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api/7529740/ : SUCCESS in 12m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/d9e1f45/ : SUCCESS in 11m 13s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/4/check/swift-probetests-centos-7/842dfd1/ : SUCCESS in 29m 05s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional/0732b00/ : SUCCESS in 30m 10s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional-ipv6/cddbda5/ : SUCCESS in 27m 22s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/4/check/swift-tox-lower-constraints/32155a5/ : FAILURE in 14m 45s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/4/check/openstack-tox-pep8/5e449ba/ : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/4/check/swift-multinode-rolling-upgrade/86b97a8/ : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":4},{"id":"02e1007d6030941a41439b928673c1ecd0c926cb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-29 20:11:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1145d27e75b552614ceb0607e15244402b78b107","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 20:44:50.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/4/check/openstack-tox-docs/efdb8cd/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/4/check/build-openstack-releasenotes/ce4a2b4/html/ : SUCCESS in 6m 28s\n- swift-tox-py27 http://logs.openstack.org/93/651893/4/check/swift-tox-py27/b56bfe4/ : SUCCESS in 10m 55s\n- swift-tox-py35 http://logs.openstack.org/93/651893/4/check/swift-tox-py35/904001a/ : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/4/check/swift-tox-py36/3c382b9/ : SUCCESS in 8m 29s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/4/check/swift-tox-py37/5d8a1c9/ : SUCCESS in 8m 50s\n- swift-tox-func http://logs.openstack.org/93/651893/4/check/swift-tox-func/1c5eca4/ : SUCCESS in 7m 54s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/4/check/swift-tox-func-encryption/e5547bc/ : SUCCESS in 8m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/4/check/swift-tox-func-domain-remap-staticweb/a4738a9/ : SUCCESS in 9m 01s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/4/check/swift-tox-func-ec/1b8cc9a/ : SUCCESS in 10m 06s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api/61e1e25/ : SUCCESS in 9m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c040399/ : SUCCESS in 10m 56s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/4/check/swift-probetests-centos-7/f6c76ec/ : FAILURE in 27m 36s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional/77891fb/ : SUCCESS in 25m 18s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional-ipv6/525fd6e/ : SUCCESS in 25m 30s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/4/check/swift-tox-lower-constraints/d9a3a2b/ : SUCCESS in 11m 09s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/4/check/openstack-tox-pep8/b0b8867/ : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/4/check/swift-multinode-rolling-upgrade/10732ef/ : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":4},{"id":"ad9f17a8ca05e13065094254091a92057edd058c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-29 20:58:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8704040af4eac6852ce758ee323f129040fc1f2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 21:29:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/4/check/openstack-tox-docs/5b6fafc/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/4/check/build-openstack-releasenotes/25e5750/html/ : SUCCESS in 6m 23s\n- swift-tox-py27 http://logs.openstack.org/93/651893/4/check/swift-tox-py27/755dd36/ : SUCCESS in 11m 37s\n- swift-tox-py35 http://logs.openstack.org/93/651893/4/check/swift-tox-py35/ae1ac3b/ : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/4/check/swift-tox-py36/98ad11b/ : SUCCESS in 8m 33s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/4/check/swift-tox-py37/354fac3/ : SUCCESS in 9m 42s\n- swift-tox-func http://logs.openstack.org/93/651893/4/check/swift-tox-func/5508407/ : SUCCESS in 9m 54s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/4/check/swift-tox-func-encryption/67a8918/ : SUCCESS in 9m 20s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/4/check/swift-tox-func-domain-remap-staticweb/0ed4501/ : SUCCESS in 8m 08s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/4/check/swift-tox-func-ec/4258e48/ : SUCCESS in 15m 56s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api/49f5eab/ : SUCCESS in 9m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/74e1189/ : SUCCESS in 10m 40s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/4/check/swift-probetests-centos-7/3838252/ : SUCCESS in 30m 52s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional/aa083e9/ : SUCCESS in 25m 03s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/4/check/swift-dsvm-functional-ipv6/61f38fc/ : SUCCESS in 23m 48s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/4/check/swift-tox-lower-constraints/5c5cd49/ : SUCCESS in 12m 48s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/4/check/openstack-tox-pep8/0d5b2b2/ : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/4/check/swift-multinode-rolling-upgrade/2b34491/ : SUCCESS in 16m 15s","accounts_in_message":[],"_revision_number":4},{"id":"88108a1d12baf01564dff203465d51d81b7ecea6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-02 23:38:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"37010476720597a62920f84198ed37940149be06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 00:15:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/5/check/openstack-tox-docs/e8276a4/html/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/5/check/build-openstack-releasenotes/16d1cf1/html/ : SUCCESS in 6m 17s\n- swift-tox-py27 http://logs.openstack.org/93/651893/5/check/swift-tox-py27/18cd760/ : SUCCESS in 15m 41s\n- swift-tox-py35 http://logs.openstack.org/93/651893/5/check/swift-tox-py35/f66fc73/ : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/93/651893/5/check/swift-tox-py36/43cfe10/ : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/93/651893/5/check/swift-tox-py37/d66c8b9/ : SUCCESS in 7m 55s\n- swift-tox-func http://logs.openstack.org/93/651893/5/check/swift-tox-func/a294efc/ : SUCCESS in 9m 27s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/5/check/swift-tox-func-encryption/34e5db7/ : SUCCESS in 8m 48s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/5/check/swift-tox-func-domain-remap-staticweb/4f0852e/ : SUCCESS in 8m 36s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/5/check/swift-tox-func-ec/be5603e/ : SUCCESS in 14m 11s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/5/check/swift-tox-func-s3api/dac79b7/ : SUCCESS in 9m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/93/651893/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/94b93fa/ : SUCCESS in 13m 46s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/5/check/swift-probetests-centos-7/9cc17bc/ : SUCCESS in 30m 58s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/5/check/swift-dsvm-functional/75a746c/ : SUCCESS in 23m 40s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/5/check/swift-dsvm-functional-ipv6/7ae1e8f/ : SUCCESS in 27m 23s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/5/check/swift-tox-lower-constraints/264e1e7/ : SUCCESS in 11m 28s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/5/check/openstack-tox-pep8/5f9ed5f/ : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/93/651893/5/check/swift-multinode-rolling-upgrade/44b096f/ : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":5},{"id":"77b0c8da81ff6b100e28a5859ed0220ab9a68d89","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 06:20:14.000000000","message":"Patch Set 5:\n\n(4 comments)\n\nRebased and resolved conflicts.","accounts_in_message":[],"_revision_number":5},{"id":"6d6800b6ac8a281855deb37a1de2de58574f0616","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-04 19:59:27.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)\n\nI think a danger of regression is low, so +A over one +2. Buck stops here!","accounts_in_message":[],"_revision_number":5},{"id":"6c3e1e1b17891f5eb6f8abb38b307b69bde1c193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 19:59:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"df63fd41069b9a95fd1125bed153704da8b68f26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:49:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6fb2d59b01bc6e2bef99b2db16f3ec0986da869f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:49:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/651893/5/gate/openstack-tox-docs/b2de3db/html/ : SUCCESS in 3m 41s\n- build-openstack-releasenotes http://logs.openstack.org/93/651893/5/gate/build-openstack-releasenotes/b7431fa/html/ : SUCCESS in 7m 38s\n- swift-tox-py27 http://logs.openstack.org/93/651893/5/gate/swift-tox-py27/51305ee/ : SUCCESS in 9m 45s\n- swift-tox-py37 http://logs.openstack.org/93/651893/5/gate/swift-tox-py37/ca63079/ : SUCCESS in 9m 54s\n- swift-tox-func http://logs.openstack.org/93/651893/5/gate/swift-tox-func/0d242fb/ : SUCCESS in 9m 19s\n- swift-tox-func-encryption http://logs.openstack.org/93/651893/5/gate/swift-tox-func-encryption/5e90212/ : SUCCESS in 12m 45s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/93/651893/5/gate/swift-tox-func-domain-remap-staticweb/32f0261/ : SUCCESS in 8m 16s\n- swift-tox-func-ec http://logs.openstack.org/93/651893/5/gate/swift-tox-func-ec/eec9715/ : SUCCESS in 9m 14s\n- swift-tox-func-s3api http://logs.openstack.org/93/651893/5/gate/swift-tox-func-s3api/2e8b0e0/ : SUCCESS in 9m 38s\n- swift-probetests-centos-7 http://logs.openstack.org/93/651893/5/gate/swift-probetests-centos-7/ff1f82c/ : SUCCESS in 30m 12s\n- swift-dsvm-functional http://logs.openstack.org/93/651893/5/gate/swift-dsvm-functional/5176425/ : SUCCESS in 29m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/93/651893/5/gate/swift-dsvm-functional-ipv6/3da21ca/ : SUCCESS in 26m 46s\n- swift-tox-lower-constraints http://logs.openstack.org/93/651893/5/gate/swift-tox-lower-constraints/2eb501e/ : SUCCESS in 12m 13s\n- openstack-tox-pep8 http://logs.openstack.org/93/651893/5/gate/openstack-tox-pep8/faa541e/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3a9f3f841957f54801c32832356c540dbef69817","revisions":{"fbdb42f8122c1c60e7b2201954a0a19184df7637":{"kind":"REWORK","_number":1,"created":"2019-04-11 17:49:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/651893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/651893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/651893/1"}}},"commit":{"parents":[{"commit":"4f45f28ea50af3d69b6bba11997ff6d14d808fe9","subject":"Merge \"Update openstack repo for libec install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f45f28ea50af3d69b6bba11997ff6d14d808fe9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-09 23:47:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-11 17:49:16.000000000","tz":-420},"subject":"py3: port s3api","message":"py3: port s3api\n\nChange-Id: Ifc35aacb2e45004f74c871f08ff3c52bc57c1463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbdb42f8122c1c60e7b2201954a0a19184df7637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbdb42f8122c1c60e7b2201954a0a19184df7637"}]},"branch":"refs/heads/master"},"8b995ae73d0076563a2c4f5bc5e0c975c8f23be1":{"kind":"REWORK","_number":2,"created":"2019-04-11 18:56:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/651893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/651893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/651893/2"}}},"commit":{"parents":[{"commit":"4f45f28ea50af3d69b6bba11997ff6d14d808fe9","subject":"Merge \"Update openstack repo for libec install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f45f28ea50af3d69b6bba11997ff6d14d808fe9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-09 23:47:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-11 18:55:11.000000000","tz":-420},"subject":"py3: port s3api","message":"py3: port s3api\n\nDrive-by: When passing a list or tuple to swob.Response as an app_iter,\ncheck that it\u0027s full of byte strings.\n\nChange-Id: Ifc35aacb2e45004f74c871f08ff3c52bc57c1463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b995ae73d0076563a2c4f5bc5e0c975c8f23be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b995ae73d0076563a2c4f5bc5e0c975c8f23be1"}]},"branch":"refs/heads/master"},"14e4d30fe097a069145470980c8688e8aa3a6ca3":{"kind":"REWORK","_number":3,"created":"2019-04-11 19:32:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/651893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/651893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/651893/3"}}},"commit":{"parents":[{"commit":"4f45f28ea50af3d69b6bba11997ff6d14d808fe9","subject":"Merge \"Update openstack repo for libec install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f45f28ea50af3d69b6bba11997ff6d14d808fe9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-09 23:47:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-11 19:32:31.000000000","tz":-420},"subject":"py3: port s3api","message":"py3: port s3api\n\nDrive-by: When passing a list or tuple to swob.Response as an app_iter,\ncheck that it\u0027s full of byte strings.\n\nChange-Id: Ifc35aacb2e45004f74c871f08ff3c52bc57c1463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14e4d30fe097a069145470980c8688e8aa3a6ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14e4d30fe097a069145470980c8688e8aa3a6ca3"}]},"branch":"refs/heads/master"},"5b85e936e23087d71ae3b062aa7cfad549533e3d":{"kind":"REWORK","_number":4,"created":"2019-04-24 17:12:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/651893/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/651893/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/651893/4"}}},"commit":{"parents":[{"commit":"4f45f28ea50af3d69b6bba11997ff6d14d808fe9","subject":"Merge \"Update openstack repo for libec install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f45f28ea50af3d69b6bba11997ff6d14d808fe9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-09 23:47:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-24 17:12:11.000000000","tz":-420},"subject":"py3: port s3api","message":"py3: port s3api\n\nDrive-by: When passing a list or tuple to swob.Response as an app_iter,\ncheck that it\u0027s full of byte strings.\n\nChange-Id: Ifc35aacb2e45004f74c871f08ff3c52bc57c1463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b85e936e23087d71ae3b062aa7cfad549533e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b85e936e23087d71ae3b062aa7cfad549533e3d"}]},"branch":"refs/heads/master"},"3a9f3f841957f54801c32832356c540dbef69817":{"kind":"REWORK","_number":5,"created":"2019-05-02 23:38:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/651893/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/651893/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/651893/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/651893/5"}}},"commit":{"parents":[{"commit":"f55167a735632cdc89fb6f83f97e4fe3ed09b7cc","subject":"Merge \"Increase node_timeout in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f55167a735632cdc89fb6f83f97e4fe3ed09b7cc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-09 23:47:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-02 23:38:39.000000000","tz":-420},"subject":"py3: port s3api","message":"py3: port s3api\n\nDrive-by: When passing a list or tuple to swob.Response as an app_iter,\ncheck that it\u0027s full of byte strings.\n\nChange-Id: Ifc35aacb2e45004f74c871f08ff3c52bc57c1463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a9f3f841957f54801c32832356c540dbef69817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a9f3f841957f54801c32832356c540dbef69817"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
